)]}'
{"id":"openstack%2Fglance~929078","triplet_id":"openstack%2Fglance~stable%2F2024.2~I7383dbfe45ba2b721d1282c2497bb3ac64632197","project":"openstack/glance","branch":"stable/2024.2","topic":"create-2024.2","hashtags":[],"change_id":"I7383dbfe45ba2b721d1282c2497bb3ac64632197","subject":"Update .gitreview for stable/2024.2","status":"MERGED","created":"2024-09-12 13:29:23.000000000","updated":"2024-09-16 16:09:07.000000000","submitted":"2024-09-16 16:07:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"929078-create-2024.2","meta_rev_id":"dd5c702ffd85542650fd4194abd626ec0e853999","_number":929078,"virtual_id_number":929078,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-16 16:07:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":2,"date":"2024-09-16 15:12:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":1,"date":"2024-09-16 15:12:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-12 13:57:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-16 15:12:01.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"}],"messages":[{"id":"c8cf67c0a382e387a37eb94d8b867b595ab814e7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2024-09-12 13:29:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"326c229757d23ae22296773980eead0e258eed24","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-12 13:57:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c31ce820736845b19fe8d749671ec48e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5e6415274064c849f99e8b3cca7347e : SUCCESS in 6m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/49d32d053d1448c3898edbc9f2074aa0 : SUCCESS in 7m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/823de19944304cc7bd6bf73504c41387 : SUCCESS in 7m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/96003d43e28548fc9faa37aa3407da79 : SUCCESS in 8m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/629658d5de3f4fa393e7d97b675a0bab : SUCCESS in 6m 01s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5b98e5c76343489eb24947ce9f1caf73 : SUCCESS in 8m 42s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/25064cec04854165af01b582cc9f2962 : SUCCESS in 8m 33s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/3d777dc2224d40ed9b7ca98ed25392df : SUCCESS in 13m 09s","accounts_in_message":[],"_revision_number":1},{"id":"75c3b0c38b6dc129751a2f4749b6fedfbeda8e51","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-12 14:39:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a8913a6842f4467aab719f9bf1da3e42\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6a3ba342cc8f4319938e86557fe883b1 : SUCCESS in 15m 08s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b38d7f6fc86f4479b870b352803b503c : SUCCESS in 11m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"73b3d37eb8bb5c29ef7c1ea124edc666f9cddf15","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-09-16 15:12:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b3c8c2992a048adf6cc725b8954d2f7f74fb7f79","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-16 15:12:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a0190c8b853c8a964ee8afe39f7973be0c21f9d3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-16 16:07:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/628bbc49a18345bfb76a5eb725466995\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c4928139e0342f3a18cc1a78c4023e1 : SUCCESS in 4m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2db57a6f2a28429888d27d4112cfd0ad : SUCCESS in 8m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7abf4c1f764f48be876b6e46da37918b : SUCCESS in 5m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd6440dbf27046e9bc5bafe71792c700 : SUCCESS in 6m 09s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9db98a746dac4fae971ecfa54b601bd2 : SUCCESS in 9m 38s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f68408d91b0f44a38fdffed88ae552e0 : SUCCESS in 9m 44s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/356726d9411e45898b38f251121e9176 : SUCCESS in 14m 11s","accounts_in_message":[],"_revision_number":1},{"id":"36c830eb9dc2000a12af1ca8d3f8fe9d1c7c8921","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-16 16:07:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"dd5c702ffd85542650fd4194abd626ec0e853999","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-16 16:09:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9120de65db41414bad031a29f3509940\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/963d3b8419b64a6db35a70e7476c1c9e : SUCCESS in 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fb6c48bd5954f17d074264f345f4588f5a05b69a","revisions":{"fb6c48bd5954f17d074264f345f4588f5a05b69a":{"kind":"REWORK","_number":1,"created":"2024-09-12 13:29:23.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/78/929078/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/78/929078/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/78/929078/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/78/929078/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/78/929078/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/78/929078/1"}}},"commit":{"parents":[{"commit":"3ea17a13a00c7fa51acce919cf3de77bfa9cf0ee","subject":"Merge \"api-ref: add \u0027compressed\u0027 container_format\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3ea17a13a00c7fa51acce919cf3de77bfa9cf0ee"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2024-09-12 13:29:22.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2024-09-12 13:29:22.000000000","tz":0},"subject":"Update .gitreview for stable/2024.2","message":"Update .gitreview for stable/2024.2\n\nChange-Id: I7383dbfe45ba2b721d1282c2497bb3ac64632197\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fb6c48bd5954f17d074264f345f4588f5a05b69a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fb6c48bd5954f17d074264f345f4588f5a05b69a"}]},"branch":"refs/heads/stable/2024.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
