)]}'
{"id":"openstack%2Fglance~929666","triplet_id":"openstack%2Fglance~unmaintained%2Fyoga~I0554706368e573e11f649c09569f7c21cbc8634b","project":"openstack/glance","branch":"unmaintained/yoga","topic":"bug/2059809","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2025-02-11 07:26:12.000000000","reason":"\u003cGERRIT_ACCOUNT_17685\u003e replied on the change","reason_account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2025-02-12 17:42:34.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-02-11 09:22:39.000000000","reason":"\u003cGERRIT_ACCOUNT_15197\u003e replied on the change","reason_account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}},"hashtags":[],"change_id":"I0554706368e573e11f649c09569f7c21cbc8634b","subject":"Reject unsafe qcow and vmdk files","status":"MERGED","created":"2024-09-17 15:47:28.000000000","updated":"2025-02-12 17:43:31.000000000","submitted":"2025-02-12 17:42:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":28,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"929666-bug/2059809","meta_rev_id":"8de0aeb97cebe9ee0ea1672a1dae71bfc905dd54","_number":929666,"virtual_id_number":929666,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-12 17:42:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2025-02-11 07:26:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2025-02-11 07:26:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-17 15:47:28.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"CC"},{"updated":"2024-09-17 17:40:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-08 11:12:30.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2024-12-13 14:58:53.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"}],"messages":[{"id":"9b5dce14e1f0b0e32be4129f32de7ab5d09f3d44","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-09-17 15:47:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8a476908080f3b1c1c4a561ce78cf473a030251b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-17 17:40:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f98f5881be45404585612ba630340be1\n\n- grenade https://zuul.opendev.org/t/openstack/build/f51ca45daa7d4c1cb9b9fb88e68fa35e : SUCCESS in 1h 15m 07s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/685bac1e035a4e96ad8db72bc8eadcbe : FAILURE in 1h 13m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/23bce75e781840dc9e4afa1b34142866 : SUCCESS in 56m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d58e607d2b0541c7829747931f59e918 : SUCCESS in 4m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6a6c4092026b43f18180331f8d5e4b00 : FAILURE in 6m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f1e6af2ddd7e4d839b2d90ee99cc2441 : FAILURE in 5m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4098580ae1164d7bb3427a709f0ba8ae : FAILURE in 3m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53201117ff7247a38783414a5892a4bb : FAILURE in 3m 34s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/149c6c4085444129b386bedc1fcf6929 : FAILURE in 5m 30s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/b962a423ab1a49119495a8c4b87dfa15 : FAILURE in 4m 16s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/6be0d8607ba4459489183fb1c26b1865 : FAILURE in 1h 37m 15s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/48577054162b4d25906d2353529edb5b : POST_FAILURE in 25m 38s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/79cc9bcb17b94ff4a4d61f21cabd3c1a : POST_FAILURE in 22m 12s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/e30e22f70fa14c62a92818017966fb0a : POST_FAILURE in 21m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8efcec94ca984cc9bc3561fc0335b844 : SUCCESS in 57m 57s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/2bf07b2b75a64a3bacf6907da8b9c4fd : FAILURE in 27m 02s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/0deae423823643d3b93585f3648c24e8 : SUCCESS in 28m 55s","accounts_in_message":[],"_revision_number":1},{"id":"ab3487e5ff6742c27f680514af81dc0a79bda526","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-10-04 09:08:48.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"d641529ad1d4c5718f8a41cbb3442ffe3698eae5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-04 09:09:07.000000000","message":"Patch Set 2: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/glance for 929665,2","accounts_in_message":[],"_revision_number":2},{"id":"674edf0405f5447c7f707d417aef498cc603e831","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-10-07 08:02:19.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"31db27f542be34b4da81560a9a176cdd65e3402a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-07 09:53:28.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/75631e06b8e7477093297adc1e341216\n\n- grenade https://zuul.opendev.org/t/openstack/build/c4637faafc834142adaf663fef8cb0e5 : SUCCESS in 52m 02s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b26de17ff9d045f18563719519b2d6fe : SUCCESS in 1h 12m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6505fb9b09a147d989e96603b9133c2b : SUCCESS in 42m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7dc34454016342a0a2d96275e50d1bad : SUCCESS in 4m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8b1e1bf175f34aa5a62c2a20e2f0755c : SUCCESS in 6m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f655aa7cc6734379af30df8c50b6f077 : SUCCESS in 5m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/26b635adab4944359d41c744951c7ab4 : SUCCESS in 6m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e79117992a3144d7aa447cda8f9451f9 : SUCCESS in 6m 04s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/eaa7dc7462cd4669a01adcc443174164 : SUCCESS in 9m 25s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/fb935189ea5a47c6a2226f7ac70150e4 : SUCCESS in 9m 39s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/8bc92db5160b4351acd9f16c6493ea64 : FAILURE in 1h 39m 43s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/3f90a2ddc42b4cffa7c78461641c2c20 : POST_FAILURE in 31m 28s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/1bca21be953c46af881e328d08fd719d : POST_FAILURE in 21m 27s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/bd7ef33c431f4e6a92afd810a93e49ab : POST_FAILURE in 27m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/18b4fa1980f740e0b4b6ba83d9a47e5b : SUCCESS in 38m 29s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/50216b479891453aa703fe05798e28e2 : FAILURE in 31m 28s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/1b524f9cb00445a8b42851384acd06e5 : SUCCESS in 36m 57s","accounts_in_message":[],"_revision_number":3},{"id":"8d1967bf2601f7f1fd217e52d887a0669d1c6a79","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-10-07 11:28:15.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"98d162bd695fd6b0d7d75aafc673bd25f9551758","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-07 13:36:56.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/10c0820d66944de88a601ccdd4f23f3b\n\n- grenade https://zuul.opendev.org/t/openstack/build/ca9eaeb0502f42b78c3ac728c0071f80 : SUCCESS in 50m 31s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ecf9c58b3c124284b6cb09e28dcdb8cc : SUCCESS in 1h 11m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/eb2c0f5d5a174fddb155f92389aa1658 : SUCCESS in 42m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91cd4c5bb38549d18f6779fc631b174f : SUCCESS in 4m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0e5fc415fe704754b4ce17709bff7211 : SUCCESS in 5m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/805b609baa2e4e4e8d3eac9959b19175 : SUCCESS in 5m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0c8327613b384a8f83ee371a6ff9af3a : SUCCESS in 5m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b7edb3533e948ffaf02e7b4c7fad14d : SUCCESS in 5m 31s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/230c51f2340c4b12b77bf06fe7fdb7b9 : SUCCESS in 9m 00s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/815d2cf5fdf24a53b7a0b4fa2b7eee93 : SUCCESS in 8m 57s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/223784c34a214b3cac4f77ef13f721cc : FAILURE in 1h 21m 43s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/68859f3fd08d4328bd545fd2a7ecf2ab : SUCCESS in 1h 36m 24s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/e4dd5a75e6754e8485ded449233b1a25 : SUCCESS in 2h 01m 52s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/97a8a941f22540b3ba88675237cf1ac2 : FAILURE in 1h 14m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/defd81c19a5f4e9399d519a0486dcf17 : SUCCESS in 58m 46s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c313c0f412684103a1cb3f21b4ba302e : SUCCESS in 41m 04s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/547411ab3fa84a85841089f1c4ab412c : SUCCESS in 19m 54s","accounts_in_message":[],"_revision_number":4},{"id":"4555d1cfc0e1bcc710939627811a7749221066aa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-10-10 09:22:32.000000000","message":"Patch Set 5: Patch Set 4 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"7872e9e818ee11686ac3207700d9ea9498fec13f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-10 11:47:29.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/398bec329eb04a859bcce6393c751014\n\n- grenade https://zuul.opendev.org/t/openstack/build/e96954523a3c457594560cf126487dd0 : SUCCESS in 1h 06m 56s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/12f6d513bf57467cba6193f569a9e1ee : SUCCESS in 1h 22m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/103a98cd0e3a466e83e56b29d7f35e3b : SUCCESS in 39m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/481193a515e04d26bd0807e34ef00e72 : SUCCESS in 6m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/81a3afd9569f4ddeb6a57bcc179d939b : SUCCESS in 5m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4ecfa97f39de45568e370fb1f2859fc5 : SUCCESS in 5m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ed8fbadf08f5487da72fdcc391a287c9 : SUCCESS in 4m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09ef2110ac3a4a63bcbaa1d2adc999f1 : SUCCESS in 5m 34s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f6bef21475b8424bbda112dc3d3052be : SUCCESS in 8m 22s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/110a7b8c008540ac986e0d85ab7f889a : SUCCESS in 10m 08s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/bf2a12a380b34cb18d8c8fb496e2f503 : FAILURE in 1h 32m 14s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/91737aa12697468b81389c309236a3cf : RETRY_LIMIT in 1m 30s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/ca00f0627037475b88acdf0a83237422 : SUCCESS in 1h 31m 31s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/d3712ad3353c452689ad63bc6423ccd7 : FAILURE in 1h 43m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/38b7f373a5064d4881df5a2c502765f4 : SUCCESS in 53m 06s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/99f8f0e52737406397e6d8b702fa1f47 : RETRY_LIMIT in 1m 29s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/8b71058f32da4b808a213256f504db87 : SUCCESS in 29m 30s","accounts_in_message":[],"_revision_number":5},{"id":"b0f3c83e69e8822041aa107f7e96a840984ec09f","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-10-14 13:09:26.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"e5112202f6c343e479a43ccabaeddc42972550ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-14 15:05:48.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3964b63acb264cf3a829932065ee1142\n\n- grenade https://zuul.opendev.org/t/openstack/build/9e3097f3ffe143e38f41dcc11a392961 : SUCCESS in 41m 32s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a4248d5e7c574344928a21ee912e1a36 : SUCCESS in 1h 14m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3eb4fef3b59c48a09188ee9d7a970d9b : SUCCESS in 35m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/757121680b174dd193c796d60bd485fa : SUCCESS in 5m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/43a3cf615f4f47728802db64e7ad2b83 : SUCCESS in 5m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3368b87017584f64a0f1810517be838a : SUCCESS in 6m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/de7e2182b0274143b1fd0433b2a75036 : SUCCESS in 4m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4be453315ab4d5a91941e061080c3b6 : SUCCESS in 5m 41s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1f8b2b34fa2c4d5f9842fe82ac09c90d : SUCCESS in 10m 01s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/545ac93963844144b92af5bd3750ffb2 : SUCCESS in 8m 55s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/c2fb53816a74486c864f4a995a3c7a7a : FAILURE in 1h 13m 31s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/7fe07ac76e924e1cbb3915acf90fa9db : SUCCESS in 1h 33m 35s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/247ed294fe924641a3d8cf74482f1ea7 : SUCCESS in 54m 08s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/31c318b7a73b4cd280e89d3681699361 : FAILURE in 1h 55m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/832d2f8219bb4b46b4ab26eff63e408d : SUCCESS in 50m 55s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/b25d28c21a264587ad8dc549f8ef8c55 : SUCCESS in 40m 30s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/57bf5cf78b6648dfa9bedabd7a238c8a : SUCCESS in 43m 57s","accounts_in_message":[],"_revision_number":5},{"id":"300beeffe049e3b27e7abe9b03520d89d116afaf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-10-15 10:01:49.000000000","message":"Patch Set 6: Patch Set 5 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"b006fc746b05b48db708d0887cb5f01deef1cb5e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-15 11:49:49.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d377620523b24860a0ab7277ca3e4eef\n\n- grenade https://zuul.opendev.org/t/openstack/build/830a3768344b4bec9130513e2370cb4a : SUCCESS in 48m 54s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/cf53d785059d40389c125297deaa7db4 : SUCCESS in 1h 18m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8e0b01c799544778a7199bc07e012ba9 : SUCCESS in 42m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a56ee8e30984304a50d4e0d95f55701 : SUCCESS in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f45cf4e7806c48d6b759634a8199515e : SUCCESS in 5m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/79f840cd12bb4c768782a50b4bdd1a90 : SUCCESS in 4m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c97d3bdfc8ca40d69c39afbf514f6348 : SUCCESS in 5m 02s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/287dd1f7fc4b449682e20d8e578be0fb : SUCCESS in 7m 09s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/5cb330a3580d4c8f9f310da4d0461297 : SUCCESS in 7m 22s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/62c7ef3f179e44608855836cb041e53e : FAILURE in 46m 59s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/45ed97e255eb45fe826b3e6ce025939e : SUCCESS in 57m 26s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/2d48d50e4b6248559e7029ce43ca90b6 : SUCCESS in 1h 37m 20s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/89f9cb387564413786a2aaeead952c94 : FAILURE in 1h 42m 34s\n- glance-multistore-cinder-import-fips https://zuul.opendev.org/t/openstack/build/f15b2e21a59f4e838500cf3be7c25547 : NODE_FAILURE Node(set) request 300-0025453144 failed in 0s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0a16805396724dcfb34fbf713b5544fc : SUCCESS in 52m 03s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/bd902f97b8c64a449f4d8b8e9a76b38a : SUCCESS in 1h 09m 42s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/948b8c2b215f48f592e732e273322671 : SUCCESS in 31m 17s","accounts_in_message":[],"_revision_number":6},{"id":"e402a35c904de20b1937c71b1bf1302f0c23f330","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-10-15 12:39:18.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"a05c6042f3fcdcbbb38f7a376b4697fdcd84d4e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-15 14:33:21.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a3b6693e1b3d4f179299a77e7c6be998\n\n- grenade https://zuul.opendev.org/t/openstack/build/2dccbed23f1d44bca9e33dbad940e2dc : SUCCESS in 59m 16s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f50d97125ff34ac581f2de3cd4bb4b96 : SUCCESS in 1h 26m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/97fc9d9c01a047f49be8a1d6e0cf422e : SUCCESS in 46m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2159030ae2f941c1b7c780de09b73b99 : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d43d202407d5402983bf6a18d0893d47 : SUCCESS in 8m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a074fb3c756842b98d3d3e70483824d4 : SUCCESS in 6m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6944ab2b71074633a16d30c0fc669e56 : SUCCESS in 4m 08s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5493241142ab41209c54e4ec83effd7a : SUCCESS in 9m 21s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/70764a711fd248a6a8ac943bdfb37762 : SUCCESS in 7m 05s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/7806ea9f70ea41a4a488aa1715eee996 : FAILURE in 1h 33m 44s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/b87ec321d5424eac900148205ab199ef : SUCCESS in 1h 39m 43s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/0a57df0678f34277baa228a9540a7e1d : SUCCESS in 58m 08s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/019cb1505c444e1a91098ea12f11ba5b : FAILURE in 1h 52m 53s\n- glance-multistore-cinder-import-fips https://zuul.opendev.org/t/openstack/build/3d35ed86830542ac9d0c1194d44f2234 : NODE_FAILURE Node(set) request 300-0025454290 failed in 0s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/09c3ab38843647fa86504643627b89c4 : SUCCESS in 49m 22s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/b0ad9411b7c147e1bb157215beca980c : SUCCESS in 1h 00m 49s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/7443d013497647e3bc72be7b6eec5f45 : SUCCESS in 34m 44s","accounts_in_message":[],"_revision_number":6},{"id":"6a406b95e45578ed2def92a10eb8c7c3fd2d2096","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-10-16 11:38:07.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"6a17e5120876a5cca88d4aef171a00c107bb332c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-16 13:51:19.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/446ee3bf6b7042f694f401ee712f6698\n\n- grenade https://zuul.opendev.org/t/openstack/build/5f6058ac726a460199e443e91d959ac7 : SUCCESS in 47m 36s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d9ffe87e75d746b4941caae1964bd9f9 : SUCCESS in 1h 12m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/65dfd953f72643c1b03c38d884d425e8 : SUCCESS in 51m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6dab94a82524ec1a7dfcc56b7363d8c : SUCCESS in 4m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/df999d66f1b64077aa77799ddf23e092 : SUCCESS in 5m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9661650787eb4ab6ace9a5cd234fe83b : SUCCESS in 4m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d6711bf94c243669de0bf7d957b5485 : SUCCESS in 5m 40s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/56c664ae1eb0417683a60cfdf985ff02 : SUCCESS in 8m 22s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/d2fbfa44dca44886986b5cc046217058 : SUCCESS in 8m 50s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/f22946114cf544bf8137b4964f9245f2 : FAILURE in 1h 07m 29s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/17c6bb1a9f914fd0a9873dfd5b8ea3f3 : SUCCESS in 1h 36m 49s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/8a6a6d8dcf74430fa5da9655feb81724 : SUCCESS in 1h 35m 12s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/6de3010d9a1b476e889f7d06cad79c06 : TIMED_OUT in 2h 05m 42s\n- glance-multistore-cinder-import-fips https://zuul.opendev.org/t/openstack/build/ffa7d527f5984676b444bd9df9e70c62 : NODE_FAILURE Node(set) request 300-0025463071 failed in 0s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dfd1187b703b44edaef4bc6a29f78051 : SUCCESS in 51m 30s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9f6a8c996b7f443e8988788d1e4f1330 : SUCCESS in 1h 01m 04s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/d4bb918b203f4eb58087ee759e298108 : SUCCESS in 34m 35s","accounts_in_message":[],"_revision_number":6},{"id":"95f412c565374173a8d2ab8a090ba798f189a7a3","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-10-16 18:14:08.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"509e448fcd1b93efd121ccca2e4dd4bf430ddbbd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-16 20:25:56.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cf1d0a6871ff4e7689d0afb81722eda2\n\n- grenade https://zuul.opendev.org/t/openstack/build/136cfac127e1400f85c1b6635d636ceb : SUCCESS in 45m 42s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5c7a6ff93fc444a380aecb75c71998e8 : SUCCESS in 1h 09m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dbc4226a83aa48eab8dd67eeadcc9010 : SUCCESS in 40m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/350cf64ab8f34e4295e7f92a59b26f29 : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8bb075300851420ea77be9a4b4f450f4 : SUCCESS in 5m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a58369c584e44b0aa7da3adf3fdb5439 : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e53ef5db29d74b808a2d72d844247b92 : SUCCESS in 5m 41s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7662b947efb64b1f80ee7d90ce7f7324 : SUCCESS in 9m 10s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/40c4f5d6b9e64ceaabd355594ca9c842 : SUCCESS in 8m 32s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/f3413c56f2b2442284f5168e8cdbb5d0 : FAILURE in 1h 27m 44s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/3270e6b690764c629e559a35fdc3609f : SUCCESS in 1h 34m 14s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/075786385d704b03829c8f0a89e5e9cd : SUCCESS in 1h 42m 47s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/0ae1c19a943540d1a30877e35cffe378 : FAILURE in 2h 03m 49s\n- glance-multistore-cinder-import-fips https://zuul.opendev.org/t/openstack/build/2b1581d7001340848db7064e41fe478c : NODE_FAILURE Node(set) request 300-0025466251 failed in 0s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2b594ac3a18a4c5aa587e1e6f38a7815 : SUCCESS in 58m 30s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/cb29974fb893422b88829144ababe450 : SUCCESS in 1h 11m 19s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/c826a09b01434aeea557c73f56b976f0 : SUCCESS in 21m 23s","accounts_in_message":[],"_revision_number":6},{"id":"dfee0ce10940c16d8f75255ead54496699589fc7","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-10-17 07:14:29.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"e565f9e3b37f975926005265a371c5bca68d7094","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-17 08:46:02.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/54d642b1cda8444788e848c48b1b1e0e\n\n- grenade https://zuul.opendev.org/t/openstack/build/5c92492db6eb466f837a786a1102c3b4 : SUCCESS in 45m 51s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9b58fa6470ad46cd9b7c197c6206387f : SUCCESS in 59m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/34f7a499ef634056b165183ade15562d : SUCCESS in 37m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb06f79881b043c6bd2b27978dbec147 : SUCCESS in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6a42d7fd2d2f4e328572e00589f28edd : SUCCESS in 4m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/be521953abc14a91aee8d0f8d78c60b9 : SUCCESS in 5m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d630ee839a649f4afd7dccdea87b4bc : SUCCESS in 5m 57s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e0a2662bfea349538c2e8fadd65858d5 : SUCCESS in 7m 24s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/c0511706fb3946a9a4d96f732663ebc3 : SUCCESS in 7m 36s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/0c56d1774d694dd481291816b1e93445 : FAILURE in 1h 22m 18s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/edbbec4d55ec49f9bce2014a29525c5b : SUCCESS in 1h 28m 11s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/ff90cb0c09df44af8cd757be7d84f4f1 : SUCCESS in 1h 13m 21s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/4a70884c40bb4ea4b36c77aac437bae1 : FAILURE in 1h 30m 27s\n- glance-multistore-cinder-import-fips https://zuul.opendev.org/t/openstack/build/7471af2542af4b7a99949869f6dd3720 : NODE_FAILURE Node(set) request 300-0025469889 failed in 0s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/759fcce5edcd48119cf42178fbfd6492 : SUCCESS in 42m 55s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/aed57d899e3148b5bd237304479f12af : SUCCESS in 55m 30s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/e86b0be534bd42958f586100d09ebfcd : SUCCESS in 26m 40s","accounts_in_message":[],"_revision_number":6},{"id":"932b682428aba68b763235612dbf22289ebf4039","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-10-17 15:18:11.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"7ea62b6abc3a96aa4b1c2961a3cf511804399895","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-17 17:09:18.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7d29d1e097954734a70c153aa80d21bb\n\n- grenade https://zuul.opendev.org/t/openstack/build/f5e3fb8179cf43af8ff039265e51beab : SUCCESS in 57m 26s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6d8442f79d114b789836e5a67d39c068 : SUCCESS in 1h 17m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4f1d1d316a7e4c6dae8df526a6239309 : SUCCESS in 51m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d271b3caab9f46529ebe649b73700d60 : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f50394cfff6c459fb05765cb72cb32b5 : SUCCESS in 5m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/318771959575471490eb74f74b28f099 : SUCCESS in 5m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b52a0a36a5d4950b80bcd04ff269bd4 : SUCCESS in 5m 30s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/dc0de1ff6708458fab9147c7c4eee41b : SUCCESS in 8m 28s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/363422635b3742d5a4ec5b53e1894750 : SUCCESS in 8m 33s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/b38a857e095344ff87b329aecb5bfeb8 : FAILURE in 1h 18m 49s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/46c815c82dab473cb005b82bb3a3e6e5 : SUCCESS in 1h 38m 11s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/cc310d30ca9c4efaa481d461b9621d99 : SUCCESS in 1h 30m 43s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/f038370a21564991bde327c33f102631 : FAILURE in 1h 45m 34s\n- glance-multistore-cinder-import-fips https://zuul.opendev.org/t/openstack/build/46459cf5a49a41cba78a42b07f46dabc : NODE_FAILURE Node(set) request 300-0025472632 failed in 0s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a68d7c2fb4ae468c8597e7e874ccab94 : SUCCESS in 43m 18s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/45e6ccea31564e27b559f9d7d24db14c : SUCCESS in 44m 25s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/83f4d2de102a4b78a3826f3f5c88d835 : SUCCESS in 22m 25s","accounts_in_message":[],"_revision_number":6},{"id":"bffa2234079760543ebb326a907b599c33cd0785","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-10-18 08:20:33.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"921d556856c817f77fbb20b1fd7b95ee9e097ce4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-18 10:31:31.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2954115a630446d6a381588838212d85\n\n- grenade https://zuul.opendev.org/t/openstack/build/2482eef8c05c4303a1b65526ef6a32e3 : SUCCESS in 1h 02m 09s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/46eb1e4b8f4c48ca9328e96babebd238 : SUCCESS in 1h 18m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2bf8ac6e0c264ad9a1eab0cae03c4100 : SUCCESS in 53m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5443eb24ab4c4a868d36818d64641c00 : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4801ea37a76242629986f0b2fb9295ac : SUCCESS in 5m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/873066d2b0724866a7f1d3a33a92abf6 : SUCCESS in 5m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd4ca29596484f7db7a06a67e64aefae : SUCCESS in 5m 08s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/fe7b60352bb848e6ae67e118d1051c2c : SUCCESS in 8m 21s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/64adf88eb628492c9a2c83aa3da6e0bb : SUCCESS in 10m 02s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/befdd41accbd4b5fbca5a7d7dfd40b70 : FAILURE in 1h 31m 22s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/7916c6ec55544e49bc7552296c054591 : TIMED_OUT in 2h 04m 20s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/1e743339ebda44188ea6f6880e3a3f45 : SUCCESS in 1h 39m 21s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/6209c3b6f17c41e48698eb0b653219ca : FAILURE in 1h 49m 42s\n- glance-multistore-cinder-import-fips https://zuul.opendev.org/t/openstack/build/54ba4f4992774454980cd1cf7edb16dd : NODE_FAILURE Node(set) request 300-0025477820 failed in 0s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2ce1e9046e824b9aa4142d680fba0731 : SUCCESS in 53m 41s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/0da913fb075c4873b0435ba6e7bd57d2 : SUCCESS in 1h 11m 09s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/b0d8947062ed402085849941fdc94d6b : SUCCESS in 38m 33s","accounts_in_message":[],"_revision_number":6},{"id":"5b8fd71ea8bdc0cd0e5700874925bacc0879546b","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-10-18 18:33:11.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"58f9c1f1caf971a3f03bfd5a5ba740bc81835b28","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-18 20:11:23.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e7b3486085e04adcba5ab8f8229b2f98\n\n- grenade https://zuul.opendev.org/t/openstack/build/0254e6b7bb2c4793b642a8fe9f09bbc9 : SUCCESS in 57m 52s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/160348585bfe42768cbc7bf2dba0936d : SUCCESS in 59m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4b1025f74bc845339bdacc64b8c31da5 : SUCCESS in 50m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d36758f7ba4640b9b413328c72f58f8b : SUCCESS in 4m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dedf7270166d43089488b34795402a14 : SUCCESS in 4m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/469c861556af470bb214415782b94a2f : SUCCESS in 5m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ae033419d364008bff3af9310751dd8 : SUCCESS in 7m 26s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/eb04167bd3864d30872382cabeb5e84d : SUCCESS in 9m 34s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/1ed0bdbc1dad465989064ee07d5a510a : SUCCESS in 8m 31s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/f4316a6b0dba4ccda4d8f21fd7002af7 : FAILURE in 1h 03m 59s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/ebc1ba92a0de4c83a3a173948db05edb : SUCCESS in 1h 24m 21s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/c83c506a671b4a72abf0a85f25cf56d9 : SUCCESS in 1h 14m 07s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/af3b8269ae4c4cd695683050aef3e2e3 : FAILURE in 1h 25m 34s\n- glance-multistore-cinder-import-fips https://zuul.opendev.org/t/openstack/build/097c18316de34186a2080a491f05f872 : NODE_FAILURE Node(set) request 300-0025482320 failed in 0s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a3d58c2a526342f4b4d8f421867e77ae : SUCCESS in 43m 54s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/40b67477a9e24447b16eeaaef8cbb0fa : SUCCESS in 1h 07m 31s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/60e3eedd2f2946498d2244d82436182f : SUCCESS in 36m 48s","accounts_in_message":[],"_revision_number":6},{"id":"c4ae79b611422f89c38ab6ebb1d93e49ed492c48","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-10-22 16:35:43.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"ec5ce94cf0b7395b9ab9a6107b80ebeb826ca7f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-22 18:24:00.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b68e9e1c722f4a0ea7a4cf60d90edf9b\n\n- grenade https://zuul.opendev.org/t/openstack/build/2e8560f757794edf830681250a62a50d : SUCCESS in 1h 06m 47s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a6c4beba677f49a892ec080f3272822c : SUCCESS in 56m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/83eecfb7a04e4d08925515b673d0e2ea : SUCCESS in 54m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f0e75fec089428ba19dd34ef60f14ee : SUCCESS in 3m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8b1bed94676b4db0aeb9a995c5cdfe54 : SUCCESS in 6m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6943a53b3518454ba69c241c01eb9c6e : SUCCESS in 3m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5197769533b449199f4d9b413f75f9e2 : SUCCESS in 5m 32s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1a6c8c10e0d7404ba5a810253fd1a74c : SUCCESS in 7m 43s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/ccd54fa6e71c4a9088b1451fa97d5182 : SUCCESS in 9m 04s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/c05531b91cb24ce697ec25d9de3984e3 : FAILURE in 1h 08m 12s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/b89cc3d9a2124f67aabc2b1c00a09b0e : SUCCESS in 1h 35m 07s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/1f7fa0980fc4405d954e56ca1b3f88f3 : SUCCESS in 1h 18m 46s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/b399cede013545978b3ffbfa621d9169 : FAILURE in 1h 46m 12s\n- glance-multistore-cinder-import-fips https://zuul.opendev.org/t/openstack/build/241958335a8b41ddb9945b4494623df6 : NODE_FAILURE Node(set) request 300-0025502025 failed in 0s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1cac8c581f024f8d84afbf8ac42d57b8 : SUCCESS in 33m 58s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ed9af1036c934092a237844d5dffbbe5 : SUCCESS in 45m 04s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/6f6de53522a44e47a72ef08193f966f8 : SUCCESS in 39m 55s","accounts_in_message":[],"_revision_number":6},{"id":"5300d3088b5b885d0d025d8e94eebbca4e4f83a1","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-10-24 13:42:29.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"c4a6937a57517715090831cd8f70556da3927bf7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-24 15:58:01.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/613a236cf0324ad298edd7f5f7036949\n\n- grenade https://zuul.opendev.org/t/openstack/build/c09af5ddfb904dd69813002be474d4b5 : SUCCESS in 1h 01m 09s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ca44db2708d04c7b8d9058d7b7be890a : SUCCESS in 1h 05m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e6bbc2ea9f0249ae995ff83adeedc4bc : SUCCESS in 41m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85641d53e259429580366684cbd6cf37 : SUCCESS in 3m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fdc3fa7053a546cc97996d7edac70933 : SUCCESS in 4m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/96987f61355844efb4dcab3003d7754b : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/752ec138092849c5a2e06b04b1c044f9 : SUCCESS in 4m 00s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2f8e6803610242f2819ba1a348068588 : SUCCESS in 8m 12s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/5f6a7cacebf647b6bce7c04ca5b79140 : SUCCESS in 8m 08s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/78e9055f368b464290be0350cd6ee66e : FAILURE in 1h 17m 13s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/31f28ccedb4845c9850688a36e8e071f : SUCCESS in 1h 24m 09s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/fca7d8f86e794ac6815919416f57557e : SUCCESS in 1h 44m 43s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/21d1a223a8454222b199c352c26b18ae : TIMED_OUT in 2h 03m 22s\n- glance-multistore-cinder-import-fips https://zuul.opendev.org/t/openstack/build/48a622d48d4149a1b2d91101c2ce90e5 : NODE_FAILURE Node(set) request 300-0025519276 failed in 0s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/aaaf4f65537d4646b72db69752906b01 : SUCCESS in 59m 04s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/097ac5535ff24126ba5be7ff09f49f5a : SUCCESS in 1h 16m 01s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/e6334c413fae4c65894c044e4d5a81dc : SUCCESS in 36m 05s","accounts_in_message":[],"_revision_number":6},{"id":"74b10cedb61e23a09f1499099df1081eb28db9b1","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-10-24 16:08:45.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"4570477bfeef05204676c6c1b66b464ca27c352f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-24 18:03:09.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/41659f265f6946c89d6bbbc7ceca82b4\n\n- grenade https://zuul.opendev.org/t/openstack/build/22df8aee698544b1a8c9892f50d3ead2 : SUCCESS in 47m 00s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/69ba582be8ae44f58ad8ad854e9b5c5a : SUCCESS in 1h 22m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/91e7af0a7d24425394e4c0b44b0be3d9 : SUCCESS in 39m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ca0068cc3494706b8d83b944f6551c1 : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6a352514a81742009434b32ac48b5e1b : SUCCESS in 8m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f620e658a9ae4857a92453719edccc91 : SUCCESS in 6m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2707d8d52344c2193de687e6f44b496 : SUCCESS in 6m 46s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d1e8fdf8fa964093959fa596eb4d945a : SUCCESS in 8m 34s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/4367c198a3084018be704787cda25144 : SUCCESS in 8m 42s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/e2295ebebe204774a338c07c57ed4960 : FAILURE in 1h 17m 25s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/4c2fff7475c24bbfb80796a9380d7205 : SUCCESS in 1h 34m 34s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/8d3b1030238942adb8ce0d3e61e9b757 : SUCCESS in 1h 46m 04s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/758e32c9237f4fecabc124c159cd6847 : FAILURE in 1h 42m 44s\n- glance-multistore-cinder-import-fips https://zuul.opendev.org/t/openstack/build/75ae151aeddc45de903d961a8fb98154 : NODE_FAILURE Node(set) request 300-0025520749 failed in 0s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e415b490762f4905becd819d358e0a05 : SUCCESS in 41m 52s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9cc26b44dd9a48adb16444552a1bedbf : SUCCESS in 1h 06m 01s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/b21e65124b2347b8988ddade987ca93f : SUCCESS in 21m 24s","accounts_in_message":[],"_revision_number":6},{"id":"453b380f5628bf21f8c6ada958c410a239ff14b2","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-10-28 07:46:26.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"df47db07b00f3bb59ad6f7601b7fc698fe426e3c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-28 08:26:58.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5fad77dd980e427c93b41cfbb5481a68\n\n- grenade https://zuul.opendev.org/t/openstack/build/a3ba3d5d1f7f4db8a6025fa972928547 : FAILURE in 21m 17s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8c2bf0520b2f4905ac29af8c57c68999 : FAILURE in 19m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/28b5b68407444459ab2ff2e4211e4842 : SUCCESS in 30m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27979297b0fa406aac338effa739c79b : SUCCESS in 3m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2a0b7445f7be4a168704e1fde61644ab : SUCCESS in 5m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/67d17778630e46b292aa6dd386d6aead : SUCCESS in 4m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f329882573f48059cca5565ea655e36 : SUCCESS in 5m 15s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/85711eadb129443d8e3d0726f566d940 : SUCCESS in 7m 39s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/3b29e77d834c4f2385052454bf9e41f5 : SUCCESS in 7m 30s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/7e2f9ebf34de4660a8ff654f95d1b33a : FAILURE in 21m 19s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/2d0a53fd94b14efea80561039a1d7def : FAILURE in 35m 12s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/e4e02e1013c949cc96cc4579ebe55dfb : FAILURE in 16m 42s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/67f2c19cb4824c3e9518b6da355ea069 : FAILURE in 34m 10s\n- glance-multistore-cinder-import-fips https://zuul.opendev.org/t/openstack/build/36a493b0d55147dea7f8aeed6443986e : NODE_FAILURE Node(set) request 300-0025543450 failed in 0s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c85d760548e94284b9752d3c3e0b0259 : FAILURE in 22m 13s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/6b994d6ca7b544c8b0119403487a00b9 : FAILURE in 23m 49s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/5bb28f5154394c7285486c5c0b966ce9 : FAILURE in 21m 56s","accounts_in_message":[],"_revision_number":6},{"id":"471fc94aff1b4a23b13fc1d9e50c287f1d5bd037","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-11-29 09:53:05.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"46787b0f34799ca888f56f2b4c42d0459efd7328","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-29 11:52:18.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/76d65ec3cd474b8d94367ed527ab122a\n\n- grenade https://zuul.opendev.org/t/openstack/build/0c3260633f004bbc97b212447f5bf0af : SUCCESS in 47m 59s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2a2101fed6ff4d42a8ab70b8d731ff03 : SUCCESS in 1h 22m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8c2991c41ff94901b35ffccdb1e83bb7 : SUCCESS in 24m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14a0d2a569b54451ad76cd55113476da : SUCCESS in 5m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f80db7c713f54bc0b9971b96547b0041 : SUCCESS in 3m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e16554ac9c5f4ff69886af4fdec7e902 : SUCCESS in 5m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18c120225e9643278ce0e01d166c9c37 : SUCCESS in 7m 06s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0d0e34aad8834a82be5822399b26accb : SUCCESS in 9m 23s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/d97ec6cc77bf4b86aa48f30d1a0c62a4 : SUCCESS in 8m 10s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/ff733b0afcb448ab9c20f43f8994b338 : FAILURE in 1h 10m 25s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/a3067180f4ab43789e4b58e651f20c98 : SUCCESS in 1h 45m 40s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/b450cd28383e4448a120f49490039661 : SUCCESS in 1h 40m 13s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/ecb4d8e2197e4621a566e858beebc92f : FAILURE in 1h 50m 21s\n- glance-multistore-cinder-import-fips https://zuul.opendev.org/t/openstack/build/022e3276a3bf49c2b64c5b51dac94971 : NODE_FAILURE Node(set) request 300-0025812874 failed in 0s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bfc73e75c7c34112a201870fdb0fbe8a : SUCCESS in 1h 00m 23s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/4fe7204523ce440ab0c62d03ceab0ecc : POST_FAILURE in 1h 11m 17s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/b4ebdd0df2dd49a5a19d1fba99eb3241 : FAILURE in 32m 12s","accounts_in_message":[],"_revision_number":6},{"id":"c50bb43e983c3e03e61cc6775721cfe4f04c2752","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-11-29 16:24:50.000000000","message":"Patch Set 7: Patch Set 6 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"ce7bd5da263318c853a66c08324be61ac9aad8bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-29 18:35:57.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c4742e3e3caa43fbb8fbd86be3de6412\n\n- grenade https://zuul.opendev.org/t/openstack/build/eb489bcb8f9446cd8d4a6c44c5345248 : SUCCESS in 33m 15s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/df44fd6b920c46d7a8373f799da2fa5a : SUCCESS in 1h 06m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b0e998c0e424456ea7d31558f793bb51 : SUCCESS in 42m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7588cbbc345946fbaec83054c40a8d02 : SUCCESS in 4m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/79fab1575b1e4426a9c53aa4d11082b1 : SUCCESS in 4m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d6ca60ee1f994d16b3914f3142a70ed0 : SUCCESS in 3m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/443edc2032e14f74be01df23e2637cad : SUCCESS in 5m 59s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ec2d0e02f5fb4e33a84826970d26a843 : SUCCESS in 7m 27s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/d51f258e4fe742a3be660494f4bc9732 : SUCCESS in 8m 51s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/b55febea3a86415596586d344216ccd0 : FAILURE in 50m 39s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/c0c4ea7c03a545d0808b94bbffaebe12 : SUCCESS in 1h 38m 51s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/8d040016fdc34504a2bfb8ff3b411619 : SUCCESS in 1h 43m 00s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/e3e23ac312d645929ce4b2a5c7709068 : FAILURE in 2h 03m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ca180084b87049a490a8194c89a45547 : SUCCESS in 1h 02m 17s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/6d8f7576f57b4847b8971556477c49ec : SUCCESS in 40m 12s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/d1a2f1a08d7944e0af0e85417bb5c524 : SUCCESS in 38m 19s","accounts_in_message":[],"_revision_number":7},{"id":"3b189f981ba1cfeb883fd36aaf93e34e17716886","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-12-13 14:58:53.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"b6500a1daf7cea116423949a8b784973d455362e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-13 17:03:37.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4b58e6fbc5f84e24883605b7c37e723d\n\n- grenade https://zuul.opendev.org/t/openstack/build/9e22de3a1eee4b55a71c6df6b51187c5 : SUCCESS in 32m 13s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ea24cd22d242476699e094d4453e8e2e : SUCCESS in 1h 24m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4e86ef0dfed246da98bfb9a149b490bc : SUCCESS in 49m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9198a0a84d70454dad1141ab1f69fe05 : SUCCESS in 4m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/234f6266ad464514bd3c81cecee7007e : SUCCESS in 5m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8eb40c4592ae455a8499a9c21c07c9d9 : SUCCESS in 5m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/948ca8d3f5ac43c09289867932a3504c : SUCCESS in 6m 32s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/61830da275a64449a26bafdfd747123a : SUCCESS in 8m 47s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/e5d2363143b24fc3a856734f02a97509 : SUCCESS in 10m 29s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/9b83c2d505f5480f8ff6e24da892e912 : SUCCESS in 1h 31m 18s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/77e5e319a5de4a04a5607ead2d2f95e9 : SUCCESS in 1h 38m 37s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/bc886ce63f7d4cf58720f0375e8e4b05 : FAILURE in 1h 08m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cbfdc027aac54cd4bcc2fc4384370479 : SUCCESS in 44m 43s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/21a5391445474e65a4eee5b087983247 : SUCCESS in 34m 49s","accounts_in_message":[],"_revision_number":7},{"id":"d0618d117399ff189486d26c488aed90dfa2db28","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-12-13 17:10:14.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"b8f983f70ae6f4880d530c4eaee83a9fa1aad39f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-13 18:50:25.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d02f9c9ce1df4a31ae6cac912e83079b\n\n- grenade https://zuul.opendev.org/t/openstack/build/74765b169a7241f692261b8251f72cc7 : SUCCESS in 28m 58s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/02ea836ae47f47038928f4448f424db9 : SUCCESS in 1h 21m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9d477fa9989f425ea97c1f7f68d3f542 : SUCCESS in 23m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/153d3ba13c4c435bb48fad57fd9bd64b : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1887b0bbf28649578e36b7b35f4e269d : SUCCESS in 5m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9ee89a3193704e109067d3b2707b105f : SUCCESS in 7m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d62964889744f2ca457137a8af022d6 : SUCCESS in 5m 03s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/8c524710e9284af89adee623ef7dc60e : SUCCESS in 9m 44s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/b90f78f1c7e94d3aa05dabd91e81be2e : SUCCESS in 8m 10s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/ee657e2e16b24814ac149164be85df73 : SUCCESS in 1h 27m 48s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/dc44ec960a2547618c3d004fb4ecfc3e : SUCCESS in 1h 31m 21s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/5d6bfd4447ee45898e30226094925331 : FAILURE in 1h 19m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e9176e3d56704daea1c9b758bd5f1ef6 : SUCCESS in 33m 54s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/41e8373de3214e74893a7262db96933d : SUCCESS in 21m 24s","accounts_in_message":[],"_revision_number":7},{"id":"8f695584fedb4e454d14fc98009c43dba03ae865","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-12-16 09:06:03.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"0b71c957dadff5e054110e600e21616d31d832c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-16 10:49:12.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/eb145ec1c6794dbc9b35223921bbdd4b\n\n- grenade https://zuul.opendev.org/t/openstack/build/4cbf3772444541cabeb59244bd08de73 : SUCCESS in 59m 57s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2002649403c044b992039e22a551be74 : SUCCESS in 1h 32m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/faf373872086405c991e462719f31dbc : SUCCESS in 58m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/579f552529e34ce1af5cd5653f48b0c7 : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/09a6fca7f6374377a4f4a4b32731b33e : SUCCESS in 5m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/51f896b3e4fc43e2aab1af42d1996845 : SUCCESS in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5f3778b968741a69d5a4b34b8842af5 : SUCCESS in 5m 46s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/39088333e9e64c5285da2e5c44c58373 : SUCCESS in 8m 20s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/b437d1a2555d490e9671d4341edf533f : SUCCESS in 6m 48s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/db6766fa38e14874bf929e9301f3a894 : SUCCESS in 1h 21m 20s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/6e63f9117f074436ab8ceafa008a97ec : SUCCESS in 56m 45s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/1c1f3361632b43c99c4499140a911480 : FAILURE in 1h 06m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e03379297354411a826ff224e53e3fd7 : SUCCESS in 54m 30s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/a2b900a524ae47cd9e6f73090631cb7a : SUCCESS in 40m 17s","accounts_in_message":[],"_revision_number":7},{"id":"9fb6fa0953cd9a0466a399c1cbca69fcdd22d31e","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-01-08 10:58:59.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"5d67ad0205df5fe63d31a80b97b34f1583e3fde5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-08 13:00:03.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e1d0f3d0652f447b8fc565a4ee9569ca\n\n- grenade https://zuul.opendev.org/t/openstack/build/65a90a38876f41708036ec84783b5100 : SUCCESS in 52m 11s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/75238d2aa45d4e9c9b3451e8fe790f9d : FAILURE in 1h 07m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a456e40f1eaf46b887b4f5e9467fd886 : SUCCESS in 37m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8bb7c70b4192492094d716cd096f1710 : SUCCESS in 5m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4ce87533b6a747cd965be6d13dabde40 : SUCCESS in 8m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7991df3d21c04cbd991eb318273c457d : SUCCESS in 6m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fffba395201741d38d882d3da5ef7cda : SUCCESS in 7m 55s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5319423c5b9c4f4da87c0ad608d664e4 : SUCCESS in 9m 35s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/abb82343d58b49c2b1b4946abc11e83c : SUCCESS in 8m 40s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/72ade45088924397a459b59bc96903e6 : SUCCESS in 1h 55m 06s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/26fcc8ef998d4cafbbf549d4b9eabb16 : SUCCESS in 1h 27m 28s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/75ec395c62434e1f85c289675e40162a : FAILURE in 1h 24m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/28866a742c394d12aca66f914edafe6a : SUCCESS in 32m 02s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/dcd00d2d8b9f47aaaeb54d9d4a989b52 : SUCCESS in 30m 29s","accounts_in_message":[],"_revision_number":7},{"id":"4c39871e470e95a9077054c7ed4b5c92eacfc251","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-01-20 22:06:43.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"89cf81908095506b94ef461fda5a56ddbbf76a34","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-21 00:11:20.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bae2759095a1495c9adf576d4ff9cd7a\n\n- grenade https://zuul.opendev.org/t/openstack/build/e345ea719f784d7e857b558cce6e4da2 : SUCCESS in 44m 22s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1ca167b4d64448179856dc52b69c8d6c : SUCCESS in 1h 06m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7a41729f376141a89adf76829b4b6bd0 : SUCCESS in 35m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/add2b7f6237f4a20b98720b1a3c8939b : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ceb968ff0c3243e596d9a5a09c25c25c : SUCCESS in 4m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0c4f5a95bcb648508179afa59ce22845 : SUCCESS in 5m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/030c024d54214b6da674b19169efc728 : SUCCESS in 5m 28s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4897a17e896f417a8fd694ffda3cf67b : SUCCESS in 7m 42s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/1293a37737724aa9bda28cbb82cdc328 : SUCCESS in 8m 09s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/020869e55c344134871230c38d8a4a8f : FAILURE in 1h 38m 58s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/7c1f2ef3a63447b0877def69b4a6d4d9 : SUCCESS in 54m 16s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/990650e6478741638e29868aa189866c : FAILURE in 1h 58m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d87098a329ef4c7f9a227740160f5b21 : SUCCESS in 51m 29s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/b2d50aa540bd4229829b4eefbefa0c03 : SUCCESS in 34m 49s","accounts_in_message":[],"_revision_number":7},{"id":"f6cf61c2fd068fdf4306b091eeb5c1a81a4fcc71","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-01-27 12:46:14.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"1cbbc119829b5bd299d48b41b16f24f3531b4fe4","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-01-27 12:47:18.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"ae4d07b754a2941c28e45c49e20307ddd6eef05a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-27 14:42:26.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a03e186c7f944e14bb14f22543993027\n\n- grenade https://zuul.opendev.org/t/openstack/build/fa7db3a5462245c08abce22cab7a0931 : SUCCESS in 55m 10s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/92126090d1c54341878d8a6e3983b8e5 : SUCCESS in 56m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f722f21bfd4b4aa7b64bd1094d32e76a : SUCCESS in 51m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab962621aff9408297a40e1470dcb015 : SUCCESS in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b17896df7b804970bfc4d7ceba2fcdcf : SUCCESS in 7m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d81018fc1a024436ae7dbeea4573d243 : SUCCESS in 5m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4063ee1cceaf4c95b4ac455a95a7fbf3 : SUCCESS in 7m 13s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1f75c23c43834f519077e43eaf46e6de : SUCCESS in 9m 21s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/282673f9a7d445e9a70f3a4f5302886a : SUCCESS in 9m 16s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/80d3684c8ef14a8d9191961876861498 : SUCCESS in 1h 36m 23s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/3a4be1ee458a4f58a577c88ed3b91ae1 : SUCCESS in 1h 19m 26s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/e1cde66d8fd144168249782effbcea1c : FAILURE in 1h 38m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2df3817531b449f992d38696c7806444 : SUCCESS in 53m 35s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/61ecbbecc697479a8d84e7f16adb9eaa : SUCCESS in 38m 03s","accounts_in_message":[],"_revision_number":7},{"id":"a2642e5c2c76e68b6268cab4ebed45f445a9b8fa","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-02-10 08:48:08.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"9036c55a2005ff97f285358c6651ae366e03446b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-10 10:45:32.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/815141b43417485eb26520cc5a61795c\n\n- grenade https://zuul.opendev.org/t/openstack/build/a8a5196f66f94139aeec8d40ce6efc70 : SUCCESS in 45m 25s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a152c552049c4eec8af6c39ebecbb5d9 : SUCCESS in 1h 15m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e6b18ee4919944139c1ef060f3fc6556 : SUCCESS in 46m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff132fecb3a646c7ac54147b784dcfc5 : SUCCESS in 3m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8a187a3bf680465c9a6d910191404df1 : SUCCESS in 4m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/13c6739005424ee59c2d7cbf242034e2 : SUCCESS in 6m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb574c8710b642218507d52405b842d9 : SUCCESS in 5m 27s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2a26d0f61dc247c1b45f5395351be081 : SUCCESS in 7m 47s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/170218bf8a454784a1a4d257aeb983dc : SUCCESS in 8m 17s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/383992da65bf41668b257e6b717f9af8 : SUCCESS in 1h 35m 27s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/5ae7db8778f24efaa8d9184670e133b2 : SUCCESS in 1h 13m 44s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/8251d5316df54f8c8801b9d45e9086c9 : SUCCESS in 1h 56m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d86d23fa667748ca8b77c22683ea4001 : SUCCESS in 56m 27s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/52bec944ca58451888e3ad9604a303af : SUCCESS in 38m 24s","accounts_in_message":[],"_revision_number":7},{"id":"98b9fd577b092585c6d508bac124753547020335","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-02-10 14:20:09.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"b3a22848d680227c890c4d6aa014b3f55f6d3232","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-02-11 07:26:12.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":7},{"id":"0f9446a124d8789443aa4320e5d469c25b65222e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 07:26:42.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"15970b3f10187d2736d495c7d66ea97a1ec9ee11","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 09:06:48.000000000","message":"Patch Set 7: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f916b9160fc14b3bbf1a01eff7a2bb34\n\n- grenade https://zuul.opendev.org/t/openstack/build/ee7d1ddcb4914d0fb3c76ff6c443d9ff : SUCCESS in 50m 09s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c0d51ba4cf51424f8531b9a9abd0c956 : SUCCESS in 1h 11m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/035e553d157d44659c6d3636ddf6fcc4 : SUCCESS in 35m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/929fffc6bd3f43ae92a0301e7c0575fc : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/41e96896a53f4259a1f5981417817b83 : SUCCESS in 5m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f700a211dafb46a99c248c37228ea8b4 : SUCCESS in 5m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8afa75713fac402f9e451585ed740719 : SUCCESS in 4m 15s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/42b273795f9f45978c06489a33df9d38 : TIMED_OUT in 30m 42s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/09a755cbff7e48c5a952f8b012696fe9 : SUCCESS in 7m 52s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/68612f90cb6e4b50a587dd6810d8ad04 : FAILURE in 1h 12m 35s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/0a175232c2dc4148a9eff42ceb62db96 : SUCCESS in 55m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f51854e0fee54046aa60d050c5d6df98 : SUCCESS in 32m 01s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f9659e69babb46038d8e016ff836d9a5 : SUCCESS in 45m 01s","accounts_in_message":[],"_revision_number":7},{"id":"e47f62b7b2c50ed9068c9f51271ffe7209505140","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-02-11 09:22:39.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"80de1bf1e832793abb854c1bb69b094c6226e506","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 12:37:33.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b8e9ebd484284312a4fd24e13f2fc7a2\n\n- grenade https://zuul.opendev.org/t/openstack/build/fb770f80382541d3815dd237edcf0c15 : SUCCESS in 47m 13s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ab2b65db9d2b40739f47806c71ff0a98 : SUCCESS in 1h 06m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c8dab339ced54a3bbfdebb4e0d7fe85d : SUCCESS in 40m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e928de6443d64824bf6dcc1b24fbf5b6 : SUCCESS in 5m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3a5dfe231d83411b97972df1cfeec767 : SUCCESS in 5m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b44a429d5b0543499c524e01ffd2fb75 : SUCCESS in 5m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e22f75c0a544aefba30dfa0089218a4 : SUCCESS in 23m 43s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2681a033c92041c9b5d85c026e633499 : FAILURE in 7m 57s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/37a0881025bf44ea9a5794b55abe4208 : SUCCESS in 8m 42s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/5dfb9a6ad86f4580a3c925768bd3064c : FAILURE in 1h 19m 32s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/1e26c7c9903c4587a7fe52843b968aca : SUCCESS in 1h 01m 11s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/b5be00242c8547859d2beec31e79e1f2 : SUCCESS in 2h 05m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1aeb905beb1244498ecbff89d052c9ca : SUCCESS in 1h 16m 24s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/16aa1f39f2334dd7b7eb78fd63e85bcc : SUCCESS in 26m 05s","accounts_in_message":[],"_revision_number":7},{"id":"79e349bdf59b6de9d35b0699ef2b5f52fb3f8b58","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-02-12 08:01:27.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"730ce5b83b2deb6f3f0b0404385ded14aae26279","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-12 10:32:18.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/42da07b2eef04aa194a8e874493e7bb0\n\n- grenade https://zuul.opendev.org/t/openstack/build/102a9834901747c3940c726339f158eb : SUCCESS in 1h 01m 44s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ae32c77e153241fbbb904e6e43e917c4 : SUCCESS in 1h 11m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/151952bb96014753ba0fc05689c5fa4e : SUCCESS in 48m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af3243aa907a4e889978a3ebda09c2b6 : SUCCESS in 4m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a2e5f82968bb47dda8542c5572471a12 : SUCCESS in 5m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cadfb3ef47c9466bbfc66e575e4a175a : SUCCESS in 5m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fdf86ea7bb794c4e8856a148ebb91e17 : SUCCESS in 5m 57s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ba58d201b2944d25b983b5a18b7490d4 : SUCCESS in 8m 38s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/e8e1dd08508043f5b6ae110e877d4f5d : SUCCESS in 9m 10s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/1a4d90e6f87c454fb4d651cba7f73d9d : SUCCESS in 1h 35m 39s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/27b15d692add4e5ea48d92d9279c1e67 : SUCCESS in 1h 25m 37s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/a7c9ba50d1e0487c8f4d0fc983d4c8c8 : SUCCESS in 2h 24m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1e1a3c954448411bbec3b7489a7d91b2 : SUCCESS in 50m 30s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/05b72a01ec7448db9e6093ee8393fbe3 : SUCCESS in 38m 29s","accounts_in_message":[],"_revision_number":7},{"id":"9adf8aec82696deb92cd7bce4960aa20735e5005","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-12 10:32:47.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"bea3a55064302be213bba69710182bc75779fdfd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-12 12:12:39.000000000","message":"Patch Set 7: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/463a01ae54f84cd1a5db8ec06916cd87\n\n- grenade https://zuul.opendev.org/t/openstack/build/fb76d6ccd6e34dcd9e5d7e98419ce234 : SUCCESS in 44m 24s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6bd2866eba4149509bc142e4c6c96972 : SUCCESS in 1h 02m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a74f9517755545d18f32bcda04f75933 : SUCCESS in 39m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be317c1828e4446b9f5509de59be1358 : SUCCESS in 4m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d756ac4fd0a04a58b0b1df7da984bd6b : SUCCESS in 4m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/09e13875be8c4c7ba064a7265cb529c1 : SUCCESS in 5m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a5180de5332457599c688fda48397ff : SUCCESS in 5m 21s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e5c1ee9978344bb3afa3cb7a64b6377a : SUCCESS in 8m 57s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/72c0bb0699fe4e07b57ea3de3b9deca2 : FAILURE in 6m 51s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/e3b28646c9364425b647386f3040f6f5 : SUCCESS in 1h 34m 52s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/cfa3c7ebad60423e93345e946624783f : SUCCESS in 52m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e010997be6cc45eb99abc25b1301ed3a : SUCCESS in 54m 48s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/dd992a7126544fc3b5f6c5b4dab8d6be : SUCCESS in 39m 04s","accounts_in_message":[],"_revision_number":7},{"id":"132b7bee7690e6b8a2c0a23631d42bac73840554","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-02-12 13:05:42.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"f6c57a93e8a59d9da296243310df77a915f7165c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-12 15:45:31.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2dcd9ff16c144a4e95faa5ad7144794c\n\n- grenade https://zuul.opendev.org/t/openstack/build/b28a2a4b5f9547cfbf9c0b593fc38b60 : SUCCESS in 45m 57s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b9a32dbe396e40ac97c09dc386362b1b : SUCCESS in 1h 05m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cad47c9599364e8882427b1dec6e2051 : SUCCESS in 25m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a7f9b0d35fd47c8b7ef82ff5b32546f : SUCCESS in 4m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8f0eab5e3caa4796aef916c86c1c0b70 : SUCCESS in 4m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/515e4f80159a4354b801536e41153799 : SUCCESS in 6m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25594738f01a4d0ab1496dfd6346f306 : SUCCESS in 5m 13s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/66a5a2dc422f46e28aaea5a85f11533a : SUCCESS in 10m 13s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/afa1499ecc25418685f40893fcedffaf : SUCCESS in 8m 54s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/782b23e220e340448a514d463e70f3e0 : SUCCESS in 1h 26m 46s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/87f8af0a5aac46ea84bdd738f4e66a7e : SUCCESS in 1h 20m 14s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/ca6ad597464e4a6aaa0b43ee4cd4c619 : SUCCESS in 2h 34m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/71697bf03b624fe0a9e1d262e5d341bc : SUCCESS in 1h 05m 14s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/b5a1729258864c4ba7ffe179db26ded1 : SUCCESS in 39m 25s","accounts_in_message":[],"_revision_number":7},{"id":"401f6124f633cd1cb34df78ac7c5b8cc1b4246a5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-12 15:46:05.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"341f6dd54aef60e6d93209eba6698910e4baa983","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-12 17:42:34.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9f1894ad75384a5abb03ed18e4bc05d0\n\n- grenade https://zuul.opendev.org/t/openstack/build/8f0a833d5293481aa56c9d777164c4f3 : SUCCESS in 52m 33s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3d8114b97acf4d7e891339677829b765 : SUCCESS in 1h 10m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/991cb205ec13447db9d2cebd64ac02ff : SUCCESS in 45m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b29c52bcc83c4f43a0fef705d459ebce : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1be8fdec644e4f229a19cfe71e38974f : SUCCESS in 5m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/11e5f8d6c1d543ecb35636509be08b62 : SUCCESS in 5m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2c6403b51cd4302855f4763145942ed : SUCCESS in 6m 26s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/82f949afb6b44a0889e84e9312f13384 : SUCCESS in 10m 19s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2f0fd94168ca47a5bfe146271be2ce48 : SUCCESS in 9m 34s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/bd53e4f2d4d24e91a1cb07ccaf35b87d : SUCCESS in 1h 50m 50s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/ef02406872c94aaa8350c283548f6afc : SUCCESS in 1h 35m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/79c8fb7719e74171b7b80318e391fe13 : SUCCESS in 48m 32s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/35f75c72e6314533a82061b0f186d2a5 : SUCCESS in 40m 28s","accounts_in_message":[],"_revision_number":7},{"id":"774b1e9a843e7017ded13d32ada17acc075b9c11","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-12 17:42:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"8de0aeb97cebe9ee0ea1672a1dae71bfc905dd54","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-12 17:43:31.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4e0407195a824a5988a2616446711844\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc6905fafe584f5d88bdc4c4f9e8ea04 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"292388520d0c0984607eb9921a751eea29b7dc29","revisions":{"1472aa1b6968faffb64eb5f9d2de9028a4e18845":{"kind":"REWORK","_number":1,"created":"2024-09-17 15:47:28.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/66/929666/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/66/929666/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/929666/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/929666/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/929666/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/66/929666/1"}}},"commit":{"parents":[{"commit":"76339c30770fbf4f67f665741f1c8f385bcb68f8","subject":"Update .gitreview for unmaintained/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/76339c30770fbf4f67f665741f1c8f385bcb68f8"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2024-04-16 18:20:48.000000000","tz":-420},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2024-09-17 15:46:23.000000000","tz":120},"subject":"Reject unsafe qcow and vmdk files","message":"Reject unsafe qcow and vmdk files\n\nThis causes us to use the format inspector to pre-examine qcow and\nvmdk files for safe configurations before even using qemu-img\non them.\n\nConflict: test_image_conversion.py\n\nDepends-On: https://review.opendev.org/c/openstack/glance/+/929665\n\nChange-Id: I0554706368e573e11f649c09569f7c21cbc8634b\nCloses-Bug: #2059809\n(cherry picked from commit a95f335bca1dfdd1c904ae475e9fe8c6806f2c56)\n(cherry picked from commit 55fc42563818fcf88b474233df242a796c918b3a)\n(cherry picked from commit f1f53075d69a9a1c006b3e25506e30eb0210de1f)\n(cherry picked from commit c1c54abeaba91ae0030d4acf01a91339b60a2d7d)\n(cherry picked from commit 63ac35e3248ceed3176a65cdccbbf06f69b3c58e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1472aa1b6968faffb64eb5f9d2de9028a4e18845"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1472aa1b6968faffb64eb5f9d2de9028a4e18845"}]},"branch":"refs/heads/unmaintained/yoga"},"7a3db5dc423c26c6924093e1847747a9997b03b2":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-10-04 09:08:48.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/66/929666/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/66/929666/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/929666/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/929666/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/929666/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/66/929666/2"}}},"commit":{"parents":[{"commit":"8fb607e002926360fc87d1eec4d5078b7699a261","subject":"Adjust integrated-storage-import jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8fb607e002926360fc87d1eec4d5078b7699a261"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2024-04-16 18:20:48.000000000","tz":-420},"committer":{"name":"Lajos Katona","email":"lajos.katona@est.tech","date":"2024-10-04 09:08:48.000000000","tz":0},"subject":"Reject unsafe qcow and vmdk files","message":"Reject unsafe qcow and vmdk files\n\nThis causes us to use the format inspector to pre-examine qcow and\nvmdk files for safe configurations before even using qemu-img\non them.\n\nConflict: test_image_conversion.py\n\nDepends-On: https://review.opendev.org/c/openstack/glance/+/929665\n\nChange-Id: I0554706368e573e11f649c09569f7c21cbc8634b\nCloses-Bug: #2059809\n(cherry picked from commit a95f335bca1dfdd1c904ae475e9fe8c6806f2c56)\n(cherry picked from commit 55fc42563818fcf88b474233df242a796c918b3a)\n(cherry picked from commit f1f53075d69a9a1c006b3e25506e30eb0210de1f)\n(cherry picked from commit c1c54abeaba91ae0030d4acf01a91339b60a2d7d)\n(cherry picked from commit 63ac35e3248ceed3176a65cdccbbf06f69b3c58e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7a3db5dc423c26c6924093e1847747a9997b03b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7a3db5dc423c26c6924093e1847747a9997b03b2"}]},"branch":"refs/heads/unmaintained/yoga","description":"Rebase"},"7bcebfda0ed93f40cb2b91dc47ca316d8fee5bc3":{"kind":"REWORK","_number":3,"created":"2024-10-07 08:02:19.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/66/929666/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/66/929666/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/929666/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/929666/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/929666/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/66/929666/3"}}},"commit":{"parents":[{"commit":"15a483ef742d0488f028be701cb37415753c7a83","subject":"Fix import job to provide valid disk-formats list to tempest","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/15a483ef742d0488f028be701cb37415753c7a83"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2024-04-16 18:20:48.000000000","tz":-420},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2024-10-07 07:59:40.000000000","tz":120},"subject":"Reject unsafe qcow and vmdk files","message":"Reject unsafe qcow and vmdk files\n\nThis causes us to use the format inspector to pre-examine qcow and\nvmdk files for safe configurations before even using qemu-img\non them.\n\nConflict: test_image_conversion.py\nConflict: api_image_import.py has propery image_disk_format in later\nbranches, that is necessary to make imge converter methods work.\n\nChange-Id: I0554706368e573e11f649c09569f7c21cbc8634b\nCloses-Bug: #2059809\n(cherry picked from commit a95f335bca1dfdd1c904ae475e9fe8c6806f2c56)\n(cherry picked from commit 55fc42563818fcf88b474233df242a796c918b3a)\n(cherry picked from commit f1f53075d69a9a1c006b3e25506e30eb0210de1f)\n(cherry picked from commit c1c54abeaba91ae0030d4acf01a91339b60a2d7d)\n(cherry picked from commit 63ac35e3248ceed3176a65cdccbbf06f69b3c58e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7bcebfda0ed93f40cb2b91dc47ca316d8fee5bc3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7bcebfda0ed93f40cb2b91dc47ca316d8fee5bc3"}]},"branch":"refs/heads/unmaintained/yoga"},"237a688d5aa9157cf5658a37f9fec78f2be575f4":{"kind":"REWORK","_number":4,"created":"2024-10-07 11:28:15.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/66/929666/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/66/929666/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/929666/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/929666/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/929666/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/66/929666/4"}}},"commit":{"parents":[{"commit":"d14adb39d80a00b1258ceed90318bcf43e3fc4cd","subject":"Fix import job to provide valid disk-formats list to tempest","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d14adb39d80a00b1258ceed90318bcf43e3fc4cd"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2024-04-16 18:20:48.000000000","tz":-420},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2024-10-07 11:27:38.000000000","tz":120},"subject":"Reject unsafe qcow and vmdk files","message":"Reject unsafe qcow and vmdk files\n\nThis causes us to use the format inspector to pre-examine qcow and\nvmdk files for safe configurations before even using qemu-img\non them.\n\nConflict: test_image_conversion.py\nConflict: api_image_import.py has propery image_disk_format in later\nbranches, that is necessary to make imge converter methods work.\nFrom Ic51c5fd87caf04d38aeaf758ad2d0e2f28098e4d\n\nChange-Id: I0554706368e573e11f649c09569f7c21cbc8634b\nCloses-Bug: #2059809\n(cherry picked from commit a95f335bca1dfdd1c904ae475e9fe8c6806f2c56)\n(cherry picked from commit 55fc42563818fcf88b474233df242a796c918b3a)\n(cherry picked from commit f1f53075d69a9a1c006b3e25506e30eb0210de1f)\n(cherry picked from commit c1c54abeaba91ae0030d4acf01a91339b60a2d7d)\n(cherry picked from commit 63ac35e3248ceed3176a65cdccbbf06f69b3c58e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/237a688d5aa9157cf5658a37f9fec78f2be575f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/237a688d5aa9157cf5658a37f9fec78f2be575f4"}]},"branch":"refs/heads/unmaintained/yoga"},"7ef504cdfa28caeba65932ec9a9724fc464aa975":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2024-10-10 09:22:32.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/66/929666/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/66/929666/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/929666/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/929666/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/929666/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/66/929666/5"}}},"commit":{"parents":[{"commit":"a36cdf5cc898357890f391dbb8042fc018a67062","subject":"Fix import job to provide valid disk-formats list to tempest","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a36cdf5cc898357890f391dbb8042fc018a67062"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2024-04-16 18:20:48.000000000","tz":-420},"committer":{"name":"Lajos Katona","email":"lajos.katona@est.tech","date":"2024-10-10 09:22:32.000000000","tz":0},"subject":"Reject unsafe qcow and vmdk files","message":"Reject unsafe qcow and vmdk files\n\nThis causes us to use the format inspector to pre-examine qcow and\nvmdk files for safe configurations before even using qemu-img\non them.\n\nConflict: test_image_conversion.py\nConflict: api_image_import.py has propery image_disk_format in later\nbranches, that is necessary to make imge converter methods work.\nFrom Ic51c5fd87caf04d38aeaf758ad2d0e2f28098e4d\n\nChange-Id: I0554706368e573e11f649c09569f7c21cbc8634b\nCloses-Bug: #2059809\n(cherry picked from commit a95f335bca1dfdd1c904ae475e9fe8c6806f2c56)\n(cherry picked from commit 55fc42563818fcf88b474233df242a796c918b3a)\n(cherry picked from commit f1f53075d69a9a1c006b3e25506e30eb0210de1f)\n(cherry picked from commit c1c54abeaba91ae0030d4acf01a91339b60a2d7d)\n(cherry picked from commit 63ac35e3248ceed3176a65cdccbbf06f69b3c58e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7ef504cdfa28caeba65932ec9a9724fc464aa975"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7ef504cdfa28caeba65932ec9a9724fc464aa975"}]},"branch":"refs/heads/unmaintained/yoga","description":"Rebase"},"0861e4bb1067c146f927da28880dad28caffce15":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2024-10-15 10:01:49.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/66/929666/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/66/929666/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/929666/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/929666/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/929666/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/66/929666/6"}}},"commit":{"parents":[{"commit":"04a230307d6463f93a12de7e4d66f97ae8f886ac","subject":"Fix import job to provide valid disk-formats list to tempest","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/04a230307d6463f93a12de7e4d66f97ae8f886ac"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2024-04-16 18:20:48.000000000","tz":-420},"committer":{"name":"Lajos Katona","email":"lajos.katona@est.tech","date":"2024-10-15 10:01:49.000000000","tz":0},"subject":"Reject unsafe qcow and vmdk files","message":"Reject unsafe qcow and vmdk files\n\nThis causes us to use the format inspector to pre-examine qcow and\nvmdk files for safe configurations before even using qemu-img\non them.\n\nConflict: test_image_conversion.py\nConflict: api_image_import.py has propery image_disk_format in later\nbranches, that is necessary to make imge converter methods work.\nFrom Ic51c5fd87caf04d38aeaf758ad2d0e2f28098e4d\n\nChange-Id: I0554706368e573e11f649c09569f7c21cbc8634b\nCloses-Bug: #2059809\n(cherry picked from commit a95f335bca1dfdd1c904ae475e9fe8c6806f2c56)\n(cherry picked from commit 55fc42563818fcf88b474233df242a796c918b3a)\n(cherry picked from commit f1f53075d69a9a1c006b3e25506e30eb0210de1f)\n(cherry picked from commit c1c54abeaba91ae0030d4acf01a91339b60a2d7d)\n(cherry picked from commit 63ac35e3248ceed3176a65cdccbbf06f69b3c58e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0861e4bb1067c146f927da28880dad28caffce15"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0861e4bb1067c146f927da28880dad28caffce15"}]},"branch":"refs/heads/unmaintained/yoga","description":"Rebase"},"292388520d0c0984607eb9921a751eea29b7dc29":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2024-11-29 16:24:50.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/66/929666/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/66/929666/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/929666/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/929666/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/929666/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/66/929666/7"}}},"commit":{"parents":[{"commit":"b732c13ee06cd5fe77f13aa153744c6dc16cf53a","subject":"[CI] Fix gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b732c13ee06cd5fe77f13aa153744c6dc16cf53a"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2024-04-16 18:20:48.000000000","tz":-420},"committer":{"name":"Lajos Katona","email":"lajos.katona@est.tech","date":"2024-11-29 16:24:50.000000000","tz":0},"subject":"Reject unsafe qcow and vmdk files","message":"Reject unsafe qcow and vmdk files\n\nThis causes us to use the format inspector to pre-examine qcow and\nvmdk files for safe configurations before even using qemu-img\non them.\n\nConflict: test_image_conversion.py\nConflict: api_image_import.py has propery image_disk_format in later\nbranches, that is necessary to make imge converter methods work.\nFrom Ic51c5fd87caf04d38aeaf758ad2d0e2f28098e4d\n\nChange-Id: I0554706368e573e11f649c09569f7c21cbc8634b\nCloses-Bug: #2059809\n(cherry picked from commit a95f335bca1dfdd1c904ae475e9fe8c6806f2c56)\n(cherry picked from commit 55fc42563818fcf88b474233df242a796c918b3a)\n(cherry picked from commit f1f53075d69a9a1c006b3e25506e30eb0210de1f)\n(cherry picked from commit c1c54abeaba91ae0030d4acf01a91339b60a2d7d)\n(cherry picked from commit 63ac35e3248ceed3176a65cdccbbf06f69b3c58e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/292388520d0c0984607eb9921a751eea29b7dc29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/292388520d0c0984607eb9921a751eea29b7dc29"}]},"branch":"refs/heads/unmaintained/yoga","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
