)]}'
{"id":"openstack%2Fglance~929720","triplet_id":"openstack%2Fglance~master~Id40ceab2a84bb7047dfd130bf8c1ac4c8073b79b","project":"openstack/glance","branch":"master","topic":"bug/2081009","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2024-10-15 17:15:53.000000000","reason":"Change was submitted"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-10-15 17:15:53.000000000","reason":"Change was submitted"},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2024-10-15 14:51:39.000000000","reason":"\u003cGERRIT_ACCOUNT_8122\u003e replied on the change","reason_account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}}},"hashtags":[],"change_id":"Id40ceab2a84bb7047dfd130bf8c1ac4c8073b79b","subject":"Do not call Enforcer.__call__ at module level","status":"MERGED","created":"2024-09-18 04:59:22.000000000","updated":"2024-10-15 17:17:09.000000000","submitted":"2024-10-15 17:15:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"929720-bug/2081009","meta_rev_id":"827daf53e2b2142c06a788d81746e67733d8b8bf","_number":929720,"virtual_id_number":929720,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-15 17:15:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2024-10-08 12:32:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2024-10-15 14:51:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2024-10-15 14:51:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-18 05:14:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-09-18 07:44:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-01 14:31:17.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"CC"},{"updated":"2024-10-08 12:32:14.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2024-10-15 14:51:39.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"}],"messages":[{"id":"12eeefca67e842a9533c134dc518e25b8347764e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-09-18 04:59:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f534dd9cdd77f2a875a933184772ba9ae0565daa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-09-18 05:00:52.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"003ab30ad0d889af72fdfce3cac741b11989f082","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-09-18 05:00:58.000000000","message":"Topic set to bug/2081009","accounts_in_message":[],"_revision_number":2},{"id":"56f797295763613d5501191521bc69235fc8fb30","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-18 05:14:50.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f7f4dfef38a34108875c655ac350128e\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5ea0584258ce48b8841a467a4b3dce4a : SUCCESS in 12m 21s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ed68f50161b544e3b4a14ad43ca10c4d : SUCCESS in 11m 08s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b5848f628ac3404f82ac86197fd1569304e3b638","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-18 07:44:43.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/852ba2cd8a2a49dda70e2fd5af3ee817\n\n- grenade https://zuul.opendev.org/t/openstack/build/092ee55918eb47fbb6eeeea533f78098 : SUCCESS in 41m 26s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d46245d63493477a968fb95b3aa47153 : FAILURE in 1h 11m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2660296816a34fc99f2acfb00dc73520 : SUCCESS in 29m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14a6e00a5954400eb81464631d2051ff : SUCCESS in 6m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/df30ddae21354a45a38d96ad749a6e7b : SUCCESS in 5m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/820aefce19d1426aa801b2fd1ab4d97c : SUCCESS in 5m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e909a55b05bf4a6091f914007e267740 : SUCCESS in 8m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d9b13b10ced44efaa9555f51d24b6c0 : SUCCESS in 5m 11s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9c834ac745dd463981af9033f4679dec : SUCCESS in 8m 43s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/fb498727044c41c6a0f38334fcdb9053 : SUCCESS in 8m 41s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/f328d4f745d848ebac97267c28883262 : SUCCESS in 10m 59s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/764d07c8b21d4a18bbea0b1268093a7a : SUCCESS in 1h 16m 00s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/e4bd82bb2c734936af33bfeaa2981b33 : SUCCESS in 1h 19m 09s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/576733fc45d045428a0954bf08ffd2fe : SUCCESS in 1h 52m 47s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/8a17b58ef63e42f885bdeabdf6a7bc9a : SUCCESS in 1h 18m 12s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/d6f3f5f2d7c040ad9a9c6dccff9d9357 : SUCCESS in 2h 38m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/25243cd9488243c5b2048e17e569e93d : SUCCESS in 43m 02s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9b4c5df767c541778546871bbce613bd : SUCCESS in 1h 32m 35s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/b244357cadda406d84d52717772e8881 : SUCCESS in 1h 50m 09s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/c3f8389c2b504adb9c340e439d19dc92 : SUCCESS in 50m 31s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/22ea6748d3a6470bb0daa317e8c27ef7 : SUCCESS in 24m 53s","accounts_in_message":[],"_revision_number":2},{"id":"ba79bef33e1c38dda970a4a50bedc9689457a96d","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-09-19 02:56:27.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bc94628e84796a3900bfd321127396430b154d1d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-19 03:41:02.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bf421aaac49349a3919bf33491c6a53b\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cd1400a406574396934c803e3da88239 : SUCCESS in 13m 44s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9b8abf573a4c484dbd68ee0f0fe09aff : SUCCESS in 9m 34s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"cbacf7097d57c56ae2a095de6f68c4666213a88b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-19 05:09:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/85770bb8297e4e6eb7df2c697cd3b950\n\n- grenade https://zuul.opendev.org/t/openstack/build/2e560b842dc4431bae75325671739dac : SUCCESS in 53m 10s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/aebd25c458154e1b9ca82dd1c9dd1dd0 : SUCCESS in 1h 26m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ef85859c10f643bc9e604191661be7e4 : SUCCESS in 31m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4810891979a94b198dc96d651d928474 : SUCCESS in 5m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/16f7ae9159504fccb4390408c9675aed : SUCCESS in 5m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/926a9b8b5b934d9a874f2ee3d24f47fa : SUCCESS in 4m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/29dee915af154e2a881e72a376a1baaf : SUCCESS in 5m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ac4bd0ce7ad4e62abd8109a2858d454 : SUCCESS in 16m 13s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/97d289734142479aa986ea9edb9bb1a0 : SUCCESS in 18m 08s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/09fb72f1fda24cf9bc0a93a1a22b2bc5 : SUCCESS in 7m 35s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/770feb43c86d4565a031d7bdd733f6c5 : SUCCESS in 21m 09s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/53af889b4b04420a80c71d22643ec644 : SUCCESS in 1h 51m 17s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/be58be39a5e94642a38ca5cdce480865 : SUCCESS in 2h 01m 24s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/7ce100af5a234cddb5ac7ca9f045a930 : SUCCESS in 2h 07m 00s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/fa962df7fe1243a691e5023ef0f9c4e4 : SUCCESS in 1h 30m 42s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/c6dda23de1824468a41d8703a9d37a44 : SUCCESS in 2h 10m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7891ce19a7024eeda74602e0f04e409f : SUCCESS in 1h 06m 43s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/8158a3c8a64b4957932604c9bf0e5fbd : SUCCESS in 1h 46m 39s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/e21975156786453081098bd7406f049e : SUCCESS in 1h 37m 23s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/80c13c1dd37144ae88f6610baa73e365 : SUCCESS in 50m 53s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/b9d2829cf722475c9b13f66d0954d7be : SUCCESS in 27m 14s","accounts_in_message":[],"_revision_number":2},{"id":"781685ff60c14179f41e654a599c27e246835e76","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-10-01 14:31:17.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c342f78ffa60c5c8d9910c1c1f155cb095df135b","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-01 14:47:09.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7c3e8b9ac98dca56e03c3ceb59513ba312b3ed36","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-10-08 12:32:14.000000000","message":"Patch Set 2: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"e9fc35ed1fd83ae89586f332f7aee1da6fc592af","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-11 15:23:33.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"ee2ea9fa42913774f37d54029366ccbb8b9746d3","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-10-15 14:51:39.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"f98e719c4fdc3135b2a909bacb5bf9726d6311c3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-15 14:52:20.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"817fa91784430b12c344e72b95b915443fd0c02d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-15 17:15:53.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/49551cbd10a0495780c31401a1f92e45\n\n- grenade https://zuul.opendev.org/t/openstack/build/5ddb6ab7545044e7aa424fd0c1b74937 : SUCCESS in 1h 03m 36s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9281e6e235e64a809c82c23c2f4ad383 : SUCCESS in 59m 02s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/95d1f31cf4fc49fca4201c00445ef267 : SUCCESS in 1h 34m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/26853d6648f5451387419feb9d96ecf9 : SUCCESS in 56m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1dcf1f6c8ca64ae0abadb81f308adfcd : SUCCESS in 3m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8890db500ac74e48ae7307d99a6a8a7d : SUCCESS in 6m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1d46b64fac0345cf8d153ebe7f633d4f : SUCCESS in 6m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d5b46a472dc46088d3e00a1c7d539b9 : SUCCESS in 6m 31s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b4c5c327143140ec8525b4d882e227ec : SUCCESS in 9m 17s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/05f0a8a590574e7ab5886e3c99e79a0c : SUCCESS in 9m 31s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/49c14a1858ac489f98058368d92bb278 : SUCCESS in 14m 46s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/e4d5c5d9aa3d4f9daa8cbd95170e8e5c : SUCCESS in 1h 41m 33s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/fe43f616b36645d1b0ff6385bafe7fb1 : SUCCESS in 1h 47m 30s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/3fcdca5040074d83bc890b091f1901de : SUCCESS in 2h 02m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5d08a7c1253e44e898e04fd2226a9f70 : SUCCESS in 48m 26s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/413e4fe5c5e249538ab90c21052f5572 : SUCCESS in 2h 09m 25s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/43d2142272034ce888748040e4fd273a : SUCCESS in 32m 32s","accounts_in_message":[],"_revision_number":2},{"id":"3917a82c2bd5fe7066f3e5c3171fed51e04e5bde","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-15 17:15:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"827daf53e2b2142c06a788d81746e67733d8b8bf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-15 17:17:09.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/676080b762e947d387e41325a5f6625f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03313cae259a4deb928ce79d431d5266 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"562a2eb48b4fc10cf97bd65d9a2f2e6d2a739eba","revisions":{"2870a30de2d9ca61e9971a335c5d9206f5d00599":{"kind":"REWORK","_number":1,"created":"2024-09-18 04:59:22.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/20/929720/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/20/929720/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/20/929720/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/20/929720/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/20/929720/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/20/929720/1"}}},"commit":{"parents":[{"commit":"3ea17a13a00c7fa51acce919cf3de77bfa9cf0ee","subject":"Merge \"api-ref: add \u0027compressed\u0027 container_format\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3ea17a13a00c7fa51acce919cf3de77bfa9cf0ee"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-09-18 04:49:47.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-09-18 04:58:10.000000000","tz":540},"subject":"Do not call Enforcer.__call__ at module level","message":"Do not call Enforcer.__call__ at module level\n\n... because the method may need to user some functionalities which can\nbe used after CONF instance is initialized and module level import\nmakes it difficult tu guarantee the order.\n\nChange-Id: Id40ceab2a84bb7047dfd130bf8c1ac4c8073b79b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2870a30de2d9ca61e9971a335c5d9206f5d00599"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2870a30de2d9ca61e9971a335c5d9206f5d00599"}]},"branch":"refs/heads/master"},"562a2eb48b4fc10cf97bd65d9a2f2e6d2a739eba":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-09-18 05:00:52.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/20/929720/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/20/929720/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/20/929720/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/20/929720/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/20/929720/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/20/929720/2"}}},"commit":{"parents":[{"commit":"3ea17a13a00c7fa51acce919cf3de77bfa9cf0ee","subject":"Merge \"api-ref: add \u0027compressed\u0027 container_format\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3ea17a13a00c7fa51acce919cf3de77bfa9cf0ee"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-09-18 04:49:47.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-09-18 05:00:52.000000000","tz":0},"subject":"Do not call Enforcer.__call__ at module level","message":"Do not call Enforcer.__call__ at module level\n\n... because the method may need to user some functionalities which can\nbe used after CONF instance is initialized and module level import\nmakes it difficult to guarantee the order.\n\nCloses-Bug: #2081009\nChange-Id: Id40ceab2a84bb7047dfd130bf8c1ac4c8073b79b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/562a2eb48b4fc10cf97bd65d9a2f2e6d2a739eba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/562a2eb48b4fc10cf97bd65d9a2f2e6d2a739eba"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
