)]}'
{"id":"openstack%2Fglance~930881","triplet_id":"openstack%2Fglance~master~Ibe2a4edf8cd6943932a46a8edbcf9587461dbb79","project":"openstack/glance","branch":"master","topic":"remove-py38","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-02-13 03:50:39.000000000","reason":"Change was submitted"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-02-13 03:50:39.000000000","reason":"Change was submitted"},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2025-02-13 03:50:39.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ibe2a4edf8cd6943932a46a8edbcf9587461dbb79","subject":"Remove python 3.8 support","status":"MERGED","created":"2024-09-30 08:47:09.000000000","updated":"2025-02-13 03:52:00.000000000","submitted":"2025-02-13 03:50:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"930881-remove-py38","meta_rev_id":"0b6a1e0656353718993aa75177ceaef2fc7f6441","_number":930881,"virtual_id_number":930881,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-13 03:50:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2025-02-13 01:56:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2025-01-08 16:07:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2025-02-13 01:56:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-30 09:03:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-09-30 10:50:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-02 02:30:21.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"CC"},{"updated":"2024-10-04 12:44:50.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"CC"},{"updated":"2025-01-08 16:07:15.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2025-02-13 01:56:46.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"}],"messages":[{"id":"733eb712d568476b1137b6b0ef143809f2c36c1b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-09-30 08:47:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2cef294fd6c4db3805b0db4c741bc8cb857abadf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-30 09:03:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/37f2bd00ff4d4caba0f04e5015539c83\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6e8fe03c4a81414ca12b817720655459 : SUCCESS in 13m 30s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/99a8d83aa4844943a17eee86d7dadc53 : SUCCESS in 14m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"86c499ba0fc04d21132247af5cad6d12d9f0eebe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-30 10:50:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec5e400cd10d4ce393ec10c46ddd9aae\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/28fd142edafc4dbaa4c05d369e770e7b : SUCCESS in 1m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/c04513c672ad48188457fea0b2eef788 : SUCCESS in 50m 53s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/831bd087d6704135b9ce101cc8e45621 : SUCCESS in 51m 12s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/229e5835fe67493185b70e094ac1e9e8 : SUCCESS in 1h 20m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b017e7e13098431b89383909f88f499e : SUCCESS in 17m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de0e252f0cc64099b4c2442b5a4f6a15 : SUCCESS in 3m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6476aa3fd0134bf0a2a56cb7358c8696 : SUCCESS in 5m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/48541c1160b64d4a8a84c8cef45677a5 : SUCCESS in 3m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad1a3e476c7f44cf96c0af96e9c7a6ba : SUCCESS in 6m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9299cc0bcd774b6b88b4defedb00adda : SUCCESS in 4m 28s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7d32c36639694e7f81c8aafd5dc4b4f3 : SUCCESS in 8m 51s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b2b3c9234f0b417487dcf25dcb43c132 : SUCCESS in 8m 06s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/e5de14543aa44aeba02c413880c20779 : SUCCESS in 10m 08s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/de6522ca8b8a4c418e66e0aa24916b4f : SUCCESS in 1h 49m 45s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/5820a365269543eead37d164162dbf5b : SUCCESS in 2h 02m 15s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/4be166ff08de47e4b8c4504c946bd798 : SUCCESS in 1h 55m 43s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/c110a574b1934e6d9030d228debe447c : SUCCESS in 1h 48m 52s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/f1ddf53c3d12416fa3046a739b18180b : SUCCESS in 1h 33m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/62951049be3c4a70a916442ff51e1a8d : SUCCESS in 1h 01m 02s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/adf57268316340afa0622085187c4d40 : SUCCESS in 1h 34m 51s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/93fa7f16aac6462d8c398e7363bd5076 : SUCCESS in 1h 33m 18s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/d2184da1c149457c8d62a96f74f0478a : SUCCESS in 1h 04m 01s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/631e0dab404146bda8b2476703e3bf49 : SUCCESS in 22m 34s","accounts_in_message":[],"_revision_number":1},{"id":"e43cd650c8771793a6b29da063660679791fa5a6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-09-30 11:14:55.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"e8dd3de702f6689d9650c7a33741afddc768480f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-30 11:26:00.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/52d4fd152580449ba86e73ba10099f35\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9d7ee30d8c22462c935ee01f930f321c : SUCCESS in 7m 42s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/91af2c931a6b44a686bb8a1b6cff2576 : SUCCESS in 9m 28s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a479fcb81cdca0c82fe327039a7f152f836e94f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-30 13:20:12.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8c92cb1c4d7a49949fc2c3ca9681a291\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/037c24c761714ee2afef998c4b8c407f : SUCCESS in 1m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/9fbf4d69cc634301b21f581fbd94174c : SUCCESS in 56m 30s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7cbb3fbe37e84509b53fca49931d6da9 : SUCCESS in 38m 24s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ab4ea6f714234e4ba55416e243da5329 : SUCCESS in 1h 22m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/668257be268f46c4ad5ec058257ce134 : SUCCESS in 30m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42eebc514b3c45c6ae0e1cf92b720c42 : SUCCESS in 4m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/29e3664bbedd4280a6e993bf84d4e6b1 : SUCCESS in 4m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/129a92b404694dc6abcf2643dec5b952 : SUCCESS in 5m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8e6675f1d0648e4bd2bb9771ccc184e : SUCCESS in 5m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0abbf0268d564eb0882645ba474d690e : SUCCESS in 6m 38s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e357eb4edf94475fb29ff6f6fbb2c11e : SUCCESS in 9m 51s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f9d229588cb54c89b0565ccad64548fc : SUCCESS in 7m 21s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/4375fce2844a45caadd42732065e4350 : SUCCESS in 11m 37s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/6d2ccceb8f4e4fb7ac32d3004fbf4951 : SUCCESS in 1h 48m 58s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/b2a115547f134267b0d8119d4683bb95 : SUCCESS in 1h 53m 20s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/3dee79e5154f4804bb960963d608e907 : SUCCESS in 1h 58m 14s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/5645e3fa0e514419a3263ac87889868e : SUCCESS in 2h 02m 33s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/5ca7e79c039d449c9408a8c7d534fa07 : FAILURE in 2h 04m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/413e5f2c6978497a9b15c7a1c1ca19dc : SUCCESS in 35m 54s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/4faed27681c94eeda236dcc7a9ce4ca4 : SUCCESS in 1h 31m 44s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/80c439cdaa1b49a4b0170f7d16a58dd8 : SUCCESS in 1h 57m 25s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/45988865a9774b5dbd05fc36c289b42c : SUCCESS in 1h 02m 12s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/3141fb8e4f8a4a61a7517f5603d301e3 : SUCCESS in 24m 36s","accounts_in_message":[],"_revision_number":2},{"id":"9f23701c8db21103c4d533e82464f80f24a89cf6","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-10-02 02:30:21.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2960431a2002088423baf20def700a358d6e9f24","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-02 11:55:04.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"243727a7be99e7bbd5d59a55839f415afdd6f3ec","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-02 12:42:49.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ec1b9f62a7b4a1a95c71c9f18bf759a\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c08c13b796f4469985e5e620aea8024d : SUCCESS in 12m 05s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0b9bf152ad9b4491970b32d6637b22e7 : SUCCESS in 14m 36s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"6d37d886ffa87a387a442ca254675786afd32703","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-02 14:02:29.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/669871e5f7474921885a8ebf9c6076a1\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/6291e7ced9724bf8a192245e4b851dd1 : SUCCESS in 1m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/5a5bbb374ca746a7a6d4873a2c196d5d : SUCCESS in 37m 02s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a65bd6d1ce3c4eed9ccde7656a9e07ca : SUCCESS in 43m 30s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/eb544f58f56b4f08bc399f33da6773e0 : SUCCESS in 1h 39m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/72c987344d7641019075e307cbbde5b0 : SUCCESS in 29m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b513da952563416bb7fd809d824b2022 : SUCCESS in 4m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/388cfc1cbe804d3291e9e25a0923902b : SUCCESS in 5m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f0d435da7095433f8c17a0c08986d0c9 : SUCCESS in 5m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1eaa6aff97d48a78ee94f139564270c : SUCCESS in 3m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/567d7f50994f46eebd4470c6dd185789 : SUCCESS in 6m 37s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/93641d1dcb354cd286543525ad302115 : SUCCESS in 7m 19s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8838422945004042b3bf5004d6a7bfc1 : SUCCESS in 7m 35s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/ec43de66c86e4c2bb761157404e40dfd : SUCCESS in 13m 50s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/99bbd7d43e8743ec876f51ec3597c6a0 : SUCCESS in 1h 51m 09s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/bc26748fcc93470ca08f6f52090f2701 : SUCCESS in 2h 01m 17s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/02cfab05ea3f47499814a476f41de013 : SUCCESS in 1h 56m 16s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/6ea2b514ab424aa09186f7f2b08eea41 : SUCCESS in 1h 18m 05s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/45075e2aa1214b8e8200b905d917395d : SUCCESS in 2h 04m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6d4590a17db747c2a0d01ceae01fe43c : SUCCESS in 38m 59s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9399de098d1c4acb9d3ef675c2c2945f : SUCCESS in 1h 41m 05s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/2ff395863ef94fd58a21ed70b2eb2b92 : SUCCESS in 1h 43m 12s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/9f40e943567d4d4d9a6df592d36ebc49 : SUCCESS in 1h 17m 47s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f23ba748a1db426d9b56c8216efd0cd0 : SUCCESS in 34m 51s","accounts_in_message":[],"_revision_number":3},{"id":"bf7f0f09d99af2027176ac94e449f7ffd7e33059","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-10-04 12:44:50.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"2dac4feb26ae2052c507f528fb14e1b6cbdf35cb","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-24 13:55:45.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"03d8f4dc249fce3303bd8afffc3433fb47f5c487","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-01-08 16:07:15.000000000","message":"Patch Set 3: Code-Review+2\n\nMakes sense, thanks for the patch!","accounts_in_message":[],"_revision_number":3},{"id":"4b7edd04b882e5484077cfb9987fe9f439ce3163","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-02-13 01:56:46.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ba7a5fe8977e1c038582f6eacf84a1c46a328310","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-13 01:57:23.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"f42bc35d22c940f63fe65afb2008509715cd298d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-13 03:50:39.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/65dbba0a029d4f9caba4a5d9262acd0f\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/bb317cb24bc34f05b34aa75824f9608a : SUCCESS in 2m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/25f06911a6e14c0b9b8ce5eff92bae6c : SUCCESS in 56m 05s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/51cac5f8391143f6b6e9622e4cce33e4 : SUCCESS in 1h 02m 43s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9533b70bc93a4e50af9d7f958258ee39 : SUCCESS in 1h 41m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/22a5531b651a48029f8c6a5cf9bc265a : SUCCESS in 1h 08m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2fd2389b374b453d8f0532ae0aabf633 : SUCCESS in 4m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/85608598b44f4cf4badf6f7396a4cda7 : SUCCESS in 5m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4fedfdebb99d4e4291ed04f798f84cd9 : SUCCESS in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/585f1218c0a742059610c334aa30659b : SUCCESS in 6m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9d24baa760cd49669a7a9cb2f365f183 : SUCCESS in 7m 52s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ee42f03508d64d9da167fa9adf8c682c : SUCCESS in 9m 36s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5ef4b127c1984a888a22753e06a81bd5 : SUCCESS in 9m 39s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/812da71b1464467f9ab34f27ac8cb5cf : SUCCESS in 11m 14s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/b8696619d5114ba7b30d0a2955d3c1c3 : SUCCESS in 1h 51m 47s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/7fc64abe2ff94a71a670474323bf2f68 : SUCCESS in 1h 49m 13s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/14093e047bc14f5390c0a89a6a759062 : SUCCESS in 1h 22m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/12faacdc466b474d8d8bfb8d5a242581 : SUCCESS in 1h 07m 46s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/8d9ba2c583ff48b6aaa13f8067b13d3f : SUCCESS in 1h 42m 29s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/b9e5ccaecfbe4d7eab1cee8c478a03a0 : SUCCESS in 26m 44s","accounts_in_message":[],"_revision_number":3},{"id":"123fbbe526ad50dc999a5bbe1f99932bb58f3c84","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-13 03:50:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"0b6a1e0656353718993aa75177ceaef2fc7f6441","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-13 03:52:00.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/16842f0b1f804adeb7eaca53f5674eb4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67a4d53715504d7baee2159ee6c9503c : SUCCESS in 57s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c8683b5efe2e40d69a2f8edbe411d393 : SUCCESS in 52s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"0e7ce177a5434234e6f7783770bf8aabae18e419","revisions":{"f6565b1b26b4ac2c3c9207c999a779417d68636a":{"kind":"REWORK","_number":1,"created":"2024-09-30 08:47:09.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/81/930881/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/81/930881/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/81/930881/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/81/930881/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/81/930881/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/81/930881/1"}}},"commit":{"parents":[{"commit":"1c4aa9d31f93df912f0c7b3d49399b946bdd24d2","subject":"Merge \"Remove workaround for eventlet \u003c0.22.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1c4aa9d31f93df912f0c7b3d49399b946bdd24d2"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-09-30 08:46:53.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-09-30 08:46:53.000000000","tz":540},"subject":"Remove python 3.8 support","message":"Remove python 3.8 support\n\nPython 3.8 is no longer be part of tested runtimes since 2024.2 .\nRemoving support for Python 3.8 allows us to replace deprecated md5\nwrapper from oslo.utils [1] by direct call of hashlib.md5.\n\n[1] https://review.opendev.org/c/openstack/oslo.utils/+/930879\n\nChange-Id: Ibe2a4edf8cd6943932a46a8edbcf9587461dbb79\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f6565b1b26b4ac2c3c9207c999a779417d68636a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f6565b1b26b4ac2c3c9207c999a779417d68636a"}]},"branch":"refs/heads/master"},"0c030eddc36a96911ca7836365026bfc4e1b5411":{"kind":"REWORK","_number":2,"created":"2024-09-30 11:14:55.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/81/930881/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/81/930881/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/81/930881/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/81/930881/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/81/930881/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/81/930881/2"}}},"commit":{"parents":[{"commit":"1c4aa9d31f93df912f0c7b3d49399b946bdd24d2","subject":"Merge \"Remove workaround for eventlet \u003c0.22.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1c4aa9d31f93df912f0c7b3d49399b946bdd24d2"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-09-30 08:46:53.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-09-30 11:14:46.000000000","tz":540},"subject":"Remove python 3.8 support","message":"Remove python 3.8 support\n\nPython 3.8 is no longer be part of tested runtimes since 2024.2 .\nRemoving support for Python 3.8 allows us to replace deprecated md5\nwrapper from oslo.utils [1] by direct call of hashlib.md5.\n\n[1] https://review.opendev.org/c/openstack/oslo.utils/+/930879\n\nChange-Id: Ibe2a4edf8cd6943932a46a8edbcf9587461dbb79\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0c030eddc36a96911ca7836365026bfc4e1b5411"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0c030eddc36a96911ca7836365026bfc4e1b5411"}]},"branch":"refs/heads/master"},"0e7ce177a5434234e6f7783770bf8aabae18e419":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-10-02 11:55:04.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/81/930881/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/81/930881/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/81/930881/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/81/930881/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/81/930881/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/81/930881/3"}}},"commit":{"parents":[{"commit":"13acba7505295cfd57830510618d867a1bb4dd8a","subject":"Merge \"api-ref: Update with links to 2.17\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/13acba7505295cfd57830510618d867a1bb4dd8a"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-09-30 08:46:53.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-02 11:55:04.000000000","tz":0},"subject":"Remove python 3.8 support","message":"Remove python 3.8 support\n\nPython 3.8 is no longer be part of tested runtimes since 2024.2 .\nRemoving support for Python 3.8 allows us to replace deprecated md5\nwrapper from oslo.utils [1] by direct call of hashlib.md5.\n\n[1] https://review.opendev.org/c/openstack/oslo.utils/+/930879\n\nChange-Id: Ibe2a4edf8cd6943932a46a8edbcf9587461dbb79\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0e7ce177a5434234e6f7783770bf8aabae18e419"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0e7ce177a5434234e6f7783770bf8aabae18e419"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
