)]}'
{"id":"openstack%2Fglance~931544","triplet_id":"openstack%2Fglance~master~I509594dd29a8e50ad94b3c53e8f55037b3548add","project":"openstack/glance","branch":"master","topic":"replace_deprecated_datetime_utcnow","attention_set":{"7634":{"account":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"last_update":"2025-06-17 11:29:18.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I509594dd29a8e50ad94b3c53e8f55037b3548add","subject":"Replace deprecated functions in datetime","status":"MERGED","created":"2024-10-05 08:40:33.000000000","updated":"2025-06-17 11:29:18.000000000","submitted":"2024-11-05 08:22:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"931544-replace_deprecated_datetime_utcnow","meta_rev_id":"7c7fe60946d16c7daa78a2ce552e63874c47c6f5","_number":931544,"virtual_id_number":931544,"owner":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-05 08:22:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"recommended":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"all":[{"value":1,"date":"2024-11-05 05:54:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":2,"date":"2024-11-05 05:55:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2024-11-02 14:27:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":1,"date":"2024-11-05 05:55:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}],"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-05 08:53:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-10-05 11:03:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-18 15:07:07.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2024-11-05 05:54:41.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"state":"REVIEWER"},{"updated":"2024-11-05 05:55:21.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2025-06-17 11:29:18.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"}],"messages":[{"id":"554eefa91fd96b99c46c096d48e8c7e5ed079343","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2024-10-05 08:40:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"58a0e61061f6feb09ce9d563c7b75c177333f2fa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-05 08:53:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7eb53b3e096a458889c75f0b5a434d2a\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/574092471a874b2f97cba3e12df87642 : SUCCESS in 9m 48s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/252ea2a14e644e99802fe000147a8c33 : SUCCESS in 11m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8dc0fc0d614096c9eba5de29d01dff9ac499b4de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-05 11:03:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d53e1d9aa254377b34f7f81c311fd2c\n\n- grenade https://zuul.opendev.org/t/openstack/build/26e1c5b3b1c64a2e8039e393751635f1 : SUCCESS in 59m 35s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/92ec5011652f4066b09f72e1361de105 : SUCCESS in 57m 34s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/eb88bf9229d34be18a807b00318b3954 : SUCCESS in 1h 26m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0f49dcbf7cf64c01855965bf82ecc205 : SUCCESS in 57m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ab98d5043eb4991b23d3b287ab5271a : SUCCESS in 5m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/915cc09dea1b44e194fbd0bdca04cc5e : SUCCESS in 4m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cb6de25172054adb8e5c2610562c241f : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f90452a6a93c4e0e936652a6cefa5acd : SUCCESS in 6m 15s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2a6ffd8bd72942f6b9418881467a676e : SUCCESS in 11m 55s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/12bef6012b134701b4b34cf24ed7ea07 : SUCCESS in 8m 22s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/4b2226a4609f43bf969b469bc91a3506 : SUCCESS in 12m 20s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/9e8865baf1b34d0b8bd93dc7da8cfe73 : SUCCESS in 1h 34m 55s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/824c27175cfe44899c387eea2c853890 : SUCCESS in 1h 40m 23s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/402579b5c8a646d287a2dcd294714600 : SUCCESS in 1h 53m 20s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/1059d49712e844e99ec4b2dde2c2ec0f : SUCCESS in 1h 53m 36s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/5a650b6aad884476aaa5a215562c6900 : SUCCESS in 2h 19m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/77641343ca8e4f25a4f05bc798125661 : SUCCESS in 37m 24s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f0210677f31844e7aad64f067efe55d8 : SUCCESS in 1h 38m 29s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/7e649ed1e26b4192823bb76727c065a3 : SUCCESS in 2h 06m 35s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/419f3f59f8744de48d09d9989037540b : SUCCESS in 52m 36s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/eb61985bf5194399961ebb77d9ea6598 : SUCCESS in 19m 39s","accounts_in_message":[],"_revision_number":1},{"id":"2bd78642e466a6a317dc9d79d07b7439366c68df","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2024-10-16 13:11:23.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"57454c2baff1950f3464e60b0a7ef40ed91c8ad6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-16 13:25:50.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d85ac937bfa14288b9766d7421fe2eb9\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/751f113a2ca74051b2f40a50cf117f20 : SUCCESS in 11m 33s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/85e9e2b9a0464da4bc6b30464d76d515 : SUCCESS in 12m 41s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"998d8d7871cad7771ce9a2c66e5562c03d27ffa1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-16 15:19:09.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/21a2521866b1468692341d8a38b5938b\n\n- grenade https://zuul.opendev.org/t/openstack/build/8486b90cd2a3458681709d0472dc7b25 : SUCCESS in 1h 04m 17s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/faaa00320dc64492b76346dfdba2c2b0 : SUCCESS in 1h 09m 33s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0b738d3efee041c8b2f2f1c79e86c3c5 : SUCCESS in 1h 34m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a3ff8725b25247bfbbd1730e76afdb47 : SUCCESS in 1h 04m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/abea72a885894178a86d50b1f473abff : SUCCESS in 4m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ea72a5b7046146549197dcf46aaa50e0 : SUCCESS in 4m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/31db8986370a45b9a45625e9b28448bc : SUCCESS in 5m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5dff9178f34545f3a2dc105bc7c87a3d : SUCCESS in 5m 28s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9a39b30fcf6c47789d3c610be79ba324 : SUCCESS in 8m 51s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/80747a6794904b37843e63437c893ac7 : SUCCESS in 8m 50s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/a11d261d175342fa9c3b3cf436e6cdbb : SUCCESS in 13m 01s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/6d683b2c6aa540ae9e0d515e13d8e6c9 : SUCCESS in 1h 48m 24s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/09e0d85b543e47da9dfdf0d170b4bc23 : SUCCESS in 1h 47m 25s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/06313931a1b24177b836dbd15f30562d : SUCCESS in 1h 48m 17s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/4513d4ee0b1c45bda8530e4778cb6fd4 : SUCCESS in 1h 21m 04s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/bd791f6b984942819d24c72dbc15475f : FAILURE in 1h 54m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a32414b3bfca499cb37b79d4e16d3fb3 : SUCCESS in 58m 06s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c9bfd3552f2540d8bce495f33a8b0985 : SUCCESS in 1h 26m 25s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/91717332e8574495a4b4d49c0ef14814 : SUCCESS in 1h 52m 55s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/3273b4cf88ad4321aabdb863d45fb664 : SUCCESS in 1h 03m 37s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/cc6f44c997064d0989c15cbdb6502fb7 : SUCCESS in 24m 32s","accounts_in_message":[],"_revision_number":2},{"id":"0957d84ef3d963bb01b40d714b60ee01049c3b09","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-10-18 15:07:07.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"43303c469120a2a9a9cd179ce8eac04ae65e5baa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2024-10-25 17:15:15.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"86211b5cb1761fe1ce413e264f1c3a6198cadfe9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-25 17:45:56.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e6f6ce91080d44489947ebdb04d69ca4\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a25932fbf0d548828451138c3ec8456e : SUCCESS in 9m 56s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c824edbe82b24e9c83ccf2156a1f0f59 : SUCCESS in 9m 56s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b3c313e72b4de739fea13085d473c646bd2b723d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-25 20:10:01.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f4482891d03a4122872c4df763f702cc\n\n- grenade https://zuul.opendev.org/t/openstack/build/086aa1ee258e435db5cbb624c3890b9a : SUCCESS in 59m 28s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5280df659b384ff5a4b1d6e9cb299369 : SUCCESS in 1h 00m 24s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4b654e3974b84055b3dacceeea44cf5a : SUCCESS in 1h 40m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2e592f87f79846e9be513cad424069d8 : SUCCESS in 59m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c55a4ab49e2942dca1e9648c1fb8db06 : SUCCESS in 3m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/60fe010b2dfe40078aba51245d60b7ef : SUCCESS in 3m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f42070f6dc814e80b980571009ef4207 : SUCCESS in 5m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4dc0ee07342477ea717f672d7466030 : SUCCESS in 5m 34s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f6211b946db94e3fa738c7f224014d6f : SUCCESS in 7m 48s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/dcfc74b7b96a4ddba1b9d915c23b4bab : SUCCESS in 8m 33s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/0a2dacda1cf94e299f7e916f82f1ed8f : SUCCESS in 9m 57s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/8ae5c6f053ed4f1ca2c6b5b033def210 : SUCCESS in 1h 21m 40s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/39e8f51e3c79462a90289a6184056f83 : SUCCESS in 1h 28m 07s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/6666d5c5533a43258a42c6c42a5dd259 : SUCCESS in 2h 12m 29s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/f2da3e522fe64bd3ad8b6bce8bfdb587 : SUCCESS in 2h 06m 21s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/8e2fb935ec384c50be32d45c855ba57d : FAILURE in 2h 30m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/77afe00b07424c82bf6a1227cf875e19 : SUCCESS in 49m 37s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/52f685f1a55b4210827b15d9a59e41b6 : SUCCESS in 1h 40m 19s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/0dbdbd645bb2426b88e19b18a027eaae : SUCCESS in 1h 56m 38s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/f35e4a987f4e4523acfda9d660ddb609 : SUCCESS in 1h 00m 28s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/8016f57267674d01864c46752477b1bd : SUCCESS in 27m 49s","accounts_in_message":[],"_revision_number":3},{"id":"bd0f4bf78137de9f45c56bcdfffe24da0ccc4b05","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2024-10-30 11:20:21.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"6916166f6ba7bd0e22d055cc8884b665e21c852b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-30 13:13:10.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b55c8a2aceb40089857854e8d19af44\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ac377288f71444f89b6b2d6cf5f8579a : SUCCESS in 15m 26s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ba103ca098254db5978e1e399e7d55e2 : SUCCESS in 14m 53s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"6d02c4aae4ccf30a6cb30602ae36bc53d6bc5b4b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-30 13:58:31.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b7ae65ef8db945efbe44ecafd6965181\n\n- grenade https://zuul.opendev.org/t/openstack/build/a28ae3003d0e41dbb3ffa6934ed840dd : SUCCESS in 49m 10s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/af3e0602274b4df8aa5d7ba4a4ea7b98 : SUCCESS in 59m 39s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8d1d666858e84962bbd985febcbfd9ef : SUCCESS in 1h 23m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ebc4fdfe7d68428eaca30bf297e5c5ed : SUCCESS in 58m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e3fef94c7dd42a68e676c170ec63a06 : SUCCESS in 4m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e0493c678f8a480b8850d4a7742bcbff : SUCCESS in 4m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/86145fd6fdf141c79af0fce2e9182293 : SUCCESS in 6m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ffedb1574d5649d3b512e1f4dcf737ad : SUCCESS in 5m 30s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ad176ac8fce747a6bb5afc4957c723b6 : SUCCESS in 8m 29s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ef196519271540738b39e64cef0e4e46 : FAILURE in 7m 03s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/10f4c2916ff7411590b69163dbdc1aff : SUCCESS in 13m 55s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/ae3b01dbfad94d0ea42f7acca91a5630 : SUCCESS in 1h 50m 03s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/b42f48d180db4cd399d43839db86fc97 : SUCCESS in 1h 44m 22s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/3e8d7236fa9d45d3872f8518d54fa1ac : SUCCESS in 1h 58m 03s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/94b97a14b5c94bee8cb7c00ecc819d0b : POST_FAILURE in 2h 34m 33s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/8d99d34eb814488b9ca71cb249e8f1b4 : FAILURE in 2h 18m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fd9f42d4cbe1439fbe0a24799a466218 : SUCCESS in 45m 05s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/dbbf15fbd749432996199fa02d830aed : SUCCESS in 1h 21m 46s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/9b71494bdfd14da89a112de5a5a6d568 : SUCCESS in 1h 43m 30s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/0ec84ed1f2e94d4496acfd835488ae6e : SUCCESS in 59m 43s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/54bd9f61ab8e4e5dae0e418a2ff6ee51 : SUCCESS in 28m 32s","accounts_in_message":[],"_revision_number":4},{"id":"96e9f4d1871ca71e7c8e3bba167f5f9d4af79c9b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2024-11-02 14:27:49.000000000","message":"Patch Set 5: Patch Set 4 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"be2d02f73c1d3a137dabaa22a3977c21a6917178","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-02 14:38:14.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eb00452b35674e8881dab02dd5e43740\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7775a81fe87e43f59d8560f7cefa8db2 : SUCCESS in 6m 21s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/75cfb04bc1354ac483b256c6b71e40b5 : SUCCESS in 8m 24s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"de2f679c8465eb43e8b03ebcc32204e450e4ba30","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-02 16:39:42.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3cec86a2861643bda1a9b8dd84f01823\n\n- grenade https://zuul.opendev.org/t/openstack/build/8082cdb9b5bf4697aab4f6f78f612c62 : SUCCESS in 1h 02m 49s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1bd95dd1838d4e2488d0ce44bc2d618a : SUCCESS in 1h 05m 31s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/fbed37c49087421b9b0b2b36f0c98475 : SUCCESS in 1h 12m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fddd4db309de4985ae2b2884dac49c52 : SUCCESS in 1h 01m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/635b622e6235455791c9b3d1e7b2f210 : SUCCESS in 4m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6d7a34473d654e3995c6f3c45ca1a146 : SUCCESS in 4m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/67ca2325653143049fcad393631d9a11 : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4c49dc0ffa04df7ae8b341d630f0bba : SUCCESS in 3m 56s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4b8c4b3da52448d798dda81999bace66 : SUCCESS in 7m 55s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/37ae55a09c194510a221c29540fd9d65 : SUCCESS in 7m 57s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/36068b4fcb684c67af773e8cf3103d52 : SUCCESS in 10m 09s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/469cc0e2722d46fc91eef04139500f9c : SUCCESS in 1h 30m 07s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/854a221880394d07bd1be1edbb641920 : SUCCESS in 1h 27m 50s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/4517ac6a063449dca09ab30e1a44b26f : SUCCESS in 1h 40m 22s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/32c79d1a27e742bd95f1cd6576289878 : SUCCESS in 2h 01m 46s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/c75a10105e674370a06e0d06e14afccc : SUCCESS in 2h 08m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2092e1ca69174ade846729c324dcb752 : SUCCESS in 55m 23s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/b8588788672542b99fc5afb2699058de : SUCCESS in 1h 48m 31s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/3e85a870e4b54d10840c59c5efb23e34 : SUCCESS in 1h 47m 30s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/184958b9a8df4693bc281450bd653981 : SUCCESS in 1h 01m 53s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/c6a29ff360934e92981b5c66659b43f8 : SUCCESS in 34m 34s","accounts_in_message":[],"_revision_number":5},{"id":"e47d6165cfcf7c80f375233ced5a668892ca2494","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-11-05 05:54:41.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"f1d769d0157f1ba7e1cd4b2fb5e53394e6136a90","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-11-05 05:55:21.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"af8629920ce64e7f850c456955c1b877bd8d6717","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-05 05:55:58.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"070be1ff5869dd34bfd6d9d5db64b8778983e962","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-05 08:22:07.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b9301eeb15c740a1bfa8580e7d6ab15c\n\n- grenade https://zuul.opendev.org/t/openstack/build/7ed279f366a642d59d755f8bbb85c904 : SUCCESS in 36m 30s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d5651ef2b45742f8b373f187682e89ba : SUCCESS in 52m 08s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8ab6fdebf6a841fbb5a454e679570a00 : SUCCESS in 1h 04m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9f3fcb3a7596426080fcb1bdcfb7fec9 : SUCCESS in 1h 02m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/636310f9919341158f6647450d27f2c4 : SUCCESS in 4m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5b8ceaccdb3748778c6b83c6bb064ab6 : SUCCESS in 5m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4d5dcc0a092e444eacc227b5e8d4d2ae : SUCCESS in 5m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8cb5f6280244607b61713716461a439 : SUCCESS in 5m 40s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/8af6cfa6a5d747dd8813396287171927 : SUCCESS in 7m 55s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/992988b6d8594fbe9d1f566af476db90 : SUCCESS in 7m 50s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/0c724e7ebd704b929faea25b9f6e2ca1 : SUCCESS in 11m 27s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/3aa008e37c1e48899a0b4c0c6120df8e : SUCCESS in 2h 00m 33s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/4ee7d696275b45d18e30fa84def04fc0 : SUCCESS in 2h 18m 45s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/90bfb321d154430bb301f7ffdf992515 : SUCCESS in 2h 03m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1cf7170245b3464987dab07fe0ed4223 : SUCCESS in 59m 48s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/1d962540886946b89afd84a454f29f4c : SUCCESS in 1h 25m 10s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/78f1775499cd46d2b6016da5d991f10b : SUCCESS in 19m 43s","accounts_in_message":[],"_revision_number":5},{"id":"da98c86aaf5bdc9a81122d674a2c2b47bfa0b420","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-05 08:22:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"a11bc21241ba32f57fc6c51ed9b1548fadd6365d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-05 08:23:17.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/970d437510bc4c189069a0c19bf4c88f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2757133fb05f4bbd80463f8832948ffe : SUCCESS in 47s","accounts_in_message":[],"_revision_number":5},{"id":"7c7fe60946d16c7daa78a2ce552e63874c47c6f5","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-06-17 11:29:18.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"1f8a84cdccec4292f6fa8d9ca4f013fe8ebc4735","revisions":{"fa44143e16afbc4f4b80efdebebdf1c134608017":{"kind":"REWORK","_number":1,"created":"2024-10-05 08:40:33.000000000","uploader":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"ref":"refs/changes/44/931544/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/44/931544/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/44/931544/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/44/931544/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/44/931544/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/44/931544/1"}}},"commit":{"parents":[{"commit":"13acba7505295cfd57830510618d867a1bb4dd8a","subject":"Merge \"api-ref: Update with links to 2.17\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/13acba7505295cfd57830510618d867a1bb4dd8a"}]}],"author":{"name":"Takashi Natsume","email":"takanattie@gmail.com","date":"2024-10-05 08:21:32.000000000","tz":540},"committer":{"name":"Takashi Natsume","email":"takanattie@gmail.com","date":"2024-10-05 08:39:52.000000000","tz":540},"subject":"Replace deprecated functions in datetime","message":"Replace deprecated functions in datetime\n\nThe datetime.utcfromtimestamp() and datetime.utcnow()\nare deprecated in Python 3.12.\nReplace datetime.utcfromtimestamp() with datetime.fromtimestamp().\nReplace datetime.utcnow() with oslo_utils.timeutils.utcnow() or\ndatetime.now().\n\nThis removes glance\u0027s in-tree \u0027delta_seconds\u0027 and \u0027utcnow\u0027\nimplementation and replaces the delta_seconds implementation\nwith oslo_utils.timeutils.delta_seconds().\n\nChange-Id: I509594dd29a8e50ad94b3c53e8f55037b3548add\nSigned-off-by: Takashi Natsume \u003ctakanattie@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fa44143e16afbc4f4b80efdebebdf1c134608017"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fa44143e16afbc4f4b80efdebebdf1c134608017"}]},"branch":"refs/heads/master"},"a910bd4c190011f841620e9033a4efd2b0489bd4":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-10-16 13:11:23.000000000","uploader":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"ref":"refs/changes/44/931544/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/44/931544/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/44/931544/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/44/931544/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/44/931544/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/44/931544/2"}}},"commit":{"parents":[{"commit":"2aa5bb356e65709795e036ca636bf46d664298ff","subject":"Merge \"Remove no-longer valid comment about IPv6 workaround\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2aa5bb356e65709795e036ca636bf46d664298ff"}]}],"author":{"name":"Takashi Natsume","email":"takanattie@gmail.com","date":"2024-10-05 08:21:32.000000000","tz":540},"committer":{"name":"Takashi Natsume","email":"takanattie@gmail.com","date":"2024-10-16 13:11:23.000000000","tz":0},"subject":"Replace deprecated functions in datetime","message":"Replace deprecated functions in datetime\n\nThe datetime.utcfromtimestamp() and datetime.utcnow()\nare deprecated in Python 3.12.\nReplace datetime.utcfromtimestamp() with datetime.fromtimestamp().\nReplace datetime.utcnow() with oslo_utils.timeutils.utcnow() or\ndatetime.now().\n\nThis removes glance\u0027s in-tree \u0027delta_seconds\u0027 and \u0027utcnow\u0027\nimplementation and replaces the delta_seconds implementation\nwith oslo_utils.timeutils.delta_seconds().\n\nChange-Id: I509594dd29a8e50ad94b3c53e8f55037b3548add\nSigned-off-by: Takashi Natsume \u003ctakanattie@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a910bd4c190011f841620e9033a4efd2b0489bd4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a910bd4c190011f841620e9033a4efd2b0489bd4"}]},"branch":"refs/heads/master","description":"Rebase"},"45e451f746cb4c05b6a7821b31e70a4c72b88a69":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-10-25 17:15:15.000000000","uploader":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"ref":"refs/changes/44/931544/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/44/931544/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/44/931544/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/44/931544/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/44/931544/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/44/931544/3"}}},"commit":{"parents":[{"commit":"7a98b9f0f6c1df88ea776c31d8c7f8a746559308","subject":"Doc: Add info about multi-attach volume types","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7a98b9f0f6c1df88ea776c31d8c7f8a746559308"}]}],"author":{"name":"Takashi Natsume","email":"takanattie@gmail.com","date":"2024-10-05 08:21:32.000000000","tz":540},"committer":{"name":"Takashi Natsume","email":"takanattie@gmail.com","date":"2024-10-25 17:15:15.000000000","tz":0},"subject":"Replace deprecated functions in datetime","message":"Replace deprecated functions in datetime\n\nThe datetime.utcfromtimestamp() and datetime.utcnow()\nare deprecated in Python 3.12.\nReplace datetime.utcfromtimestamp() with datetime.fromtimestamp().\nReplace datetime.utcnow() with oslo_utils.timeutils.utcnow() or\ndatetime.now().\n\nThis removes glance\u0027s in-tree \u0027delta_seconds\u0027 and \u0027utcnow\u0027\nimplementation and replaces the delta_seconds implementation\nwith oslo_utils.timeutils.delta_seconds().\n\nChange-Id: I509594dd29a8e50ad94b3c53e8f55037b3548add\nSigned-off-by: Takashi Natsume \u003ctakanattie@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/45e451f746cb4c05b6a7821b31e70a4c72b88a69"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/45e451f746cb4c05b6a7821b31e70a4c72b88a69"}]},"branch":"refs/heads/master","description":"Rebase"},"641e7208bce8651f0cbb0d431ed5b5e6e2c6a7e3":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-10-30 11:20:21.000000000","uploader":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"ref":"refs/changes/44/931544/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/44/931544/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/44/931544/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/44/931544/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/44/931544/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/44/931544/4"}}},"commit":{"parents":[{"commit":"704b24fd67b6d3dee5f1f42f9b793832b4b47bf4","subject":"Increase per project quota to 50","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/704b24fd67b6d3dee5f1f42f9b793832b4b47bf4"}]}],"author":{"name":"Takashi Natsume","email":"takanattie@gmail.com","date":"2024-10-05 08:21:32.000000000","tz":540},"committer":{"name":"Takashi Natsume","email":"takanattie@gmail.com","date":"2024-10-30 11:20:21.000000000","tz":0},"subject":"Replace deprecated functions in datetime","message":"Replace deprecated functions in datetime\n\nThe datetime.utcfromtimestamp() and datetime.utcnow()\nare deprecated in Python 3.12.\nReplace datetime.utcfromtimestamp() with datetime.fromtimestamp().\nReplace datetime.utcnow() with oslo_utils.timeutils.utcnow() or\ndatetime.now().\n\nThis removes glance\u0027s in-tree \u0027delta_seconds\u0027 and \u0027utcnow\u0027\nimplementation and replaces the delta_seconds implementation\nwith oslo_utils.timeutils.delta_seconds().\n\nChange-Id: I509594dd29a8e50ad94b3c53e8f55037b3548add\nSigned-off-by: Takashi Natsume \u003ctakanattie@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/641e7208bce8651f0cbb0d431ed5b5e6e2c6a7e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/641e7208bce8651f0cbb0d431ed5b5e6e2c6a7e3"}]},"branch":"refs/heads/master","description":"Rebase"},"1f8a84cdccec4292f6fa8d9ca4f013fe8ebc4735":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2024-11-02 14:27:49.000000000","uploader":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"ref":"refs/changes/44/931544/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/44/931544/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/44/931544/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/44/931544/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/44/931544/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/44/931544/5"}}},"commit":{"parents":[{"commit":"acab9351a1fa124f2bdeb58d2f407070071dc9d2","subject":"Remove Windows OS support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/acab9351a1fa124f2bdeb58d2f407070071dc9d2"}]}],"author":{"name":"Takashi Natsume","email":"takanattie@gmail.com","date":"2024-10-05 08:21:32.000000000","tz":540},"committer":{"name":"Takashi Natsume","email":"takanattie@gmail.com","date":"2024-11-02 14:27:49.000000000","tz":0},"subject":"Replace deprecated functions in datetime","message":"Replace deprecated functions in datetime\n\nThe datetime.utcfromtimestamp() and datetime.utcnow()\nare deprecated in Python 3.12.\nReplace datetime.utcfromtimestamp() with datetime.fromtimestamp().\nReplace datetime.utcnow() with oslo_utils.timeutils.utcnow() or\ndatetime.now().\n\nThis removes glance\u0027s in-tree \u0027delta_seconds\u0027 and \u0027utcnow\u0027\nimplementation and replaces the delta_seconds implementation\nwith oslo_utils.timeutils.delta_seconds().\n\nChange-Id: I509594dd29a8e50ad94b3c53e8f55037b3548add\nSigned-off-by: Takashi Natsume \u003ctakanattie@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1f8a84cdccec4292f6fa8d9ca4f013fe8ebc4735"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1f8a84cdccec4292f6fa8d9ca4f013fe8ebc4735"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
