)]}'
{"id":"openstack%2Fglance~932630","triplet_id":"openstack%2Fglance~master~I8d992b88043d80249ca6e246ed3068a4e6a8b54f","project":"openstack/glance","branch":"master","topic":"remove-wsgi_scripts","attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2024-10-18 14:55:31.000000000","reason":"\u003cGERRIT_ACCOUNT_8122\u003e replied on the change","reason_account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}}},"removed_from_attention_set":{"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2024-10-18 14:55:23.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I8d992b88043d80249ca6e246ed3068a4e6a8b54f","subject":"Add wsgi_api.py file","status":"ABANDONED","created":"2024-10-17 15:41:49.000000000","updated":"2024-10-18 14:55:31.000000000","total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"935a1c4dbbe8123ef299325cdcad4c8e71c1fc52","_number":932630,"virtual_id_number":932630,"owner":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-10-17 17:49:46.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-17 15:49:40.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2024-10-17 16:23:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-10-17 17:49:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"97079ad228ccaf47f77326116824fc077fe2a8fb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-10-17 15:41:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6e89fd35463aea1dacea20eaf34fde9360759067","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-10-17 15:49:40.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"075e930523ed6f6781f40e6f0eec6139b3a059fe","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-17 16:23:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f78ed9ae09d4da490ac6c9bb9a9de81\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/907519eddb0445fba08f934418e4f450 : SUCCESS in 11m 01s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9ea653c382e342898aa0cb12613e8c3f : SUCCESS in 14m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"61493d0bed63533c7bb26c1784121370c6cd434f","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-10-17 16:49:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dd87a69d337c6edd8ee5ec0c1af9f0b4fc2d2942","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-10-17 16:50:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3ad4170f22f094097b745ebab570490a75143f00","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-17 17:49:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/386d7fbcda2e4ecb96ebc84739099e04\n\n- grenade https://zuul.opendev.org/t/openstack/build/aea43a22ae224affbc816c9895ae79d4 : SUCCESS in 49m 59s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ddc14e476eed445c92b72db3091ef040 : SUCCESS in 46m 03s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4ca4a51deccd48398fd1d927b9295640 : SUCCESS in 1h 17m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bb3777c2b5964261b3c28200d427935a : SUCCESS in 45m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a4f836494fe4c0ea80b0f50b396b669 : SUCCESS in 4m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e62a25e749de463fba5625b0a2d704c8 : SUCCESS in 4m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f5f31e5627cd4dc8b6b8bcbf5ae2f42a : SUCCESS in 5m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c3fb410886c4130815b1c60753670df : SUCCESS in 4m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f301a6d38f7048b7860a774ba59d0bfc : SUCCESS in 6m 18s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/801bfc5f34af493e82c96818ecb0483d : SUCCESS in 8m 16s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/487227b0cf924f85aa8226ec59c716e8 : SUCCESS in 7m 51s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/496b52987e3747488cece21f24fabd79 : SUCCESS in 15m 35s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/a0558ae3924b4c31a6f5126d8019cda8 : SUCCESS in 1h 35m 23s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/c36d06b0307f403d93781c2c62ce7578 : SUCCESS in 2h 07m 14s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/8eca219ae41e4e22af7971fba3955234 : SUCCESS in 1h 20m 31s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/413e054da36c465ea820bef18035bfc7 : SUCCESS in 1h 44m 53s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/80c6afa1d3f348aa8153da17ef0c687b : FAILURE in 1h 58m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e831e2d03f564d51b6739e86c61247cb : SUCCESS in 47m 38s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/653a2e45ff4e459c8b6118af2e3e2ace : SUCCESS in 1h 50m 34s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/fb17e91e12ca46a190655cd207177598 : SUCCESS in 1h 45m 03s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/0c43d3d18c204e229fbe10208cfe5703 : SUCCESS in 1h 00m 03s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/d3193b4d5db24d779a408ee23bb1cccd : SUCCESS in 21m 45s","accounts_in_message":[],"_revision_number":1},{"id":"383d9ff538b47dc5b627fa926dd1eb6a64c39102","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-10-17 20:04:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"990ad20f18aaed00888da2fae4cdc923cc775751","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-10-18 12:42:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8ee5b048c21e53677b266a4bba5aee42f98515cb","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-10-18 12:46:06.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"faca17bbc0c16f0a768cf44f5564304aa3036002","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-10-18 14:55:23.000000000","message":"Abandoned\n\nIn favor of https://review.opendev.org/c/openstack/glance/+/932701","accounts_in_message":[],"_revision_number":1},{"id":"935a1c4dbbe8123ef299325cdcad4c8e71c1fc52","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-10-18 14:55:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b27ac643aa14ec7ce987eb55f4fb8e69aeb3c9e1","revisions":{"b27ac643aa14ec7ce987eb55f4fb8e69aeb3c9e1":{"kind":"REWORK","_number":1,"created":"2024-10-17 15:41:49.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/30/932630/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/30/932630/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/30/932630/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/30/932630/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/30/932630/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/30/932630/1"}}},"commit":{"parents":[{"commit":"2aa5bb356e65709795e036ca636bf46d664298ff","subject":"Merge \"Remove no-longer valid comment about IPv6 workaround\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2aa5bb356e65709795e036ca636bf46d664298ff"}]}],"author":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2024-10-17 15:25:36.000000000","tz":120},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2024-10-17 15:31:36.000000000","tz":120},"subject":"Add wsgi_api.py file","message":"Add wsgi_api.py file\n\nThis allows deployment tooling to easily switch from passing a binary\npath to passing a Python module path. We\u0027ll use it shortly.\n\nChange-Id: I8d992b88043d80249ca6e246ed3068a4e6a8b54f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b27ac643aa14ec7ce987eb55f4fb8e69aeb3c9e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b27ac643aa14ec7ce987eb55f4fb8e69aeb3c9e1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
