)]}'
{"id":"openstack%2Fglance~932768","triplet_id":"openstack%2Fglance~master~I502b5c246de1631193d604143ea075bc5ca20dd7","project":"openstack/glance","branch":"master","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2026-04-28 18:56:51.000000000","reason":"Change was submitted"},"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2025-12-03 06:29:15.000000000","reason":"\u003cGERRIT_ACCOUNT_9303\u003e replied on the change","reason_account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-04-28 18:56:51.000000000","reason":"Change was submitted"},"4393":{"account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"last_update":"2026-04-28 18:56:51.000000000","reason":"Change was submitted"},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2026-04-28 15:42:57.000000000","reason":"\u003cGERRIT_ACCOUNT_8122\u003e replied on the change","reason_account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2025-12-02 17:12:49.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_37598\u003e by clicking the attention icon","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}},"hashtags":[],"change_id":"I502b5c246de1631193d604143ea075bc5ca20dd7","subject":"swift: Update description about auth version options","status":"MERGED","created":"2024-10-19 07:46:49.000000000","updated":"2026-04-28 18:58:03.000000000","submitted":"2026-04-28 18:56:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"932768","meta_rev_id":"0a7386562fff6ef9aa9c68052e4aa109996c73f2","_number":932768,"virtual_id_number":932768,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-28 18:56:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2025-12-03 06:29:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2026-04-28 15:42:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-02-05 15:50:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2026-04-28 15:42:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-19 08:01:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-20 15:07:31.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2025-03-20 15:07:31.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2025-03-20 15:07:31.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2025-04-12 14:24:06.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2026-02-05 15:50:41.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"}],"messages":[{"id":"9f478244f746e14952f918d6e734353bebb7a740","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-19 07:46:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"973569f69e07227597e1312be94cfd4f18775053","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-19 07:48:25.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"c96e41530ed5639f56a12b07d728c9f4efb4df60","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-19 07:52:48.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"a18b75cbda91ea131fa719ec135034609a4f9f1d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-19 07:53:07.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"926852458191d3be3865189c44cbf56f77da3be4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-19 08:01:42.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1841129ce23f4261b7ce119478b7f10f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24e4f20920fc449fa8a4ccc4ea5fefe2 : SUCCESS in 6m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e85de9ce6128420786677023746415b6 : SUCCESS in 5m 10s","accounts_in_message":[],"_revision_number":3},{"id":"65b23a3bbdac83e55fc8b99cea6d6dd1d5ebdc6a","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-03-20 15:07:31.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"0ea4d28affc196d459218664b2a7186ea1bcc017","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-07-01 14:17:23.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f269fca0592ddae5b96bb2a33bfdedfba0d6eecc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-09-22 13:04:31.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"0b4020b24f20371121154f250721db4d9827e6a8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-09-22 13:05:55.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"e44514d8a9b871b9d193a6b7c66f1fe7ea13c19b","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-09-22 13:06:29.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"4700705cb685d34efe59f649fdfd469fd51d257b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-22 13:59:21.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7aabb4b84f884f93a939a874e70adcf7\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c1b41fd6b91747c78de454bc4da0cfd4 : SUCCESS in 6m 25s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/61620f89974b44cfa8c300ccb2fe44b2 : SUCCESS in 11m 08s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/547651cd9d8846d9a6290fc82393a5b6 : SUCCESS in 14m 48s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"085ae405d7f85e2505fc1ddaa88966966cf9f66b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-22 15:28:58.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3541ced2e60c4787ad7b0aa4c41fc8b7\n\n- grenade https://zuul.opendev.org/t/openstack/build/218a8cb85b784af3bfe06b05ae9a05c5 : SUCCESS in 1h 03m 23s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/fb736a91535743728c3af79041a22526 : SUCCESS in 57m 44s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/41c8992d8148478db5d2b066f985580d : SUCCESS in 1h 24m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/881f50d125ed4575bedf6d1213cd58ff : SUCCESS in 56m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6bcb8a4b27e4e9398ff458eab3395ac : SUCCESS in 6m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fa738694fdb747aba16f03c7dfe0e3c9 : SUCCESS in 4m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ee3942f915984507a2cf34bbcb0e3587 : SUCCESS in 7m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f32a16c248144d6892612f3c406907bb : SUCCESS in 8m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe6e24c7acb649bb84657b7ee4e66ae1 : SUCCESS in 6m 23s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c56dbc3dae4e4fb7b8390cc651f90300 : SUCCESS in 5m 17s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/d10d3ac691a8453dad64c84865b899e4 : SUCCESS in 12m 03s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/728118e2eaac4dfd90042d22465770e0 : SUCCESS in 1h 47m 31s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/a6165d82971c44b086abeec9e6d35f5a : SUCCESS in 2h 13m 59s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/037d912ce0c347d8b1d8bdea5f5566b7 : SUCCESS in 1h 53m 15s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/f9e29da6bcd04b1dab26743fc0c086ed : SUCCESS in 2h 07m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/48dc323402a44e718cc78e41ab227c36 : SUCCESS in 51m 30s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/7cca8d298f1a40238994b13a9a30aca5 : SUCCESS in 2h 00m 38s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/caa66bbdfa4141b0969ebe35c2b9da44 : FAILURE in 27m 37s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/18e0ff9707d94c17bf1ba782c8a1781a : SUCCESS in 1h 59m 50s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/98d666b711af45acb23b94689599019e : SUCCESS in 22m 21s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/d5922b0fcd9649d294db2bc60d2550da : SUCCESS in 48m 06s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"29e01de2f59715dd6c831083517876c951fdee91","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-12-03 06:29:15.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"aa589352856fc7346d14b22d2236ffcd61937276","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-05 15:50:41.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"5c3ca390fb1dc093eb66f69ac46126db89f5d1c4","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-04-28 15:42:57.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"710f9e33d9ce8e65deb980df5cb3b2df621f1b70","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-28 15:44:03.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"908cda156f4f39bfa6531aa96f1ef6dbf10a8687","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-28 18:56:51.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/62ea91c3ba6a465d820a17929d0b1a8e\n\n- grenade https://zuul.opendev.org/t/openstack/build/6c9f2e6fbe644f438616efc5d556f51d : SUCCESS in 35m 47s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f089aad4528943adb19165c47c0162b2 : SUCCESS in 1h 04m 18s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d984caaeda264b07adf883c6925f8869 : SUCCESS in 1h 28m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/235a62a7ee844b1e95e5a6f8322c4e3e : SUCCESS in 1h 05m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4a12a8467584be28e6494868238a93e : SUCCESS in 6m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fd7da80c26e0412e974b62331c678c11 : SUCCESS in 5m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/49ed5f7b4e274a0b9e6b6f9046a8b18f : SUCCESS in 5m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90453f87f5024a1eb7bc270f0d46995f : SUCCESS in 6m 05s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/06425759bc8b4696b736c0972812f1f0 : SUCCESS in 8m 22s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/22cdbdad91524fa291560424325a990d : SUCCESS in 6m 55s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/76aa5b04904942f1a381bcb1be3b7368 : SUCCESS in 1h 37m 37s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/5d50ec591d7b4e7fbb7841bcac444060 : SUCCESS in 2h 12m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/69eca2e6497a4f2884bdc97bbbf8c0be : SUCCESS in 1h 01m 03s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/8afc29ade9b14fde9be1d22677315b1d : SUCCESS in 1h 41m 41s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/eebb49c13fc2468fb285be939017adc7 : SUCCESS in 42m 56s","accounts_in_message":[],"_revision_number":5},{"id":"37e4a557994126ee0a9248302be295102ff0698e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-28 18:56:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"0a7386562fff6ef9aa9c68052e4aa109996c73f2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-28 18:58:03.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d702e18e25ad4d02b0d1310c5d06cb34\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a039a869003640d59d2d93b093cd8a86 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"2db180e518cdb76348e9e09de2073b819a2350c8","revisions":{"486473ca668816cdf07cf93bf4a9eb1e856242f8":{"kind":"REWORK","_number":1,"created":"2024-10-19 07:46:49.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/68/932768/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/68/932768/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/932768/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/932768/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/932768/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/68/932768/1"}}},"commit":{"parents":[{"commit":"13acba7505295cfd57830510618d867a1bb4dd8a","subject":"Merge \"api-ref: Update with links to 2.17\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/13acba7505295cfd57830510618d867a1bb4dd8a"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-19 07:43:49.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-19 07:46:36.000000000","tz":540},"subject":"swift: Update description about auth version options","message":"swift: Update description about auth version options\n\nThese options were deprecated in glance_store and accept only \u00273\u0027 now.\nUpdate the documentation to reflect that change.\n\nCloses-Bug: #1323646\nCloses-Bug; #1480330\nRelated-Bug: #1947926\nDepends-on: https://review.opendev.org/c/openstack/glance_store/+/932251\nChange-Id: Ie8c522ba9df91722eb6777578eba1fad22fc0bac\n\nChange-Id: I502b5c246de1631193d604143ea075bc5ca20dd7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/486473ca668816cdf07cf93bf4a9eb1e856242f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/486473ca668816cdf07cf93bf4a9eb1e856242f8"}]},"branch":"refs/heads/master"},"2e91f25e754d081b2c830362e1138d1caf368d46":{"kind":"REWORK","_number":2,"created":"2024-10-19 07:52:48.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/68/932768/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/68/932768/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/932768/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/932768/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/932768/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/68/932768/2"}}},"commit":{"parents":[{"commit":"13acba7505295cfd57830510618d867a1bb4dd8a","subject":"Merge \"api-ref: Update with links to 2.17\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/13acba7505295cfd57830510618d867a1bb4dd8a"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-19 07:43:49.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-19 07:52:27.000000000","tz":540},"subject":"swift: Update description about auth version options","message":"swift: Update description about auth version options\n\nThese options were deprecated in glance_store and accept only \u00273\u0027 now.\nUpdate the documentation to reflect that change.\n\nCloses-Bug: #1323646\nCloses-Bug; #1480330\nRelated-Bug: #1947926\nDepends-on: https://review.opendev.org/c/openstack/glance_store/+/932251\n\nChange-Id: I502b5c246de1631193d604143ea075bc5ca20dd7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2e91f25e754d081b2c830362e1138d1caf368d46"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2e91f25e754d081b2c830362e1138d1caf368d46"}]},"branch":"refs/heads/master"},"d98f9848a9b88b6d6cb97921b4e09e6ae4e39b62":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2024-10-19 07:53:07.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/68/932768/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/68/932768/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/932768/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/932768/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/932768/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/68/932768/3"}}},"commit":{"parents":[{"commit":"13acba7505295cfd57830510618d867a1bb4dd8a","subject":"Merge \"api-ref: Update with links to 2.17\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/13acba7505295cfd57830510618d867a1bb4dd8a"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-19 07:43:49.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-19 07:52:52.000000000","tz":540},"subject":"swift: Update description about auth version options","message":"swift: Update description about auth version options\n\nThese options were deprecated in glance_store and accept only \u00273\u0027 now.\nUpdate the documentation to reflect that change.\n\nCloses-Bug: #1323646\nCloses-Bug: #1480330\nRelated-Bug: #1947926\nDepends-on: https://review.opendev.org/c/openstack/glance_store/+/932251\nChange-Id: I502b5c246de1631193d604143ea075bc5ca20dd7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d98f9848a9b88b6d6cb97921b4e09e6ae4e39b62"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d98f9848a9b88b6d6cb97921b4e09e6ae4e39b62"}]},"branch":"refs/heads/master"},"fb5e492ba20eeec0ba04a01762365057b64bd317":{"kind":"REWORK","_number":4,"created":"2025-09-22 13:04:31.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/68/932768/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/68/932768/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/932768/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/932768/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/932768/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/68/932768/4"}}},"commit":{"parents":[{"commit":"60a2f055d54eb28413921c84afa7ed400bdb0695","subject":"Merge \"Use Ubuntu Noble instead of Jammy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/60a2f055d54eb28413921c84afa7ed400bdb0695"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-22 12:58:37.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-22 13:03:54.000000000","tz":540},"subject":"swift: Update description about auth version options","message":"swift: Update description about auth version options\n\nThese options were deprecated in glance_store and accept only \u00273\u0027 now.\nUpdate the documentation to reflect that change.\n\nCloses-Bug: #1323646\nCloses-Bug: #1480330\nRelated-Bug: #1947926\nDepends-on: https://review.opendev.org/c/openstack/glance_store/+/932251\nChange-Id: I502b5c246de1631193d604143ea075bc5ca20dd7\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fb5e492ba20eeec0ba04a01762365057b64bd317"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fb5e492ba20eeec0ba04a01762365057b64bd317"}]},"branch":"refs/heads/master"},"2db180e518cdb76348e9e09de2073b819a2350c8":{"kind":"REWORK","_number":5,"created":"2025-09-22 13:05:55.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/68/932768/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/68/932768/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/932768/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/932768/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/932768/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/68/932768/5"}}},"commit":{"parents":[{"commit":"60a2f055d54eb28413921c84afa7ed400bdb0695","subject":"Merge \"Use Ubuntu Noble instead of Jammy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/60a2f055d54eb28413921c84afa7ed400bdb0695"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-22 12:58:37.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-22 13:05:46.000000000","tz":540},"subject":"swift: Update description about auth version options","message":"swift: Update description about auth version options\n\nThese options were deprecated in glance_store and accept only \u00273\u0027 now.\nUpdate the documentation to reflect that change.\n\nCloses-Bug: #1323646\nCloses-Bug: #1480330\nRelated-Bug: #1947926\nDepends-on: https://review.opendev.org/c/openstack/glance_store/+/932251\nChange-Id: I502b5c246de1631193d604143ea075bc5ca20dd7\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2db180e518cdb76348e9e09de2073b819a2350c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2db180e518cdb76348e9e09de2073b819a2350c8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
