)]}'
{"id":"openstack%2Fglance~933509","triplet_id":"openstack%2Fglance~master~I51530aa3d723d04358db1a7eeb7dff996c4fb507","project":"openstack/glance","branch":"master","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2025-04-21 13:01:38.000000000","reason":"\u003cGERRIT_ACCOUNT_5314\u003e replied on the change","reason_account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2025-04-23 08:36:32.000000000","reason":"\u003cGERRIT_ACCOUNT_9303\u003e replied on the change","reason_account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-04-26 14:37:43.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I51530aa3d723d04358db1a7eeb7dff996c4fb507","subject":"Add note about requirements lower bounds","status":"MERGED","created":"2024-10-26 15:35:11.000000000","updated":"2025-04-26 16:30:10.000000000","submitted":"2025-04-26 16:28:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"933509","meta_rev_id":"c7f6ed8e78b667a0f91f497dc60983a6e9576b07","_number":933509,"virtual_id_number":933509,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-04-26 16:28:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2025-04-21 13:01:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2025-01-08 16:04:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-04-12 14:12:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2025-04-21 13:01:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-26 15:46:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-10-26 16:45:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-08 16:04:14.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2025-04-12 14:12:40.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2025-04-12 14:12:40.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2025-04-12 14:24:21.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"}],"messages":[{"id":"1faf90e774acdfb3bd539ec927a9e0b404e6169e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-26 15:35:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5a17ad9a8bb1fc2656913fa3ed10859de59e7fb4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-26 15:46:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b0fa7f1b2f7b48a8b6a49537b13f25ac\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6f4441cdca2346a4a52e1aae1351caa8 : SUCCESS in 9m 18s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/fb3ec218937d48278253829cd5406694 : SUCCESS in 9m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"badd0a0c51beb3bdfde1dedc2dce9fdee3d8c475","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-26 16:45:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0db3b917913c43648856377f46193c2c\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/a24002bbc5ca462bb4e7398db6f33df4 : SUCCESS in 2m 27s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2f8ec0589f99494fbb05a20e2930d9da : SUCCESS in 59m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b30344afae2f4e47b9a6f2a3bc874a40 : SUCCESS in 59m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/332a384190de4139b5e1e993fd497f6a : SUCCESS in 3m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a59122d38b8d4e2b8102a8f46172729f : SUCCESS in 4m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d6ec4acd109a459a87c7628275aab8b9 : SUCCESS in 3m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51c5721efe8e40649805ffc247358ca5 : SUCCESS in 3m 25s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e1d5ef977b5348b489fea1481e8ffbdc : FAILURE in 6m 47s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b6eb209c226a421d99215b4f11670500 : SUCCESS in 6m 44s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/cf03cd0f0e6e4c64829866b27e1e4d54 : SUCCESS in 9m 17s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/70329498462440deb67574aef7816180 : SUCCESS in 19m 26s","accounts_in_message":[],"_revision_number":1},{"id":"1a86edd590a1f91255b809f99e633814a1175d72","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-26 18:30:27.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"3d61657f6c16f4a26a26f161fb501366e5d35e3e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-26 18:42:03.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b146aa642745429ebe6723f62de5f0b5\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6a88d0b132f04f3892a4271d9ea19b98 : SUCCESS in 9m 55s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/886b29d43f2544059f743339218f0207 : SUCCESS in 9m 23s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"07cdabfdf6c07dad45e7d7ceb51f00eaa5481ca6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-26 19:33:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fbdf7b1e17b746068a124853d279cae0\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/d9b31548e8544d259455cbf345efdcfd : SUCCESS in 2m 28s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b24d70b4a79d407092fb86ea1cdb2f86 : SUCCESS in 37m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/23712198d87343859c03ca8a81ad2f9f : SUCCESS in 52m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df6e82b58f3348d2975463fb314b3d0f : SUCCESS in 4m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c3e99893005b41e3b9d7862b18ac3066 : SUCCESS in 3m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/acb10225e08f41d185f71fef7dd52bc7 : SUCCESS in 5m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f14badc40aa403aad057375a806de50 : SUCCESS in 6m 46s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/eb55121870c14b6ebd49ffecd42ff804 : SUCCESS in 7m 31s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/342d456cc65147f69109521d3cad3b1d : SUCCESS in 8m 38s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/995839b1f97849ff993a020971240122 : SUCCESS in 10m 11s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/6a77ccff71554f3082e53f731893e362 : SUCCESS in 20m 10s","accounts_in_message":[],"_revision_number":2},{"id":"d92274b1fbdb68c5195ce7aff77c446a8c1cff2c","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-01-08 16:04:14.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"826573a6a24c9947c127535718bc887e689c8594","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-04-12 14:12:40.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"fc60eb2bf7d3cfd902498fab91f53e48b0434b97","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-04-21 13:01:38.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e993a7287726517308f77843b25607103a5a7253","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-21 13:02:05.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"013ecc91c72e2f422c096b0e89c69c2e8bd8e29d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-21 14:15:50.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/49ea08ff03cf4ec08c5750e807e4e815\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/88f709941b084747934c57daaae0dace : SUCCESS in 2m 27s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a1349081a2df47bd8fe20aeed079394e : SUCCESS in 44m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8d3f6c2cecc240dda177d9617886810c : SUCCESS in 1h 05m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06a75bf6dad14af9ab1378719d38afe4 : SUCCESS in 2m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/78b42d23b5f64110832b658733fb9672 : SUCCESS in 5m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f59fb6728cd44a16a96648f04514b8d9 : SUCCESS in 4m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/027b4a4aebc04159a9df6d306df86110 : SUCCESS in 4m 45s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f2999ef76a0c40acbd667a1f991bd90b : SUCCESS in 11m 18s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6c142b362d434ea399803c6be67e7c46 : SUCCESS in 7m 22s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/f94064ddbce44eee9a37fcce17ac4e2f : FAILURE in 7m 04s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/7782666b8ebe452c83052607e91616b5 : SUCCESS in 27m 54s","accounts_in_message":[],"_revision_number":2},{"id":"1f0ddb7856ee22c792e83f7b16a6a984c53ad6a6","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-04-21 15:50:35.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5273f4811ec727c9058fd54f3f940607e205baa9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-21 16:00:11.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a208164d76e74d0487bb1d78166b0d08\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a8ced80769a04559b77f7399d55fe9ee : SUCCESS in 5m 41s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a9f99d86f599459ea387b80de5e6f308 : SUCCESS in 5m 09s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e74ef3f865574eeab5b41b3ae6ce9b3a : SUCCESS in 7m 50s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"80f1ae483435bc532422bd9dc4e5a7dd76cd1969","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-21 16:40:22.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1f48e0f0c9974040a2b0f5263f0c2038\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/2130bba90e7046a5a4d50b59ba2d3477 : SUCCESS in 2m 58s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/57f5a5aa6f6d4266803c3a34510984bd : SUCCESS in 40m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ddce84431021467a8404598fe1294ea4 : SUCCESS in 49m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9f0afb0f2b3482ba0a8eca621e303d1 : SUCCESS in 4m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ec10d2b5aafb4d46a25b35e67f547519 : SUCCESS in 4m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9ef58756afd74db9adc4e3405dc52aec : SUCCESS in 5m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/30b3ca260e0b44758dfaa14e57a27cfa : SUCCESS in 6m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f88ab0d4dd349ed9f7eff014535b67f : SUCCESS in 5m 22s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/baa67fb365054e908027fa8a20074db5 : SUCCESS in 7m 34s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ae85c1e0f60141ca890d681fa260ba00 : FAILURE in 5m 55s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/fa3d510108914634b1ae08521b8b0c93 : SUCCESS in 11m 01s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/df3616071ea146c08e71837fae996226 : SUCCESS in 17m 58s","accounts_in_message":[],"_revision_number":2},{"id":"a087e6fdb014ffa06d7c0e33730d609ec14a59f6","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-04-21 16:41:59.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7ce169cd6db5cbf95bc6e73d5d94eb4647c0f20c","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-04-23 08:36:32.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"15926c60d1d2a657a457a13f8cde83bbfdd7a81c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-23 08:49:51.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aad82bdfced94ce8a030edcde669d87b\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e378a4c3b0c14c9189d3e3a6cf37725c : SUCCESS in 7m 15s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/21636ef5059f4a1fa43048fbdbd0052b : SUCCESS in 7m 47s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/39dfe30b3525449baf0297d5398d5404 : SUCCESS in 11m 29s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"11e42f7de546186d9e9e0a2c437d01cd690a6429","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-23 09:09:57.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6deb3e94140849e6a68b2df4849271fb\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/0e04a9e52c35493ba7be6c286a0ca47b : SUCCESS in 1m 43s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/826a10dc0606481981d333fdc89527ec : SUCCESS in 26m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fa2538b327194af0a1a59474249fdf84 : SUCCESS in 32m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9417e9620539471aa145e8d948255764 : SUCCESS in 3m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a4139843d9634be4abb6d293838ce6b2 : SUCCESS in 2m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cc43eeed49dd4910813f3be21f4fae29 : SUCCESS in 4m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/dab9d69423dc41eb985697939db5e225 : SUCCESS in 6m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39451ad1659440c7a35abc52fdb00713 : SUCCESS in 3m 22s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2b9ed5ec7d0f41868fe6130506d8eda2 : SUCCESS in 7m 43s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e9a8cede349346b1a792b4a7af6ac4f7 : FAILURE in 5m 57s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/228d35862cd54c9eaf77e1975bf6644e : SUCCESS in 10m 17s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/fbc8246a8375439ca46aed0530f8fd5b : SUCCESS in 21m 46s","accounts_in_message":[],"_revision_number":2},{"id":"7595b1608b65f3f41876dd12b23cc9ebcbdf9a88","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-04-26 14:37:43.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3aa97b75874ce626ce56cae6d5f9aff7c8ee86eb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-26 14:51:18.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2d2b85a6f9ca4a519f9421ad8c8cdea0\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8313330cbeeb47ba99a88381a10cb297 : SUCCESS in 9m 51s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/bd045ea628804e86a10d817d4fb3749a : SUCCESS in 10m 22s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d835de80c1004dd1898d48a31991285a : SUCCESS in 13m 23s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f684663fca266c374b59694a446be69b486aa47e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-26 15:41:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/815112d3e58d457987e55d89c73b4532\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/8e67df3f450f42fa93fa1eab5c6dbfd0 : SUCCESS in 2m 56s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3f3fe960704c4946a39261345f2e6a7c : SUCCESS in 57m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f460fbb1154f466a88623ba5ca3f4c7b : SUCCESS in 59m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3da8eee913f545789286b7e403e8691e : SUCCESS in 2m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4d3dd43ce4e84ec9b67a0f6864fda3ad : SUCCESS in 2m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3df0a78ed2c1425eafabd4a15a1d5b68 : SUCCESS in 3m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/97b99a191a6647a4a6a95ae4ec5c9db9 : SUCCESS in 7m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60e1194181d64cf9a36c0f74f4c0bb24 : SUCCESS in 5m 45s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1909c84909484a8db9d3874eb58d633e : SUCCESS in 7m 45s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/10de21a0bbf44a22a264b5653bc4b3ef : SUCCESS in 6m 46s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/20891cbea15d443e86a0c599eb64dd31 : SUCCESS in 9m 16s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/25eb070e626d40a38aee5afd6581985e : SUCCESS in 22m 40s","accounts_in_message":[],"_revision_number":2},{"id":"8a7378191c37db8097140de33a16e45ed7da1feb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-26 15:41:47.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"20a7b06c5c7cd56f3f7c0cdb0fee198be95ce2e3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-26 16:28:55.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b593a3cd57243d09496b86573095b3e\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/5b23fcd15dbb4dbe824888bdc0a22a36 : SUCCESS in 2m 34s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7747917f81314fd9a8cdcdc09902a2bd : SUCCESS in 38m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/74e9512a82624f91a6bba789ed6b7519 : SUCCESS in 46m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d82769874700492a944815a040dd1086 : SUCCESS in 4m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/040c3e197d9442e795fa9b3cc22cf233 : SUCCESS in 3m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cf886924cc51403aabb2fd5c3459b258 : SUCCESS in 4m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b2a0dbced90450c8cde5ec4aea279ce : SUCCESS in 5m 51s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3f05913f853848089e9f501660449c52 : SUCCESS in 8m 12s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7ae91cc2e6b849eeb668cb1294a6b3a0 : SUCCESS in 5m 34s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/64ba9eb5b6b147e8b0c5e1472272a083 : SUCCESS in 7m 10s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/da6fc3883bdf4d33b293fbcdc50bfd97 : SUCCESS in 14m 18s","accounts_in_message":[],"_revision_number":2},{"id":"51123516f3bb7b2cd706416385b0a671736d0ca8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-26 16:28:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"c7f6ed8e78b667a0f91f497dc60983a6e9576b07","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-26 16:30:10.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4814a7794f0c41859f9ad6b65226fcb5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70185123413b4f0b852a4962a9f0fe5e : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2f1ff005eb4fd1958a67570c1301b8437ef9a6a8","revisions":{"0f74f03a7602c61bb2d1c03a22f8b829bc45fad8":{"kind":"REWORK","_number":1,"created":"2024-10-26 15:35:11.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/09/933509/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/09/933509/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/933509/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/933509/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/933509/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/09/933509/1"}}},"commit":{"parents":[{"commit":"7a98b9f0f6c1df88ea776c31d8c7f8a746559308","subject":"Doc: Add info about multi-attach volume types","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7a98b9f0f6c1df88ea776c31d8c7f8a746559308"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-26 15:35:02.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-26 15:35:02.000000000","tz":540},"subject":"Add note about requirements lower bounds","message":"Add note about requirements lower bounds\n\nSince lower constraints job was removed the bounds are not actually\ntested. Add a note to explain the bounds are maintained on bast effort\nbasis.\n\nChange-Id: I51530aa3d723d04358db1a7eeb7dff996c4fb507\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0f74f03a7602c61bb2d1c03a22f8b829bc45fad8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0f74f03a7602c61bb2d1c03a22f8b829bc45fad8"}]},"branch":"refs/heads/master"},"2f1ff005eb4fd1958a67570c1301b8437ef9a6a8":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-10-26 18:30:27.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/09/933509/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/09/933509/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/933509/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/933509/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/933509/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/09/933509/2"}}},"commit":{"parents":[{"commit":"7a98b9f0f6c1df88ea776c31d8c7f8a746559308","subject":"Doc: Add info about multi-attach volume types","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7a98b9f0f6c1df88ea776c31d8c7f8a746559308"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-26 15:35:02.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-26 18:30:27.000000000","tz":0},"subject":"Add note about requirements lower bounds","message":"Add note about requirements lower bounds\n\nSince lower constraints job was removed the bounds are not actually\ntested. Add a note to explain the bounds are maintained on best effort\nbasis.\n\nChange-Id: I51530aa3d723d04358db1a7eeb7dff996c4fb507\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2f1ff005eb4fd1958a67570c1301b8437ef9a6a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2f1ff005eb4fd1958a67570c1301b8437ef9a6a8"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
