)]}'
{"id":"openstack%2Fglance~933550","triplet_id":"openstack%2Fglance~master~Iaafcbd4ff858bf905b51d80bfda89e8a59b235e4","project":"openstack/glance","branch":"master","attention_set":{},"removed_from_attention_set":{"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2024-10-28 21:56:35.000000000","reason":"\u003cGERRIT_ACCOUNT_27615\u003e replied on the change","reason_account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}}},"hashtags":[],"change_id":"Iaafcbd4ff858bf905b51d80bfda89e8a59b235e4","subject":"Increase per project quota to 50","status":"MERGED","created":"2024-10-28 09:08:32.000000000","updated":"2024-10-29 16:17:37.000000000","submitted":"2024-10-29 02:20:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"933550","meta_rev_id":"41e85f3dc96c54baf0f9eb26e20fa43ecf470722","_number":933550,"virtual_id_number":933550,"owner":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-29 02:20:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2024-10-28 19:15:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2024-10-28 19:15:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-28 09:20:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-10-28 12:31:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-28 14:06:57.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"CC"},{"updated":"2024-10-28 19:15:06.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"c500687732fa7ab06c6ab4405501ee770e191d77","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-10-28 09:08:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fe2b87df6a17a320303e01b89a7b69b05bf87133","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-28 09:20:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2212bec03f6c4025a888d5fe3db70975\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/21f476248a234a9eb218e0016427e48f : SUCCESS in 7m 58s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b9b6e8c08e1c40548861acae8167ef5d : SUCCESS in 9m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"85034089da74ebe4e115bb051692dbe752de0c2a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-10-28 10:08:39.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"431bb72c31f22e60272e5bee9860153b1395a7ef","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-28 10:18:56.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c00defd229f34d9c8069ed4e0c6ab010\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/491ff066fd7548c68031098ea594235f : SUCCESS in 6m 38s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3e2b92869f27449bbf74d7ecb12d41b2 : SUCCESS in 7m 52s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0ac823708bbc007e61d0f4f9389f507e6c272952","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-28 12:31:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/226d73ab21ef453c83b7da0d3c6ede33\n\n- grenade https://zuul.opendev.org/t/openstack/build/9a060176b4f943a58dbb8aec0ffa25bc : SUCCESS in 55m 47s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f1b77b902b0e47d3a4faad9e89a859f6 : SUCCESS in 58m 10s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/eded56e615e44509beaa2d05c642ed7c : SUCCESS in 1h 27m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a917bd801dc24e718fef5d274cb636e9 : SUCCESS in 58m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b18063f219834cc1946b40d3edc3bfc6 : SUCCESS in 2m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/49962ec47aa84312ab3fe6508051a8dd : SUCCESS in 3m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/55bfc66d95214e8aa482c3600c0bb0fa : SUCCESS in 7m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03b27ddfae3c4772892229ce7f7d58f0 : SUCCESS in 5m 47s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0c4f9216be0246a9a327a244a5e8ca96 : SUCCESS in 10m 10s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e8f58ef15b52436191573ec4e2a36388 : SUCCESS in 8m 05s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/4e2cb330bc3b40ad84c51a99cb660b1d : SUCCESS in 12m 50s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/427f088d2c4c470bbb9aa4a70fa707bb : SUCCESS in 1h 50m 24s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/23f6aa678ebd44ceb6883857a7bf8735 : SUCCESS in 2h 10m 45s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/9846a47b9dc64cba925ffa7ee90521d6 : SUCCESS in 1h 40m 37s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/23e931416b3b4c5d9c6b0e6f6c66a5ae : SUCCESS in 1h 53m 50s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/d497fa1ae5e94542b851432247787fd2 : SUCCESS in 2h 12m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/55d58438790549b3a143046f914323e6 : SUCCESS in 52m 20s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/0e7e4adea7864774a1f5f749d190475c : SUCCESS in 1h 10m 40s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/278fb985b4484fbd912663efba4421fd : SUCCESS in 2h 08m 17s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/421f1d04aa71425a96c6702bfb1a9156 : SUCCESS in 1h 00m 50s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/72f19ff44ab441c4848a127d3944d705 : SUCCESS in 30m 08s","accounts_in_message":[],"_revision_number":2},{"id":"ac76f8fae7a1b48706849430cd59c3ee99d532ad","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-10-28 14:06:57.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"054505872e3930719a303643c47504ed0f75744a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-28 14:19:36.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d1d53025ae9740faa823714f323ebcc5\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/30c11c06407a468fa7dc9c284c162e05 : SUCCESS in 8m 53s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/44cfea8a94a84031b3f575019782140f : SUCCESS in 9m 47s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3337778a5fd51fa394d2fe15d23661b0bf12d656","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-28 17:16:30.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a1ab63ada74944babefc3f089834ef41\n\n- grenade https://zuul.opendev.org/t/openstack/build/b948dd0e59584ceaaea6f4ffa7758db9 : SUCCESS in 1h 05m 50s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/511651ce2b8d4fa2bc1a7fc2c99bfd51 : SUCCESS in 1h 02m 42s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/786e035e9be5465c88d03ae471918e77 : SUCCESS in 1h 30m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/951674a5b0134524a8e1a599d1f51a3d : SUCCESS in 1h 03m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17f28f6452594754b0828c713f2a97e9 : SUCCESS in 4m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/33e075b8129b4dc0a0e45bc13d46083e : SUCCESS in 5m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d6b02e29b50b4e53828b8f53e963d292 : SUCCESS in 5m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efa6492c7b9b4246a3a51f157300e293 : SUCCESS in 5m 47s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a022b7bbac7241ec853e6f48fc7819c0 : SUCCESS in 8m 47s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/888867be44fd44ca9da0d9a2af063b3f : SUCCESS in 7m 27s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/23b7a112eceb479db0f16755756962e5 : SUCCESS in 11m 42s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/272e52c525bc446c912f5b2260675069 : SUCCESS in 1h 34m 07s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/9639cbbbe00b43fc93d05d79a4a5d6ea : SUCCESS in 1h 36m 26s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/f970ece325884111999028b3fa6caf39 : SUCCESS in 1h 27m 43s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/007715b6df5a4260942deeaaaeefaaf9 : SUCCESS in 1h 19m 59s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/410d6645ac6043129f81d9157995aef6 : SUCCESS in 2h 03m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e3202c0feb4249d5b318367312f3edcf : SUCCESS in 41m 48s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/cd20267130944b92b54832ae4289e60a : SUCCESS in 1h 16m 09s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/004fad2cf2234913b019fd3193950571 : SUCCESS in 1h 52m 59s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/03441f81b5fa464a97b0439f8511bfcb : SUCCESS in 1h 01m 50s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/6463f04bd3d744c6a23b65eafee27799 : SUCCESS in 23m 27s","accounts_in_message":[],"_revision_number":2},{"id":"4dd593fc98aaab79b1dff1777d7b437667fac5bb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-10-28 19:12:34.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"e9371fbf02804248c87451fc6cea59d78edc9901","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-10-28 19:15:06.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ecd4b2a8783d58772708ad5de143c735bc9b278f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-28 19:24:01.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d6a7462ba5574b229a851f16634c6952\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/78973874495c4bf59807f46299b8ab7a : SUCCESS in 9m 15s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/fe4cea62dda24924889bee6cdf1552df : SUCCESS in 8m 53s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"9c9a7805ef15c5a918bd6cd4ba093344471bbd4c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-28 21:24:05.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/26ebf199a0e3466783b2ba2f34da7221\n\n- grenade https://zuul.opendev.org/t/openstack/build/766953bd372847aba6f64d3ad50c4766 : SUCCESS in 41m 12s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6329a433195c45ef99bcd6c9da08f855 : SUCCESS in 40m 38s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8fb1bdae1bf041f9af31b8dd60151cee : SUCCESS in 1h 39m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/18e4830106824444b460d5dc1d7491f1 : SUCCESS in 42m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3a9087c5d584cb0801ddd7f5259e424 : SUCCESS in 3m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/60f311808ced4dd0ae9df5f633b5c600 : SUCCESS in 3m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9b72229aac334adc929a9125d2f2e082 : SUCCESS in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5651c50980ad44c99ee91403dc6988ec : SUCCESS in 5m 25s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ec000c02c8f84086b4a315be48adf9e3 : SUCCESS in 7m 29s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/946da3c9da7d4817b4f96af5b6f57899 : SUCCESS in 8m 53s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/b18372c7ad864a2fa1a7fd9a8317a46e : SUCCESS in 12m 25s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/27afe4d9662b4f9f8c881bdeb5d31673 : FAILURE in 1h 20m 51s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/0b9238fc4a3d4accb83233935e9f49d4 : SUCCESS in 1h 48m 49s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/06f1d0eae5d04f2c89eea41fc9b23cb3 : SUCCESS in 1h 58m 47s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/ced82992cea342e9adb4da97ad34ca9c : SUCCESS in 1h 22m 06s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/d3e0a6b623b04edc91e9e69cf243f856 : FAILURE in 1h 19m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bd72cfc920a04718a3b0f2e824e804c2 : SUCCESS in 56m 53s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/15346ed39c49411fa592c08926907461 : SUCCESS in 1h 45m 21s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/01e04f12bef54c30bd3c6f6c2f989f2e : SUCCESS in 1h 39m 57s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/0ab7151d2d5149f9b071c3b4df796ca6 : SUCCESS in 55m 19s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/d517aaa43a1b49d98b9fb039c067d561 : SUCCESS in 34m 17s","accounts_in_message":[],"_revision_number":3},{"id":"df50598f158cdf0f5d33b1dae092afba7e0fde73","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-10-28 21:44:47.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8f1f87a3482a1b0ec368bc35ba3011468faf87ad","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-28 21:56:27.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bd475d3cd00a4c039a6fbe71341da74f\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/538a7df75689439aa7296e4a0e8a1369 : SUCCESS in 9m 33s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/fbc5f5d8e1034a91aa6aa9e71bebf064 : SUCCESS in 8m 39s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"488f238fa65be957223ef5e2a0a140b66308edba","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-10-28 21:56:35.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c2a859a1af8af64ca18fcdb235417e1ad8634e51","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-29 00:12:53.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ce516568be0b4a28bd46a78627e83f71\n\n- grenade https://zuul.opendev.org/t/openstack/build/0e494e3642d043aaa8f3d2010b2132a7 : SUCCESS in 1h 07m 36s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/542f57467ac14c7193f05affdfdb3a7b : SUCCESS in 1h 09m 21s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/882a83de8b134acfb204e9b22259edc3 : SUCCESS in 1h 34m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/723637c1d16e49f78ddcf254eb9af8bd : SUCCESS in 1h 03m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8185dcefd9014d9bbca40c40922d9861 : SUCCESS in 4m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6bc509915cfd4e3d8432b39ea37bed39 : SUCCESS in 5m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f8aab9ce30f0490db126361eb0bcdb30 : SUCCESS in 5m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3105c93dac074a37abb7b2d5d9f4bbec : SUCCESS in 5m 21s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/48150b198d5945a287441bad7343bc6b : SUCCESS in 8m 58s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e659ffbf2ce5459eb1a5187949049fc0 : SUCCESS in 8m 26s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/aacf69be7eef4d45a28c6b50228a261a : SUCCESS in 12m 12s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/e6bd34cff48c442a8af34ecb3177db98 : SUCCESS in 1h 17m 09s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/c568f507c7f34c998c30d85ae1d6749d : SUCCESS in 1h 55m 54s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/f1ef377bb9f64b88b12fd38bebad4aa9 : SUCCESS in 1h 54m 02s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/d1450b98678c4472b5c4c08fb646c4a7 : SUCCESS in 1h 21m 38s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/7dffad7bc24d448d9a2ab6bf2c6d1d37 : SUCCESS in 2h 19m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d85cc40172b5436f828f2a78e7d5f7e1 : SUCCESS in 37m 04s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ff8f52bdf8054d4f9c30a90af8062266 : SUCCESS in 1h 19m 55s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/42775aa45d9b4120a9921477e273858c : SUCCESS in 1h 48m 03s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/d238d0ded96c4a72914b1edba6c0e341 : SUCCESS in 43m 29s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/3313936fd50a4562990ec12054a471f4 : SUCCESS in 33m 14s","accounts_in_message":[],"_revision_number":3},{"id":"8796b1de9d88098a1558eb850bce30e3e284eb68","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-29 00:13:37.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"3a5dd000891e3560a22cabae3770914d98b31fa2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-29 02:20:48.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/46f6a446ee714af5bf407a5bed7b8b0b\n\n- grenade https://zuul.opendev.org/t/openstack/build/9b5af068233a43daac7fb0d3bef2df9c : SUCCESS in 46m 00s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b0cae39766b7429dbc7a99f047819760 : SUCCESS in 39m 56s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8dd3906188d54aac9c86de5ce599f8b6 : SUCCESS in 1h 16m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/da777241936b4d1ebff5ef308ed6065f : SUCCESS in 1h 04m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99e78ef1527647dd8831f3c1711aebf5 : SUCCESS in 3m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c7f8597ab4b94c9eaf74e9d9b5eb9fce : SUCCESS in 4m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b8fa9420a7e144fb8aca7cff0a58d26b : SUCCESS in 4m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cbfdaaaaa7a84ea9afae55085b107685 : SUCCESS in 4m 33s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a57305c565984492a573a4e7a0826817 : SUCCESS in 7m 44s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/db5199d492c644b6b417f547b229dda2 : SUCCESS in 6m 50s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/437c59572e914a7982e71b749fb7bfce : SUCCESS in 12m 54s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/469b1cd601554370af1c20233b3649ef : SUCCESS in 1h 36m 52s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/d7edcf92c4204da4aa7436928a0c971e : SUCCESS in 1h 18m 04s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/bc916a2a2c78497bac4dd157407f78ea : SUCCESS in 1h 58m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e5dab56d1e0842ea92cbdd59ed633686 : SUCCESS in 41m 01s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/adda5ad71e5947fc88a6cace648477c0 : SUCCESS in 1h 38m 52s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/a9fee2f53c3c48e897ae7b216c2ab059 : SUCCESS in 20m 49s","accounts_in_message":[],"_revision_number":3},{"id":"4f0cd6d6c3aa7eb42bd8fb690bf00b6242f8080a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-29 02:20:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"3c38ca687afec88b103af8e2df3a145128b3f94b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-29 02:24:16.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c752a518e35e46d39ff101f725e8c84a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4ea661ae0dd48ab8ce0a509363b13a3 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":3},{"id":"41e85f3dc96c54baf0f9eb26e20fa43ecf470722","tag":"autogenerated:gerrit:revert","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-10-29 16:17:37.000000000","message":"Created a revert of this change as I013e5a90a4dd5b6ba436b9f57ffe88fdc29eb326","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"704b24fd67b6d3dee5f1f42f9b793832b4b47bf4","revisions":{"7717b4a19330f92808a8d444dfb76648e1ca6308":{"kind":"REWORK","_number":1,"created":"2024-10-28 09:08:32.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/50/933550/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/50/933550/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/50/933550/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/50/933550/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/50/933550/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/50/933550/1"}}},"commit":{"parents":[{"commit":"2aa5bb356e65709795e036ca636bf46d664298ff","subject":"Merge \"Remove no-longer valid comment about IPv6 workaround\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2aa5bb356e65709795e036ca636bf46d664298ff"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2024-10-27 09:35:54.000000000","tz":330},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2024-10-27 09:36:40.000000000","tz":330},"subject":"WIP: Test increasing per project quota to 50","message":"WIP: Test increasing per project quota to 50\n\nChange-Id: Iaafcbd4ff858bf905b51d80bfda89e8a59b235e4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7717b4a19330f92808a8d444dfb76648e1ca6308"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7717b4a19330f92808a8d444dfb76648e1ca6308"}]},"branch":"refs/heads/master"},"234ff7392b40232f237eb54e985965f6a1ea90c3":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-10-28 10:08:39.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/50/933550/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/50/933550/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/50/933550/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/50/933550/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/50/933550/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/50/933550/2"}}},"commit":{"parents":[{"commit":"296398df18c4d310345dc58f08a07b4200406fa4","subject":"Add new glance.wsgi module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/296398df18c4d310345dc58f08a07b4200406fa4"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2024-10-27 09:35:54.000000000","tz":330},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2024-10-28 10:08:39.000000000","tz":0},"subject":"WIP: Test increasing per project quota to 50","message":"WIP: Test increasing per project quota to 50\n\nChange-Id: Iaafcbd4ff858bf905b51d80bfda89e8a59b235e4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/234ff7392b40232f237eb54e985965f6a1ea90c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/234ff7392b40232f237eb54e985965f6a1ea90c3"}]},"branch":"refs/heads/master","description":"Rebase"},"704b24fd67b6d3dee5f1f42f9b793832b4b47bf4":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":3,"created":"2024-10-28 19:12:34.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/50/933550/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/50/933550/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/50/933550/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/50/933550/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/50/933550/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/50/933550/3"}}},"commit":{"parents":[{"commit":"7a98b9f0f6c1df88ea776c31d8c7f8a746559308","subject":"Doc: Add info about multi-attach volume types","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7a98b9f0f6c1df88ea776c31d8c7f8a746559308"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2024-10-27 09:35:54.000000000","tz":330},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2024-10-28 15:36:00.000000000","tz":330},"subject":"Increase per project quota to 50","message":"Increase per project quota to 50\n\nThis is a workaround to set the default volumes quota to 50.\nThe correct way is to release the openstackclient fix[1].\n\nUntil the OSC fix is released, this patch will unblock gate.\n\n[1] https://review.opendev.org/c/openstack/releases/+/933578\n\nChange-Id: Iaafcbd4ff858bf905b51d80bfda89e8a59b235e4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/704b24fd67b6d3dee5f1f42f9b793832b4b47bf4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/704b24fd67b6d3dee5f1f42f9b793832b4b47bf4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
