)]}'
{"id":"openstack%2Fglance~933601","triplet_id":"openstack%2Fglance~master~I32416d38e295d32ae24652ec83716ca82eb76b06","project":"openstack/glance","branch":"master","topic":"glance-as-defender","attention_set":{"19138":{"account":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"last_update":"2026-10-09 13:08:03.000000000","reason":"A robot voted negatively on a label"},"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2025-05-15 14:55:52.000000000","reason":"Added by \u003cGERRIT_ACCOUNT_37598\u003e using the hovercard menu","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},"11655":{"account":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"last_update":"2026-10-09 11:31:00.000000000","reason":"Vote got outdated and was removed: Code-Review+1"},"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2026-10-09 11:31:00.000000000","reason":"Vote got outdated and was removed: Code-Review+2"},"4393":{"account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"last_update":"2026-10-09 13:08:03.000000000","reason":"A robot voted negatively on a label"},"16137":{"account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"last_update":"2025-04-12 14:13:35.000000000","reason":"\u003cGERRIT_ACCOUNT_37598\u003e replied on the change","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2025-08-01 15:05:50.000000000","reason":"Added by \u003cGERRIT_ACCOUNT_37598\u003e using the hovercard menu","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-10-09 11:31:00.000000000","reason":"Vote got outdated and was removed: Code-Review+1"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I32416d38e295d32ae24652ec83716ca82eb76b06","subject":"Add GPT disk_format","status":"NEW","created":"2024-10-28 17:24:02.000000000","updated":"2026-10-09 14:04:19.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":16,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"94aaa9531c1689870af3fe3d12f99fc19987c775","_number":933601,"virtual_id_number":933601,"owner":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-10-09 13:08:03.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"CC":[{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-28 17:37:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-10-28 19:34:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-25 06:33:02.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2025-02-21 12:19:39.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2025-02-24 18:22:34.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2025-02-25 18:33:46.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"},{"updated":"2025-04-12 14:13:35.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2025-04-12 14:23:41.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2026-10-09 11:32:03.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"state":"CC"}],"messages":[{"id":"9e65e2bd27638d4c40f499484c991f147a00aeea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-10-28 17:24:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f538b9cede8c88c3f1e1fb271e9522c1a7294c56","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-28 17:37:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d4154f07f674a3eb9f3776211b3d3cd\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/669d0b7bc56a453c9c4febaf107d13d7 : SUCCESS in 11m 33s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/44ea448d5caa46609955363977186542 : SUCCESS in 10m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c5329ce9a7695629150061e6b30aa2cc6ad2a07a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-28 19:34:24.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/230120d7536e403caaa89da3369df42f\n\n- grenade https://zuul.opendev.org/t/openstack/build/0675f8989b9e42eb9405a37e0ea3f558 : SUCCESS in 59m 49s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5a0fc5f73d49454cb0a8a6ca8af2d488 : SUCCESS in 1h 02m 58s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b15e710f310445f4a70fffe7a5e3cfd5 : FAILURE in 1h 29m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/877f61a524cc4585b194314ca0975523 : SUCCESS in 59m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27689246630c46d687d32abd94383732 : FAILURE in 3m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d35be54acdb540fd9f84f4a0498972da : SUCCESS in 3m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9fbdc662a8ac41b18941a9ae1ad819f7 : SUCCESS in 5m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b023a1a72f3413c8b347ba770a1380d : SUCCESS in 6m 28s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c3f3fb3bdb074db3bc8ede67449c8563 : SUCCESS in 7m 30s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a666cb07c57a456da2fb9ba349d6c03f : SUCCESS in 8m 10s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/0c7dc0d26d8a4b2ea078d0790b671d0f : SUCCESS in 13m 46s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/facd5883c8094797b46b1f75c58c519a : SUCCESS in 1h 11m 45s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/2d398b4cd92840728164e74057cd5ccb : FAILURE in 1h 21m 54s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/38835475716c4629a5ddb33c8e3f81ad : FAILURE in 1h 25m 49s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/a0370178f5ed43d584816b4d9ce91d6c : FAILURE in 1h 23m 03s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/6ef6f0cc19a74a81a9df92b20aac2d07 : FAILURE in 1h 33m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f28351bd210340e2a2e76a4e3aec78ea : SUCCESS in 45m 06s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c25dc4709bd04343bd74665d64ffd57b : SUCCESS in 2h 04m 14s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/d5f5639ba1f44778817e2fc3dea93456 : FAILURE in 1h 55m 23s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/c385dcff3c3645f1baea3b985c139619 : SUCCESS in 1h 05m 53s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/ba5da3cf1c2a4b5e960da9d4d1bbe4cd : SUCCESS in 33m 26s","accounts_in_message":[],"_revision_number":1},{"id":"f7eb258f419a1e42c38adac6a30c4c3a3a8b092f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-10-30 15:40:19.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"ae2e962bb0a6423e1555f8b3e0d4f87c2504672e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-30 18:37:03.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8130106a17a8476b9c597d793a755269\n\n- grenade https://zuul.opendev.org/t/openstack/build/7e1a4598ff144fb28881fa9d415d75fb : SUCCESS in 46m 54s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b082da2a5abc4b36b38a1ce005d237cf : SUCCESS in 1h 01m 45s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9fc98c02c7d84ee6886dac6865bb6cb6 : FAILURE in 1h 29m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9d88ff3c7efa403ab975693846033642 : SUCCESS in 51m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/446182c925574cd0b135afb3e7192370 : SUCCESS in 4m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e49d499e3e3847bfa9f98e9511b2efa2 : SUCCESS in 5m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1d04e0d1be074b768c449c8a9510a643 : SUCCESS in 5m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f30f0c3854c4e09b4cbf14a4d29e4f4 : SUCCESS in 5m 37s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e31473e23e1147f9b9335a5f2b084e61 : SUCCESS in 20m 17s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/889232f72f3245139a27a67db1c177eb : SUCCESS in 8m 10s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/3625131fb93e4d9f9add6584ea6b115a : SUCCESS in 24m 22s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/cf719c1a05804c569019f7022b30c5dc : SUCCESS in 1h 51m 52s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/3d3e9d3ff0284a2481a7b43bb59a65d3 : FAILURE in 1h 43m 42s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/6e9677a4eb594b57b7a0aa80685ac290 : FAILURE in 1h 45m 52s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/be5b11a2ba98409081d66a9a6621ca97 : FAILURE in 1h 45m 32s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/60e7e064568b444cb28db37a1979f9b5 : FAILURE in 1h 13m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9cfde3d0c2224850af08283e9ba3d7bf : SUCCESS in 1h 03m 47s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/3306de9315a0434da8cf657bd585d68c : SUCCESS in 1h 51m 35s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/11b054b01f2e433c8d249625208e55c5 : FAILURE in 1h 52m 20s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/e4845af6f2e34cfdb2dd5e22010090d3 : SUCCESS in 1h 02m 31s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/0cfa1f7200fd4aab9f28ba6a97bd3926 : SUCCESS in 36m 06s","accounts_in_message":[],"_revision_number":2},{"id":"e8ddd6f147a3fe59d037498547be2c8e39906a89","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-30 19:13:52.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d48d2d74a7ac4600bbbd90aafe97144f\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/04db17bb392d4da19ef2f5153cdd1e9f : SUCCESS in 14m 20s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9a48e094384941f4ac2596bf2c478e80 : SUCCESS in 13m 12s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8005760db47dca7f514b596aba32a9ecaaf1aa5f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-10-31 14:27:18.000000000","message":"Uploaded patch set 3: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"8ddb27422cd470e4fdf92755d177c45a7a75345a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-10-31 14:34:06.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"c47f544ba9b5ea0ee386f6610d005bcfd556b952","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-31 16:58:02.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5b3e9ca038044a108604491a2019a001\n\n- grenade https://zuul.opendev.org/t/openstack/build/32e255b33a6a444f8c1077ca6f57a032 : SUCCESS in 1h 04m 19s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/85ef4268b5e04a9fb3f57b8f447072cb : SUCCESS in 1h 00m 45s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a6cf50d6cca042debed4042df5b01d47 : FAILURE in 1h 36m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7896564d0e1d46d58313e35090536765 : SUCCESS in 1h 10m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0870583e13a48cca3837aa8bece82ad : SUCCESS in 4m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/619ca59d037845a3aa1d24ed7056b436 : SUCCESS in 5m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/40a7d735a12c42a9aa3fd1644e8b6929 : SUCCESS in 5m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99cad7252f8e44538049e87e2ab7377f : SUCCESS in 5m 44s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e6f64fdb2a4f4f67b5d3a0848d42f40b : SUCCESS in 8m 46s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/af1930a1b09042d8864e0f5f3bedcdda : SUCCESS in 9m 21s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/6e16ca8e74514befb5d37e76a99a3fd2 : SUCCESS in 13m 15s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/da5ef130f9c446ad9e5513e306ec1b2e : SUCCESS in 1h 40m 36s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/4b4a69a76cfe4ffe9763a9b0cba71bf2 : FAILURE in 1h 46m 55s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/931b9f3d3ea040468e3423c9b4637add : POST_FAILURE in 1h 01m 47s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/8709e922319d46de9899f8473205980d : POST_FAILURE in 1h 16m 07s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/6d1ea47615cf4387823122b957196805 : POST_FAILURE in 1h 14m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d37b1a3726dd4afea4e13317ba0fd83c : SUCCESS in 1h 00m 51s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/5a16fd74165e4787be2077d7238669fc : SUCCESS in 2h 10m 53s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/b17b5106d7c14302a300fe6b89279d1b : POST_FAILURE in 1h 11m 28s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/3054d2e908d44dcb8e552c4bd1268a2e : SUCCESS in 1h 20m 02s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/b07332c4d1ae4fdeb1624189e5b74639 : SUCCESS in 39m 11s","accounts_in_message":[],"_revision_number":4},{"id":"6b52b66f19073a90fa1770ab349cbec65634b51f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-01 06:57:21.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aab196b111494bada31644ad3ed3537e\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8ad48c00b8bf4f92ab6ca9056b9148a7 : SUCCESS in 28m 04s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7cfbaad1bc3648289260d4b0aa5efaaf : SUCCESS in 28m 17s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"8c2ba997198d69f90748aa4ffa61d26d95c73907","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-11-01 14:31:03.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"1499be73171b86d018d1cb5769f47579546c4bec","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-01 14:45:13.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1483cf744856414ea88c79edda9ddea9\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fd28c3b16e494aab9879717b4be0a7cb : SUCCESS in 11m 05s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ce52133033e74bfaac9ab1819d4de723 : SUCCESS in 11m 18s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"2b448af857d409541dc012590693c22a6e0282b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-01 16:23:41.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/71883526cfa64264b012404ea10c17d3\n\n- grenade https://zuul.opendev.org/t/openstack/build/e262c73a74a949c8aa35a8e1abc367fd : SUCCESS in 1h 02m 55s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2e38067709e445eda46a799bcae0d5de : SUCCESS in 1h 01m 37s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/32bbf7b024a0405e82251a9126ceacc7 : SUCCESS in 1h 29m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9f3fad6a82a5419d90bb82f992ec792b : SUCCESS in 58m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e0e3fe6711e4d57a54b7e9fa6da4192 : SUCCESS in 3m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0f67b4f704534cf3a2688e833da09500 : SUCCESS in 5m 27s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/82acd229e49d4f578ca1218232157144 : SUCCESS in 5m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8da4177cc32a4f368b505483223617b8 : SUCCESS in 5m 43s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a0900c3497484a46a5a5cbdc941342e2 : SUCCESS in 7m 54s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/25dde7064d5743b0b758a578e160e3c6 : SUCCESS in 7m 55s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/8d71d3806fc24c3796bd853b67b03d28 : SUCCESS in 12m 56s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/cf0ac0422e17468d882fc9e23bcd8e71 : SUCCESS in 1h 37m 33s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/7fa030d30f4145c49fa3f4efac12103a : SUCCESS in 1h 46m 41s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/abbbca79d17a4c20b60a1f3b287fc37d : POST_FAILURE in 1h 08m 21s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/90e7e4d504d7492092cbfb5a24ee1d1b : POST_FAILURE in 1h 07m 54s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/a6ef18d42e954e87be7e3e8c02b61ab1 : POST_FAILURE in 48m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a185da9f8e6d438399fde2ae3e6353e2 : SUCCESS in 55m 07s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9e2158adc3c8433385bd289acd5a78b0 : SUCCESS in 1h 42m 54s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/386b648c26024c6dbf380119bbb495ac : POST_FAILURE in 52m 42s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/a92282b7b5c146e2936851b98ff1bb75 : SUCCESS in 54m 34s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/38925cd052314b3780b3c5f28ba67f84 : FAILURE in 28m 37s","accounts_in_message":[],"_revision_number":4},{"id":"6ca43c7863fe9b0f6062c01eeb05749ff733ac4e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-11-01 16:53:39.000000000","message":"Uploaded patch set 5: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"29bd6e4e06dd1e968f320e2a15c3f7eacf161472","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-01 17:52:29.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bd2fdf393c56419a9ca1e3333cf19f21\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/05dbbfc79012434298a3bd6a2bda56de : SUCCESS in 13m 56s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/70ccb666c74f47898312c9bfdc34c2a9 : SUCCESS in 12m 38s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"8c9e6562bb810dd99e9d0a7c64e72f9fd09ecef6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-01 18:25:34.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0fee3820bd8e4386b6cab8d11ca752c7\n\n- grenade https://zuul.opendev.org/t/openstack/build/77939be69d3142eda22a3b3c27dc54b4 : FAILURE in 1h 06m 28s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/49e8b0579ebe4c4fa537f1de496dab71 : FAILURE in 1h 04m 28s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f12f9ea52dcb4a438908f5367736107e : FAILURE in 50m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d29a769bddd04358b8357f83560af9a0 : FAILURE in 47m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f85076f440548828c883dbb6f1c0e23 : SUCCESS in 4m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/947678401f9445c1adfca7dbb15ba12c : SUCCESS in 5m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cee4d0b3998a48ddb9cb3bfd0048b6b0 : SUCCESS in 4m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/399f6c0571d947a094fbccf100cf5abc : SUCCESS in 5m 29s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e55cb9dfcb7a4e3b9e8346c7e4376f0f : SUCCESS in 9m 51s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d5b2bbb0e37b4ce6880542a459a19ac5 : SUCCESS in 8m 36s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/d69f4e93f8854367b58cab7cd735b69f : SUCCESS in 11m 59s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/540b0c531ebc4499a9f73061c0cffc32 : FAILURE in 1h 09m 42s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/579c8d8d2c454286a1c18463cc4753ff : FAILURE in 1h 08m 55s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/ecf6be060bcd4538b04e47aaf09f8322 : POST_FAILURE in 1h 03m 14s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/7ac82683144a4327bb437a83b3eb3727 : POST_FAILURE in 38m 32s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/83855b89f4d64d80a7553c448bd9d033 : POST_FAILURE in 1h 09m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a05aa96515174fec9983a59d3d5aebbe : FAILURE in 27m 22s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/7a139a2900e24f9b8b64bc2ee36908a6 : SUCCESS in 1h 29m 05s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/859c845345a04064a7898bc45ac619ea : POST_FAILURE in 39m 10s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/7a3c00757c564a9f9bb03691917eef24 : SUCCESS in 50m 43s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/b5b0316625c54890a32922b3d057b516 : SUCCESS in 34m 41s","accounts_in_message":[],"_revision_number":5},{"id":"2d34d082497b587a7197fbbdb1fb7f36e4484f68","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-11-06 15:23:08.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"a3de73b888361414bcdefa375e9fd179d9b33505","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-06 16:13:59.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a996407feb0940d5b353584d3ef91f65\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/34cb674b420744aaaf84cee8cfe8a846 : SUCCESS in 16m 52s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ce08b739836c4897853818ca26135631 : SUCCESS in 14m 18s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"5ae0cf738789b722cdd482b512702f950e1d063d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-06 17:28:05.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/897d29ff492841bfa3db8339dcdbe9c4\n\n- grenade https://zuul.opendev.org/t/openstack/build/b8d5e84668324dcaa48e03cb941da35f : SUCCESS in 1h 09m 20s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/79e721f9769b458785ecedf6d6da773f : SUCCESS in 55m 41s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1e0c70671da04f36b472221afec34401 : SUCCESS in 1h 35m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1edfbb765365490a9b117cb520af7e67 : SUCCESS in 1h 02m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/043cf33e43e84f369a8a20808c164e0b : SUCCESS in 4m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1b13af406f9a4762b6f31d16922e49f2 : SUCCESS in 4m 28s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ffbf88c255104b65afe46c6082a92829 : SUCCESS in 6m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c87f238fcb944a57acc158f5eec3ab49 : SUCCESS in 5m 29s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a230b30ee2344622a373a44d6fec263c : SUCCESS in 8m 29s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ae9cab4e332648359c463607ba30e467 : SUCCESS in 9m 24s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/4b07e473c3c94944a7616f17ae33396a : SUCCESS in 11m 50s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/40df9e91853d4fda9eaa20db2c6e20f9 : TIMED_OUT in 2h 03m 18s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/e39bda5502f6434d9a4361d6e707ca21 : SUCCESS in 1h 55m 10s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/5eeddb3f633447cbbe92c23e40c62ac7 : POST_FAILURE in 1h 49m 03s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/0d38e7e9f7994b438441673cff44d0ed : POST_FAILURE in 1h 39m 03s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/fb9d64c3f99a4b798f5eb7af5923e799 : POST_FAILURE in 1h 42m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e07c52fd859d429b92663febf0a57f17 : SUCCESS in 1h 01m 01s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/406fdccc21c24f2e936f2e84d1c63e2b : SUCCESS in 1h 41m 13s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/9f106315d83c4d59abaa58b156c10c9b : POST_FAILURE in 1h 53m 24s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/a04a9aa806a44099b7414308c5700d61 : SUCCESS in 1h 00m 21s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/7aa5ce5377f9482dadf21030e7bddd42 : SUCCESS in 21m 44s","accounts_in_message":[],"_revision_number":5},{"id":"603849d04debab40190397b1815de695c15bdd4b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-11-06 17:43:22.000000000","message":"Uploaded patch set 6: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"1ea6f481d50c87fccc87990b125dff5b6b52aa6c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-06 17:58:38.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3191602667cd41648b707e4e60dd776f\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b03bce0e98c64056be166de47da42e66 : SUCCESS in 10m 54s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2860265c4bfe4b6aabf1dfd2121c56ad : SUCCESS in 12m 14s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"b0ef015316f1969153e6631e00602056c2d814d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-06 19:36:07.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9c8d22a0a03e44fbb60dcf393235a1c7\n\n- grenade https://zuul.opendev.org/t/openstack/build/d4840ffae27a4d3694aecdf8873e980d : SUCCESS in 41m 43s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c3a625689a634f859b6f322ec3a0fbe3 : SUCCESS in 53m 39s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/77eec60fd4164765985f1caf60c33436 : SUCCESS in 1h 25m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3f2a09f2336a421085a342614c3fd45d : SUCCESS in 1h 13m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ac6f5102837400e9278e9f5174b7d4e : SUCCESS in 4m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/252ad52f54bb4480b368f6e8d57162a1 : SUCCESS in 5m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/48d5b7c02d2c47478e92dd9bb3678951 : SUCCESS in 4m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/720fb8ef54464105b1e103b95273d929 : SUCCESS in 5m 37s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/828e8f97c77b43e4b7dfbd6b49443446 : SUCCESS in 8m 59s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2deb29cf9ca345c2bebc192f49b0027a : SUCCESS in 7m 19s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/eac32dced7c0461fbe5c9c33a98eb89c : SUCCESS in 9m 10s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/1052a4980a9a4c3cb3f9c9c29dad3944 : SUCCESS in 1h 38m 42s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/bd9c8c6cc68b4349b34bf8d9493a8390 : SUCCESS in 1h 48m 16s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/91c2ad7d14a94ece881401c1fbc2c264 : POST_FAILURE in 1h 19m 44s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/3da3f7f189d644d29b3ef4e599a3f765 : POST_FAILURE in 1h 28m 20s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/26d7cf33edc14b639a6d937e4796190d : POST_FAILURE in 1h 24m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b45f26663b67447ca9df70ef94828be1 : SUCCESS in 49m 41s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/4a9f0ebb7e1a4927ac166bf6f5bccf71 : SUCCESS in 1h 44m 41s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/21b50d496de64ffaa377f7f572d7c084 : POST_FAILURE in 1h 16m 32s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/972a807c20864f9592ade51c2c8d06cd : SUCCESS in 1h 03m 18s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/6121ca3cf75c4a0eb4c270c4a5cb2aca : SUCCESS in 32m 52s","accounts_in_message":[],"_revision_number":6},{"id":"73d0811ab17b7f0dc4e5950ac0d1759abc8bac33","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-11-08 15:29:37.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"c90c809590f115a589a1540a0eaf7fddb0ae9165","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-08 17:31:36.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3f364d519bfd48158c2258646de66288\n\n- grenade https://zuul.opendev.org/t/openstack/build/d01abbeb53f44bbbbfa0b28459deecca : SUCCESS in 48m 23s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/073e39317f864a37b10903834eb89a17 : SUCCESS in 57m 19s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/910599469abd4de18be4a13b7c1bb0b8 : SUCCESS in 1h 10m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/078fd6f498774e308bf7ceda50dde92b : SUCCESS in 1h 09m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be765349efd3435b9e26956b30e6aea3 : SUCCESS in 3m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9f5787da1de54cc68db1323003b795af : SUCCESS in 5m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/26bb5740c84945aabd8657118ef79975 : SUCCESS in 6m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b96460cef2f446a1af89904b9e5e491a : SUCCESS in 6m 52s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f9ec79485efb4e649f0e5923efad41d3 : SUCCESS in 8m 56s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/175d77c99d294c678d32e1c845451ccf : SUCCESS in 7m 35s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/4b825c98a02e47e89f585cec9771868a : SUCCESS in 12m 18s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/deefd19930bf477ebfbff830a0011777 : SUCCESS in 1h 48m 22s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/d2674ba621b544649a8dff7d9c8b55c9 : SUCCESS in 1h 23m 53s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/32e393f309cd480b96406bdda1b5f7f7 : POST_FAILURE in 1h 55m 43s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/f8cb954cc2bd445897f9dfe3a0cdf95e : POST_FAILURE in 1h 09m 03s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/22f7da69fdca452ca39ec9bfdee05605 : POST_FAILURE in 1h 30m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/55774e44042448ffb93767a54164e52b : SUCCESS in 48m 36s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f07515ea5da8484d9937446e1d543b9d : SUCCESS in 1h 47m 35s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/bdc8cafa97d74517a7413139f36ca774 : POST_FAILURE in 1h 26m 24s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/33193ac0f88f478799be65ac4f259a0c : SUCCESS in 1h 04m 08s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/7780518eb35549c795a633304025e59f : SUCCESS in 24m 48s","accounts_in_message":[],"_revision_number":6},{"id":"7185d8a2686142e8c2a16c4b70d274c614c1fe8a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-08 18:49:58.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b603d612362f4e3fa24c87459ce3c295\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/62eb25c1081a42ba9f5c571e7328bc59 : SUCCESS in 15m 31s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f244085e2ec74a468bd3c7237df3975c : SUCCESS in 13m 22s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"d8f4c89e586df6945f5f6433d371418f6b3ecd48","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-11-12 15:52:52.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"94ceb52ea674769464361885e4d4482fabce1ac4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-12 16:07:15.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1f33f1c31a674ccfaac9b4812428929a\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/365bd385949442e4935d6ff8f0e2f68d : SUCCESS in 9m 54s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/63527070c72648869624865d7408425f : SUCCESS in 11m 19s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"e2e35f7669d5b1940c3e19f8f310376c18cc5239","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-12 17:55:39.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d0b498eaa34d4f7b9a456a277129f150\n\n- grenade https://zuul.opendev.org/t/openstack/build/6c92810c49f741c598fa669a55d31631 : SUCCESS in 1h 06m 34s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/36d7580ac61743078f90411efe63e1ec : SUCCESS in 39m 20s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0cfa18a2f8b94d08ba85046333a08718 : SUCCESS in 1h 20m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e752732c2f514f55b481ffd5c28055a9 : SUCCESS in 43m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bfb45de66b4c4796a4758fc7eb082374 : SUCCESS in 4m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b3a02ab8083c4d85a3c0b183572cd21c : SUCCESS in 4m 51s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c3dcdeafeef34f2494694638e075dfc9 : SUCCESS in 5m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e5306dbe9dc4867a2f5f237427bbd1b : SUCCESS in 5m 27s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/36c0229acf8b4b9a8f1a516eb82cc767 : SUCCESS in 8m 10s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/80964c765e67483ea7d1962745481b8f : SUCCESS in 7m 59s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/59a23a5ec79e465f8b55eed28dd05d14 : SUCCESS in 14m 16s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/c702f53318184308ae419d609a13f4f1 : SUCCESS in 1h 40m 44s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/9a7c931adc4648d7b6335ff22ad8b8a5 : POST_FAILURE in 1h 49m 57s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/c0b799eadda1494ca342e55b11f6d70e : POST_FAILURE in 1h 51m 26s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/ab7e00a6adfe43d3b6d7742ece664ae5 : POST_FAILURE in 1h 40m 36s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/f52d54d8f4ef4953ac4cc30a129b2ae1 : POST_FAILURE in 1h 37m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c5e3273e9ee74f4087469c491e80afc8 : SUCCESS in 55m 14s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ca1215b3a4f5446da7c49e17a58a5c98 : POST_FAILURE in 1h 34m 39s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/5e11b0dfa25c40ffba3cdb5dff701ebb : POST_FAILURE in 1h 43m 41s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/38092b0967384de681b3ef415e3f18ed : SUCCESS in 1h 00m 29s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/902a9b0a42be41b69eb2514b34c53236 : SUCCESS in 30m 13s","accounts_in_message":[],"_revision_number":6},{"id":"ddaa5180640d0eb6c80fb9e9d13c396057cb2125","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-11-12 18:01:31.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"83f9e38529720664e73614094842cce829a10a2b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-12 18:14:44.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/afa6e39227e64dde9238c028d0702327\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/20390d63f2dd43a48c3251407ca52839 : SUCCESS in 7m 38s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/759496902c5f45d48b3d613c8db156fa : SUCCESS in 10m 16s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"4bdad5911dbdd2d38bcbd77e899bfc4d211a6c01","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-12 20:16:24.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bc2009695aa84bd5a05650254f4827f7\n\n- grenade https://zuul.opendev.org/t/openstack/build/81f91a41cf51448c943d64b3857f3851 : SUCCESS in 47m 57s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/adef366889c64ec09521642f393f115a : SUCCESS in 56m 18s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6351cb87ccf94a2bb81b556ee3163edc : SUCCESS in 1h 19m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9b6bbfc5acc34359ab74389f006c1275 : SUCCESS in 57m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d5971c76a754d5e857d9337fcd59f14 : SUCCESS in 5m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/986bd67d6347480ea221e51d91e7abb6 : SUCCESS in 6m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9b6414d777264275b81ba4be27ad37fc : SUCCESS in 6m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4f5b3e0b1c8417682c9af64a5512928 : SUCCESS in 5m 28s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5ce04bb9bfd74fc3969ae7a7d95e7a96 : SUCCESS in 9m 20s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/44691550a0ba42e89bf4f1997f861061 : SUCCESS in 8m 51s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/0c3d89cbe4384c2d87a2e99e8d1586cc : SUCCESS in 9m 49s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/59a781aecd184aa2aced3e63df7e34fc : SUCCESS in 1h 50m 11s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/1f735bd4dbb54169adc0c5bad001cd54 : FAILURE in 1h 54m 51s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/2ebbead7200540dfb21e31e722152551 : POST_FAILURE in 1h 16m 11s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/b052b31a05514e03a960cabcd88e70e7 : POST_FAILURE in 1h 25m 13s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/6c489294959d4ee59cfbbc62e4543bdf : POST_FAILURE in 2h 03m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dd71650f547d4cbda98e8aa32179b93b : SUCCESS in 58m 31s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/b029a51988e24a88ad03d7d71ac4c550 : SUCCESS in 1h 29m 38s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/00daaa8ab8614806b4e4eb8eba023034 : POST_FAILURE in 1h 26m 17s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/fa6fdab678f84e30839cd8fdda5b450c : SUCCESS in 1h 10m 20s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/a32f8cf82f3a4ee6b53fe1988482cbd9 : SUCCESS in 23m 17s","accounts_in_message":[],"_revision_number":6},{"id":"00a277d9aac9ce1b52a9141b1865e6a2cee68e08","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-11-13 15:33:04.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"48987871b0ed713e70daed55e0e8b350d6a47ac3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-13 16:00:10.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/07a35654a8af4893a15e1a5f9d6d8b32\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fea030255f0d4fa5b7b62a056c125e7d : SUCCESS in 10m 16s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0b2ecc854b9b4efa80d310b91ac05cc1 : SUCCESS in 12m 17s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"65f41cafc9d86ed2f60405c2086dcdba1a988351","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-13 17:43:41.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/becf9059ad8746b5a498d7d031d0bb76\n\n- grenade https://zuul.opendev.org/t/openstack/build/ce6b9266d6094cd89cc59576b3821fd1 : SUCCESS in 55m 17s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/66dc089faf7d4837a19996d284923e68 : SUCCESS in 57m 30s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d7f7a27541fb41449b1819714055716d : SUCCESS in 1h 39m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/085521bb54ff4755af93b642d664a8ec : SUCCESS in 1h 03m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b623ffb4e4e40b58922aa4ed773e691 : SUCCESS in 4m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5cbc3664fc394e3d82908278b456521a : SUCCESS in 5m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f4fe5ae2bfdb4bdc8f0f97a367429649 : SUCCESS in 6m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df46c2339b534404bcbb340352fc78b2 : SUCCESS in 5m 39s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/691e4a13686949889a8dc4ba9300ebeb : SUCCESS in 7m 25s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6d61430d94dd4661a501779899857bbc : SUCCESS in 7m 34s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/181ddfd681ec453085b92d61badf1ec0 : SUCCESS in 13m 40s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/3fee86a699ed4619b90742c7e2974469 : SUCCESS in 1h 39m 42s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/c0506b6b24fb44e3953e5cd3bf901fbb : SUCCESS in 1h 24m 46s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/4fee2f2fb69b41538771522856f04f0f : POST_FAILURE in 1h 57m 19s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/3e7734831dcb4b16868db30d2c2dfb08 : POST_FAILURE in 1h 35m 48s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/2a092b78500c4eaf98b7ec0d0a50425d : POST_FAILURE in 1h 59m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1c11850a9f43470bbc5e7a7303670095 : SUCCESS in 39m 13s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/4398dad7ae5d443686ddc0cfdef92bd1 : SUCCESS in 1h 34m 15s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/c08e521f31f24c31b83d1f27d820ab70 : POST_FAILURE in 1h 54m 22s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/2911b323ac0c4385bc003e3dd8975342 : SUCCESS in 1h 02m 41s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/cd5a50c98fc14ece998c9bc10b2f2920 : SUCCESS in 36m 28s","accounts_in_message":[],"_revision_number":6},{"id":"18cef4f184478c8a92dc994fde343e4ba0594217","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-11-13 17:56:40.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"ea0f4da183992ca83943e83cd8c5420a4b4d3654","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-13 17:58:19.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc884f98632b47f78c3b7d0a78c93189\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/38b810e45c7a442e97a831516a2c405a : NODE_FAILURE Node(set) request 300-0025673697 failed in 0s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/08df573788154eafbc348c1d2d7978b0 : NODE_FAILURE Node(set) request 300-0025673698 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"2d9dbb3b6c81e4cea772b131bd2b9ae929079823","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-13 19:57:36.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/49c9a0a3aaed402799056d555557095c\n\n- grenade https://zuul.opendev.org/t/openstack/build/65da4a930f07474b849af1daa12252eb : SUCCESS in 1h 00m 37s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/dc2b290769914f7e90584a246181bb10 : SUCCESS in 1h 02m 14s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/127f3327e3d048669e95a2a1a5856485 : SUCCESS in 1h 22m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9ab317eabbf444ef822496268b9e067e : SUCCESS in 59m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f06a57b06dd24c17bb8b7b81d44bee7f : SUCCESS in 3m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6a0762d5336f45b5b9450c5f62faa11d : SUCCESS in 3m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e90766f8476543da9dfe1ca32a5933d3 : SUCCESS in 3m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59a8ee8c70804c23970b2fa5c833bf14 : SUCCESS in 3m 52s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3710e90bc2ef4a299462af7e2a45dea5 : FAILURE in 7m 16s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/39c243ae09fd490ea860f419b36d25a4 : SUCCESS in 8m 57s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/002dc96ca60f44309385914b0d5d1181 : SUCCESS in 10m 47s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/edd7a86c6afb41c7a4da0203b0f66e89 : SUCCESS in 1h 39m 42s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/67b43616faf6427c95736b5d10bde1ad : SUCCESS in 1h 45m 22s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/9c651eac223a429dab40f9a2db41ec94 : SUCCESS in 1h 53m 41s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/bd26cf647fe6466cb80ae713b2780b45 : SUCCESS in 1h 41m 03s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/893b73792325477097933d0a914e8464 : SUCCESS in 1h 59m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/af253b9a72744e4c8b9a47f9764911ae : SUCCESS in 53m 10s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/5080ffb5dae84450b4261e44bf0bd64f : SUCCESS in 1h 11m 54s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/7b1d519b965546339d3a85d94760fcdc : SUCCESS in 1h 53m 42s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/d29d6b34cdba4620a0f14f7d7ddb7867 : SUCCESS in 33m 03s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/3d8054884ac841509506e329494cabe1 : SUCCESS in 24m 07s","accounts_in_message":[],"_revision_number":7},{"id":"28cf017cadd781ad44169a60cab8c254e8690e5a","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-11-13 20:10:12.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"f14cff53e5c4f2e84e33e7ed4f607a00cbfedbaa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-13 20:26:52.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f0da86b46cd491184e5c0ace4d7ea77\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b7f3576ca0724e0bbbd1833e9b2f3898 : SUCCESS in 12m 42s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/aac8ebb269244c5f84f20eea6da76417 : SUCCESS in 13m 53s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"8ca5e218a627fa07fde9029c8bb659d7431fe80b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-13 22:19:11.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e6619e982b2c40c89c7ba21371c643f7\n\n- grenade https://zuul.opendev.org/t/openstack/build/218cf2be3e9c46c997c0e07b5930dcbe : SUCCESS in 37m 00s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/411199532d1d4826a08fa6d1322d36d3 : SUCCESS in 38m 25s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d9dc47bcbe6d4d1bbe0a4464caaa51ba : SUCCESS in 1h 34m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e4249fe26c4342eb8c581cf1a916cc41 : SUCCESS in 1h 08m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c3e9bcdcee848dab716ee3aefa36ab5 : SUCCESS in 4m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f29ce95176e647a097cb6e645afd5b1d : SUCCESS in 5m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/900b1eafb53447e3a2d2b77acad202d9 : SUCCESS in 5m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d74aa76a177446adaf5cd495cee931e6 : SUCCESS in 6m 57s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a4de0bc5ece74ad4b034d4e8e56b0021 : SUCCESS in 8m 39s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a5133f9fd66e4492897453f8d4c60360 : SUCCESS in 9m 45s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/f6e65d96152c42c8aba71ca4845e7fd8 : SUCCESS in 12m 33s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/881cb251f00d47af8ad14fa3f0d2fcfc : SUCCESS in 2h 00m 59s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/dc46d1fdcc1c40da9a582553548c6585 : SUCCESS in 2h 00m 38s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/a672a17a31bd482b8215b45da5fcef97 : SUCCESS in 1h 57m 39s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/19566de9f8fc4a02bea2e08161375dd6 : SUCCESS in 2h 01m 47s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/a16725e61bd44884a5b7d68e9a324554 : FAILURE in 1h 54m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/85bac96fcb1a437788493180dd334945 : SUCCESS in 1h 03m 55s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ef083b0e4eee46bf932724db536b05c2 : SUCCESS in 1h 54m 59s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/b57986ba7c5e4ad08e4b73fce2f96511 : SUCCESS in 1h 59m 33s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/33ed31525b2e48e39a0f64bd2618c4d9 : SUCCESS in 1h 06m 03s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/2f0404148486455cac426a86c14b34d9 : SUCCESS in 39m 29s","accounts_in_message":[],"_revision_number":7},{"id":"0f9adb6f280ee0a37a9832dbd0ad8251615caf6c","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-11-14 00:01:20.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"3934cd9b740abf76295847a72eeeba478ed88411","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-14 00:13:23.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ef3591195f924e0fb25bf346f3f183e4\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1012785e984a43889937954b5abfaee0 : SUCCESS in 9m 47s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0857c167418d4c04bb821c414c9f1bac : SUCCESS in 8m 51s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"decb705618e933fe4ff15592403987e383cea9de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-14 02:08:23.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f3886f9345284bd6ba1a66d8a72a903a\n\n- grenade https://zuul.opendev.org/t/openstack/build/bbee8080e69e435996327cdca1755605 : SUCCESS in 40m 29s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ef0b5fe24f334b5c9e0db5df2adfde24 : SUCCESS in 38m 07s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9f237dddd8d74312aa2612630f50cbb4 : SUCCESS in 1h 06m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dae44c2ed460410c929aa81354dfbc58 : SUCCESS in 43m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26236fb3715a47acbf79664fc50ca2fb : SUCCESS in 3m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/39547d6f40694242a887d151306c614f : SUCCESS in 4m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5c4fedea49654634bc4c1ff7d7ad2513 : SUCCESS in 4m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9923b12aa5d548dfae0c1e2eb9ed7d5e : SUCCESS in 4m 13s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1482040cb9e2454e962396b80599dbf5 : SUCCESS in 7m 44s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/90b5fefd8b2d4f8f867ac82fdbbe0f88 : SUCCESS in 7m 36s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/ffc23a4339e644e68cfe934b4ed88328 : SUCCESS in 12m 07s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/0447adcd0aa84f12b13f79a3c2e301ab : SUCCESS in 1h 12m 06s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/c3cf55b74a9b49e6830ae5f3351052be : SUCCESS in 1h 50m 35s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/5af66bb6bd16465abc3728d8950b1706 : SUCCESS in 1h 16m 02s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/1b62cfd3b0dd4b43b0e91048b35569e1 : SUCCESS in 1h 50m 33s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/b487265c8c9741ccb8b0cf3762b04159 : SUCCESS in 1h 38m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/728e5252458947358229cd3dcf7fecd2 : SUCCESS in 1h 05m 59s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/2cdfbbe7aec04f6181c32e5bf467ef64 : SUCCESS in 1h 43m 21s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/6c22078aa8964fea85fcef240ce65364 : SUCCESS in 2h 00m 18s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/256a19762384414b8b2b361deacf4a4e : SUCCESS in 1h 07m 18s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/5980e7a3e2b347908dd20712fb8890b7 : SUCCESS in 19m 59s","accounts_in_message":[],"_revision_number":7},{"id":"1e124fb4e85cc102f6997e8b8148ae97d249fb65","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-11-25 06:33:02.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"625f4d9eae1f174e0dc1fbe9e573c59708d9956d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2025-01-14 14:33:36.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"b5a600c90964bf9ebcb67cb824b7863dcf899979","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-14 15:32:54.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f685d44ba7324febbc53b8fd4e9bc208\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/055ec34c1edb4b029adb38fd7920a16c : SUCCESS in 12m 03s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b4de9cc8148d4c869666e34b68dbc189 : SUCCESS in 13m 21s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"74585db24f41a8f1a772e1fde8122f158240afdc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-14 17:03:01.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8f408a5f802848958b647d5128b3ab3f\n\n- grenade https://zuul.opendev.org/t/openstack/build/dac507fa37da443980ada62bf9ad50de : SUCCESS in 1h 12m 13s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d9435fdf80e345d3869f06e0f1738f46 : SUCCESS in 1h 01m 25s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/34e4c219fcf24daf894a3ac9d097073a : SUCCESS in 1h 32m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3b1c3fe9f8a84086b7dc35a58568b2da : SUCCESS in 1h 10m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1b160ff2d554e039b81dd6a7fff4ac9 : SUCCESS in 5m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f12e339efb664745943ae09081607224 : SUCCESS in 5m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/52a8817e921a4939a6e6405a212d0506 : SUCCESS in 6m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30b1f513e0244544ac988e9e140ac1a6 : SUCCESS in 5m 57s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/90465a2ddb0443a19849ed2df7b8516e : SUCCESS in 7m 56s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/bf2b25225fb14f4b92eb1cd8709d1211 : SUCCESS in 8m 10s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/d8aa21b7ee724583bb83d6db2a31442a : SUCCESS in 13m 00s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/4b13b61350b741a79c680521b3716df8 : TIMED_OUT in 2h 04m 15s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/0796d5c1f3a64b76949acbe143efa55f : SUCCESS in 2h 11m 58s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/468654d782294d5188e18f6d913b588a : SUCCESS in 2h 03m 30s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/795758e324ac4fc2a725e2d6547437fa : FAILURE in 2h 04m 27s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/03f87e79bc514417a74b446ad21d0892 : SUCCESS in 2h 16m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8127e3a556454217bdb72c84b427bb00 : SUCCESS in 36m 24s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/14fa3b13546c4629800ff62cf6c3c62a : SUCCESS in 2h 11m 14s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/87b1aad2d0274f53898b4f787b3baf06 : SUCCESS in 1h 52m 13s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/087abb4c5aca437ea89e5ba6a7fcd9f1 : SUCCESS in 1h 07m 53s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/761da13c99f84eb1821a3a89a70e4399 : SUCCESS in 1h 59m 01s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/4b9f29c9de7d4b1b968ee6c992a902c7 : SUCCESS in 36m 14s","accounts_in_message":[],"_revision_number":8},{"id":"e97f03f0837b0749fed7d7d0c3e60c7de011d574","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2025-01-14 19:45:19.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"ac63f7511dff81d48850ef94fb289166d9cddf23","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-14 20:02:25.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/166e17123d234692a5c13feb0f5ae3fb\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/24cc6372ffb840569c42245c063a5982 : SUCCESS in 8m 05s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/bc81118294064fea9de540261c618069 : SUCCESS in 14m 05s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"4578b7272efc150ec418fed0b5ff8368bc82c95c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-14 22:10:07.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0f3944bba2024c049f6155f9bd5bab5a\n\n- grenade https://zuul.opendev.org/t/openstack/build/8fc31bfffeef4365965991914e5ecd9f : SUCCESS in 58m 02s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9656642c31094dc0b8515dea875a1d89 : SUCCESS in 45m 35s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5a993938594e4aff85e450797f5d866d : SUCCESS in 1h 32m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2f7fca296e0f48d38befb73b1594e7c5 : SUCCESS in 46m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83526e3c46ee44d6967e7d14c72bc770 : SUCCESS in 5m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/95964e0d640a446995a5e07f4da81f75 : SUCCESS in 4m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e881327749124c9bb497ffef74abb06e : SUCCESS in 5m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df4f735f215945fab666bb3b2eb8a67e : SUCCESS in 6m 26s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/635fcd7f5c8d48668563cc4c5f91c0d2 : SUCCESS in 8m 51s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/173bb407ff2a4bda919abef6382ad4c8 : SUCCESS in 8m 20s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/263d757a48c14fa0846294d1627c4c59 : SUCCESS in 11m 49s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/4f7e13d2faea4a8d9b644aa70d1b583f : TIMED_OUT in 2h 04m 38s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/891a49df709b43fd97ea00f43a89489e : SUCCESS in 1h 37m 14s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/308a2ac23a0d418dad97245ae71c4515 : SUCCESS in 1h 45m 07s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/6071b1491b984d6c855d1514a0f81dcf : SUCCESS in 2h 10m 24s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/98d2a5a2e164467b98456544644286d7 : SUCCESS in 2h 16m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ceeec0fad9fe49599f6111844a1bf50b : SUCCESS in 1h 05m 59s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/0c0ccb26ea6740789b50c8ae19c28664 : SUCCESS in 1h 48m 56s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/b6c5ba2b21884815936b3963676abf2c : SUCCESS in 1h 43m 08s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/cd6a156fc5464e6788d4cba6841e9867 : SUCCESS in 1h 11m 22s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/33f130b443e04156bf2e0a7cc2bbf43a : SUCCESS in 2h 07m 51s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/548c1d467e5543769c1163d247cbd44b : SUCCESS in 36m 11s","accounts_in_message":[],"_revision_number":8},{"id":"bf213935cb9e231d83a6de98323b3fc14e82d6c8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2025-01-15 17:36:19.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"a60d4fa2b2d19a01a45128fe97f19c1ee0ba3dc7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-15 18:02:24.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d308475bd68e44918b124de319d080ca\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/beb0896b3ab949cb8948b15e430c5b1c : SUCCESS in 16m 13s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/10a872ad2d0d40c0813d83b785fedd52 : SUCCESS in 21m 17s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"c1d80b3dc225d648e349eafccd0c2594830d4a12","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-15 20:11:30.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/719d77de812844b8946409d2ec280bd5\n\n- grenade https://zuul.opendev.org/t/openstack/build/35b272f9062646fa96f07b5c7d654f7f : SUCCESS in 37m 49s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/20649eaf164b447990ce1be505b04a79 : SUCCESS in 34m 35s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ff8d196dbca34dc5b2df2f3cf853a31d : SUCCESS in 1h 48m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/33ad91b839f84ac28d275c796fa7f2f0 : SUCCESS in 1h 07m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7518400b9bf40968d3317643808b597 : SUCCESS in 7m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5fb79263d2da40088f76aeba47a3f403 : SUCCESS in 5m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/aeb55808edc34c479721bff60e80a4ba : SUCCESS in 5m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54ab8c2c5d624e4eae499764adf68b3f : SUCCESS in 6m 29s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5578e9d647c54af1bdc84fc1b0aa8c90 : SUCCESS in 8m 32s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/75612c0f7b1a4a63b49450af507e388a : SUCCESS in 8m 58s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/ef64766b5f9c4dd39647e368438c6567 : SUCCESS in 12m 59s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/20bb0eb74e2742beb2341e117583910f : SUCCESS in 1h 38m 29s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/a69a2c33a1f3446991703bcd8ebd754f : SUCCESS in 1h 55m 13s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/37ee1a82daa248e5b2052a909fa1bc19 : SUCCESS in 2h 27m 54s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/a2e3b29d5bf848b98ade077fa0b9feb0 : SUCCESS in 1h 55m 24s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/fe5eea2ceda8421a89981ba91326ce86 : FAILURE in 2h 07m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b2c8f3cc2d88425bb43ce8bda194930c : SUCCESS in 1h 00m 24s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/633114baece745d9bc7704373f0cbbac : SUCCESS in 1h 21m 06s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/86961ebd42554dae92eae340055b2816 : SUCCESS in 2h 04m 15s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/da5655f3e2b34ad894ec0aac9e1e4959 : SUCCESS in 54m 34s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/ac91afcf9ddc40b686dca035b2dc629c : SUCCESS in 1h 30m 20s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/a553288670134a439b11befe046ea4b6 : SUCCESS in 33m 24s","accounts_in_message":[],"_revision_number":9},{"id":"472e2a42eba75f02674ed54c3bffa0de7a82b4cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2025-01-16 15:23:07.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"323f9a9652c603b0fa0608446bf4f0cd0e77ff03","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-16 18:45:14.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2f45eae547854e4a9c190691db51c63d\n\n- grenade https://zuul.opendev.org/t/openstack/build/74f0ea21aeb447feba94c93d6691f9cc : SUCCESS in 47m 47s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/715d3486cdd741968cb7005f7422caf1 : SUCCESS in 51m 55s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4729024c2c9a4f448305389354ef6274 : SUCCESS in 1h 42m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/01d200e69f55463cac48d1542af895dc : SUCCESS in 1h 13m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59dc475dce654780b8b4802cc054f8f6 : SUCCESS in 4m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2f1910ca515e4c58963c067723e1c51b : SUCCESS in 5m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d4d92684024b421285537cea7e31d66f : SUCCESS in 3m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55e17efd40d745b08ed0af7500a1bda5 : SUCCESS in 6m 40s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a5cc3ca50fcc499ab72f4d5e8ddf63fb : SUCCESS in 8m 33s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6a82cd52cc484ef482351d5c29a19da2 : SUCCESS in 8m 25s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/b69d565fd30545b38ff2b7a279130cd4 : SUCCESS in 13m 10s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/8c4d10e01eea427c885be785c3203c47 : FAILURE in 1h 54m 34s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/b2843adfcd44485485a78de882fc9b3c : SUCCESS in 2h 07m 38s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/9eb327f61d254a26b693811cf19b1dde : SUCCESS in 1h 14m 23s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/589c53a4a0244b20bac86715f13a54f6 : SUCCESS in 1h 42m 11s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/ab9b8d0e42424a76812ce9194a583ae4 : SUCCESS in 2h 17m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8dc056e5f35545b7b24fbb8c8d6fae08 : SUCCESS in 58m 57s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/0924de5f5559448faed1ceb1c4225cfe : SUCCESS in 1h 38m 56s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/65fb9cb2317a4fd3bc9687cb5109f776 : SUCCESS in 1h 12m 13s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/91554b07117d423caf3a317323861997 : SUCCESS in 1h 07m 37s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/296bd1e44d2d41cea89bf0dc309f30b4 : SUCCESS in 1h 58m 31s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/22d17e8f48574f49af3e4da896cf6231 : SUCCESS in 27m 54s","accounts_in_message":[],"_revision_number":10},{"id":"9aaf79da69c6edc329d35b02707997e322b11b81","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-16 19:22:11.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aac18b8380094760b453cfb5bc2aa612\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ea5e794720ba4db58576d0bd2c5c0b88 : SUCCESS in 15m 47s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c82ff85173344b70a2ec519422e84b7e : SUCCESS in 20m 35s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"e5cd9341e1c07971aa3c6bc90ea91a002b425bb1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2025-02-20 21:13:58.000000000","message":"Uploaded patch set 11: Patch Set 10 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"e6ae331cea3d31af079dde08ceb57950aac2dce7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 21:40:13.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3933adb305084f35af73d77ec4e4a71a\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0261a29468484329b39f2916aef21d41 : SUCCESS in 19m 35s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c7c8a418c17e483f8d49c8afe610d667 : SUCCESS in 21m 13s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"a2e82246aeca818426c1bb9f6e0fd4b75d8ec80d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-21 00:01:35.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/259433b8a147458e8c523bc92acda016\n\n- grenade https://zuul.opendev.org/t/openstack/build/ed0f7ce6f9d84b91ae1c5c90747564c5 : SUCCESS in 1h 07m 36s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/fc210d95cb0f4546a4f7d0c6e8ee8c51 : SUCCESS in 1h 00m 51s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/11fc3cb31c7b4f02abc1c5c8651af5e3 : SUCCESS in 1h 37m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ef82d9cd36a94f03bf70d3776524f579 : SUCCESS in 1h 06m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a960fb5490bc452b9f6848329ef94ba1 : FAILURE in 6m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1915dc79365847b89d6ce37d6fa1c3cf : SUCCESS in 6m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/041c16565c3a4a35b84616bcb0aac999 : SUCCESS in 8m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33ffcf442c7b40e59f1e9a696bcbbf22 : SUCCESS in 7m 39s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5901a6ce45b04b3bb0e3d23af9d4684f : SUCCESS in 8m 59s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2dd67ed6d1ea492a8fa829848ded58ca : SUCCESS in 8m 13s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/15b8e40e5d7a478d98dd4131f23c9502 : SUCCESS in 11m 38s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/d80288560f2a40fc82ed9506e8b76810 : SUCCESS in 1h 48m 30s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/af8f4466b5594400af60d1ad79124526 : SUCCESS in 2h 01m 50s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/10db80bab4c848b991011bf886803a73 : SUCCESS in 2h 17m 01s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/9f1f97e965c04ee0876993f4d901bc19 : SUCCESS in 2h 23m 10s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/c4d2cc7802f349699e43b5d096057af2 : SUCCESS in 2h 39m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b5c68425220041e2a9b6d3b095f9e9b3 : SUCCESS in 54m 57s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/75eb905b37844eba8cec06ae6fd97262 : SUCCESS in 1h 45m 43s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/cd9d7190604a46e08a63e4e1024ff407 : SUCCESS in 1h 37m 59s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/a23f3d9ab10b45d3baa83c01d65dba84 : SUCCESS in 1h 05m 49s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/59b05c222c604a0fa8c05f5768685a60 : SUCCESS in 1h 51m 19s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/1bcfb71cf9114e6abe685d71bc392f88 : SUCCESS in 37m 19s\n\nWarning:\n  Comments left for invalid file glance/location.py","accounts_in_message":[],"_revision_number":11},{"id":"0e9cc795303601d20e571fee248d782063c41234","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-02-21 12:19:39.000000000","message":"Patch Set 11: Code-Review+1","accounts_in_message":[],"_revision_number":11},{"id":"b7015f6db129121427b60228b61d20c372daa1a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2025-02-24 14:32:47.000000000","message":"Uploaded patch set 12: Patch Set 11 was rebased.\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":12},{"id":"4a4389ce7587d5b97d13ed49de5bc531af7c098f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-24 14:58:54.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/450102e177ca42b69c287f4ca5a0294d\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/97efdee39d5d4f8ea9544386fb0e9593 : SUCCESS in 13m 36s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ae8ee39398e84b5cba56661490429d54 : SUCCESS in 22m 00s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"6fca550e547b25f196a69b14b924e42d4c15c443","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-02-24 18:22:34.000000000","message":"Patch Set 12: Code-Review+1","accounts_in_message":[],"_revision_number":12},{"id":"69071a78c1263b09ce0acabfc3a0dbd8217d33be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-24 19:24:47.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/49f5808c35714865bfae95fda887efb1\n\n- grenade https://zuul.opendev.org/t/openstack/build/7d5d718b6ed54bd0bd548fd0b2f82bb8 : SUCCESS in 51m 03s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1e4bd03789864ff295e74c655a9b6c90 : SUCCESS in 47m 01s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0479b67f48f64bdcbf3a36183e9ea6b7 : TIMED_OUT in 2h 05m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3bcc3ae7f85b44688d33c7c599de203a : SUCCESS in 1h 06m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/948e7eae8c944f68b33fff4d603f3cab : SUCCESS in 5m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a8b90e68a0ee46639cdf7915f4e844a9 : SUCCESS in 4m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2b2d43098a5a416d9ad69d8342688c70 : SUCCESS in 5m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf72a12b25be4f569ab2d5629c0e7654 : SUCCESS in 5m 25s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e3a238cb57a94afdaf658b67a93d7165 : SUCCESS in 8m 18s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/890e3eaca3ae483d9cb4ac5aab0dfcff : SUCCESS in 9m 15s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/4ed0651e5e14441284a7f8beac6e6cee : SUCCESS in 11m 45s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/d6d149548f6440259228b597740ad8a0 : SUCCESS in 1h 26m 32s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/4f0fbde6a94f43b8a97e7e2b5924ea06 : SUCCESS in 1h 51m 28s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/94b17b45138048e0b7026922c77db902 : SUCCESS in 1h 12m 51s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/e0a4918dd13d4cc1bb42b48cd83dd0e6 : SUCCESS in 1h 44m 19s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/d62605286fb34f708e0faa87da02741f : FAILURE in 2h 30m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1f2e761bb71141d285b537ffdfb2f616 : SUCCESS in 46m 05s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a39e7e40fcfd4448a87408dbcaa5ba1c : SUCCESS in 1h 42m 52s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/437cd56f8e294a65979523b1a3082485 : SUCCESS in 1h 57m 24s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/c68139b47919460595d509a244ae66a0 : SUCCESS in 59m 56s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/9585514be58040a9b0b3cbc2705a94bd : SUCCESS in 1h 16m 00s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/820eaafe29954db292f08fa72256dfa7 : SUCCESS in 35m 30s","accounts_in_message":[],"_revision_number":12},{"id":"ab10bb313df2b30bb1c8fa545c8176c70bebe650","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-02-24 20:54:59.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"bfc4e7772f2c0a9b022e0b9ee08369399e12f2c0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-24 21:32:02.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e7e90fe926cc41e0a566406b5d5886a3\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/95a510af1fa8409a8a1fe17a60cdcaff : SUCCESS in 24m 11s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f62168e8f4dc43debe739777a2e23933 : SUCCESS in 27m 24s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"3b20c11113369b7f4b43d94d7f3657bb9b2bc4a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-24 23:57:41.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fb90562022014da99e849224ab0cafbd\n\n- grenade https://zuul.opendev.org/t/openstack/build/c589bf58a25244d7afaaab5ab740a58b : SUCCESS in 51m 44s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b140bc22c4a64eb3abbd363959ede059 : SUCCESS in 57m 30s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/26d32b2c43b149c6b0680ff7e3390203 : SUCCESS in 1h 36m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bdb5523d6926451c8671fdc7ac38d566 : SUCCESS in 39m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75f03831861d4d829b1971add92076b2 : SUCCESS in 5m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0849e3664e604f0ba249405ff6581626 : SUCCESS in 6m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9469924f3f1e447fbab914d124e6a945 : SUCCESS in 5m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54de7ff0ea9948be8aaa64786d878b60 : SUCCESS in 6m 17s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/12fefce8275c4c2eb9f9e1ffa186d2c0 : SUCCESS in 9m 03s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/210a6ec129664e56884c760fe01aa84b : SUCCESS in 9m 26s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/a7ece563052f4bdda483a10fb1e22ef0 : SUCCESS in 11m 52s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/30045a81369c4c76a3e20ea381176dca : SUCCESS in 1h 55m 12s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/3b4c973c9cdf4f83819aee63bde5e0a4 : SUCCESS in 2h 10m 55s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/f3736e78947d4217a9eaa23d3069c436 : SUCCESS in 1h 42m 27s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/8a017df718ec4fa683b2b268b6d6f029 : SUCCESS in 2h 03m 54s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/a4016759a6904c989b15ffe0dbdecce0 : SUCCESS in 2h 54m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7beddcb927ab4146a2e37824584d56ce : SUCCESS in 59m 52s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/3edfa298c45044f3a63116330bf36b09 : SUCCESS in 2h 02m 09s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/c7275ca5400e4440b0bcc8be2da24d95 : SUCCESS in 1h 56m 00s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/2035a330537d4cffb18a9841c8310675 : SUCCESS in 1h 24m 55s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/23b4928ce26e41d698d44bf3257bdd7f : SUCCESS in 1h 45m 32s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/01d1b0c56ca64245b3dd4ae288ade617 : SUCCESS in 23m 39s","accounts_in_message":[],"_revision_number":12},{"id":"ee3c4e6c524776c30478a37a56921b780c6991b6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2025-02-25 16:19:21.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":13},{"id":"6aeb047a441c4e8d533a1b2650b2e315489c1b57","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-02-25 16:20:27.000000000","message":"Patch Set 13: Code-Review+1","accounts_in_message":[],"_revision_number":13},{"id":"12a16d7550f6c1abee3579d0ea9323c62cc36767","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-25 18:10:05.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff8b9d48b5ee41a1b5eb6e66d2da9ca6\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/dd5f74f2d6fe4bd68ffb9179c4ba1986 : FAILURE in 18m 30s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0a9a38547e7948efb3c7bf94b2cad6ab : FAILURE in 29m 59s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"99799ee899f790123000a8edd9ccdad027e07e01","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2025-02-25 18:31:59.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"9794abb09e4a2766be24610c2a591deda0135183","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2025-02-25 18:33:46.000000000","message":"Patch Set 13: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"715105f370afb454b7ac14b7c85f806f9ece3208","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-25 18:55:59.000000000","message":"Patch Set 13: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ede3852c3a8e497c88fd49bc5dd88035\n\n- grenade https://zuul.opendev.org/t/openstack/build/4b9a65a06dba4418a1b7cea650ed7f59 : SUCCESS in 1h 01m 48s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/fd3667eccdb2435fb9c31166c48f10fa : SUCCESS in 59m 22s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/28f9ac7aa450478298a816d297daea4a : SUCCESS in 1h 24m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cdd58bf51dc24812bd34a413327e0b94 : SUCCESS in 59m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bbfe82ee5df849fd9749ff3cfe4650a7 : FAILURE in 6m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cea248b98383401284e694686ec7fc11 : FAILURE in 4m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8baed30ec4f44a54aae0191f1355f7ef : FAILURE in 4m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ef7c76f6c654ce88858cc8ad5865bf8 : SUCCESS in 6m 07s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2949ad0f8b6f42158691b67bed549ba9 : SUCCESS in 10m 34s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/053ba37517ef4a4f81126a95ad6bd871 : SUCCESS in 9m 24s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/ca14a1cb954b43e69d79501a90a618e9 : SUCCESS in 11m 31s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/19183d189f094c8692863ead90fb404f : SUCCESS in 2h 05m 34s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/48c8fff6c1d14636902e229aab665c9e : SUCCESS in 1h 57m 28s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/1b202e8c3c7844838a1eab978bee43db : SUCCESS in 2h 09m 25s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/b4deddc9c7234f24a8190be00e32fc4a : SUCCESS in 1h 57m 07s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/86eabdb14b35455580b4bdacf2724e7b : SUCCESS in 2h 17m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6c47aa2227e14d2cb075c5c77bff6931 : SUCCESS in 1h 03m 49s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a819258274984dc2908bd815d5045677 : SUCCESS in 2h 19m 53s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/65f468ccf1f542d2b3c7f827dd315239 : SUCCESS in 2h 16m 03s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/49aae628225f4d8ea7e1cf92e26430ea : SUCCESS in 1h 16m 17s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/35aef34ac50e4dbe98e23545b5b9546b : SUCCESS in 2h 27m 33s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/b81c2afc25894fdebeba8dfeef93e2fb : SUCCESS in 51m 13s","accounts_in_message":[],"_revision_number":13},{"id":"3782396208441246e26eaae48c496ff35ff66f63","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2025-02-25 19:12:46.000000000","message":"Uploaded patch set 14.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":14},{"id":"140ef378e7d85e1369a661f76dc10337cae8a952","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-25 19:47:39.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/819fe376effd4b66b9cbdc091f0f6f31\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/57c02e1de1424890bc598f9455903a96 : SUCCESS in 19m 01s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/beefe86459e24702a7d10a8248660e84 : SUCCESS in 21m 08s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"19720c73bcf196d8b1ea1a3c38a3c31e026d8f98","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-02-25 20:37:17.000000000","message":"Patch Set 14: Code-Review+1","accounts_in_message":[],"_revision_number":14},{"id":"776bb14777107249138b76217cf85dd394afc64b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-25 21:40:51.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5f7ecf96746c4d7f8530b0d5696c6584\n\n- grenade https://zuul.opendev.org/t/openstack/build/991fdaf8a4444b029bc53ae35ad0ad40 : SUCCESS in 44m 52s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9bb841d62a504290893aa618fdade36b : SUCCESS in 55m 09s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f2ae25c545be4094ac999be4726aa63d : SUCCESS in 1h 50m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c8b880040a56482f9a9cf2f8c2cecbe3 : SUCCESS in 1h 00m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9a2c595371b4ff4899cf322b25283d9 : SUCCESS in 3m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7275c992034c4c36bf79de3ab5036c80 : SUCCESS in 5m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/441feaf1c09f466b9564e01c0045e5f5 : SUCCESS in 5m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef4fa8eb26e84f408329526e4f1b6063 : SUCCESS in 6m 41s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c62c24f07b924e1ea2423d3ab2a3d33f : SUCCESS in 8m 51s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ca8b2dbe642d42d5b7fe01c65c35d3b0 : SUCCESS in 8m 11s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/dc0f68e9f6b94a7286c32d67e66d871e : SUCCESS in 12m 36s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/689cb93ec5a7420499b05137805a8a7f : SUCCESS in 1h 46m 57s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/8c141c3427ff4e0f8acc0bfbb4f12ce1 : SUCCESS in 1h 44m 05s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/1d2b77ad1d30459e8f87331b9f3de223 : SUCCESS in 2h 05m 37s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/547799c31b2348048f895a43ae76da94 : SUCCESS in 2h 16m 49s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/e10b7e2618a84a4f86d9d73c93423a0f : SUCCESS in 1h 53m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dbe03b6a0ef34f61bbc354216fc150b6 : SUCCESS in 1h 13m 08s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9d7bbb11181044eab476aac4cf8be99a : SUCCESS in 2h 08m 37s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/7bcad39c724d4dd0bf98402e6a49d253 : SUCCESS in 2h 01m 09s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/1022f0b06f5d416a87e2b66d60f81491 : SUCCESS in 1h 01m 52s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/9a6649d5fa944aaf91083777d6dfd002 : SUCCESS in 1h 51m 32s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/35c2df4fa7ae4a1785df9561ab7fe8ba : SUCCESS in 39m 04s","accounts_in_message":[],"_revision_number":14},{"id":"750c7a40eae4b988c2b4442dbf5c9a254ee37aad","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-02-26 04:38:39.000000000","message":"Patch Set 14: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"b6a127ab04d4d01dae85fbb5e3c960db9884a1e1","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-03-26 18:46:42.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"c5f498fceceb8993bb77ab4ea2579e2bce7e949a","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2025-06-16 17:58:48.000000000","message":"Patch Set 14: Code-Review+1","accounts_in_message":[],"_revision_number":14},{"id":"3eab23feb3d5d782d7abd64d10183d101eed3e60","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-24 18:01:24.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"5504e78157cbfff4b80b9e5043fc1a5a6b125dc1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2026-10-09 11:31:00.000000000","message":"Uploaded patch set 15.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":15},{"id":"fdb88db2f8b2df63172e8b315f284a91fd997e05","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2026-10-09 11:32:03.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"26bf97e7b99e9da95dfc726bb785403e76246664","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-09 13:08:03.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4b6f313322944f24b5cb2707883243d3\n\n- grenade https://zuul.opendev.org/t/openstack/build/7fcbbb239f3f499aaf9b816cb76b16a6 : SUCCESS in 34m 24s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/dcbe55de83374c29954baa3376dfebbe : SUCCESS in 37m 29s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e55eb45f2b33494ca07b803d5044494d : FAILURE in 1h 07m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e7a6af84ba254e2ba849a606e534fb91 : SUCCESS in 33m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42563d9bcf7944a5a8d54fcf3d86e44c : SUCCESS in 3m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d28cf200ab394c5781859502297f5835 : SUCCESS in 3m 51s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/7e53ba7bc286410b8416fd56ab0f7e8a : SUCCESS in 5m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30105b2b90674cc0a1a5764648ce07a8 : SUCCESS in 3m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/89ef7a7ff0734b42ab00d51aeafa36b1 : SUCCESS in 5m 20s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/88522fb23c87478c96c603670a5d2eb2 : SUCCESS in 6m 27s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/7cfd3abf2fe64672a7d38b4b807d42af : SUCCESS in 8m 08s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/b0e4d59a9d6b45d7baf500cf2579e5e2 : SUCCESS in 54m 43s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/69ed2bbb1d2e46d6b584b08c10ad0620 : FAILURE in 1h 10m 33s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/b4d913b9d4324a589f6f4790bdd2cffa : FAILURE in 1h 31m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/aa1193cf353f4f528549b3f086b9084c : SUCCESS in 31m 05s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/904e54299e3e4d7fad3f054eac62c740 : SUCCESS in 1h 17m 17s\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/5e32018e96294af5ab07bee5ef4ec5c4 : FAILURE in 20m 28s (non-voting)\n- glance-multistore-add-location-do-secure-hash-false https://zuul.opendev.org/t/openstack/build/f8feba30538448a99dde506a1de997ff : FAILURE in 15m 02s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/5b2268df1dc24e5d95400c13019f1916 : SUCCESS in 15m 55s\n- glance-rally-task-uwsgi https://zuul.opendev.org/t/openstack/build/17272ee81de347c4bff0c66633fcd584 : SUCCESS in 35m 13s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"94aaa9531c1689870af3fe3d12f99fc19987c775","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-09 14:04:19.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/32f7bb8e7d504d05bec49a80dde9b7af\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e2f608134887412c9c908c23daccd4fb : SUCCESS in 9m 02s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/d703e841bee248309687b4bed3bbefaf : SUCCESS in 14m 46s (non-voting)","accounts_in_message":[],"_revision_number":15}],"current_revision_number":15,"current_revision":"0567588c39a3f64da96d57d30f5738365cd22fab","revisions":{"74280f193726b22b3c17d484e5c771b4167b652f":{"kind":"REWORK","_number":1,"created":"2024-10-28 17:24:02.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/01/933601/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/01/933601/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/01/933601/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/01/933601/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/01/933601/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/01/933601/1"}}},"commit":{"parents":[{"commit":"60ba0a966ed475586267b1f563d016cfc7214d8b","subject":"Run safety checks on uploaded content","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/60ba0a966ed475586267b1f563d016cfc7214d8b"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2024-10-28 15:26:24.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2024-10-28 17:20:30.000000000","tz":-420},"subject":"WIP: Add GPT disk_format","message":"WIP: Add GPT disk_format\n\nThis needs tests.\n\nThis adds gpt as a valid disk_format for glance. For image conversion,\nit makes us:\n\n- Allow gpt as a target format, but use raw if the thing being\n  converted is not actually a GPT in final form. This is what users\n  will want for compatibility\n- Inspect the image after conversion to make the above determination.\n  Note that this actually increases safety because it prevents hiding\n  one format inside another (this could be broken out into its own\n  patch if we want).\n- Makes us treat gpt as raw when talking to qemu-img, since it does\n  not know the difference\n\nRelated to blueprint glance-as-defender\n\nChange-Id: I32416d38e295d32ae24652ec83716ca82eb76b06\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/74280f193726b22b3c17d484e5c771b4167b652f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/74280f193726b22b3c17d484e5c771b4167b652f"}]},"branch":"refs/heads/master"},"e37ef1aa9c94fe77e6480ade2dab62dfc2e4d6ee":{"kind":"REWORK","_number":2,"created":"2024-10-30 15:40:19.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/01/933601/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/01/933601/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/01/933601/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/01/933601/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/01/933601/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/01/933601/2"}}},"commit":{"parents":[{"commit":"a568b2772a17bb90cda00293490f7ac7fe6f2785","subject":"Run safety checks on uploaded content","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a568b2772a17bb90cda00293490f7ac7fe6f2785"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2024-10-28 15:26:24.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2024-10-30 15:39:25.000000000","tz":-420},"subject":"Add GPT disk_format","message":"Add GPT disk_format\n\nThis adds gpt as a valid disk_format for glance. For image conversion,\nit makes us:\n\n- Allow gpt as a target format, but use raw if the thing being\n  converted is not actually a GPT in final form. This is what users\n  will want for compatibility\n- Inspect the image after conversion to make the above determination.\n  Note that this actually increases safety because it prevents hiding\n  one format inside another (this could be broken out into its own\n  patch if we want).\n- Makes us treat gpt as raw when talking to qemu-img, since it does\n  not know the difference\n\nRelated to blueprint glance-as-defender\n\nChange-Id: I32416d38e295d32ae24652ec83716ca82eb76b06\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e37ef1aa9c94fe77e6480ade2dab62dfc2e4d6ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e37ef1aa9c94fe77e6480ade2dab62dfc2e4d6ee"}]},"branch":"refs/heads/master"},"06d0624c824d6ee853da1b6199a9c093e9ca1c70":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2024-10-31 14:27:18.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/01/933601/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/01/933601/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/01/933601/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/01/933601/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/01/933601/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/01/933601/3"}}},"commit":{"parents":[{"commit":"a568b2772a17bb90cda00293490f7ac7fe6f2785","subject":"Run safety checks on uploaded content","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a568b2772a17bb90cda00293490f7ac7fe6f2785"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2024-10-28 15:26:24.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2024-10-31 14:26:56.000000000","tz":-420},"subject":"Add GPT disk_format","message":"Add GPT disk_format\n\nThis adds gpt as a valid disk_format for glance. For image conversion,\nit makes us:\n\n- Allow gpt as a target format, but use raw if the thing being\n  converted is not actually a GPT in final form. This is what users\n  will want for compatibility\n- Inspect the image after conversion to make the above determination.\n  Note that this actually increases safety because it prevents hiding\n  one format inside another (this could be broken out into its own\n  patch if we want).\n- Makes us treat gpt as raw when talking to qemu-img, since it does\n  not know the difference\n\nRelated to blueprint glance-as-defender\n\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/933870\nChange-Id: I32416d38e295d32ae24652ec83716ca82eb76b06\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/06d0624c824d6ee853da1b6199a9c093e9ca1c70"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/06d0624c824d6ee853da1b6199a9c093e9ca1c70"}]},"branch":"refs/heads/master"},"648013ff2e355e59b7286775f9cfba18e6cf51d1":{"kind":"REWORK","_number":4,"created":"2024-10-31 14:34:06.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/01/933601/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/01/933601/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/01/933601/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/01/933601/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/01/933601/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/01/933601/4"}}},"commit":{"parents":[{"commit":"a568b2772a17bb90cda00293490f7ac7fe6f2785","subject":"Run safety checks on uploaded content","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a568b2772a17bb90cda00293490f7ac7fe6f2785"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2024-10-28 15:26:24.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2024-10-31 14:33:01.000000000","tz":-420},"subject":"Add GPT disk_format","message":"Add GPT disk_format\n\nThis adds gpt as a valid disk_format for glance. For image conversion,\nit makes us:\n\n- Allow gpt as a target format, but use raw if the thing being\n  converted is not actually a GPT in final form. This is what users\n  will want for compatibility\n- Inspect the image after conversion to make the above determination.\n  Note that this actually increases safety because it prevents hiding\n  one format inside another (this could be broken out into its own\n  patch if we want).\n- Makes us treat gpt as raw when talking to qemu-img, since it does\n  not know the difference\n\nThis also changes the import job to target \u0027gpt\u0027 as the output format\nfor image conversion of the devstack cirros image, which should end\nup with us using a gpt image for the tempest run.\n\nRelated to blueprint glance-as-defender\n\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/933870\nChange-Id: I32416d38e295d32ae24652ec83716ca82eb76b06\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/648013ff2e355e59b7286775f9cfba18e6cf51d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/648013ff2e355e59b7286775f9cfba18e6cf51d1"}]},"branch":"refs/heads/master"},"6d0803ada2478ac371b3c8ab1e7af27c06358caf":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2024-11-01 16:53:39.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/01/933601/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/01/933601/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/01/933601/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/01/933601/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/01/933601/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/01/933601/5"}}},"commit":{"parents":[{"commit":"a568b2772a17bb90cda00293490f7ac7fe6f2785","subject":"Run safety checks on uploaded content","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a568b2772a17bb90cda00293490f7ac7fe6f2785"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2024-10-28 15:26:24.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2024-11-01 16:53:21.000000000","tz":-420},"subject":"Add GPT disk_format","message":"Add GPT disk_format\n\nThis adds gpt as a valid disk_format for glance. For image conversion,\nit makes us:\n\n- Allow gpt as a target format, but use raw if the thing being\n  converted is not actually a GPT in final form. This is what users\n  will want for compatibility\n- Inspect the image after conversion to make the above determination.\n  Note that this actually increases safety because it prevents hiding\n  one format inside another (this could be broken out into its own\n  patch if we want).\n- Makes us treat gpt as raw when talking to qemu-img, since it does\n  not know the difference\n\nThis also changes the import job to target \u0027gpt\u0027 as the output format\nfor image conversion of the devstack cirros image, which should end\nup with us using a gpt image for the tempest run.\n\nRelated to blueprint glance-as-defender\n\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/933870\nDepends-On: https://review.opendev.org/c/openstack/nova/+/933928\nChange-Id: I32416d38e295d32ae24652ec83716ca82eb76b06\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6d0803ada2478ac371b3c8ab1e7af27c06358caf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6d0803ada2478ac371b3c8ab1e7af27c06358caf"}]},"branch":"refs/heads/master"},"711d6bcb53f788edb7b1331c8de914a633294fba":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2024-11-06 17:43:22.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/01/933601/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/01/933601/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/01/933601/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/01/933601/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/01/933601/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/01/933601/6"}}},"commit":{"parents":[{"commit":"a568b2772a17bb90cda00293490f7ac7fe6f2785","subject":"Run safety checks on uploaded content","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a568b2772a17bb90cda00293490f7ac7fe6f2785"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2024-10-28 15:26:24.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2024-11-06 17:42:57.000000000","tz":-480},"subject":"Add GPT disk_format","message":"Add GPT disk_format\n\nThis adds gpt as a valid disk_format for glance. For image conversion,\nit makes us:\n\n- Allow gpt as a target format, but use raw if the thing being\n  converted is not actually a GPT in final form. This is what users\n  will want for compatibility\n- Inspect the image after conversion to make the above determination.\n  Note that this actually increases safety because it prevents hiding\n  one format inside another (this could be broken out into its own\n  patch if we want).\n- Makes us treat gpt as raw when talking to qemu-img, since it does\n  not know the difference\n\nThis also changes the import job to target \u0027gpt\u0027 as the output format\nfor image conversion of the devstack cirros image, which should end\nup with us using a gpt image for the tempest run.\n\nRelated to blueprint glance-as-defender\n\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/933870\nDepends-On: https://review.opendev.org/c/openstack/nova/+/933928\nDepends-On: https://review.opendev.org/c/openstack/cinder/+/934261\nChange-Id: I32416d38e295d32ae24652ec83716ca82eb76b06\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/711d6bcb53f788edb7b1331c8de914a633294fba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/711d6bcb53f788edb7b1331c8de914a633294fba"}]},"branch":"refs/heads/master"},"d51e24a60adb14b22db1a649c3c9ca6ff00abdc6":{"kind":"REWORK","_number":7,"created":"2024-11-13 17:56:40.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/01/933601/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/01/933601/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/01/933601/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/01/933601/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/01/933601/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/01/933601/7"}}},"commit":{"parents":[{"commit":"a568b2772a17bb90cda00293490f7ac7fe6f2785","subject":"Run safety checks on uploaded content","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a568b2772a17bb90cda00293490f7ac7fe6f2785"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2024-10-28 15:26:24.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2024-11-13 17:56:31.000000000","tz":-480},"subject":"Add GPT disk_format","message":"Add GPT disk_format\n\nThis adds gpt as a valid disk_format for glance. For image conversion,\nit makes us:\n\n- Allow gpt as a target format, but use raw if the thing being\n  converted is not actually a GPT in final form. This is what users\n  will want for compatibility\n- Inspect the image after conversion to make the above determination.\n  Note that this actually increases safety because it prevents hiding\n  one format inside another (this could be broken out into its own\n  patch if we want).\n- Makes us treat gpt as raw when talking to qemu-img, since it does\n  not know the difference\n\nThis also changes the import job to target \u0027gpt\u0027 as the output format\nfor image conversion of the devstack cirros image, which should end\nup with us using a gpt image for the tempest run.\n\nRelated to blueprint glance-as-defender\n\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/933870\nDepends-On: https://review.opendev.org/c/openstack/nova/+/933928\nDepends-On: https://review.opendev.org/c/openstack/cinder/+/934261\nChange-Id: I32416d38e295d32ae24652ec83716ca82eb76b06\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d51e24a60adb14b22db1a649c3c9ca6ff00abdc6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d51e24a60adb14b22db1a649c3c9ca6ff00abdc6"}]},"branch":"refs/heads/master"},"a551c54a93e678d6e07e4af300485aa49f1b34db":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2025-01-14 14:33:36.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/01/933601/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/01/933601/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/01/933601/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/01/933601/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/01/933601/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/01/933601/8"}}},"commit":{"parents":[{"commit":"e695b932235cc0da4b499981154a64aff4bc3eb1","subject":"DNM: Test glance against latest oslo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e695b932235cc0da4b499981154a64aff4bc3eb1"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2024-10-28 15:26:24.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2025-01-14 14:33:28.000000000","tz":-480},"subject":"Add GPT disk_format","message":"Add GPT disk_format\n\nThis adds gpt as a valid disk_format for glance. For image conversion,\nit makes us:\n\n- Allow gpt as a target format, but use raw if the thing being\n  converted is not actually a GPT in final form. This is what users\n  will want for compatibility\n- Inspect the image after conversion to make the above determination.\n  Note that this actually increases safety because it prevents hiding\n  one format inside another (this could be broken out into its own\n  patch if we want).\n- Makes us treat gpt as raw when talking to qemu-img, since it does\n  not know the difference\n\nThis also changes the import job to target \u0027gpt\u0027 as the output format\nfor image conversion of the devstack cirros image, which should end\nup with us using a gpt image for the tempest run.\n\nRelated to blueprint glance-as-defender\n\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/933870\nDepends-On: https://review.opendev.org/c/openstack/nova/+/933928\nDepends-On: https://review.opendev.org/c/openstack/cinder/+/934261\nChange-Id: I32416d38e295d32ae24652ec83716ca82eb76b06\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a551c54a93e678d6e07e4af300485aa49f1b34db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a551c54a93e678d6e07e4af300485aa49f1b34db"}]},"branch":"refs/heads/master"},"d1d5343910b79311160f607f705edc5bf3bc354f":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2025-01-15 17:36:19.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/01/933601/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/01/933601/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/01/933601/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/01/933601/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/01/933601/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/01/933601/9"}}},"commit":{"parents":[{"commit":"d6208e330546b2fe43b1abd9188b729da0d87557","subject":"DNM: Test glance against latest oslo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d6208e330546b2fe43b1abd9188b729da0d87557"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2024-10-28 15:26:24.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2025-01-15 14:41:07.000000000","tz":-480},"subject":"Add GPT disk_format","message":"Add GPT disk_format\n\nThis adds gpt as a valid disk_format for glance. For image conversion,\nit makes us:\n\n- Allow gpt as a target format, but use raw if the thing being\n  converted is not actually a GPT in final form. This is what users\n  will want for compatibility\n- Inspect the image after conversion to make the above determination.\n  Note that this actually increases safety because it prevents hiding\n  one format inside another (this could be broken out into its own\n  patch if we want).\n- Makes us treat gpt as raw when talking to qemu-img, since it does\n  not know the difference\n\nThis also changes the import job to target \u0027gpt\u0027 as the output format\nfor image conversion of the devstack cirros image, which should end\nup with us using a gpt image for the tempest run.\n\nRelated to blueprint glance-as-defender\n\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/933870\nDepends-On: https://review.opendev.org/c/openstack/nova/+/933928\nDepends-On: https://review.opendev.org/c/openstack/cinder/+/934261\nChange-Id: I32416d38e295d32ae24652ec83716ca82eb76b06\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d1d5343910b79311160f607f705edc5bf3bc354f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d1d5343910b79311160f607f705edc5bf3bc354f"}]},"branch":"refs/heads/master"},"aaf6299b7639ae7fdb18a231d50cd27c3871dc60":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2025-01-16 15:23:07.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/01/933601/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/01/933601/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/01/933601/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/01/933601/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/01/933601/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/01/933601/10"}}},"commit":{"parents":[{"commit":"4a6455590d17b7de25d9a44275ffde5a507bdd46","subject":"Run safety checks on uploaded content","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4a6455590d17b7de25d9a44275ffde5a507bdd46"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2024-10-28 15:26:24.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2025-01-16 15:22:56.000000000","tz":-480},"subject":"Add GPT disk_format","message":"Add GPT disk_format\n\nThis adds gpt as a valid disk_format for glance. For image conversion,\nit makes us:\n\n- Allow gpt as a target format, but use raw if the thing being\n  converted is not actually a GPT in final form. This is what users\n  will want for compatibility\n- Inspect the image after conversion to make the above determination.\n  Note that this actually increases safety because it prevents hiding\n  one format inside another (this could be broken out into its own\n  patch if we want).\n- Makes us treat gpt as raw when talking to qemu-img, since it does\n  not know the difference\n\nThis also changes the import job to target \u0027gpt\u0027 as the output format\nfor image conversion of the devstack cirros image, which should end\nup with us using a gpt image for the tempest run.\n\nRelated to blueprint glance-as-defender\n\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/933870\nDepends-On: https://review.opendev.org/c/openstack/nova/+/933928\nDepends-On: https://review.opendev.org/c/openstack/cinder/+/934261\nChange-Id: I32416d38e295d32ae24652ec83716ca82eb76b06\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/aaf6299b7639ae7fdb18a231d50cd27c3871dc60"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/aaf6299b7639ae7fdb18a231d50cd27c3871dc60"}]},"branch":"refs/heads/master"},"9eee7832827e9112b5bda748a1bcd2a172843dd7":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2025-02-20 21:13:58.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/01/933601/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/01/933601/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/01/933601/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/01/933601/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/01/933601/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/01/933601/11"}}},"commit":{"parents":[{"commit":"d09ac6da8ca5c6b5cfae203a4389b5c0864f4428","subject":"Run safety checks on uploaded content","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d09ac6da8ca5c6b5cfae203a4389b5c0864f4428"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2024-10-28 15:26:24.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2025-02-20 21:13:48.000000000","tz":-480},"subject":"Add GPT disk_format","message":"Add GPT disk_format\n\nThis adds gpt as a valid disk_format for glance. For image conversion,\nit makes us:\n\n- Allow gpt as a target format, but use raw if the thing being\n  converted is not actually a GPT in final form. This is what users\n  will want for compatibility\n- Inspect the image after conversion to make the above determination.\n  Note that this actually increases safety because it prevents hiding\n  one format inside another (this could be broken out into its own\n  patch if we want).\n- Makes us treat gpt as raw when talking to qemu-img, since it does\n  not know the difference\n\nThis also changes the import job to target \u0027gpt\u0027 as the output format\nfor image conversion of the devstack cirros image, which should end\nup with us using a gpt image for the tempest run.\n\nRelated to blueprint glance-as-defender\n\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/933870\nDepends-On: https://review.opendev.org/c/openstack/nova/+/933928\nDepends-On: https://review.opendev.org/c/openstack/cinder/+/934261\nChange-Id: I32416d38e295d32ae24652ec83716ca82eb76b06\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9eee7832827e9112b5bda748a1bcd2a172843dd7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9eee7832827e9112b5bda748a1bcd2a172843dd7"}]},"branch":"refs/heads/master"},"4b78a0c9ef32b1b106e44574c0d41a02307cba07":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2025-02-24 14:32:47.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/01/933601/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/01/933601/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/01/933601/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/01/933601/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/01/933601/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/01/933601/12"}}},"commit":{"parents":[{"commit":"75a8851c10fb1c59381eac2eedd7a87c0e1c0bd7","subject":"Run safety checks on uploaded content","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/75a8851c10fb1c59381eac2eedd7a87c0e1c0bd7"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2024-10-28 15:26:24.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2025-02-24 14:32:46.000000000","tz":-480},"subject":"Add GPT disk_format","message":"Add GPT disk_format\n\nThis adds gpt as a valid disk_format for glance. For image conversion,\nit makes us:\n\n- Allow gpt as a target format, but use raw if the thing being\n  converted is not actually a GPT in final form. This is what users\n  will want for compatibility\n- Inspect the image after conversion to make the above determination.\n  Note that this actually increases safety because it prevents hiding\n  one format inside another (this could be broken out into its own\n  patch if we want).\n- Makes us treat gpt as raw when talking to qemu-img, since it does\n  not know the difference\n\nThis also changes the import job to target \u0027gpt\u0027 as the output format\nfor image conversion of the devstack cirros image, which should end\nup with us using a gpt image for the tempest run.\n\nRelated to blueprint glance-as-defender\n\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/933870\nDepends-On: https://review.opendev.org/c/openstack/nova/+/933928\nDepends-On: https://review.opendev.org/c/openstack/cinder/+/934261\nChange-Id: I32416d38e295d32ae24652ec83716ca82eb76b06\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4b78a0c9ef32b1b106e44574c0d41a02307cba07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4b78a0c9ef32b1b106e44574c0d41a02307cba07"}]},"branch":"refs/heads/master"},"7b1ca0c43c763c93295fd7d0b28d14205ba00377":{"kind":"REWORK","_number":13,"created":"2025-02-25 16:19:21.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/01/933601/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/01/933601/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/01/933601/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/01/933601/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/01/933601/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/01/933601/13"}}},"commit":{"parents":[{"commit":"3e98709c3f40f20dd1688909811fd5927c44e5f8","subject":"Safety check output of image conversion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3e98709c3f40f20dd1688909811fd5927c44e5f8"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2024-10-28 15:26:24.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2025-02-25 16:18:29.000000000","tz":-480},"subject":"Add GPT disk_format","message":"Add GPT disk_format\n\nThis adds gpt as a valid disk_format for glance. For image conversion,\nit makes us:\n\n- Allow gpt as a target format, but use raw if the thing being\n  converted is not actually a GPT in final form. This is what users\n  will want for compatibility\n- Inspect the image after conversion to make the above determination.\n- Makes us treat gpt as raw when talking to qemu-img, since it does\n  not know the difference\n\nThis also changes the import job to target \u0027gpt\u0027 as the output format\nfor image conversion of the devstack cirros image, which should end\nup with us using a gpt image for the tempest run.\n\nRelated to blueprint glance-as-defender\n\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/933870\nDepends-On: https://review.opendev.org/c/openstack/nova/+/933928\nDepends-On: https://review.opendev.org/c/openstack/cinder/+/934261\nChange-Id: I32416d38e295d32ae24652ec83716ca82eb76b06\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7b1ca0c43c763c93295fd7d0b28d14205ba00377"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7b1ca0c43c763c93295fd7d0b28d14205ba00377"}]},"branch":"refs/heads/master"},"def11f3b3adf832bf96711c84f520d1231e4b0d1":{"kind":"REWORK","_number":14,"created":"2025-02-25 19:12:46.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/01/933601/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/01/933601/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/01/933601/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/01/933601/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/01/933601/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/01/933601/14"}}},"commit":{"parents":[{"commit":"f54f6b3b12a6eb6c50a58de5f40d9b96f1da5184","subject":"Safety check output of image conversion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f54f6b3b12a6eb6c50a58de5f40d9b96f1da5184"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2024-10-28 15:26:24.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2025-02-25 18:36:37.000000000","tz":-480},"subject":"Add GPT disk_format","message":"Add GPT disk_format\n\nThis adds gpt as a valid disk_format for glance. For image conversion,\nit makes us:\n\n- Allow gpt as a target format, but use raw if the thing being\n  converted is not actually a GPT in final form. This is what users\n  will want for compatibility\n- Inspect the image after conversion to make the above determination.\n- Makes us treat gpt as raw when talking to qemu-img, since it does\n  not know the difference\n\nThis also changes the import job to target \u0027gpt\u0027 as the output format\nfor image conversion of the devstack cirros image, which should end\nup with us using a gpt image for the tempest run.\n\nRelated to blueprint glance-as-defender\n\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/933870\nDepends-On: https://review.opendev.org/c/openstack/nova/+/933928\nDepends-On: https://review.opendev.org/c/openstack/cinder/+/934261\nChange-Id: I32416d38e295d32ae24652ec83716ca82eb76b06\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/def11f3b3adf832bf96711c84f520d1231e4b0d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/def11f3b3adf832bf96711c84f520d1231e4b0d1"}]},"branch":"refs/heads/master"},"0567588c39a3f64da96d57d30f5738365cd22fab":{"kind":"REWORK","_number":15,"created":"2026-10-09 11:31:00.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/01/933601/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/01/933601/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/01/933601/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/01/933601/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/01/933601/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/01/933601/15"}}},"commit":{"parents":[{"commit":"2b987bfb3caafdf09fe30115fa65df5f563962fe","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2b987bfb3caafdf09fe30115fa65df5f563962fe"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2024-10-28 15:26:24.000000000","tz":-420},"committer":{"name":"PranaliD","email":"pdeore@redhat.com","date":"2026-10-09 11:29:38.000000000","tz":0},"subject":"Add GPT disk_format","message":"Add GPT disk_format\n\nThis adds gpt as a valid disk_format for glance. For image conversion,\nit makes us:\n\n- Allow gpt as a target format, but use raw if the thing being\n  converted is not actually a GPT in final form. This is what users\n  will want for compatibility\n- Inspect the image after conversion to make the above determination.\n- Makes us treat gpt as raw when talking to qemu-img, since it does\n  not know the difference\n\nThis also changes the import job to target \u0027gpt\u0027 as the output format\nfor image conversion of the devstack cirros image, which should end\nup with us using a gpt image for the tempest run.\n\nRelated to blueprint glance-as-defender\n\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/933870\nDepends-On: https://review.opendev.org/c/openstack/nova/+/933928\nDepends-On: https://review.opendev.org/c/openstack/cinder/+/934261\nChange-Id: I32416d38e295d32ae24652ec83716ca82eb76b06\nSigned-off-by: PranaliD \u003cpdeore@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0567588c39a3f64da96d57d30f5738365cd22fab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0567588c39a3f64da96d57d30f5738365cd22fab"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
