)]}'
{"id":"openstack%2Fglance~933667","triplet_id":"openstack%2Fglance~master~I013e5a90a4dd5b6ba436b9f57ffe88fdc29eb326","project":"openstack/glance","branch":"master","attention_set":{},"removed_from_attention_set":{"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2024-10-29 16:37:25.000000000","reason":"\u003cGERRIT_ACCOUNT_9303\u003e replied on the change","reason_account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2024-12-27 08:33:02.000000000","reason":"\u003cGERRIT_ACCOUNT_27615\u003e replied on the change","reason_account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}}},"hashtags":[],"change_id":"I013e5a90a4dd5b6ba436b9f57ffe88fdc29eb326","subject":"Revert \"Increase per project quota to 50\"","status":"MERGED","created":"2024-10-29 16:17:37.000000000","updated":"2024-12-27 13:53:08.000000000","submitted":"2024-12-27 13:51:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"revert_of":933550,"submission_id":"933667","meta_rev_id":"424c7319885f3510318d33e5422aaae5af1da770","_number":933667,"virtual_id_number":933667,"owner":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-12-27 13:51:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2024-10-29 16:37:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2024-10-29 16:37:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2024-12-27 08:33:02.000000000","_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-29 16:17:37.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2024-10-29 16:17:37.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"44ef9e03dd4847667948c88b63e0c72d9472d0a8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-10-29 16:17:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f47a1a1d1a2ae5fb6117f987429668fb6b805360","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-10-29 16:37:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a872d2bc139dc52db535786911374510c50248bf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-29 17:45:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a7f63d6a03644c91b4f8e5dec9bfc82e\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8812f67cb52140e9816a18b9eb384469 : SUCCESS in 18m 03s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e78caa637a8e4902845c410379b1def8 : SUCCESS in 21m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fcd72acc06346d2889d0519e6e7f6f2bd5b58c58","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-29 18:34:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/294e41f8612f4ff786af6e94fb4d47e2\n\n- grenade https://zuul.opendev.org/t/openstack/build/2e18d188f3c44a5a9620250498955b43 : SUCCESS in 57m 29s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/76c87745f99c419799764f5254c906a4 : SUCCESS in 1h 00m 12s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/cfdd64aeb703478bb55ab3b5dcfc4990 : SUCCESS in 1h 18m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1857d41ad67342918e34a1c6a00f1118 : SUCCESS in 1h 00m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa44008f5d9f4be8a0ddfaab55d28a16 : SUCCESS in 4m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8b968bba94254b2ea64420f4c8fe2f26 : SUCCESS in 4m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3d7ece7a603945d2bfd71f802611cf40 : SUCCESS in 5m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e8e814669c14f948fea095d47182627 : SUCCESS in 6m 16s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/de8f1a8c97874ce182c388ea502b906c : SUCCESS in 8m 24s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/58a40cf6f8a34b03a6d6ccb6068b2faf : SUCCESS in 8m 44s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/3cf7561ea46146829a5aef5899c62ed7 : SUCCESS in 13m 17s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/a9163ebf06c74f4d9d39937367a1c2d7 : SUCCESS in 1h 36m 33s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/1b86fa174f6c4083ad9543b92fbef894 : SUCCESS in 1h 51m 34s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/e0e5cdc778f842f292f418d8eea28294 : SUCCESS in 1h 55m 13s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/1b0a4ed38cff46f4a4465ed4fede3aeb : SUCCESS in 1h 53m 52s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/1638a23b022e4cb395b93b6310ce7f65 : FAILURE in 2h 10m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5a61f98b94f0409592c317f610a8b7ab : SUCCESS in 1h 00m 56s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d398099b41574276a2563e139cab8fe3 : SUCCESS in 1h 46m 23s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/30c454d76a264dc3b1351a5460f1174c : SUCCESS in 2h 10m 51s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/9e2922ce9d1b4b61b44f0a3f4f0ba89d : SUCCESS in 59m 20s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/3f8e5933a2564980968c5cb060c5d17f : SUCCESS in 28m 36s","accounts_in_message":[],"_revision_number":1},{"id":"34af531b16c8ace52e19dcf9930cb18071ab3edf","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-10-29 18:50:49.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"31588be335d1726489d71ee7742411e216f4b2d9","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-10-29 18:55:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"519d923b2b8b350c861f8bd889d44572abaa0630","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-10-29 18:58:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3411e5e5a32806a94643d0d6a671cf05f1dc2e88","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-11-04 07:07:19.000000000","message":"Patch Set 1: -Workflow\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"318130dee14ab6116b72cd83c271125bfc7fdb37","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-04 07:18:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f0982fceb96248cabf1769500facdb86\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/223058b61bcb47c28e509ce872290e5c : SUCCESS in 7m 33s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/05be924448dc43948220682f5ba49399 : SUCCESS in 8m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5003470ed687f3c1611e1fc7f3e73fa6b7afa617","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-04 09:15:48.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c60c08e0f4f448b0b898d850df097dd1\n\n- grenade https://zuul.opendev.org/t/openstack/build/0c159e7cfd0e45f78c969c31f4ce61e7 : SUCCESS in 1h 03m 14s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/09c2ff2d009e4708b7b16c25c2b24462 : SUCCESS in 1h 02m 48s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/bb32bcb4f84347c6b00ae6a03bac1433 : SUCCESS in 1h 42m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/39d14f1e6d0146c1b21534a70a606601 : SUCCESS in 59m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5e8aa11df404c6ea1b71435a819905a : SUCCESS in 3m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0d8c28c3cf76410c9185fe9cf11157f9 : SUCCESS in 5m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a897fd2b2326401299bbab686f30f1bc : POST_FAILURE in 6m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6ecfd6b734c47c4a8d91fd9b72dc0f9 : SUCCESS in 5m 33s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d959593107ce483e9c93d1effe9f8f2e : SUCCESS in 8m 59s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7eda4c12b1b3478d9ccda5aad374cb2e : SUCCESS in 7m 08s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/c0cf9b9870bb45e6a991492eb4eeb015 : SUCCESS in 10m 32s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/8407d9677050427caa996d86b5f4eed8 : SUCCESS in 1h 14m 09s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/4436a241c70740daa96160196566d56a : SUCCESS in 2h 07m 18s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/ecb2c2bd2d0a40d6af38b8c90e613e7c : FAILURE in 1h 18m 11s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/c881be8fff8a4cd98727ea2239fc5b13 : SUCCESS in 1h 53m 58s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/ff1e3f3c609347d6a0b151b6a70e47ab : SUCCESS in 1h 29m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2e22f72bbe964ba0bfe5d80f8a08112a : SUCCESS in 52m 44s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/51e510d7e7fa4c4494867628ce5e8c96 : SUCCESS in 1h 12m 49s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/af86c65ca8aa4ef5964f5cd36bac60e5 : SUCCESS in 1h 18m 52s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/c2454880838e415f8945359988168693 : SUCCESS in 59m 17s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/19c6724144d34ebaad40111b71d229db : SUCCESS in 21m 11s","accounts_in_message":[],"_revision_number":1},{"id":"62c566da6ebbe209160639495562e069a296cd04","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-11-05 05:55:02.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"aa4a5a641b790451b478a0f3a267b8e9a729b15e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-05 06:06:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6848e28b460648e38a10bfc0945ed7a6\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cb45d27dd41042c2b5514f9add4fb1b0 : SUCCESS in 8m 54s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7dea5d9fda32489280c66b7ff65d3dd3 : SUCCESS in 9m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1de35c78bee8f8f813ccf6c14475361c973507ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-05 07:54:04.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f77cec41cec44eabaacfb71b7d1d37b4\n\n- grenade https://zuul.opendev.org/t/openstack/build/438a6587a5cd461cbf04428efc18ceb2 : SUCCESS in 1h 02m 57s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0d6096d2119f47288ce23cd7db706c83 : SUCCESS in 44m 03s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0b448df456b843fa9c8348998c5eab3a : SUCCESS in 1h 12m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/63cc1462468343ffbd22639fb73425d7 : SUCCESS in 42m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5997e5abb19147deb875f76b244ab4e7 : SUCCESS in 3m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c8d4732b617b4ed982d1bc27da27469e : SUCCESS in 3m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/70e97c980b9c460d84536ad8871853df : SUCCESS in 6m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3efb4fd35cd34492a832affe408b86d3 : SUCCESS in 6m 58s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5d74879006da48febb6eb942174cda56 : FAILURE in 6m 29s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9cc5a55d41204c69b768ed661fc5b460 : SUCCESS in 7m 12s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/a5510a8114514f10b5dbdb2c93b63ea6 : SUCCESS in 10m 18s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/3c0c605fe05740faac40ff22219377d7 : SUCCESS in 1h 39m 42s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/78f9907debeb423d9a590433ea21d302 : SUCCESS in 1h 43m 02s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/9e19eea69e12439fa975718f74005713 : SUCCESS in 1h 26m 51s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/97087268313e4a1588c0cdffa8fee4ef : SUCCESS in 1h 51m 46s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/b10984de64cd4cab8fb73191a2fcfd6d : SUCCESS in 1h 46m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/baa5250880694c21bdea4044bd6bdfe7 : SUCCESS in 58m 31s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/1fc85561161d4fa7a4988b56d0f95665 : SUCCESS in 1h 41m 38s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/70d25573d4a84f7ba1c3acfec030dc84 : SUCCESS in 1h 52m 30s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/c475401de1d34462b95cdbe8fd24383a : SUCCESS in 1h 08m 20s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/c778f638b0cf4c0f85b50d966e7e25de : SUCCESS in 34m 10s","accounts_in_message":[],"_revision_number":1},{"id":"1627b9c54a16da9f1206288855a7931858d462eb","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-11-13 14:00:42.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5d529b7591583d628520d7390799088ff054a6b6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-13 14:39:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2f1b95bc59114f7f950bafaf89b0ed60\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/70484623c1d94b9d849f77fa34beaf9f : SUCCESS in 11m 28s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d91b6708b0624b078488f41b0f4ea435 : SUCCESS in 12m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fdb0e4c97839c6ba910f14e27815a9cc7ef0a04b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-13 16:17:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/989074b613814da7bbe1988ca5c5fa5e\n\n- grenade https://zuul.opendev.org/t/openstack/build/820f0a00c5c64ea6bef3e64333a42607 : SUCCESS in 57m 44s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/59d8b078f1ac425da9a76480fb59decc : SUCCESS in 1h 06m 11s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/639f7fc1044d4621b17a3c9c1c8b709c : SUCCESS in 1h 40m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ed3c07a1d39040bfa70cdf7d0c4f35e8 : SUCCESS in 54m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07fb7d1256824c91a6a74fc83e5ffa0b : SUCCESS in 4m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e7064a1d60e1479487fb902e9f245d27 : SUCCESS in 5m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/980182c9e319417e81313a0c16127e39 : SUCCESS in 5m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24fe62556acd4742845752008a2c67ce : SUCCESS in 5m 34s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4a5632c9f0e04425a7de882f5e3b585c : SUCCESS in 8m 07s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/45e05f8fe90c43f8869f56cef1f2a2f8 : SUCCESS in 8m 47s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/0f8c0fb168e04d9da9f1bfce88fa6892 : SUCCESS in 12m 56s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/4fd005f1810246b682d792ff3b4f71f0 : SUCCESS in 1h 43m 41s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/1ea4f9665be04ff48ad4029dfcde5450 : SUCCESS in 1h 36m 22s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/050ce4288f124ad7b3b808fef5c67aa2 : SUCCESS in 2h 02m 57s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/d5a64c9da9694abaad7533c49f2a663c : SUCCESS in 1h 08m 05s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/9ce7ddfeb05d4f6fbf8b43ec81e6746f : SUCCESS in 2h 09m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/486d06de23c1488f9d714ac7731fc4d0 : SUCCESS in 1h 03m 27s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/65f042a54b9744a184106546405785fe : SUCCESS in 1h 43m 59s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/fbb25951d86841d78163633351065cba : SUCCESS in 1h 20m 06s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/cd1c7d6da7ee426cb4821e37bab8d0f5 : SUCCESS in 53m 55s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/859bdcec5aa54278bce62d79ed2ad27c : SUCCESS in 35m 26s","accounts_in_message":[],"_revision_number":1},{"id":"c7a0f654a98b92db3e44f3558291cb6a098e9dd1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-13 16:17:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"752c2b4b24f126337c167c9b0d1c57ca1b48e385","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-13 18:32:41.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/932100d718c24131a72328b4d9cd5b1b\n\n- grenade https://zuul.opendev.org/t/openstack/build/6df521dcf1a1489683e9391f9204b3fb : SUCCESS in 42m 08s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1b0d6381d93e4ef5bccb0fbac791aee4 : SUCCESS in 1h 02m 30s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/af759fffd7c54fc38b3e27874512ff10 : SUCCESS in 1h 51m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fc02ed362aa245de84a893152f503300 : SUCCESS in 1h 00m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0951e85b25a415b812c4b814c44485d : SUCCESS in 4m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ce0015102c1d482baf0921da8887279f : SUCCESS in 5m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ea69c68c294842999b3da2c92e0f29a3 : SUCCESS in 6m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3042796560b848319d0e70be803f682d : SUCCESS in 5m 35s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0aa7b1919a16419f864d03e8951c31d1 : SUCCESS in 7m 57s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5a040f9f29f74e1a920e7822683b182d : SUCCESS in 8m 27s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/547e4ef5339e4a04857b58b358ffcbdd : SUCCESS in 13m 00s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/b3590855720e4655a2a7112ff1cebd6c : SUCCESS in 1h 44m 33s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/3c5739e83cc849f9b8572ecbf08fa039 : SUCCESS in 2h 09m 07s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/f6a29198f8604f0bba9cea3cda228343 : SUCCESS in 1h 48m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/377d65cd806c49168c5e247bd2c3012d : SUCCESS in 42m 03s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/0e50c3eb404b4aba93a896e3ef58fa6c : SUCCESS in 1h 37m 10s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/9f1a006a8b5247a29caf89ab3c3a6f7b : FAILURE in 17m 33s","accounts_in_message":[],"_revision_number":1},{"id":"5e72bb974064138f31ceca76c64b0aac62f650b4","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-11-14 11:41:55.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d1b9455f0d672b4e72a849fc4492802eaf63872b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-14 11:52:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dbefec5e7e03462a849df2c0042de545\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8d0fc6218e96405591158b944a4c85d4 : SUCCESS in 7m 35s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5828dc01ca34428da79981f728937970 : SUCCESS in 8m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d8dd7fe15584c49454841263dccc138d6ff95cdc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-14 13:40:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/620398d2299e486aa0c89164249e0bd9\n\n- grenade https://zuul.opendev.org/t/openstack/build/f211d528687643468790d6c11668563b : SUCCESS in 38m 27s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/83ac3975eb5940d387be3bd000ca0531 : SUCCESS in 1h 01m 41s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/84a5888695224311b97defe94f2f4f6c : SUCCESS in 1h 24m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/376b219cbe4e4effab7f37ab52c1dc65 : FAILURE in 35m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5131fe41ec2046e2b806b66269b8ef38 : SUCCESS in 4m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b44b9008110d48088dab76722c777824 : SUCCESS in 5m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a45510604b8341068fa3b50a3125287f : SUCCESS in 5m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19b020a44e7c4e70a1fe4ec619ab43b5 : SUCCESS in 5m 34s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a58cefd3d1ec44468b756d5c44814fec : SUCCESS in 8m 16s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/09cc382c00a3417681bd3c60e15bc4cd : SUCCESS in 7m 52s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/1d6b7df51eb149a7a4221fae903c6ee3 : SUCCESS in 12m 47s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/317651a2f7904248b8d8a89421dc27c7 : SUCCESS in 1h 48m 00s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/85d441885a6f4c35b61ff161983df193 : SUCCESS in 1h 36m 02s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/1a8f2ce43f0d4b019153cf4cb9ce0041 : SUCCESS in 1h 53m 40s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/93f2b3c8e23c4b0eb5526c18d8a61aca : SUCCESS in 1h 44m 52s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/82517d7212d446c19dca520849e94d11 : SUCCESS in 1h 52m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d7d92652dd074991bd0c0dc0d279ef67 : SUCCESS in 46m 14s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/03068dc3514243ef83d80445ea59504f : SUCCESS in 1h 41m 40s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/c71850f2f75e4bd386fb1c2a23fdf09f : SUCCESS in 1h 53m 30s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/b4093e3b939540118c3cc122684e41b3 : SUCCESS in 1h 01m 27s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/71114150cffc46259f8720a475532e9a : SUCCESS in 24m 13s","accounts_in_message":[],"_revision_number":1},{"id":"0a7a0262b0d48930b2c2c592961c6546a0439908","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-11-15 08:09:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2e7b2f35a9fb2aa3fa855653173fa5969572f170","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 08:21:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/203e5fc268c646319fcc67ed781e9d07\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c3844cda68fa4d3b94f8914dc93f1ddb : SUCCESS in 6m 54s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a23190a9a3c8477ba029a9fdfceb79b6 : SUCCESS in 9m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"61958cb8ae6e22723d32e59042ba331589be24d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 10:27:35.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1c0851495a7449a6a24fff0b70c86ef0\n\n- grenade https://zuul.opendev.org/t/openstack/build/85ea3ee593c644248622b8b30cc28d26 : SUCCESS in 37m 14s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/46664789c8f6475ea6768a95e9fc2990 : SUCCESS in 45m 25s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9d15796fc4de478cbee579148a829021 : SUCCESS in 1h 08m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9463122b0b7349f8afaf644af04615f6 : SUCCESS in 1h 02m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eaeb7423c33d4dafa22a566df85779ff : SUCCESS in 5m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d68f37adb0eb45e7b7fe80d22d2d2f15 : SUCCESS in 4m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/253dd05c26104daea61ba7081df1c4b8 : SUCCESS in 7m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48bb89c06f1f422191f4fbcb383427d6 : SUCCESS in 5m 59s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/38a562638a574e5aacd6618c4dc2b344 : FAILURE in 11m 39s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b11dd96838d8413a84c1687fbdd2a5bc : SUCCESS in 7m 28s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/7dc2c97b6d7742d6beb9413aa5557cf9 : SUCCESS in 12m 09s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/6b0461cfba504d5f9ae708a69d4912a9 : SUCCESS in 1h 46m 49s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/43fe6b1770b34a59bc17f20af5673c54 : SUCCESS in 1h 48m 29s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/fe1037182c3348e196fcf39136653132 : SUCCESS in 1h 15m 02s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/8a0173a8ef25461a9b2cb90510d7fc54 : SUCCESS in 2h 05m 06s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/f1d2e4c44b1f41d8ac40917ca8b02661 : SUCCESS in 1h 42m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b8744cca9755464a8b2cb5a710fdcb06 : SUCCESS in 37m 40s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/0f9a6d0cdf8247348822d859d43519cd : SUCCESS in 1h 45m 30s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/d4dac234628f4cb9a03d68ec97a90d42 : SUCCESS in 1h 31m 27s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/ef3349b2f9294a7bbd87e477397df8d8 : SUCCESS in 55m 15s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/32a1031eddf443eeb7464ce217e00d67 : SUCCESS in 32m 30s","accounts_in_message":[],"_revision_number":1},{"id":"79730b88d6c712aba927d8c08b452c65b9a9ba88","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-12-27 08:33:02.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b5681c37713b4946824531284bf27a12912ce319","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-27 08:45:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/230af31de4074ec3a273eb843a501be7\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9ba77358836247a6a7139316efd00a85 : SUCCESS in 8m 50s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a3a8de7603c94e519eed5977fb9fc567 : SUCCESS in 10m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f85591be8245d907abb3fe91ccf6d946ea009819","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-27 11:19:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d83496477674f0a94c8b6124beb2baa\n\n- grenade https://zuul.opendev.org/t/openstack/build/91db61f5dce940aabfeea02b377dcb06 : SUCCESS in 57m 09s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/96c2ea0585fe41bf9a29605153ca53bc : SUCCESS in 57m 38s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/cbf5db7ac05140c384560b56b8f6f42b : SUCCESS in 1h 34m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d9798d66834848ed98883b6c731a7ae6 : SUCCESS in 1h 02m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9ebed682853416f8da180bd7c83954e : SUCCESS in 4m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d983527331f6476187632e9e9e61685b : SUCCESS in 5m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fda1e2c03886430fbe97048af820d7c3 : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c1c8c6ad5744bf8b35bcde18e7f6a47 : SUCCESS in 5m 18s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3245abaf3ba448a7ae889183f58a33fe : SUCCESS in 8m 32s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8198d96719d448a195ec3af96768036f : SUCCESS in 7m 52s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/6e4378356a4c4a788182cf265f862b3d : SUCCESS in 11m 11s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/2e40235b491c4d84b6099dba0d22dd3d : TIMED_OUT in 2h 03m 40s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/aa458644c56b411f84719b0531b91535 : SUCCESS in 1h 51m 41s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/76934b51afcd41f181ab3aecabbf3278 : SUCCESS in 2h 00m 03s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/97eab12b21c94efbaafae715b996db28 : SUCCESS in 1h 49m 42s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/9385620af50d4f9fb54814c765bf5642 : SUCCESS in 1h 54m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bdf6708f1b4f4a4c9bd78c84dcc74efa : SUCCESS in 57m 24s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/eb2beece1efc4735bc0e0ed28dce04eb : SUCCESS in 2h 01m 33s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/3534b4a261664da7b4944084a6cc246d : SUCCESS in 1h 57m 57s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/b8974580847a4f31b03e789d42769be0 : SUCCESS in 1h 04m 09s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/0adbe64e208c4c70926e65504815b6fe : SUCCESS in 1h 47m 53s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/4559850b0e154759bf1729a5c75d4f50 : SUCCESS in 37m 14s","accounts_in_message":[],"_revision_number":1},{"id":"d93a5df1c6e7cd541b16c7ce71e6121a4c31f5dc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-27 11:19:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"28c9a57dd2238f823007a2c9c4a909dc538f74a6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-27 13:51:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/44cd816b1d574b6593067b4cf142757f\n\n- grenade https://zuul.opendev.org/t/openstack/build/8153b13872a548b2ad7126c36e188e80 : SUCCESS in 59m 59s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d768702e6968496ca48b7419dcc7d949 : SUCCESS in 41m 23s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/de7b5b4055394e299b34f7c0d16f71bb : SUCCESS in 1h 06m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5f4d34d4ba114c47954143052f38c967 : SUCCESS in 47m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ac81e8b914346f195c914e2d0a4a4ba : SUCCESS in 4m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/63a8b7d8f6b84ae58eab379c5f61249f : SUCCESS in 5m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/54c89027cdee4689a0227df92a4f49f6 : SUCCESS in 6m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c42cff9ac3e44548a6de98cd553ef49b : SUCCESS in 6m 08s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1485220f21064264a8aad1ff8b2b5603 : SUCCESS in 8m 38s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/57ba8fb72ada4d8d9780a6cc2e58df38 : SUCCESS in 8m 05s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/638ed3cf1db34ce18bc58fd225bca376 : SUCCESS in 11m 42s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/e8a9286cb6bf4736958aaaa1a603136d : SUCCESS in 1h 37m 47s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/c10ed9871f7849079864ee44273f6ba1 : SUCCESS in 2h 20m 22s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/a0262a8f2e6649f3b276612d2eb9c932 : SUCCESS in 2h 26m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4247493d7d4f48a6910426d5f97c76ae : SUCCESS in 50m 12s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/0c2b12a6a1144b4fb91d861ac6a4631c : SUCCESS in 1h 57m 52s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/9ad4c94f7df7400f8208ea90998dc881 : SUCCESS in 35m 51s","accounts_in_message":[],"_revision_number":1},{"id":"499a0b0da1c6939988aa1d619bf03fb6d79d5d07","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-27 13:51:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"424c7319885f3510318d33e5422aaae5af1da770","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-27 13:53:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1905b23f77f5449e995cbf162d3efe52\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af2aea0f3da84912865692dac0ccd771 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"26b9e49ab93d71307ccbfc6acea397b9e7ad6924","revisions":{"26b9e49ab93d71307ccbfc6acea397b9e7ad6924":{"kind":"REWORK","_number":1,"created":"2024-10-29 16:17:37.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/67/933667/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/67/933667/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/67/933667/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/67/933667/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/67/933667/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/67/933667/1"}}},"commit":{"parents":[{"commit":"704b24fd67b6d3dee5f1f42f9b793832b4b47bf4","subject":"Increase per project quota to 50","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/704b24fd67b6d3dee5f1f42f9b793832b4b47bf4"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2024-10-29 16:17:37.000000000","tz":0},"committer":{"name":"Gerrit Code Review","email":"review@openstack.org","date":"2024-10-29 16:17:37.000000000","tz":0},"subject":"Revert \"Increase per project quota to 50\"","message":"Revert \"Increase per project quota to 50\"\n\nThis reverts commit 704b24fd67b6d3dee5f1f42f9b793832b4b47bf4.\n\nReason for revert: We can revert this as the main fix is released now\n\nhttps://review.opendev.org/c/openstack/releases/+/933578\n\nChange-Id: I013e5a90a4dd5b6ba436b9f57ffe88fdc29eb326\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/26b9e49ab93d71307ccbfc6acea397b9e7ad6924"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/26b9e49ab93d71307ccbfc6acea397b9e7ad6924"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
