)]}'
{"id":"openstack%2Fglance~933905","triplet_id":"openstack%2Fglance~master~Ie5708bbbfe1e4bd8badb11786b0d38ffe765a942","project":"openstack/glance","branch":"master","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2025-05-07 17:03:10.000000000","reason":"Change was submitted"},"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2025-05-07 15:02:04.000000000","reason":"\u003cGERRIT_ACCOUNT_9303\u003e replied on the change","reason_account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},"16137":{"account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"last_update":"2025-05-07 17:03:10.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie5708bbbfe1e4bd8badb11786b0d38ffe765a942","subject":"Replace links to HypervisorSupportMatrix","status":"MERGED","created":"2024-11-01 09:41:06.000000000","updated":"2025-05-07 17:04:31.000000000","submitted":"2025-05-07 17:03:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"933905","meta_rev_id":"8689f6ae24a26386e0224f455f26eb6ce703e91e","_number":933905,"virtual_id_number":933905,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-07 17:03:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2025-05-07 15:02:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2025-01-07 15:54:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-03-13 18:04:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2025-05-07 15:02:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-01 10:47:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-11-01 12:15:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-07 15:54:44.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2025-03-13 18:04:47.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2025-03-13 18:04:47.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2025-04-12 14:24:14.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"}],"messages":[{"id":"e8303e129b5dabf3b150bd7d292ccfaa6940f80b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-11-01 09:41:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c673d42e942396a27e94f84c994acab661e4d3a4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-01 10:47:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4a2038b047cc47108c2235cc5205ec74\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9619afdf81da4503ba8072158a1e7524 : SUCCESS in 13m 09s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/57ed7ed0fb73431198be757e7279cb76 : SUCCESS in 17m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5a2574562f69ebcb2534d0bf552582961613c8ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-01 12:15:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7173072cb0df4e24baf0c0d8f6763776\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/94f858708f7947c0a81f40b100626d4b : SUCCESS in 2m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/6c41f8476ee546e9ac43d060cea17b49 : SUCCESS in 46m 13s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/933d0ba9fd95496083821d9bf78759d6 : SUCCESS in 47m 57s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d6e07023aa6f4b4eb68dbc89ce783719 : SUCCESS in 1h 53m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ac23d714a6494abcb49bda45f6d6ee4d : SUCCESS in 48m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f2b76f7f73d49f1ae3bb13f9599f637 : SUCCESS in 3m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2e0853abb1c44c29ad3910327db80597 : SUCCESS in 4m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e57ad8988e624a3695cd63cafe575715 : SUCCESS in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc1f5df311b64e18b7b0fbf7fc0aeae5 : SUCCESS in 4m 45s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e631a3083eaf4de99818aca16c11be28 : SUCCESS in 8m 01s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/25c04c80e10645b69bfeea5508a4d023 : SUCCESS in 7m 51s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/a299d1ba38024379a12b5432e48ae7e9 : SUCCESS in 10m 29s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/e0fd1881e29640e186f9edf03105f9a5 : SUCCESS in 1h 29m 39s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/1ff845edf5a84dcd9d29c03c164a6e1a : SUCCESS in 1h 32m 47s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/308bc823d5654d9aaa08fb865581e405 : SUCCESS in 1h 35m 05s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/ea13a7d265ae4c8cb5f3cdda47fc411d : SUCCESS in 2h 22m 29s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/6a44b809b6444727b4e3109b1f3ae451 : SUCCESS in 1h 40m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1ca38a19588b4ba59925a8e7fd19dbef : SUCCESS in 1h 01m 28s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/b64b0b66893a4728a501fcff1f86a176 : SUCCESS in 1h 08m 01s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/ceee5f00485d4f82a5cb72e4c84f8e48 : SUCCESS in 1h 56m 13s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/b163598ae21a45cdba08513ac23aaba2 : SUCCESS in 40m 25s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/05187ca0833349eca24723c45776afe3 : FAILURE in 15m 13s","accounts_in_message":[],"_revision_number":1},{"id":"d1b6de7ff4c4837699a8a0772f45b9f7b291fcb3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-12-09 10:00:00.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"5b45e77841a50d46e8b3166b054f23f65a3ed2cf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-09 10:45:02.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2f4c033095454961bc05fbbe7baec565\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/04be0df849774cb4a42eeb834c4387f9 : SUCCESS in 7m 57s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5b2f4e5acb4943c98cf4243bbba2cad6 : SUCCESS in 9m 08s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"362d8c680ab25f85cb05c22f5bbcd2b3306aa93f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-09 12:50:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/21df85a826774dcda8e2ca10ecf80219\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/5b34a8af89af43e6a4349f2d6f64699b : SUCCESS in 4m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/0c7f309b2b3c417a81670548b23e342e : SUCCESS in 47m 51s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d1939282de9341dd82e370542895fe11 : SUCCESS in 54m 55s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/878a1afa782344fb90aa56cd6567cdf6 : SUCCESS in 1h 57m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0965b065e5ab4466b68809bda65eaada : SUCCESS in 51m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee1732f0bc874a5a8821b762521dc37f : SUCCESS in 6m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/98fbc4da179c4b3ba4190004f09473b0 : SUCCESS in 5m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/98619366fd4948d796f6f8a6491fed74 : SUCCESS in 5m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b24b4d2ff594019b91e30911b7a8934 : SUCCESS in 8m 41s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d4d93b8bcb24496ab6cea505e27f607d : SUCCESS in 8m 05s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6b1e36e96b7344dea226daa44d2279ea : SUCCESS in 7m 29s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/3d5b25bdb2854705b3d9f9af4eeddbbc : SUCCESS in 10m 52s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/47fc0e49eed64748ab4ba880323e209f : SUCCESS in 1h 43m 49s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/e9229b393c214e5cad4cc0b8f7275801 : SUCCESS in 1h 40m 18s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/d8e607fff5284d1bb12226a490be2f48 : SUCCESS in 2h 35m 05s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/a5ffa816956a493cb2c3297c97427aba : SUCCESS in 1h 46m 25s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/0d08ff2135544e41833b5739853c5680 : SUCCESS in 1h 57m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/974cf676ba7f4fcab943ea2e3a96d32b : SUCCESS in 58m 22s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/79e6e0be673f4b258590c1c83550b766 : SUCCESS in 1h 30m 45s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/c7d0ad36303a462482f95d62c4e2cffa : SUCCESS in 1h 52m 25s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/8340e5f957cc43b6b4c41be2cfe01da7 : SUCCESS in 1h 14m 28s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/22edeb7932a84a4489f056c9ddb59ea3 : SUCCESS in 1h 30m 06s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f792a9da78e148189ba73dcca1f81800 : SUCCESS in 34m 04s","accounts_in_message":[],"_revision_number":2},{"id":"13d52ef5bc44fd790c47027002ec8e38b0063406","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-01-07 15:54:44.000000000","message":"Patch Set 2: Code-Review+2\n\nThanks!","accounts_in_message":[],"_revision_number":2},{"id":"2a590acdcc014262078c94a4d9ce42f3f1917004","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-03-13 18:04:47.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"e680f37ab1d46e2915f81af9cd564d4d4b650c10","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-05-07 15:02:04.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"3efb9bb817a8b373771723c1e83695b79afc8971","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-07 15:02:30.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b5502d68b4dc369d017dbbe3d937b9f12aaa9a61","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-07 17:03:10.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/db155676eca24519ae94212ab467c432\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/a4f5db6b005741118259aab131380c44 : SUCCESS in 4m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/64018b778dcd48cc9177c2e76a3bf7f8 : SUCCESS in 55m 22s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/07967feebb654dcc8c0c663f39981e29 : SUCCESS in 54m 41s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/59b91f526f06497ea36535c4cb7e87f5 : SUCCESS in 1h 28m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bb644d6e981a4c17ada64ce38fcc0180 : SUCCESS in 58m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f812b2ee6f4845868f4ee74129ddf335 : SUCCESS in 4m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4c30693d291e417493800a99e7e4dd03 : SUCCESS in 3m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7d101f1af797444e92cf79f817a8bb5b : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/345658d29fca4ea2aa7822c16e56933e : SUCCESS in 5m 47s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1f116aa310aa41348532e65fb7539cd5 : SUCCESS in 7m 25s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/6611684f32f843469b54b1ed8a95c38b : SUCCESS in 8m 44s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/60c3092cdbab43e1906fb59c7a798ba4 : SUCCESS in 1h 42m 53s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/031d2525f39c4dceb99c30c196f3ae62 : SUCCESS in 1h 08m 16s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/23e324172c23426ca98f66ac71f313de : SUCCESS in 1h 56m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/01b68b2d83f3488fbaaac8f25b319f4a : SUCCESS in 49m 44s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a04b2695ea834543ba1ca8271ba88c78 : SUCCESS in 1h 00m 10s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/8776cb0360cb4682926550352dc4b353 : SUCCESS in 29m 33s","accounts_in_message":[],"_revision_number":2},{"id":"bc5860e2e9c84a62600e66d912b96bae97d4798f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-07 17:03:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"8689f6ae24a26386e0224f455f26eb6ce703e91e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-07 17:04:31.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/833eec57c9d840a6827d4585887d91a2\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/25f817c0521442b188f0c2a38e11785c : SUCCESS in 44s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93ed098714b64943bf03e335b87b77bf : SUCCESS in 51s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"20e1fee5836ebfdb08e92315471057f9f2e4416b","revisions":{"b1e6b511bcabb656344818b60c21e6571a85ca5f":{"kind":"REWORK","_number":1,"created":"2024-11-01 09:41:06.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/05/933905/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/05/933905/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/05/933905/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/05/933905/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/05/933905/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/05/933905/1"}}},"commit":{"parents":[{"commit":"acab9351a1fa124f2bdeb58d2f407070071dc9d2","subject":"Remove Windows OS support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/acab9351a1fa124f2bdeb58d2f407070071dc9d2"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2024-11-01 09:40:10.000000000","tz":60},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2024-11-01 09:40:10.000000000","tz":60},"subject":"Replace links to HypervisorSupportMatrix","message":"Replace links to HypervisorSupportMatrix\n\nThe HypervisorSupportMatrix document [1] on\nwiki.openstack.org is not longer updated and\nthe support matrix in the Nova documentation [2]\nis the correct place.\n\n[1] https://wiki.openstack.org/wiki/HypervisorSupportMatrix\n[2] https://docs.openstack.org/nova/latest/user/support-matrix.html\n\nChange-Id: Ie5708bbbfe1e4bd8badb11786b0d38ffe765a942\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b1e6b511bcabb656344818b60c21e6571a85ca5f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b1e6b511bcabb656344818b60c21e6571a85ca5f"}]},"branch":"refs/heads/master"},"20e1fee5836ebfdb08e92315471057f9f2e4416b":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-12-09 10:00:00.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/05/933905/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/05/933905/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/05/933905/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/05/933905/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/05/933905/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/05/933905/2"}}},"commit":{"parents":[{"commit":"e0f5bb97c2a8b6fa83d985a0c6f795b24f1cc981","subject":"Merge \"Add ``fsid`` to RBD store properties\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e0f5bb97c2a8b6fa83d985a0c6f795b24f1cc981"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2024-11-01 09:40:10.000000000","tz":60},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2024-12-09 10:00:00.000000000","tz":0},"subject":"Replace links to HypervisorSupportMatrix","message":"Replace links to HypervisorSupportMatrix\n\nThe HypervisorSupportMatrix document [1] on\nwiki.openstack.org is not longer updated and\nthe support matrix in the Nova documentation [2]\nis the correct place.\n\n[1] https://wiki.openstack.org/wiki/HypervisorSupportMatrix\n[2] https://docs.openstack.org/nova/latest/user/support-matrix.html\n\nChange-Id: Ie5708bbbfe1e4bd8badb11786b0d38ffe765a942\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/20e1fee5836ebfdb08e92315471057f9f2e4416b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/20e1fee5836ebfdb08e92315471057f9f2e4416b"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
