)]}'
{"id":"openstack%2Fglance~934101","triplet_id":"openstack%2Fglance~master~I7a69dea2443244264d037e7657d59230dc6749a2","project":"openstack/glance","branch":"master","topic":"db-migration-version","attention_set":{},"removed_from_attention_set":{"19138":{"account":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"last_update":"2024-11-07 16:22:00.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7a69dea2443244264d037e7657d59230dc6749a2","subject":"Change DB migration constant to 2025_1","status":"MERGED","created":"2024-11-05 06:54:56.000000000","updated":"2024-11-07 16:23:24.000000000","submitted":"2024-11-07 16:22:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"934101-db-migration-version","meta_rev_id":"3f8a504450ccbd0e9de214d36be206fc2eaab260","_number":934101,"virtual_id_number":934101,"owner":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-07 16:21:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2024-11-07 14:05:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2024-11-07 14:05:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-05 07:18:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-11-05 08:58:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-07 14:05:12.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"80ca46e3b08bed29deaaa9426f17c84d8187ad7f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-11-05 06:54:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d7246a1488a97e086e840f9b02c67bfe6aec41c0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-05 07:18:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/476fc483e28649818525034b78d05569\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/643ccfb4b9a348bcbda42fc138a35666 : SUCCESS in 9m 56s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/51242396922847ff8464783042e3e3b2 : SUCCESS in 11m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d2a581deeea33824c6ba39e43b7eee3e7f05f619","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-05 08:58:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7650744dc17340339bfb7b3199c4ca36\n\n- grenade https://zuul.opendev.org/t/openstack/build/56e07545b3f94a7f80cd9692577ff32b : SUCCESS in 1h 04m 53s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2c3de2758dd341a1bf7b3fe09ee3cdf8 : SUCCESS in 54m 20s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/55b37044bea24f7e945828dceb856070 : SUCCESS in 1h 33m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b32defca16e24570af93bd4de4392b37 : SUCCESS in 58m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8315f2e0a91484599716bff1cb01712 : SUCCESS in 4m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eb24cc383d614ec2a2cf1822c93f8b85 : SUCCESS in 4m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6a95f1677b8845ecaafe18be8786a74f : SUCCESS in 5m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34f6b39f677749648574e8089e57c472 : SUCCESS in 5m 23s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b7bb0adfde51478c9b7e047768c92825 : SUCCESS in 8m 32s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f685b3c6ca1a4ba9a705c2f1ca63c73d : SUCCESS in 6m 38s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/79b202fc91e04a2b9429c595fca2ecd7 : SUCCESS in 12m 19s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/6424818262354954b3eee00f65bd9084 : SUCCESS in 1h 07m 56s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/0b89376d5e2b4aa19178677772e88f11 : SUCCESS in 1h 29m 17s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/25abeff8507a43859b0127b9d14de32d : SUCCESS in 2h 01m 34s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/2e78c00e89134ab4a3ab19f4b47854c6 : SUCCESS in 1h 54m 25s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/e4ba2143bce84624a4e05e6f2cfdf5bb : SUCCESS in 1h 55m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f92ad60a943149919d26ced7dd5eb127 : SUCCESS in 34m 25s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/3abc30903b634a5c8d84ebb87d034190 : SUCCESS in 1h 42m 37s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/83bdaf46f65341f7911f1b5ef98e939f : SUCCESS in 1h 44m 36s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/8f674080f06749f9a33020e8e07419ff : SUCCESS in 43m 28s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/18d01527e15e4bb593a0d0456d01d693 : SUCCESS in 38m 33s","accounts_in_message":[],"_revision_number":1},{"id":"a27ec205638871c5f120716dccaed514301cc818","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-11-07 14:05:12.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"32b7f5b9f8e2fe1ab52b7f3a7987d3a9867dca84","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-07 14:05:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3d6d7e996b2fdcb725ecc69b01a30f04aa837e48","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-07 16:21:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9721a4a743ce4a5d98ed0890fae39194\n\n- grenade https://zuul.opendev.org/t/openstack/build/2804d44b19b14494806ea3945c070cec : SUCCESS in 54m 51s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0fe3d6cd6c954846aec1b0975ce3f67a : SUCCESS in 44m 03s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/09ef2a565c7f4d51855495e1d3884a12 : SUCCESS in 1h 32m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1571262f5a2c4a78a4d4e7bd4b026069 : SUCCESS in 50m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b265e52266b547f4a73c9b841fce7a69 : SUCCESS in 6m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/50368a2ae07b49fdb2f6f4ce83587e3f : SUCCESS in 4m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/22f6f07d74574c9d8704b95c761db427 : SUCCESS in 7m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f42ab56ddfb400ca11398ceaca02d44 : SUCCESS in 5m 34s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2b35d7dc4838498f9a0cd219137a7785 : SUCCESS in 7m 34s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/cd5aa9255a8a4b3f9834d341c16c0137 : SUCCESS in 7m 20s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/efdd1851a8b2497c928a7320d32f3769 : SUCCESS in 11m 55s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/a90fc68e444b4505b1c156de5ed01241 : SUCCESS in 1h 50m 21s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/534527ba2e304c6b9584eb2116a75a76 : SUCCESS in 1h 17m 55s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/114661cc69ac45eaaa40e82e867d7f2b : SUCCESS in 2h 09m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7814e2b3215242149e17df3770f8f2a0 : SUCCESS in 47m 59s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/916066ec2e9546f3a6d49851c460ea03 : SUCCESS in 1h 38m 56s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/5b38a5f8131849d893879d2f1d8796e6 : SUCCESS in 20m 44s","accounts_in_message":[],"_revision_number":1},{"id":"c0d2e3fa4688abb116a3975a1cabe6a39f2afbbd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-07 16:22:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"3f8a504450ccbd0e9de214d36be206fc2eaab260","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-07 16:23:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a2beb6a7e7c045d680e4d0ced6f7f632\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3bdf4dc1951b46cc844e980c24d67d0e : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"293cb2c0b5d623f0e5b2ab12c8583d36444a32ce","revisions":{"293cb2c0b5d623f0e5b2ab12c8583d36444a32ce":{"kind":"REWORK","_number":1,"created":"2024-11-05 06:54:56.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/01/934101/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/01/934101/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/01/934101/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/01/934101/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/01/934101/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/01/934101/1"}}},"commit":{"parents":[{"commit":"acab9351a1fa124f2bdeb58d2f407070071dc9d2","subject":"Remove Windows OS support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/acab9351a1fa124f2bdeb58d2f407070071dc9d2"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-11-05 06:51:19.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-11-05 06:51:19.000000000","tz":0},"subject":"Change DB migration constant to 2025_1","message":"Change DB migration constant to 2025_1\n\nChange-Id: I7a69dea2443244264d037e7657d59230dc6749a2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/293cb2c0b5d623f0e5b2ab12c8583d36444a32ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/293cb2c0b5d623f0e5b2ab12c8583d36444a32ce"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
