)]}'
{"id":"openstack%2Fglance~934562","triplet_id":"openstack%2Fglance~master~Ia48842d0e0254bc2185716a3199ec7f810374f5d","project":"openstack/glance","branch":"master","topic":"policy-yaml-file","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2024-11-26 19:32:22.000000000","reason":"Change was submitted"},"19138":{"account":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"last_update":"2024-11-26 19:32:22.000000000","reason":"Change was submitted"},"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2024-11-25 06:22:53.000000000","reason":"\u003cGERRIT_ACCOUNT_9303\u003e replied on the change","reason_account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-11-24 16:58:35.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2024-11-26 15:51:06.000000000","reason":"\u003cGERRIT_ACCOUNT_8122\u003e replied on the change","reason_account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2024-11-26 19:32:22.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia48842d0e0254bc2185716a3199ec7f810374f5d","subject":"Remove default override for config options policy_file","status":"MERGED","created":"2024-11-11 05:40:15.000000000","updated":"2024-11-26 19:33:37.000000000","submitted":"2024-11-26 19:32:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"934562-policy-yaml-file","meta_rev_id":"d91782ac3b51771e7f3bef62c5ea6628898e4c55","_number":934562,"virtual_id_number":934562,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-26 19:32:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":2,"date":"2024-11-25 06:22:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2024-11-26 15:51:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2024-11-26 15:51:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-11 05:54:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-11-11 07:39:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-16 03:27:33.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2024-11-18 06:49:41.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"state":"CC"},{"updated":"2024-11-18 09:06:27.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2024-11-18 09:47:51.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"state":"REVIEWER"},{"updated":"2024-11-24 16:58:35.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2024-11-24 16:58:35.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"}],"messages":[{"id":"522048808a7e5d4a1657be80e83997e67f31573a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-11-11 05:40:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0a9de7add4930899a42c8d9730647bc0f5a774ad","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-11 05:54:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7e30a1936ca7416aa8da0ddcb542056b\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e9fe80143acb4123a5d857b2ab0edf37 : SUCCESS in 11m 19s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/37cd25643a00430098137a60c17eb607 : SUCCESS in 12m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d4a39f86428adb50ccf27acaba9ab4d1c0f08a21","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-11-11 05:55:33.000000000","message":"Topic policy-json-file removed","accounts_in_message":[],"_revision_number":1},{"id":"38d99e6763c596b58362efc22d56560206f17ba5","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-11-11 05:55:37.000000000","message":"Topic set to policy-yaml-file","accounts_in_message":[],"_revision_number":1},{"id":"3405980ef6fbfda119ba7121eb3f81625d222938","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-11 07:39:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8019e9ac460b4cccabaff25e41c2c0e0\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/ca28ff220440499494dba11de426fc1f : SUCCESS in 2m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/cda76d155d15450cbcde3a342f502b89 : SUCCESS in 1h 01m 40s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/15585ad15dc740c48107f5287919d5a9 : SUCCESS in 54m 09s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7687c3f675074b128683fd32a595e9f5 : SUCCESS in 1h 28m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/83bc1a6d1afa4697a6fa48e3945cb8ce : SUCCESS in 1h 04m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b634497242d3430cb7ac10d8028a9f0e : SUCCESS in 5m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/77fe933b56824d0b80764a961810f068 : SUCCESS in 5m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/eac085d502f04b648e832c507f13eaf8 : SUCCESS in 6m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6a70f18fb244638b56e0b51347a5b64 : SUCCESS in 5m 59s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/11688f7c04944a7d9ce33ab2b09a8da7 : SUCCESS in 8m 41s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/00b4a603c56e41ec962e70a459cc62e8 : SUCCESS in 9m 41s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/01432d6506734d00a41acfef240ef399 : SUCCESS in 12m 38s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/2aa2dffda76a4cfd8947bbd3d2c4df13 : SUCCESS in 1h 15m 04s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/504f0750aa5d4d4097611e9351dde0c0 : SUCCESS in 1h 48m 23s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/08c007bf03134059a2f9d5aeee2f2976 : SUCCESS in 1h 32m 27s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/a2af181ba1ed4b3287162a7ee3b2be91 : SUCCESS in 1h 35m 40s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/bbb251667db6446ea566158fc0070975 : SUCCESS in 1h 55m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8e6a157fc10c42678d616dfb97584710 : SUCCESS in 57m 17s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/0db9ca7b17064541931007dae49a5129 : SUCCESS in 1h 50m 47s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/3f40259fdcdd4ed1aa9f8fdb021d88d1 : SUCCESS in 1h 14m 25s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/f13b93b9161c41f69e3edb3ea272bacc : SUCCESS in 49m 17s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/de8a92f8d02e4746b7438941e556d9f5 : SUCCESS in 26m 59s","accounts_in_message":[],"_revision_number":1},{"id":"562e821d7193d219d3a18a681c7882024edfda6e","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-16 03:27:33.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7c6dfea81dd174e58ffa18dbda756e1353fea8f5","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-11-18 06:49:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d0b6a802143b3214bb3469992ba1a545d01192ef","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-18 07:07:07.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"02545d04c8518d497d732a9a9ec24cb9d8a37d46","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-18 08:55:03.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"9cbdc80f4e262f68725b32c4ceb03a7a5b8a307a","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-11-18 09:06:27.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"150a16c89629a30b178600eab945e1bd9a398ccc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-18 09:09:02.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/df5fad1a74a9497db82e0440d0ba158d\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/51450fa331e1403182d725532e37b3c1 : FAILURE in 8m 34s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1f3c0c8b3f0a41038050259c0557bd4e : FAILURE in 11m 52s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6d5e621bce14df166b47fd7b63339fae4e573761","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-11-18 09:47:51.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5ad042cc3ae9d1a1843e2739cd6c98913a229862","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-18 11:09:31.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bb4272bbfa7b4dbda472722b8a2cc25e\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/6aa4ddb00b18423bae17ae8ac3ad7863 : SUCCESS in 2m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/024e8089177c4b04a6bf69a3056afadc : SUCCESS in 55m 56s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/dd9ec923efd8462a957096878b802ccb : SUCCESS in 35m 37s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4d1b19de35214e60b1ec49c44689da37 : SUCCESS in 1h 32m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e1468a52fbcf43069515e056521e943e : SUCCESS in 59m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3da3b6136a7344cda1464a32f3bfbd02 : SUCCESS in 4m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ee8467661bfa4bfeb715a4831921077e : FAILURE in 5m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7f1ba3b4906745958cb98985919be6b0 : FAILURE in 5m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb8c61a5bcf249f7afcc2fd4044f9084 : SUCCESS in 6m 08s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5a12712fccf8438bbf69e3e3d200e08b : FAILURE in 4m 22s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/155e1fa7402345d1923c489a70de2a0b : FAILURE in 4m 19s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/602a5b8fde3944a1aff45c71cee4119e : FAILURE in 7m 40s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/169522f568c74678a7856573e51b4da8 : SUCCESS in 1h 37m 07s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/baf723a8f27d444c9392ee94696943fc : SUCCESS in 1h 54m 22s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/4d5b0fb05bf54286bb2c7584af9d09eb : SUCCESS in 2h 04m 28s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/894cba18d979432bb6945534a5fbd3d9 : SUCCESS in 1h 42m 25s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/1615f8c86b1b4b4cbc62d5f3e7b2f6e6 : SUCCESS in 2h 08m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/852a6fd623ed46458aa07d819bbf51d5 : SUCCESS in 44m 44s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/b5bbac3927ea443ebcd266f055831de1 : SUCCESS in 1h 45m 06s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/089b9cf3d44d4630aa41bb4585ad2c56 : SUCCESS in 1h 48m 49s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/0ea1376e08e9474fa14630cdf3848f89 : SUCCESS in 48m 39s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/e05c0ac08591448e89fbf2d2fcf68086 : SUCCESS in 29m 14s","accounts_in_message":[],"_revision_number":2},{"id":"6af7b57ace3dad4b190e882fe822f0003dc6e913","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-11-19 19:41:59.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"86ac65baed398648e8461c19446cc4d76416e75d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-11-19 20:28:53.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"6ce6d9c15ef814a3340fa2739929d8df8338bf75","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-11-19 20:30:35.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"badb411f8d10e1ae1fa893430fd1a09cf006d319","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-11-19 20:31:04.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"f46e73cb4cdcafe2ae7b748f3916ffe4223b7e67","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-19 21:44:29.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5054ee0038ef4b97a75d458382363ac1\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c81bf796c85e428f817cb5c928f6503b : SUCCESS in 11m 51s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2508939139ca4303ac1eabc29aec9d02 : SUCCESS in 12m 14s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"26eb155262a9ec0655b23427726d645138662ae2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-19 22:57:38.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/545cee78dd954986a9d2e2fce3b2b327\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/5155a36a1e2c4a05916eb4998dbfe9e6 : SUCCESS in 2m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/2ca20955ffe64934a9ba8d2128007b51 : SUCCESS in 54m 40s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/dbed522540674a4fba97d99d048f97ed : SUCCESS in 1h 11m 16s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3fee6f27146b4c669c77a8bde5200069 : SUCCESS in 1h 18m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bb77977eaa8041b6a750bbb692cc415c : SUCCESS in 1h 06m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd701149e50c432d86814b32b18ac227 : SUCCESS in 4m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0d082767ac64407c9675f1b49ee81786 : SUCCESS in 4m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7f284fca70664393aeb5b5a8a5b81167 : SUCCESS in 6m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aabca6da80904f2ba6462a7a3c54f3db : SUCCESS in 6m 40s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/dc85e3bc90d6473ca8afcd1ee9a9be32 : SUCCESS in 8m 22s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/70ec0c3666e9413d8be24514f5a30cf1 : SUCCESS in 8m 19s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/daef99df8a73424193ef5b220cdd496e : SUCCESS in 12m 05s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/4ee9b7d48fae4f598c052c24e13fe50a : SUCCESS in 1h 21m 51s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/97cb4ec4085f43d3bde37623157b5d71 : SUCCESS in 1h 10m 40s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/89fa967e150743079a0605eb13aec7e5 : SUCCESS in 1h 40m 30s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/b655868abdfd4790b0f07ce062a16d86 : SUCCESS in 2h 00m 54s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/edc1e5530ca149569063ee83e10de696 : SUCCESS in 2h 09m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a09a401effe8421694d724cd39357a5d : SUCCESS in 49m 41s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/75f5104e3acf4422b1cd0d461ac96737 : SUCCESS in 1h 48m 42s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/af9ed2468d1942858f800c22a84c907a : SUCCESS in 2h 09m 01s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/8f3dce21c238440683dae26f84f7dff1 : SUCCESS in 1h 15m 48s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/da689b662f44425386c9d8c643b5afc6 : SUCCESS in 24m 06s","accounts_in_message":[],"_revision_number":5},{"id":"434dc8c37441db8e710a4ffc2e0e3d6b4d8265dd","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-11-25 06:22:53.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"018bb11d9b7407a830cecf5fbb63e7840e0d2e2e","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-11-26 15:51:06.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"bbad97681e5f10b30436d04ec1dc3cf2a09ce97a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-26 15:51:51.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"d41143b54446c69d5d0ae4b396a6c5a735523a2e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-26 19:32:22.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cdda85c19af44eb4b28873e6c13ebf0b\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/342ed3ddc64a418b8cdc61eeb7e3275d : SUCCESS in 4m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/3a19b9c25d5f41f4b32a89d04466dc1f : SUCCESS in 1h 06m 35s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/962b7c132ce843afb134905cd89c24ec : SUCCESS in 55m 32s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ce5acd5cbcae4f87951f3e9e21bb2c1d : SUCCESS in 1h 25m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/38d10eaa96f44afcafc7db2137aa5ad0 : SUCCESS in 42m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93675d5609c9496eb08bd1bb772f87e0 : SUCCESS in 4m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5558e8605ff7431db0f200e3d7cd7971 : SUCCESS in 3m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/90b58a84f7fa407b8b9b465b92d1803d : SUCCESS in 5m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c728e4b45cf4fe29b7e59937af69f6c : SUCCESS in 7m 00s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f10c7dce3a5d4dbf9a9572fde4f1437a : SUCCESS in 8m 55s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2ef8d116d5ba45f88be3f5ed34000f3d : SUCCESS in 8m 44s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/62c09e94c82741c3a56662a8b30b196f : SUCCESS in 14m 06s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/2fee6f4afa4c4bbb9a56e62ae58cb48e : SUCCESS in 1h 11m 26s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/15de5b4956b9452abd10f40dd19e4239 : SUCCESS in 1h 14m 53s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/5ee6989fe51040caa29ca7da481681ef : SUCCESS in 1h 11m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/374eeb953d5f480a8fe51466e7f8e21f : SUCCESS in 40m 47s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ea1637b68952475ca612559fd9161c11 : SUCCESS in 1h 48m 08s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/01fe805b51d3401da7d8f2c1624a5aa6 : SUCCESS in 33m 38s","accounts_in_message":[],"_revision_number":5},{"id":"0592f7a60ea9c9a75b8bc142f823df91f3e94ac4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-26 19:32:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"d91782ac3b51771e7f3bef62c5ea6628898e4c55","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-26 19:33:37.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e036736a1aac4342a1dfa9d784f4bb99\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aceceaf7cef04fd7a65909f9312b4480 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"7eb5e2948cf6674e5627bbf4a91b52ee4e6942d5","revisions":{"8338020d5d3efd04a8bf1d817995f714057b9136":{"kind":"REWORK","_number":1,"created":"2024-11-11 05:40:15.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/62/934562/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/62/934562/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/62/934562/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/62/934562/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/62/934562/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/62/934562/1"}}},"commit":{"parents":[{"commit":"c7d70ae763667ff772cc4fb13706076641f4fd31","subject":"Merge \"Imported Translations from Zanata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c7d70ae763667ff772cc4fb13706076641f4fd31"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-11-11 05:37:59.000000000","tz":-480},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-11-11 05:37:59.000000000","tz":-480},"subject":"Remove default override for config options policy_file","message":"Remove default override for config options policy_file\n\nolso.policy 4.5.0[1] changed the config options policy_file\ndefault value to \u0027policy.yaml\u0027, which means it is changed\nfor all the OpenStack services and they do not need to\noverride the default anymore.\n\nNOTE: There is no change in behaviour here, oslo.policy provides\nthe same configuration that services have overridden till now.\n\n[1] https://review.opendev.org/c/openstack/releases/+/934012\n[2] https://review.opendev.org/c/openstack/requirements/+/934295\n\nChange-Id: Ia48842d0e0254bc2185716a3199ec7f810374f5d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8338020d5d3efd04a8bf1d817995f714057b9136"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8338020d5d3efd04a8bf1d817995f714057b9136"}]},"branch":"refs/heads/master"},"16d2411ac86ed4158d3fe0bc11edd0e52c819920":{"kind":"REWORK","_number":2,"created":"2024-11-18 08:55:03.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/62/934562/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/62/934562/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/62/934562/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/62/934562/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/62/934562/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/62/934562/2"}}},"commit":{"parents":[{"commit":"5167b21c60f9a3b670b286828373e64a220a104b","subject":"Merge \"Remove workaround for eventlet bug #592\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5167b21c60f9a3b670b286828373e64a220a104b"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-11-11 05:37:59.000000000","tz":-480},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-18 07:05:16.000000000","tz":540},"subject":"Remove default override for config options policy_file","message":"Remove default override for config options policy_file\n\nolso.policy 4.5.0[1] changed the config options policy_file\ndefault value to \u0027policy.yaml\u0027, which means it is changed\nfor all the OpenStack services and they do not need to\noverride the default anymore.\n\nNOTE: There is no change in behaviour here, oslo.policy provides\nthe same configuration that services have overridden till now.\n\n[1] https://review.opendev.org/c/openstack/releases/+/934012\n[2] https://review.opendev.org/c/openstack/requirements/+/934295\n\nChange-Id: Ia48842d0e0254bc2185716a3199ec7f810374f5d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/16d2411ac86ed4158d3fe0bc11edd0e52c819920"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/16d2411ac86ed4158d3fe0bc11edd0e52c819920"}]},"branch":"refs/heads/master"},"67ab2775c0072b5384107db282c362dcb42f4db7":{"kind":"REWORK","_number":3,"created":"2024-11-19 19:41:59.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/62/934562/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/62/934562/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/62/934562/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/62/934562/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/62/934562/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/62/934562/3"}}},"commit":{"parents":[{"commit":"5167b21c60f9a3b670b286828373e64a220a104b","subject":"Merge \"Remove workaround for eventlet bug #592\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5167b21c60f9a3b670b286828373e64a220a104b"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-11-11 05:37:59.000000000","tz":-480},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2024-11-19 19:41:59.000000000","tz":0},"subject":"Remove default override for config options policy_file","message":"Remove default override for config options policy_file\n\nolso.policy 4.5.0[1] changed the config options policy_file\ndefault value to \u0027policy.yaml\u0027, which means it is changed\nfor all the OpenStack services and they do not need to\noverride the default anymore.\n\nNOTE: There is no change in behaviour here, oslo.policy provides\nthe same configuration that services have overridden till now.\n\n[1] https://review.opendev.org/c/openstack/releases/+/934012\n[2] https://review.opendev.org/c/openstack/requirements/+/934295\n\nChange-Id: Ia48842d0e0254bc2185716a3199ec7f810374f5d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/67ab2775c0072b5384107db282c362dcb42f4db7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/67ab2775c0072b5384107db282c362dcb42f4db7"}]},"branch":"refs/heads/master"},"61f6bcdfec89fe461191319f98ec150926968266":{"kind":"REWORK","_number":4,"created":"2024-11-19 20:28:53.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/62/934562/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/62/934562/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/62/934562/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/62/934562/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/62/934562/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/62/934562/4"}}},"commit":{"parents":[{"commit":"5167b21c60f9a3b670b286828373e64a220a104b","subject":"Merge \"Remove workaround for eventlet bug #592\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5167b21c60f9a3b670b286828373e64a220a104b"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-11-11 05:37:59.000000000","tz":-480},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-11-19 20:28:44.000000000","tz":-480},"subject":"Remove default override for config options policy_file","message":"Remove default override for config options policy_file\n\nolso.policy 4.5.0[1] changed the config options policy_file\ndefault value to \u0027policy.yaml\u0027, which means it is changed\nfor all the OpenStack services and they do not need to\noverride the default anymore.\n\nNOTE: There is no change in behaviour here, oslo.policy provides\nthe same configuration that services have overridden till now.\n\n[1] https://review.opendev.org/c/openstack/releases/+/934012\n[2] https://review.opendev.org/c/openstack/requirements/+/934295\n\nChange-Id: Ia48842d0e0254bc2185716a3199ec7f810374f5d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/61f6bcdfec89fe461191319f98ec150926968266"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/61f6bcdfec89fe461191319f98ec150926968266"}]},"branch":"refs/heads/master"},"7eb5e2948cf6674e5627bbf4a91b52ee4e6942d5":{"kind":"REWORK","_number":5,"created":"2024-11-19 20:30:35.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/62/934562/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/62/934562/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/62/934562/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/62/934562/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/62/934562/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/62/934562/5"}}},"commit":{"parents":[{"commit":"5167b21c60f9a3b670b286828373e64a220a104b","subject":"Merge \"Remove workaround for eventlet bug #592\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5167b21c60f9a3b670b286828373e64a220a104b"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-11-11 05:37:59.000000000","tz":-480},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-11-19 20:30:23.000000000","tz":-480},"subject":"Remove default override for config options policy_file","message":"Remove default override for config options policy_file\n\nolso.policy 4.5.0[1] changed the config options policy_file\ndefault value to \u0027policy.yaml\u0027, which means it is changed\nfor all the OpenStack services and they do not need to\noverride the default anymore.\n\nNOTE: There is no change in behaviour here, oslo.policy provides\nthe same configuration that services have overridden till now.\n\n[1] https://review.opendev.org/c/openstack/releases/+/934012\n[2] https://review.opendev.org/c/openstack/requirements/+/934295\n\nChange-Id: Ia48842d0e0254bc2185716a3199ec7f810374f5d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7eb5e2948cf6674e5627bbf4a91b52ee4e6942d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7eb5e2948cf6674e5627bbf4a91b52ee4e6942d5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
