)]}'
{"id":"openstack%2Fglance~934716","triplet_id":"openstack%2Fglance~master~I662ba19697e03917ca999920ea7be93a0b2a8296","project":"openstack/glance","branch":"master","topic":"bug/2086675","attention_set":{"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2025-01-11 03:08:26.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I662ba19697e03917ca999920ea7be93a0b2a8296","subject":"Fix performance glitch while sorting image locations","status":"MERGED","created":"2024-11-12 06:01:30.000000000","updated":"2025-01-11 03:08:26.000000000","submitted":"2025-01-10 15:41:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"934716-bug/2086675","meta_rev_id":"152b828a0160238170b3999c28ed09901e3aa27e","_number":934716,"virtual_id_number":934716,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-01-10 15:41:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"recommended":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"all":[{"value":1,"date":"2025-01-06 08:48:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":2,"date":"2025-01-10 15:01:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":2,"date":"2025-01-07 14:20:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":1,"date":"2025-01-10 13:32:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"all":[{"value":1,"date":"2025-01-06 08:48:58.000000000","_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}],"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-12 06:15:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-11-12 07:53:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-03 07:07:54.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"state":"CC"},{"updated":"2025-01-06 08:48:58.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"state":"REVIEWER"},{"updated":"2025-01-07 14:20:20.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2025-01-10 15:01:24.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2025-01-10 15:58:44.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"}],"messages":[{"id":"2b260fdb4ac8295bdadfe442e3e8eb4931b4d4d2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-11-12 06:01:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f7aeba97269f16f2502c85d6194197f7b91cf005","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-12 06:15:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/062585dc0fb743d0bf78da03a2c427c7\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/416ce6d0a46848b39b9cc1ca05ae0aae : FAILURE in 10m 03s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ac95a94a97634da5be8ef4a380c6e0da : FAILURE in 12m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9409a91eeb12ac9e444d452954f210fa99427a8d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-12 07:53:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e46e9a033fa84c558e3d750844d2bea5\n\n- grenade https://zuul.opendev.org/t/openstack/build/fc8aefff9f1e48978f84f39eb5f24efb : POST_FAILURE in 52m 15s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a26da6203c2c4410ae7988383facc32f : POST_FAILURE in 45m 15s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6c8fbd0261bb4c2a874b77d3b82b9cb3 : POST_FAILURE in 1h 37m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a4bc58ee381e461a83ffdeeb24c27a18 : POST_FAILURE in 53m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0f4c1d3b68947a2971c2be962aa3fb0 : SUCCESS in 4m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/caee96191f6444778d9ba0069eb02ef3 : FAILURE in 5m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6e210a8f812c4bed8d770f5d8f79c9f6 : FAILURE in 3m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cdba428938aa460bb7ebb79619623be3 : SUCCESS in 3m 52s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e99bbb3edfbc4295bdc77c147200398f : FAILURE in 9m 49s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9024a65ed6e742828b8d5490272483c8 : FAILURE in 9m 54s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/87cf14b4d88d4c5083c9691d58bfded4 : FAILURE in 13m 35s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/6d215dd6447b4a5e9b10c4907f11989d : SUCCESS in 1h 26m 10s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/6b1dc398099b453c80535463d68462a6 : SUCCESS in 1h 44m 28s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/9aec0a3c610645b5b9caea8c5b535845 : SUCCESS in 1h 42m 11s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/acd7238300864b45a545ef2778fac164 : POST_FAILURE in 1h 40m 55s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/34a0f8c78fc34dc48d2de712d0b8cb2b : FAILURE in 19m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c8b081cb87df4ab68648835f1bbdc19f : SUCCESS in 38m 22s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e817d1c414c0427f878f732bc01e8e64 : FAILURE in 17m 15s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/2738fcb49c1b49f38ea5306728eb6db7 : SUCCESS in 1h 31m 30s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/ef400b7a19d64709b6185a0786711581 : SUCCESS in 34m 46s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/2bc4a976cf2c4ef9adcdbaa259c7a9ee : SUCCESS in 26m 10s","accounts_in_message":[],"_revision_number":1},{"id":"49b51f26c83255fb7b675e021c0cc1967cf067a7","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-11-12 08:01:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dcf22624944e2483dcbe5480819ec0e8f1505336","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2025-01-03 07:07:54.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6a1d4f8917631961e2b723649d0b096c8d0a2ee0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-03 07:20:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c1f22601a78b41a5a2cf8ed1d090a931\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/69cfe581895046fe8992d3e0cf8a41fd : SUCCESS in 8m 26s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d0eb39f7f4784e4fbf394ca9f05ef773 : SUCCESS in 10m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f20725242a2301effa551c260a6effcdaf66d847","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-03 09:08:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c79a5cd5e27b45ff9289e5a2edd2d37c\n\n- grenade https://zuul.opendev.org/t/openstack/build/9d041bf523f44648bc23c41026701006 : SUCCESS in 1h 06m 36s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/627074cdb6a042abad8623116ceaf0b4 : SUCCESS in 57m 17s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/092ddbd75f054e2082ecaa349d22449c : SUCCESS in 1h 32m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ebaf4c3ac4414b4aaac8ae3d43666b5b : SUCCESS in 1h 12m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1623db22f22341c193bd70dd343fb62c : SUCCESS in 4m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0204158b6bc444c287fbf8e4b3e6520f : SUCCESS in 4m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fe30bcf4a1094321a70833df0c75f4a5 : SUCCESS in 6m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ef7360b5f69476b85472e57cd562184 : SUCCESS in 5m 27s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/57d50216b0c34302a25633fbbe94d224 : SUCCESS in 8m 09s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e1ef4ac0e9b94a8c9863c0847a2363d9 : SUCCESS in 7m 05s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/7604964b47c3473cb8c55aefd46c79ee : SUCCESS in 9m 03s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/556acbe1014d44a7963c499fa469c599 : SUCCESS in 1h 33m 14s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/2c0d936f511b433fb36a7c46a33ed067 : SUCCESS in 1h 55m 02s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/cc691c29f5e045688e32c3999fca32ca : SUCCESS in 1h 12m 55s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/633eff88f68c45659f5ed8dfbf674993 : SUCCESS in 1h 30m 27s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/67d0aad9fb274226b6e6cb9c8674bfdb : SUCCESS in 1h 44m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/477453361cb84342a5c732194b54c1c2 : SUCCESS in 38m 01s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/5b399ec36717418a8b835650d1b9b080 : SUCCESS in 1h 36m 48s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/07bc09142e674e018f14e64591acc81a : SUCCESS in 1h 14m 49s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/ccbeaf9fc4cd4f68b7ba9691aab3c264 : SUCCESS in 43m 33s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/e3b9b1e3335641a4be7be5bb0f330c76 : SUCCESS in 1h 18m 51s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/762a2269d9b74f759030bde695fd6bfa : SUCCESS in 39m 25s","accounts_in_message":[],"_revision_number":1},{"id":"56484ee773cc3f0e9c456b6bdb6974f1e4602e76","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2025-01-06 08:48:58.000000000","message":"Patch Set 1: Code-Review+1 Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d25f35bcbd96c9076536eee8038889ce6061f780","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-01-07 14:20:20.000000000","message":"Patch Set 1: Code-Review+2\n\nThanks","accounts_in_message":[],"_revision_number":1},{"id":"91c1d5b1467e4f88a8f074a0c1c19b89f6792069","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-01-10 13:32:25.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0d17bdf175aea44b8e42b5435314a2d015b3e9c5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-10 13:33:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9009d0fae68d5c0e9cd26ec730f988550cc57285","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2025-01-10 15:01:24.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c327763256f8bc67112731e9147f817f27a1da13","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-10 15:41:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/981d53bb003a4fcd9ebfed7bc78f092e\n\n- grenade https://zuul.opendev.org/t/openstack/build/c11a599f8089465db1ee7eb5d6e5f0ae : SUCCESS in 1h 02m 09s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/27bd77c93aed4e6893c3e3b88701a104 : SUCCESS in 34m 08s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/90116312c6a540ba8a441cbc274b1336 : SUCCESS in 1h 31m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/34eb8e0c3e8d49b68097a1fbf2eaab43 : SUCCESS in 1h 00m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9005e31bfe5c4026bfdabd48cff5e3c6 : SUCCESS in 5m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c1f96ed204bc4dc8ad7f5cf813e2d61e : SUCCESS in 5m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8593ad9d5f884add948bcafbeafa134c : SUCCESS in 5m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bdeb4b6b7b434bc4b1072ad47161526f : SUCCESS in 5m 15s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/30174b8981c944dc9a240f7f3d620010 : SUCCESS in 8m 09s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/06f056e0abca41e8992e3338dd2cf972 : SUCCESS in 8m 14s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/58341ebf298e407cbcba0d40f0705264 : SUCCESS in 12m 09s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/f21bc3bc9f1f43db8d4f03998535f584 : SUCCESS in 2h 00m 41s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/5a7d9879cd8a45e9bb4fef0608e03222 : SUCCESS in 2h 02m 35s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/a01147eef97b43a38b9da0ab956fed65 : SUCCESS in 2h 01m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/38cd287121a543fc830e33a56e835eae : SUCCESS in 1h 00m 44s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/93f86fe186d3474a99f6abf2cfc11f7d : SUCCESS in 2h 00m 45s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/fc23499573774550aefc02f1c0965a61 : SUCCESS in 38m 36s","accounts_in_message":[],"_revision_number":1},{"id":"939653860f2134eadf4369be7af4776a8294a2da","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-10 15:41:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"dfd80684ebb4e908f1ca8b4f4b976d49e7fc4dda","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-10 15:42:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/11361894a6f845d09d15197acb6cee18\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98990f4721614b9495202132cf0b6c4f : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1},{"id":"47399a2eb47f0a601c1bf6fee95f313f20cf3c28","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-01-10 15:58:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"45dcae2cf177f4194dbbac46221c86cc7cb2d4d6","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2025-01-10 16:48:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"152b828a0160238170b3999c28ed09901e3aa27e","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-01-11 03:08:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"45202b1e044f52aa863ed135564cc1fc74b43145","revisions":{"45202b1e044f52aa863ed135564cc1fc74b43145":{"kind":"REWORK","_number":1,"created":"2024-11-12 06:01:30.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/16/934716/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/16/934716/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/16/934716/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/16/934716/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/16/934716/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/16/934716/1"}}},"commit":{"parents":[{"commit":"cfa36c1b55ff6422430be07fba1e851ac1fe50d8","subject":"Merge \"Change DB migration constant to 2025_1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/cfa36c1b55ff6422430be07fba1e851ac1fe50d8"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-11-12 05:49:23.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-11-12 05:49:23.000000000","tz":0},"subject":"Fix performance glitch while sorting image locations","message":"Fix performance glitch while sorting image locations\n\nSome of the available glance stores like file, cinder etc has\ncapability to reuse already initiated driver (DRIVER_REUSABLE \u003d 0b01000000). In\nCaracal we have added a feature to sort image locations based on store\nweight. As RBD driver of glance does not have this reuse capability, during\nimage list API call it initializes the RBD driver for each of the available\nimage which is causing noticable delay in list call.\n\nTo avoid this, using new interface added in glance_store which will directly\nget the weight of the store from memory and return it back to user.\n\nDepends-On: https://review.opendev.org/c/openstack/glance_store/+/934362\nCloses-Bug: #2086675\nChange-Id: I662ba19697e03917ca999920ea7be93a0b2a8296\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/45202b1e044f52aa863ed135564cc1fc74b43145"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/45202b1e044f52aa863ed135564cc1fc74b43145"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
