)]}'
{"id":"openstack%2Fglance~934724","triplet_id":"openstack%2Fglance~master~Ica8c21aed4c88d85594fd242116cb92d5ce1484b","project":"openstack/glance","branch":"master","topic":"glance-epoxy-m1","attention_set":{},"removed_from_attention_set":{"19138":{"account":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"last_update":"2025-01-06 06:16:05.000000000","reason":"Change was abandoned"},"33765":{"account":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"last_update":"2025-01-06 06:16:05.000000000","reason":"Change was abandoned"},"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2024-11-15 06:47:00.000000000","reason":"\u003cGERRIT_ACCOUNT_9303\u003e replied on the change","reason_account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2024-11-14 21:19:16.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"Ica8c21aed4c88d85594fd242116cb92d5ce1484b","subject":"Refresh Glance example configs for epoxy m1","status":"ABANDONED","created":"2024-11-12 07:33:59.000000000","updated":"2025-01-06 06:16:05.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"3b7b227bee3f27e87597078b485ab6a0a42caae8","_number":934724,"virtual_id_number":934724,"owner":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:check","value":1,"date":"2024-11-14 08:54:47.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},{"value":-1,"date":"2024-11-15 06:47:00.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2024-11-14 21:19:16.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-12 07:34:59.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"state":"REVIEWER"},{"updated":"2024-11-12 07:34:59.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2024-11-12 07:34:59.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2024-11-12 07:46:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-11-12 09:49:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"91546495ad0afb75f16d33cfa69e15601666b96d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-11-12 07:33:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"caa27240c2f3470f0619c1100acf287998a2aab4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-12 07:46:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/209ff0ba4d9e4e9ab1695109fea52a1c\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/257b48edd557456887f3d7171de7c283 : SUCCESS in 8m 42s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/19cbc4c52d64415492232a0933fbedcb : POST_FAILURE in 10m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c5bb27d171f026c48d643edf2b15180fa3a49fe3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-12 09:49:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c62343d8f2054998a2345e8532f98dbf\n\n- grenade https://zuul.opendev.org/t/openstack/build/264e3807661043c7a460cf7d5599dca7 : SUCCESS in 38m 12s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/fe80b3db62d44ab89c5b2ae055eff359 : SUCCESS in 57m 29s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b8cb327676a94a96a0a0ecbcdc6d06d3 : SUCCESS in 1h 45m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6fd8db9f787546f09c654e072aba5f07 : SUCCESS in 1h 00m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c9394bc7bbc4cffbd6959f4337541a5 : POST_FAILURE in 6m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/46f335f482a34ef1a314bf8740415443 : POST_FAILURE in 6m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/85adffd528874aafb677cca430810cba : POST_FAILURE in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29e653a3437a4ded879e8fba75b314ee : POST_FAILURE in 6m 17s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/06d6ea45a7bb49c9a5b9b6f17b37407c : SUCCESS in 9m 02s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3d45b722d9df49678f5fcb397f82987a : SUCCESS in 8m 49s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/0ebea6f9104946acb2bbe386af177488 : POST_FAILURE in 9m 33s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/b322dcb854854fdba8f8a75c57eff09e : SUCCESS in 1h 47m 51s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/e2834a338e284ed4a46be032d44d20ea : SUCCESS in 1h 12m 44s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/ee7b9f11c71b42dd93c4a0667793a447 : SUCCESS in 1h 51m 10s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/6e1ec01a901d4b6380f94fe337189e90 : SUCCESS in 1h 36m 52s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/f808b4196cc041969373552a4b7bd7c8 : SUCCESS in 2h 13m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/af6b92825fac4e48953e3ec5c75519f9 : SUCCESS in 48m 18s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/662ea2aa437b47b1916ffd104b53cb20 : SUCCESS in 1h 50m 03s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/0b6acd21f82f4975a1aaf00487286bc3 : SUCCESS in 1h 42m 52s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/345bec27542e487bbbb42834c312b3b8 : SUCCESS in 45m 46s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/2d76f7de9a0542e28ca94cb661f5a4a4 : POST_FAILURE in 18m 20s","accounts_in_message":[],"_revision_number":1},{"id":"d05e25709e9ae42490b62446b9a04197e1d4e92a","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-11-14 06:26:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d948dea6419ff2fbbac5aef4a20632eac6f66d57","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-14 06:39:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eaa9904f27c74d1390c225e43a079df6\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/dbf9672766d24eaa8ab042eb46b13b39 : SUCCESS in 9m 02s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/546257996e3f4aefb703f066d603090d : SUCCESS in 10m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4b87620ff299506cdc7d4bfdfa0c5f1ed1d5f4d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-14 08:54:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c22511cba9534b0c9fafb6e38a7fb2bd\n\n- grenade https://zuul.opendev.org/t/openstack/build/2d226686fcee4eafaf05d485661b7763 : SUCCESS in 37m 30s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8afd907eeabe45b4aa97d694bd0a9ebd : SUCCESS in 37m 35s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/bf631fa1e02c48e29605e84c1f0de68d : SUCCESS in 1h 32m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d6f6335b77b043f9aad8814201f7ac2f : SUCCESS in 51m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7bf800afa76b40b0b14d9d0de56f0e9f : SUCCESS in 4m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/609ce39bf77845f690850b80fe9b324a : SUCCESS in 4m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/511cfe59ab2643d6ac325fdc1d7cecc7 : SUCCESS in 5m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e762cb51bdea4b7ebe1a176529b6a7e2 : SUCCESS in 6m 38s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e47338a915a64abb900c40d394d7f032 : SUCCESS in 8m 43s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c8823f961ddb4cfcb1704c687a7a514c : SUCCESS in 7m 26s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/7a020d301a564316ae3d9f74ddfe97a9 : SUCCESS in 12m 03s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/6dd34d6e6e0f4cf29384b41907b53d51 : SUCCESS in 1h 29m 08s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/e2a67b15ca3840e0af726752a650d181 : SUCCESS in 1h 56m 16s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/f054e2bf82c140c491a62b82152cc221 : SUCCESS in 1h 33m 15s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/2620e220a83c48f2870dfe0aca8338e8 : SUCCESS in 1h 36m 04s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/2ce392fcb0c74bdca41b5faf2eb4e099 : SUCCESS in 2h 25m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8769ba75b78b4a628610f416d816d569 : SUCCESS in 40m 03s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/6ff2549ad23441a1b8dae93639066a56 : SUCCESS in 1h 46m 26s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/6643b89493204195a217f8277c385a42 : SUCCESS in 2h 00m 57s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/e9b9e8c37456491bba7f911a388ba36a : SUCCESS in 48m 29s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/4dd1dcd5231d4a7b9abf2ec6ecdf60ed : SUCCESS in 32m 51s","accounts_in_message":[],"_revision_number":1},{"id":"6155edea83296639a09187b3462875ca363347c6","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-11-14 21:19:16.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d8626b3b1bee6fbd49cab25800e88be25359ecdb","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-11-15 06:47:00.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3b7b227bee3f27e87597078b485ab6a0a42caae8","tag":"autogenerated:gerrit:abandon","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2025-01-06 06:16:05.000000000","message":"Abandoned\n\nAbandoning, no longer needed since we skipped tagging m1.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c8e7203db6727428cbb3bf46195d849751c3318a","revisions":{"c8e7203db6727428cbb3bf46195d849751c3318a":{"kind":"REWORK","_number":1,"created":"2024-11-12 07:33:59.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/24/934724/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/24/934724/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/24/934724/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/24/934724/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/24/934724/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/24/934724/1"}}},"commit":{"parents":[{"commit":"4b90684fbb81adb8e86d38696f2e5db41a5bdaf4","subject":"Release notes for Epoxy Milestone 1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4b90684fbb81adb8e86d38696f2e5db41a5bdaf4"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-11-12 07:30:25.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-11-12 07:30:25.000000000","tz":0},"subject":"Refresh Glance example configs for epoxy m1","message":"Refresh Glance example configs for epoxy m1\n\nChange-Id: Ica8c21aed4c88d85594fd242116cb92d5ce1484b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c8e7203db6727428cbb3bf46195d849751c3318a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c8e7203db6727428cbb3bf46195d849751c3318a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
