)]}'
{"id":"openstack%2Fglance~935227","triplet_id":"openstack%2Fglance~master~I15d984b83a5be4a5aa9c00576377d0a515ccc2c5","project":"openstack/glance","branch":"master","topic":"eventlet-removal-functional-tests","attention_set":{},"removed_from_attention_set":{"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2025-03-18 13:40:19.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I15d984b83a5be4a5aa9c00576377d0a515ccc2c5","subject":"Functional tests migration: test_schemas.py","status":"MERGED","created":"2024-11-14 21:18:18.000000000","updated":"2025-03-18 15:25:51.000000000","submitted":"2025-03-18 15:24:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"935227-eventlet-removal-functional-tests","meta_rev_id":"4c43b4afdcebc002f8b0143482bf6211e74ee6a6","_number":935227,"virtual_id_number":935227,"owner":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-18 15:24:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2025-02-11 16:47:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2025-02-14 16:12:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2025-02-14 16:12:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}],"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-14 21:29:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-11-14 22:29:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-18 09:07:49.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"CC"},{"updated":"2024-12-06 13:02:57.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2025-01-20 10:49:22.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"CC"},{"updated":"2025-02-14 16:12:58.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"}],"messages":[{"id":"8e77b4284304b8411d45550e9a3338a60456c4ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-11-14 21:18:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"aac4a514b7fe1833d476011791647dc834f5ee90","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-14 21:29:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1929168221814f809a478059b0087de0\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f90142d8441b4a4fa764806cfc1912ee : SUCCESS in 7m 08s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/faec14580c9244529838155408b0c6c2 : SUCCESS in 9m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ee79cd84a2c52729db1f4ee78ed6a3d19b507dca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-14 22:29:27.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d1c4c06809144a16984e73e0e46c63f5\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1875a36895e0415a9ab1764b5f4e213f : SUCCESS in 1h 03m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8440a20a5d7e4a03a494d9406a6c1075 : SUCCESS in 1h 04m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c15b333737f4a23b04f772c546501ef : FAILURE in 4m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/103656bb25034c11884a20f86930ee5c : SUCCESS in 5m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a84de5c7a380467e95d4c72dded0fce6 : SUCCESS in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1611e5bd84b487c8dfdeffcf473ba7f : SUCCESS in 5m 29s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1787e8febdc1423cbcd833caadd91d12 : SUCCESS in 7m 42s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0cadd45ae2fe40ccb5dd558d7257715f : SUCCESS in 7m 14s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/1774846c9b7c410f9db88564b457e1cc : SUCCESS in 11m 46s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/055c28a96b6e47d184a017490fdef331 : SUCCESS in 33m 27s","accounts_in_message":[],"_revision_number":1},{"id":"306315044eb132a83e2a0a823881fb1125ec4bc6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-11-15 18:23:42.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"611fef8643a3b3f0cc2b7489da3efccd58133109","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 18:42:06.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/701b8b2796b14c11a9028489998de744\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a400163160ad48cea972152378e19a6e : SUCCESS in 7m 34s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5085db2343ec44a8ba7d378a43a69649 : SUCCESS in 9m 05s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0196edb0a1daa84ed6cf94036ebbbd73e7809306","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 19:20:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8436f304b8f6499096ab9b4569630eb7\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6da3627451e64d4fba9d5c8863a95926 : SUCCESS in 41m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8de26a39b1d84ecd9716c805d8281811 : SUCCESS in 47m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d07454fa4f1f499f8a873f92f6206d0a : SUCCESS in 4m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9b7ebe407b8d475f9e14274cf3ab6d3b : SUCCESS in 5m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f3a5d2461900407ca8e1dfff79b9453b : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84bf32f31b8f4cbb957ecd27be0f7ce1 : SUCCESS in 5m 22s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7c2e97cf569e4afc8533190427a3ee73 : SUCCESS in 7m 19s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f437a58143be4723a20e0e4a3e2ed1f5 : SUCCESS in 8m 30s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/ed09817afbcf40498813cdb79a1df70e : SUCCESS in 9m 56s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/3943bded90bb41cf92f396c95b1b428a : SUCCESS in 35m 48s","accounts_in_message":[],"_revision_number":2},{"id":"3edbe2692b18d3c971d767a5615f05ec2bed7a9f","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-11-18 09:07:49.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e75574cc96aedfac7a2259eebaa6183a013a74d7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-12-05 13:59:17.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"3380a346f18ad4fefe64e94d90fcec8f77998f14","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-05 14:14:01.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a7c88d77b7f8499fa916b7ae9fd18b03\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b665189e240d4806a7503e3f82358f13 : SUCCESS in 10m 56s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d678c5d12a844fbcb9e96b2ecb1d3248 : SUCCESS in 11m 58s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e595600fc8904a8ecc5684dea607ee1adff27f94","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-05 15:03:15.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5f4b32fa20e845568d6e5f010dad12cb\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e501078c2f954a74b50f5778b304dc08 : SUCCESS in 58m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4ca67c82b2004339a32f20e4fc84afe0 : SUCCESS in 1h 02m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd64e153d318481fbbbb866417588281 : SUCCESS in 4m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1cb12a7a05d8420fb0f0d62968ed7cb6 : SUCCESS in 4m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2b50289ecbb14b51b6ec9c9f4fed3584 : SUCCESS in 7m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/143b8d16fe754c36820cc617a6a86b71 : SUCCESS in 6m 34s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/949caa3af1234737949c56d2eba6aab1 : SUCCESS in 10m 52s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/63db09956ec84b7bb23fd0a444b1974b : SUCCESS in 9m 15s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/6f361d8687024a0299fb177c49beb1f8 : SUCCESS in 13m 40s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/16c08c2b59de4ac7bb5110050ee3697d : SUCCESS in 33m 32s","accounts_in_message":[],"_revision_number":3},{"id":"d195dea79bdcb307e0959e1cc9eab8d447774a29","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-12-06 13:02:57.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"908d538acd11ca11ee6d6acaad81276c6cadf417","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2025-01-20 10:49:22.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"98b952de8e6881aa3cd35448fdb9e48f45430b58","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-02-05 19:37:30.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"9cff76c02fa66ad2351e953f76b04d50c472f0a6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-05 19:52:22.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/45d83750d6cd488a9619937b080c2daa\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/adc5764c377241f4bcec9b2b4212405f : SUCCESS in 10m 31s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/fe9662accc6343889a24693fad59cda1 : SUCCESS in 12m 34s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"2fd67f201bf6b989ac070bc4f68a51157a17d334","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-05 21:01:44.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/beca9dfcacf8485fb9cacef171067384\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/aef51b3ea78e453bb958f2c873082839 : SUCCESS in 1h 17m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/261e79915b9f4b05a56cc0bb053531c2 : SUCCESS in 1h 07m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2de3e3451547427caae5a34dcd3a2235 : FAILURE in 5m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/50d1a96e5a0a4129a3a3983dca785171 : SUCCESS in 4m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6ad47e8ef8694f3c8c430e45fdf7813a : SUCCESS in 5m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4700427946aa4370a9a78f3ea1ee09fb : SUCCESS in 8m 52s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/dd11b99d43b840f5a82e6d736f8133ae : SUCCESS in 8m 25s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5e041d87849e4d8abf0b5f488c71df9a : SUCCESS in 8m 01s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/9c8558f02c9f45bc9cec08d14c78ad80 : SUCCESS in 11m 26s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/b15befa14b674988bd809567553990e9 : SUCCESS in 35m 12s\n\nWarning:\n  Comments left for invalid file glance/tests/functional/db/base.py","accounts_in_message":[],"_revision_number":4},{"id":"b9042a5b231ce254a95138c1e0a5be2d5abc208a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-02-06 16:12:30.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"acb7a1dd76afbdc53b19bd50879b41758cbd492e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-06 17:12:22.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/97233b09b08a44c38cce9609d0f7218d\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b9629d7536f94c0785bd2a9da824d927 : SUCCESS in 58m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a48f69d9d5ef4624947b0b52ff7cfb09 : SUCCESS in 41m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/408c56b94d26474794444789494d43a4 : SUCCESS in 4m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e91d6c1a2242404a8dc1683eca7a0b2d : SUCCESS in 5m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/110d7bcd0d014ec7b28123f559179ed5 : SUCCESS in 6m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8db68a5f5e1a4c6b9ffd1f35e115cf3b : SUCCESS in 5m 28s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b3b8c77e99cd4fdc9856f787d9fb5715 : SUCCESS in 10m 15s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/667de8bbe86a4086b586f7608b639589 : SUCCESS in 7m 12s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/d152c1f897554bd1808c7443cd5d3887 : SUCCESS in 12m 37s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/a2de526d9b724428a5a0b45341e8d679 : SUCCESS in 33m 44s","accounts_in_message":[],"_revision_number":5},{"id":"7cff18de91d106fa5f9a7e238ed820826eb4d91b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-06 17:37:46.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c01ea4e9916d48bb80053a3c8ede1726\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/759d5d8564d54318a4512eb3f209cad8 : SUCCESS in 14m 44s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/749ec1bc12614275825a12c004bdf494 : SUCCESS in 17m 01s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"9fc385deec9a80af63314cb6b193574e74054788","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-02-11 16:47:40.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"29ecb53c2e9972189e65fcc63a02fade8c65281f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 17:00:20.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0a29969feec9439aa44802d962011efe\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fe309e1e54184727a35517b2a8c45860 : SUCCESS in 9m 01s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4ad5fec394c04e038540a4b814331173 : SUCCESS in 10m 44s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"e775daf7f7d00799b837191a5c827b3d6bdaca84","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 17:55:15.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e4e35ce7e9be42e38d9ae4d63b89d23d\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/bd4c346dddb74d47addd0a8bd47b42b0 : SUCCESS in 1h 01m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6edc400a93ae44239eddc639e4634196 : SUCCESS in 56m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce604736d71a4939b0b52cdc0cc1a647 : SUCCESS in 3m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/90db3753b09e439383218e94515fa3ca : SUCCESS in 4m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6c0213842d7e42388c63147767609137 : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e99f284148a5411d9a9b9ef535480e39 : SUCCESS in 5m 30s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/cd062fce9a044493ba39680eaf0df567 : SUCCESS in 6m 34s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0c3ef1c418da473db79b65335865454f : SUCCESS in 8m 34s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/fd9d90e537a9419198cf7440f83e9583 : SUCCESS in 14m 50s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/935d6dff5be34d9e9a277dd68dad5d2d : SUCCESS in 51m 40s","accounts_in_message":[],"_revision_number":6},{"id":"dc5a6709ec0c2fa1121f2ca5ab7a972f817278e1","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2025-02-14 16:12:58.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"1636a3f0698bb14c6a4937d737a09485c943f880","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-14 16:13:48.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"518a48713a7419c2a53426a66ad7973af132c975","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-14 17:30:47.000000000","message":"Patch Set 6: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/937181 is needed.","accounts_in_message":[],"_revision_number":6},{"id":"3ee346b828f50c0067b21101ef4b5d621510f18b","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-03-18 13:40:19.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"1ac95edf448011442165e7edceaedb292b380a5c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-18 13:46:47.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e501e123ca094f748cde5a0f234ecb4f\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4e2d175734f14d65916d089ff51e80a1 : SUCCESS in 6m 07s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3467f20a92044e47a7459fee6ecb9360 : SUCCESS in 5m 42s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"ea16fce47bc6e50661f04f7448e05a67b651af6a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-18 14:18:34.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/35ded9925fd944428ea51eadfda18f5f\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0b32333d99ef4b9c9e597d150774e88c : SUCCESS in 28m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/068678be96d246388b304fd2389eedd4 : SUCCESS in 36m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb5bfa28e9db49c7973d3d1ee424062c : SUCCESS in 2m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0cd8cb19f4ac49429e731f3264a00f6b : SUCCESS in 3m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/20ac8f2fbc3341a6bce221ebad991093 : SUCCESS in 5m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8a8cff56cf649f1b7bd529b9158b75c : SUCCESS in 5m 52s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e600d23df93346b3a6a487b3dcb195ac : SUCCESS in 8m 04s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3ad633097b7a4f9ca1bfa390d8bc609c : SUCCESS in 7m 41s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/31ab6a7191ad4c4bbf91bbc42ddb4fa6 : SUCCESS in 13m 02s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/375cb7dea5934c92928cecbb08cb7d3a : SUCCESS in 29m 32s","accounts_in_message":[],"_revision_number":6},{"id":"43ee2c6b39e799ea5a7d1a4042e42f65d89ffa4c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-18 14:19:10.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"ab79ac4edd08331daf5f58883b93f3295695c02e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-18 15:24:41.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4a3ba85158d54884b20b5905add24f2d\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0167ceac92be456c889c247cb5ec821f : SUCCESS in 53m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/44930f2232914aab9d000950afbf961f : SUCCESS in 1h 05m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f359033c988240ef9ea32f4959142b2c : SUCCESS in 4m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/63effa69164645489a7885de591a0661 : SUCCESS in 2m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ea142fb860194821aa05744515afceb5 : SUCCESS in 5m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dca90c03690a40a290e46327a100759b : SUCCESS in 6m 00s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b16e2da751f546f1aadafd6d80fd71ec : SUCCESS in 8m 38s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/add3d78b7ffe4bc68bd1118db8a44579 : SUCCESS in 8m 45s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/2477635d5fac49fab55fd778ea9fa0eb : SUCCESS in 11m 16s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/6a7bac89bb1d427b80f94e4508f821ae : SUCCESS in 14m 09s","accounts_in_message":[],"_revision_number":6},{"id":"726521d2ea313a77fe0e0c97741633ce3b298f61","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-18 15:24:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"4c43b4afdcebc002f8b0143482bf6211e74ee6a6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-18 15:25:51.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d595cfe1688f4faea45bd2d760bc62c0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1be31c11c621498db9ee283eb57df9ce : SUCCESS in 52s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"7194b038b9115a1f420db275e8d25a6e25b143a5","revisions":{"6262dd9d57ae009c929f4bca92da968c0cab90a5":{"kind":"REWORK","_number":1,"created":"2024-11-14 21:18:18.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/27/935227/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/27/935227/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/935227/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/935227/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/935227/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/27/935227/1"}}},"commit":{"parents":[{"commit":"c7d70ae763667ff772cc4fb13706076641f4fd31","subject":"Merge \"Imported Translations from Zanata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c7d70ae763667ff772cc4fb13706076641f4fd31"}]}],"author":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2024-11-14 01:15:42.000000000","tz":60},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2024-11-14 21:11:24.000000000","tz":60},"subject":"[WIP] Functional tests: do not really start a server","message":"[WIP] Functional tests: do not really start a server\n\nCurrently, most of the functional tests need to run Glance in a separate\nprocess. This is always done using eventlet (not wsgi). Using the\nSynchronousAPIBase allows us to no longer depend on eventlet.\n\nThis patch is a proof of concept that only fixes *1* test. I believe we\nneed to fix 387 tests, from the following files:\n\nglance/tests/functional/db/base.py:190\nglance/tests/functional/v2/test_images.py:62\nglance/tests/functional/db/base_metadef.py:38\nglance/tests/functional/db/test_sqlalchemy.py:25\nglance/tests/functional/test_api.py:9\nglance/tests/functional/serial/test_scrubber.py:9\nglance/tests/functional/test_client_exceptions.py:7\nglance/tests/functional/test_glance_manage.py:6\nglance/tests/functional/test_client_redirects.py:6\nglance/tests/functional/test_cors_middleware.py:4\nglance/tests/functional/test_cache_middleware.py:4\nglance/tests/functional/db/test_migrations.py:4\nglance/tests/functional/v2/test_metadef_namespaces.py:3\nglance/tests/functional/test_logging.py:3\nglance/tests/functional/v2/test_tasks.py:2\nglance/tests/functional/v2/test_metadef_tags.py:2\nglance/tests/functional/v2/test_metadef_resourcetypes.py:2\nglance/tests/functional/v2/test_metadef_properties.py:2\nglance/tests/functional/v2/test_metadef_objects.py:2\nglance/tests/functional/test_wsgi.py:2\nglance/tests/functional/test_healthcheck_middleware.py:2\nglance/tests/functional/test_sqlite.py:1\nglance/tests/functional/test_reload.py:1\nglance/tests/functional/test_gzip_middleware.py:1\n\nChange-Id: I15d984b83a5be4a5aa9c00576377d0a515ccc2c5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6262dd9d57ae009c929f4bca92da968c0cab90a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6262dd9d57ae009c929f4bca92da968c0cab90a5"}]},"branch":"refs/heads/master"},"a96daebabcd7da5b75a201548b8e922dfd613a2f":{"kind":"REWORK","_number":2,"created":"2024-11-15 18:23:42.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/27/935227/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/27/935227/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/935227/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/935227/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/935227/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/27/935227/2"}}},"commit":{"parents":[{"commit":"c7d70ae763667ff772cc4fb13706076641f4fd31","subject":"Merge \"Imported Translations from Zanata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c7d70ae763667ff772cc4fb13706076641f4fd31"}]}],"author":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2024-11-14 01:15:42.000000000","tz":60},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2024-11-15 01:40:05.000000000","tz":60},"subject":"[WIP] Functional tests: do not really start a server","message":"[WIP] Functional tests: do not really start a server\n\nCurrently, most of the functional tests need to run Glance in a separate\nprocess. This is always done using eventlet (not wsgi). Using the\nSynchronousAPIBase allows us to no longer depend on eventlet.\n\nThis patch is a proof of concept that only fixes *1* test. I believe we\nneed to fix 387 tests, from the following files:\n\nglance/tests/functional/db/base.py:190\nglance/tests/functional/v2/test_images.py:62\nglance/tests/functional/db/base_metadef.py:38\nglance/tests/functional/db/test_sqlalchemy.py:25\nglance/tests/functional/test_api.py:9\nglance/tests/functional/serial/test_scrubber.py:9\nglance/tests/functional/test_client_exceptions.py:7\nglance/tests/functional/test_glance_manage.py:6\nglance/tests/functional/test_client_redirects.py:6\nglance/tests/functional/test_cors_middleware.py:4\nglance/tests/functional/test_cache_middleware.py:4\nglance/tests/functional/db/test_migrations.py:4\nglance/tests/functional/v2/test_metadef_namespaces.py:3\nglance/tests/functional/test_logging.py:3\nglance/tests/functional/v2/test_tasks.py:2\nglance/tests/functional/v2/test_metadef_tags.py:2\nglance/tests/functional/v2/test_metadef_resourcetypes.py:2\nglance/tests/functional/v2/test_metadef_properties.py:2\nglance/tests/functional/v2/test_metadef_objects.py:2\nglance/tests/functional/test_wsgi.py:2\nglance/tests/functional/test_healthcheck_middleware.py:2\nglance/tests/functional/test_sqlite.py:1\nglance/tests/functional/test_reload.py:1\nglance/tests/functional/test_gzip_middleware.py:1\n\nChange-Id: I15d984b83a5be4a5aa9c00576377d0a515ccc2c5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a96daebabcd7da5b75a201548b8e922dfd613a2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a96daebabcd7da5b75a201548b8e922dfd613a2f"}]},"branch":"refs/heads/master"},"8d3a3be0a3a61693aabd2264b43919b48e270a5e":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":3,"created":"2024-12-05 13:59:17.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/27/935227/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/27/935227/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/935227/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/935227/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/935227/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/27/935227/3"}}},"commit":{"parents":[{"commit":"1c2f9d82316c3ac5008f8f1b340344a331087f86","subject":"Functional tests migration: test_sqlite.py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1c2f9d82316c3ac5008f8f1b340344a331087f86"}]}],"author":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2024-11-14 01:15:42.000000000","tz":60},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2024-12-05 13:47:54.000000000","tz":60},"subject":"Functional tests migration: test_schemas.py","message":"Functional tests migration: test_schemas.py\n\nUse SynchronousAPIBase instead of FunctionalTest as the base class for\nthe TestSchemas class, so we can run the tests it defines without\nrunning eventlet or running a \"real\" WSGI server.\n\nChange-Id: I15d984b83a5be4a5aa9c00576377d0a515ccc2c5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8d3a3be0a3a61693aabd2264b43919b48e270a5e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8d3a3be0a3a61693aabd2264b43919b48e270a5e"}]},"branch":"refs/heads/master"},"e6bfd1adfef8ee5e4c2b3ab19098ebff5e45b30d":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2025-02-05 19:37:30.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/27/935227/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/27/935227/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/935227/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/935227/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/935227/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/27/935227/4"}}},"commit":{"parents":[{"commit":"a2cb5ff5e19f369ce74a3a3be4529c22f314c60d","subject":"Functional tests migration: test_sqlite.py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a2cb5ff5e19f369ce74a3a3be4529c22f314c60d"}]}],"author":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2024-11-14 01:15:42.000000000","tz":60},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2025-02-05 19:36:18.000000000","tz":60},"subject":"Functional tests migration: test_schemas.py","message":"Functional tests migration: test_schemas.py\n\nUse SynchronousAPIBase instead of FunctionalTest as the base class for\nthe TestSchemas class, so we can run the tests it defines without\nrunning eventlet or running a \"real\" WSGI server.\n\nChange-Id: I15d984b83a5be4a5aa9c00576377d0a515ccc2c5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e6bfd1adfef8ee5e4c2b3ab19098ebff5e45b30d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e6bfd1adfef8ee5e4c2b3ab19098ebff5e45b30d"}]},"branch":"refs/heads/master"},"2f4dfafd2a43e5b9a78188fb6bd7a6d31299a9f3":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2025-02-06 16:12:30.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/27/935227/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/27/935227/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/935227/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/935227/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/935227/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/27/935227/5"}}},"commit":{"parents":[{"commit":"3b82a341a1a632190c613f45f7d22c4a7785e393","subject":"Functional tests migration: test_sqlite.py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3b82a341a1a632190c613f45f7d22c4a7785e393"}]}],"author":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2024-11-14 01:15:42.000000000","tz":60},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2025-02-06 16:12:22.000000000","tz":60},"subject":"Functional tests migration: test_schemas.py","message":"Functional tests migration: test_schemas.py\n\nUse SynchronousAPIBase instead of FunctionalTest as the base class for\nthe TestSchemas class, so we can run the tests it defines without\nrunning eventlet or running a \"real\" WSGI server.\n\nChange-Id: I15d984b83a5be4a5aa9c00576377d0a515ccc2c5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2f4dfafd2a43e5b9a78188fb6bd7a6d31299a9f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2f4dfafd2a43e5b9a78188fb6bd7a6d31299a9f3"}]},"branch":"refs/heads/master"},"7194b038b9115a1f420db275e8d25a6e25b143a5":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2025-02-11 16:47:40.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/27/935227/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/27/935227/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/935227/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/935227/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/935227/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/27/935227/6"}}},"commit":{"parents":[{"commit":"e0818db7f336a7c0fb11dceefdabd5932b2b62b0","subject":"Functional tests migration: test_sqlite.py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e0818db7f336a7c0fb11dceefdabd5932b2b62b0"}]}],"author":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2024-11-14 01:15:42.000000000","tz":60},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2025-02-11 16:31:15.000000000","tz":60},"subject":"Functional tests migration: test_schemas.py","message":"Functional tests migration: test_schemas.py\n\nUse SynchronousAPIBase instead of FunctionalTest as the base class for\nthe TestSchemas class, so we can run the tests it defines without\nrunning eventlet or running a \"real\" WSGI server.\n\nChange-Id: I15d984b83a5be4a5aa9c00576377d0a515ccc2c5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7194b038b9115a1f420db275e8d25a6e25b143a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7194b038b9115a1f420db275e8d25a6e25b143a5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
