)]}'
{"id":"openstack%2Fglance~936393","triplet_id":"openstack%2Fglance~master~Iea70c348478be1c82e1a087b68c954d13afa7770","project":"openstack/glance","branch":"master","topic":"fix-policy-opts","hashtags":[],"change_id":"Iea70c348478be1c82e1a087b68c954d13afa7770","subject":"Fix missing oslo.policy options","status":"ABANDONED","created":"2024-11-27 16:42:26.000000000","updated":"2026-05-10 07:01:04.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2e461610d5735c850df67446ed9592f4e14b4c45","_number":936393,"virtual_id_number":936393,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:check","value":1,"date":"2024-11-27 19:15:34.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":-1,"date":"2024-11-27 18:26:49.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-27 19:15:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1d1e1f08a93aa05504568c402d7ea7dc0c49787b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-27 16:42:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bcda0273b019cf7e17a20f036ed8f96e9110ef91","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-27 16:48:37.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"7cf76549faea69c83d62f95fc54b0cd8ef561915","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-27 18:26:49.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"6931fcf33fd3c8ce581b8e087b22949785efb300","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-27 19:15:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a8235633096a4232a36094254ff0b0eb\n\n- grenade https://zuul.opendev.org/t/openstack/build/545859a5048743f091087063dc805c76 : SUCCESS in 52m 17s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b6d2ea53084f42b5b87b97cc69e88d36 : SUCCESS in 55m 53s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0b3709bf2087499f9c4e274ab7e2e066 : SUCCESS in 1h 27m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c4f2cb59ac8547648cb418806514d78a : SUCCESS in 1h 12m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1c3f3042e9549c2baba3d9afa7b9dee : SUCCESS in 4m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cebed76bf37c499f810875fcf3e5b37f : SUCCESS in 3m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/38c622e5824144c185851a25b7378e58 : SUCCESS in 7m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30679821c00c40c98561575b321ead76 : SUCCESS in 7m 43s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ad8dd5a9d7e845a5bca35d5b2031ab53 : SUCCESS in 9m 01s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3cda9b663f8b40d7b2a1dbc1139c3b26 : SUCCESS in 9m 01s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/4b147bf04bca4eac99c75fb271e64f12 : SUCCESS in 11m 52s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/7ad3d475146f42d79e08127ec8b78e64 : SUCCESS in 2h 04m 15s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/b97a40e1062f49489d265c36cc6d29b5 : SUCCESS in 1h 48m 21s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/52f6f0ddbf844d68b62c8ce7bc3c36ee : SUCCESS in 1h 53m 19s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/62cf175e0d0946dd81bcccb21e753d8c : SUCCESS in 1h 14m 00s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/6ef0fedd64e54dfeacae7416d5e3c220 : SUCCESS in 2h 11m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b324bfd7f9b2454e83d532b3141fc0c4 : SUCCESS in 59m 33s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a1c31fb27b3f4c3cb728827cc8311f0d : SUCCESS in 1h 39m 38s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/20d7bcdde8104565bfdf02dbdf8afd25 : SUCCESS in 1h 33m 12s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/624d05fd9376476a8615ca7487df3f1f : SUCCESS in 1h 05m 33s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/b76d946cca064c42958a4cd5da265bbd : SUCCESS in 20m 38s","accounts_in_message":[],"_revision_number":2},{"id":"806cd567e335a90f826da19b843898ab23bffb3d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-27 21:13:08.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe559835f6a645edac68bda1c49340e0\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/570ae48d667e4ea69533a71b5719d87c : SUCCESS in 12m 46s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0e41b2384bbb4e16bf1e00bc7f06fa5b : SUCCESS in 14m 08s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2e461610d5735c850df67446ed9592f4e14b4c45","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-05-10 07:01:04.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2620879dde3e5e3a28dd97e989d473d0ab967e04","revisions":{"38cc1d9c9513189e8e3085697baef8b9bc5bb330":{"kind":"REWORK","_number":1,"created":"2024-11-27 16:42:26.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/93/936393/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/93/936393/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/93/936393/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/93/936393/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/93/936393/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/93/936393/1"}}},"commit":{"parents":[{"commit":"7eb5e2948cf6674e5627bbf4a91b52ee4e6942d5","subject":"Remove default override for config options policy_file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7eb5e2948cf6674e5627bbf4a91b52ee4e6942d5"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-27 16:42:09.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-27 16:42:09.000000000","tz":540},"subject":"Fix missing oslo.policy options","message":"Fix missing oslo.policy options\n\nThe set_defaults function should be still called without any overrides\nbecause it registers oslo.policy options.\n\nChange-Id: Iea70c348478be1c82e1a087b68c954d13afa7770\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/38cc1d9c9513189e8e3085697baef8b9bc5bb330"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/38cc1d9c9513189e8e3085697baef8b9bc5bb330"}]},"branch":"refs/heads/master"},"2620879dde3e5e3a28dd97e989d473d0ab967e04":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-11-27 16:48:37.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/93/936393/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/93/936393/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/93/936393/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/93/936393/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/93/936393/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/93/936393/2"}}},"commit":{"parents":[{"commit":"7eb5e2948cf6674e5627bbf4a91b52ee4e6942d5","subject":"Remove default override for config options policy_file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7eb5e2948cf6674e5627bbf4a91b52ee4e6942d5"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-27 16:42:09.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-27 16:48:23.000000000","tz":540},"subject":"Fix missing oslo.policy options","message":"Fix missing oslo.policy options\n\nThis partially reverts 7eb5e2948cf6674e5627bbf4a91b52ee4e6942d5 .\n\nThe set_defaults function should be still called without any overrides\nbecause it registers oslo.policy options.\n\nChange-Id: Iea70c348478be1c82e1a087b68c954d13afa7770\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2620879dde3e5e3a28dd97e989d473d0ab967e04"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2620879dde3e5e3a28dd97e989d473d0ab967e04"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
