)]}'
{"id":"openstack%2Fglance~937181","triplet_id":"openstack%2Fglance~master~I2cae99af61196388ec842bb0d7f4c6c065ac18e8","project":"openstack/glance","branch":"master","topic":"eventlet-removal-functional-tests","attention_set":{},"removed_from_attention_set":{"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2025-02-12 06:13:05.000000000","reason":"\u003cGERRIT_ACCOUNT_9303\u003e replied on the change","reason_account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},"4393":{"account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"last_update":"2025-02-14 16:12:12.000000000","reason":"\u003cGERRIT_ACCOUNT_4393\u003e replied on the change","reason_account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2025-03-17 20:16:22.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I2cae99af61196388ec842bb0d7f4c6c065ac18e8","subject":"Functional tests migration: test_sqlite.py","status":"MERGED","created":"2024-12-05 13:59:17.000000000","updated":"2025-03-17 23:39:07.000000000","submitted":"2025-03-17 23:37:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"937181-eventlet-removal-functional-tests","meta_rev_id":"53ff705a9558618c27a4a2bbfdc9559476e0d913","_number":937181,"virtual_id_number":937181,"owner":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-17 23:37:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2025-02-12 06:13:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2025-02-14 16:12:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2025-02-14 16:12:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-05 14:28:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-12-05 15:09:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-12-06 13:02:41.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2025-01-09 15:23:10.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"CC"},{"updated":"2025-01-20 10:49:18.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"CC"},{"updated":"2025-02-07 15:58:08.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"}],"messages":[{"id":"6ebca909f576695119f7091ee434b2ee67114796","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-12-05 13:59:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ca90bc6144829eddfc10048558d3c4d8e90637aa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-05 14:28:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/30ae3d63528d49a49bc61d1f2478da13\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6dfaeed25a8d4935a730cdb615d29be3 : SUCCESS in 12m 28s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ef9b2667b1f74d30a2e8fca2cedb1f7f : SUCCESS in 10m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b93972eb5e0c41596e575bac9143b8d40b8fb5ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-05 15:09:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc73cea3595c464296ceb3deaf7a10ab\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/411aa804450b4d17b149ab0669c4f07a : SUCCESS in 1h 04m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/11359debf63c4481975f44298eaba283 : SUCCESS in 1h 05m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d06a84fd63e46f4bb8a3133446d6a6c : SUCCESS in 4m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/910e417eedd9417690e50ca79f256021 : SUCCESS in 5m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3ac194310bf2402789c16e8baaeecaae : SUCCESS in 7m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6fe74540cc884b7ba5498a488a0fd2f7 : SUCCESS in 5m 50s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2861db8c86c9461990f49b78b3815d56 : SUCCESS in 8m 20s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/059d3643eaeb40a4acd33941699ffd63 : SUCCESS in 10m 14s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/da857e3ee7bf49a79dddc66d0d88dcf4 : SUCCESS in 14m 17s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/25817d45eaf84ccc963e6b97cfc3e40e : SUCCESS in 38m 35s","accounts_in_message":[],"_revision_number":1},{"id":"03ff1472ab4466b57e44f7a1cc56b9cf9cb68419","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-12-06 13:02:41.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"47e974ff728784911d2b9246e34540f74cb55737","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2025-01-09 15:23:10.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"499684b5ea6d0ee60f610bca74d2faaac284605f","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2025-01-20 10:49:18.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"db32789bada71a5b9129d8d7594f7bf7acaea9bb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-02-05 19:37:30.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"12f90ca3422d43702337e8786d3e2e6b2e163c6f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-05 19:52:10.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/15faafcf5af14d64ada2b5cba12e3201\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/60c19079e60a4556a95d283542979f64 : SUCCESS in 12m 34s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/80230f91e2dc43b998e6e56c37a7c700 : SUCCESS in 10m 47s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c6988542575486d38d39b3d7fb1630efe5fbcd85","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-05 20:47:38.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/528985ca28fa490cb0c7cdc7bd657e12\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/cbfa25ff41554996adb11ef5075a1fce : SUCCESS in 58m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6ac984212ed641a2912c3765ebde1941 : SUCCESS in 1h 04m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/815f3343b0044007a101c702d940c4a9 : FAILURE in 5m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ff9b8738f3414074929244ec0d645f92 : SUCCESS in 5m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/30c7470f74c944b68585b2bd83491c91 : SUCCESS in 6m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5da2b3ca30b40068b8bb422799f6d68 : SUCCESS in 5m 26s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b001c358ac4446ffb86106709326c3ef : SUCCESS in 7m 37s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e18a9e5e29e14e68a9df51a647ffe3a1 : SUCCESS in 8m 21s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/13ad70b4b8f24ac99b3da66be45e2b3e : SUCCESS in 15m 52s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/492cbe28c1df4a2b9df772da1d1be6b7 : SUCCESS in 24m 09s\n\nWarning:\n  Comments left for invalid file glance/tests/functional/db/base.py","accounts_in_message":[],"_revision_number":2},{"id":"2c5d23a86c025d6618033d308a4df6576056fafb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-02-06 16:12:30.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"bf9b05b8d536609625cdf1070d365effe8227d1d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-06 17:21:23.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f07607a230e74a6981c7e1202d850371\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ea9d17d5f11f4534b975edbf49424a61 : SUCCESS in 51m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fd239139398a459a8ef4699a1741c1c5 : SUCCESS in 1h 03m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f539cb7ce38a4bcf9be09ed9f02f9eb6 : SUCCESS in 5m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/959ba142337246e094d657093baf2ed8 : SUCCESS in 3m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/57443d5dea814be69c953fc441680cca : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4fa80db790234ae7aed505ba5ed7eee7 : SUCCESS in 5m 26s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/75fe0698e3bd466abdd00947c041e8bd : SUCCESS in 9m 47s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/165449ba815f45479d59ec50db4fb45a : SUCCESS in 10m 17s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/4398044f6cb5422ba855ebc59107c9de : SUCCESS in 14m 11s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/e33cf637bb364bc8a7dc811ce65f97f7 : SUCCESS in 33m 51s","accounts_in_message":[],"_revision_number":3},{"id":"f7642c30c908ab02c6b373a894418fb831dc6d5a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-06 17:29:30.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/18f26c9f411b48d3a522012d0ab8006e\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9ca75ea12b15409b97c1b3b8946ac433 : SUCCESS in 17m 15s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1c59dc56c8694f108b5279c6bfb3e8c5 : SUCCESS in 14m 52s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a15afe0f3de4ce6922b74d2daf25c1fc6d059826","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2025-02-07 15:58:08.000000000","message":"Patch Set 3: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"76bba3c6a475951ce383026ca964d2997e78e390","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-02-11 16:47:00.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"7456ab8d918693c69dcc30909e400d65f83427c2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-02-11 16:47:40.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"7941d47c77c94ca3de8e8268564b05a3c4ec3d56","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 16:58:25.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/73b530cba85b4c6aaf842a244a105a21\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/36a8e4f667d14ffe927020202c7c3c5f : SUCCESS in 8m 52s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/29304d78fcc24c51ae4a22d4348f3f69 : SUCCESS in 8m 55s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"5d4b7d3d6d22008497ffeae298e077697154fabb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 17:51:18.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/80b2148c564d448e9006e603be86b420\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e5169527f0304823bc10ff2c10d90ced : SUCCESS in 53m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3b5c340be6494fb6921e018512c911d4 : SUCCESS in 1h 01m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/752fce69c0f84240966c9794ec77b2e8 : SUCCESS in 5m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/02eadc2542ba45c4b23486bc4ba67970 : SUCCESS in 6m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4d85456c6eb14b5ea38bb1b55431a6d1 : SUCCESS in 6m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b35da93658d427683ac48d07532cd64 : SUCCESS in 4m 55s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/96d65ec5ac9349fea366534137c23764 : SUCCESS in 8m 39s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/627a43102d8e4246b7d4d1c00c2890aa : SUCCESS in 7m 17s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/202801d94cae494c92fa7ff3a9087fa3 : SUCCESS in 11m 27s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/8ea57a17409340ce887daaa45f242c99 : SUCCESS in 24m 35s","accounts_in_message":[],"_revision_number":4},{"id":"0db8249d3928baa3da1fd4271f6f173fc5cd9870","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-02-12 06:13:05.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"955f6e6d06ba6bddde27aa4348cd90af761a251e","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2025-02-14 16:12:12.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"dcaafd3c1f9404dd0837b05096e7f59adeb03df1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-14 16:13:14.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"e1eccedf2fda07988c4249212528989b6290709b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-14 17:30:46.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/23dda0909d774fdeb5e12251a88e2391\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3a1b2159853e4f2aabf2e8d7e3e1144d : SUCCESS in 1h 06m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/707868c7c39d4541a52aa8c6d3cd22f9 : SUCCESS in 1h 04m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad1a9f13a99f4c40a23bf88cba37c74a : SUCCESS in 5m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d1cf681f3200473b8ecad6fb08ef938a : SUCCESS in 6m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/61174992b6c54e7da6b41c63f6a6752f : SUCCESS in 6m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3ffd42d607a4290a8283a5848866161 : SUCCESS in 5m 36s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/70b25aa215aa44858395eee5d3095acf : FAILURE in 9m 25s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d449e87742ad400e8b213a6583988fb4 : SUCCESS in 8m 04s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/d237484626314ad3885d47a46189b126 : SUCCESS in 10m 41s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/ecb3d693d7aa453a9e220b690b9f80cb : SUCCESS in 45m 16s","accounts_in_message":[],"_revision_number":4},{"id":"10f37569d27cba78193ad109a1352b468c9dcfe5","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-03-11 14:16:14.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"109e15cac72a18dec774b74393711a1526b2af61","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-11 14:26:52.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1ed63d919f8a4a77b9161bae7d6d77e4\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/354d9c385e774088a5e8bab403ef0340 : SUCCESS in 7m 50s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/52f276e3b2fd4d5096adfb28d63861f9 : SUCCESS in 8m 44s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"d41a0bfe424e461b35aa3e9d95bf9a43b73fbf21","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-11 14:59:17.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/980da6eb47514d5fb356166c4f03cf3a\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5b8202776ad5499f9f2bb9f186e9b446 : SUCCESS in 34m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/94eb34c6c5bd4c3d84514a429e2ea179 : SUCCESS in 39m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b323b8d6d36948028462deb0d6395336 : SUCCESS in 3m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3344219836ee4b72ab6570087e2f1a6a : SUCCESS in 3m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bb4e7cc66fa0411cbb01b935c26a5a89 : SUCCESS in 2m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e502bbd67f3441dbfb025d8f3a0b4e4 : SUCCESS in 3m 47s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/975cd75a24a14c8788512d43239bbefe : FAILURE in 6m 28s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4b6fc3f4f8624aa7959ddda759be96b4 : FAILURE in 5m 38s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/b9359809559b4635ad34495ede685b2f : SUCCESS in 12m 33s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/a83d61bbaa7a44188a905a9e5f3da194 : SUCCESS in 36m 57s","accounts_in_message":[],"_revision_number":4},{"id":"9862cbca7d3008e4b5c5946ab9ea7d65845876a7","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-03-17 15:02:45.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"4ffb33988205cd1be9e6fc61ed19140abe7f0b31","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-17 15:11:28.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d9f13e3280084b478a38ffe0690e6047\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/346b3dac695f419f8d7fe94e3e9ebde4 : SUCCESS in 6m 33s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9354cf6634d14e1588363238aa70fcc5 : SUCCESS in 6m 56s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"b55763b79a17b871b31a4593e393bce4f6a26d07","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-17 16:03:41.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d9b4c70399fe4a6a8bcae624bb81bc66\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d4e3c8dfc7fb448b984ede8c04ca4a89 : SUCCESS in 44m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/78fe6b49e3914fd0989340897d5aebf9 : SUCCESS in 59m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/109d9f2cda7348e59448877469c99168 : SUCCESS in 3m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cd7dfe01d1304f7788baf712d0d32ef6 : SUCCESS in 3m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3433fb04e33c461f902bbf48e64c1dae : SUCCESS in 6m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6cb0cf9afa79446babdb733230d39d96 : SUCCESS in 6m 08s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f94643df3b3e4a398faa71503fee0c13 : SUCCESS in 9m 58s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f30b4e92c661461d96dfa052128b8d9f : SUCCESS in 8m 55s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/e85514655fee4faf9b3863d0e241db5f : SUCCESS in 12m 56s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/00f1e4c335254b4cab0f937fa27ab55c : SUCCESS in 47m 20s","accounts_in_message":[],"_revision_number":4},{"id":"69a8d7b4fce73c12fc04294aee61b45131447476","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-17 16:04:21.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"fca2baf1c0ae6cddcc2691fd80e990299ceb21ad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-17 17:15:12.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/29a1947d690943c6ba0fd58fd76a6824\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/95632e0723874ec5b024f5dac4a1084a : SUCCESS in 1h 04m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/78ded1fe4ca84cb0bc3253ee2e5d3f9c : SUCCESS in 1h 05m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4fadd69f53314a649492a053d2f0de59 : SUCCESS in 6m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5f9f1174ef6e4a4aa602d0805372db3a : SUCCESS in 5m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8c9648a5200b4c55a9445e25edc28454 : SUCCESS in 5m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dfacbc8bcf664a1d86915c0743612be6 : SUCCESS in 2m 59s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/620827fa57744fd8a8081253433f14fe : SUCCESS in 8m 35s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4cf6f2ecdf644631b55e4f27ee51c15e : SUCCESS in 8m 40s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/129ebe3af9884e5984947a039a1fe93a : SUCCESS in 12m 32s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/5a52332bbade4e9da210889010514f33 : POST_FAILURE in 33m 29s","accounts_in_message":[],"_revision_number":4},{"id":"fbbbce7857135e21ef0ed73f2000e14565663042","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-03-17 20:16:22.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"197ec122561f9d4c089c03d9fe5ff7754d6b9524","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-17 20:24:49.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e8641d16637d414582b367a2dfcb5169\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d092217340f84e85b0c707ea32beef8b : SUCCESS in 5m 32s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1d3ba75d963f45ae8edd80bbb0e05800 : SUCCESS in 6m 37s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"e6236e75f5bc23667548b9490496a23c982c464e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-17 21:24:41.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4710b9e3befd4c888afaddaded053b4a\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/00331ef02c9a4971bb17bba31416ac7c : SUCCESS in 49m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/aee10e54ebb541f39d0aaddc4c08a1f6 : SUCCESS in 1h 02m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/848c4640bd6149059ddcd0eea827f0cb : SUCCESS in 5m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9fcaf499b88f4a52ac705da6861c399d : SUCCESS in 4m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b4621eb383bc45128de20edfb6073346 : SUCCESS in 5m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cbc177181e5b482a8c6c6d821d9f3e10 : SUCCESS in 5m 37s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9c511423fde6445dbba3783b47d723c4 : SUCCESS in 6m 52s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/42ccbec601ab4c179dfdee0388ad8b2b : SUCCESS in 8m 19s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/697da08117c043258f37edc564ecdf47 : SUCCESS in 10m 14s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/9301b26b9be64b0da87b3ce02694cd91 : SUCCESS in 25m 52s","accounts_in_message":[],"_revision_number":4},{"id":"e0b996d0ef9aafa577ba33603f59a9634aeb015e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-17 21:25:15.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"e4c1d40379354ab5022a4678d91c3b53103c7a26","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-17 23:37:53.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb8f7c05be3648fc86c2fa13c3e1ebed\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8fa0be61a64f43deb0a0e95bf6390d71 : SUCCESS in 57m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c1d211bba7d545848e5fb1de27ed8e58 : SUCCESS in 1h 59m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14e43ca6162a430bb87e5d0ef3a54783 : SUCCESS in 4m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/89c5f2b39d464099ad2ef6b6dcc51cd9 : SUCCESS in 4m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/aa5a5fb6a8634a24bb671e851b5a465a : SUCCESS in 5m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35fdc3bcd64a4d9982614cd9bb604766 : SUCCESS in 6m 50s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ad7a6436b4274ddebdcf25f1f3a3ed8f : SUCCESS in 11m 25s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/670f43f408914b7595ba56bc39fdfeea : SUCCESS in 9m 46s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/703b88b1a68e4a00baedfbf22bcf232c : SUCCESS in 10m 11s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/c2c7f61c45a04f9991d0f6b6f6d723a3 : SUCCESS in 33m 45s","accounts_in_message":[],"_revision_number":4},{"id":"1db5d0a9da02e082cbecb7607b7e26d932ef748a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-17 23:37:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"53ff705a9558618c27a4a2bbfdc9559476e0d913","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-17 23:39:07.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f01c34ee468e4df4966b2af346f25890\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9833293687d4b23bb15ca65cf08e2ce : SUCCESS in 55s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"e0818db7f336a7c0fb11dceefdabd5932b2b62b0","revisions":{"1c2f9d82316c3ac5008f8f1b340344a331087f86":{"kind":"REWORK","_number":1,"created":"2024-12-05 13:59:17.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/81/937181/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/81/937181/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/81/937181/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/81/937181/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/81/937181/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/81/937181/1"}}},"commit":{"parents":[{"commit":"758cc76c26124d71c71ab357c7273ce1af7c2333","subject":"FunctionalInitWrapper: inherit from SynchronousAPIBase","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/758cc76c26124d71c71ab357c7273ce1af7c2333"}]}],"author":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2024-12-04 23:17:49.000000000","tz":60},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2024-12-04 23:58:08.000000000","tz":60},"subject":"Functional tests migration: test_sqlite.py","message":"Functional tests migration: test_sqlite.py\n\nUse SynchronousAPIBase instead of FunctionalTest as the base class for\nthe TestSqlite class, so we can run the tests it defines without running\neventlet or running a \"real\" WSGI server.\n\nChange-Id: I2cae99af61196388ec842bb0d7f4c6c065ac18e8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1c2f9d82316c3ac5008f8f1b340344a331087f86"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1c2f9d82316c3ac5008f8f1b340344a331087f86"}]},"branch":"refs/heads/master"},"a2cb5ff5e19f369ce74a3a3be4529c22f314c60d":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-02-05 19:37:30.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/81/937181/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/81/937181/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/81/937181/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/81/937181/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/81/937181/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/81/937181/2"}}},"commit":{"parents":[{"commit":"028180816f26cd190ebc9c443606b3a5b7d7ceb7","subject":"Remove FunctionalInitWrapper","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/028180816f26cd190ebc9c443606b3a5b7d7ceb7"}]}],"author":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2024-12-04 23:17:49.000000000","tz":60},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2025-02-05 19:36:18.000000000","tz":60},"subject":"Functional tests migration: test_sqlite.py","message":"Functional tests migration: test_sqlite.py\n\nUse SynchronousAPIBase instead of FunctionalTest as the base class for\nthe TestSqlite class, so we can run the tests it defines without running\neventlet or running a \"real\" WSGI server.\n\nChange-Id: I2cae99af61196388ec842bb0d7f4c6c065ac18e8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a2cb5ff5e19f369ce74a3a3be4529c22f314c60d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a2cb5ff5e19f369ce74a3a3be4529c22f314c60d"}]},"branch":"refs/heads/master"},"3b82a341a1a632190c613f45f7d22c4a7785e393":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2025-02-06 16:12:30.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/81/937181/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/81/937181/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/81/937181/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/81/937181/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/81/937181/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/81/937181/3"}}},"commit":{"parents":[{"commit":"7da454e8c0d2df81bf4c24208ac661fe3f8ed8a2","subject":"Remove FunctionalInitWrapper","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7da454e8c0d2df81bf4c24208ac661fe3f8ed8a2"}]}],"author":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2024-12-04 23:17:49.000000000","tz":60},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2025-02-06 16:12:22.000000000","tz":60},"subject":"Functional tests migration: test_sqlite.py","message":"Functional tests migration: test_sqlite.py\n\nUse SynchronousAPIBase instead of FunctionalTest as the base class for\nthe TestSqlite class, so we can run the tests it defines without running\neventlet or running a \"real\" WSGI server.\n\nChange-Id: I2cae99af61196388ec842bb0d7f4c6c065ac18e8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3b82a341a1a632190c613f45f7d22c4a7785e393"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3b82a341a1a632190c613f45f7d22c4a7785e393"}]},"branch":"refs/heads/master"},"e0818db7f336a7c0fb11dceefdabd5932b2b62b0":{"kind":"REWORK","_number":4,"created":"2025-02-11 16:47:40.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/81/937181/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/81/937181/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/81/937181/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/81/937181/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/81/937181/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/81/937181/4"}}},"commit":{"parents":[{"commit":"5cfecc79995c6a9524f0dbc4553b4a88761b382f","subject":"Remove FunctionalInitWrapper","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5cfecc79995c6a9524f0dbc4553b4a88761b382f"}]}],"author":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2024-12-04 23:17:49.000000000","tz":60},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2025-02-11 16:31:10.000000000","tz":60},"subject":"Functional tests migration: test_sqlite.py","message":"Functional tests migration: test_sqlite.py\n\nUse SynchronousAPIBase instead of FunctionalTest as the base class for\nthe TestSqlite class, so we can run the tests it defines without running\neventlet or running a \"real\" WSGI server.\n\nThis also fixes this test which had not been doing anything for a while.\nBecause the database file \"tests.sqlite\" did not exist (it should have\nbeen \"{self.test_dir}/tests.sqlite\"), the returned schema was empty.\nObviously, \u0027BIGINT\u0027 could not be found in that, so the test passed.\n\nChange-Id: I2cae99af61196388ec842bb0d7f4c6c065ac18e8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e0818db7f336a7c0fb11dceefdabd5932b2b62b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e0818db7f336a7c0fb11dceefdabd5932b2b62b0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
