)]}'
{"id":"openstack%2Fglance~937182","triplet_id":"openstack%2Fglance~master~I1ffc642cd071c262ee45b81743f6e9ed832db811","project":"openstack/glance","branch":"master","topic":"eventlet-removal-functional-tests","attention_set":{},"removed_from_attention_set":{"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2025-03-18 23:38:31.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_37598\u003e by clicking the attention icon","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}},"hashtags":[],"change_id":"I1ffc642cd071c262ee45b81743f6e9ed832db811","subject":"Functional test migration: test_tasks.py","status":"MERGED","created":"2024-12-05 13:59:17.000000000","updated":"2025-03-19 03:03:56.000000000","submitted":"2025-03-19 03:02:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"937182-eventlet-removal-functional-tests","meta_rev_id":"97eb6bbb956cf7c9cc5f72eb191712f2b7999a2e","_number":937182,"virtual_id_number":937182,"owner":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-19 03:02:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2025-02-11 16:47:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2025-02-14 16:14:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-03-18 23:38:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2025-02-14 16:14:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"CC":[{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-05 14:36:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-12-05 15:05:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-12-06 13:03:30.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2025-01-20 10:49:26.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"CC"},{"updated":"2025-02-14 16:14:32.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2025-03-18 23:38:20.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"}],"messages":[{"id":"5407f81ad74a631e8faf8432c6ead487a83394fe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-12-05 13:59:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d737ff146241f36c0b333dafbb6914fa7f6e104b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-05 14:36:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/00a3029f8b6a4cf28fa74b34f55ec9dd\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/989d7a30be604581a36a02bd45abecd4 : SUCCESS in 10m 03s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ea9d7534f6544d88b8dc5d2f1c8b965b : SUCCESS in 10m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"be39f4346d59644479c2a3d1b7c3dad6c8eb5544","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-05 15:05:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ce7d31e647b48098eb31d01bbb022df\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6369d0f743314b34b54b4d7497d97707 : SUCCESS in 54m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e91aa38be6174f49ba497efc9ae058e5 : SUCCESS in 1h 05m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b8b0547a6fe4b58b11acbc51cc20b6c : SUCCESS in 6m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c3d0b59945b843c9bdd2f437140a9854 : SUCCESS in 6m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a473ab45ecff463dabd0275bbc0d6d63 : SUCCESS in 5m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b25d963b408403395306a52d3fd8090 : SUCCESS in 6m 01s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/95c8b57e81954690b0ed29b54b0e3466 : SUCCESS in 8m 25s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/02c7d4e6cf6847b1ac9e2c2bd2cce9c5 : SUCCESS in 9m 51s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/cb622677891f4ef8925d46c5a93e0a02 : SUCCESS in 13m 53s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f7c47d33979a4333897d179b2fe7b29d : SUCCESS in 33m 45s","accounts_in_message":[],"_revision_number":1},{"id":"8c6bc665844a5caa8693d117dac31c588464e6e6","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-12-06 13:03:30.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"98cd8e12b05e142ad4f8aec3f2fad8fd071b17ae","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2025-01-20 10:49:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ec2b435fe52cf42f172e5825e64f6d4899e163cb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-02-05 19:37:30.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"020ea1d16bba99ae635f27ff62180f430eb651aa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-05 19:50:53.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/54e30e0b362d44949bbec73cacee2792\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6cea1c57abf44f64b286a7845c4bb903 : SUCCESS in 11m 15s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5a1a3456c5f3452ba1b44e0adc99c862 : SUCCESS in 10m 15s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"fc7040933aa796bc354eff28e05a9843cc4975bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-05 20:31:06.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c0a190f3806846979ba82dcbd4827cda\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ee9e2d0b6e6f47428fad3a72b7b2db80 : SUCCESS in 44m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/118eafe6f8104dc79defcb5332b0913e : SUCCESS in 52m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54ae248623a640a789c30277ab624d88 : FAILURE in 6m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b9b2fd3d94254e38a57d0114cd2d608d : SUCCESS in 5m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e1a7b6123f2e449691d7c947c5623be4 : SUCCESS in 5m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/023f0347325d444698275a8cda947785 : SUCCESS in 8m 20s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b64c4db1e3ea40e2b546f2e3f13ede22 : SUCCESS in 8m 09s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/51b907cb42ce4a04b2b4269e8ff9ec71 : SUCCESS in 7m 51s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/b5150b60be5f4bc18e5f8a7c63f8f208 : SUCCESS in 11m 56s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/ef8f4aaad428483ca28dbc59cc2c64d0 : SUCCESS in 34m 59s\n\nWarning:\n  Comments left for invalid file glance/tests/functional/db/base.py","accounts_in_message":[],"_revision_number":2},{"id":"0eefef24d7b92cbee4aae57f9a43e35f4bd98a17","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-02-06 16:12:30.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"a1acacb4bdfc787dd818f51382de8474e85eaa7b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-06 17:27:05.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b6028079e90d4324b2f5f016a89f43a5\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/012d9ddd25af43b1b8d477945d9ba66c : SUCCESS in 58m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d73a708a26c44f98b4b5c1e467ea7b66 : SUCCESS in 1h 08m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7e26883988645f7b3a871261b4257e4 : SUCCESS in 7m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a94d434fd31649beab8411a966d98b8a : SUCCESS in 5m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d53e6c9434124395a4fecf19b25c1982 : SUCCESS in 7m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f6385342536460390f1900e0d731375 : SUCCESS in 9m 27s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/289a324e21b64db48d738fb7bbd40451 : SUCCESS in 8m 41s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7724a0eb325e4150a4d9b7d2ff6eb021 : SUCCESS in 8m 11s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/fbe080d97ce343d5bdeaaa7e9b2c9643 : SUCCESS in 12m 18s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/667ad7f7e91949b5a5ddcdb76efad035 : SUCCESS in 38m 36s","accounts_in_message":[],"_revision_number":3},{"id":"234910f2e09cc06b8f66def144ca16d1fcfa8222","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-06 17:51:07.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4c26d4bf4abd4a8593a3cb1d01fff3be\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6e4052c5cfe540d1a03b047460c01b4a : SUCCESS in 10m 27s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2da0368346424423b483553138ff09ef : SUCCESS in 14m 24s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"4e1dad9a86e7af2bdc31aaac5ff1edee1c1610a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-02-11 16:47:40.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"5f127fb5c4d64c90506ef7e952db65171c48f980","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 17:01:00.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/72e1e99574e34979971f50fc5359cdc4\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/705bdba042ca47cbb2416c957083b077 : SUCCESS in 9m 59s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/bf6b9358230147d3baadb22a7e9822cb : SUCCESS in 11m 24s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"c226f4e8a8bd24864bd98b120f963016b6db4c5e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 18:02:37.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f99c2b58fc0d44da82f004909c487aa7\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/84ac2d33e9e4425aa2b7c48ac09ae5f2 : SUCCESS in 47m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cdcca235ffdd46c38803105143e9a5e7 : SUCCESS in 1h 08m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6bef166626c4b9d9e75069bffafcb56 : SUCCESS in 5m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5217117448704a6d8ec6ffadbe4f8967 : SUCCESS in 4m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fa9bceca05be4ad787c243916fae75e7 : SUCCESS in 6m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4b5c410d68a40d8ab6ebeb997a7ce69 : SUCCESS in 6m 12s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e1d9a42f40d54e0c9c5310e1a0f7bb79 : SUCCESS in 8m 43s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a52c9de791fa49468da3bb5e2853fda5 : SUCCESS in 8m 34s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/bfd4c67b63a1470bbab3825f165a93b6 : SUCCESS in 11m 51s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/66cd78d964d84a2f8b164cb8ce61666b : SUCCESS in 30m 15s","accounts_in_message":[],"_revision_number":4},{"id":"0dd1e838c5a27c84493598b8a7b10b690cf927f3","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2025-02-14 16:14:32.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"b56b4dbbc3beb9ba4c4a3706fafb4f31f02c6848","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-14 16:15:22.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"4d22024664cb54834a0f0d923bc815cd3afb2aea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-14 17:30:48.000000000","message":"Patch Set 4: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/935227 is needed.","accounts_in_message":[],"_revision_number":4},{"id":"71125dbfa6c1ab4e10cb9c3330deb10c39e6f149","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-03-18 16:59:38.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"14a11af4f84eb5bf030f05f29d44aca530280cb5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-18 17:09:52.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/49ded77580254d1b9a5ac7063dff4e24\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8413ec2442e5496eb78ccf5a52e5137c : SUCCESS in 7m 25s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a9b3c3ef7299401789ff46a9cb36f96b : SUCCESS in 8m 26s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"1389a039f44205215ae092d416b5e06b8d40fd13","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-18 17:58:06.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/365c5041af2c4efa9c0f9bf03e30fc88\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ce4f89947649405ebae3ec97a8ce4389 : SUCCESS in 49m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/47dce6ac7e1d4f2a83ea6fec9e55b4c2 : SUCCESS in 49m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b083713eadec451496a9fe4202e899e7 : SUCCESS in 2m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9403681a9f894a5094f52fed211f799f : SUCCESS in 5m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8deda5d1636f4d0d8f315dc1aafad97e : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4ca657de9ad4e12be9c2f38af420b0e : SUCCESS in 7m 29s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3eb601de58a84071bff4af1966d62f94 : SUCCESS in 7m 30s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8f2d877e8f434447adf10158df9326ab : FAILURE in 6m 03s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/98465101527e4d79833488851623c6b6 : SUCCESS in 11m 40s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/51820c8efed24d0a8ca113b8d35bd3df : SUCCESS in 18m 07s","accounts_in_message":[],"_revision_number":4},{"id":"16e18de564fe704da356019409cd82837567578c","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-03-18 22:23:13.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"cfd727738f2f86093d1690d64c6c0d8204f1d4cf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-18 22:33:05.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2bc470e86e164a45b98565c1c701dd82\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/19c06f5253e441099b5ed1ead126cff5 : SUCCESS in 7m 07s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b0474b7ef0df4a939678d7d164fea4a6 : SUCCESS in 8m 00s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"dcce71e039025cd3d4e85049a45b344df0ce2009","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-18 23:28:08.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/63ec01ceaf564734a693a8a85ecf21cb\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0a7a393dc87045469347d2ea18f46983 : SUCCESS in 58m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2549684bf78747f8918e84c4460593d6 : SUCCESS in 1h 02m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08b1a660112642d4b7d09980bc618f83 : SUCCESS in 6m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ce659fdc4a2c4e1cb5c1ed929f395aca : SUCCESS in 5m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2c023be1943c4dcd98dbe0c59f3a687b : SUCCESS in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7ecf87d3f34483bb15e37a1b914b4e1 : SUCCESS in 5m 10s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b0d614f18ff447eaaabbea8c0c9f2b61 : SUCCESS in 7m 25s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/77ad8aef507543558a7280a87c9cdd0c : FAILURE in 5m 31s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/3259cec78fbf467ba362101d63884892 : SUCCESS in 11m 01s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/d2ceefe1ce6d45bc8a5dc7c773528185 : SUCCESS in 18m 01s","accounts_in_message":[],"_revision_number":4},{"id":"2db7af694cadd5483e322aadc8519af014b03865","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-03-18 23:38:20.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"8cede9527c2c8222548de9d337354519d3e8ae93","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-03-18 23:38:29.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"2738dc8492322faca537def631118c98027372e2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-18 23:48:43.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b0dd4c091e8a4630a20633bde2a1adce\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9b629972e7c448de89411adadfc9c125 : SUCCESS in 7m 08s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a2ef1e0907d940738b8f25f32b598e1d : SUCCESS in 8m 30s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"c737944126b62c0de7a7c548df4decf00514e41e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-19 00:42:51.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/428801da8052404cbeacaf1a8273201d\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/42dcc6c04c40485b947df3207dcd86f9 : SUCCESS in 53m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/09d4d1567f314b89970852f2a5df347d : SUCCESS in 59m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/626e892c23a546e4851c22aacf24f8d0 : SUCCESS in 4m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fdf54ada052d4be298da9f1eaac3b61e : SUCCESS in 4m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/33d04889463d4e36baeb51699f2c83b6 : SUCCESS in 2m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c328bd2dbd4f49ce8f0f369d10bb34f5 : SUCCESS in 5m 17s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/80ae5c76f2b640c4bf9bc5e72216aae4 : SUCCESS in 8m 02s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8e970b690ad44d10bc8145af4386ad73 : SUCCESS in 6m 34s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/24df46f6071548bfa4b117ccfe7f007c : SUCCESS in 9m 46s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/01d9f75e57ab4d4fb677ad8e6d97e6ae : SUCCESS in 30m 48s","accounts_in_message":[],"_revision_number":4},{"id":"9d1db5a4b802eb737027800eb95af64eaeadd24b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-19 00:43:30.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"f569e2bef0ca42e1c22bc933f265ec942c749288","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-19 03:02:21.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aa35b0dd391c4401a444b460fa8faa90\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9565a979cf3141d69372664ac26d3c76 : SUCCESS in 25m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ab094ceaac0f4be98a19fd9cb040dcb5 : SUCCESS in 58m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4c0a00fcf5b4eab9c199b7725bc826a : SUCCESS in 2m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/feb73a8a4efa4ee3beb5104bf7d5579e : SUCCESS in 4m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/993b7383477d412894638fd5e6af8786 : SUCCESS in 5m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/083da73f63ce4357b78779231a193a84 : SUCCESS in 5m 29s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1254e75d1b0c411d80068def3a38b49d : SUCCESS in 8m 43s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3ade66098919491ba8574376b20ab653 : SUCCESS in 8m 09s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/75eb052174e84724bdb239f181e32a66 : SUCCESS in 15m 21s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/c1086938ff394f3faa2c8bd749c0865d : SUCCESS in 25m 56s","accounts_in_message":[],"_revision_number":4},{"id":"badc112d2c746b385ae2b142b65ca38c3b8a6d33","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-19 03:02:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"97eb6bbb956cf7c9cc5f72eb191712f2b7999a2e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-19 03:03:56.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/106d0891c4f741c493168dd1b54af887\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c43ad50b5ea0457fb0ffbff479b0b6bd : SUCCESS in 51s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"01686b80f23ef5b23f418c879d865e9765aaacf3","revisions":{"94874d329cf4c2897740de29827271f96a663bcd":{"kind":"REWORK","_number":1,"created":"2024-12-05 13:59:17.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/82/937182/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/82/937182/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/82/937182/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/82/937182/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/82/937182/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/82/937182/1"}}},"commit":{"parents":[{"commit":"8d3a3be0a3a61693aabd2264b43919b48e270a5e","subject":"Functional tests migration: test_schemas.py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8d3a3be0a3a61693aabd2264b43919b48e270a5e"}]}],"author":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2024-12-04 22:19:13.000000000","tz":60},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2024-12-05 13:49:19.000000000","tz":60},"subject":"Functional test migration: test_tasks.py","message":"Functional test migration: test_tasks.py\n\nMove all two tests from glance/tests/functional/v2/test_tasks.py from\nFunctionalTest to SynchronousAPIBase as part of our eventlet removal\neffort.\n\nChange-Id: I1ffc642cd071c262ee45b81743f6e9ed832db811\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/94874d329cf4c2897740de29827271f96a663bcd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/94874d329cf4c2897740de29827271f96a663bcd"}]},"branch":"refs/heads/master"},"e0e52f042ec759de5b9915f540b75559d73afc7d":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-02-05 19:37:30.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/82/937182/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/82/937182/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/82/937182/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/82/937182/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/82/937182/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/82/937182/2"}}},"commit":{"parents":[{"commit":"e6bfd1adfef8ee5e4c2b3ab19098ebff5e45b30d","subject":"Functional tests migration: test_schemas.py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e6bfd1adfef8ee5e4c2b3ab19098ebff5e45b30d"}]}],"author":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2024-12-04 22:19:13.000000000","tz":60},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2025-02-05 19:36:18.000000000","tz":60},"subject":"Functional test migration: test_tasks.py","message":"Functional test migration: test_tasks.py\n\nMove all two tests from glance/tests/functional/v2/test_tasks.py from\nFunctionalTest to SynchronousAPIBase as part of our eventlet removal\neffort.\n\nChange-Id: I1ffc642cd071c262ee45b81743f6e9ed832db811\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e0e52f042ec759de5b9915f540b75559d73afc7d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e0e52f042ec759de5b9915f540b75559d73afc7d"}]},"branch":"refs/heads/master"},"a3c65f8981c80c043d8f105f9aaf4e4855b6b04c":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2025-02-06 16:12:30.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/82/937182/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/82/937182/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/82/937182/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/82/937182/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/82/937182/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/82/937182/3"}}},"commit":{"parents":[{"commit":"2f4dfafd2a43e5b9a78188fb6bd7a6d31299a9f3","subject":"Functional tests migration: test_schemas.py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2f4dfafd2a43e5b9a78188fb6bd7a6d31299a9f3"}]}],"author":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2024-12-04 22:19:13.000000000","tz":60},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2025-02-06 16:12:22.000000000","tz":60},"subject":"Functional test migration: test_tasks.py","message":"Functional test migration: test_tasks.py\n\nMove all two tests from glance/tests/functional/v2/test_tasks.py from\nFunctionalTest to SynchronousAPIBase as part of our eventlet removal\neffort.\n\nChange-Id: I1ffc642cd071c262ee45b81743f6e9ed832db811\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a3c65f8981c80c043d8f105f9aaf4e4855b6b04c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a3c65f8981c80c043d8f105f9aaf4e4855b6b04c"}]},"branch":"refs/heads/master"},"01686b80f23ef5b23f418c879d865e9765aaacf3":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2025-02-11 16:47:40.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/82/937182/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/82/937182/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/82/937182/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/82/937182/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/82/937182/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/82/937182/4"}}},"commit":{"parents":[{"commit":"7194b038b9115a1f420db275e8d25a6e25b143a5","subject":"Functional tests migration: test_schemas.py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7194b038b9115a1f420db275e8d25a6e25b143a5"}]}],"author":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2024-12-04 22:19:13.000000000","tz":60},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2025-02-11 16:31:15.000000000","tz":60},"subject":"Functional test migration: test_tasks.py","message":"Functional test migration: test_tasks.py\n\nMove all two tests from glance/tests/functional/v2/test_tasks.py from\nFunctionalTest to SynchronousAPIBase as part of our eventlet removal\neffort.\n\nChange-Id: I1ffc642cd071c262ee45b81743f6e9ed832db811\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/01686b80f23ef5b23f418c879d865e9765aaacf3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/01686b80f23ef5b23f418c879d865e9765aaacf3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
