)]}'
{"id":"openstack%2Fglance~937254","triplet_id":"openstack%2Fglance~master~I07aea6a11e9ff34127e96129cbf610703e8d90ee","project":"openstack/glance","branch":"master","topic":"eventlet-removal-functional-tests","attention_set":{},"removed_from_attention_set":{"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2025-03-28 03:55:11.000000000","reason":"\u003cGERRIT_ACCOUNT_9303\u003e replied on the change","reason_account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},"4393":{"account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"last_update":"2025-04-02 13:36:42.000000000","reason":"\u003cGERRIT_ACCOUNT_4393\u003e replied on the change","reason_account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2025-04-02 14:44:51.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I07aea6a11e9ff34127e96129cbf610703e8d90ee","subject":"Functional tests migration: test_healthcheck_middleware","status":"MERGED","created":"2024-12-06 13:59:39.000000000","updated":"2025-04-02 14:46:24.000000000","submitted":"2025-04-02 14:44:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"937254-eventlet-removal-functional-tests","meta_rev_id":"752f3ccffa1c75dbaa77dcf161e5c63d2136262b","_number":937254,"virtual_id_number":937254,"owner":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-04-02 14:44:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2025-04-02 04:40:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2025-04-02 13:36:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2025-04-02 13:36:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-06 14:14:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-12-06 14:25:31.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2024-12-06 15:02:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-20 10:49:30.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"CC"},{"updated":"2025-02-14 16:16:47.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"}],"messages":[{"id":"4b6633e1ddef002adb4e0a507a917a80e5ad40d5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-12-06 13:59:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a00427ced7beee8bce253945844be9a9a5e47bc6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-06 14:14:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/907e9438e834492d8c094a830b909aef\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/12a52c6c86a34c2a9de1cf6f46fa31b8 : SUCCESS in 12m 44s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2bf211ed7c3e4d6c885b48325df8b358 : SUCCESS in 11m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fb061c006a87e48c5afd6cc77a63ab502f35711e","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-12-06 14:25:31.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b1c7cad02ae83c5a7f10b027e683ef667d40920b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-06 15:02:16.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5217f9e7f6e14dd188cd6054b8b73aee\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d47f4379eac945bd94a25fcaa8d0da27 : SUCCESS in 52m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dba82be8e74247288628057834de7e67 : SUCCESS in 1h 01m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0206ae61792949a2815e8b8e4e5a7571 : FAILURE in 5m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cef46eddf3404041b1648f608a916c45 : SUCCESS in 3m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d192eeee32754eb798635a587d5f9b88 : SUCCESS in 7m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5fb1c1d0ba7495c9f325370227ca295 : SUCCESS in 5m 40s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f4b7d12b4bc44d82b7bf6ede6300964c : SUCCESS in 10m 26s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9cf5e830668c4c218b0475068c66f258 : SUCCESS in 7m 23s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/b94492aa0dab48c9a1ade2a16787beb5 : SUCCESS in 12m 50s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/6ccb9a213dc043b68cedcc082d3b69e2 : SUCCESS in 36m 05s","accounts_in_message":[],"_revision_number":1},{"id":"121b55924485fda96f08a8bcbafb9e35c51c9dd4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-12-06 15:22:07.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"5ba4995e8c9ee00c6f28782dc1961594d3fbdd3c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-06 15:42:36.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fce81f90927c46c1851f785bf14066cf\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0670ddaf810f4477b5802dfe3c0b495b : SUCCESS in 11m 19s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b6ffc66f13024a069ece62d9af35be3d : SUCCESS in 14m 23s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"dceee1b38d70ddffd41b364afcf01dfeb66e39eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-06 16:55:19.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/558f303f843848ebb12bcae82e2e1d23\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/596528920e6841afa1d204fe76984523 : SUCCESS in 54m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9218cf34b5f245758251d72204819efb : SUCCESS in 1h 23m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40d4aab60598490182adaa3d525109de : SUCCESS in 5m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/730c4344eb3542259a28d03c7d7800c8 : SUCCESS in 22m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6e38a4644ea14f918f39ce87813e6679 : SUCCESS in 23m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/437434b695c64f6ea916050cc6171fe4 : SUCCESS in 24m 25s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/369e842936da47efb932bb4a952a6b26 : SUCCESS in 28m 28s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/000b3ff9b7fe48a6974aabf32948626f : SUCCESS in 26m 25s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/31f2567565f74510954bbbaaa4dc5fe6 : TIMED_OUT in 31m 18s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/7f484a81812f4ebba81c99d86d2994a9 : SUCCESS in 59m 01s","accounts_in_message":[],"_revision_number":2},{"id":"6aced7dba78a1be495ee222706ea2fbcdc50f548","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-01-07 14:03:15.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"7362f2d1d0bb4c33949a6335818f1abefaef02b0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-07 14:15:34.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2d4ba176ffd34ba690f3f1661026833c\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d18f340677fd4690840b6886dc88e1f6 : SUCCESS in 9m 47s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/cea21ea41c6f4855909a55bfa573cfaf : SUCCESS in 9m 51s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"df08412e22f0f2150281a2fd1d7b5fa390e2cb49","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-07 15:12:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/69a66475cfbb4382a76c4e60c0f613dd\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/41cbccc2b7cc49839f3b9e9f2ee1de28 : SUCCESS in 1h 02m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dfa1b29fe78640768582096891531197 : SUCCESS in 1h 02m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45505669ecb14de382cebfff1209f6b8 : SUCCESS in 6m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/63f7ee126c0943d4b71b04aa33aa41b8 : SUCCESS in 5m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/41aed9c655654834bb000f7d5e5577b7 : SUCCESS in 5m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28e42dac11274e8394c41d6b0c9c2e20 : SUCCESS in 5m 34s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/21ed25b7340f4024b2fe0dc9332f9e1a : SUCCESS in 9m 03s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/fc836d0664e6462db0f4355d5298b516 : SUCCESS in 10m 22s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/7f700232c35049cc811b849f7acf3401 : SUCCESS in 10m 32s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/8460255b8c6c458aab5fc9f0162817c7 : SUCCESS in 42m 42s","accounts_in_message":[],"_revision_number":2},{"id":"991c95e204dcd6c911d95fc11bfd8f0aeeb366cc","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-01-08 12:04:15.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"02faae8251d81085c1b6a6ce4c239f02db787975","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2025-01-20 10:49:30.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"489e4e5dba37592577ec54eee285d4cc934b3fce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-02-05 19:37:30.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"009af32f5b7d1300f94831c618a4fcd0863bcf3c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-05 20:02:51.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b87c692752164367869b88fff5129a9d\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c3d5f5efbff346d195950926850abae1 : SUCCESS in 9m 07s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f75b8f4a23fb4d098bc0f1bd30d2a524 : SUCCESS in 10m 42s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c72e688001250cc8342c37a94fb6e87b368ee177","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-05 20:46:53.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b8edff2435834188b8ae1056b73ceb72\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/59e60a6c038e445a944287aa82ebc44e : SUCCESS in 1h 03m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6c5874b0a0984c60b6d71f983be014a6 : SUCCESS in 1h 03m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26f7bfbd661d4b2897b2b3455a471d28 : FAILURE in 5m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0b9a349f1b744cdcb84766ddda3f8f0d : SUCCESS in 5m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0a137a61baa24baeb12ad02c4e024398 : SUCCESS in 6m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54cc1b5f920e4bc2a2b0c13701ee4d70 : SUCCESS in 5m 03s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a34a66de09e0450f877f9853257a68a9 : SUCCESS in 7m 10s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8be1f822108845d29237e00d26c9076e : SUCCESS in 8m 23s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/ce0de482be0a411f8a117bdfa4c0f13f : SUCCESS in 13m 36s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/a757aa0ecbc44155bb8c3461bca1e01a : SUCCESS in 39m 27s\n\nWarning:\n  Comments left for invalid file glance/tests/functional/db/base.py","accounts_in_message":[],"_revision_number":3},{"id":"812b01e02f27d81997db675d98a1d0b7ee8c2e6d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-02-06 16:12:30.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"2005af30951d3f2fac7f19ae7920481730466761","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-06 17:19:04.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a693700493eb42e088cb876585726bc0\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b13d3ae891bf4300b8f97291ff676f93 : SUCCESS in 51m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a07e7b22d32b4a1186267fb526e75002 : SUCCESS in 1h 03m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15d3bb71a0d64285842905e05264deca : SUCCESS in 5m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f9c306858e7148dbb40e6e9c066bc18c : SUCCESS in 7m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d5d5443524a3473bbbedd41f303c5763 : SUCCESS in 7m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38884061911349b2803553076b74055b : SUCCESS in 6m 19s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/8bc7c40c75524b5fb3275ca7aa64f801 : SUCCESS in 9m 22s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c0898d226bb04f60908280970bbb5bf1 : SUCCESS in 9m 33s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/4da471c325b445e3a99dc66e4a90d9fe : SUCCESS in 13m 29s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/791c75db41c643a28959aa7f37e582de : SUCCESS in 29m 37s","accounts_in_message":[],"_revision_number":4},{"id":"1bc2cb018e422abfe74ac1b23757d8d242b12bed","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-06 17:54:21.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e991eea93834bfea2612f209a7fc34f\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7f8acf2ce5e648cf8207fdf6254cd6e1 : SUCCESS in 10m 45s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3cc42aba0f274dc9bf91c1f45d2b0b09 : SUCCESS in 14m 04s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"02ab66b542eb4fcd34b4c2a54176fe1b0e7e2c23","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-02-11 16:47:40.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"12c851d42424ac96f2ead41f0c873d0e334a7c14","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 16:58:43.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a8292dc1057c43a49741d8a64ef22a57\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5bf3d7e5a0264c5dab0b9c45312d7e48 : SUCCESS in 9m 11s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6c7a45cd421d45d090f93d4f4f4b8ed5 : SUCCESS in 8m 34s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"56797cc999eeac99f0abe01fcb4d0a6afc0ca5ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 17:44:16.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f638ae084b04451c970d85e5dc9793fb\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/40ee93e161794bc5bbe361e21a525763 : SUCCESS in 45m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a4707d59fb334acbba496966af078476 : SUCCESS in 51m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e114379bf9a4e3b9de64675ee49eb08 : SUCCESS in 4m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b3ccf7870175473abc31b0c388c87204 : SUCCESS in 5m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3e2ef370418147ae975498b9d4dac6e5 : SUCCESS in 5m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a2f0a7521b040b4b0c70e7327bff45e : SUCCESS in 5m 44s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2c4c28345be0460ebb03db3e960a8357 : SUCCESS in 8m 21s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0b0dc5d329314305937b201e0c162bcb : SUCCESS in 8m 12s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/7b1643a09bc049609eaab70bac62552a : SUCCESS in 11m 37s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/e964b4192a094baf84cbeffe59faef50 : SUCCESS in 40m 51s","accounts_in_message":[],"_revision_number":5},{"id":"a066860d86bcf1817a94648f0b8d4bad3c38dc48","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2025-02-14 16:16:47.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"18b728eacf2d867c9e8f1e4e5e702616db380f3a","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-03-27 19:17:19.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"b8784aae305d5791986708645c0ccfc99d1c8b91","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-03-27 19:17:27.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"277faf5f4e0832bc39adc056feb7030e181f1a0d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-27 19:25:39.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/05c0f6bc9af34d09933a6b8e0beded5d\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0253d40bb8e045bd9b420347693c9fc2 : SUCCESS in 6m 04s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/525b2046cc804843accf4ecb7c53cd91 : SUCCESS in 6m 20s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"72295b0ab0f7bdac28448931a3da0eab2d87be96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-27 20:16:14.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4291eaf33fad4272ab95df4d705b94ab\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a94e29534af54fb9841e66ddc20c5afe : SUCCESS in 30m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b0acb869f3564ebba2f37f61a569311f : SUCCESS in 57m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be5ec1f63e53414e9aae984214f549db : SUCCESS in 2m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/60ad1982feef4e958e606c53a5d1b879 : SUCCESS in 3m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ef4fdd0872d047579176ef5ddfb455ba : SUCCESS in 5m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31c1f738ba74409d9d5fe6148c947f94 : SUCCESS in 2m 48s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/577f18ead1ac4c89954b725ae18f5e98 : FAILURE in 6m 36s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ddba3fde4c3046b380bbb03c4363cc37 : SUCCESS in 8m 00s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/fd3f77ee52664d629aaf77d642cb2286 : FAILURE in 7m 49s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/521c66fbf3904e0d889556fa11b629a8 : SUCCESS in 17m 30s","accounts_in_message":[],"_revision_number":6},{"id":"096adc4b38e07ffa3c32a6147088bb9519fe1008","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-03-28 03:55:11.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"e8852677dbd44efda0e376346358e33c6b2e9c46","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-28 04:03:02.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1bb412e59af14d338ae7a520994d59d2\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d9f16f7a8f8d45f9bd5987846f6cc663 : SUCCESS in 6m 00s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/32a8b13ff5e04b5bbb98d70728aef7b5 : SUCCESS in 6m 01s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"240ad674109f1796dea3c0de530008090dba01f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-28 05:05:56.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a729c1d4bb81436f82a0c9b578c11bab\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6dfc5b9e64174dcda5b6bee2ffabf7dc : SUCCESS in 1h 00m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0789049f44ac4eda9f73c183f60ef326 : SUCCESS in 57m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48761f8611a24408943b334b51d9e157 : SUCCESS in 4m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/92ffb00297044cc4b6bb62c97405eb75 : SUCCESS in 4m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b7932fa8d67641c59b15e0fbcfd341d2 : SUCCESS in 5m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a32cea1145aa4db68e33387ad566933c : SUCCESS in 5m 24s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/92ba00d1551d41c992c806c3afbfb3be : SUCCESS in 8m 13s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7042ecc3e58b484681f851cd7a483572 : FAILURE in 5m 50s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/92919a5cd7c34fcc8875ab53ee83c455 : SUCCESS in 13m 37s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/3c47983251544a4baebc39c5d9c3b872 : SUCCESS in 26m 25s","accounts_in_message":[],"_revision_number":6},{"id":"05d99675bc51497e5a96acaecb2f7b957de2f244","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-03-28 05:46:43.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"6b4051f0ab3cd00a1c291d7e94157781f9acab6e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-28 05:57:02.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ea15fc57ca5d47b9ba6b50dcffa4fbec\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/49ba4610314c4c80b1a873c0030a1456 : SUCCESS in 6m 13s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/34cee01bc1944ff686be3436f4bcbe23 : SUCCESS in 8m 24s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"4fe6327886cab05084d4236190d798e173eeb7a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-28 06:29:29.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b217f082e85457fa749d101841aeaee\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f535e4cde77d4094b163669af0a714f8 : SUCCESS in 41m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a38bf7b085ab4e738a1cdcece883c680 : SUCCESS in 30m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fcc3d7a0a5944a0abe6bacad576f0e26 : SUCCESS in 4m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9a7dbcb5972b46dea65aa178aca57951 : SUCCESS in 2m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6ca330b2b76f4ce4a648b3bdcb1b842d : SUCCESS in 4m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b6332ec5f84405d998eec2a574e0a0e : SUCCESS in 2m 55s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/298a556413064c2795567d8775144c01 : SUCCESS in 7m 50s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3c77ae73ece24c0ea021486e8568d0f0 : SUCCESS in 7m 41s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/080ea8f4e51b4abca2d758c90dacd559 : SUCCESS in 9m 05s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/643be968e2884372aa8b74173b3be76e : SUCCESS in 21m 36s","accounts_in_message":[],"_revision_number":6},{"id":"d27524ea0cce48ae314866606169ebe24b955723","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-04-02 04:40:53.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"9b5eac09adda6f71be78528f124da2d1ea5ebc49","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2025-04-02 13:36:42.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"f41498036b94bb9cbd0f72fa99351d91e450cd6d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-02 13:37:32.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"3127c361143b1b2ebcb22d6ce4a40e641e1180e4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-02 14:44:51.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0656a254f794418ba71ed31bdfbe06a5\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/81b1c180d3b64029bb72c4ddd375167a : SUCCESS in 55m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a4950156356749e8904f86bbde1bb0fb : SUCCESS in 59m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e603cfc4cacd402cb4e292fd910b56a2 : SUCCESS in 5m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6fa7a5514a824cbc831f306b6bb073aa : SUCCESS in 5m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/118c065d283c4688bc3cc1f1cede6c45 : SUCCESS in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1eece5e37ca455280f78932ee80032f : SUCCESS in 7m 44s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b405c1952b2f4889840f5465a90d1bfc : SUCCESS in 7m 28s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/932f4f911d9f4e4881cf1bcc00766650 : SUCCESS in 8m 30s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/b46413b4dba7453d956d3982ba7146ff : SUCCESS in 11m 14s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/01707a114bb945b085eb14a3bcea414e : SUCCESS in 40m 53s","accounts_in_message":[],"_revision_number":6},{"id":"d9a14017e13c09c5d40086e9d8f12ad2532bde79","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-02 14:44:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"752f3ccffa1c75dbaa77dcf161e5c63d2136262b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-02 14:46:24.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d75e908d76444a9aaa0de9935ea93cae\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/973552065f8944b5b6b929f1d8f70272 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"3ed84f79ef00b30ec5874625034bfabc304e4f3e","revisions":{"dc0b61dffbb90c6a87185f276a32d2c51849991d":{"kind":"REWORK","_number":1,"created":"2024-12-06 13:59:39.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/54/937254/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/54/937254/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/54/937254/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/54/937254/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/54/937254/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/54/937254/1"}}},"commit":{"parents":[{"commit":"94874d329cf4c2897740de29827271f96a663bcd","subject":"Functional test migration: test_tasks.py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/94874d329cf4c2897740de29827271f96a663bcd"}]}],"author":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2024-12-06 13:57:44.000000000","tz":60},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2024-12-06 13:57:44.000000000","tz":60},"subject":"Functional tests migration: test_healthcheck_middleware","message":"Functional tests migration: test_healthcheck_middleware\n\nUse SynchronousAPIBase instead of FunctionalTest as the base class for\nthe HealthcheckMiddlewareTest class, so we can run the tests it defines\nwithout running eventlet or running a \"real\" WSGI server.\n\nChange-Id: I07aea6a11e9ff34127e96129cbf610703e8d90ee\nCo-Authored-By: Abhishek Kekane \u003cakekane@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/dc0b61dffbb90c6a87185f276a32d2c51849991d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/dc0b61dffbb90c6a87185f276a32d2c51849991d"}]},"branch":"refs/heads/master"},"bd1744a67f4b449bf3ff7cac34462808e27fc7e3":{"kind":"REWORK","_number":2,"created":"2024-12-06 15:22:07.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/54/937254/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/54/937254/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/54/937254/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/54/937254/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/54/937254/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/54/937254/2"}}},"commit":{"parents":[{"commit":"94874d329cf4c2897740de29827271f96a663bcd","subject":"Functional test migration: test_tasks.py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/94874d329cf4c2897740de29827271f96a663bcd"}]}],"author":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2024-12-06 13:57:44.000000000","tz":60},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2024-12-06 15:21:55.000000000","tz":60},"subject":"Functional tests migration: test_healthcheck_middleware","message":"Functional tests migration: test_healthcheck_middleware\n\nUse SynchronousAPIBase instead of FunctionalTest as the base class for\nthe HealthcheckMiddlewareTest class, so we can run the tests it defines\nwithout running eventlet or running a \"real\" WSGI server.\n\nChange-Id: I07aea6a11e9ff34127e96129cbf610703e8d90ee\nCo-Authored-By: Abhishek Kekane \u003cakekane@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bd1744a67f4b449bf3ff7cac34462808e27fc7e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bd1744a67f4b449bf3ff7cac34462808e27fc7e3"}]},"branch":"refs/heads/master"},"ff90c0ef9cc2b4d5d52ccda685f161b1f24506ed":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2025-02-05 19:37:30.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/54/937254/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/54/937254/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/54/937254/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/54/937254/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/54/937254/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/54/937254/3"}}},"commit":{"parents":[{"commit":"e0e52f042ec759de5b9915f540b75559d73afc7d","subject":"Functional test migration: test_tasks.py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e0e52f042ec759de5b9915f540b75559d73afc7d"}]}],"author":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2024-12-06 13:57:44.000000000","tz":60},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2025-02-05 19:36:18.000000000","tz":60},"subject":"Functional tests migration: test_healthcheck_middleware","message":"Functional tests migration: test_healthcheck_middleware\n\nUse SynchronousAPIBase instead of FunctionalTest as the base class for\nthe HealthcheckMiddlewareTest class, so we can run the tests it defines\nwithout running eventlet or running a \"real\" WSGI server.\n\nChange-Id: I07aea6a11e9ff34127e96129cbf610703e8d90ee\nCo-Authored-By: Abhishek Kekane \u003cakekane@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ff90c0ef9cc2b4d5d52ccda685f161b1f24506ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ff90c0ef9cc2b4d5d52ccda685f161b1f24506ed"}]},"branch":"refs/heads/master"},"e2d1980f0a59678072a94d3ddc9f687a053f9676":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2025-02-06 16:12:30.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/54/937254/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/54/937254/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/54/937254/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/54/937254/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/54/937254/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/54/937254/4"}}},"commit":{"parents":[{"commit":"a3c65f8981c80c043d8f105f9aaf4e4855b6b04c","subject":"Functional test migration: test_tasks.py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a3c65f8981c80c043d8f105f9aaf4e4855b6b04c"}]}],"author":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2024-12-06 13:57:44.000000000","tz":60},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2025-02-06 16:12:22.000000000","tz":60},"subject":"Functional tests migration: test_healthcheck_middleware","message":"Functional tests migration: test_healthcheck_middleware\n\nUse SynchronousAPIBase instead of FunctionalTest as the base class for\nthe HealthcheckMiddlewareTest class, so we can run the tests it defines\nwithout running eventlet or running a \"real\" WSGI server.\n\nChange-Id: I07aea6a11e9ff34127e96129cbf610703e8d90ee\nCo-Authored-By: Abhishek Kekane \u003cakekane@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e2d1980f0a59678072a94d3ddc9f687a053f9676"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e2d1980f0a59678072a94d3ddc9f687a053f9676"}]},"branch":"refs/heads/master"},"4933303cd90f7f058318dcd120cb146a85a87436":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2025-02-11 16:47:40.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/54/937254/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/54/937254/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/54/937254/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/54/937254/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/54/937254/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/54/937254/5"}}},"commit":{"parents":[{"commit":"01686b80f23ef5b23f418c879d865e9765aaacf3","subject":"Functional test migration: test_tasks.py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/01686b80f23ef5b23f418c879d865e9765aaacf3"}]}],"author":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2024-12-06 13:57:44.000000000","tz":60},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2025-02-11 16:31:15.000000000","tz":60},"subject":"Functional tests migration: test_healthcheck_middleware","message":"Functional tests migration: test_healthcheck_middleware\n\nUse SynchronousAPIBase instead of FunctionalTest as the base class for\nthe HealthcheckMiddlewareTest class, so we can run the tests it defines\nwithout running eventlet or running a \"real\" WSGI server.\n\nChange-Id: I07aea6a11e9ff34127e96129cbf610703e8d90ee\nCo-Authored-By: Abhishek Kekane \u003cakekane@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4933303cd90f7f058318dcd120cb146a85a87436"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4933303cd90f7f058318dcd120cb146a85a87436"}]},"branch":"refs/heads/master"},"3ed84f79ef00b30ec5874625034bfabc304e4f3e":{"kind":"REWORK","_number":6,"created":"2025-03-27 19:17:27.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/54/937254/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/54/937254/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/54/937254/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/54/937254/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/54/937254/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/54/937254/6"}}},"commit":{"parents":[{"commit":"1179ce80caa525f8e6268c56c20e89972dcfd95a","subject":"Merge \"Fix functional tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1179ce80caa525f8e6268c56c20e89972dcfd95a"}]}],"author":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2024-12-06 13:57:44.000000000","tz":60},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2025-03-27 19:16:20.000000000","tz":60},"subject":"Functional tests migration: test_healthcheck_middleware","message":"Functional tests migration: test_healthcheck_middleware\n\nUse SynchronousAPIBase instead of FunctionalTest as the base class for\nthe HealthcheckMiddlewareTest class, so we can run the tests it defines\nwithout running eventlet or running a \"real\" WSGI server.\n\nChange-Id: I07aea6a11e9ff34127e96129cbf610703e8d90ee\nCo-Authored-By: Abhishek Kekane \u003cakekane@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3ed84f79ef00b30ec5874625034bfabc304e4f3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3ed84f79ef00b30ec5874625034bfabc304e4f3e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
