)]}'
{"id":"openstack%2Fglance~937580","triplet_id":"openstack%2Fglance~master~I46cab5543226446da08ef4e04700352c076e2e62","project":"openstack/glance","branch":"master","topic":"scrubber-removal-split","hashtags":[],"change_id":"I46cab5543226446da08ef4e04700352c076e2e62","subject":"[WIP] Remove glance-scrubber: remove code","status":"ABANDONED","created":"2024-12-11 23:59:27.000000000","updated":"2025-05-12 14:17:06.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ab7b0d283fd924758e5c3590b7edece2d50763e9","_number":937580,"virtual_id_number":937580,"owner":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-12-12 02:10:12.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-12 00:11:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-12-12 02:10:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b06f807b1b106911067f0f76ef7b51c9ed634518","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-12-11 23:59:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c48db832504b56f3a2986884ca9e9aea07d1a0bf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-12 00:11:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1ef1847ed7834eac85e960cbf3ba05da\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/856ec9f254654e9a9240aa8459aea967 : SUCCESS in 7m 32s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/32300353458f47f0b1e4ded0a3eb0408 : SUCCESS in 9m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3fcb105406acbd75fd4552b0548135a37aef6a8e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-12 02:10:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/91d5c8ee5a3d4516bffed0d5155c98c6\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a0c90875e1844407b2f5872a6794fc5d : SUCCESS in 1m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/de1ac56adc11469a9db0231ba25b7ba2 : SUCCESS in 57m 13s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/dcaa8181ebb046ce93c8f67be86b2418 : SUCCESS in 1h 01m 55s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8862dfefc2554334a297d208127abe98 : SUCCESS in 1h 12m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/812b5502248f42768126ed8ffc5cc30b : SUCCESS in 45m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e4892434fdc4a1189402174788a0851 : SUCCESS in 4m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e57ad0d5176b403fa6d540b302bdde0a : SUCCESS in 3m 27s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ffd91abb49e34579abf38c6427844490 : SUCCESS in 5m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dcee0c58072a4cddba2d503955b4bdcd : SUCCESS in 5m 57s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c8c0f0dffe1845ea8398e2712517f7a1 : SUCCESS in 7m 57s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ccd0066e36044b619f916268deb97b62 : SUCCESS in 5m 53s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/3aa754528b8842e980ce1e73bf0c4109 : SUCCESS in 9m 54s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/6c1013ef67924af0a507d9c7cf28d676 : SUCCESS in 2h 03m 43s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/4aac9764a0ce45b6a1816c69f19b7032 : SUCCESS in 1h 23m 08s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/650a7041d06a40089442add7a168f6ec : SUCCESS in 1h 09m 52s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/45746796446246319e6d6362d5a96bd1 : SUCCESS in 1h 58m 10s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/ae3adba8958346789eb87e6c6a0f0866 : SUCCESS in 1h 56m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/50bccc4c0cd5475b88db3ba93a283e3b : SUCCESS in 1h 05m 15s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/55752ca1f34d47dd9aa103f738b4f31f : SUCCESS in 1h 30m 19s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/fc00bae77e464acb881acac7cfe48056 : SUCCESS in 1h 36m 38s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/08248d26ad2f4c54a7571766cef40d7a : SUCCESS in 42m 02s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/243106333a894d2e8529132b80da1d69 : SUCCESS in 1h 37m 34s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/4a9b376c2579466caa1e70ed2f6072e5 : SUCCESS in 21m 46s","accounts_in_message":[],"_revision_number":1},{"id":"ab7b0d283fd924758e5c3590b7edece2d50763e9","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-05-12 14:17:06.000000000","message":"Abandoned\n\nWe will not remove the scrubber in the end","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d6751871e0ff16c260137cee03d1396fa8eaf789","revisions":{"d6751871e0ff16c260137cee03d1396fa8eaf789":{"kind":"REWORK","_number":1,"created":"2024-12-11 23:59:27.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/80/937580/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/80/937580/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/80/937580/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/80/937580/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/80/937580/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/80/937580/1"}}},"commit":{"parents":[{"commit":"6d9d9be49072cdecf693721326c99c2289c391dd","subject":"[WIP] Remove glance-scrubber: remove tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6d9d9be49072cdecf693721326c99c2289c391dd"}]}],"author":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2024-12-11 23:58:45.000000000","tz":60},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2024-12-11 23:58:45.000000000","tz":60},"subject":"[WIP] Remove glance-scrubber: remove code","message":"[WIP] Remove glance-scrubber: remove code\n\nThis is a split version of a bigger patch[1], in four parts:\n\n- Remove tests\n- Remove actual code\n- Remove sample configuration\n- Remove documentation and api-ref\n\nThis might be easier to review, but in the end we may want to\nmerge a single patch.\n\n[1] https://review.opendev.org/c/openstack/glance/+/930241 .\n\nChange-Id: I46cab5543226446da08ef4e04700352c076e2e62\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d6751871e0ff16c260137cee03d1396fa8eaf789"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d6751871e0ff16c260137cee03d1396fa8eaf789"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
