)]}'
{"id":"openstack%2Fglance~937581","triplet_id":"openstack%2Fglance~master~I2df926780e2d1879259ab44cd5d169b7f6bd22bd","project":"openstack/glance","branch":"master","topic":"scrubber-removal-split","attention_set":{},"removed_from_attention_set":{"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2025-05-12 14:17:02.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I2df926780e2d1879259ab44cd5d169b7f6bd22bd","subject":"[WIP] Remove glance-scrubber: remove the sample configuration","status":"ABANDONED","created":"2024-12-11 23:59:27.000000000","updated":"2025-05-12 14:17:02.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7beb2281ac657f25364ab607409519eed1b9afc3","_number":937581,"virtual_id_number":937581,"owner":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-12-12 03:22:05.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-12 00:10:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-12-12 03:22:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"439c5edd1fdc3a54be6bfdfe1137344ab5af4755","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-12-11 23:59:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"91ea5c7cf218a1a6c7b89d136a8b1b4820e1fd19","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-12 00:10:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/695a9c897f774445b866ea08e7983964\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3a7fa5f254cd45e0a1fff0108416738a : SUCCESS in 7m 26s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8562775e6dd840d18f7c445ae85e3afc : SUCCESS in 8m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4253eafb345594e9b6d2f5c077eee9f194061845","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-12 03:22:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2333cdad945b489581f273440d601919\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e2fd55853dfb41d2a31a09be90a5a6b2 : SUCCESS in 2m 35s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/c3dc4128fc2048c3bd5e3ef95be92383 : SUCCESS in 4m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/4998f644ce294f289f80c26f6133344b : SUCCESS in 59m 11s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/61ebba6cacd749f2b22ea31bd154b69b : SUCCESS in 32m 28s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9d9b3352b0c74c5cb54a96ce1a8b4fda : SUCCESS in 1h 26m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cf2dc2e59d4149d0865eb3e56c63cc99 : SUCCESS in 47m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad2df2073c9b4a73a87fd35cae999400 : SUCCESS in 4m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8ab2aee9346d45c79e1536709a085643 : SUCCESS in 4m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d6335b155d514c52b42b27ef718978d3 : SUCCESS in 5m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c92ae96993c946bab2baaf881e44e691 : FAILURE in 2m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/050e545324f44595a775bcfa3a76b4ed : SUCCESS in 7m 13s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e1e1fe3507a9485a86e088f308d6f98c : SUCCESS in 6m 40s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4d4d47b6ac0b4ea1b056bdf3d1829ea8 : SUCCESS in 5m 54s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/a528901a02c249c69391593253263677 : SUCCESS in 9m 47s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/5e2ca0427dad437e89550370daa1bd6b : SUCCESS in 1h 38m 32s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/cb1a25284b904d969d404cc89b942bcf : SUCCESS in 1h 24m 04s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/fe96e6595f834130b18a968d098a19ff : SUCCESS in 1h 48m 39s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/8b6982ebec8242eca1e97a8001bfbaf4 : SUCCESS in 1h 04m 48s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/7b0305838315471abc1aa05e028bf10d : TIMED_OUT in 3h 09m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/45a9732f14194dc78926a598125b3fb9 : SUCCESS in 57m 41s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c1f04213d4cb456dbd6cd483725ac8b8 : SUCCESS in 1h 47m 23s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/09e972705e594c0b841e1bdf89369afe : SUCCESS in 1h 37m 03s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/63338a2d9a5f48d6928278023e624e51 : SUCCESS in 44m 49s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/a509f7d0d5d54b0a9b44911fb6e2f918 : SUCCESS in 1h 21m 56s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/a40de01a358e469f9b3e27029e22d14a : SUCCESS in 22m 23s","accounts_in_message":[],"_revision_number":1},{"id":"7beb2281ac657f25364ab607409519eed1b9afc3","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-05-12 14:17:02.000000000","message":"Abandoned\n\nWe will not remove the scrubber in the end","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b22b90e11e44ccf1b629cfa6c246af1325072f82","revisions":{"b22b90e11e44ccf1b629cfa6c246af1325072f82":{"kind":"REWORK","_number":1,"created":"2024-12-11 23:59:27.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/81/937581/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/81/937581/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/81/937581/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/81/937581/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/81/937581/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/81/937581/1"}}},"commit":{"parents":[{"commit":"d6751871e0ff16c260137cee03d1396fa8eaf789","subject":"[WIP] Remove glance-scrubber: remove code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d6751871e0ff16c260137cee03d1396fa8eaf789"}]}],"author":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2024-12-11 23:58:45.000000000","tz":60},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2024-12-11 23:58:45.000000000","tz":60},"subject":"[WIP] Remove glance-scrubber: remove the sample configuration","message":"[WIP] Remove glance-scrubber: remove the sample configuration\n\nThis is a split version of a bigger patch[1], in four parts:\n\n- Remove tests\n- Remove actual code\n- Remove sample configuration\n- Remove documentation and api-ref\n\nThis might be easier to review, but in the end we may want to\nmerge a single patch.\n\n[1] https://review.opendev.org/c/openstack/glance/+/930241 .\n\nChange-Id: I2df926780e2d1879259ab44cd5d169b7f6bd22bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b22b90e11e44ccf1b629cfa6c246af1325072f82"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b22b90e11e44ccf1b629cfa6c246af1325072f82"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
