)]}'
{"id":"openstack%2Fglance~937844","triplet_id":"openstack%2Fglance~master~Iabd3efbebf7534dd6b7ebdf961bdff6804710c6c","project":"openstack/glance","branch":"master","topic":"remove-py311-tips-jobs","attention_set":{},"removed_from_attention_set":{"19138":{"account":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"last_update":"2025-01-16 19:15:35.000000000","reason":"Change was submitted"},"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2025-01-16 19:15:35.000000000","reason":"Change was submitted"},"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2024-12-18 05:18:38.000000000","reason":"\u003cGERRIT_ACCOUNT_9303\u003e replied on the change","reason_account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2025-01-16 14:11:22.000000000","reason":"\u003cGERRIT_ACCOUNT_8122\u003e replied on the change","reason_account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}}},"hashtags":[],"change_id":"Iabd3efbebf7534dd6b7ebdf961bdff6804710c6c","subject":"Move all periodic tips jobs from py311 to py313","status":"MERGED","created":"2024-12-17 08:39:49.000000000","updated":"2025-01-16 19:19:15.000000000","submitted":"2025-01-16 19:15:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"937844-remove-py311-tips-jobs","meta_rev_id":"6875746f75a443678a0ecbb024a9372c270617b8","_number":937844,"virtual_id_number":937844,"owner":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-01-16 19:15:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":2,"date":"2025-01-16 14:11:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":1,"date":"2025-01-16 14:11:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-17 08:40:42.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2024-12-17 08:40:42.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2024-12-17 08:40:42.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2024-12-17 09:13:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-12-17 13:25:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f9f3d23ee4bcbf59c107f8fa436ae381bdac9fd8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-12-17 08:39:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1befa6f91788fc17f8273d07de8596593593a195","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-12-17 08:43:28.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2f5973fc7fd2b2b964451e22b87932bd49f54a55","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-17 09:13:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a6e7fe3411dc40939ae7da89318ce78b\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f5813c7c9335401c9484d4bae91fde16 : SUCCESS in 15m 24s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c970064424cf43768629aa2f5add156d : SUCCESS in 17m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ab4b53ca20d02e6c659316f348c05df258416487","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-12-17 10:56:59.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6bf8078fb59130f852900049a6cd40ca02ca8caa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-12-17 11:05:11.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5c1fb59e83ccafb26cdb8637cef82ee51f95115d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-12-17 11:12:04.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"7f2d7ca2b7bd9f05c7e4d60f8b3be0a9875ce248","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-17 11:43:34.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ad9831dc47b4b97ac97493e672af59e\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e42b2c070b464570a54a8b1a76a80201 : SUCCESS in 10m 27s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/dee82f0cf15c474fb11bf9c0de1f4a1a : SUCCESS in 10m 02s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"de317dfadcc23d09ac1db1f589a88f7da27bd999","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-17 13:25:29.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/dae5f8679bb4494094a363c49ee57460\n\n- grenade https://zuul.opendev.org/t/openstack/build/36c84d9e236a45b2a82e743e23959a2f : SUCCESS in 1h 07m 11s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e1f36c4c6b6245ad9f267d3cc90e6250 : SUCCESS in 1h 00m 36s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2603f626caca494a8d28942981cd5d16 : SUCCESS in 1h 40m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/15f263e8249949d2b361e35abffdbd89 : SUCCESS in 56m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8718271720a24ada998697be914549c1 : SUCCESS in 3m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c05c4cf486864caca2f51619e35683f3 : SUCCESS in 5m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4bf09379a89549e69ba48bb0d15be73f : SUCCESS in 5m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4adf272453ed45099014b2baf942c37b : SUCCESS in 4m 58s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/21e8432886c142199a2aa318b3dc90ca : SUCCESS in 7m 48s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/97df847075eb4067ab1d63fd15e162d5 : SUCCESS in 8m 26s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/0d15f01f57e14c1a977f6a1470be5374 : SUCCESS in 9m 23s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/9a1082a240fc4555ba75291652b95e21 : SUCCESS in 1h 55m 52s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/57a2b7ced9ea44c99b052463f135c98e : SUCCESS in 1h 21m 01s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/bdba5f4b735446d09dc2939d2fd6e856 : SUCCESS in 1h 58m 08s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/19c337cdd1024cc7a743da918caf279b : SUCCESS in 1h 58m 30s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/2164893119414a8f9bb77455e823c46c : FAILURE in 2h 01m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5674d411a152434a82fd6b3c16e98a6a : SUCCESS in 58m 55s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/86176d8893b04b5cb23af02a57dbb6cb : SUCCESS in 2h 02m 31s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/2d0f1715b2a047219a01c310025ec461 : SUCCESS in 1h 22m 39s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/d1bb42f71e2a46709cd25f02d7cbfbe3 : SUCCESS in 35m 15s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/2c722eff4da943b0b4f733f08204da18 : SUCCESS in 2h 00m 02s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/65587341a700449b83cf4f61cf25cbce : SUCCESS in 28m 55s","accounts_in_message":[],"_revision_number":3},{"id":"b204aa9375080bb436e4feeee1a51f6c43542462","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-12-18 05:18:38.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"9a46a9940dbeda59fbc14475061a169421edaf0b","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-12-18 05:18:49.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"168e236123fff8aa42c4f8b102afa750f8b4e49f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-18 05:34:48.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2a9d430fe8534f02a39599c51f12a2be\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b6703e4c6a224b8ca47799257645f41f : SUCCESS in 9m 28s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e234f105cc0d4b068c16ed47f41c744a : SUCCESS in 13m 14s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8d324827979098e9fcb1db8939af484bf5a942f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-18 07:21:04.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cea50749742048fe8ead6564afc0c83e\n\n- grenade https://zuul.opendev.org/t/openstack/build/549ab64d3e174b24a94a5b9907bebd4f : SUCCESS in 51m 18s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b1a7e3a7c79642dba034b675f30918ad : SUCCESS in 47m 16s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5ba73c7cd9c74903abc2fe1296c44cb3 : SUCCESS in 1h 27m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/95aa132a9bbb43e38f0c1c6b072a6193 : SUCCESS in 55m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b4aa38e4c584e579ad7ac775dd57a4e : SUCCESS in 4m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3bc609cd74cf4f698ee9f3c9c488ea27 : SUCCESS in 5m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0d4c4892460e4007917f6564c588999a : SUCCESS in 5m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0ce2ae705994d33a4254144d2fe9fdd : SUCCESS in 5m 43s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6c0d1c2196f743bea92610b7c8530089 : FAILURE in 6m 41s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e94dc4a5cce54a808fc3e2563e729612 : SUCCESS in 7m 47s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/3e746fbdcb9e450fb811b868427d3748 : SUCCESS in 10m 50s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/45aa8ee00492478995566bee29d35395 : SUCCESS in 1h 25m 13s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/37031bd1111d4906b525c56592343263 : SUCCESS in 1h 18m 26s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/e675a2e5204943d0becf8f1bacc43f9d : SUCCESS in 1h 51m 36s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/c96b2db0cb4b45939ac53384b57f488b : SUCCESS in 1h 35m 47s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/51602834e0ba49e0a08a2a7d475b84fe : SUCCESS in 2h 01m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8ac1d8ba363646abb0309daeb6fec3ff : SUCCESS in 41m 08s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/91ba39d515484825a4a1e5cdc8548119 : SUCCESS in 1h 50m 09s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/b4a34d2e999049e88a681bac157528cf : SUCCESS in 1h 42m 33s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/976c5bb122f046a887183997b0332c7e : SUCCESS in 46m 01s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/70d6d43f8a90490faaba4e76e1ae0008 : SUCCESS in 1h 31m 33s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/c00f5929ac834b148a680ee300cb7afb : SUCCESS in 42m 15s","accounts_in_message":[],"_revision_number":3},{"id":"36be695717060f68ce4c2a7d7de366f3a7854479","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-12-24 06:19:21.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"56fb43ac1a08ade6e1da5d52aee1bbdcd31da8e2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-24 06:31:55.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff2e427d698c4b62a48efec665db4e10\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a27c05c8d7d349f39dffeeb24248b0d2 : SUCCESS in 10m 30s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ab8a15336a1a4ea4a061bdf63a067cb5 : SUCCESS in 10m 17s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d819e921f3f31018f0660349fd1d469106539f64","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-24 09:07:17.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/576b7616567c46d1b40100623f13ac8c\n\n- grenade https://zuul.opendev.org/t/openstack/build/46146710665349b7b16cef6f1d2f90ea : SUCCESS in 1h 01m 58s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/78b13d2a21b44a5891beb4292d754f8f : SUCCESS in 1h 03m 42s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c6d9d480c04041d6b87ff04094d6e00c : SUCCESS in 1h 39m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2bb02d482ea44be4b55cd98076d261e1 : SUCCESS in 1h 11m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3caeeb827b734faba903d1ac985ac323 : SUCCESS in 4m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/57ff3fb3f5224957be10ff853fa5c9a6 : SUCCESS in 7m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9a643851164d4180bf7023b2566693dc : SUCCESS in 6m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a85388be63a4dd6957a7164412e31c5 : SUCCESS in 4m 27s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/94a54af0945e4ef7985f4a971952c23c : SUCCESS in 9m 05s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2f416212c1634a15838b15e0ec7e2331 : FAILURE in 6m 38s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/e8d78775b4db492d80eb74fa4edea222 : SUCCESS in 9m 29s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/3d866cd13bf94c768a33fbb649b51daf : SUCCESS in 1h 56m 20s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/b6fe26ca42394675b6e6dcfce36e90c7 : SUCCESS in 1h 49m 25s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/baf4cf4d77a14deb88409bc905c4803e : SUCCESS in 1h 49m 14s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/64a8ec9589704618ac4add6ab7645907 : SUCCESS in 1h 25m 59s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/93bf4580ad7444b2a561a97346b43ecc : SUCCESS in 2h 00m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/55f9a605b8044ecf8bd3f3a936d1f211 : SUCCESS in 58m 32s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f3e5e9bffabf4291baf4c58942d6ef60 : SUCCESS in 1h 06m 27s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/6948931379ba4fc1b4a14097fe239809 : SUCCESS in 2h 06m 43s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/0c1e3d4b883e4227bda0b30af643ad99 : SUCCESS in 42m 28s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/4350058670ad43df941b07556637cc94 : SUCCESS in 1h 34m 41s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f958974659d5495d9db1bab97c9e00b3 : SUCCESS in 21m 48s","accounts_in_message":[],"_revision_number":3},{"id":"55cbd0c30bbb053dcf7daa93ef1b30719e977ea5","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2025-01-02 12:24:07.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"2ebbdacc20732cfdbfff6409397be75ba1570f7c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-02 12:39:59.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa4f2efc7b444f0e800ea768dd6e870f\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6e2ef75b6da84060b5cb0c666b0cb5c2 : SUCCESS in 10m 32s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/fc12978ffe1b48d7bc319b8808491e65 : SUCCESS in 12m 53s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a6e838ab677ce11d9cb9157de269ef71f9672fef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-02 14:53:34.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/202fabe77bcb4a6b9e6136e958ec4d87\n\n- grenade https://zuul.opendev.org/t/openstack/build/0d34c31846b34d6ea0d3a2bb04a149da : SUCCESS in 35m 23s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/86a909c4670f4f5f8f11929c7ac86bb5 : SUCCESS in 35m 57s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a10f98c2bbbd4adb878e0859dee0931e : SUCCESS in 1h 32m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ab335be1a0ed459b92a73d4353b5194e : SUCCESS in 1h 02m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7008a39863c4ed7a20ad825c7e29749 : SUCCESS in 4m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f1a8d0b0af384323acd40e4614f27d90 : SUCCESS in 3m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/197a09486f194643b89abef89469c068 : SUCCESS in 5m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b2433fb2ca74a77838c553cfeb3f6e7 : SUCCESS in 6m 16s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/475d383d4b404323a5bf0468d3a27e12 : SUCCESS in 6m 54s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/349faa55be7a469a9361738d445e56dd : SUCCESS in 8m 39s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/9162b6d862df46448207d8b76f861d0d : SUCCESS in 8m 36s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/25d173a3921e4742aa856a154f7cb64c : SUCCESS in 1h 38m 40s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/652dff01503a46f3b098fb400f272c28 : SUCCESS in 1h 47m 46s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/46f92b01b4f442d8aa6f12a270bc6804 : SUCCESS in 2h 06m 35s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/2f9841533cbe4981974b1ef56e47ffed : FAILURE in 1h 41m 56s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/b3de43f7765b4b229f5100b8086fd10d : SUCCESS in 2h 14m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3343f2ac1172469dbdeae5ea7e23a8be : SUCCESS in 56m 32s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/aa03860fa66a4216b97cc9c5421e194a : SUCCESS in 1h 48m 24s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/0b4db358181f48ca8986939c71bc3709 : SUCCESS in 1h 52m 18s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/55968c79ea7446b89c60dde55bd72a42 : SUCCESS in 44m 49s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/d8c43d4cf9454f9095c16698a760fe0d : SUCCESS in 1h 57m 14s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/9f193ff373c5433fac109cfe80fcd661 : SUCCESS in 23m 31s","accounts_in_message":[],"_revision_number":3},{"id":"85ee95f838e706388b9d9af270c1f28a627b6c64","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-01-07 14:36:14.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"90f89d94fd1f20b1cf37a8c99d52c9dea4136031","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2025-01-09 05:59:15.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"fd514e3c314cbf939d6da441ecdd46d3a2dc09f3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2025-01-09 06:18:56.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"8e44d356e881029ae7671dbf86b9c41c4344afa5","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2025-01-09 06:19:45.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"3a3a0013381eed390d7c67794e36cb04baff1caf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-09 06:33:25.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cde53c2c893d4e1685c988ca4f3b9d6a\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6ba5e4c23f0a4d759c23268e2ded8b1c : SUCCESS in 12m 20s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/437388df26434e0bac36ea0ed20bcdc6 : SUCCESS in 10m 34s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"bc4b301d9f8694bc343a2dc58e691e7a639a3e0e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-09 08:19:32.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a5f70dda07e844edadbf45d581b98bf2\n\n- grenade https://zuul.opendev.org/t/openstack/build/163f5fa804764f58bf22342c8782ac05 : SUCCESS in 1h 02m 53s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a93a591d0d5e48f2aa5638e3e70b5089 : SUCCESS in 1h 01m 57s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8fa1515d2cb2427aa1a305f2c9905645 : SUCCESS in 1h 46m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9044f0884b2e4d968f8b6fcbbbd3659f : SUCCESS in 1h 04m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/652722c71e5040fc9b0c5e435be870a6 : SUCCESS in 4m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9dc33bd4c3c9443aaa84e2df3642fe90 : SUCCESS in 4m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/044bbcac6a7946e4bfb476025814a952 : SUCCESS in 5m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f786c2e16f644daa7f2a2f1e6487ccb : SUCCESS in 5m 31s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2e89d7370bf545d8920cda8f2fd64dc7 : SUCCESS in 8m 55s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/da3457d1da8c4465aff89f0355e5e4d8 : SUCCESS in 7m 41s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/de128be08f9d4d81a8b57dd4c5fdbd6c : SUCCESS in 8m 43s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/46c4af23413d47949e6210837f0afb53 : SUCCESS in 1h 14m 26s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/db5ca4eaf6364990b1b4713c9cf421af : SUCCESS in 1h 13m 08s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/a772f0c7b51143cc8adc0b63acfb8b49 : FAILURE in 1h 08m 21s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/dab19fd72dfd42908281d1f7d382c6bf : SUCCESS in 1h 56m 01s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/ceea4f5401d04bb881ecd1e868a51e4f : SUCCESS in 1h 51m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ed9d3f9d5d8b4bde9dbf907899ee4b80 : SUCCESS in 52m 27s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9aa071f201474e269bafffa540df45c5 : SUCCESS in 1h 35m 41s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/82c4908c108e41a8815abb08154e5a85 : SUCCESS in 1h 58m 35s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/d9de6fcf60d64aaa955004de36a8a2ce : SUCCESS in 41m 02s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/f52342a88836484a9c2121e4683a6bea : SUCCESS in 1h 08m 04s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/c9b1df10857f4c5bb81e5e6e79dd7558 : SUCCESS in 22m 51s","accounts_in_message":[],"_revision_number":4},{"id":"89c583f66795568246ba150aee8ad207cd7d529c","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2025-01-09 08:27:59.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"dd231b5b84ed4113fbd38c5cccfe567dbf197cbc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-09 08:42:50.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a33d32f9120844ec94f717d5bef00782\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fdce20bfd08b45adad0e6b8b602f8ed4 : SUCCESS in 10m 47s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7b1c1cbb786f4a8b96dac718beb262ed : SUCCESS in 11m 57s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"aa109fe5bba8fea93d4ee5d3703317347b64eccc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-09 10:48:26.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1f2582c2ddba46749d1927274561e386\n\n- grenade https://zuul.opendev.org/t/openstack/build/00570a2c9b184863957c429b2ad7d413 : SUCCESS in 1h 05m 26s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/bd01d1fe12b34ed7a3972490ae51fada : SUCCESS in 47m 32s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d79c41cd8a9948fb848f758e926f396a : SUCCESS in 1h 13m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dbff8a203dc9407e948ff44fc2f1b231 : SUCCESS in 1h 07m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16806d99d36744b694e55c78988bdea8 : SUCCESS in 4m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f4af9e7a91c04288a4d8075e769cdb59 : SUCCESS in 4m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d287ace0b94f45fba7331739b617cb3e : SUCCESS in 5m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/291f561b26714576ada032122834efd4 : SUCCESS in 4m 21s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e810e491d2c44cf3bdb24c93886e85ca : SUCCESS in 8m 42s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ce85465d7e1144ecbd54174d1c7baf72 : SUCCESS in 7m 50s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/51f4f3a56bf6491389bca59d2abb383e : SUCCESS in 9m 32s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/f5222addcf7a41d18742d7c53d5e9efa : SUCCESS in 1h 52m 46s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/220b47539bba4a2abb64d6cafb82e140 : SUCCESS in 2h 19m 36s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/63fd38ed605744f883745c1834d66c11 : SUCCESS in 1h 59m 05s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/20553564fefe487cbda2c69ef8cbe065 : SUCCESS in 2h 13m 55s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/0fae351e75ae4426bcd793b99a252ea0 : SUCCESS in 2h 11m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c153aa01a60a4b0a9b60a802691ff6ba : SUCCESS in 54m 53s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/6a48ae62d97e4ea5b0283f9afc438123 : SUCCESS in 1h 44m 21s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/33117800d1fd4f618bc0b4572d674afc : SUCCESS in 1h 58m 35s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/28b0aca7462842d0b70b84b67c7ed8d8 : SUCCESS in 59m 45s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/7ca140e599c140ce91cf23ffd9116da0 : SUCCESS in 1h 46m 12s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/3d3bbafe38da432fa68f98965d66fb77 : SUCCESS in 31m 35s","accounts_in_message":[],"_revision_number":4},{"id":"d1c977095347819ea83fd792e27a5d824ccaef2a","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-01-16 14:11:22.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"5273d8abfff29e0b80b306c24eab2838d1e5f67a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-16 14:12:16.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"9016c057818feba8174eea38f7c8b2a769f8f221","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-16 19:15:35.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/645121edc83d46c5a6dcafa00df60a3c\n\n- grenade https://zuul.opendev.org/t/openstack/build/bd24200a432146359966a1f2e92ec7b4 : SUCCESS in 1h 02m 48s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8c985a14ee9d41518e93421cf73a7509 : SUCCESS in 1h 05m 58s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0df2dc823f3f4ca7b63d413b0ba4ec04 : SUCCESS in 1h 28m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3cd1a1c02c1943e08461ffb48bb66b5f : SUCCESS in 1h 15m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c8c549f3a244c3aaa56445690697484 : SUCCESS in 5m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d6aae301fede49d98e5861514ef763e0 : SUCCESS in 5m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7fa3abb092f54506bd5dde110a1a091c : SUCCESS in 5m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0bf4f4fdd98140b9bf06f3be9d705e32 : SUCCESS in 6m 00s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/03a0536ad2c648779b4713b87d409da4 : SUCCESS in 8m 02s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c4fdd56a5e2a43fabc5304d1238cf0d3 : SUCCESS in 10m 40s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/4921d4449e5e48289c9b157375c8625e : SUCCESS in 11m 17s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/07958f7e24e7476794b6008ffa5fa38b : SUCCESS in 1h 45m 52s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/70f570dec48d47c5a9e98f4409697d8d : SUCCESS in 1h 43m 49s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/37cc784548c04064a603db86f45181c0 : SUCCESS in 1h 47m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/afd90d3388be4e3a90cfdfd725aa38bb : SUCCESS in 53m 08s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e03ee58675ac46ab934b3e546260f734 : SUCCESS in 1h 46m 17s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/e52ead0d657b457faa9bc53514b1df12 : SUCCESS in 35m 05s","accounts_in_message":[],"_revision_number":4},{"id":"5a7e49a82b37d4911400924f7fa6c035c3f1c5c7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-16 19:15:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"6875746f75a443678a0ecbb024a9372c270617b8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-16 19:19:15.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cd99592787474a5586cef15bc28e95f2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51529c9ac83f4464ba2d5c5c9ef1aafb : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"0addd1a99b1be32cc11301730e86234ddfb3c3c1","revisions":{"70d52b27e668b4e9c98b51079b753fbe1c6fbd9b":{"kind":"REWORK","_number":1,"created":"2024-12-17 08:39:49.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/44/937844/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/44/937844/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/44/937844/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/44/937844/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/44/937844/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/44/937844/1"}}},"commit":{"parents":[{"commit":"fbf4cd0c4cf081085a9d5400de49c7cdf85a7c62","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fbf4cd0c4cf081085a9d5400de49c7cdf85a7c62"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-12-17 08:25:58.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-12-17 08:34:11.000000000","tz":0},"subject":"Keep only Python 3.12 for tips priodic jobs","message":"Keep only Python 3.12 for tips priodic jobs\n\nWith migration from ubuntu jammy to noble, python3.11\nis not available anymore. This makes the 311 tips jobs\nto fail.\n\nSo keeping only py3.12 tips jobs since ubuntu-noble\ndefaults ships with py3.12[1].\n\n[1]: https://governance.openstack.org/tc/reference/runtimes/2025.1.html\n\nChange-Id: Iabd3efbebf7534dd6b7ebdf961bdff6804710c6c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/70d52b27e668b4e9c98b51079b753fbe1c6fbd9b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/70d52b27e668b4e9c98b51079b753fbe1c6fbd9b"}]},"branch":"refs/heads/master"},"89c307dbe2003de0f4b72e2548f733e73a1a0bb2":{"kind":"REWORK","_number":2,"created":"2024-12-17 11:05:11.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/44/937844/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/44/937844/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/44/937844/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/44/937844/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/44/937844/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/44/937844/2"}}},"commit":{"parents":[{"commit":"fbf4cd0c4cf081085a9d5400de49c7cdf85a7c62","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fbf4cd0c4cf081085a9d5400de49c7cdf85a7c62"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-12-17 08:25:58.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-12-17 10:59:09.000000000","tz":0},"subject":"Keep only Python 3.12 for tips priodic jobs","message":"Keep only Python 3.12 for tips priodic jobs\n\nWith migration from ubuntu jammy to noble, python3.11\nis not available anymore. This makes the 311 tips jobs\nto fail.\n\nSo keeping only py3.12 tips jobs since ubuntu-noble\ndefaults ships with py3.12[1].\n\n[1]: https://governance.openstack.org/tc/reference/runtimes/2025.1.html\n\nChange-Id: Iabd3efbebf7534dd6b7ebdf961bdff6804710c6c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/89c307dbe2003de0f4b72e2548f733e73a1a0bb2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/89c307dbe2003de0f4b72e2548f733e73a1a0bb2"}]},"branch":"refs/heads/master"},"b319123ff7b471efb716581e671d0cb13a3d1deb":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2024-12-17 11:12:04.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/44/937844/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/44/937844/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/44/937844/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/44/937844/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/44/937844/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/44/937844/3"}}},"commit":{"parents":[{"commit":"fbf4cd0c4cf081085a9d5400de49c7cdf85a7c62","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fbf4cd0c4cf081085a9d5400de49c7cdf85a7c62"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-12-17 08:25:58.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-12-17 11:00:58.000000000","tz":0},"subject":"Move all periodic tips jobs from py311 to py313","message":"Move all periodic tips jobs from py311 to py313\n\nWith migration from ubuntu jammy to noble, python3.11\nis not available anymore. This makes the 311 tips jobs\nto fail.\n\nSince ubuntu-noble defaults ships with py3.12[1] and\nsupports py3.13, moving py311 jobs to py313.\n\n[1]: https://governance.openstack.org/tc/reference/runtimes/2025.1.html\n\nChange-Id: Iabd3efbebf7534dd6b7ebdf961bdff6804710c6c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b319123ff7b471efb716581e671d0cb13a3d1deb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b319123ff7b471efb716581e671d0cb13a3d1deb"}]},"branch":"refs/heads/master"},"0addd1a99b1be32cc11301730e86234ddfb3c3c1":{"kind":"REWORK","_number":4,"created":"2025-01-09 06:18:56.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/44/937844/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/44/937844/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/44/937844/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/44/937844/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/44/937844/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/44/937844/4"}}},"commit":{"parents":[{"commit":"c28988720306b16cd6c211cbf017939fcef6828d","subject":"Merge \"Revert \"Increase per project quota to 50\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c28988720306b16cd6c211cbf017939fcef6828d"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-12-17 08:25:58.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2025-01-09 05:58:25.000000000","tz":0},"subject":"Move all periodic tips jobs from py311 to py313","message":"Move all periodic tips jobs from py311 to py313\n\nWith migration from ubuntu jammy to noble, python3.11\nis not available anymore. This makes the 311 tips jobs\nto fail.\n\nSince ubuntu-noble defaults ships with py3.12[1] and\nsupports py3.13, moving py311 jobs to py313.\n\n[1]: https://governance.openstack.org/tc/reference/runtimes/2025.1.html\n\nChange-Id: Iabd3efbebf7534dd6b7ebdf961bdff6804710c6c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0addd1a99b1be32cc11301730e86234ddfb3c3c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0addd1a99b1be32cc11301730e86234ddfb3c3c1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
