)]}'
{"id":"openstack%2Fglance~938433","triplet_id":"openstack%2Fglance~master~Idfffbf2fa5ea96a6fd9b784532c7dd379df77502","project":"openstack/glance","branch":"master","topic":"epoxy-milestone-2","attention_set":{},"removed_from_attention_set":{"19138":{"account":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"last_update":"2025-01-13 05:40:09.000000000","reason":"\u003cGERRIT_ACCOUNT_19138\u003e replied on the change","reason_account":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"}}},"hashtags":[],"change_id":"Idfffbf2fa5ea96a6fd9b784532c7dd379df77502","subject":"Refresh Glance example configs for epoxy m2","status":"MERGED","created":"2025-01-06 06:14:23.000000000","updated":"2025-01-13 10:13:26.000000000","submitted":"2025-01-13 10:10:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"938433-epoxy-milestone-2","meta_rev_id":"f5bb33fb222df6596e1bcb0ae238888e19f2a4a8","_number":938433,"virtual_id_number":938433,"owner":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15527,"name":"foundjem","display_name":"Armstrong","email":"foundjem@ieee.org","username":"foundjem"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-01-13 10:10:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"recommended":{"_account_id":15527,"name":"foundjem","display_name":"Armstrong","email":"foundjem@ieee.org","username":"foundjem"},"all":[{"value":1,"date":"2025-01-10 15:31:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15527,"name":"foundjem","display_name":"Armstrong","email":"foundjem@ieee.org","username":"foundjem"},{"value":2,"date":"2025-01-09 16:19:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":0,"_account_id":15527,"name":"foundjem","display_name":"Armstrong","email":"foundjem@ieee.org","username":"foundjem"},{"value":1,"date":"2025-01-09 16:19:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":15527,"name":"foundjem","display_name":"Armstrong","email":"foundjem@ieee.org","username":"foundjem"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":15527,"name":"foundjem","display_name":"Armstrong","email":"foundjem@ieee.org","username":"foundjem"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-06 06:25:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-01-06 08:28:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-07 14:30:01.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2025-01-10 15:30:51.000000000","updated_by":{"_account_id":15527,"name":"foundjem","display_name":"Armstrong","email":"foundjem@ieee.org","username":"foundjem"},"reviewer":{"_account_id":15527,"name":"foundjem","display_name":"Armstrong","email":"foundjem@ieee.org","username":"foundjem"},"state":"REVIEWER"}],"messages":[{"id":"c55373bdd126836fa71627ee05ed6b2595bb9c11","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2025-01-06 06:14:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c57a492d590aa1094a5a2939ae218cd9d8cba774","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-06 06:25:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/909c80227b934194ba849dfc54d73296\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7f7c1498b8f540db9ae9039783e7e57d : SUCCESS in 9m 06s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4ad12c7b0ccf4a1cb4c7761acb8db764 : SUCCESS in 8m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c7b38e6d8a9f5f72d2ff649be7b4222a47d25233","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-06 08:28:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6688442e478f4041a641e3fcd67eb3e5\n\n- grenade https://zuul.opendev.org/t/openstack/build/8169cdf026304ee2b13c894846b831be : SUCCESS in 1h 09m 18s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/32010e0c540148e4aa76e33caa4f8da1 : SUCCESS in 41m 58s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/74d14c7d39134fb4a6e18d94fb46feb7 : SUCCESS in 1h 10m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cd9dfd4ae3cb46aeaf92cf2e984543a7 : SUCCESS in 1h 06m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b37e832c65f8485c935f30b03f68da8b : SUCCESS in 4m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/77e66c55d717402bb1af4a4c693c5cd2 : SUCCESS in 4m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/21f2f2fa50964d83bc068fcb15e48cb1 : SUCCESS in 3m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a5095c0b58a42fca3a4280180c04ef4 : SUCCESS in 5m 16s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6dedf19739d945888ca13d466fa73497 : SUCCESS in 8m 47s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/fb6e328d49b0445ea366b189c46c3f66 : SUCCESS in 7m 23s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/da24ab2952b641f8a5f0b0e56f3827c1 : SUCCESS in 9m 26s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/2f1d5de83339403ca212d836108c7d4d : SUCCESS in 1h 32m 36s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/122c693a3c594c4ea79750adbf75ae71 : SUCCESS in 1h 12m 20s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/d3e2540fcd424a958dc20a5f85fffcb8 : SUCCESS in 2h 13m 08s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/9618a918623949ba9fd6766f490e47b2 : SUCCESS in 2h 08m 03s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/0f40b142e1ac493480e6668e86c24c15 : SUCCESS in 2h 00m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fda604f7fc0f49418a76992c6ef926d6 : SUCCESS in 1h 02m 19s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/791a9c514622461f9d0af04b49ae7600 : SUCCESS in 1h 35m 46s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/18272f9ad35848f88b61a6f2b276a49e : SUCCESS in 2h 08m 13s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/57b5875d15d546e4958a6c8dc95621c5 : SUCCESS in 1h 03m 04s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/ad30e416c2174c9290164a43b3b632b5 : SUCCESS in 2h 02m 37s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/13b6af881a844774be85e5de442941dc : SUCCESS in 38m 15s","accounts_in_message":[],"_revision_number":1},{"id":"77aae646e5290b9e5ec162b5e7351fc8dd399007","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-01-07 14:30:01.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"37d1e6fb61eb00dfc046e8deb51148e5839b9d80","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2025-01-09 14:37:01.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"6a70a1184c7dea1159fd8896813e618beb25612a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-09 15:09:06.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/80f6e88daf854d56899436433ef637ce\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d21385c6a98f4808b74a30ebdeea5368 : SUCCESS in 14m 41s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8a6b9e3085ff4c059b789793af7decce : SUCCESS in 14m 45s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3c4bcefc86b71c0b8db640806350ccf67c213dee","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-01-09 16:19:21.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c2082582f2a0b1d030acd0b1825055ba900be7c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-09 17:10:13.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d3ceea8a3a444d39bf29122d6944f5cc\n\n- grenade https://zuul.opendev.org/t/openstack/build/13f23a00ea2a455ba0a9ea6de875bd6c : SUCCESS in 59m 09s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ed6f78c0ff514a14b4b3d23a8372759e : SUCCESS in 56m 53s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e1819c2f306646a3930668256e17e5a2 : SUCCESS in 1h 32m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9a325629d20f47acb59561a021cf9ea9 : SUCCESS in 1h 09m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c24a5adce9f46b5b967b378d5fc36ba : SUCCESS in 6m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4a480bf4236b402fbcccafb5d6847038 : SUCCESS in 4m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4102850d2b4844d0b08523438d5539be : SUCCESS in 7m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a022cbcfe1f54770b27fc8946f48321c : SUCCESS in 6m 36s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d3740867329d4839bb4927c69dac47c2 : SUCCESS in 9m 22s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b648ff24b1e046d491f8b2d7ab3dee5b : SUCCESS in 8m 38s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/29a18bc753a246098d801a0a5f3f6c5a : SUCCESS in 12m 38s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/49cb41a25bb746cc9247135601d8e7fb : SUCCESS in 1h 57m 38s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/7bbf576e4a194a8db5680d2a9adc99e7 : SUCCESS in 1h 55m 52s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/95b0230bacd4438fba73a5a4c5c9313c : SUCCESS in 2h 05m 55s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/bfe539d0ca834041a05c36d1bef7581a : SUCCESS in 1h 14m 47s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/8790f5107b7041f3bfcebed6fbf814d6 : FAILURE in 1h 49m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2597cac6c4f142e695e90d7a5e1cb641 : SUCCESS in 56m 02s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/2f5544f54c824f0b9fb98d6f890c3190 : SUCCESS in 1h 47m 25s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/cf32ee1c9fa7431ea069eafac0e93f69 : SUCCESS in 2h 26m 34s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/e33a5519cf49468daa987992aee06fe8 : SUCCESS in 53m 31s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/90b82d7b8cb84a0bbeb77347b609f971 : SUCCESS in 2h 02m 23s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/4e071c97374d497b8803dac87578eeb5 : SUCCESS in 38m 28s","accounts_in_message":[],"_revision_number":2},{"id":"5222ddba3b5b59ebb05cde0bfc7e936293a75982","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2025-01-10 05:31:56.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7a46486ec571e5f05afc8009b6b1c38fe5f9806f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-10 05:47:40.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/531fe8a356664790804d7fa53930c56f\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b2d3539f3d904d00b6ee92594f7261d7 : SUCCESS in 9m 14s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/499f8e570dc943afab020a74829f0f69 : SUCCESS in 13m 34s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c2948f31306d9eadef805d37c7bb197e311506aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-10 07:38:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/682b54d4b3ec4a2eb5b48017f74c348b\n\n- grenade https://zuul.opendev.org/t/openstack/build/6d0a1746731c403cadcd68742438ec75 : SUCCESS in 41m 59s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/98d6c9eac2d744da85d11876a4a01e1a : SUCCESS in 42m 05s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/fff42d99f42c4b459d75e85bb66adaac : SUCCESS in 1h 32m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1574dabd53b44631a91cba06a2d23a8a : SUCCESS in 45m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6fda9dd5e6704b3f96f62b85ab0268d7 : SUCCESS in 4m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c6691c9e8ecb4898aaf2221b2ec19059 : SUCCESS in 4m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/640b99c892944f3fae1f4cddae9911cf : SUCCESS in 5m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d09dc80dc23d49bf88d35d9053bf27ef : SUCCESS in 5m 37s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a784f178e2df429982f547dceacade92 : SUCCESS in 8m 30s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d0f3528cca754ea59370d2ffa744ca10 : SUCCESS in 8m 34s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/775bb38681bd4dbca653ee54494213c3 : SUCCESS in 11m 56s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/043a3c32d8e5418ea4d633dd57ea9180 : SUCCESS in 2h 00m 26s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/110cb34ad7d7469993c3d9ef640482fd : SUCCESS in 1h 56m 10s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/33269b7bcab1401babcc28611c173b00 : SUCCESS in 1h 26m 20s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/970ed4f3a9b6480388b3449cb533ac7c : SUCCESS in 1h 47m 31s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/969ab078d23144fc8522138e05d22eeb : SUCCESS in 1h 27m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/56f740077ddb4badafddad257a185c8a : SUCCESS in 43m 35s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d1bf8d951d9447e4944c3c77b3d9bd77 : SUCCESS in 1h 04m 13s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/d31ec6d8f2b84aa696a68a271712cc80 : SUCCESS in 1h 31m 41s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/1c7e0ca2eb234cb3aaf2db5fbcff58cc : SUCCESS in 33m 18s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/69c5a9d7d4fc40c8944acdedcc60a844 : SUCCESS in 1h 35m 53s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/48a90fd5eb884071b8785666c575a64a : SUCCESS in 35m 10s","accounts_in_message":[],"_revision_number":2},{"id":"bbdc444b6c71cec0e9a5e9dd513ba76be06623fc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-10 07:39:36.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e2e54e6974cadefd711336d37d380aff678a06ea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-10 09:49:51.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/756d27c0a9fd416a96a8879ab8e1eae0\n\n- grenade https://zuul.opendev.org/t/openstack/build/5d2c2df98a174446b8c3d538e981d58b : SUCCESS in 35m 30s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c89f2e0229e04ae0a923c4413866f3f3 : SUCCESS in 57m 33s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/dd79dd95c27e4d28bd532b168523eb4d : SUCCESS in 1h 32m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/30a44edb3d944fa98d79cfc8d5ea230c : SUCCESS in 47m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39914c95d44041bfa5e671533cc29af4 : SUCCESS in 3m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7c5a463253054974a8457a0ded8d88d3 : SUCCESS in 3m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/23aa277cdb774d76845559b4f69d9840 : SUCCESS in 3m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a5c3968986a45b182d93dcf5cfba72e : SUCCESS in 4m 17s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/232ecd8bac654837acd2fb6e4c2d074f : SUCCESS in 8m 55s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/074460b583044dcb8b17166f13d0f35c : SUCCESS in 7m 51s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/d6ab6255fd3e4161a3bb3ddc33e02dc7 : SUCCESS in 11m 42s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/7d054f368c594ff29329d92764db7e79 : SUCCESS in 2h 10m 12s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/647e380541af42d997c3f63d563c762d : SUCCESS in 2h 04m 04s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/897576e5727e40ab80b412059507b180 : FAILURE in 1h 02m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/34256b84ca904e4998021e1226fa68a5 : SUCCESS in 1h 07m 00s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/affb8c5737f940cf96f4a279d95abc3b : SUCCESS in 1h 05m 45s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/aca71cec80be40c8be7403b81b887125 : SUCCESS in 36m 59s","accounts_in_message":[],"_revision_number":2},{"id":"20d685a1ff49240ef487fe7d6c22e0a155b2d181","author":{"_account_id":15527,"name":"foundjem","display_name":"Armstrong","email":"foundjem@ieee.org","username":"foundjem"},"date":"2025-01-10 15:30:51.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2ddded1f8702a8c50eff04ed5b07ccb58a981d61","author":{"_account_id":15527,"name":"foundjem","display_name":"Armstrong","email":"foundjem@ieee.org","username":"foundjem"},"date":"2025-01-10 15:31:11.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"baacd19b25364b433155643bb79b046752cdab2a","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2025-01-13 05:40:09.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5f905e60c298fdf617d289427488efb1acfe1d40","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-13 05:52:22.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c9658dfde4a7426e91af20c5779881b3\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6b660da7c69c4a1f93b9343ebc15866e : SUCCESS in 8m 16s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6815bc8243744ee195b18568c40ef3f9 : SUCCESS in 10m 01s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"cbac175eb8d469663a23085f72a677f61fa8f051","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-13 08:07:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7e15d13cd3ca4a74a892f86d9249f3fa\n\n- grenade https://zuul.opendev.org/t/openstack/build/1615c4df149b47349d9d16315698b5d5 : SUCCESS in 54m 16s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/93650023385046f4b77a347005bd725e : SUCCESS in 51m 04s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/44bdec4ae60641918108642639ad96a9 : SUCCESS in 1h 26m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5d241d8e730e4e60a4b2e588e3ec0178 : SUCCESS in 47m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db66931880384a87b119ae9299d4a587 : SUCCESS in 4m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/51ef7c79963642ed888f75db6dab1236 : SUCCESS in 4m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/32fca96628e943709c151d9ea2207cb5 : SUCCESS in 5m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b225b927f934c72b646c6ba37046fd0 : SUCCESS in 4m 44s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3e82c837cfd942e4800f34e5169f14cc : SUCCESS in 7m 51s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9871131ea7e347edb0237b3f2dbb422b : SUCCESS in 8m 13s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/0ed65fe5c5204035b8bd47d630fdad31 : SUCCESS in 12m 25s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/0abe15aac085486d8f69556469d7c953 : SUCCESS in 1h 47m 29s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/8dc22cd096ea45ffaadce634dd579b26 : SUCCESS in 2h 17m 25s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/7206dbcce4cb4f1b9d9caccf6e977dda : SUCCESS in 1h 24m 17s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/e5cf8ab5d65e494eaa04aa1148c87ebd : SUCCESS in 1h 58m 23s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/14c2de785ac4421682ce6cfac4a5864e : SUCCESS in 2h 04m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5a712c64092a445bb219c631395cab3d : SUCCESS in 44m 23s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9c4ff27f0f8749659aa370810ccfebf1 : SUCCESS in 1h 47m 37s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/02e389d7d4ae4b478d18a67a33e8a5d4 : SUCCESS in 1h 54m 10s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/b7de36f13aa1487e907d10d6165031d7 : SUCCESS in 57m 57s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/23cf6a6ab6854c2c8edaeae96b2d7e16 : SUCCESS in 1h 14m 06s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/58b00ff8183e414f87a109b454f0fc2d : SUCCESS in 36m 10s","accounts_in_message":[],"_revision_number":2},{"id":"7fd8fc75651d0eab050b202ed7b4a87cf5b9f8c9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-13 08:08:32.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9534bea7868789e307ddea4b8cc8231de182803c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-13 10:10:56.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/679ed92021f94b5b90fb930969c856b7\n\n- grenade https://zuul.opendev.org/t/openstack/build/b734f6e685c642549e16c57d6f91f706 : SUCCESS in 42m 31s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/23f97cb042344957aceea7f4b393c52e : SUCCESS in 57m 41s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/722ed9379cb94647bf38e4ee7e100900 : SUCCESS in 1h 28m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/85f88894864243e9a4df6ccee8251a6d : SUCCESS in 1h 03m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a9d4f9baf3141c19c4706e353b9864f : SUCCESS in 4m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bc8b44a5a5624ce58b1f87d0ef17412a : SUCCESS in 4m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/599545fda43c4c2db2e8b8d9451cbc33 : SUCCESS in 4m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b670cd3f75bc4bdf80cd330b615aa037 : SUCCESS in 5m 20s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/79a2f013d0a045679eaaf16dc31f9726 : SUCCESS in 9m 45s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1a6d418b950c436c98d97a73fd4ec65a : SUCCESS in 9m 32s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/b61f303be9c74d1ba108dea1a050f9f8 : SUCCESS in 9m 36s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/58b34756959d4305a77cc7924b0a434b : SUCCESS in 1h 50m 38s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/4b20b6b4bdac4a6981916f8b33bdd8af : SUCCESS in 1h 39m 01s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/0ea5e6d944094ad484e28329e9022b8d : SUCCESS in 1h 41m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8d2f003201df40ad9060870240f5e923 : SUCCESS in 1h 02m 22s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/4d81f1b0e22c41d59ab2fd02c241a7c0 : SUCCESS in 1h 19m 15s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f08d02353ca64c76985ba0ad6e04a4b9 : SUCCESS in 30m 15s","accounts_in_message":[],"_revision_number":2},{"id":"385b2d0bfa96b7385d568b1786ce0a6f234b4552","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-13 10:10:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"f5bb33fb222df6596e1bcb0ae238888e19f2a4a8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-13 10:13:26.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d62a22524dfe40f7ac099d8ca7b21a4a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f09c5cfa960e4ef399265f4dc607c706 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5e3d32e1fcaf4a1b90547875d39f89eba1177d4e","revisions":{"75db09666d981ca5b8f96fd3f3f15a6b7443535e":{"kind":"REWORK","_number":1,"created":"2025-01-06 06:14:23.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/33/938433/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/33/938433/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/33/938433/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/33/938433/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/33/938433/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/33/938433/1"}}},"commit":{"parents":[{"commit":"05539df736700b34eb068553238c3e3e4be03b79","subject":"Release notes for Epoxy Milestone 2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/05539df736700b34eb068553238c3e3e4be03b79"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2025-01-06 06:07:46.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2025-01-06 06:07:46.000000000","tz":0},"subject":"Refresh Glance example configs for epoxy m2","message":"Refresh Glance example configs for epoxy m2\n\nChange-Id: Idfffbf2fa5ea96a6fd9b784532c7dd379df77502\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/75db09666d981ca5b8f96fd3f3f15a6b7443535e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/75db09666d981ca5b8f96fd3f3f15a6b7443535e"}]},"branch":"refs/heads/master"},"5e3d32e1fcaf4a1b90547875d39f89eba1177d4e":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-01-09 14:37:01.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/33/938433/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/33/938433/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/33/938433/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/33/938433/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/33/938433/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/33/938433/2"}}},"commit":{"parents":[{"commit":"69cb639f01dbb9e61cbfd83f2eabd6aa96b5bf5f","subject":"Release notes for Epoxy Milestone 2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/69cb639f01dbb9e61cbfd83f2eabd6aa96b5bf5f"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2025-01-06 06:07:46.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2025-01-09 14:33:41.000000000","tz":0},"subject":"Refresh Glance example configs for epoxy m2","message":"Refresh Glance example configs for epoxy m2\n\nChange-Id: Idfffbf2fa5ea96a6fd9b784532c7dd379df77502\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5e3d32e1fcaf4a1b90547875d39f89eba1177d4e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5e3d32e1fcaf4a1b90547875d39f89eba1177d4e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15527,"name":"foundjem","display_name":"Armstrong","email":"foundjem@ieee.org","username":"foundjem"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
