)]}'
{"id":"openstack%2Fglance~939243","triplet_id":"openstack%2Fglance~master~Ibf9f5eb4e2378ddd77ebe61170b8e921ada792d7","project":"openstack/glance","branch":"master","topic":"bp/glance-as-defender","attention_set":{},"removed_from_attention_set":{"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2025-02-25 17:18:31.000000000","reason":"\u003cGERRIT_ACCOUNT_9303\u003e replied on the change","reason_account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},"4393":{"account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"last_update":"2025-02-25 20:37:01.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_37598\u003e by clicking the attention icon","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2025-02-25 21:19:00.000000000","reason":"Change was submitted"},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2025-02-25 16:13:38.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"Ibf9f5eb4e2378ddd77ebe61170b8e921ada792d7","subject":"Allow ignoring specific gpt safety checks","status":"MERGED","created":"2025-01-14 14:33:36.000000000","updated":"2025-02-25 21:20:30.000000000","submitted":"2025-02-25 21:19:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"939243-bp/glance-as-defender","meta_rev_id":"6fc3ab8bb01b4c75355d2996814c4d93c57e5314","_number":939243,"virtual_id_number":939243,"owner":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-25 21:19:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"recommended":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"all":[{"value":1,"date":"2025-02-25 18:19:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":2,"date":"2025-02-25 17:18:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2025-02-25 16:12:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-02-25 16:13:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":1,"date":"2025-02-25 18:37:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-14 15:34:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-01-14 15:56:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-20 15:23:20.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"CC"},{"updated":"2025-02-20 15:33:41.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2025-02-21 12:17:31.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2025-02-24 17:38:57.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2025-02-25 16:12:14.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"},{"updated":"2025-02-25 18:19:13.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"REVIEWER"}],"messages":[{"id":"7c284ac7544ade0c9cab3bc764a649b0a954d44c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2025-01-14 14:33:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2b4034695eda08e1853f44b88ba78e3e4d4e13fe","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-14 15:34:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b7c3a892f77449259b800c1dade5d308\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5b0e1d520b8f45b0831ddc6bad9ff5d2 : FAILURE in 11m 30s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/28dd77e9b7294fe884768fa24923cbba : FAILURE in 12m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bbb8b20a78eeda64f811c158407a7c374660bb00","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-14 15:56:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c195b4d9a07e451691e41c4c502179f9\n\n- grenade https://zuul.opendev.org/t/openstack/build/b0c11cb4bdf3406c8d69c2f37bd3d088 : FAILURE in 40m 44s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/08ab9bdc095b485c94edb5611904a0e4 : FAILURE in 43m 15s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1598d3f735af4c7f88c2e9773dbb6c78 : FAILURE in 28m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/59d0e8c823a241b3b6601d96c604be5c : POST_FAILURE in 31m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2788cb4c91c541489c27025527e9c36e : SUCCESS in 5m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3167497a88514c4ab2b6d8c5015619b8 : FAILURE in 4m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a944870bf7c248e6a8c4d0450a3a192f : FAILURE in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a10aeeb3b60b45f4866556d2c643b04c : FAILURE in 4m 51s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/cf7512657d19416dabacd63b98d2bf03 : FAILURE in 4m 16s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3eba8b93e4424c5c90aa44f05391b9b1 : FAILURE in 4m 21s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/fa30a785c4df42818f24e2262e30d85c : FAILURE in 7m 14s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/fa5ebcf951e7457c99e2afb67acdabad : FAILURE in 42m 07s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/57990499e0774b80a2db08ea5cd95a8b : FAILURE in 32m 25s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/e15e317511054cf7a1a5e26b5326af63 : FAILURE in 36m 16s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/f866c2b1cec34d9f8f9e29128261ac78 : FAILURE in 29m 21s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/ccc801adc3c042fca4fd426dd683e832 : FAILURE in 32m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b586f31886c54373980a1a50a13eb6e4 : FAILURE in 30m 58s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9778f6a88d274f8f87ba290ba2e1fb89 : FAILURE in 43m 21s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/de66bd332b974866b12861d6c232248c : FAILURE in 28m 07s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/4fe1a3ee867b4a66895784d40d2569a8 : FAILURE in 1h 13m 18s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/73ec140c3f2641d6bc12e8fac1d54759 : FAILURE in 39m 34s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/dc7e9a308561482e8c8eea30caeeb6b0 : FAILURE in 22m 44s","accounts_in_message":[],"_revision_number":1},{"id":"4a47c849e6eb7c871f85734767fbdf80e59a4269","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2025-01-15 17:36:19.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"8cf5fe2503bbc36f3c2a35dce4f23722de2809c5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-15 18:06:32.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d39d1a8ab5b49d7a9297ec1a777d77d\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/69be1dca9e674687890012e1040d5a5c : SUCCESS in 16m 34s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/db72c607d55e4b77af33926b0b652bfc : SUCCESS in 18m 54s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"504295bd6b16f507051ba6d6e706fd958a301551","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-15 20:04:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f8b15e57ca2f489baada25d5532deb7b\n\n- grenade https://zuul.opendev.org/t/openstack/build/df0f7bc6e38b4b2c8ee0925eaac77ab2 : SUCCESS in 1h 04m 24s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7e72b8911da84785a66849a2ebc32893 : SUCCESS in 52m 04s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e9094d23fd08401b84a075899eb3d06c : SUCCESS in 1h 29m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8088702365b14415ac893d5a765b4b8d : SUCCESS in 59m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8b7ea54bbd8465d9d9a9bf1fdbeea8a : SUCCESS in 5m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a87cb656f68b48a9b1fe7b1f9325f795 : SUCCESS in 5m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cb83eab771f24a54b510076bcf0b068a : SUCCESS in 7m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cef28841ded7424e9544c903da4b41ef : SUCCESS in 6m 03s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ce7f1df5afdb49b7bca3889a197d7d3f : SUCCESS in 7m 32s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b1c3458f9adc4933b58c6eda90ee402e : SUCCESS in 8m 14s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/1ac82a752513422aab03917599689995 : SUCCESS in 12m 43s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/2a92e3268bb34fbbabc4338150bb8010 : SUCCESS in 1h 58m 59s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/53f0e256ffe647368bcdf0bd7742407a : SUCCESS in 2h 15m 18s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/651c2535572d49daa863ac6221c577d7 : SUCCESS in 1h 38m 56s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/f205c3550eea4bebaa572456bd03967b : SUCCESS in 1h 59m 05s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/151b627155fe4ba1b544a3ecd9f6f278 : SUCCESS in 2h 08m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e291c3375e80451fa7c0175d9ab3454e : SUCCESS in 51m 27s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/b7674b07cd754566aeada4b42cf58d22 : SUCCESS in 1h 34m 40s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/ff5829c51987435ca2a130730871db91 : SUCCESS in 1h 57m 06s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/d12d4789dec149d0ac9627540a265db3 : SUCCESS in 56m 35s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/a9437dade9f24c268b665ea43cc1b93b : SUCCESS in 1h 50m 53s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/05370f3fc021443492605740d4bd2d9e : SUCCESS in 28m 49s","accounts_in_message":[],"_revision_number":2},{"id":"5f85916140f5a0d08c31dae958a3c2b06651079d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2025-01-16 15:23:07.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"85b351cc356707aa50fdc6e0a70fdd96e6dbb448","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-16 18:53:49.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/71184e073e5548dcb3b1d4218fe5d788\n\n- grenade https://zuul.opendev.org/t/openstack/build/e6b6f97c979f47249394f5ed2dbb6cf1 : SUCCESS in 36m 06s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/74e86b313f26416cae8ad836436f61ea : SUCCESS in 34m 56s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8552ac2de4144b2d952a5e6740e03b69 : SUCCESS in 1h 29m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e33c8897e3954016ba2a1ed93f3b3f7c : SUCCESS in 39m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1efcf3275d1432698a8eb99d5dd9164 : FAILURE in 3m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a32587a1b58c4f6687a794fab1e1b4ea : SUCCESS in 6m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3ed136f28d8c4e36bc844eb5a80ffa22 : SUCCESS in 5m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83b7c22decbd401ab926fee185243436 : SUCCESS in 5m 40s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2425fbe570974ead94d89516f06a0f4e : SUCCESS in 7m 09s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e4993dd20ab8467e8b1615f061bf1156 : SUCCESS in 9m 52s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/46a5967eebc140fc8f2ec2e0ffaccb01 : SUCCESS in 12m 37s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/c8df66266c2b4743af04029eb3d8be93 : SUCCESS in 1h 09m 29s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/85d618ba0dba4a1aa13dd9ca6091f324 : SUCCESS in 1h 56m 17s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/0812413d1d4f4f9fafa94863c33a9f61 : SUCCESS in 1h 55m 50s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/27147d5913f4420aacdd2378b3eea843 : SUCCESS in 1h 36m 12s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/19bbc270a0084ff7a946033e66fe0de7 : SUCCESS in 2h 06m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/033c7f70f0e54186b33ec6082635da7d : SUCCESS in 53m 37s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/40020722d903411cbb5270f02db4e12f : SUCCESS in 1h 45m 30s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/3037f568e04a4088b353bd31a5f0ce28 : SUCCESS in 2h 10m 21s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/a1642d02f3b342cf85636fe8d945af1c : SUCCESS in 57m 53s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/c1cb9f1c666242d884244000934dd023 : SUCCESS in 2h 02m 32s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/9a17f9b39cdf479a8ffe56711d97e1bc : SUCCESS in 31m 00s","accounts_in_message":[],"_revision_number":3},{"id":"f9d439ce619e650fca647aa429cdedf6793d8411","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-16 19:43:28.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e672361196e4489f83820011ae7d04ee\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9ca6bdb2f447420f99aa90ef97f8933e : SUCCESS in 12m 40s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ad5ee7eb18e244dcb7a9cce55c869ccc : SUCCESS in 16m 11s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"2b0a068f6efaaea9ec9c0a2ce716b31c8a90e523","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2025-01-21 15:45:39.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"2a5aa3aa59a1a558514297fc43814b84d43c0076","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-21 18:02:33.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/574fe0d0abb7460086955f3fc7f9459b\n\n- grenade https://zuul.opendev.org/t/openstack/build/271dfb170ce145caa134eda2de8aecfd : SUCCESS in 1h 04m 50s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9b5e870d0e864fd3988c9802774ac66e : SUCCESS in 1h 14m 20s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/72f46db7c3e64994ab0afa12972e6f20 : SUCCESS in 1h 41m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/73c3a7c27a8f4469978a3ef6958e65c5 : SUCCESS in 56m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f67877dcba74a4ba478aa79e647a5da : SUCCESS in 5m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9e8e0f69cec04d4bbb6b89d020c0a2ab : SUCCESS in 5m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/15691eac4d8f493192e16eb755511730 : SUCCESS in 6m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87c9779c54ca4204b4bc416a4bed1fe3 : SUCCESS in 5m 15s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/27cf061fe96c4a59bcc2d8466510b467 : SUCCESS in 7m 38s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a33808c49f1845788cd78ac50343fd4b : SUCCESS in 9m 39s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/466a303db7874536b615abb0616e9f7f : SUCCESS in 12m 05s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/886d489a98c14fcb887059c12d5194c8 : SUCCESS in 1h 45m 23s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/e98e65b26e334ac7bf72c9c3e7c4dd7b : SUCCESS in 1h 44m 26s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/3ad55ad048684328b0fe046070764608 : SUCCESS in 2h 01m 14s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/88f41bb287794deaa107d05bd12bea93 : SUCCESS in 1h 53m 39s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/a67f02daa355455795f053bbbe07baf3 : SUCCESS in 1h 48m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/89df5e179fe4408d8da819d30204d7fb : SUCCESS in 39m 41s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a7c1ec893cf24375b39909cacca56eaf : SUCCESS in 1h 32m 35s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/e96204f1fd524f148a3df2b5d94d2160 : SUCCESS in 1h 56m 10s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/6d3e05cbec794472bb4542703e37a19f : SUCCESS in 1h 02m 00s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/474a302937014ca6a6b2fe843be091d9 : SUCCESS in 1h 33m 51s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/58cabdfb6a82452394ae928ccedb24eb : SUCCESS in 29m 29s","accounts_in_message":[],"_revision_number":4},{"id":"0d32b0b647e5b74b3ec90516ef343489ce4cf3e8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-21 19:29:41.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c41d72c15cb24b12ac5ee8809656d9ca\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8972b3deb6b4449c94dbab800be8d6f9 : SUCCESS in 18m 00s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/61288bbbcda349d0818b5818d1e5a5da : SUCCESS in 27m 32s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"8f065d6908b2735f6afab259922d21e7817e7e48","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-02-20 15:23:20.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a017c6890baec9d2faa1369140435629e5ddc6a3","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2025-02-20 15:30:43.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"30ade4da4ffe93ea94979d84ab764cab01340997","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-02-20 15:33:41.000000000","message":"Patch Set 4: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"5ce02f42de90380d36e6f7b810d723cab4a02c4d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2025-02-20 21:13:58.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"5711d99b92ecf2213ca24a61719febcdf9aadad2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 21:43:24.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/39d114bd897645c9bd3739d434cecd3f\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7db84e27626449d4b51451a9723c7222 : SUCCESS in 12m 49s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/31a1a46b854841c8a0f754e2d442c85e : SUCCESS in 23m 57s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"8e785377d74b87e4687b415aaa84acbcb3a5e26c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 23:21:46.000000000","message":"Patch Set 5: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d430348921f84ce79f920f75dea2efb0\n\n- grenade https://zuul.opendev.org/t/openstack/build/b5b4882468f34882b6bb5cc882f1bd3f : SUCCESS in 56m 13s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/beba3e16ee1645a8818ac98afdb0ba16 : SUCCESS in 1h 02m 39s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e9b9a76fe6f643ada3b8854a14f967bd : SUCCESS in 1h 44m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/47152acfeab94123bc7a2ff9494a2145 : SUCCESS in 58m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2adf69a1d9ae4f25b0e9fe86e3e5386f : FAILURE in 3m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/915645ca74f84a5388019b8fd9258b9e : SUCCESS in 7m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4416a2cb4faa4c408b147be129948696 : SUCCESS in 4m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1c9ceefd0964ea7b07877024f3032dc : SUCCESS in 7m 59s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a528b6671e4e49e1a7af1522f64a39e3 : SUCCESS in 11m 32s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5313ea4922bd41bca69d888e900c2fdc : SUCCESS in 8m 19s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/bec99b6d6e6941d68998eed15300da1c : SUCCESS in 10m 21s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/c3dfde60ee17454d854a4811044a94fc : SUCCESS in 1h 59m 16s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/25463e9f271e44b2b5627f4b356f671b : SUCCESS in 1h 32m 19s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/0ebae5c9fa734bf88dd5c71d5a9ff6a3 : SUCCESS in 2h 02m 43s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/b719c976e4354543a01e65ca1b9d56c2 : SUCCESS in 1h 55m 24s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/b0e04806173d4a61a6095edd4af983e9 : SUCCESS in 1h 56m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ad5a8f1158ca464e989a4d6170aa708b : SUCCESS in 47m 27s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a406e51a1d784ba391ce212e848a812b : SUCCESS in 1h 59m 51s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/1bbecf792c4e4e35a8f10df010c97d33 : SUCCESS in 1h 56m 47s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/62fe4287524d4eb0901c3f271be17480 : SUCCESS in 52m 06s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/38f60d8183d64b9fa36f3c708857dcb8 : SUCCESS in 1h 26m 57s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/935c73d72e8c414eb863cd36c3444db7 : SUCCESS in 27m 10s","accounts_in_message":[],"_revision_number":5},{"id":"cd4533a039969bd28c309850143557b413f4c4a4","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-02-21 12:17:31.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"c3ea52add0c1850716b6714df0bd23460c64bca9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2025-02-24 14:32:47.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"db14b56f625c98a6bc9010ed6beec50d9cd7d93c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-24 15:38:54.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0ecf773aac6941449f5e85d35508e5be\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f7d08473fab64023aa0f7b4b835adc99 : SUCCESS in 13m 59s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6026048b4539434aa4923ec8c9af7669 : SUCCESS in 35m 11s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"cbd4be5e6d2e1c2af32fa5b65bf7060f3e6e6f36","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-24 17:36:11.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/efed0a21fd2144bfba42f75625627c19\n\n- grenade https://zuul.opendev.org/t/openstack/build/d1c46d2258b942e09c8b909f49378661 : SUCCESS in 55m 45s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5558a7061232491e828793969454144a : SUCCESS in 1h 03m 37s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/946b4fc5b4f94de4b7b9f72b71e33adf : SUCCESS in 1h 34m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5be02333b17d4f3a864a9368d80045ec : SUCCESS in 56m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/459ac7f9ef1741cd87dbcd6c82fb77c0 : SUCCESS in 5m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/950e0ef4edf146bcbb229fa4f021bb2a : SUCCESS in 5m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8e6c9570eb5a444e90e2deda64605fc5 : SUCCESS in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b609b2ba975e4b30bdf455740880cd28 : SUCCESS in 5m 14s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2423c04c69a84f9ebbc19f2a8316ee2a : SUCCESS in 7m 45s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/81be2002b029435f898295d45bb7af6c : SUCCESS in 7m 58s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/a1ba6b4a0bcd4d679fb250512532e9c7 : SUCCESS in 16m 51s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/9f4fa82c047d413cbfde7456aba26394 : SUCCESS in 2h 29m 45s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/3d03a78112e944a886563af9d1279e1e : SUCCESS in 2h 06m 56s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/74641998ad0548b8801bb48624435572 : SUCCESS in 2h 18m 50s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/6faf8439b7ba4c238c2921fc4a503fe5 : SUCCESS in 1h 45m 12s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/36419181c58f4b00a1fb2fa44b4d4a81 : SUCCESS in 2h 10m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/57081a74176e44f69e346b2c59e53089 : SUCCESS in 52m 22s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/bc8f357708a94d95a91aba679a3b24c6 : SUCCESS in 1h 56m 34s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/94cfcc6e9e7447b28559e0e8bb245c68 : SUCCESS in 2h 14m 00s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/cd3dc76c561540bdb0c4e42a16ce5b06 : SUCCESS in 1h 06m 57s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/18a381ec27e54f74af129f19fda07aa4 : SUCCESS in 1h 52m 19s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/6cc5a225372142a1ad175d73ee05b964 : SUCCESS in 38m 15s","accounts_in_message":[],"_revision_number":6},{"id":"7f059396481bc332755c4f44a7f4d80cd1f65c96","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-02-24 17:38:57.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"eb240fb459b6cd23260f80da58656ae788474f24","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2025-02-25 16:01:19.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"21acd8c0fbe1e449aaa605fc5546edd3eebc1599","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2025-02-25 16:12:14.000000000","message":"Patch Set 7: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"db76b1e3614e23b03350311bd99bf98f51c42a7e","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-02-25 16:13:38.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"44b7e3ed98544bdd829d3e6811f2ee0695fc301c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-25 17:01:29.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6ecaddf791654b65adff9c57812a3c0b\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/dd4735f043b345419f625a9a743b265d : SUCCESS in 22m 38s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/694a42fb546046a8a06a403b68a06bab : SUCCESS in 36m 14s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"bb8a513b7c80ae02eedd7e83f620ad29b33a7704","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-02-25 17:18:31.000000000","message":"Patch Set 7: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"4025cfce903379396e5c3b42aa01f46f261f9455","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2025-02-25 17:34:39.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"1bafc2f1962c73a0df841f06b921ca4f620948f0","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2025-02-25 18:19:13.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"14ec3faf6050f8aa06dac192f5975409060ad828","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-25 18:30:50.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ead44a28ab534361b80876cb65bfc67d\n\n- grenade https://zuul.opendev.org/t/openstack/build/a54063399c2348cab926f4f3f199944c : SUCCESS in 39m 43s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/fce4816458eb4deb9c27cc33dccfaeda : SUCCESS in 54m 04s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/517f09203c98494b8086735f8d5854eb : SUCCESS in 1h 45m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9a6daebafaec4e8bb4f064bac5586904 : SUCCESS in 1h 01m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1673eefbbc854058a46e7a296814dbd7 : SUCCESS in 5m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0a697ab151554801987ab0cedd371d46 : SUCCESS in 5m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3b664a9b779f48dfa70bae4075e76b99 : SUCCESS in 6m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54f7d8cd594e48789e8c5cce4c500242 : SUCCESS in 7m 03s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/db8076f995a44d00a2953a1b01e5858c : SUCCESS in 10m 03s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8b45652a40ae49ddb1da419b8cf741ee : SUCCESS in 9m 34s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/35d6a5d7daee4a92b78cbd5c9a79d3b0 : SUCCESS in 13m 28s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/e2a9ae37c4b54fcb90ac78224f106c59 : SUCCESS in 1h 47m 45s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/3abe01874e0145f88d2a36e3bd1940df : SUCCESS in 1h 57m 42s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/bba5f96e634d467188cf5cbf50012eb0 : SUCCESS in 1h 52m 18s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/133e783b62974fb09896e6a2dd4c8e6c : SUCCESS in 1h 44m 08s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/5826f64fcf244e0a928273e0f5c29120 : SUCCESS in 2h 28m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0e60c8ee64434f7a9aeb36c7d46e7d7a : SUCCESS in 48m 43s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/39d59c5315d04180b487ded14117e72c : SUCCESS in 1h 36m 05s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/033a49f9eac54f4e8c33b3e141a341cf : SUCCESS in 1h 54m 30s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/976f1ac4133546a281144e1991b0a408 : SUCCESS in 51m 40s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/cdf674709a504597820afd3ea8d0883e : SUCCESS in 1h 35m 04s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/3bb98e9b24bf4d20afaaac15ed0fe593 : SUCCESS in 28m 16s","accounts_in_message":[],"_revision_number":7},{"id":"42bd61c2a7bd022f2409e0037db0048dcb157f9c","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-02-25 18:37:34.000000000","message":"Patch Set 7: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"d17072318be675f4865dcf97d9dd99305ce08325","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-25 18:38:10.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"47b2aa6dcb010c409330257dfd582478f6ddfa16","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-25 21:19:00.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a318e9c8ca794d75986e3710bd499a11\n\n- grenade https://zuul.opendev.org/t/openstack/build/5bf452649cb8434a82b31dc1ec355f2f : SUCCESS in 52m 39s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2342de9bf0fa49d088939cbdea250759 : SUCCESS in 1h 03m 06s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/be5d595d5b5c436192a88c92a249dd53 : SUCCESS in 1h 59m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/eabf2aaab2b34884941a6cab3f675641 : SUCCESS in 58m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c085a0c49bab485082efe9d5378a79a6 : SUCCESS in 5m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f7acd7a9067d47ca94fcfa8f3dd55193 : SUCCESS in 6m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/99f3a5220128446f8e8d984f1b29281a : SUCCESS in 6m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9800a61b445a4d40a5554d73c71fbf19 : SUCCESS in 5m 11s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/819944940362405fb67681e71a957335 : SUCCESS in 6m 51s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ba5e150df3234a3ea03d8e3bad67266e : SUCCESS in 7m 16s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/cd9f0f54a79e4d0ca9d47bc0234221c0 : SUCCESS in 12m 14s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/e3170c24965241438605bc927fb456c6 : SUCCESS in 1h 56m 52s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/54fee039379f49b0b0f78e5c1fcff9ff : SUCCESS in 1h 42m 10s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/1480369b45eb484ba700d801204444ab : SUCCESS in 1h 53m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2208323786964809b8b3d04cb098283c : SUCCESS in 36m 25s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/0445df0eaf2f4677943431dc48b80018 : SUCCESS in 2h 35m 13s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/a7e74956c5a7477da69ee108c56b0387 : SUCCESS in 26m 13s","accounts_in_message":[],"_revision_number":7},{"id":"fe9136d11e278d08b69c3396d7c7ac5bd7d35573","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-25 21:19:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"6fc3ab8bb01b4c75355d2996814c4d93c57e5314","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-25 21:20:30.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3cdbd1f65fa04b2a8fd29d4a19934f89\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/540d5d61cfac49bd87951932af4f8383 : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"046c738c4e4905eef600d86617d0eeaea6e2dc13","revisions":{"71a9633306b8ea6882732ab46bcb52b4e0db4158":{"kind":"REWORK","_number":1,"created":"2025-01-14 14:33:36.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/43/939243/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/43/939243/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/43/939243/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/43/939243/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/43/939243/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/43/939243/1"}}},"commit":{"parents":[{"commit":"a551c54a93e678d6e07e4af300485aa49f1b34db","subject":"Add GPT disk_format","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a551c54a93e678d6e07e4af300485aa49f1b34db"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2025-01-14 14:31:00.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2025-01-14 14:33:28.000000000","tz":-480},"subject":"WIP: Allow ignoring specific gpt safety checks","message":"WIP: Allow ignoring specific gpt safety checks\n\nChange-Id: Ibf9f5eb4e2378ddd77ebe61170b8e921ada792d7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/71a9633306b8ea6882732ab46bcb52b4e0db4158"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/71a9633306b8ea6882732ab46bcb52b4e0db4158"}]},"branch":"refs/heads/master"},"51157744db0a1d1aaf4dc04862b9f1ebe71063b8":{"kind":"REWORK","_number":2,"created":"2025-01-15 17:36:19.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/43/939243/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/43/939243/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/43/939243/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/43/939243/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/43/939243/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/43/939243/2"}}},"commit":{"parents":[{"commit":"d1d5343910b79311160f607f705edc5bf3bc354f","subject":"Add GPT disk_format","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d1d5343910b79311160f607f705edc5bf3bc354f"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2025-01-14 14:31:00.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2025-01-15 15:38:33.000000000","tz":-480},"subject":"WIP: Allow ignoring specific gpt safety checks","message":"WIP: Allow ignoring specific gpt safety checks\n\nChange-Id: Ibf9f5eb4e2378ddd77ebe61170b8e921ada792d7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/51157744db0a1d1aaf4dc04862b9f1ebe71063b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/51157744db0a1d1aaf4dc04862b9f1ebe71063b8"}]},"branch":"refs/heads/master"},"3a1be1db6cda957f8c308d2bc186a646d4b4df87":{"kind":"REWORK","_number":3,"created":"2025-01-16 15:23:07.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/43/939243/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/43/939243/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/43/939243/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/43/939243/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/43/939243/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/43/939243/3"}}},"commit":{"parents":[{"commit":"aaf6299b7639ae7fdb18a231d50cd27c3871dc60","subject":"Add GPT disk_format","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/aaf6299b7639ae7fdb18a231d50cd27c3871dc60"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2025-01-14 14:31:00.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2025-01-16 15:22:56.000000000","tz":-480},"subject":"Allow ignoring specific gpt safety checks","message":"Allow ignoring specific gpt safety checks\n\nThis allows the operator to configure specific GPT safety checks to be\nnon-fatal. This could be necessary if an operator finds that legit,\nbut non-spec-compliant images are already present in the system and\nneeds time to fix them. See this nova bug for an example of a widely-\ndistributed image that violates the GPT spec for example:\n\nhttps://bugs.launchpad.net/nova/+bug/2091114\n\nRelated to blueprint glance-as-defender\n\nChange-Id: Ibf9f5eb4e2378ddd77ebe61170b8e921ada792d7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3a1be1db6cda957f8c308d2bc186a646d4b4df87"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3a1be1db6cda957f8c308d2bc186a646d4b4df87"}]},"branch":"refs/heads/master"},"efe5441ee0f096e024d3a9d23e6581953d3843ef":{"kind":"REWORK","_number":4,"created":"2025-01-21 15:45:39.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/43/939243/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/43/939243/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/43/939243/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/43/939243/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/43/939243/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/43/939243/4"}}},"commit":{"parents":[{"commit":"aaf6299b7639ae7fdb18a231d50cd27c3871dc60","subject":"Add GPT disk_format","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/aaf6299b7639ae7fdb18a231d50cd27c3871dc60"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2025-01-14 14:31:00.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2025-01-17 15:34:54.000000000","tz":-480},"subject":"Allow ignoring specific gpt safety checks","message":"Allow ignoring specific gpt safety checks\n\nThis allows the operator to configure specific GPT safety checks to be\nnon-fatal. This could be necessary if an operator finds that legit,\nbut non-spec-compliant images are already present in the system and\nneeds time to fix them. See this nova bug for an example of a widely-\ndistributed image that violates the GPT spec for example:\n\nhttps://bugs.launchpad.net/nova/+bug/2091114\n\nRelated to blueprint glance-as-defender\n\nChange-Id: Ibf9f5eb4e2378ddd77ebe61170b8e921ada792d7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/efe5441ee0f096e024d3a9d23e6581953d3843ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/efe5441ee0f096e024d3a9d23e6581953d3843ef"}]},"branch":"refs/heads/master"},"51af50eaa7b190680dd95afe69633884781bfac6":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2025-02-20 21:13:58.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/43/939243/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/43/939243/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/43/939243/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/43/939243/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/43/939243/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/43/939243/5"}}},"commit":{"parents":[{"commit":"9eee7832827e9112b5bda748a1bcd2a172843dd7","subject":"Add GPT disk_format","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9eee7832827e9112b5bda748a1bcd2a172843dd7"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2025-01-14 14:31:00.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2025-02-20 21:13:48.000000000","tz":-480},"subject":"Allow ignoring specific gpt safety checks","message":"Allow ignoring specific gpt safety checks\n\nThis allows the operator to configure specific GPT safety checks to be\nnon-fatal. This could be necessary if an operator finds that legit,\nbut non-spec-compliant images are already present in the system and\nneeds time to fix them. See this nova bug for an example of a widely-\ndistributed image that violates the GPT spec for example:\n\nhttps://bugs.launchpad.net/nova/+bug/2091114\n\nRelated to blueprint glance-as-defender\n\nChange-Id: Ibf9f5eb4e2378ddd77ebe61170b8e921ada792d7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/51af50eaa7b190680dd95afe69633884781bfac6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/51af50eaa7b190680dd95afe69633884781bfac6"}]},"branch":"refs/heads/master"},"4405a990095da842cc12520c1214816df70a39bf":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2025-02-24 14:32:47.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/43/939243/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/43/939243/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/43/939243/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/43/939243/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/43/939243/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/43/939243/6"}}},"commit":{"parents":[{"commit":"4b78a0c9ef32b1b106e44574c0d41a02307cba07","subject":"Add GPT disk_format","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4b78a0c9ef32b1b106e44574c0d41a02307cba07"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2025-01-14 14:31:00.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2025-02-24 14:32:46.000000000","tz":-480},"subject":"Allow ignoring specific gpt safety checks","message":"Allow ignoring specific gpt safety checks\n\nThis allows the operator to configure specific GPT safety checks to be\nnon-fatal. This could be necessary if an operator finds that legit,\nbut non-spec-compliant images are already present in the system and\nneeds time to fix them. See this nova bug for an example of a widely-\ndistributed image that violates the GPT spec for example:\n\nhttps://bugs.launchpad.net/nova/+bug/2091114\n\nRelated to blueprint glance-as-defender\n\nChange-Id: Ibf9f5eb4e2378ddd77ebe61170b8e921ada792d7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4405a990095da842cc12520c1214816df70a39bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4405a990095da842cc12520c1214816df70a39bf"}]},"branch":"refs/heads/master"},"046c738c4e4905eef600d86617d0eeaea6e2dc13":{"kind":"REWORK","_number":7,"created":"2025-02-25 16:01:19.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/43/939243/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/43/939243/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/43/939243/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/43/939243/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/43/939243/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/43/939243/7"}}},"commit":{"parents":[{"commit":"1040997b3313a10a9c6a3518d47cb143c7fe01c3","subject":"Merge \"Run safety checks on uploaded content\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1040997b3313a10a9c6a3518d47cb143c7fe01c3"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2025-01-14 14:31:00.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2025-02-25 15:49:26.000000000","tz":-480},"subject":"Allow ignoring specific gpt safety checks","message":"Allow ignoring specific gpt safety checks\n\nThis allows the operator to configure specific GPT safety checks to be\nnon-fatal. This could be necessary if an operator finds that legit,\nbut non-spec-compliant images are already present in the system and\nneeds time to fix them. See this nova bug for an example of a widely-\ndistributed image that violates the GPT spec for example:\n\nhttps://bugs.launchpad.net/nova/+bug/2091114\n\nFor maximum compatibility, we ignore the mbr check by default for the\nfirst release with this check.\n\nRelated to blueprint glance-as-defender\n\nChange-Id: Ibf9f5eb4e2378ddd77ebe61170b8e921ada792d7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/046c738c4e4905eef600d86617d0eeaea6e2dc13"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/046c738c4e4905eef600d86617d0eeaea6e2dc13"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
