)]}'
{"id":"openstack%2Fglance~939250","triplet_id":"openstack%2Fglance~master~Ic56d897c441004d710ec89a536ad57b2fe480a41","project":"openstack/glance","branch":"master","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-02-20 08:23:59.000000000","reason":"Change was submitted"},"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2025-02-20 05:51:11.000000000","reason":"\u003cGERRIT_ACCOUNT_9303\u003e replied on the change","reason_account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},"16137":{"account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"last_update":"2025-02-20 08:23:59.000000000","reason":"Change was submitted"},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2025-02-18 16:43:17.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"Ic56d897c441004d710ec89a536ad57b2fe480a41","subject":"Fix broken metadata function calls","status":"MERGED","created":"2025-01-14 15:46:13.000000000","updated":"2025-02-20 08:25:05.000000000","submitted":"2025-02-20 08:23:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"939250","meta_rev_id":"ef11f4fede7c736b7717f8f8424c89dfd8d05f35","_number":939250,"virtual_id_number":939250,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-20 08:23:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"recommended":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2025-02-17 13:39:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2025-02-20 05:51:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2025-02-17 13:39:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2025-02-18 16:43:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":1,"date":"2025-02-20 05:51:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-14 15:48:20.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2025-01-14 15:48:20.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2025-01-14 15:48:20.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2025-01-14 15:58:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-01-14 18:19:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-17 13:39:56.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"80c45b654f34dd9c8e3d94966994b3757ca003a0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-01-14 15:46:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dd69d14b4b669caeb2c669d437608fb491a94207","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-14 15:58:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3dd23e87fb644a01836443886f87790c\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/14e95d69fd4f4b239e20a70197bc09db : SUCCESS in 8m 12s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/cd03ade571ae47659760e46f299aa93e : SUCCESS in 9m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4b302f1f44f36727828f560a201b03ac2dd143de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-14 18:19:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ae7d3ecd7eea47b495d0cd9dfd2561da\n\n- grenade https://zuul.opendev.org/t/openstack/build/f213583f670549b6a2a6a28eab0c88e8 : SUCCESS in 1h 08m 02s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/352fc6c0587241739b266af862d5270e : SUCCESS in 35m 23s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e62a99a34a594853b0a738438b93582b : SUCCESS in 1h 27m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4c383e22b8224a02bce38ce81f6159c2 : SUCCESS in 1h 07m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c3359e565e6460e9fb27dd9f0d7f0f6 : SUCCESS in 3m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/166d92ad451340479d3e7d0958b1905e : SUCCESS in 5m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/52b399d7dce649c0830490528aefbf89 : SUCCESS in 5m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b4aa9ce027a466cb635860d5e55e2d9 : SUCCESS in 5m 24s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ebbbf5dd285348b4b0da09072b920f07 : SUCCESS in 8m 10s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b0e3c52780cf4619ab1943f05e5eca51 : SUCCESS in 6m 47s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/1650ac7cceea4097ae63b486e36d23e9 : SUCCESS in 11m 55s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/80c0f326528d429ab2cf1e7e3caaed28 : SUCCESS in 2h 01m 12s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/25881c6da9e642dcaa1e256624bbb2e1 : SUCCESS in 1h 14m 59s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/672fe81d2bf846ddb9fd61d0795ccabe : SUCCESS in 2h 01m 36s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/c28961456a1d4ede93460c48febe0cdd : SUCCESS in 1h 59m 28s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/83b7d41b8144497aa0ed7d7e03a1c310 : SUCCESS in 2h 02m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/54c933214cb44d8bbc760de005398dbc : SUCCESS in 44m 42s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/8379efab1f784a38a86ceea5ba0f09cc : SUCCESS in 1h 10m 19s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/a959813a58c14f72af008c9ae5db506e : SUCCESS in 1h 57m 39s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/2745386c2d93442fa4ba2954bee3468b : SUCCESS in 55m 34s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/02b0aa88d69548e5ac3edc831b6e3a8f : TIMED_OUT in 2h 32m 58s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/a94cdad5adb641178539c09e2a763273 : SUCCESS in 24m 57s","accounts_in_message":[],"_revision_number":1},{"id":"7bf00839ae36e5ecc654254b283db056a1c47a49","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-01-16 12:15:51.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d866142307934d672dab9e67f5e3f764b1d6ba29","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-01-16 12:16:02.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"8b28d482d4bc2fbda0ea5887f805535b738ff4fc","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-01-16 12:16:14.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b489921c3ed44ed28aa7274eb4eefd79d03a85a4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-16 13:25:52.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee7e683ed9cb401cac2e22a533a7e836\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/065e76eb0e7646fd929f5dcde8a37353 : SUCCESS in 15m 05s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/bcfec53116b64d38ad9fdbe0c060ddc3 : SUCCESS in 14m 16s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8d48a6611906208ab07351a8f7278f13a5161337","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-16 15:08:27.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/508a91683190409a806ac06ba43237b5\n\n- grenade https://zuul.opendev.org/t/openstack/build/9ee2e59491f341b6bc212a9a8391e5ca : SUCCESS in 58m 26s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/899e21f2234e4ff88b18e6b7be295c09 : SUCCESS in 54m 43s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3fbf4112bdf64360ac51ac0f8a674d80 : SUCCESS in 1h 24m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0a245dc720ac449ea591b99d31b787e2 : SUCCESS in 44m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92aa8aba7c844682842c6347ad8569de : SUCCESS in 5m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/253c1461d55947f98254b77448a4ce77 : SUCCESS in 5m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1829e4cba07a460ea92480658e12d85c : SUCCESS in 3m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e3c2ce679834776a1a5732972cdfe6e : SUCCESS in 4m 57s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/52fee308d2a7414589d8ed35edef4f74 : SUCCESS in 7m 57s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/94221d4b277949b6bf6ff783504ac2db : SUCCESS in 7m 05s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/55c4fc57cf5841c6a8dbf1abb3a012c3 : SUCCESS in 14m 08s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/b45732186e714092ab727baf45aac888 : TIMED_OUT in 2h 03m 53s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/d1e183773266472fb828c515ae41acb2 : SUCCESS in 1h 35m 45s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/40c75240f6ff44cc9554eaf2b300a0fc : SUCCESS in 2h 03m 13s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/0dc5634928ad437c9d7f48d7748c2f4b : SUCCESS in 1h 51m 05s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/da4e80ab46b14e668befb1300353b00d : SUCCESS in 2h 05m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6e8827b80ac043a3bb4ce5e27297e727 : SUCCESS in 1h 01m 47s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/0189a45c84e34378ad990062c0d2aa70 : TIMED_OUT in 2h 37m 26s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/e6f7967949224528b909cde5c21f4e23 : SUCCESS in 1h 50m 29s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/daa5d69e67844c5889401ef2f7d0191f : SUCCESS in 1h 18m 21s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/2307f4e6f930451c8a174ba82592b688 : TIMED_OUT in 2h 37m 47s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/ac8de2c3f25342748bda190696daede1 : SUCCESS in 35m 32s","accounts_in_message":[],"_revision_number":2},{"id":"ebabd0e8117171e26bc8fa49b1ba13e787492229","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-02-17 13:39:56.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"c9740dc0c9ede27b63fd165cd8cbee5c9f9ad7f1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"real_author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-02-17 13:39:59.000000000","message":"Patch Set 3: Patch Set 2 was rebased on behalf of \u003cGERRIT_ACCOUNT_15334\u003e\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}],"_revision_number":3},{"id":"241d71a6312cac7f72756b7af327cd7f3f4abc02","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-17 13:41:24.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/00e666708e704c24bea69171c27b00c4\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e48749a960b240af9099fdce71e95c17 : NODE_FAILURE Node(set) request 300-0026357679 failed in 0s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d415ae99ab2946a6811af4a50a17cbf7 : NODE_FAILURE Node(set) request 300-0026357680 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"37b6cba68a4172b62da0cbebdde548f52e2efe1d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-17 17:14:46.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2db3c680c6d741b098de6543c5a48385\n\n- grenade https://zuul.opendev.org/t/openstack/build/228871b895364159b4fa8bdd7e3d58e0 : SUCCESS in 53m 24s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/70cd7cbab8b5444cb1de580a9f7ffc1a : SUCCESS in 43m 49s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f38a9efd9ef14bcb8a696bdb1c8bb891 : SUCCESS in 1h 31m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0a938c39c30f44b9a2326724f16f9112 : SUCCESS in 53m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e076521f00c042c898d4c0eb5a313abf : SUCCESS in 4m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4f3cee80a21245f4887f4dd852f3a99e : SUCCESS in 5m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6387c0813a1d489bbfa73550627b9db6 : SUCCESS in 5m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a4c027dfe9d48adbfeb6db3c01c821f : SUCCESS in 4m 59s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/458d534039b04150a3710d56e9f917a2 : SUCCESS in 9m 34s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7e8c36427bee4f098df977ed4b0be5d7 : SUCCESS in 8m 02s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/1b4f104dc9a74cc79a96462f8ec7e781 : SUCCESS in 12m 11s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/c5d1df40805b43e99d725038514014ae : SUCCESS in 2h 05m 02s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/95dde9c26c014d27b640ccbb592ee35e : SUCCESS in 1h 47m 28s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/821a38d487a54c1693a333fc8cacb598 : FAILURE in 1h 52m 47s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/39c8bca4a8f647849e74e5b2dd80dded : SUCCESS in 2h 53m 47s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/0fce42e1c4de4d79bd8c9014d6522ca4 : SUCCESS in 2h 13m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/abc8febbc8764d349e6f00b605ab3658 : SUCCESS in 59m 24s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9ca29ba0aa0146548e7ceb36fe66b771 : SUCCESS in 2h 31m 02s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/f81953c283e54a0181a3d53d88831d6d : SUCCESS in 2h 01m 13s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/67f82b3ffebc4e70ae88b9e658a42e3c : SUCCESS in 1h 11m 19s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/682fc89fe04c4300bc8020acc182f07b : TIMED_OUT in 2h 34m 34s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/a56c9d2be3c048bbab8ccdad84817693 : SUCCESS in 59m 07s","accounts_in_message":[],"_revision_number":3},{"id":"b7790e043d8c2b28666192de5fd65e748b65e5ca","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-02-18 16:43:17.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"7641a3a95e596aa114dfb8ebebea36465b21dac0","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-02-18 16:43:22.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"3a39b4520b356eb47a0c70a56b55a8255924a005","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-18 16:55:49.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/36a380c9b6a843adaeb8f0200e54ecf2\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1ae4145d1e694330a703288278515d46 : SUCCESS in 9m 24s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4e3bf12bd8a844acb939bf0aec46e78a : SUCCESS in 10m 12s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"69e07e589d00d75c42d450320bd9d86501b29946","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-18 19:22:57.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b31a4d33bad431b9fcf04cb6e5c9b2b\n\n- grenade https://zuul.opendev.org/t/openstack/build/3433063454764da4b38d38ac3d54dd11 : SUCCESS in 1h 03m 10s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c080c627fc954791b855d83db58161e5 : SUCCESS in 1h 01m 14s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/32b0cc34df7b4cbe8037e6d29cabac5d : SUCCESS in 1h 29m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5ece02beac6d4cd2883ef00f6cc8392c : SUCCESS in 56m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ab07d9902d247e0bf114572ba139232 : SUCCESS in 6m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b1ffc4c6463a4f8a9fe76e0c528b50f2 : SUCCESS in 6m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3e61bdde1dc0420080c31fc9526b7630 : SUCCESS in 6m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/acac48b5af154b4e8c18755166ddd13e : SUCCESS in 5m 03s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/09a2876c192349f0865068ad82b510c4 : SUCCESS in 9m 19s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e4e236ce07e84f75b380313eee2c8a3e : SUCCESS in 8m 04s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/432caa59b10644aaa418d958c9a8809c : SUCCESS in 12m 55s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/c4c5aa85197f4f50bd2c45d672843c53 : SUCCESS in 1h 49m 32s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/f5c7e0bd814048c5a4f735f13d468267 : SUCCESS in 2h 08m 01s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/3001544690d94147aab83ac99c7c787a : SUCCESS in 2h 10m 28s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/1e248df5fc584953a1199359c55d4242 : SUCCESS in 2h 08m 13s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/ecd6370815564ba2bbb452196cc77141 : SUCCESS in 2h 24m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/453f99d8dead4c21b43966ee8063dd49 : SUCCESS in 51m 48s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/64b1f3d6fc93441487f2e2929cb25d62 : SUCCESS in 1h 42m 08s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/3496a818fbbb45089f73d951decabad6 : SUCCESS in 1h 44m 13s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/a1bcafe8fada4a0cbac9c4699dfc3d67 : SUCCESS in 1h 01m 22s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/c3fd95ef95044e84a52991dee6eb86e7 : SUCCESS in 2h 02m 33s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/a45509349f8c4fb1aa2bf743a002dcd6 : SUCCESS in 36m 46s","accounts_in_message":[],"_revision_number":3},{"id":"a8cd35ca3f4b4edccf302082fc9d15ae08907d66","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-02-20 05:51:11.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"5272e0a7e254d983f985accafa068b8f90328d84","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 05:51:44.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"bdf6334be3f389e750f1ef9287b72d896cf0ce0d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 08:23:59.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c870c2a57cd74a63b33967ec6bef9b24\n\n- grenade https://zuul.opendev.org/t/openstack/build/0fc70457448b40e1869a6c5250a3f484 : SUCCESS in 1h 02m 18s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/80d32ce8195649aeb5eee3524e046594 : SUCCESS in 1h 00m 00s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0e97985ea6c3466d93f27e9a91af1b7a : SUCCESS in 1h 15m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e78ae1ab6d6c4a49aa377fe6f4a4fb80 : SUCCESS in 52m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f4bbfd866a642dca87a067dfeb70d55 : SUCCESS in 4m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3f88c007ffd942e29f3f21e2b80b3386 : SUCCESS in 4m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f44a66127fbf4b5eabdb4aed09045f7a : SUCCESS in 5m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12be1d8fdf2a4bab9960cd8fca1bd3c9 : SUCCESS in 6m 06s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/be213be1272741aca30a217107f5f0ca : SUCCESS in 8m 30s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d459faa894cc4e1789209cee63a8418b : SUCCESS in 9m 28s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/6ba2ca0816054df9a498472b63346e79 : SUCCESS in 10m 42s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/bf0afcaf541b47b89199e99621badba2 : SUCCESS in 2h 26m 08s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/4bf52d5ae62d41fbb02c07762bee52b4 : SUCCESS in 2h 18m 18s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/9d62d8f36c5941819b011feee822588c : SUCCESS in 2h 22m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dc0f84736da94cd8aee6e81903ffa855 : SUCCESS in 1h 02m 44s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/811354c2e80e4f0d9a2b04272fef4bb9 : SUCCESS in 1h 30m 04s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/b7f1695ddc47408398bf9d74df11c1aa : SUCCESS in 36m 49s","accounts_in_message":[],"_revision_number":3},{"id":"23b4c60c27720f4837a4aaefe3b3104960c4fb87","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 08:23:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"ef11f4fede7c736b7717f8f8424c89dfd8d05f35","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 08:25:05.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/080e52464ec84895892e87e2988aa049\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e615845c887f43c691531cb37c4c95ab : SUCCESS in 47s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"e791ed565e9b1a17743306caaf5c7a086fc47aee","revisions":{"25d717658702d45ba1b41e70b6043c9b173cfcb9":{"kind":"REWORK","_number":1,"created":"2025-01-14 15:46:13.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/50/939250/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/50/939250/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/50/939250/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/50/939250/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/50/939250/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/50/939250/1"}}},"commit":{"parents":[{"commit":"8b7c393423441437c1adb34f0d6b945f9fcc7648","subject":"Merge \"Refresh Glance example configs for epoxy m2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8b7c393423441437c1adb34f0d6b945f9fcc7648"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2025-01-14 15:43:16.000000000","tz":60},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2025-01-14 15:43:16.000000000","tz":60},"subject":"Fix broken metadata function calls","message":"Fix broken metadata function calls\n\nThe patch [1] cleanup so that the database\nconnection was re-used but forgot to update\nall function calls causing functionality to\nbe broken.\n\n[1] https://review.opendev.org/c/openstack/glance/+/889380\nCloses-Bug: #2094816\n\nChange-Id: Ic56d897c441004d710ec89a536ad57b2fe480a41\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/25d717658702d45ba1b41e70b6043c9b173cfcb9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/25d717658702d45ba1b41e70b6043c9b173cfcb9"}]},"branch":"refs/heads/master"},"b61a85f9f82bedb4f11b57b7fa961c627467f69f":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-01-16 12:16:02.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/50/939250/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/50/939250/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/50/939250/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/50/939250/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/50/939250/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/50/939250/2"}}},"commit":{"parents":[{"commit":"8b7c393423441437c1adb34f0d6b945f9fcc7648","subject":"Merge \"Refresh Glance example configs for epoxy m2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8b7c393423441437c1adb34f0d6b945f9fcc7648"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2025-01-14 15:43:16.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-01-16 12:16:02.000000000","tz":0},"subject":"Fix broken metadata function calls","message":"Fix broken metadata function calls\n\nThe patch [1] cleanup so that the database\nconnection was re-used but forgot to update\nall function calls causing functionality to\nbe broken.\n\n[1] https://review.opendev.org/c/openstack/glance/+/889380\n\nChange-Id: Ic56d897c441004d710ec89a536ad57b2fe480a41\nCloses-Bug: #2094816\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b61a85f9f82bedb4f11b57b7fa961c627467f69f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b61a85f9f82bedb4f11b57b7fa961c627467f69f"}]},"branch":"refs/heads/master","description":"Edit commit message"},"e791ed565e9b1a17743306caaf5c7a086fc47aee":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2025-02-17 13:39:59.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"real_uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/50/939250/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/50/939250/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/50/939250/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/50/939250/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/50/939250/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/50/939250/3"}}},"commit":{"parents":[{"commit":"f64de3200e7e12d70543f22086b9460cd50dc240","subject":"Merge \"Remove python 3.8 support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f64de3200e7e12d70543f22086b9460cd50dc240"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2025-01-14 15:43:16.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-02-17 13:39:59.000000000","tz":0},"subject":"Fix broken metadata function calls","message":"Fix broken metadata function calls\n\nThe patch [1] cleanup so that the database\nconnection was re-used but forgot to update\nall function calls causing functionality to\nbe broken.\n\n[1] https://review.opendev.org/c/openstack/glance/+/889380\n\nChange-Id: Ic56d897c441004d710ec89a536ad57b2fe480a41\nCloses-Bug: #2094816\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e791ed565e9b1a17743306caaf5c7a086fc47aee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e791ed565e9b1a17743306caaf5c7a086fc47aee"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
