)]}'
{"id":"openstack%2Fglance~939684","triplet_id":"openstack%2Fglance~master~I71d921e0530aad0a7fafcafca1b131297cfa920a","project":"openstack/glance","branch":"master","topic":"spec/list-nodes-image-cached","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2026-05-12 19:08:47.000000000","reason":"Change was submitted"},"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2026-05-12 19:08:47.000000000","reason":"Change was submitted"},"4393":{"account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"last_update":"2026-05-11 13:47:18.000000000","reason":"\u003cGERRIT_ACCOUNT_4393\u003e replied on the change","reason_account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2026-05-11 02:15:39.000000000","reason":"\u003cGERRIT_ACCOUNT_8122\u003e replied on the change","reason_account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-05-11 21:41:42.000000000","reason":"\u003cGERRIT_ACCOUNT_37598\u003e replied on the change","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}},"hashtags":[],"change_id":"I71d921e0530aad0a7fafcafca1b131297cfa920a","subject":"List cached nodes","status":"MERGED","created":"2025-01-21 06:54:12.000000000","updated":"2026-05-12 19:10:56.000000000","submitted":"2026-05-12 19:08:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":37,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"939684-spec/list-nodes-image-cached","meta_rev_id":"6e60d90cb12c7f103b0a09dcc25a600d7a2c259e","_number":939684,"virtual_id_number":939684,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-12 19:08:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2026-05-11 13:47:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":2,"date":"2026-05-11 02:15:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-05-11 21:41:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2026-05-11 13:47:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-21 07:14:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-01-21 09:30:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-25 10:37:45.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2025-08-25 10:37:45.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2025-08-25 10:37:45.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2026-04-28 16:35:22.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"}],"messages":[{"id":"81ce4c2bb5f4e02ad0ddf7980ae1b7061e9dc84a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-01-21 06:54:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"47e5cc079db4a193841becc2a684c979b2fbd434","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-01-21 06:55:00.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6a2b9e7eca18807af06e0fa5e5833ad7d909170f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-21 07:14:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b3150f8a6dbb4c43b022d063188f257d\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3744beb596b34cf29a5be63f8129ecac : SUCCESS in 14m 31s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/91c2df55f10745b388dd0853519d64e7 : SUCCESS in 19m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f14141a9cc5d158901908819028bc8cf0563fbed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-21 09:30:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3e2745edd6a4496b8fa65fe66baf2b4a\n\n- grenade https://zuul.opendev.org/t/openstack/build/cd3a4f41fffc4bf98a3c64ffde8dcf19 : SUCCESS in 52m 11s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8191b9598596428e9c2c03609d4de062 : SUCCESS in 38m 17s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/726264e35c0d48e095083db2d1bb4ee7 : SUCCESS in 1h 27m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4e8a0596a0714b62ba1a76fe8b506b89 : SUCCESS in 1h 13m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ce88e34ebc14a3bb69d010b3d45cc0a : SUCCESS in 4m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/324be9ea9ecc4e1790a79a14c9edf654 : SUCCESS in 5m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7374dbe284934b96a7bd40f156ee3412 : SUCCESS in 5m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/931dbb21a00a4cee83f747c125d3784b : SUCCESS in 5m 24s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4f9bbcc7027f49e7a9fde2a784fc7a52 : SUCCESS in 8m 46s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/fd01e6b0daac48ddbd2a13ca74f901c1 : SUCCESS in 9m 30s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/4c59a386bb4a4669bd284b1abb7d3309 : SUCCESS in 10m 22s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/460c4bd4e0624b498206de09ed601972 : TIMED_OUT in 2h 04m 17s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/db715d15b5af46c1aecbdb60d600b543 : SUCCESS in 2h 26m 07s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/0218c83b4004466588dfcdfe3986ed3f : SUCCESS in 1h 39m 29s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/a9968c7d5a7e4eb8a6e6e1774edd1a4b : SUCCESS in 1h 52m 36s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/7c83814705b846c49acedca1f4582e64 : SUCCESS in 1h 40m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/68b94e45352f40c3a0a04162b36dd14d : SUCCESS in 1h 02m 19s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d7aa2cbfe5ec47a1afefc78973780400 : SUCCESS in 1h 50m 36s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/cb4d87a734fd4e5381f8556b16066c87 : SUCCESS in 1h 58m 33s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/44e01a13f84f404e9fef56fa76e4f752 : SUCCESS in 1h 06m 05s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/3be3bbb28c924585a8404ae858bfd0aa : SUCCESS in 2h 03m 33s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/84495a6a27ea4e099d9ba8b0f783e83e : SUCCESS in 26m 43s","accounts_in_message":[],"_revision_number":1},{"id":"b62679dafc70857829237e09a9e04d7d837dd4da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-02-12 08:35:21.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"b874162c7beff8f13eaa19331d1fab11ffbb923f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-12 08:49:34.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3f3f595ac12a4071b2546aef392ec41b\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/84e2a056bd6948fb8f5f042571b3472e : SUCCESS in 12m 28s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f32ad27ef6ce47ec9449ca779a6b0ca7 : SUCCESS in 11m 56s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a59588c6f89c676f23d0f7ace7b9f417e9142ac7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-12 10:45:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d7955c68db046ffa36b989ff74b4b18\n\n- grenade https://zuul.opendev.org/t/openstack/build/b443e039c23f4c4b823bf92c44de5fc5 : SUCCESS in 1h 00m 58s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8e5ba503032f409d98725d401aec785c : SUCCESS in 1h 02m 55s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a00e0d67625148dcb7c9582d716dcc07 : SUCCESS in 1h 50m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ea44aa6f19974d0abf7a415443001f8d : SUCCESS in 54m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12ffda2439ac47b2ad93c61211c5746e : SUCCESS in 6m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6cfb900c486b407e90d47763b9ab8586 : SUCCESS in 6m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/027a817d79fc45a5b70857de6d42edce : SUCCESS in 5m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43385cd9dfc5450386305dd00bca98e5 : SUCCESS in 3m 39s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/79556c41059342b2b8a639e402c43f1f : SUCCESS in 7m 14s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f0d237b5035b43df991dea5afbb1a872 : SUCCESS in 7m 12s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/9279d9e8ec714095acd2a815649876a2 : SUCCESS in 12m 32s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/101fa8b0dace4ff39502d23761998b79 : SUCCESS in 1h 33m 12s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/bd2bf0b0ed894c4e823806c4dadb7858 : SUCCESS in 1h 19m 55s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/65bb827b6999490eb7c9a6472c3227c3 : SUCCESS in 2h 03m 33s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/b19c2ea3c24d467fbedc796d2a56e380 : SUCCESS in 1h 55m 36s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/7fd0aa86789a4349b156046a10da51ae : SUCCESS in 1h 21m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fdb3b90a0d3c4d57bec307301493fb60 : SUCCESS in 59m 11s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/dd751db20caa4051b9230bfdfa5849ed : SUCCESS in 1h 39m 35s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/a19f5478b4e74cac9a4cede06d5f6076 : SUCCESS in 1h 53m 29s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/2d0dd638410c401cbfb3a9eb2f691234 : SUCCESS in 55m 55s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/222128a4cd554906b432aea45a0b41df : SUCCESS in 1h 50m 07s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/dbc3e42ae1724c3394578c3abad91f3f : SUCCESS in 37m 35s","accounts_in_message":[],"_revision_number":2},{"id":"aeed0acea25fa58c4d809e7cb4eacdfce443ee65","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2025-08-26 13:46:33.000000000","message":"Patch Set 2: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":2},{"id":"37560f73ce68e0a0ec362271f14b2b0840b2cd4c","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-08-28 14:54:59.000000000","message":"Patch Set 2:\n\n(7 comments)","accounts_in_message":[],"_revision_number":2},{"id":"5be9fd364f3852344a0e2fd653a1222d20ac4d6f","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2025-08-28 15:10:09.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7c846e592437ff77717faae3eba067c9b2768cfb","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-08-28 15:20:59.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"47b4b7f71da07807ad6d6032dc44ebbba2bfe3fd","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-08-28 15:26:11.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e2429633f205aeb75439bb40fcfbc2220ca66ce7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-08-28 16:58:50.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"331d419c76e15040ce90c8628354e76db4c578b6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-28 19:11:37.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3f4abc703ed2455a98ceaed7ba0f9ec4\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9d2bda814f774e35acff9a88e3bb0542 : SUCCESS in 9m 03s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8ea7f291de2940cca6e0ffd8498c76b4 : SUCCESS in 13m 26s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/5258c5eb1b7b42309e2da105f138916a : SUCCESS in 16m 14s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"29ec2090975f2e8893d8bf2f7c1c39d5a1fe06da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-28 19:29:33.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7bab3df4f32c4c24972afa84493ae3c1\n\n- grenade https://zuul.opendev.org/t/openstack/build/2f63bc88a28148f482d0c33598c35ef4 : SUCCESS in 33m 26s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e12d89730fc1463dad48ec2bca1d22e6 : SUCCESS in 41m 50s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c16fefb523e543cf8a2ce025b32f1f66 : SUCCESS in 1h 32m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6475bcae7a3049649dc3ebfe99803f7c : SUCCESS in 31m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a2609b7101d4e7da6c3a2eda60eddad : SUCCESS in 2m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8ca41eebd3e1406cb8438cac0b01798a : SUCCESS in 4m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/dee563c1f4fa425783d47e88129ecfe7 : SUCCESS in 3m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/27d4b8560f4744be997c4562e3a1d5e0 : SUCCESS in 6m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9dc74dbdb0784cb1895fdc216bc4c0f8 : SUCCESS in 5m 05s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0261c195118c454e88e3ed8409cb9f9c : SUCCESS in 6m 01s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/8d1ed437ce9b424ca6f2343e2b2fdf47 : SUCCESS in 12m 15s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/a69b87b90e1d4b069cfb69b33de89622 : SUCCESS in 2h 23m 59s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/5f0846358d554eb3bc72e7455d4424e9 : SUCCESS in 1h 55m 53s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/ee8b7079eb574c3ea8e44011ad0fa0b6 : SUCCESS in 1h 47m 18s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/b1ddb063bad94a96a02c58f2ee64cc99 : SUCCESS in 2h 11m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d61a5736c02844189b26d704ebe0676d : SUCCESS in 56m 10s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c297e1a005064951a77edb4785480b94 : SUCCESS in 1h 51m 17s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/bb4168757dfb43b18e923d899b476903 : FAILURE in 16m 05s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/a4d327b40b514dfdaff92f2acf023054 : SUCCESS in 1h 48m 24s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/58335e4709cb4aeabff0a3f97b53c0d4 : SUCCESS in 25m 07s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/8bce7baa56204a6082476b51c17f3713 : SUCCESS in 33m 33s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"293e48e9aa809387378dc5aedca01a88654d60d8","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-11-27 16:24:02.000000000","message":"Patch Set 3: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":3},{"id":"faf007164cc5af436feaa5b7a8847bd3a72892cb","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-11-27 16:37:30.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"32bbe1515856cd2ca10a28322672702c22c84a5e","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-11-27 16:42:04.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"9987f29fa513527659d8f0029cc9e968f848df8e","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-11-27 16:44:02.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"5794b6184385d5d7a63325b9e7828c1b8ddab5ce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-11-28 07:22:06.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"c33c1775edac174f5b8c85b9f3bf54af0bffe982","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-11-28 07:24:03.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"add667428de0ee293bc1a344defc2b1949d3c546","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-28 07:33:53.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/35f6b9804cf3490fbec04cbf3b18c829\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/27d825336e164db3b624ec387dfd7ba1 : SUCCESS in 6m 00s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f27b4f7ae2624d5f86aebcae54458692 : SUCCESS in 8m 04s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ba33683bd2af4174bba1880b673663f5 : SUCCESS in 10m 17s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"546ec71cec9eb20154a4ebf9c774ec5677f4cf03","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-28 09:23:47.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/09af80e515564443a35a790a9b84c750\n\n- grenade https://zuul.opendev.org/t/openstack/build/3e2eea4cdb564ecb921fc234d04a8286 : SUCCESS in 44m 00s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/12762f767f6e480c823599c553815530 : SUCCESS in 1h 02m 24s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/39889f2df84a48ea852404c5023209dc : SUCCESS in 1h 19m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dea38a92cc9941d3aa241e703b0dfb52 : SUCCESS in 1h 06m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78f6f21f89974e229142ab1e11a58a2b : SUCCESS in 4m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/61081a9729db4425ba687ee29e6f7d5f : SUCCESS in 5m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7a165ff4be51497da844e468db8a93fe : SUCCESS in 6m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62c8d6f8bd804860bde2b5d4974b73fe : SUCCESS in 5m 32s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1696bb56341c4ad99f8b9677bdd10c65 : SUCCESS in 6m 40s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/5c29519492e5468fac0e943b8b9eb3e7 : SUCCESS in 8m 19s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/7930801f9798450bad6ed43ef3ad9ed7 : SUCCESS in 1h 57m 08s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/c79b62cfaa3646199eeff0d537c0bd16 : SUCCESS in 1h 09m 14s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/671bfb4e0f35466b85273710f18cb9f5 : SUCCESS in 1h 36m 49s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/4b4af751f0724bdbb4849425fd2c9c13 : SUCCESS in 1h 43m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/349c92672d9045c091f824982910dde6 : SUCCESS in 55m 44s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/b296877c9b62438db187febf3014f66d : SUCCESS in 1h 26m 00s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/e437cb42772a409aa4677a8867360a52 : FAILURE in 21m 44s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/fa58e0a860d243bb8270437d422149bd : FAILURE in 17m 57s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/baf8d68f4687482d926bc96c64bd62f6 : SUCCESS in 32m 40s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/704a5059e23d4b1f9fb458357b2aa84c : SUCCESS in 1h 14m 16s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"549c009618b8679e3946e3e89d34015369746a84","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-11-28 15:24:29.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a0d753d1fa3f52700ce51ba48cf4dc9d3b322936","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-04-28 16:35:22.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"285d67cd088383bcb12a83bc13274e700cb09eca","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2026-05-07 15:55:04.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"bf82e1924217a5fa9629c542c44449cde1a45d2b","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-05-08 07:02:21.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"29be53129631b5b23a33436e46150bfcd65ea4b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-05-08 09:00:22.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"60b8131c998dda187b774fd2655bae681bf7dff5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-08 09:31:47.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ce52f24cda21499794be8ed13e02e6f5\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3939f45eadc944a6b308dd4e9c755a22 : FAILURE in 5m 57s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/22edd72f7e13460c858173609be0409a : SUCCESS in 5m 42s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/b476139c1ff1405286a9f38f2c350dff : SUCCESS in 8m 10s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"9de9e75e143106b8c3e6ec805e19c0de61892054","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-08 11:01:18.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f5887948c9b41429fa247695e188623\n\n- grenade https://zuul.opendev.org/t/openstack/build/5963179fd6de4c24b04369dcb783282b : SUCCESS in 55m 25s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e6ced7faa31046a68d9411db3f12e79b : SUCCESS in 58m 57s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/70b7bc6f05474a24b70d3d1128ca590b : SUCCESS in 1h 52m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cf9645743e2545a2bca52d8af73db2b7 : SUCCESS in 58m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0b69e002d27490aa357a1767e13a718 : SUCCESS in 4m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1a05dd846f9d4620a1a7206d7586f28f : SUCCESS in 4m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/03b20097ce62466dbb7c96456cda6b02 : SUCCESS in 4m 49s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/87b5d8f69680476b8a8f720fe184398a : SUCCESS in 7m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d096f3e9ce647ad86403cda98838717 : SUCCESS in 6m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2e8b51e6d7bc442197071740ccf2a364 : SUCCESS in 5m 54s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/c4b467321cea43b580e079fab5be5820 : SUCCESS in 8m 08s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/89868b98db1949c78798ef3f0ec49f13 : SUCCESS in 8m 08s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/e0dee1113f424ca48651ccd3ddd74c43 : SUCCESS in 1h 45m 41s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/aa24506ab62b4300a43e092bff110f5f : SUCCESS in 1h 02m 46s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/2275533866f94338a7ed3e29769d8b90 : SUCCESS in 1h 06m 43s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/ddfaea26563e4f23b8bc204a7fce5c7f : SUCCESS in 1h 10m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d9cea911514d4c11a294039069ef7ad3 : SUCCESS in 34m 02s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/5fd6be2976e24dd493d2f95931679fb7 : SUCCESS in 1h 55m 11s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/84bdd803476348eb8dac2fc600963eb8 : FAILURE in 9m 26s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/cbb609a726774821aa64e6f3f5a42c80 : SUCCESS in 1h 28m 19s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/2beddcec06f045a59b72a4fcadf1cacc : SUCCESS in 29m 14s","accounts_in_message":[],"_revision_number":5},{"id":"8263124f12fee16cf56812bfb46fe19dfd2bedf7","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-05-11 02:15:39.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"944ffd4133b569f827940b969a195ebd454e67f4","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2026-05-11 13:47:18.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"324add312a271df20b4f15dccebf3e63004f6fff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-11 13:48:30.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"5dd09b3e0e44847e28b17bfc18795cd254c964a4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-11 15:47:59.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c936e0cb83784c15b668d54c0601e0d1\n\n- grenade https://zuul.opendev.org/t/openstack/build/743bce3d47ac449592eb23dcbdc5b3c2 : SUCCESS in 59m 38s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/293ef7cc08cd4299a0ad2a9cb7a0d0a9 : SUCCESS in 1h 03m 23s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/935f70c812684048a6c4d88758efda0f : SUCCESS in 1h 32m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/03b02bea2d3c41be906c50a90e3e2562 : SUCCESS in 52m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e5cc76c8fb24f098f383c55e03702bc : SUCCESS in 4m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/342bfa4b478a4f7087da54a1c74a6506 : SUCCESS in 4m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9940ffbc390b4920bd2602ddb4e9c327 : SUCCESS in 5m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74c2caa8640c4a3fa674dbbc8d6f370d : SUCCESS in 5m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b07a145277ad423c987c29516b1ea6ea : SUCCESS in 7m 06s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/1e549faeb2904649aeb9346ceed8eaa9 : SUCCESS in 6m 47s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/b0ea46418a494e44ad925fddb9b0afa4 : SUCCESS in 6m 46s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/14709ef7f140460cb3250d8c111ffcf7 : SUCCESS in 1h 57m 44s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/fe49e90433104cc58fec12e93c97c948 : FAILURE in 1h 44m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/522af7a6f5cd4eedb84f828c82c0c779 : SUCCESS in 56m 51s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/efd394c768d245fb85d63769dd468c61 : SUCCESS in 1h 50m 26s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/e0078ca0e0144752b20867f2437be751 : SUCCESS in 23m 18s","accounts_in_message":[],"_revision_number":5},{"id":"5d01794fc8d8ce96416a30345437aa3f2808f465","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-11 21:41:42.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"98f14344bb263586f728ee37d42dc418b08c8f47","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-11 21:54:03.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c9e155ca06f34c93bfa697cde91c28b9\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/677c39cfa9bf45d6869e5e3df9843151 : FAILURE in 5m 29s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a26fd8a4390e46e0b52bf247662cbe84 : SUCCESS in 7m 06s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/47cb1408f10b4e3ea1a092ed41990f9d : SUCCESS in 10m 42s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"ee455117d93c3ed89990eab690a803841ec80676","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-11 23:54:33.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f7263955c52a4d49898fad7124be6d21\n\n- grenade https://zuul.opendev.org/t/openstack/build/1c11bfd9451d4aecbdf287c546687c35 : SUCCESS in 40m 08s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/cc0b856b4e994403a54a5424149bdea7 : SUCCESS in 1h 03m 50s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4c8d8ce981e4492fb36adac75f829fbe : SUCCESS in 1h 22m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e7fbe523509947a9bd49c9759a872c75 : SUCCESS in 48m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c4671741dc04b40b7f7692ba053a887 : SUCCESS in 4m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6ec660c7bd47431c876a32a5107acd1f : SUCCESS in 4m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/eefe83c4d11d427787ba10534f92892e : SUCCESS in 4m 29s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d1fa4cbf6ec9424ba77c26a484fe0286 : SUCCESS in 7m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3311418e982f4a16a4a6581488d95f7d : SUCCESS in 3m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8b7b90d90f68480690954e40b604d6f5 : SUCCESS in 7m 21s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/df7b69374e7d49808f4921c8eecf95c8 : SUCCESS in 8m 02s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/18b0a9803abd4681a5f1378813069b7d : SUCCESS in 6m 26s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/828c11f7962a4c518c47f690aa6a2b3b : SUCCESS in 1h 55m 39s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/21710336a46e4244b439ce1e995a00a8 : SUCCESS in 1h 01m 24s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/c14c6e8c62284b798bb08e8bf90c61e5 : FAILURE in 2h 07m 26s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/7d86a76ea0e048dd846b792422527f90 : SUCCESS in 2h 05m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e6d82a2470544aa8a42763c1ef9ed06e : SUCCESS in 1h 00m 12s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/160447d0a910444f9834044b31ec1538 : SUCCESS in 1h 38m 07s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/22a9b990a64244e9a37908e4d6437571 : FAILURE in 14m 56s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/5b6962b1e55b44a5890083c89b09b447 : SUCCESS in 1h 52m 05s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/13bea258951d4a82b8c001a54e379b3f : SUCCESS in 33m 19s","accounts_in_message":[],"_revision_number":5},{"id":"adbd4603711c2a67363bd41c98f777d183b7238a","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-05-12 06:44:16.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"a718e20b5c39b08fbfc8c89de308f25bedbce620","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-12 07:13:48.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0c4b5d9825b44c4fa3d3b77071caa9ca\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f10d85c7b78b43bf89862faeed9643e4 : FAILURE in 5m 02s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/8d6e0fd79b314ae7b97fd1d2eaa3b9b7 : SUCCESS in 6m 19s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/ff266270bd4a4fd8935d8efe71e1366d : SUCCESS in 8m 55s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"9c531f27ce533b58121d069865ea6e0f2b9d4805","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-12 10:00:38.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5d41dfb6262b4af99dec25dad72bde40\n\n- grenade https://zuul.opendev.org/t/openstack/build/2fb4c869513143e5a8ff9be8c123c2e1 : SUCCESS in 1h 00m 11s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/74f09590e1ac452d85251438b5fe5ed6 : SUCCESS in 30m 51s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8b745eb9e52345feabc5850fab682165 : TIMED_OUT in 2h 06m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/01d64a141fec4e6687fa42debd81ff66 : SUCCESS in 50m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75c5c00e61c04dfca2a7f64d982cb0f2 : SUCCESS in 5m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/502ce5fc31d645e389d71151f6c89177 : SUCCESS in 4m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1b96b5bb45024a969f2930456a4ab49e : SUCCESS in 2m 50s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/72dcbb676e074722bd8a083875a4bcb4 : SUCCESS in 6m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f375e8265aca42e4b7c955267a9960fa : SUCCESS in 6m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/636488ee1fb6462ba03003bf7641fd00 : SUCCESS in 7m 44s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/3c0e69752c3140f5a06c7db9d345a869 : SUCCESS in 6m 20s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/9f7e0400ee81410c98672317f9dd8a65 : SUCCESS in 5m 57s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/46dbb7e114e94d07b3cd41eb0e25bd6f : SUCCESS in 1h 34m 03s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/bf94dbc3cd79477981a569a7515ccc3d : SUCCESS in 2h 09m 11s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/f94ca91390cd47d583353ac8a9a14c8b : SUCCESS in 1h 08m 35s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/ab8a5f0961d04e7bb9ea9ac8f864b14a : TIMED_OUT in 3h 06m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a58b158d95a4449eb0ecc50846051dec : SUCCESS in 42m 38s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ca4d78a45d904ec4beb6cc0217968892 : SUCCESS in 1h 29m 06s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/5ded545434534a3ebe6c8f5099a6863e : FAILURE in 7m 42s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/a494d7d6161c45159be2083f7c4c51e1 : SUCCESS in 1h 35m 33s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/8ad6e73ceec04bc4831d840a41be891f : SUCCESS in 37m 25s","accounts_in_message":[],"_revision_number":5},{"id":"b9e75d1ec5f05bb1ad91b69cbfaedee006b5f1fe","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-05-12 14:05:39.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"6f26e5d838b6f50817d1d142a9a17210c272c873","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-12 14:18:47.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c3207e57cba247f79cbf19d0e528b55e\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3612862bb3df4048ae0e8e4b08596073 : FAILURE in 5m 15s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d6d3f9c0d8084649858dc8ecf1e92300 : SUCCESS in 5m 34s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/056a6fe03b4e4b7080961ff3dff11b7d : SUCCESS in 10m 37s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"0d20d9d38a5d77922e31dd3f284c8f55894c315b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-12 16:24:34.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/617ea0761c034c11906677cd30814b6f\n\n- grenade https://zuul.opendev.org/t/openstack/build/963b5037d1634b6e908e381fb5cee2d7 : SUCCESS in 1h 04m 26s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5ec724f031304ad6b9bedb918ae996c1 : SUCCESS in 1h 19m 58s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5f3b94a1951f4258966955a311b6ef58 : SUCCESS in 1h 16m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bf13e992ff2844499a2270dd54b3d495 : SUCCESS in 41m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc752ae8d5a04961bc6c71f1622df6fa : SUCCESS in 4m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c7a9835064cc4335a5b32aebe595ff63 : SUCCESS in 3m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e7157557906745da9d10ae31892eb41b : SUCCESS in 5m 12s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8a77fa0eabf646a9bc70648aca0cd5ab : SUCCESS in 9m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dfeb2ec789524ee4ae1f465479255e93 : SUCCESS in 6m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/28535c826541497ca69534c7a7a4a5ba : SUCCESS in 6m 15s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/c462bf735be14672ba7cb52ee526e8af : SUCCESS in 7m 20s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/a93efc7ebafe4246982cb9a28d20abaf : SUCCESS in 5m 29s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/d67c733b1e4b4d23a9e6fb4ec14be8f7 : SUCCESS in 1h 24m 19s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/6a3237591bc54f648add1c10d500b306 : SUCCESS in 2h 09m 24s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/0c58432b90234429856b6021c1630d7c : SUCCESS in 1h 12m 10s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/846cca0cea8149fbbd780c6265306e0d : SUCCESS in 1h 38m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6f84c8ef90334fafa87532f256329e89 : SUCCESS in 1h 03m 27s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/20c154edae364d77a088e6f21aaca52c : SUCCESS in 1h 48m 58s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/faea16886d80443c8718f967950e4efb : FAILURE in 9m 27s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/5364c3e28302411b8a5b4ac4c7cf08f5 : SUCCESS in 1h 56m 09s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/861d17db9d7d45a9b00476ed8e59e1bd : SUCCESS in 36m 26s","accounts_in_message":[],"_revision_number":5},{"id":"63e2c5116d09dab603771f63a0633510eeb3ccd4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-12 16:25:07.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"1d2cbb733a441d5c5c089c95615182de8bfb026e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-12 19:08:47.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eec2daf6e042459a958c5b08a4e1bc59\n\n- grenade https://zuul.opendev.org/t/openstack/build/bc467fce972e4c3f9b0f1116c9b3d1d6 : SUCCESS in 1h 18m 10s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5336a25e29414d9dabab4848afe85f7c : SUCCESS in 40m 22s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/577502a2f2eb4ab78b4ff7e8ccf4721a : SUCCESS in 1h 33m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/aa11840b6d394bcb8e947109c272701d : SUCCESS in 1h 05m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0eb6c4a346354a0697cb50d52848c0d0 : SUCCESS in 3m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/eef0d1b08d974016922d16b75c9ed7a9 : SUCCESS in 4m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e376854749594d00b91d2602df8c30db : SUCCESS in 5m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eacd608b8fd14c9b8c72fa8a066f588d : SUCCESS in 6m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/70e0d60966764ba4a2ff59f2e71fa3ca : SUCCESS in 8m 28s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/5daad15539494648be425ca7d6e89fd2 : SUCCESS in 6m 24s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/d9554f6de5ae42589cdcf9feb31f64c6 : SUCCESS in 5m 16s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/1610166778c946deb66c7649e6c2bb04 : SUCCESS in 2h 36m 50s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/0d23700a8dbb415abbbc2a79186b3d85 : SUCCESS in 1h 20m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/684da106c7aa4f89be532017838d70cd : SUCCESS in 57m 23s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/cb133b6cec3e4d48974c6895dda77956 : SUCCESS in 2h 10m 53s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/a308102ef8f44762b5e5493a7e52a038 : SUCCESS in 22m 04s","accounts_in_message":[],"_revision_number":5},{"id":"0aa8f1cbde48d4adeeae0d33f5243eb0ebf3b0c5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-12 19:08:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"6e60d90cb12c7f103b0a09dcc25a600d7a2c259e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-12 19:10:56.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/68d1e94370b04e56a2468f43631ac6d7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef5e02bec5c44a4299cc09c9c691ba53 : SUCCESS in 45s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c4612aaa3a574bb981ece76b512daaff : SUCCESS in 43s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"783f9f937eaaadad8428171895639f884a712fd2","revisions":{"9be74c76e8b6b6d2534882eb28fb64c675616f40":{"kind":"REWORK","_number":1,"created":"2025-01-21 06:54:12.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/84/939684/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/84/939684/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/84/939684/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/84/939684/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/84/939684/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/84/939684/1"}}},"commit":{"parents":[{"commit":"faa436bee114334314103044a37a7b6125ce896f","subject":"Merge \"Move all periodic tips jobs from py311 to py313\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/faa436bee114334314103044a37a7b6125ce896f"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-01-20 06:22:18.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-01-21 06:53:01.000000000","tz":0},"subject":"DNM List cached nodes","message":"DNM List cached nodes\n\nChange-Id: I71d921e0530aad0a7fafcafca1b131297cfa920a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9be74c76e8b6b6d2534882eb28fb64c675616f40"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9be74c76e8b6b6d2534882eb28fb64c675616f40"}]},"branch":"refs/heads/master"},"d10591ed61a35812208a00265d3128d24c136b72":{"kind":"REWORK","_number":2,"created":"2025-02-12 08:35:21.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/84/939684/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/84/939684/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/84/939684/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/84/939684/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/84/939684/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/84/939684/2"}}},"commit":{"parents":[{"commit":"a293432c3bfe61691b642fcc96688608b6c4f5d1","subject":"Merge \"docs: Fix possible values for enabled_backends\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a293432c3bfe61691b642fcc96688608b6c4f5d1"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-01-20 06:22:18.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-02-12 08:28:32.000000000","tz":0},"subject":"List cached nodes","message":"List cached nodes\n\nAdded new API /v2/cache/nodes/{image_id} to get nodes where\nimaeg is cached.\n\nImplements: blueprint list-nodes-image-cached\nChange-Id: I71d921e0530aad0a7fafcafca1b131297cfa920a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d10591ed61a35812208a00265d3128d24c136b72"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d10591ed61a35812208a00265d3128d24c136b72"}]},"branch":"refs/heads/master"},"25cc0da288860a97ced73164f5a7ba8da1594d96":{"kind":"REWORK","_number":3,"created":"2025-08-28 16:58:50.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/84/939684/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/84/939684/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/84/939684/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/84/939684/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/84/939684/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/84/939684/3"}}},"commit":{"parents":[{"commit":"6c33a667a9f5ddce07b6131f4a5cb7460a4bdf17","subject":"Merge \"Replaced usage outdate egrep to grep\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6c33a667a9f5ddce07b6131f4a5cb7460a4bdf17"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-08-28 15:02:36.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-08-28 16:57:42.000000000","tz":0},"subject":"List cached nodes","message":"List cached nodes\n\nAdded new API /v2/cache/nodes/{image_id} to get nodes where image is cached.\n\nImplements: blueprint list-nodes-image-cached\n\nChange-Id: I71d921e0530aad0a7fafcafca1b131297cfa920a\nSigned-off-by: Abhishek Kekane \u003cakekane@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/25cc0da288860a97ced73164f5a7ba8da1594d96"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/25cc0da288860a97ced73164f5a7ba8da1594d96"}]},"branch":"refs/heads/master"},"1fe36f662a1753445d7ccf4dbf54510bb520fe01":{"kind":"REWORK","_number":4,"created":"2025-11-28 07:22:06.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/84/939684/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/84/939684/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/84/939684/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/84/939684/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/84/939684/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/84/939684/4"}}},"commit":{"parents":[{"commit":"c98146c422dc1ad24b37098fbb2fdf779f3c00df","subject":"Merge \"Inject os_glance.* section in the property protection file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c98146c422dc1ad24b37098fbb2fdf779f3c00df"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-08-28 15:02:36.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-11-28 07:21:52.000000000","tz":0},"subject":"List cached nodes","message":"List cached nodes\n\nAdded new API /v2/cache/nodes/{image_id} to get nodes where image is cached.\n\nImplements: blueprint list-nodes-image-cached\n\nChange-Id: I71d921e0530aad0a7fafcafca1b131297cfa920a\nSigned-off-by: Abhishek Kekane \u003cakekane@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1fe36f662a1753445d7ccf4dbf54510bb520fe01"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1fe36f662a1753445d7ccf4dbf54510bb520fe01"}]},"branch":"refs/heads/master"},"783f9f937eaaadad8428171895639f884a712fd2":{"kind":"REWORK","_number":5,"created":"2026-05-08 09:00:22.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/84/939684/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/84/939684/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/84/939684/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/84/939684/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/84/939684/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/84/939684/5"}}},"commit":{"parents":[{"commit":"dc5b35d66dd04a88092cdd1ee310ef76d2df47f1","subject":"Merge \"Change DB migration constant to 2026_2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/dc5b35d66dd04a88092cdd1ee310ef76d2df47f1"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-08-28 15:02:36.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2026-05-08 08:58:45.000000000","tz":0},"subject":"List cached nodes","message":"List cached nodes\n\nAdded new API /v2/cache/nodes/{image_id} to get nodes where image is cached.\n\nImplements: blueprint list-nodes-image-cached\n\nChange-Id: I71d921e0530aad0a7fafcafca1b131297cfa920a\nSigned-off-by: Abhishek Kekane \u003cakekane@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/783f9f937eaaadad8428171895639f884a712fd2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/783f9f937eaaadad8428171895639f884a712fd2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
