)]}'
{"id":"openstack%2Fglance~940528","triplet_id":"openstack%2Fglance~stable%2F2024.2~I662ba19697e03917ca999920ea7be93a0b2a8296","project":"openstack/glance","branch":"stable/2024.2","topic":"cherrypick-bug/2086675-2kfaj5717mc","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2025-04-25 13:30:16.000000000","reason":"\u003cGERRIT_ACCOUNT_5314\u003e replied on the change","reason_account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},"4393":{"account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"last_update":"2025-05-12 09:57:20.000000000","reason":"Change was abandoned"},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2025-05-12 09:57:20.000000000","reason":"Change was abandoned"},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2025-05-12 09:56:04.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}}},"hashtags":[],"change_id":"I662ba19697e03917ca999920ea7be93a0b2a8296","subject":"Fix performance glitch while sorting image locations","status":"ABANDONED","created":"2025-01-31 14:46:41.000000000","updated":"2025-05-12 09:57:20.000000000","total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":934716,"cherry_pick_of_patch_set":1,"meta_rev_id":"d877887ff8362ec874ce5c9098763a2943b40887","_number":940528,"virtual_id_number":940528,"owner":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-01-31 23:08:12.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":-2,"date":"2025-04-25 13:30:16.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}],"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-31 14:46:41.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"CC"},{"updated":"2025-01-31 14:57:01.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2025-01-31 14:57:01.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2025-01-31 14:57:01.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2025-01-31 15:06:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-01-31 16:50:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"53861c2188f00e40225d81eda234bde52ce802b6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-01-31 14:46:41.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"8700cd340ccea7e0f9b792809234591f059335f9","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-01-31 14:57:01.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a703a3b0e22b4ae0d35a1882e84e11472175fdeb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-31 15:06:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/55cd39d798d14d849a1ef9e72e11cb73\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ffa338aa883e4517aa56e63026b20d87 : FAILURE in 14m 37s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/822bfd7c944240b78ebd2d4263c255cc : FAILURE in 10m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"05a13fe6f06f601b4d2c261ede392f81bcd0c775","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-31 16:50:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/296dc4bf07a74c14bad5889ff35c120f\n\n- grenade https://zuul.opendev.org/t/openstack/build/5b11bd7e18d341f48cdd147024376ad3 : SUCCESS in 1h 15m 12s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2028a56d1ee746628c592c08d54ba47b : SUCCESS in 1h 25m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9f2b2e4aa0cc45eabebc89ecef98f273 : SUCCESS in 59m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c0d53c9162d443d82762321254dfa6a : SUCCESS in 4m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/71f76b997033479dbd876cf9096011f6 : FAILURE in 6m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3a5d4530acbe4353a788ff6955b35bc9 : FAILURE in 5m 28s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8bcb19077c384a97bc5f786d63745dfe : FAILURE in 6m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/323c53a4ad364ed6afad8a438d80b28d : SUCCESS in 5m 40s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/fcb97e142bfd48e5b66f41de71b7551a : FAILURE in 8m 23s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f6f1367e79cb47e597c459688973d7d8 : FAILURE in 8m 47s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/9e800d27cdc54e208a0f22b68458cf50 : FAILURE in 11m 52s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/004dc8d11a6c49ae9b58bc72336d1db9 : SUCCESS in 1h 37m 53s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/e0c87246a3764240a1cfee33cb175ed5 : SUCCESS in 2h 00m 18s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/0aa560de181a45d7bfa10264b319ae5c : SUCCESS in 1h 53m 42s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/fa8ae8c7d6724448b33af79b19956c3d : FAILURE in 1h 49m 24s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/b61d5c5154f84f97ba4d666011f3d03f : FAILURE in 21m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c7e99884fc26477187f66a8f0efc922e : SUCCESS in 52m 14s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/4c8d6953f8bd4b64bb94bf446db0bccf : FAILURE in 25m 01s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/e7b4a58069f7403b970885fc53e9bb96 : SUCCESS in 1h 57m 30s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/7913ff684ced4d46930aaf4e7d3a9833 : SUCCESS in 1h 06m 06s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/e7593c9c7f81462db9bc9b794fe511af : SUCCESS in 33m 17s","accounts_in_message":[],"_revision_number":1},{"id":"3aaf568476819964aa7cfd2e48a2da5ce71912a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-01-31 20:45:54.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"8392ffcac1884ed618709ad203e85ba88bab269a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-31 21:02:37.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/885342cf34804b70aa2958626755b4b7\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/03050bca726f4087b08f51788b542eeb : FAILURE in 14m 12s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ec032cdd0d82464aaadd64a99696e3b8 : FAILURE in 10m 38s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"dfea2b63b916a93e8fbd4ffc8053a35586a3f027","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-31 23:08:12.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2c2b50818bf04aa6a77efd2a63171249\n\n- grenade https://zuul.opendev.org/t/openstack/build/51965866a1714698bda91da7046fdbbe : SUCCESS in 1h 01m 24s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d1e6dd93e0ce40929d045c05b47f9e94 : SUCCESS in 1h 25m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/907caf487a0e48e09903645bbec94b48 : SUCCESS in 1h 14m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f0db7ab88444c51beb07887a58596fc : SUCCESS in 4m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2772557bd4914c1e96182b91c4653100 : FAILURE in 5m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7b5bfe1c35094c0abe16661ce05184da : FAILURE in 5m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f9d35e747ea1420c8e980d2159a499b7 : FAILURE in 5m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/914e738f94a749f7be265799017e8251 : SUCCESS in 5m 31s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/705db65668944208ab1ca4b759bc4650 : FAILURE in 8m 37s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f2514244fa554b2bbeb08798d0b8f553 : FAILURE in 8m 12s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/59ee045868774afbb96e7581d12a2a71 : FAILURE in 13m 59s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/9c8d97436b79468e9bc4aa2277ae08d8 : SUCCESS in 1h 24m 08s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/e892118c69c944c2a236b153f88926b6 : SUCCESS in 1h 22m 56s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/20a467f7ca9e4c009b7bdd2173e65f58 : SUCCESS in 2h 15m 40s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/f1d3b79fc0da469e81fb481b0c4c9bf3 : SUCCESS in 1h 18m 00s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/34c0ec5cc471474f86d2b88f9d9f9d71 : FAILURE in 15m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ae5b53f8f0af4c908fdc314cf5bd3d0c : SUCCESS in 48m 58s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/5ea7b719a65f4feca83b0b5c34a8e894 : FAILURE in 32m 10s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/3b115c3a3ea44ec4ab30ade773fbe4e9 : SUCCESS in 1h 12m 56s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/c65c1ed4618b4bb6b89b54bc96211c87 : SUCCESS in 1h 12m 09s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/5a9ae8fd90f1403abd5676eefc3544b1 : SUCCESS in 34m 14s","accounts_in_message":[],"_revision_number":2},{"id":"c965a0365cd7ec1e3051ab42ccc930c4001bce24","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-03-10 22:50:51.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"0f5313d8d4cf53f0a01d064c92e076d21bf3c88d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-10 23:14:09.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/46fb1a20ff884e2a93c378c11f206558\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e65db47c3d9949f7b085b5a320c7053c : FAILURE in 20m 44s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/75a381e3ba3d4f5383415f8ce4e63ccc : FAILURE in 13m 20s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e77cb54d4ba78a696d2c78dd5007bb349bf0560b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-11 01:09:37.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4d3c8532850247fb871782d4fae6db49\n\n- grenade https://zuul.opendev.org/t/openstack/build/6571b33f78af45c0a0d672279454995c : SUCCESS in 58m 39s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/22aa16900c804e008f7b8d9f0f3fd97f : SUCCESS in 1h 15m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9a6e40cda1624eb2adfd70862491925f : SUCCESS in 58m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e96c9e10b7b64512b820c6a184d4aa1c : SUCCESS in 4m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4598bef7a5634c11b7a54c90c2550406 : FAILURE in 2m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fd0dcf2e4e5e4840a8df6d266cf65b5f : FAILURE in 5m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5785b0186a7442b3a4374fe1e96ed045 : FAILURE in 7m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a86373614124e90aca40a0ca3568a74 : SUCCESS in 2m 50s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4c935164612d485cbf8152e81813d723 : FAILURE in 9m 04s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7fe46cccd10541b784bde991ee7d8a84 : FAILURE in 9m 19s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/ed67cd273f504bfda3bfc16db3405191 : FAILURE in 8m 01s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/d6a62a0f35aa41a891c8e33178d8df5e : SUCCESS in 1h 22m 13s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/2b8a4554fb594716b2989220c4b77da5 : SUCCESS in 1h 45m 56s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/63dc14446ba0418aaf27d72a5f1a600e : SUCCESS in 1h 58m 16s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/b5590552d5cd43a481dd70b1ff42b0f1 : SUCCESS in 2h 12m 13s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/7d80f03a14a94d8aaf29a94c240dc3ee : FAILURE in 27m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6191f0d0fd2a4973b73e28c00c8ac43d : SUCCESS in 1h 05m 06s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/34ef774d307d4940add4572aa09e5f71 : FAILURE in 15m 47s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/df730eb815224c7bbab0b80b4394ed2b : SUCCESS in 2h 01m 43s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/36a7971ff49f414cb7e63aff29c777a0 : SUCCESS in 34m 09s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/7221941e7d204f6da7f7420a4a1f03bb : SUCCESS in 33m 35s","accounts_in_message":[],"_revision_number":2},{"id":"7e8df5a84692d9d85e7ae5587698a51dbee761b9","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-04-02 14:05:23.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5b089d0e393688b4551b014e789d35bdb6a2656e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-02 15:02:30.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3f62b9234cc540d5b821f23bcef665a1\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a0403faf4fee4d98ab21368f89086dfa : SUCCESS in 8m 37s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/80482f0e172d42acbb1a020145f6d84c : SUCCESS in 6m 56s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"dcdcf1ab180d984f82142c07ddcd3416b5511b63","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-02 17:10:08.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cc20403b1aca4cb9af6583747bd720a5\n\n- grenade https://zuul.opendev.org/t/openstack/build/570a11894f814602a407e7090d654aa0 : SUCCESS in 57m 11s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7246bb1da31e4d6abddafe53bb18b496 : SUCCESS in 1h 14m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/79f17ef196b34c63aca44331487f8578 : SUCCESS in 29m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3480a11967ec4557916e6a5a8f900c92 : SUCCESS in 4m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/47b03bbd860940a08fad68b72ba9b65d : SUCCESS in 6m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d2a1d479fff447389fd71f6944471485 : SUCCESS in 6m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9ee907a6c1d14f7a828c79d7103be70a : SUCCESS in 7m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/428b854266504176a5eca966e5c8e6af : SUCCESS in 5m 22s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/bc5dd6bc0f064bf69bfd70225ad12558 : SUCCESS in 9m 29s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0d0b9c616dad429dae989e431a41c86c : SUCCESS in 7m 44s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/7017697510114ef78ca6274c7fe1a8ba : SUCCESS in 12m 49s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/0683a06d95534f5f86750ae059622429 : FAILURE in 1h 48m 41s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/d2e0e26370e146b7a55ad57b7e4b763b : SUCCESS in 1h 55m 01s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/1a1352ec07ba406396f130c83dd59107 : SUCCESS in 1h 51m 43s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/2445a9dcb6d64fc7aed1b948e31840b9 : SUCCESS in 1h 46m 04s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/9e67229d38e848c3969e883560b960d8 : SUCCESS in 1h 44m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8eecd4e3154b41dab674f172e73a78fe : SUCCESS in 46m 22s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c6078dcefee9436fb489c073eeebc4c7 : FAILURE in 1h 27m 06s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/7dfdcc3fb2584dba99bb5634106c45b3 : SUCCESS in 1h 51m 28s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/9afbd8208f10493eb1f586d4a29f6e54 : SUCCESS in 1h 05m 20s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/a6edb13d14f2433fa56ab283e4a7a3ae : SUCCESS in 23m 27s","accounts_in_message":[],"_revision_number":2},{"id":"5c8c80d82ea6ac9c9202eaf705ece8b7da7be70a","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-04-03 10:51:36.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"901233d4a864982d1d0a6c646583afa0fb4ce20d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-03 11:02:16.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bafdb0d8f5524f909356a71784a2c79c\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fb49ead755de446bb1d766894c50a714 : SUCCESS in 8m 46s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b04e74b52c1a40d3933cd3b10332447b : SUCCESS in 8m 11s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6bcd2b366051a5a4b241d024cdba3f64aa39cdfc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-03 12:57:20.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b76f06140a0640adad285655acacd805\n\n- grenade https://zuul.opendev.org/t/openstack/build/f6ea8deef43345d09f0ba77f294a98a5 : SUCCESS in 1h 00m 57s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9869cc573d5244fa84a3d2cb9f8e51ee : SUCCESS in 1h 38m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/737cf55d85bb4847aaf39023bebd8ab5 : SUCCESS in 59m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d83f1d73005947f987f371a2f73a5d2e : SUCCESS in 5m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d9de5ec26b614176b8bdfd5f5a75625a : SUCCESS in 6m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/60f5b98e4a4842f6a70c281242ef2c43 : SUCCESS in 6m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3565f346179044c0a0bb7a419b626f72 : SUCCESS in 6m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c54f35f483a432ba5a076952136c8aa : SUCCESS in 5m 08s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/34ae69a263bd40a59de3769414786205 : SUCCESS in 10m 39s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f0e6bce314a142a4b8c6c6d017814484 : SUCCESS in 7m 54s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/12c512fc7a85468eb04dc7311cff7283 : SUCCESS in 9m 45s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/fbbe5c24e2604292a758c2576f7c86f9 : FAILURE in 1h 59m 36s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/3f7bc336fa46423f997f97ced1a45785 : SUCCESS in 1h 35m 11s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/46bb42b42241419caa857da4e971778c : SUCCESS in 1h 39m 42s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/363938b384934ab980ef4a4e05281c33 : SUCCESS in 1h 57m 14s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/220980dcaa7b46ef83f81c7364159e3f : SUCCESS in 1h 36m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a079ba79b010483384a27427050345dc : SUCCESS in 43m 13s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/64c57cecbc48464cb445256317b1131b : FAILURE in 1h 35m 24s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/e59cf2add38c4e3eaab3075857e433ef : SUCCESS in 1h 00m 15s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/31834819292242e69442395a8ce7dd28 : SUCCESS in 26m 10s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/fb971af31ba6453fbbc1c43234c4893d : SUCCESS in 21m 06s","accounts_in_message":[],"_revision_number":2},{"id":"37d68418c72203f217f3a33bba862954991fb420","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-04-23 18:39:50.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"c328f613a9c093569ee40989e998d0cc0dbf96a0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-23 18:52:42.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa0c772567324aee9f07f6f6f602097d\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/32ca2acd88054964b2f6f6ff1616d8a9 : SUCCESS in 11m 21s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3b00efe990e94e2b8e6f2f353a4c66f6 : SUCCESS in 9m 31s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"cabe0f43c19b8dd030e57b8822c51ab2d510dcb4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-23 21:05:22.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/073efb4d324b46758cb14ca59e027984\n\n- grenade https://zuul.opendev.org/t/openstack/build/e1ba9a4f93b14ec1a6136d582842a285 : SUCCESS in 39m 20s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/067e1ed4452c426684445e18c5a73e05 : SUCCESS in 1h 01m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/745a510436fe4ef68cc0a111ff25c9d8 : SUCCESS in 41m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f76dcd89ad84492b86773a38d17dac4f : SUCCESS in 3m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e75bd40ec03e4038a02dd8b743c8a8ca : SUCCESS in 5m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3e6bd30493394cf8823089d8e3f549aa : SUCCESS in 5m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d9bc56770943407cbc3f9c0e74122d9d : SUCCESS in 6m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5da2306fba5443e85bfdd71e6abc0b6 : SUCCESS in 4m 51s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a46c88daf69145bd9818a6c7561e9195 : SUCCESS in 8m 21s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/76809f9067c54cc38e25da94188bcb07 : FAILURE in 5m 58s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/c3c93802933b46529aedd69136677b3e : SUCCESS in 10m 37s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/295a57523e144aa9873a58ad979944ce : SUCCESS in 1h 04m 23s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/43ba786b54b64cb7a779758fb78e6bb7 : FAILURE in 1h 44m 01s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/a571f43efdee4587832dda086a9e77c2 : SUCCESS in 1h 51m 44s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/aa38648e10d94ed6becb0a1ec7af6969 : SUCCESS in 2h 19m 05s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/7cc8f70ac8de4c76ac1557200600970e : SUCCESS in 2h 13m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5fefe9e25b34455d9732e41864b7baf9 : SUCCESS in 47m 34s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/341b26e9c00a476f8bbb6af8cf85b44a : SUCCESS in 1h 38m 13s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/9bca531c8cd24dba8e0ef44a97522adf : SUCCESS in 1h 40m 57s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/a3ecd25d7a2d48f4a87e2f2596b1ddd3 : SUCCESS in 1h 01m 34s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/98d3687eaf3b460c88ccf9cd78863909 : SUCCESS in 28m 28s","accounts_in_message":[],"_revision_number":2},{"id":"88135e881a035a0017bc10050713ed38ee230144","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-04-25 13:30:16.000000000","message":"Patch Set 2: Code-Review-2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"fca331f8c0e6f5cbb41b1462a9ad537794b5bda4","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-04-28 15:44:41.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ad9c939bfb96f73aa66972ae400441054f2d82d3","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-05-12 09:56:04.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"a9768b4a9c0d5fc2c1cf649d7e51086c342bd454","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-05-12 09:57:04.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d877887ff8362ec874ce5c9098763a2943b40887","tag":"autogenerated:gerrit:abandon","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-05-12 09:57:20.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"025a02b498b36e50ea9451b44ed33a9048b5c96e","revisions":{"c4e3deb0f5fa0076bf8e97ca497a78a2f330a40b":{"kind":"REWORK","_number":1,"created":"2025-01-31 14:46:41.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/28/940528/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/28/940528/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/28/940528/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/28/940528/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/28/940528/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/28/940528/1"}}},"commit":{"parents":[{"commit":"0dcb9aff2c8f96e33a3dc1f40b5cb3c4cee758ad","subject":"api-ref: Update with links to 2.17","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0dcb9aff2c8f96e33a3dc1f40b5cb3c4cee758ad"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-11-12 05:49:23.000000000","tz":0},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-01-31 14:46:41.000000000","tz":0},"subject":"Fix performance glitch while sorting image locations","message":"Fix performance glitch while sorting image locations\n\nSome of the available glance stores like file, cinder etc has\ncapability to reuse already initiated driver (DRIVER_REUSABLE \u003d 0b01000000). In\nCaracal we have added a feature to sort image locations based on store\nweight. As RBD driver of glance does not have this reuse capability, during\nimage list API call it initializes the RBD driver for each of the available\nimage which is causing noticable delay in list call.\n\nTo avoid this, using new interface added in glance_store which will directly\nget the weight of the store from memory and return it back to user.\n\nDepends-On: https://review.opendev.org/c/openstack/glance_store/+/934362\nCloses-Bug: #2086675\nChange-Id: I662ba19697e03917ca999920ea7be93a0b2a8296\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c4e3deb0f5fa0076bf8e97ca497a78a2f330a40b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c4e3deb0f5fa0076bf8e97ca497a78a2f330a40b"}]},"branch":"refs/heads/stable/2024.2"},"025a02b498b36e50ea9451b44ed33a9048b5c96e":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-01-31 20:45:54.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/28/940528/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/28/940528/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/28/940528/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/28/940528/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/28/940528/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/28/940528/2"}}},"commit":{"parents":[{"commit":"0dcb9aff2c8f96e33a3dc1f40b5cb3c4cee758ad","subject":"api-ref: Update with links to 2.17","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0dcb9aff2c8f96e33a3dc1f40b5cb3c4cee758ad"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-11-12 05:49:23.000000000","tz":0},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-01-31 20:45:54.000000000","tz":0},"subject":"Fix performance glitch while sorting image locations","message":"Fix performance glitch while sorting image locations\n\nSome of the available glance stores like file, cinder etc has\ncapability to reuse already initiated driver (DRIVER_REUSABLE \u003d 0b01000000). In\nCaracal we have added a feature to sort image locations based on store\nweight. As RBD driver of glance does not have this reuse capability, during\nimage list API call it initializes the RBD driver for each of the available\nimage which is causing noticable delay in list call.\n\nTo avoid this, using new interface added in glance_store which will directly\nget the weight of the store from memory and return it back to user.\n\nDepends-On: https://review.opendev.org/c/openstack/glance_store/+/940529\nCloses-Bug: #2086675\nChange-Id: I662ba19697e03917ca999920ea7be93a0b2a8296\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/025a02b498b36e50ea9451b44ed33a9048b5c96e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/025a02b498b36e50ea9451b44ed33a9048b5c96e"}]},"branch":"refs/heads/stable/2024.2","description":"Edit commit message"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Code-Review\u003dMIN"],"failing_atoms":["label:Code-Review\u003dMAX"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
