)]}'
{"id":"openstack%2Fglance~940530","triplet_id":"openstack%2Fglance~stable%2F2024.1~I662ba19697e03917ca999920ea7be93a0b2a8296","project":"openstack/glance","branch":"stable/2024.1","topic":"cherrypick-bug/2086675-qbf76jmjphi","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2025-05-12 09:57:24.000000000","reason":"Change was abandoned"},"4393":{"account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"last_update":"2025-05-12 09:57:24.000000000","reason":"Change was abandoned"},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2025-03-10 22:50:41.000000000","reason":"removed on reply"},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2025-05-12 09:57:24.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I662ba19697e03917ca999920ea7be93a0b2a8296","subject":"Fix performance glitch while sorting image locations","status":"ABANDONED","created":"2025-01-31 14:48:18.000000000","updated":"2025-05-12 09:57:24.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":934716,"cherry_pick_of_patch_set":1,"meta_rev_id":"034c975c9e753b55a646d24eb0fe592bed4c155d","_number":940530,"virtual_id_number":940530,"owner":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-01-31 22:50:37.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-31 14:48:18.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"CC"},{"updated":"2025-01-31 14:56:14.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"CC"},{"updated":"2025-01-31 14:56:14.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"CC"},{"updated":"2025-01-31 14:56:14.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"CC"},{"updated":"2025-01-31 14:57:57.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2025-01-31 14:57:57.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2025-01-31 14:57:57.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2025-01-31 15:02:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-01-31 16:50:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c79a60e9094417f5aee82800ef55a58c1846b694","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-01-31 14:48:18.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"3424e087f95d135c4291cd0b6149df09c28fe7b8","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-01-31 14:56:14.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"73cf1d3defb3eab07c5cf65bb9de095397b0b2eb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-31 15:02:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/efa26813d55a49e0b53ea20e8054b295\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/baf931c4c0f34adf884a750fcfb4d265 : FAILURE in 11m 12s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c6796cebb3c242efbb7a1c4aa046256e : FAILURE in 10m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"346c4150b35b782d61865852988db2e247967705","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-31 16:50:03.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ae21e897d1ae43b1b31562105d04e64c\n\n- grenade https://zuul.opendev.org/t/openstack/build/39d522ff89c644cfbd925e4cb5f951a0 : SUCCESS in 53m 47s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/99ab0a53b994403ca26ac33f596306f3 : SUCCESS in 51m 26s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/93983c7cc8e946c8a166390ffc820341 : SUCCESS in 1h 30m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0fc67020afd44d6bb5ebb21c8d298858 : SUCCESS in 56m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/841ffc27f6f6491a85ac9c9ec6f94d61 : SUCCESS in 3m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4bebc05b8cc74682a1d6b249e6b98d73 : FAILURE in 5m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d2e62d41b398447381195a226b7838f6 : FAILURE in 5m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2fef9eb3c2684a2c9195dd5dd0a1d7cc : SUCCESS in 5m 27s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0e44a74524434025984fd05596e9afd6 : FAILURE in 9m 24s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8b4e1be64e834c3e87904eecc41eee40 : FAILURE in 9m 30s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/34f2835168f64d438499d74d7dad6644 : FAILURE in 7m 34s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/eb889a4aa1f44883a3e725b73c674f24 : SUCCESS in 1h 33m 26s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/3f67f3d71315484782d4ce183ee3c103 : SUCCESS in 1h 43m 30s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/21b0d08b9062490f86d81620a4072df1 : SUCCESS in 1h 55m 45s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/1629cc00c54a4b11860363025c047e36 : SUCCESS in 1h 51m 34s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/0eb2422a9b7b4757a53819ccc09008a0 : FAILURE in 22m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1f2e670114364659b2b10afeeeeb04fb : SUCCESS in 53m 38s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/5fc3269cd7784136b39e0e85bc57b052 : FAILURE in 29m 58s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/33daa0a27be24c2aaaf1738d604c7ebc : SUCCESS in 1h 24m 12s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/5ed4df4d6c664fd39a374f70f0f07a4a : SUCCESS in 1h 07m 25s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/e4ff5b292350431fa0627977dce0fbf4 : SUCCESS in 23m 26s","accounts_in_message":[],"_revision_number":1},{"id":"310a1d41d17e14616d0612440506d7ca4b3605da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-01-31 20:46:21.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"cf91796c96b923d95952cd1eeea48f9f70eb8357","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-31 21:01:11.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d443025eb7c4d8fb77b1247b92ef617\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0cb29589044640d69c7da7d184e24358 : FAILURE in 12m 42s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9f6fab8f7ee84bed9b5c99bcde2584f2 : FAILURE in 11m 48s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"907a0ab7273eab57d6e2ca0a5ba6ee260452af48","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-31 22:50:37.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/42d00e975b7f453c9689a7b4cc679ee2\n\n- grenade https://zuul.opendev.org/t/openstack/build/ecc80db9ce624a09afcfe17f278bca80 : SUCCESS in 1h 16m 26s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/be69b5e48aaa4dfc85d14365f46b5ad7 : SUCCESS in 1h 01m 41s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/23df9f4e188d41d19eee3a6f8d078b03 : SUCCESS in 1h 43m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/76f9bdbed105449abd1f4aa4db0a5159 : SUCCESS in 1h 07m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2b41d22e6c345de8239df5b95d06e86 : SUCCESS in 3m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e0db52d73b04484d8ec98740e6c604eb : FAILURE in 5m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a8e222c787f842409da69da0dfb261c2 : FAILURE in 6m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fca9ecb3d7a549a3996790d33e6db471 : SUCCESS in 5m 21s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b789fa82c0144524a6fcd658982b91fd : FAILURE in 9m 05s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/69706892505e4270919da6f4032c9856 : FAILURE in 8m 58s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/d31d180acb1b457c84bfe8480c6921d3 : FAILURE in 8m 41s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/127117ffbaf3477593d7c1d6f81c85d9 : SUCCESS in 1h 43m 29s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/3b54c59e9f6d47efb90aeb37ceaa049c : SUCCESS in 1h 24m 16s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/0b8c988482234e49b0ee80d87895e47c : SUCCESS in 1h 16m 34s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/840b57e63aad430d9f222f27ea1ac70b : SUCCESS in 1h 57m 47s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/63d1218967a9488d9e9b874b47129043 : FAILURE in 16m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6617ae3edc9d45cf9c6fed5590edd61c : SUCCESS in 57m 57s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/8c68dd14c1cb466db0712a7a2a7407df : FAILURE in 18m 19s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/0efb0461ae1c4e2b93779b716d14a1c3 : SUCCESS in 1h 36m 59s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/4b5493d69f7044ccaeef2d392998f728 : SUCCESS in 42m 12s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/91d13e88c5834cd8a936bb1b5fab3009 : SUCCESS in 39m 17s","accounts_in_message":[],"_revision_number":2},{"id":"47efebabc7f53060731723c6c45f84d3932f37cc","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-03-10 22:50:41.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"8778a9c3f810d84035ae63b9225ec3c09ed5d53b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-10 23:09:38.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/27a2be045ffd4255957f1f53b6a16b18\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ade6a6dced784f6ab0ae14661767aef8 : FAILURE in 16m 25s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/4cc4c8c64cba460481921d5751517b4d : FAILURE in 12m 30s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d46db48b43dbc211a142fcf63e40a494c742234d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-11 01:11:06.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6310193eb6664f9686975ab130b0d50f\n\n- grenade https://zuul.opendev.org/t/openstack/build/69299cfc5eb345dc8edffad6a1e915a5 : SUCCESS in 34m 59s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/d604e21a3f224fcfa5233d4d1fdfd6b7 : SUCCESS in 32m 16s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/dd2655d46acd468f9e1de4e38cea2334 : SUCCESS in 1h 12m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c9d8e57643cd4da4a1e3ba7c3eaef697 : SUCCESS in 26m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e4ada297d814194b9540ae9b6b6ec7b : SUCCESS in 2m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/248afa499e154c25834215291778ad94 : FAILURE in 2m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/87519623b39949e0811b3128b13f995c : FAILURE in 3m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5ce5dfdaec844ef883bf25695e3ca87 : SUCCESS in 4m 48s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f027faf364d042bd87798e52d237e8d7 : FAILURE in 5m 36s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/baaacb4243f747168dde1d782e5ac295 : FAILURE in 8m 01s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/842f0bd70c6c42218851929f19da76b5 : FAILURE in 8m 22s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/abf18834e8cd4ccd98aa7ce2c278b5ac : SUCCESS in 1h 22m 04s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/c22faf5eed344a61b8fe763124e403bb : SUCCESS in 51m 47s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/18b0c7d178c6421c90c0ce6e22bb637d : SUCCESS in 2h 19m 46s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/63a437f6a8f34f82afab95cdd26e7e58 : SUCCESS in 1h 20m 42s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/74811c398f9a412e87a13ef276de9e80 : FAILURE in 35m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/915c2ea4a1e648e3966bfdd4d2bad5c1 : SUCCESS in 26m 57s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/965360cc38764309894c6d26400eddc6 : FAILURE in 21m 03s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/08c7ccceab9e48b98171465360e559a0 : SUCCESS in 1h 18m 52s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/1913285e4ec1403ba579c06f407ce2f2 : SUCCESS in 33m 15s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/7045c0cc82b94827aeff9805ee957f37 : SUCCESS in 26m 54s","accounts_in_message":[],"_revision_number":2},{"id":"034c975c9e753b55a646d24eb0fe592bed4c155d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-05-12 09:57:24.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ce537e073b2e7efb332d719bb45bf20c403c140a","revisions":{"74685599359fe5630b9c867ae4876cdda2262993":{"kind":"REWORK","_number":1,"created":"2025-01-31 14:48:18.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/30/940530/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/30/940530/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/30/940530/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/30/940530/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/30/940530/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/30/940530/1"}}},"commit":{"parents":[{"commit":"d0be1bededcc482946970312d92832f085d35e17","subject":"Fixed typo in discovery.py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d0be1bededcc482946970312d92832f085d35e17"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-11-12 05:49:23.000000000","tz":0},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-01-31 14:48:18.000000000","tz":0},"subject":"Fix performance glitch while sorting image locations","message":"Fix performance glitch while sorting image locations\n\nSome of the available glance stores like file, cinder etc has\ncapability to reuse already initiated driver (DRIVER_REUSABLE \u003d 0b01000000). In\nCaracal we have added a feature to sort image locations based on store\nweight. As RBD driver of glance does not have this reuse capability, during\nimage list API call it initializes the RBD driver for each of the available\nimage which is causing noticable delay in list call.\n\nTo avoid this, using new interface added in glance_store which will directly\nget the weight of the store from memory and return it back to user.\n\nDepends-On: https://review.opendev.org/c/openstack/glance_store/+/934362\nCloses-Bug: #2086675\nChange-Id: I662ba19697e03917ca999920ea7be93a0b2a8296\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/74685599359fe5630b9c867ae4876cdda2262993"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/74685599359fe5630b9c867ae4876cdda2262993"}]},"branch":"refs/heads/stable/2024.1"},"ce537e073b2e7efb332d719bb45bf20c403c140a":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-01-31 20:46:21.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/30/940530/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/30/940530/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/30/940530/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/30/940530/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/30/940530/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/30/940530/2"}}},"commit":{"parents":[{"commit":"d0be1bededcc482946970312d92832f085d35e17","subject":"Fixed typo in discovery.py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d0be1bededcc482946970312d92832f085d35e17"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-11-12 05:49:23.000000000","tz":0},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-01-31 20:46:21.000000000","tz":0},"subject":"Fix performance glitch while sorting image locations","message":"Fix performance glitch while sorting image locations\n\nSome of the available glance stores like file, cinder etc has\ncapability to reuse already initiated driver (DRIVER_REUSABLE \u003d 0b01000000). In\nCaracal we have added a feature to sort image locations based on store\nweight. As RBD driver of glance does not have this reuse capability, during\nimage list API call it initializes the RBD driver for each of the available\nimage which is causing noticable delay in list call.\n\nTo avoid this, using new interface added in glance_store which will directly\nget the weight of the store from memory and return it back to user.\n\nDepends-On: https://review.opendev.org/c/openstack/glance_store/+/940531\nCloses-Bug: #2086675\nChange-Id: I662ba19697e03917ca999920ea7be93a0b2a8296\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ce537e073b2e7efb332d719bb45bf20c403c140a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ce537e073b2e7efb332d719bb45bf20c403c140a"}]},"branch":"refs/heads/stable/2024.1","description":"Edit commit message"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
