)]}'
{"id":"openstack%2Fglance~941076","triplet_id":"openstack%2Fglance~stable%2F2024.1~I24073c1b1e5ea92357d9a774e6c9c9cbf0980a44","project":"openstack/glance","branch":"stable/2024.1","topic":"bug/2095304-stable/2024.2-stable/2024.1","attention_set":{},"removed_from_attention_set":{"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2025-03-13 05:50:30.000000000","reason":"\u003cGERRIT_ACCOUNT_9303\u003e replied on the change","reason_account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}}},"hashtags":[],"change_id":"I24073c1b1e5ea92357d9a774e6c9c9cbf0980a44","subject":"Remove S3 credentials from debug log","status":"MERGED","created":"2025-02-10 06:23:48.000000000","updated":"2025-03-13 10:17:32.000000000","submitted":"2025-03-13 10:15:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"941076-bug/2095304-stable/2024.2-stable/2024.1","cherry_pick_of_change":940697,"cherry_pick_of_patch_set":1,"meta_rev_id":"2199ff51debf5f31c3ad7954a102a7d18b258bfa","_number":941076,"virtual_id_number":941076,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-13 10:15:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"recommended":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"all":[{"value":1,"date":"2025-02-27 05:52:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":2,"date":"2025-02-28 17:31:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2025-02-26 18:19:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":1,"date":"2025-02-28 17:31:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-10 06:39:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-02-10 08:16:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-26 18:19:46.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2025-02-27 05:52:11.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"state":"REVIEWER"},{"updated":"2025-02-28 17:31:46.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"}],"messages":[{"id":"5cd168093e36b486f887bd91bc5da4241e359909","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-02-10 06:23:48.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2024.2.","accounts_in_message":[],"_revision_number":1},{"id":"6fc54fbcd7c394acebf5b44b05b32a64b78229b5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-10 06:39:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/caea54e6b3da40c6937e18528b8682de\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7064a506cac5463da2b38deae7de91c0 : SUCCESS in 13m 18s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/1de16cf32ae34152a489b1ba7ff9a537 : SUCCESS in 11m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4e9f487e7ee056dba5e2dc76c645961dc1b9a793","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-10 08:16:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d10b8bf943243ceac3e628c047a2d88\n\n- grenade https://zuul.opendev.org/t/openstack/build/9d31f3cd0173488cbeb838762720309b : SUCCESS in 32m 54s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/ba23985062e54a2782e4b9ba221bab4e : SUCCESS in 31m 14s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/096045b8dca84553a11015a8fc6cb3d7 : SUCCESS in 1h 18m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f29acf0a03b245d58361dc38f675ec86 : SUCCESS in 1h 03m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/415a48a76039411fa6222ec0f3b8aeb6 : SUCCESS in 2m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4c6c8fb73ce24efc8981284a184cd2b1 : SUCCESS in 4m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d4fbe89923934f43b4e9c14e2c67b006 : SUCCESS in 3m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/357882eb99b84646ac4d0a6f84cc4b41 : SUCCESS in 5m 38s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/155d1b5b93784f37aef3377698faae62 : SUCCESS in 8m 34s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6ab7e975424d4e678560eba30acded24 : SUCCESS in 6m 19s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/1da6213ed8c942fa9775f8b30147ffdd : SUCCESS in 7m 52s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/4cb15f7ccbac4e7ea00d26239597dbc1 : SUCCESS in 56m 40s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/32bc5e0d28fa43adbdcfcf03d178ab91 : SUCCESS in 1h 44m 40s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/5ac2263a5d8f49418a475b10759951d3 : SUCCESS in 1h 34m 55s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/eea7c1cac0c645aeb36d111363d77950 : SUCCESS in 1h 46m 03s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/1584dbb5fe7344878fb3c66106c52f10 : SUCCESS in 1h 39m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ad0f8019a80f44b0959b2f1d52c2f3fe : SUCCESS in 46m 02s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/7ae4d443739a40d3b97c330ac5f8556a : SUCCESS in 1h 31m 49s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/bb78174da94f4524a242adb6ce8d22aa : SUCCESS in 1h 32m 28s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/b4b88ca3942d4ff598a9802294f8d27c : SUCCESS in 1h 01m 38s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/e9f65ec10145449db9614042185dd39f : SUCCESS in 20m 41s","accounts_in_message":[],"_revision_number":1},{"id":"7fa7bfad452688af6ff477b944aa8822f523bc5e","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-02-26 18:19:46.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e2fa5c6d6af84e7240329ebc8deeb505064a8f3a","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2025-02-27 05:52:11.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"0677f965933ee69da70ffa4187ca3808122d297c","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-02-28 17:31:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"431ba4a037ac7c9e823775174f8e94ba88c18ebb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-28 17:32:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"27b0395c7912a01afe9bef0f29048ef8c18ec261","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-28 23:14:12.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c83d7c0d8e514fe4840a2f11e6c28d43\n\n- grenade https://zuul.opendev.org/t/openstack/build/6572125da7eb4244861639eae9dc172e : SUCCESS in 56m 50s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/a44f03264b3e4e3dad8ff3790baae0bf : SUCCESS in 1h 03m 02s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b4c3342ec1b04588a6ada92e4af2a328 : SUCCESS in 1h 19m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/69969694d74a435aa70c9e116857b2cf : POST_FAILURE in 9m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5904e8eeb30145f4a28b9ee297e84229 : SUCCESS in 4m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b70fdf8bd99d43e699337f8f892097a5 : SUCCESS in 5m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/dbafe11964354920b6509f604a8a03a0 : SUCCESS in 4m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f248700dcf1240109dfff0d319733657 : SUCCESS in 6m 18s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d0707f6bfa114e3ba243a350a6351272 : SUCCESS in 7m 48s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0d69c1ef67544bc7bff70b7d189d503b : SUCCESS in 6m 29s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/7c6bfb0491b94f9baa6c7deb3db63601 : SUCCESS in 2h 15m 13s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/7ad0e0546d8348088ee8b2acc559b724 : SUCCESS in 1h 21m 19s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/7c94c16e503b4cad94a3d42f962513d3 : SUCCESS in 1h 36m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/98cc3ab5208a41bc89172f09b7e127ef : SUCCESS in 56m 20s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/6d7853d984044f469cf8a7d3023928fe : SUCCESS in 1h 27m 59s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/8736b6ff10c143f29933fec409973af7 : SUCCESS in 26m 14s","accounts_in_message":[],"_revision_number":1},{"id":"a46d486edc43623d5688050f4489f54e323b3d0f","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-03-10 14:10:58.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c0476a0161ebc4d479bfab83eb5da7460f960238","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-10 14:24:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/74b20958eb084d01af6f80b03c0bd5c0\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f5b399c92f424677810a75629aaa8d50 : SUCCESS in 11m 00s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/11b99bb08d8442938cc1ca8ba898b69c : SUCCESS in 10m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f446c37d6634e41c32c0c42cb4df10ce57fb2631","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-10 17:15:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7e21688aeaf0497095172070b86735d5\n\n- grenade https://zuul.opendev.org/t/openstack/build/3ce7a858c6394f2aa06f867c8cef9f6f : SUCCESS in 2h 14m 55s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/223943538de249bdbc62d323a965ea97 : SUCCESS in 43m 15s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/83f31f001afa40a2a79df653c6cfaab1 : FAILURE in 1h 21m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0ab4a26ab3214536a68474e1ee2bf57b : SUCCESS in 56m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e2c474cb26e472e831b32909e70d6c0 : SUCCESS in 3m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ebbfd15c28604c82baac88a32151b39b : SUCCESS in 4m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d10e6302f2f94ee88c3b5104c0a8c0e0 : SUCCESS in 2m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5198f6d17ad04021af1231b3a2016413 : SUCCESS in 6m 00s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5bf6e13054fa407ead2910e0e7618b03 : SUCCESS in 6m 25s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d97b6c846b504700bf14b1a5dc517f6b : SUCCESS in 7m 28s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/d8d9d6ba4e2546b088951bfc47a8de0e : SUCCESS in 8m 25s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/77889fcb4c144e87a37d31b615ad443f : SUCCESS in 1h 22m 53s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/b0a57b2ac5de4f8f870a770e06917a5f : SUCCESS in 54m 50s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/8355c607d6bb48fe89eaf7c31e3fb585 : SUCCESS in 1h 38m 13s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/89d5b4f9350b4c7c9935c1914b5112ca : SUCCESS in 2h 59m 06s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/af05494b0d324081b207e22623b12800 : SUCCESS in 1h 06m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/89b1bd89029147568db1e0a4f47ff797 : SUCCESS in 40m 37s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/23ada73a9f2e4fd68146050ebd4db796 : SUCCESS in 1h 37m 38s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/765c252777694ca2a64e8b4adcee317c : SUCCESS in 50m 22s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/e61f1ae3de144d4ea64f98ab6f5df130 : SUCCESS in 59m 59s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/0d931b2f6d8b4464ba97dbe23c5a30fd : SUCCESS in 25m 06s","accounts_in_message":[],"_revision_number":1},{"id":"07156bd5b73fe7d50f30e81e5c5e881bb7d69edf","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-03-11 16:18:25.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"447aca0d95cd251b18a38ae7be9236e5d7a8b02e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-11 16:31:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8516789b14974dca93c09db046adcfad\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/31d40bdf1308445d94762836a5396397 : SUCCESS in 11m 00s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a8c516614f9d4ce183d318aa89e0b2bc : SUCCESS in 11m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8223eed934e2d4f6cb2f37c8d1c5eda72e7f6820","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-11 18:36:46.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/38502905e9094825837a7282d3cc310f\n\n- grenade https://zuul.opendev.org/t/openstack/build/13b022c7e08a4ac8a47c794c2bf0a1c3 : SUCCESS in 1h 03m 20s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/dbd94c3636c442e591309968e1691c5a : SUCCESS in 52m 17s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/802543a0a7b54e1bbb32a2765a0078a5 : SUCCESS in 1h 14m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c9e510e0c2e04a5b99063a06411f9033 : SUCCESS in 48m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d860d6d021d4214a2ef09295addecbd : SUCCESS in 3m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dd188b1ead8b44cd9d029413ecee328e : SUCCESS in 4m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0d2cbf74a35c43a9b440a181ca15f70e : SUCCESS in 3m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49a441dd4629478a85d0eb6046ab8c76 : SUCCESS in 6m 01s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ea36dc9a7cda40e0a5ca5f88c44b1416 : FAILURE in 6m 20s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1cbb84762c404abda24f1900d0853b75 : SUCCESS in 9m 41s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/3bb85ef83c9b4ef3a0b4a8f050a06356 : FAILURE in 5m 36s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/7467a7c9464043f7b6438e719b6a526e : SUCCESS in 1h 59m 50s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/2e8e44eb8dfb4731a36d942cf029a18c : SUCCESS in 1h 00m 41s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/d6e9bbea5e1146d2a6b917379ba8820a : SUCCESS in 2h 09m 24s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/50b0db30700f4e74be83103e7cfb05b7 : SUCCESS in 1h 57m 54s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/7bc82d01f7444548a3dcf9b793403c23 : SUCCESS in 1h 20m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7d96bd94029c4b47a25b4067bf633326 : SUCCESS in 51m 56s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9b958d87578d4fd5a5d3e3ed1a6e5365 : SUCCESS in 1h 22m 10s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/0c224a68daae4bfaac9b1bed20b52146 : SUCCESS in 1h 45m 31s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/3df6e39a90554912906e74c8b887d532 : SUCCESS in 35m 01s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/7dbe305871364c39b13f9ea2bc61b1e2 : SUCCESS in 25m 27s","accounts_in_message":[],"_revision_number":1},{"id":"018b746201c3b40cdab71594f9607cef938e0d61","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-03-13 05:50:30.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8bc598fb12caff0601f53d49c16f85b2628c0e84","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-13 06:02:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/93566f1a535a4099bbc8331fefb0cfda\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/05c2e2fe07d44fab9e5f25612e643ac8 : SUCCESS in 9m 59s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6a12412f1a944756a44c05757c734a5b : SUCCESS in 9m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0cb6cfc4cb981b1ee724b73249aa937a736135d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-13 08:01:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3a6a01a815d545fa920f4ab721f9d7d7\n\n- grenade https://zuul.opendev.org/t/openstack/build/09a859f871d146ee8af80cb938617276 : SUCCESS in 1h 02m 08s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/a576a877202e457291624b8ff2afe617 : SUCCESS in 58m 06s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/99b3d4d0d2254c23a43057e4fb1e3874 : SUCCESS in 1h 37m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/23ee3d0e78c145e5a72784b31eee7979 : SUCCESS in 1h 13m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9274f251cd1942079f5d93ffc7dbfd80 : SUCCESS in 4m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bdbe835b41454992845e56b271ab480a : SUCCESS in 5m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e61e5903c9954dfab29552bb315b3e67 : SUCCESS in 5m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62e9b68ac1d245aca1460991d9509ed4 : SUCCESS in 5m 44s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/00c105c205864ee981e6fd6f9cbf5f46 : SUCCESS in 8m 37s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1fce1d7c30cb4e8dab9923b986682fec : SUCCESS in 8m 50s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/cab1035f200e4845adf74d2e5f076fd2 : SUCCESS in 8m 44s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/bf8c9d225bda4e2d984927b46e000a3d : SUCCESS in 1h 40m 04s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/b378851c6d604f6ea5170ee44e8bfb5d : SUCCESS in 48m 15s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/e9d89977bcc54e99b5c70517acb0d603 : SUCCESS in 2h 00m 09s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/a5ea0a55befb4c60afa48ed7fafd124b : SUCCESS in 47m 39s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/f9b4a3d2ef734b83a6995198cba1abed : SUCCESS in 1h 48m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/535277e355414af0a268cb5ebcedacb0 : SUCCESS in 1h 02m 36s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/3cc62336baaf4f19aa3819604c3bbfd1 : SUCCESS in 1h 31m 40s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/1dc0840828564b95b9b9bd97b7859abc : SUCCESS in 1h 30m 42s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/f9829a95cf7041ccb2c7e8784d851c46 : SUCCESS in 1h 13m 08s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/fd46dd34ff494de39cead5b9860d8634 : SUCCESS in 15m 22s","accounts_in_message":[],"_revision_number":1},{"id":"bff33e943d30921678f76f007383fa95fba2922b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-13 08:02:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"284b15f6d7cd07001888e5a5fe106f617b1701c3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-13 10:15:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3d68b54576094c6e86d3415d6fe153f4\n\n- grenade https://zuul.opendev.org/t/openstack/build/da2b9a45d0d94a299b8ec480ee622a2c : SUCCESS in 59m 27s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/ca0ef8c653ac4d05a982662d596feef1 : SUCCESS in 50m 45s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/23feb8dfb56e45abbfade80a3431449f : SUCCESS in 1h 30m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/484884d25c4f4579850dc68c08e85084 : SUCCESS in 47m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d72451cadd344e4a124807bab80c441 : SUCCESS in 5m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5256463b0d8a4a5b80ad406bb6f5cf22 : SUCCESS in 5m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b14310966bf147d8be7ecb13bd309257 : SUCCESS in 5m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95e5dfcc9e3e4cd894922ecfb784a4ec : SUCCESS in 3m 19s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a4251a49000b4cf199b25cf697257e41 : SUCCESS in 8m 29s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/fabe4fdf068e4b9b830815edb190204f : SUCCESS in 8m 34s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/1259e443ee0a4587a5c6250c44e4b1d0 : SUCCESS in 49m 02s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/6443b37152b649e08e4312ee76ac28d8 : SUCCESS in 2h 02m 33s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/6d6c4c59b41c4db6bcb6a6ff9a92e27c : SUCCESS in 1h 32m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2fb00e8fab394888bcae533eb7adfa05 : SUCCESS in 53m 33s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e83e22bf94ed4b22a509cce77a0489cf : SUCCESS in 1h 27m 13s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/c510c8f3586e4a47a4e389891af0a02b : SUCCESS in 24m 48s","accounts_in_message":[],"_revision_number":1},{"id":"b64beb22fff0bdd31a6fb1fccfb05f5d7c8b06f1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-13 10:15:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"2199ff51debf5f31c3ad7954a102a7d18b258bfa","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-13 10:17:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e4916e561b604d2e8634b4aae0b8934b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bfa6e0f294d045b79ec9ce1dfdb220f3 : SUCCESS in 1m 12s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"265ad47616f1ef8d762448705a0637931e04c33a","revisions":{"265ad47616f1ef8d762448705a0637931e04c33a":{"kind":"REWORK","_number":1,"created":"2025-02-10 06:23:48.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/76/941076/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/76/941076/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/76/941076/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/76/941076/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/76/941076/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/76/941076/1"}}},"commit":{"parents":[{"commit":"d0be1bededcc482946970312d92832f085d35e17","subject":"Fixed typo in discovery.py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d0be1bededcc482946970312d92832f085d35e17"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-01-22 07:03:28.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-02-10 06:23:48.000000000","tz":0},"subject":"Remove S3 credentials from debug log","message":"Remove S3 credentials from debug log\n\nWhile sorting image locations using store weight, glance logs a\ndebug message which logs secret and access key for s3 backend.\n\nRemoving the debug log to avoid leaking of the s3 credentials.\n\nCloses-Bug: #2095304\nSecurityImpact\n\nChange-Id: I24073c1b1e5ea92357d9a774e6c9c9cbf0980a44\n(cherry picked from commit 33fa9596ebbd7ed16e2bbdba5fab2f6eeb8eb5c2)\n(cherry picked from commit d1cc917a29c9d2e87b1bad51a33a8a2500eb69c6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/265ad47616f1ef8d762448705a0637931e04c33a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/265ad47616f1ef8d762448705a0637931e04c33a"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
