)]}'
{"id":"openstack%2Fglance~941243","triplet_id":"openstack%2Fglance~master~I93e024f01f3027647c5677c317839f1a682b3382","project":"openstack/glance","branch":"master","topic":"fix-no-qemu-img","attention_set":{},"removed_from_attention_set":{"36765":{"account":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"last_update":"2025-02-12 08:19:48.000000000","reason":"Change was marked work in progress"}},"hashtags":[],"change_id":"I93e024f01f3027647c5677c317839f1a682b3382","subject":"Fix qemu-img package not getting installed for testing","status":"NEW","created":"2025-02-11 13:37:40.000000000","updated":"2025-02-12 09:52:58.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"20e9c33e26ffe58afcff6773e1f4b7c63b2ea84d","_number":941243,"virtual_id_number":941243,"owner":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-02-11 16:08:47.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{"CC":[{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"reviewer_updates":[{"updated":"2025-02-11 15:24:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-02-11 16:08:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-12 09:52:58.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"}],"messages":[{"id":"5411078f5808246a130710fe53a41f9fade6adea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2025-02-11 13:37:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"df15a59f9e7a231a695fcc6ce7fda94c1c0b2f4e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 15:24:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/329876def18b4bb7a9cbfd53d59bb7b4\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/87af4307365e4e9abbe642629568e1f3 : SUCCESS in 11m 49s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/bcf50366507e44a0b7448f3755844b7b : RETRY_LIMIT in 2m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bcc4ce9cea7564a790c70a0e05ddace75a092ae4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 16:08:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9d6c6d95108c45c597771fcf435bc447\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/d34f27ec05184d16b3c29933ed60a7bb : SUCCESS in 3m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/3bc5901eb1de4441977788b798c6d06b : SUCCESS in 1h 08m 17s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/25af2d79564b4a5bb02c51a86201d9ff : SUCCESS in 1h 03m 29s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/937dbfdda02b49f6a681d35d85c6c5bd : SUCCESS in 1h 55m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e8b0f84a772d428c9cfade83eadb6f56 : SUCCESS in 1h 04m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/458638a16b984784aeda44d0e6d57854 : RETRY_LIMIT in 1m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6c50bead66bf4380bb7a9483c587933d : SUCCESS in 5m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c9f2feaea86947388bb21459bf1978a0 : RETRY_LIMIT in 2m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22b08b7f598945b6a706865994cbfbf6 : SUCCESS in 3m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8c1888ffec3e4bf4926733659f15de62 : SUCCESS in 5m 27s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/162638b7d23c4ea7a6998e50711da82c : SUCCESS in 6m 50s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/72c839abf36c466e97d906067c61e954 : SUCCESS in 9m 49s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/62ddb6fd08a24327b6bede96b078db61 : RETRY_LIMIT in 1m 38s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/ef8c2af5bacd4bbcb0d78d5a34c84b78 : SUCCESS in 2h 08m 30s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/236a0f0a099a47abaf06052f72d75afa : SUCCESS in 1h 55m 33s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/7775abe676d44842bdbaee7d8d5ea8c0 : SUCCESS in 2h 17m 06s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/d38b9adbc7524d64b8f6d8e7e256865e : SUCCESS in 2h 11m 16s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/e2724541dcec47ab81e01839ecc8e51c : SUCCESS in 2h 25m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/94527b214fb247cd92f0857122b50297 : SUCCESS in 49m 21s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/2213ddaf0bcb471f9f7e1d4a6d301676 : SUCCESS in 2h 05m 58s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/aac225f2cc0a4ce98c2ff93ac54ba428 : SUCCESS in 2h 05m 04s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/9ef8997c8094455a887002e17dc369cf : SUCCESS in 1h 11m 41s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/85e43795962d4830bb96c2353710ae25 : SUCCESS in 1h 48m 41s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/d3bdf6f13dc94628907994e689dab8ee : SUCCESS in 36m 31s","accounts_in_message":[],"_revision_number":1},{"id":"2a1240addd2bab5cb3858b120b1a3ab3b2f6d53c","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2025-02-12 08:19:48.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a2f25fde4828f7e1e2773ca458391685d16c32ec","revisions":{"a2f25fde4828f7e1e2773ca458391685d16c32ec":{"kind":"REWORK","_number":1,"created":"2025-02-11 13:37:40.000000000","uploader":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"ref":"refs/changes/43/941243/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/43/941243/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/43/941243/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/43/941243/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/43/941243/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/43/941243/1"}}},"commit":{"parents":[{"commit":"a293432c3bfe61691b642fcc96688608b6c4f5d1","subject":"Merge \"docs: Fix possible values for enabled_backends\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a293432c3bfe61691b642fcc96688608b6c4f5d1"}]}],"author":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2025-02-11 13:34:48.000000000","tz":0},"committer":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2025-02-11 13:37:37.000000000","tz":0},"subject":"Fix qemu-img package not getting installed for testing","message":"Fix qemu-img package not getting installed for testing\n\nCloses-Bug: #2097890\nChange-Id: I93e024f01f3027647c5677c317839f1a682b3382\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a2f25fde4828f7e1e2773ca458391685d16c32ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a2f25fde4828f7e1e2773ca458391685d16c32ec"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
