)]}'
{"id":"openstack%2Fglance~941291","triplet_id":"openstack%2Fglance~stable%2F2024.1~I8a47d701c195d9418040b4163a3d58af8a4468bc","project":"openstack/glance","branch":"stable/2024.1","attention_set":{},"removed_from_attention_set":{"16137":{"account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"last_update":"2025-10-16 19:07:23.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8a47d701c195d9418040b4163a3d58af8a4468bc","subject":"Update glance-store requirement to \u003e\u003d 4.7.0","status":"MERGED","created":"2025-02-11 18:51:27.000000000","updated":"2025-10-16 19:08:15.000000000","submitted":"2025-10-16 19:07:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"941291","meta_rev_id":"edbda202c9707502490e76f67a24285d77c9f2f8","_number":941291,"virtual_id_number":941291,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-16 19:07:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2025-10-16 16:18:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2025-10-16 15:42:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2025-10-16 16:18:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-11 19:10:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-02-11 19:57:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-12 09:52:02.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"},{"updated":"2025-10-16 15:42:54.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2025-10-16 16:18:12.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"b547b0b637251ac09bd71272f7b15fcd5a368bb8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-02-11 18:51:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"52ce583fb239607cdbffab7d27ed85c6137bb895","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 19:10:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7bb88f01f13448b58c9b9450de1b8792\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9f1488887b224821b26ec8181c077e12 : SUCCESS in 17m 06s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b48828145c0540f599043b5c16db6e26 : SUCCESS in 15m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f92f06e668fbc209ee307576cb754745c44eb235","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 19:57:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4915f9ce48ab4194b32db7e37d38c776\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/fd4e7d7bc0f446acaba24d2e68314a80 : SUCCESS in 3m 11s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/c0408eb012e14cc9bc8f265505d0b79e : SUCCESS in 52m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e2d4c44eaeb84ef1b580501eb0ddae2c : SUCCESS in 1h 03m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d3c4c3502934acfa6bd1e78e3b0dde1 : SUCCESS in 4m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9af3ff08d1054c5f8e6fb4f83eb0ab13 : SUCCESS in 6m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/38e1654240eb46b5b63d52f3011b43ad : SUCCESS in 7m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e63c7a95f81749c0a76610c352ba6ec6 : SUCCESS in 6m 26s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c52892c0d57144d194d9f0b413c30e7a : SUCCESS in 8m 54s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/cdb6023d60c94c8c9a2198a2be8aabfb : SUCCESS in 7m 07s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/bba1098de4b74f72a91feac57e3d84a1 : SUCCESS in 7m 41s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/764a6377576f433ba4dd950dc677eba3 : SUCCESS in 29m 20s","accounts_in_message":[],"_revision_number":1},{"id":"2c4946503bd7d9fcf9327a70a1f79952f41c55a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-08-29 08:04:39.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"e727602978f75163772c8b5a2e22b1ed3b430481","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-29 09:11:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5a35cc2d338f4923ae106002ae9f5ba0\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/228da7e3412e4ef5a19a9fe20d24901d : SUCCESS in 2m 56s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/fdd4cbe0a8b14298a846f7f9c317d2f2 : SUCCESS in 51m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5e2c0d3e7a1d4c7fa52ffa6c0908cc7f : SUCCESS in 58m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b0be9af08aa45beb013736c8b494b6d : SUCCESS in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cfb151a8be6041cbae4e741291d3c319 : SUCCESS in 5m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0d0bc5cd46a0489dbe27b3c41a8892b6 : SUCCESS in 3m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b71964941c0a4df48d0a6a8f3854b936 : SUCCESS in 3m 32s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/25f96775ecf946efa859c9386dd52362 : SUCCESS in 8m 56s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/10f8cfb76a2b4e3191ebb6b413deba2d : SUCCESS in 6m 28s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/e66cc1f5012d4055bb975379f4254060 : SUCCESS in 12m 16s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/cee1099158fc44f3abf64c310d6a605e : SUCCESS in 36m 03s","accounts_in_message":[],"_revision_number":2},{"id":"88e54ffad76e5b36050dbeacb9d912c0ac452137","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-29 12:13:52.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3612926e6ac5422380e005f6dc461165\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/49f6e822120d454e88adfa716f23dd4e : SUCCESS in 11m 45s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/19f485875ba046c49edc918279d7e921 : SUCCESS in 13m 03s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"70c7ca6e755950cbd0bd83a4b24b131c7335583f","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-10-16 15:42:54.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"0ab56db455e6ae76c47abf96625042cb5e009995","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-10-16 16:18:12.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"2c1d2e19a2a221d9f16e894d2d725fa07097b7ac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-16 16:18:37.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"fbc5173c9c3c5293a651ed82f57d3a0ad401e5e0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-16 19:07:22.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/415b4e45a39444bdb3afd62dccb05f93\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/f248fd38c0564c43911da5e88a88e3a8 : SUCCESS in 1m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a464b02453cb4df9bf820fe2c7ce43e7 : SUCCESS in 1h 00m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa389962ef60445f8b7d2aeb1eeaa9ee : SUCCESS in 4m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7b2c4f0f671b4391a9845e796d5272d1 : SUCCESS in 6m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fece8c64153b4aab95690bfa0eca5366 : SUCCESS in 6m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2b5a4e040ce415abdc3ddef98414cc4 : SUCCESS in 5m 20s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4bb92ea74c7f4a97af38bed5aafe39da : SUCCESS in 12m 51s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a99753cee05844948a7aa852ecd11ff1 : SUCCESS in 8m 47s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/921711b1bc214a5783a40947ce600e80 : SUCCESS in 36m 45s","accounts_in_message":[],"_revision_number":2},{"id":"c35bfe30d4b28cea27429e5f969b028aec3e827d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-16 19:07:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"edbda202c9707502490e76f67a24285d77c9f2f8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-16 19:08:15.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2a456ce4d1764ac9ad1c8d644e57645f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de78af74ff654d9ca0fa4f0b9f5c89ca : SUCCESS in 41s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"de73d00ae0db0c07d13dd9a027639923bc565c19","revisions":{"bbea3ed6c09c641abf4bc157e860488aae2ecce0":{"kind":"REWORK","_number":1,"created":"2025-02-11 18:51:27.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/91/941291/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/91/941291/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/91/941291/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/91/941291/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/91/941291/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/91/941291/1"}}},"commit":{"parents":[{"commit":"d0be1bededcc482946970312d92832f085d35e17","subject":"Fixed typo in discovery.py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d0be1bededcc482946970312d92832f085d35e17"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2025-02-11 18:50:19.000000000","tz":60},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2025-02-11 18:50:19.000000000","tz":60},"subject":"Update glance-store requirement to \u003e\u003d 4.7.0","message":"Update glance-store requirement to \u003e\u003d 4.7.0\n\nThe glance-api service doesn\u0027t start unless there\nis a newer glance-store version installed, when doing\na SLURP upgrade from Antelope to Caracal glance-api\ndid not start until glance-store was upgraded to a\nnewer version.\n\nChange-Id: I8a47d701c195d9418040b4163a3d58af8a4468bc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bbea3ed6c09c641abf4bc157e860488aae2ecce0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bbea3ed6c09c641abf4bc157e860488aae2ecce0"}]},"branch":"refs/heads/stable/2024.1"},"de73d00ae0db0c07d13dd9a027639923bc565c19":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":2,"created":"2025-08-29 08:04:39.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/91/941291/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/91/941291/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/91/941291/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/91/941291/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/91/941291/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/91/941291/2"}}},"commit":{"parents":[{"commit":"4914bb9db5453f0721e37a631bd87f24135d5e19","subject":"Docs: be more explicit regarding oslo_limit.endpoint_id","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4914bb9db5453f0721e37a631bd87f24135d5e19"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2025-02-11 18:50:19.000000000","tz":60},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2025-08-29 08:04:33.000000000","tz":120},"subject":"Update glance-store requirement to \u003e\u003d 4.7.0","message":"Update glance-store requirement to \u003e\u003d 4.7.0\n\nThe glance-api service doesn\u0027t start unless there\nis a newer glance-store version installed, when doing\na SLURP upgrade from Antelope to Caracal glance-api\ndid not start until glance-store was upgraded to a\nnewer version.\n\nChange-Id: I8a47d701c195d9418040b4163a3d58af8a4468bc\nSigned-off-by: Tobias Urdin \u003ctobias.urdin@binero.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/de73d00ae0db0c07d13dd9a027639923bc565c19"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/de73d00ae0db0c07d13dd9a027639923bc565c19"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
