)]}'
{"id":"openstack%2Fglance~941692","triplet_id":"openstack%2Fglance~unmaintained%2Fxena~I6326a3e85c1ba4cb1da944a4323769f2399ed2c1","project":"openstack/glance","branch":"unmaintained/xena","topic":"bug/2059809-unmaintained/xena","attention_set":{},"removed_from_attention_set":{"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2025-12-05 13:49:33.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I6326a3e85c1ba4cb1da944a4323769f2399ed2c1","subject":"Reject qcow files with data-file attributes","status":"ABANDONED","created":"2025-02-14 16:50:22.000000000","updated":"2025-12-05 13:49:33.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":931639,"cherry_pick_of_patch_set":1,"meta_rev_id":"c3ac229017c6136e8792dd79c8b62af673f711ae","_number":941692,"virtual_id_number":941692,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-09-05 15:08:05.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-14 16:50:22.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"CC"},{"updated":"2025-02-14 18:06:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"37311d84844c1f5e030e79ce373004fe1eb11440","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-02-14 16:50:22.000000000","message":"Patch Set 1: Cherry Picked from branch unmaintained/yoga.","accounts_in_message":[],"_revision_number":1},{"id":"6e62170018b7ca522468d228eabd75a31813ea97","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-14 18:06:50.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/dd4cc47c558e41dd92bb94d0499f20a7\n\n- grenade https://zuul.opendev.org/t/openstack/build/d058326454844e1b9a2bfa5a0c7e175b : SUCCESS in 55m 54s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7688eed573c9418c8875ecfa474123a6 : FAILURE in 52m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/56e826fc987c4bc894ab573399566778 : SUCCESS in 49m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e7a66bdac9848dbbf8815edc4816212 : FAILURE in 4m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5d71fe3a957840c4b9ff05edffe95dd7 : FAILURE in 5m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fa7f3b2379de45438a4414bd0b32c2ff : FAILURE in 4m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d30e87ab9938442fa02a31e95bda6c7c : FAILURE in 5m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0792517079c4f6aa25fb0506a7c7e3b : SUCCESS in 5m 06s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/627928d7059a4c0cbdc9f45d3f5d4c2c : SUCCESS in 8m 54s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/89d745266d11441dba75eeda2bea03d5 : SUCCESS in 7m 55s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/275aede0b45d425f9c44eb8bc5d1d837 : SUCCESS in 6m 55s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/881ca84a5a3046a3915467138676a6a8 : POST_FAILURE in 1h 08m 09s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/2466d7d2e88e41f0a8d45b88a293aec4 : POST_FAILURE in 1h 09m 23s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/582fc16b1ddf4373a0633b8c2e9dace1 : POST_FAILURE in 1h 09m 49s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1546f4f3df31422889ae10debe54e98d : SUCCESS in 54m 30s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/2b669b974cd845cfb1d41bc511a9b545 : FAILURE in 25m 20s","accounts_in_message":[],"_revision_number":1},{"id":"017751584869d58e9b3dbec6f3b2df21d56108e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-02-19 13:11:21.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"0a28e7de2e7c217daab18d19e35934460dfe4a58","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-19 14:37:13.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e24d9eea326d4a118419bbbdff478f22\n\n- grenade https://zuul.opendev.org/t/openstack/build/590e11e627054803bb4019c64e148c5a : SUCCESS in 1h 01m 35s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/49a1563013f347ba87508048fea4d71a : FAILURE in 58m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a3543d721d8140bbabe78caad0730d76 : SUCCESS in 41m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2bbfef096c374811bd88d90582731f58 : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3b71e6e2ba1b4c5a98781fe2d4bea2d9 : SUCCESS in 6m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0abc93e245f94b0d9cb64077f71191ae : SUCCESS in 5m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d6e01581e0b94451adf9a9528d200a70 : SUCCESS in 6m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e873e75929404afba6807300839f9cdd : SUCCESS in 5m 33s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/fd62af588938432b8269f37d17e6a159 : TIMED_OUT in 30m 40s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f710a26515c94f81909361a8e888736a : SUCCESS in 8m 22s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/ada33f4e81374444947e0665b18eb6e8 : SUCCESS in 8m 27s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/80be249408ee44cb996eb9d447b69c35 : POST_FAILURE in 1h 04m 41s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/c87ebb2400a4425d86c03222051c20d8 : POST_FAILURE in 1h 00m 35s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/d0fbece870854d698ec0370b75176096 : POST_FAILURE in 1h 05m 39s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/266e34da6a1d4299836117cb320ad47d : SUCCESS in 54m 12s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/e3abc9131ef44cf0abe7713c17c58ab7 : FAILURE in 38m 18s","accounts_in_message":[],"_revision_number":2},{"id":"315482d906c05fc941981af9912278f95afbf784","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-02-20 08:40:14.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4a271bed197b06d763b6b3a350a1490fa83e42e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 10:23:57.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/803bcb2d393a44149410885ef87c0892\n\n- grenade https://zuul.opendev.org/t/openstack/build/191b7737f2ad43a5ab142c965378e730 : SUCCESS in 1h 06m 43s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/41b6b348125647efa59f3a6764a2ae2e : FAILURE in 1h 34m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/51742838893c4df89b820e655e5c5a71 : SUCCESS in 48m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84ea6e0916c64be78210e52d035635ec : SUCCESS in 5m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/16a558f804bc4a3b9692b5f0ee4bbbe0 : SUCCESS in 10m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5a5dc323aa1f4666b79406eba4609b08 : SUCCESS in 4m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f1fdc4cd4ad84a6593b32d819b224be7 : SUCCESS in 5m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64562ec695934013a4143ab063587cc8 : SUCCESS in 3m 57s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4080d45287d742a5ad5ea7d88dd86f29 : SUCCESS in 13m 11s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/6a4cde25d0d642fc95b43f67d6351e51 : SUCCESS in 7m 39s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/0d5521d3633c44869e2e674dbaacbcbf : SUCCESS in 7m 11s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/0ed3cfb764814792b599a6f2be7874e7 : POST_FAILURE in 1h 32m 31s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/63a44f96535d4534a0e6d0051a1476a3 : POST_FAILURE in 1h 15m 26s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/8acfdbc497fd48a39966e97d61d83193 : POST_FAILURE in 1h 01m 38s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f4880c35429e428592b037ccd6e78fc2 : SUCCESS in 40m 16s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/00a458f74c854b7fbbb23bd6938b66f9 : FAILURE in 36m 09s","accounts_in_message":[],"_revision_number":2},{"id":"2b34d0ba3965499f3c9ee0382d13dd3ae30b91e2","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-02-25 09:42:50.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b88e6b92812b16670cbf7bc9bcf860f9d42e9e6e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-25 11:00:50.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c0269d7769ea407fa28d5859c163e4af\n\n- grenade https://zuul.opendev.org/t/openstack/build/8ac12cb505b64bbd8b369ea975875352 : FAILURE in 55m 17s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1d7ac883a14d436aa4c9c3b91a241d73 : FAILURE in 1h 00m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/53e25f0173c9476eb9149069b89f9ab3 : SUCCESS in 46m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c50d672b6132419e9abf0c0f37d2e01b : SUCCESS in 3m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b4214df2adf045179288983bbc798de2 : SUCCESS in 7m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4d4b6ed12f744d3caf87c541f0dec622 : SUCCESS in 4m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5837e7f96e594e78aed55b18ddb88e72 : SUCCESS in 5m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/458467d9a1fb4e3fa5ccfa37a01055a1 : SUCCESS in 5m 11s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6151886e98304efbb095760556be5e6d : TIMED_OUT in 30m 45s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5e6a7fcef2d5466599045222ccaf399c : SUCCESS in 6m 44s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/ed1906f3371246278b50cace85bc0d36 : SUCCESS in 8m 24s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/241a1b41398440748a0af7b1fd865939 : POST_FAILURE in 1h 12m 36s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/bcf6bbcc74eb4cc28457818558cc3385 : POST_FAILURE in 1h 08m 57s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/55cf4751b5a5477596130bd202bd8d67 : POST_FAILURE in 1h 12m 04s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d5f9be13ab4743bfb75e2fb97b1f12b6 : SUCCESS in 41m 30s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/28d46a04982c4ddfb902c4818a5e4097 : FAILURE in 19m 26s","accounts_in_message":[],"_revision_number":2},{"id":"fa4aee4206dc91b0d1808020bb04383cd62ee3d7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-09-05 14:08:15.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"a4d418a7c8ad8c11138f0c4a1e7a9883aa9d1871","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-05 15:08:05.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e069307536a14b9d8886013b774010d4\n\n- grenade https://zuul.opendev.org/t/openstack/build/acf650a78474421ba880ae41fb1a81e5 : FAILURE in 20m 55s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c36956186e684310aa8948bd99e1c355 : FAILURE in 48m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cb23d744504d45bfbe93885cee6729b3 : SUCCESS in 21m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e1ed043a76c4350842fa929b2ecc509 : SUCCESS in 3m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2ef83a8c1bbd4c9cbdda0c54392c428d : RETRY_LIMIT in 9s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/937ee5335d8249fa93321d2883e22401 : SUCCESS in 3m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7cc9883734b54323bf99e7083aebb5a2 : SUCCESS in 7m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dbf0817379374835a37b4571cbb1ecca : SUCCESS in 4m 38s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6910f30e005f41a6b890a6c9558f8724 : RETRY_LIMIT in 9s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/6e525c7023e9463ca75a8aa3f10bfc81 : FAILURE in 6m 43s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/1791d4ca454c4974874cc9851b5f2807 : SUCCESS in 7m 18s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/0037bbcfb7a945adbf9710fb72ab3976 : POST_FAILURE in 41m 48s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/710a2ea64693455f92715aa1dda6b17c : POST_FAILURE in 51m 00s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/fc74a8f43492452bbe08a25a649f9e70 : POST_FAILURE in 45m 32s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cddcc62876d345c7a2dbd80d2cd9f137 : SUCCESS in 45m 22s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/e373078b68a44e3a8da2333c5eef6fd4 : FAILURE in 22m 18s","accounts_in_message":[],"_revision_number":3},{"id":"c3ac229017c6136e8792dd79c8b62af673f711ae","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-12-05 13:49:33.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"d4465d1570fccddc33d62868e621b3108dceb5ea","revisions":{"d0f6064eaceb9c6cda2f0db1234d931e51ec8eea":{"kind":"REWORK","_number":1,"created":"2025-02-14 16:50:22.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/92/941692/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/92/941692/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/92/941692/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/92/941692/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/92/941692/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/92/941692/1"}}},"commit":{"parents":[{"commit":"e113723f3fa85cfbbab9447b39e5106d4603f285","subject":"Update .gitreview for unmaintained/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e113723f3fa85cfbbab9447b39e5106d4603f285"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2024-04-01 15:06:31.000000000","tz":-420},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2025-02-14 16:50:22.000000000","tz":0},"subject":"Reject qcow files with data-file attributes","message":"Reject qcow files with data-file attributes\n\nChange-Id: I6326a3e85c1ba4cb1da944a4323769f2399ed2c1\nCloses-Bug: #2059809\n(cherry picked from commit 2ca29af4433e9fa99a0a48e230d8d25d6eaa4a87)\n(cherry picked from commit c3586f3a122f6cb0663217b12b52203e74e2e4fa)\n(cherry picked from commit a92c438fb5ba55440b38cae7c8b4361b58daa9dd)\n(cherry picked from commit dba3bdb458aa8a5d0193f12b7f1e374a89ed34a2)\n(cherry picked from commit 6a38aef8baaf5caecbd8c866f1cf922d939dfbcc)\n(cherry picked from commit f7f7249b343bb7c1f6425a3dc11d926ecbbd40ba)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d0f6064eaceb9c6cda2f0db1234d931e51ec8eea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d0f6064eaceb9c6cda2f0db1234d931e51ec8eea"}]},"branch":"refs/heads/unmaintained/xena"},"aa81aaf5c5a30bf6088d377cbc350b02e6974bbd":{"kind":"REWORK","_number":2,"created":"2025-02-19 13:11:21.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/92/941692/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/92/941692/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/92/941692/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/92/941692/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/92/941692/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/92/941692/2"}}},"commit":{"parents":[{"commit":"e113723f3fa85cfbbab9447b39e5106d4603f285","subject":"Update .gitreview for unmaintained/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e113723f3fa85cfbbab9447b39e5106d4603f285"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2024-04-01 15:06:31.000000000","tz":-420},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2025-02-19 13:10:39.000000000","tz":60},"subject":"Reject qcow files with data-file attributes","message":"Reject qcow files with data-file attributes\n\nChange-Id: I6326a3e85c1ba4cb1da944a4323769f2399ed2c1\nCloses-Bug: #2059809\n(cherry picked from commit 2ca29af4433e9fa99a0a48e230d8d25d6eaa4a87)\n(cherry picked from commit c3586f3a122f6cb0663217b12b52203e74e2e4fa)\n(cherry picked from commit a92c438fb5ba55440b38cae7c8b4361b58daa9dd)\n(cherry picked from commit dba3bdb458aa8a5d0193f12b7f1e374a89ed34a2)\n(cherry picked from commit 6a38aef8baaf5caecbd8c866f1cf922d939dfbcc)\n(cherry picked from commit f7f7249b343bb7c1f6425a3dc11d926ecbbd40ba)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/aa81aaf5c5a30bf6088d377cbc350b02e6974bbd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/aa81aaf5c5a30bf6088d377cbc350b02e6974bbd"}]},"branch":"refs/heads/unmaintained/xena"},"d4465d1570fccddc33d62868e621b3108dceb5ea":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":3,"created":"2025-09-05 14:08:15.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/92/941692/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/92/941692/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/92/941692/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/92/941692/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/92/941692/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/92/941692/3"}}},"commit":{"parents":[{"commit":"0c11c6fa2bf677ea2e336609a49f21ad9821d217","subject":"Unmaintained only: make grenade non-voting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0c11c6fa2bf677ea2e336609a49f21ad9821d217"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2024-04-01 15:06:31.000000000","tz":-420},"committer":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2025-09-05 14:07:02.000000000","tz":120},"subject":"Reject qcow files with data-file attributes","message":"Reject qcow files with data-file attributes\n\nChange-Id: I6326a3e85c1ba4cb1da944a4323769f2399ed2c1\nCloses-Bug: #2059809\n(cherry picked from commit 2ca29af4433e9fa99a0a48e230d8d25d6eaa4a87)\n(cherry picked from commit c3586f3a122f6cb0663217b12b52203e74e2e4fa)\n(cherry picked from commit a92c438fb5ba55440b38cae7c8b4361b58daa9dd)\n(cherry picked from commit dba3bdb458aa8a5d0193f12b7f1e374a89ed34a2)\n(cherry picked from commit 6a38aef8baaf5caecbd8c866f1cf922d939dfbcc)\n(cherry picked from commit f7f7249b343bb7c1f6425a3dc11d926ecbbd40ba)\nSigned-off-by: lajoskatona \u003clajos.katona@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d4465d1570fccddc33d62868e621b3108dceb5ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d4465d1570fccddc33d62868e621b3108dceb5ea"}]},"branch":"refs/heads/unmaintained/xena"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
