)]}'
{"id":"openstack%2Fglance~941693","triplet_id":"openstack%2Fglance~unmaintained%2Fxena~Iaf86b525397d41bd116999cabe0954a0a7efac65","project":"openstack/glance","branch":"unmaintained/xena","topic":"bug/2059809-unmaintained/xena","attention_set":{},"removed_from_attention_set":{"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2025-12-05 13:49:15.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Iaf86b525397d41bd116999cabe0954a0a7efac65","subject":"Extend format_inspector for QCOW safety","status":"ABANDONED","created":"2025-02-14 17:12:59.000000000","updated":"2025-12-05 13:49:15.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"593301db6721c7672c17f557d7383924381c9124","_number":941693,"virtual_id_number":941693,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-09-05 15:13:34.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-14 17:12:59.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"CC"},{"updated":"2025-02-14 18:40:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e225cdb821afaa3b82d7c85f38f0407793387bba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-02-14 17:12:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"32a5e9cfda43a7b3740b8160fddfc5f611138337","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-02-14 17:16:48.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"52b93722cdd11f00a0891249a571c1fa6b9342d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-14 18:40:40.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c96df9a0430a403f8862a33bf11b3d5a\n\n- grenade https://zuul.opendev.org/t/openstack/build/6ad5b13c5a8a44a58ee4ca0d19d0e558 : SUCCESS in 55m 15s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f7684138c0ef45d899a82b36e1802952 : FAILURE in 48m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5f8d18ae8a3543ce841a9cf42a67c11a : FAILURE in 25m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df390276b30d44cb919ff16850dec346 : FAILURE in 2m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6094e2c3b71f4dd084529634e3ca6237 : FAILURE in 4m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/74fda560d0894e85864d19b74e6c2a19 : FAILURE in 3m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3b4a1a9e16ee4e0485a3154834740352 : FAILURE in 5m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a91e421dd30455db47367d477c9182e : SUCCESS in 5m 24s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/040cb224b4e245c19e5fdce211b86e9e : FAILURE in 6m 22s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/98c79a855e82487da5c0b71cd8975ec5 : SUCCESS in 8m 50s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/9194c36b086b4ccdaa40e798ac2c4421 : FAILURE in 6m 28s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/e1d1e96ba6e2494291792dc1aaa4ee06 : POST_FAILURE in 57m 27s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/36d4502bf213402281069d39df7c4183 : POST_FAILURE in 1h 16m 27s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/8a329bcf34934d78a056632a826861fc : POST_FAILURE in 1h 08m 19s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8a8381d46fb7408fa4cc9ad5fab51999 : SUCCESS in 58m 47s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/e6003b9833ba4f53ac4d1c0404205124 : FAILURE in 32m 34s\n\nWarning:\n  Comments left for invalid file glance/tests/unit/async_/flows/test_import.py","accounts_in_message":[],"_revision_number":2},{"id":"8cabd1ef2b7a9758d273df8669ae4b2ca9562f06","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-02-20 08:44:19.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"d399149292d62ec9f6089bb6e139633096659956","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 10:06:05.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/dcfafdb5a7f1408f9cb611e8635a987c\n\n- grenade https://zuul.opendev.org/t/openstack/build/ba53045c38dc4921bb0ebb4790c0a761 : SUCCESS in 49m 47s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/63441b544a404d23b8ab52b1af90e481 : FAILURE in 50m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6eeca7e4c6d04aa2ae2903f38676c34a : SUCCESS in 1h 11m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9fd2264c5e2c4ea79423bc6ae7587175 : FAILURE in 2m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/345355fd7e35415e8b64b8dbcc679c95 : SUCCESS in 10m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/249e495ad333431da38a275383f5a6cd : SUCCESS in 3m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c40555acd599439c893b8eb80c27a031 : SUCCESS in 12m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/267419e6b40e4572b67166b3a64d08d1 : SUCCESS in 6m 20s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2cb19bea13c144189575834d381935b5 : TIMED_OUT in 30m 54s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/45b18e6346254c90bd47ca0cc0e9f059 : SUCCESS in 9m 08s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/01ca8aabe2e242b79d4243b50410f097 : SUCCESS in 7m 34s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/e132f4f2a787419d8721382543a951de : POST_FAILURE in 1h 05m 27s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/10abbfe9d6774cbe928e737b5c5a3554 : POST_FAILURE in 1h 11m 46s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/3c1dd296de0549739de727cfb2670936 : POST_FAILURE in 1h 02m 39s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/028fd069a4e840abaf202d65c7254b20 : SUCCESS in 49m 39s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/493e2998baee4ae9b82454ade3203e2d : FAILURE in 25m 34s","accounts_in_message":[],"_revision_number":3},{"id":"0809838e291f0e0be5b18d36d865a0740e473f1e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-02-20 13:25:54.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"9fe60a09aad079aa9dd10ef4a5b2c478c0fec1c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 15:04:39.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7a7223b4848f4e0eb2af5df1d500786a\n\n- grenade https://zuul.opendev.org/t/openstack/build/e022892fd18342bdbda69a257034661c : SUCCESS in 1h 19m 35s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/40490cccff41493889ced4a901b598d4 : FAILURE in 1h 08m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ee846cef262d4575b9878a89d3679d29 : SUCCESS in 45m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/384c652b49f247648ce2e7aef04adc29 : SUCCESS in 8m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0e98c016555d40e89ee6ae0fc66448b3 : SUCCESS in 9m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/69b18a124e8b494c89d827114e3e4f72 : SUCCESS in 5m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/29cbd15f9a7e409b82bc7411658ecbf0 : SUCCESS in 5m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53abb57569ed432c9def95880ce5c55b : SUCCESS in 5m 50s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2ce7138b58e7450bbe905d122638a804 : SUCCESS in 8m 14s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/35ac27d30c43436b908e63cc9e58af80 : SUCCESS in 7m 01s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/999b50c3071c4a609c0e65e128ecafff : SUCCESS in 8m 09s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/f9e76a6dabe5423da1c30887526d597c : POST_FAILURE in 1h 31m 14s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/a4113af58e6e4e478c312002f66a8409 : POST_FAILURE in 1h 03m 27s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/4048c04c9a284c8bbf9425d8113fe899 : POST_FAILURE in 59m 40s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5201396a0d124d2ea20906c58405ced2 : SUCCESS in 54m 27s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/7adbaf75f9954002a7e144d568e07e6f : FAILURE in 21m 54s","accounts_in_message":[],"_revision_number":4},{"id":"57abdc1b04b80aa08cb2ba732e85b729783c530e","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-02-25 09:43:05.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"440045585f498b43f93893e78f87c21c19277904","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-25 11:06:23.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b928d86293c144a4a90790e995f01d1c\n\n- grenade https://zuul.opendev.org/t/openstack/build/8fb387006beb4c508259a5b77117d8ac : FAILURE in 41m 54s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c175f06e63a347568cbe76b96e6e6386 : FAILURE in 55m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8d16bac3092140a5bb6e18b9e1756c35 : SUCCESS in 52m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dad35164e8164d838ea60254c3d3ac70 : SUCCESS in 3m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3e2399ef9c2d46feba86b36d8e6c9b3e : SUCCESS in 5m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/df2e015249ef40669b66259d29f653ce : SUCCESS in 3m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6128b74e9ae34efc874ef5c63bb5b81a : SUCCESS in 5m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3535f7dd72b8484fbe82d040eb979f97 : SUCCESS in 4m 50s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ea366c1243344107935501dde26bc4b1 : SUCCESS in 8m 20s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/faf0481ff3a44a988d74a789bb7fbf8a : SUCCESS in 9m 43s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/902e08c8c2e040a6bf71c2c42441bc03 : SUCCESS in 7m 55s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/d45035236cc743c99541682ad96afa62 : POST_FAILURE in 1h 10m 23s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/e9d66e5cf1f74de693ecf4d1432392e6 : POST_FAILURE in 1h 15m 10s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/493486d941af42b59c45bbf199d5dc55 : POST_FAILURE in 1h 17m 32s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6eb5dd2ee72b497db9543a9f2017664b : SUCCESS in 44m 13s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/170a6f955b31498b97ab10b620ada3a9 : FAILURE in 20m 26s","accounts_in_message":[],"_revision_number":4},{"id":"55735d22178b0a8b5cc6306eae7c90b6f1b1fede","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-09-05 14:08:15.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"57f26e896795750a8ab7b524091e22bb5030d561","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-05 15:13:34.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/968308cc3b1d48c8a3e7336c9120da06\n\n- grenade https://zuul.opendev.org/t/openstack/build/7a10df3631124d54be6cd0dd44cff67d : FAILURE in 17m 16s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a2fb15da628d43209226a7f9267f13f1 : FAILURE in 42m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fb9ba9891863403e88180a5945068d86 : SUCCESS in 18m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8838f6fa6712448290de2c959d9415db : SUCCESS in 4m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d37e5394035041ebbe3e579cc7cd7ead : RETRY_LIMIT in 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8cb82bd3f65a4bc2befcfdf1d97ad63a : SUCCESS in 4m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b30afe5938134b8dbbd3aff36022791c : SUCCESS in 7m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ac72f10521b4357a46574186330ed5a : SUCCESS in 6m 02s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4904419f329b437fadb05db0fd291477 : RETRY_LIMIT in 12s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/cd619f613aa543acad77807340af05bd : FAILURE in 6m 42s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/91a07f4aeb2e4438a2f31c690b3a0fac : SUCCESS in 8m 30s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/98e71a3158e24ee2a96e077f0177ee32 : POST_FAILURE in 39m 32s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/2b880ac78d3c413da6257ee8383a57ff : POST_FAILURE in 52m 10s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/bbc8188d00ee4ac498495460dc36ebc3 : POST_FAILURE in 57m 45s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9a661160c47748f098e6e04a1827d419 : SUCCESS in 45m 07s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/b514cfd21df5446bbcdd807b0dc48616 : FAILURE in 12m 05s","accounts_in_message":[],"_revision_number":5},{"id":"593301db6721c7672c17f557d7383924381c9124","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-12-05 13:49:15.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"d5b45582cdb0436c456fe6a985b95ad0117a0cdc","revisions":{"86e07d083037a800fc726f2f82038e505d4d898d":{"kind":"REWORK","_number":1,"created":"2025-02-14 17:12:59.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/93/941693/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/93/941693/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/93/941693/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/93/941693/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/93/941693/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/93/941693/1"}}},"commit":{"parents":[{"commit":"d0f6064eaceb9c6cda2f0db1234d931e51ec8eea","subject":"Reject qcow files with data-file attributes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d0f6064eaceb9c6cda2f0db1234d931e51ec8eea"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2024-04-16 17:29:10.000000000","tz":-420},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2025-02-14 17:12:16.000000000","tz":60},"subject":"Extend format_inspector for QCOW safety","message":"Extend format_inspector for QCOW safety\n\nThis adds two properties to the QcowInspector that makes it able to\nindicate whether the file specifies a backing_file or data_file in the\nheader. Both conditions are considered unsafe for our usage. To\nease checking of this condition, a classmethod is added that takes\na local filename and digests just enough of the file to assert that\nboth conditions are false.\n\nNote for yoga backport:With older qemu installed one of the qemu-img\ncreate commands fails, let\u0027s skip it from unmaintained/yoga and\nbelow that.\n\nChange-Id: Iaf86b525397d41bd116999cabe0954a0a7efac65\nRelated-Bug: #2059809\n(cherry picked from commit ae536bb394793c9a7a219cb498e03d5c81dbbbb7)\n(cherry picked from commit 2eba54e0821106097dfeceb424e53943fd090483)\n(cherry picked from commit 89dbbc838d606f461087e1494d19ddbcf9db0a38)\n(cherry picked from commit 4860024286256b028fabc5ed50274934c3dfdd8a)\n(cherry picked from commit f32d5b8ad865113d499a36f7507a085f583514f9)\n(cherry picked from commit a28d96bb8ca65e50be1efea3f3c0db82eed0bb62)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/86e07d083037a800fc726f2f82038e505d4d898d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/86e07d083037a800fc726f2f82038e505d4d898d"}]},"branch":"refs/heads/unmaintained/xena"},"41c3d99ab0cd9d40ed5811fec95024bd6bd68553":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-02-14 17:16:48.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/93/941693/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/93/941693/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/93/941693/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/93/941693/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/93/941693/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/93/941693/2"}}},"commit":{"parents":[{"commit":"d0f6064eaceb9c6cda2f0db1234d931e51ec8eea","subject":"Reject qcow files with data-file attributes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d0f6064eaceb9c6cda2f0db1234d931e51ec8eea"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2024-04-16 17:29:10.000000000","tz":-420},"committer":{"name":"Lajos Katona","email":"lajos.katona@est.tech","date":"2025-02-14 17:16:48.000000000","tz":0},"subject":"Extend format_inspector for QCOW safety","message":"Extend format_inspector for QCOW safety\n\nThis adds two properties to the QcowInspector that makes it able to\nindicate whether the file specifies a backing_file or data_file in the\nheader. Both conditions are considered unsafe for our usage. To\nease checking of this condition, a classmethod is added that takes\na local filename and digests just enough of the file to assert that\nboth conditions are false.\n\nNote for yoga backport:With older qemu installed one of the qemu-img\ncreate commands fails, let\u0027s skip it from unmaintained/yoga and\nbelow that.\n\nConflicts:\n* glance/tests/unit/common/test_format_inspector.py\n\nChange-Id: Iaf86b525397d41bd116999cabe0954a0a7efac65\nRelated-Bug: #2059809\n(cherry picked from commit ae536bb394793c9a7a219cb498e03d5c81dbbbb7)\n(cherry picked from commit 2eba54e0821106097dfeceb424e53943fd090483)\n(cherry picked from commit 89dbbc838d606f461087e1494d19ddbcf9db0a38)\n(cherry picked from commit 4860024286256b028fabc5ed50274934c3dfdd8a)\n(cherry picked from commit f32d5b8ad865113d499a36f7507a085f583514f9)\n(cherry picked from commit a28d96bb8ca65e50be1efea3f3c0db82eed0bb62)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/41c3d99ab0cd9d40ed5811fec95024bd6bd68553"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/41c3d99ab0cd9d40ed5811fec95024bd6bd68553"}]},"branch":"refs/heads/unmaintained/xena","description":"Edit commit message"},"b5ce136e915347f56a3aa0ad9b7eb72da1ff1527":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2025-02-20 08:44:19.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/93/941693/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/93/941693/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/93/941693/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/93/941693/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/93/941693/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/93/941693/3"}}},"commit":{"parents":[{"commit":"aa81aaf5c5a30bf6088d377cbc350b02e6974bbd","subject":"Reject qcow files with data-file attributes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/aa81aaf5c5a30bf6088d377cbc350b02e6974bbd"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2024-04-16 17:29:10.000000000","tz":-420},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2025-02-20 08:43:42.000000000","tz":60},"subject":"Extend format_inspector for QCOW safety","message":"Extend format_inspector for QCOW safety\n\nThis adds two properties to the QcowInspector that makes it able to\nindicate whether the file specifies a backing_file or data_file in the\nheader. Both conditions are considered unsafe for our usage. To\nease checking of this condition, a classmethod is added that takes\na local filename and digests just enough of the file to assert that\nboth conditions are false.\n\nNote for yoga backport:With older qemu installed one of the qemu-img\ncreate commands fails, let\u0027s skip it from unmaintained/yoga and\nbelow that.\n\nConflicts:\n* glance/tests/unit/common/test_format_inspector.py\n\nChange-Id: Iaf86b525397d41bd116999cabe0954a0a7efac65\nRelated-Bug: #2059809\n(cherry picked from commit ae536bb394793c9a7a219cb498e03d5c81dbbbb7)\n(cherry picked from commit 2eba54e0821106097dfeceb424e53943fd090483)\n(cherry picked from commit 89dbbc838d606f461087e1494d19ddbcf9db0a38)\n(cherry picked from commit 4860024286256b028fabc5ed50274934c3dfdd8a)\n(cherry picked from commit f32d5b8ad865113d499a36f7507a085f583514f9)\n(cherry picked from commit a28d96bb8ca65e50be1efea3f3c0db82eed0bb62)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b5ce136e915347f56a3aa0ad9b7eb72da1ff1527"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b5ce136e915347f56a3aa0ad9b7eb72da1ff1527"}]},"branch":"refs/heads/unmaintained/xena"},"62716dbb77bc8c743eb804368c3f4c381888e131":{"kind":"REWORK","_number":4,"created":"2025-02-20 13:25:54.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/93/941693/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/93/941693/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/93/941693/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/93/941693/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/93/941693/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/93/941693/4"}}},"commit":{"parents":[{"commit":"aa81aaf5c5a30bf6088d377cbc350b02e6974bbd","subject":"Reject qcow files with data-file attributes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/aa81aaf5c5a30bf6088d377cbc350b02e6974bbd"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2024-04-16 17:29:10.000000000","tz":-420},"committer":{"name":"Lajos Katona","email":"lajos.katona@est.tech","date":"2025-02-20 13:25:53.000000000","tz":0},"subject":"Extend format_inspector for QCOW safety","message":"Extend format_inspector for QCOW safety\n\nThis adds two properties to the QcowInspector that makes it able to\nindicate whether the file specifies a backing_file or data_file in the\nheader. Both conditions are considered unsafe for our usage. To\nease checking of this condition, a classmethod is added that takes\na local filename and digests just enough of the file to assert that\nboth conditions are false.\n\nNote for yoga backport:With older qemu installed one of the qemu-img\ncreate commands fails, let\u0027s skip it from unmaintained/yoga and\nbelow that.\n\nConflicts:\n* glance/tests/unit/common/test_format_inspector.py\n\nChange-Id: Iaf86b525397d41bd116999cabe0954a0a7efac65\nRelated-Bug: #2059809\n(cherry picked from commit ae536bb394793c9a7a219cb498e03d5c81dbbbb7)\n(cherry picked from commit 2eba54e0821106097dfeceb424e53943fd090483)\n(cherry picked from commit 89dbbc838d606f461087e1494d19ddbcf9db0a38)\n(cherry picked from commit 4860024286256b028fabc5ed50274934c3dfdd8a)\n(cherry picked from commit f32d5b8ad865113d499a36f7507a085f583514f9)\n(cherry picked from commit a28d96bb8ca65e50be1efea3f3c0db82eed0bb62)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/62716dbb77bc8c743eb804368c3f4c381888e131"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/62716dbb77bc8c743eb804368c3f4c381888e131"}]},"branch":"refs/heads/unmaintained/xena"},"d5b45582cdb0436c456fe6a985b95ad0117a0cdc":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":5,"created":"2025-09-05 14:08:15.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/93/941693/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/93/941693/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/93/941693/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/93/941693/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/93/941693/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/93/941693/5"}}},"commit":{"parents":[{"commit":"d4465d1570fccddc33d62868e621b3108dceb5ea","subject":"Reject qcow files with data-file attributes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d4465d1570fccddc33d62868e621b3108dceb5ea"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2024-04-16 17:29:10.000000000","tz":-420},"committer":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2025-09-05 14:07:16.000000000","tz":120},"subject":"Extend format_inspector for QCOW safety","message":"Extend format_inspector for QCOW safety\n\nThis adds two properties to the QcowInspector that makes it able to\nindicate whether the file specifies a backing_file or data_file in the\nheader. Both conditions are considered unsafe for our usage. To\nease checking of this condition, a classmethod is added that takes\na local filename and digests just enough of the file to assert that\nboth conditions are false.\n\nNote for yoga backport:With older qemu installed one of the qemu-img\ncreate commands fails, let\u0027s skip it from unmaintained/yoga and\nbelow that.\n\nConflicts:\n* glance/tests/unit/common/test_format_inspector.py\n\nChange-Id: Iaf86b525397d41bd116999cabe0954a0a7efac65\nRelated-Bug: #2059809\n(cherry picked from commit ae536bb394793c9a7a219cb498e03d5c81dbbbb7)\n(cherry picked from commit 2eba54e0821106097dfeceb424e53943fd090483)\n(cherry picked from commit 89dbbc838d606f461087e1494d19ddbcf9db0a38)\n(cherry picked from commit 4860024286256b028fabc5ed50274934c3dfdd8a)\n(cherry picked from commit f32d5b8ad865113d499a36f7507a085f583514f9)\n(cherry picked from commit a28d96bb8ca65e50be1efea3f3c0db82eed0bb62)\nSigned-off-by: lajoskatona \u003clajos.katona@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d5b45582cdb0436c456fe6a985b95ad0117a0cdc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d5b45582cdb0436c456fe6a985b95ad0117a0cdc"}]},"branch":"refs/heads/unmaintained/xena"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
