)]}'
{"id":"openstack%2Fglance~941820","triplet_id":"openstack%2Fglance~master~Ibac1794880e9e5dd9cfecf605d868bdafffaccae","project":"openstack/glance","branch":"master","attention_set":{},"removed_from_attention_set":{"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2025-03-20 14:30:10.000000000","reason":"\u003cGERRIT_ACCOUNT_9303\u003e replied on the change","reason_account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-03-21 06:20:00.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_37598\u003e by clicking the attention icon","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2025-02-18 16:46:04.000000000","reason":"\u003cGERRIT_ACCOUNT_8122\u003e replied on the change","reason_account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}}},"hashtags":[],"change_id":"Ibac1794880e9e5dd9cfecf605d868bdafffaccae","subject":"Use common environment name for ostestr testpath","status":"MERGED","created":"2025-02-15 16:05:59.000000000","updated":"2025-03-21 10:49:36.000000000","submitted":"2025-03-21 10:48:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"941820","meta_rev_id":"0fe6279acb17dfe51ad22a9353bb95bdb96ad14e","_number":941820,"virtual_id_number":941820,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-21 10:48:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2025-03-20 14:30:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2025-02-18 16:46:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-02-16 08:18:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2025-03-20 14:30:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-15 16:06:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-02-15 18:32:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-16 08:18:10.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2025-02-16 08:18:39.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2025-02-20 18:43:25.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"2de66684574778d54ba6a43da85bbd0e0a0d9077","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-02-15 16:05:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8863db66bccb8254d1eb6a15c6a21d91854376e7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-15 16:06:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5bddf5a948cf4929a3aaed72857edc35\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/11dfc005fe4a4a368a5e4a2dfbd81772 : NODE_FAILURE Node(set) request 300-0026348538 failed in 0s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/df72c1e71bfd4bb08f87259a8b81346b : NODE_FAILURE Node(set) request 300-0026348539 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0aaf7e1980ca1c86af892714be7f7c089f9773a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-15 18:32:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0e1a9c9fdbe5460282fc739facda0014\n\n- grenade https://zuul.opendev.org/t/openstack/build/82324d485eca41dc9d8a63eddcb0bea5 : SUCCESS in 57m 37s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/debd4de1958d4a98b361b38fd1360a92 : SUCCESS in 43m 58s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/097f51752e274c7b80d45245c302def6 : SUCCESS in 1h 23m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9d08625eaadf413ba3e88efa7430b65c : SUCCESS in 1h 05m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f13612990e84ed1876371f69e0fa807 : SUCCESS in 6m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ac709bd6d51f433182e94eb933096235 : SUCCESS in 4m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b0bc0312a5974e49a9d60cf6685ef0cb : SUCCESS in 3m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/725e7297ea59468883b74e665d16ab02 : SUCCESS in 5m 01s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0c55f2709a8d45b181633e2303f27c24 : SUCCESS in 5m 53s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/dbea914412354f378b2971524b363d15 : SUCCESS in 3m 53s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/b94e1d65dd2a4a78b9f01f171cd55ce1 : SUCCESS in 8m 56s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/61ed4ce9e76a48d39c64ae23b0b3ed6e : SUCCESS in 1h 48m 31s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/65071721d7984ca6a5ca6dba9d346a26 : SUCCESS in 1h 29m 35s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/18350df24b29476984cd525fb77d8b5c : SUCCESS in 1h 25m 44s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/2543313d361c44a6ba2cae1652c8266d : SUCCESS in 1h 31m 49s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/09ca80e9ddb7478291dfe349276e6b3e : SUCCESS in 2h 20m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a180453fb06e4cf3aabe060f951417a6 : SUCCESS in 1h 04m 52s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/432bdbea47134f3faae5dd3b6d43502c : SUCCESS in 1h 37m 44s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/6b5e34bf93af4f84a6e6a452c09b0084 : SUCCESS in 1h 32m 20s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/3523ca98e4794610929cb8bfe2f73240 : SUCCESS in 41m 44s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/1e2de44e24c449c296896245143bbd76 : SUCCESS in 1h 16m 42s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/e0f135ae353148ffa3b49dcfed26ad53 : SUCCESS in 36m 59s","accounts_in_message":[],"_revision_number":1},{"id":"3d1c7848fefdc113ce8c58264afe6ecf7969e82f","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-02-16 08:18:10.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"4b1d2fdffc0b703d3774260c13ddd355a20509b6","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-02-18 16:46:04.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"388d3ca38b70a2a4467bb6f687d0b13d95947fe9","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-03-20 14:30:10.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fc4c233d1ca767bc87393ef7462b7e380ef8e29d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-20 14:30:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f59f92a7d11700fc30573e1616265ff010966471","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-20 17:53:48.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4184c6f8fb374d6d83db3ae41ed38c08\n\n- grenade https://zuul.opendev.org/t/openstack/build/85149c10eae94094826c166ca44479bd : SUCCESS in 57m 02s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8a1048494a664740b267825644ad898e : SUCCESS in 28m 39s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8024281090354f34ae97207ae025f4bb : SUCCESS in 1h 28m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4e162e229afd42628194d804f4a65d21 : SUCCESS in 1h 00m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c560a41318f4a40b8406a2b2e252cee : SUCCESS in 5m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9a9c464f831a49dbadbef7c414ac01bd : SUCCESS in 7m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8c968fd586c542d89c7316c97f98162c : SUCCESS in 5m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0715f17e36b04991908a44364b577ad6 : SUCCESS in 9m 08s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6d9bea03bedf4cad90ce9789e4eafe61 : SUCCESS in 5m 23s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/24765bdde54e459499129b2e084f0b15 : SUCCESS in 7m 30s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/5e701ad27e4c4e208da1131da95b6c76 : SUCCESS in 9m 18s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/a31ed54058c5459d94c0b0c38086580c : SUCCESS in 2h 10m 10s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/bac8de611b874e8bbb6ef2fae41b2883 : FAILURE in 2h 20m 31s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/5baef74820774444aa0bf4d1cc2d2ad2 : SUCCESS in 2h 27m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7367d9e183224697aa7831e585a03034 : SUCCESS in 59m 32s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/8bf5700f2b05412d8251c5cf487f842d : SUCCESS in 1h 58m 32s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/7c6307e7d81844a0852e2112b6a6c6d4 : SUCCESS in 31m 24s","accounts_in_message":[],"_revision_number":1},{"id":"8dda903445e024a2b77397fcdc1bd1d8a1503a6c","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-03-20 18:28:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d330804b235386a7e7017ecf8216bd92d6eb50ed","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-20 18:44:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b90303886fe845bc8db31135577f2601\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b937bf38dd8f49828ec1cf1b6e23555b : SUCCESS in 7m 31s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/69aa6e2cfdc7480cbe47cbe21dcb0b21 : SUCCESS in 9m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"23ed19276d246bfc931599eb947c0ba4a5f7da25","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-20 20:50:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e9761779e07542c6992960f68f16e56f\n\n- grenade https://zuul.opendev.org/t/openstack/build/e5f3d208e40041b6b1b7d5ce34e242d3 : SUCCESS in 47m 08s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/04c3a5d0fc59456c864bce8b38c86902 : SUCCESS in 39m 23s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e4715422f3104418b55237548f13ddcc : SUCCESS in 1h 25m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/16b2b9c6d4f141458b1e237db5ef4662 : SUCCESS in 1h 09m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ef24684f8e5429db16f19b6312926d1 : SUCCESS in 3m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8230f3505cae47b2af124961168e960e : SUCCESS in 20m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/27aa7fa1c9a54217b6306ad23a1dc892 : SUCCESS in 5m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2de6797ac074796b4a1f2a1a8ce8fb1 : SUCCESS in 7m 00s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2151d12feded4300a509593f2dd01e81 : SUCCESS in 4m 59s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/04ce41633bed471fade6865af3abf12c : SUCCESS in 3m 18s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/b321547d2b2447bb9ddb9fd2323ea986 : SUCCESS in 23m 38s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/f53d6405568d4985a4037317a001b771 : SUCCESS in 1h 58m 37s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/9e96c97b2dad4f6eae16911ef3bbd217 : SUCCESS in 1h 57m 17s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/997472d5e2b249619b55cb9af849bef3 : SUCCESS in 2h 01m 32s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/4b10c054ad6e4890aea97807bd3343e9 : SUCCESS in 1h 37m 18s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/42588142347d4aa0bd3c15f9744c741a : SUCCESS in 2h 19m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/309e2865b93b45b692a0eb90e04206a9 : SUCCESS in 49m 39s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/711c164644cc4c0ab73ac6a2c316ea70 : SUCCESS in 1h 06m 58s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/b1ce183049bb4351a6489817150ba630 : SUCCESS in 1h 58m 08s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/ad0292d377354dd29f5cb73d46913f8e : SUCCESS in 29m 08s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/13e6f84ef5e6456ca3f9c437755c8ff1 : SUCCESS in 1h 09m 29s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/8cde1c7b1f1b414ea0f224dffe56865f : SUCCESS in 16m 11s","accounts_in_message":[],"_revision_number":1},{"id":"3232edf7ed49baf1bb382dec3fdcc0d1facf5ef8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-20 20:50:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8b0e65e29578f2bd46e9d5901be90ddcc120a83d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-20 23:12:43.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a0daf4f961d24b5c84bf5ab61a0d2c1d\n\n- grenade https://zuul.opendev.org/t/openstack/build/8f539493a4524b64bbbcf8fca20a2622 : SUCCESS in 46m 54s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6c14407c80d14723b31b04b51e2f12dc : SUCCESS in 45m 57s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4529e1c9dddc4935a1265bb42fc34c5c : SUCCESS in 1h 25m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3d2ff85e0ef449ee92ea1f4164635d42 : SUCCESS in 36m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bbc4ffe8c223471f9ec80ab27f9acbfe : SUCCESS in 5m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/67f154435a10430bac838f947daf218e : SUCCESS in 3m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/406201d5f5fc4fc38e448c8af2ba0a07 : SUCCESS in 3m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0cb8fc6d9f514b2ca9405e843f0cd073 : SUCCESS in 5m 23s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/69acd4dc413a47c795dfc5d8478543bf : SUCCESS in 4m 43s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/cfb95ae90f2c49b1a93614dbbf0a3bc0 : SUCCESS in 6m 14s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/339fdee8a1dc4e83889d304f003f58fd : SUCCESS in 7m 46s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/f532ccf579104eeb8a0c94e1cb0cb240 : FAILURE in 1h 40m 27s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/de5e85cd1ba348eb88f10d3b11f9c508 : SUCCESS in 1h 53m 52s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/9d00a1be653d44f49efb6f51e22dcd37 : SUCCESS in 1h 58m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/117e7ec387384d99bbe42e7ecc0684d8 : SUCCESS in 46m 29s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/07bf2d48e1bb494fa52babf5ab833de4 : SUCCESS in 1h 46m 27s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/1779628f37c6445c84a5e02377c07c43 : SUCCESS in 25m 38s","accounts_in_message":[],"_revision_number":1},{"id":"a46f3c1db63927cfd59cd66a11ec863ba6a41fd4","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-03-21 00:53:48.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"836b7abefc808f44eaf6589261791ae17b1ad6b9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-21 01:03:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5718c9087a154064a475fbf842e0fee0\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/30485f1d4f7a4f41bd2d7ba8c0264526 : SUCCESS in 6m 59s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0ee1816dc9334ca5a74bb36b7184e4de : SUCCESS in 7m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1efdfb9fc5e22d1eab74fcd117e3af47cbdd4bf6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-21 03:28:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6ef349c511ea48be850c70b2f9aa69d1\n\n- grenade https://zuul.opendev.org/t/openstack/build/f1980ff813a1474a98ca16d5b8247447 : SUCCESS in 55m 48s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e062e486d82d4269b6337ddec67c3a3f : SUCCESS in 50m 14s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/396911b9b2814d5cabc42d5b470e6d20 : SUCCESS in 1h 44m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/02e172fe871549e480798f677811e6dd : SUCCESS in 1h 08m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9d1c0463b3e4d74a546c020173c5961 : SUCCESS in 4m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a96ad20738574870bc5cc9d0b4d7cab4 : SUCCESS in 2m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/10e57fa6156d4bd5b4cf7f8768edc54c : SUCCESS in 5m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a5db72d765440658c0f10686bed812f : SUCCESS in 5m 12s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f75e36b621314f53a3b8839dc89b27b7 : SUCCESS in 2m 40s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/65edcadd54074053b45da3d328a36b3a : SUCCESS in 4m 08s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/ccf8cf12cb314c15b2c406b905058e6b : SUCCESS in 6m 25s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/7e3971132c354d1593846ee3a7e3d68f : SUCCESS in 1h 41m 15s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/ad3309d8923a4bce8a5c7798c30106e6 : SUCCESS in 1h 58m 06s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/65c5befce0df44e6bd150f793e879bb6 : SUCCESS in 1h 41m 59s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/bfb6c645dc4d425eaee2a648e5ec8cc6 : FAILURE in 1h 50m 58s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/9059f2c329894331aa6aeb6fbd95da6c : SUCCESS in 2h 28m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f7af0703583d4009b8e862f9f96081ff : SUCCESS in 28m 03s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/7b2512c404914325967296bc97892100 : SUCCESS in 2h 05m 09s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/0b64f7f09dd944acbf237bd8a3853cfa : SUCCESS in 1h 48m 24s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/749540306eac46a78fa2a115aac8c0e1 : SUCCESS in 59m 05s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/79bbfd6ca132441ea16d321082200f89 : SUCCESS in 1h 08m 12s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/eaa62b85e0724f5d81e8a649f0d262b0 : SUCCESS in 24m 16s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/4bfbbbfc47eb4b55bc2271288209fc84 : SUCCESS in 57m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"af7908672fc20cbb38b5e7df402b83fcb12cca5a","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-03-21 06:19:58.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b38354be19babba18ac9dac6852141c3a0ff1361","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-21 06:29:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d5f12545267b4d7eb13bc49c5e6e38bb\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e95eaa96e07c42d99a6cc7ba4802b8b9 : SUCCESS in 7m 32s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4f6bbbf5c4984190888e854e165d391a : SUCCESS in 5m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0c9150bf1d6148d674b2c0fed6b5c9f63dcb09e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-21 08:37:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d787fafd8704fbaa1d3e0b7ffb13bf8\n\n- grenade https://zuul.opendev.org/t/openstack/build/a982c39f123148aca1a1a1c0cea54900 : SUCCESS in 57m 59s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/848effa14cba48ce9c3073300040210e : SUCCESS in 58m 00s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d4ac3fb76e264e3bbdfaa262699847df : SUCCESS in 1h 28m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/63d3e0f5d0ae4626b97cf6b2f2834f20 : SUCCESS in 1h 00m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84db3750c0414627b1ff8d735629570e : SUCCESS in 2m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/393b8d21f8834fea9e76abae15004531 : SUCCESS in 5m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1ff21c25fcd549ea83219eaf175f981b : SUCCESS in 2m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/346b994447204ec98523289627bfca78 : SUCCESS in 5m 28s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3fa2363b1d914dac9ae6fce80dc2fedb : SUCCESS in 5m 12s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d314a1dbbce942a8b8c2f04ba8daf7e2 : SUCCESS in 2m 26s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/a881a02e12c34c9085b3ae4cb16951f0 : SUCCESS in 8m 53s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/1c6a7f4d897f4a568456bf82f2e2c4c8 : SUCCESS in 2h 00m 13s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/87267285be1948c29d79b5fcd67f07fd : SUCCESS in 1h 01m 41s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/f087f2c45b144c86961a8010d5a71eb4 : SUCCESS in 1h 55m 23s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/18f02bdf3ef94058bfc8fa1fcdc73594 : SUCCESS in 2h 08m 13s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/74866fa782454dfd967ded25a75081eb : SUCCESS in 1h 26m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/36da673ac295490ca07320e67c805593 : SUCCESS in 29m 14s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/5d46b39f8f574c1a98e72cbb0e55e98c : SUCCESS in 1h 47m 59s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/128c2bb79067450195c090c634853e2f : SUCCESS in 2h 12m 06s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/45ca375cd3c94621a9454b7e3c08bffe : SUCCESS in 57m 55s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/4e42a0d5c40c4fc9834ddfb066c1dd0b : SUCCESS in 1h 06m 30s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/36939af5ae964068955fb51826fc334d : SUCCESS in 28m 51s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/4a38c288338c4c13b3baf016331ed559 : SUCCESS in 57m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6be7d7c526a32afbc66fc4106d7cc6a0a672b821","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-21 08:38:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"891db0aff7686e47d30c2f3df38446244c7d2491","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-21 10:48:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/92fcd7291f564477a33e393b394418ea\n\n- grenade https://zuul.opendev.org/t/openstack/build/e78b8df45fa3434bb62db54aa7693fce : SUCCESS in 1h 21m 36s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ec705863b4aa4294a552ef9f2ca09834 : SUCCESS in 1h 12m 50s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/fdbce6fca68a4efb8cecabfa5a19daf0 : SUCCESS in 1h 58m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/723980e508174edda1f9d5ff09aae492 : SUCCESS in 1h 03m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8f5b67acb8349eb9f99ec1d727b1680 : SUCCESS in 11m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/76de9bb9b7a245fb80a8b150fa5aea97 : SUCCESS in 5m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/103b60bfb3ce4f0e95034d16375f4988 : SUCCESS in 6m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/885a01335d4549f6ae8a6b990ac62b5d : SUCCESS in 6m 57s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/942e237466c34f619962f691fc71638d : SUCCESS in 4m 46s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/614a6dcbb94042d3bff74469d277eee9 : SUCCESS in 5m 09s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/d59c5bacee29412f8ba8096165926bed : SUCCESS in 4m 15s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/04188641a3124bf9acfb8a3738698050 : SUCCESS in 1h 41m 47s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/d7628dcd451b40c7a651ad878ac48eb6 : SUCCESS in 1h 47m 41s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/d324e92d733a45059564e6a72c0b6b68 : SUCCESS in 1h 52m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/604d13b9d3cc4aabb90aa01be65cb6be : SUCCESS in 56m 41s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/73c8388a21724df1a4d6d83e850c401d : SUCCESS in 1h 48m 36s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/12a8ec1c56ed4ac0b89b50f704861139 : SUCCESS in 56m 24s","accounts_in_message":[],"_revision_number":1},{"id":"aa8d6a574b9201361f0ee0008de191452a9a0f38","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-21 10:48:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"0fe6279acb17dfe51ad22a9353bb95bdb96ad14e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-21 10:49:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/47be19c9c2134e08aa5befeb520b8ea0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c03d2af00d0d47f4b317e7d1f99d47c6 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b5a3075c8a9ea53958020b4edb5a8c70362d4965","revisions":{"b5a3075c8a9ea53958020b4edb5a8c70362d4965":{"kind":"REWORK","_number":1,"created":"2025-02-15 16:05:59.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/20/941820/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/20/941820/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/20/941820/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/20/941820/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/20/941820/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/20/941820/1"}}},"commit":{"parents":[{"commit":"f64de3200e7e12d70543f22086b9460cd50dc240","subject":"Merge \"Remove python 3.8 support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f64de3200e7e12d70543f22086b9460cd50dc240"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-02-15 16:05:19.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-02-15 16:05:19.000000000","tz":540},"subject":"Use common environment name for ostestr testpath","message":"Use common environment name for ostestr testpath\n\nOS_TEST_PATH, instead of TEST_PATH, is used widely in OpenStack repos\n(even the other glance-related repos).\n\nChange-Id: Ibac1794880e9e5dd9cfecf605d868bdafffaccae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b5a3075c8a9ea53958020b4edb5a8c70362d4965"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b5a3075c8a9ea53958020b4edb5a8c70362d4965"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
