)]}'
{"id":"openstack%2Fglance~942882","triplet_id":"openstack%2Fglance~master~I5292952556a09c6899ac63db718dd553fae46450","project":"openstack/glance","branch":"master","topic":"ironic-job","attention_set":{},"removed_from_attention_set":{"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2025-03-20 19:50:05.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5292952556a09c6899ac63db718dd553fae46450","subject":"Add cross project Ironic non voting job","status":"MERGED","created":"2025-02-27 08:21:13.000000000","updated":"2025-03-20 19:54:07.000000000","submitted":"2025-03-20 19:50:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"942882-ironic-job","meta_rev_id":"f2985631b749c957eba005716972443e670159b4","_number":942882,"virtual_id_number":942882,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-20 19:50:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"recommended":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"all":[{"value":1,"date":"2025-03-20 14:29:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},{"value":2,"date":"2025-03-14 12:37:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":0,"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},{"value":1,"date":"2025-03-20 14:30:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-27 08:54:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-02-27 11:10:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-14 12:37:18.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2025-03-20 14:29:41.000000000","updated_by":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"reviewer":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"state":"REVIEWER"}],"messages":[{"id":"d4b530801c4c34f3c56d2c7d3f5e91fbef6b62f8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-02-27 08:21:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8f77ca01b37b49563fd0a9580001cb3140e7c7a4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-27 08:54:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/66332226dc17461fbeee32bdd4a4807c\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/399fe98253d6419d93963dc5b0e34b26 : SUCCESS in 13m 38s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a8ef72c5796740d2b67f192c5654782e : SUCCESS in 12m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ad472d17f1a3b165ccde9233d75347ca64dd44c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-27 11:10:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dd1d99de16794c5ca04a5987cd84f87e\n\n- grenade https://zuul.opendev.org/t/openstack/build/e35cec53a77a48519e3759e6e0a830e7 : SUCCESS in 47m 46s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7110bc0ed7f446ff95f73da082f6ec28 : SUCCESS in 48m 46s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7d5b5bfb281642d9a2d0a61c659d9189 : SUCCESS in 1h 30m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ac80fb493d934c779d4aa21b1812fcb7 : SUCCESS in 50m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/200af322e3f841c1adde44c00ae773c9 : SUCCESS in 5m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/65075b416f0c43bfbc36413fe9ed0a8b : SUCCESS in 5m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/24c7bc79061149cd85929cb438376c66 : SUCCESS in 5m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d8e79e98d44475db2effb17ce1b731b : SUCCESS in 21m 58s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0eada0ee6c8441eb86dbc81f958ed752 : SUCCESS in 9m 18s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/39e14e6c9c774020b0f5925694688db5 : SUCCESS in 24m 53s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/22297035fbee4ee0a1215c2ffe460fae : SUCCESS in 11m 55s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/7920c3de875742c5807a0ae9c36d1bbf : SUCCESS in 2h 41m 35s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/497c70d3ec504391b3857da145c55628 : SUCCESS in 2h 39m 09s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/b5d34e4bc4874b679ff03929feba23ce : SUCCESS in 1h 48m 31s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/015c824894624e24a4d878b4bd9a20bb : SUCCESS in 2h 43m 59s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/c3420c8309b146659e3affc4726dfb94 : SUCCESS in 1h 20m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d535cbb93a054734aeecd0bba3088b3e : SUCCESS in 1h 55m 28s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/83b529c54ea94d46b8341dd148c24eca : SUCCESS in 1h 24m 43s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/93ded5244813412d8bfd8c43e381eb9c : FAILURE in 1h 09m 37s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/383a8302fc2b47299b49ede208ac0e30 : SUCCESS in 2h 26m 57s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/c004d0cdad2d4946b2c0587325b53459 : SUCCESS in 1h 52m 22s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/94cec450984c482db7e9584fab64ecd8 : SUCCESS in 39m 00s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/9a30bf41ef834bf9858b8ae93adedd6d : FAILURE in 1h 14m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6a7915b70c1ad60600a6c74a8ca6d70d9c2f5854","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-03-14 12:37:18.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"bef3fb91fad57e7f7d06e84563ee312036dca590","author":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"date":"2025-03-20 14:29:41.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"36879e85bf079336387bb1dd86bf225e7eb33cc7","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-03-20 14:30:50.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"60d3c8b9ddfd1119255cc801dcca52f8c41b9ffa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-20 14:31:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b90406f7e43ef31a1d7a184e869a0881f4511314","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-20 19:50:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/658417658f504401a1fd684e73f1da9b\n\n- grenade https://zuul.opendev.org/t/openstack/build/2a21dacf57da47f09aca05ead2ef1bc9 : SUCCESS in 53m 56s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/26258358b3d549ca92743215f5f2fe7e : SUCCESS in 53m 07s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/251f6625a7c64ee19eaadf3d11c31a9f : SUCCESS in 1h 41m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0b802909314345dab9d70e6ebbc29eca : SUCCESS in 56m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ced7eee8a474943a11e00d50416307e : SUCCESS in 4m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eb3c595f073a400589bcc34a5ebcfee8 : SUCCESS in 5m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3d2317b81f8a488a8b19c2a5a1bd6b99 : SUCCESS in 5m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/930e27f9c1a2419fa6aec3962d971b50 : SUCCESS in 5m 14s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/69713b30f681420285a3f51e6bad28ec : SUCCESS in 7m 47s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/824a4c60f0354f55a4d837be60d26a57 : SUCCESS in 9m 02s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/a16aec09a5c3402aa8c2f5b60bc29e6a : SUCCESS in 11m 31s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/97c99a62d99642dda3cbe9161268e5b2 : SUCCESS in 1h 53m 05s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/f5643da566d1455a9e1f757cde5f1ae7 : SUCCESS in 2h 20m 56s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/51a92548666c4f0aa2551d8ea8842593 : SUCCESS in 1h 47m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1b9a4ddcfaba419686507061a3a954f2 : SUCCESS in 1h 00m 11s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/08271517362c4cb3b3d08b68c6c29a0a : SUCCESS in 1h 42m 07s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/7e4bed55b3584ead9f6fa69ed53dad30 : SUCCESS in 34m 22s","accounts_in_message":[],"_revision_number":1},{"id":"df792fbdd98dd0373ae535ab8c09773567b0bfb3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-20 19:50:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"f2985631b749c957eba005716972443e670159b4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-20 19:54:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8249b6ac62ad4f60b12f10f185725525\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c8057fb334f48dfb9556f4d2dd74a36 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5c1da616bb172c93247ba992e42af88ba1092095","revisions":{"5c1da616bb172c93247ba992e42af88ba1092095":{"kind":"REWORK","_number":1,"created":"2025-02-27 08:21:13.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/82/942882/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/82/942882/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/82/942882/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/82/942882/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/82/942882/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/82/942882/1"}}},"commit":{"parents":[{"commit":"f54f6b3b12a6eb6c50a58de5f40d9b96f1da5184","subject":"Safety check output of image conversion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f54f6b3b12a6eb6c50a58de5f40d9b96f1da5184"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-02-27 08:04:51.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-02-27 08:04:51.000000000","tz":0},"subject":"Add cross project Ironic non voting job","message":"Add cross project Ironic non voting job\n\nChange-Id: I5292952556a09c6899ac63db718dd553fae46450\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5c1da616bb172c93247ba992e42af88ba1092095"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5c1da616bb172c93247ba992e42af88ba1092095"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
