)]}'
{"id":"openstack%2Fglance~943932","triplet_id":"openstack%2Fglance~master~Idabe0164a82b6aacaab313aa16e96799d847b074","project":"openstack/glance","branch":"master","topic":"refresh-configs","attention_set":{},"removed_from_attention_set":{"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2025-03-11 16:17:48.000000000","reason":"\u003cGERRIT_ACCOUNT_8122\u003e replied on the change","reason_account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}}},"hashtags":[],"change_id":"Idabe0164a82b6aacaab313aa16e96799d847b074","subject":"Refresh Glance example configs for Epoxy rc1","status":"MERGED","created":"2025-03-10 13:03:39.000000000","updated":"2025-03-11 20:57:55.000000000","submitted":"2025-03-11 20:56:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"943932-refresh-configs","meta_rev_id":"7f4dbf9e3541cc0ac71308048608f7399fc9b59a","_number":943932,"virtual_id_number":943932,"owner":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-11 20:56:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2025-03-10 14:09:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2025-03-10 14:09:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-10 13:19:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-03-10 14:09:33.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2025-03-10 15:12:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3912e6fd9e75f7d99b2465553c089b100566f5a8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-03-10 13:03:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"47e2c3b6f88d82c922f810f61c8c65b3e74a8253","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-10 13:19:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a0700556e674e4a845550371b020cc2\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9114dba942ab4b8dacb3cff1f5a781c8 : SUCCESS in 10m 49s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/fa33f4c1f6de4c77a00a29015680dc78 : SUCCESS in 13m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2396d74ea4a918faf7134fc0d321a28a67f96a81","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-03-10 14:09:33.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8fe5c68d8d708bce8f17382e92582782043f4951","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-10 15:12:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f3b40c780e6645fd978fbc7e5dd44944\n\n- grenade https://zuul.opendev.org/t/openstack/build/e1ac9ff5243c424187873d49f4802771 : SUCCESS in 1h 12m 19s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1e82244f7e974397aa1429059d4329b0 : SUCCESS in 32m 48s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/10d14f375ee149bca2a11df4691ba3d3 : SUCCESS in 1h 51m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/710c4057c72e449bb0350bce4ccd37f2 : SUCCESS in 40m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0fb92dff27a04b6da05b732fb90951a8 : SUCCESS in 5m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c447da3aaf9a45be9b4c97c39340659a : SUCCESS in 3m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bfb3009bdc854e1491937d10b0d44701 : SUCCESS in 5m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8100972311a0435f8b2ac7939b5062d6 : SUCCESS in 3m 45s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/565257bb60664403a0de2eeca7b253a1 : SUCCESS in 9m 08s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/dff8628524714e128b41c18264cae7a8 : FAILURE in 5m 43s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/c52b27e82d3142a08239efbd515f85dd : SUCCESS in 12m 56s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/dc95a9dd71064e34b55f0c413cd4fe17 : SUCCESS in 1h 51m 25s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/3ade43b314e24c8387acaa8456ff3ec3 : SUCCESS in 1h 49m 50s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/1089284148d84ad4914d0541335de286 : SUCCESS in 1h 10m 42s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/4f36c3db6e8f490ca19108c0fc0f68ac : SUCCESS in 2h 01m 46s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/96a66a61d6444998a50e151a2ddaa3b4 : SUCCESS in 1h 14m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/acadf30f85c84b9386794840854e00bb : SUCCESS in 58m 49s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/643fb4e1bed644aa98d9c3f412d9ae26 : SUCCESS in 1h 10m 35s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/8ca3f80fcd0d4701bb57667b5bc98819 : SUCCESS in 2h 02m 49s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/461c58fc29cc4040ab5bb4cd5602014b : SUCCESS in 28m 53s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/97f77176209d493f9616f744577857f9 : SUCCESS in 1h 55m 05s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/fc7860a5bcf44a6f8f3664ee3b970600 : SUCCESS in 41m 02s","accounts_in_message":[],"_revision_number":1},{"id":"ed6dd7fdef0eaef16ba13481164645328cfdd054","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-03-10 15:14:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e7c8e2f3b2915b5f54a7d9c02f07a0d8ff982bb6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-10 15:48:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d6bc537601d144b69de33331ed7ffed3\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1a28c41709414389a83fc4e284eb0f14 : SUCCESS in 14m 04s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6e8da1affe5f440cb56f738863713e5f : SUCCESS in 14m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"02634331dc344ba5c2e0455910e424a4f52bf484","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-10 17:54:49.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f193a75be94e4287a000e4098d0a143b\n\n- grenade https://zuul.opendev.org/t/openstack/build/da1b727489a6441dab7d2dbde6538d52 : SUCCESS in 56m 15s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ca994106398240ceb0f074ca7a8eb1a1 : SUCCESS in 57m 44s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e21831ddf8124ff088a4c34c73ce51c2 : SUCCESS in 1h 41m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/827e97165d0e4488abdd30c3ffcf658b : SUCCESS in 38m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9b8bbef55574ca9bcb98974e9cca728 : SUCCESS in 4m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0f7aeaf0bb5042588fcf1d8569a0cd78 : SUCCESS in 4m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0318668762f943a9ba121f2c1d5f336e : SUCCESS in 5m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a5cde5f78474798b4e24d84512c68cb : SUCCESS in 3m 51s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e413faecdc324d6798f9e6b653d14a68 : SUCCESS in 9m 31s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/78b4a503a37547cc8154a38cddc52f41 : SUCCESS in 7m 59s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/dea0eee9df3f4fa59603defba4709702 : SUCCESS in 9m 13s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/a4bdf8d117334e8e80b7ee91862d6c1f : SUCCESS in 1h 59m 48s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/1ae4efb3fba8441194920b1fcc5c8a19 : SUCCESS in 1h 50m 32s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/e8c440655ad144d9a144532ac3b816a8 : SUCCESS in 1h 43m 32s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/b5b2e15e04b549e19a54389daba052d9 : SUCCESS in 1h 45m 55s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/89ccd8bb32e94875b9b0efb62dfc9f92 : SUCCESS in 2h 15m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6000ec0aa45943a98bbe3e37a6c12d5c : SUCCESS in 46m 11s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/37c9a8fb7c824b6c9340368276c576d6 : FAILURE in 1h 13m 44s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/40251162638343a69a461cee9b925b0f : SUCCESS in 2h 19m 23s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/e2b0388132b047499240897e250bf0b3 : SUCCESS in 27m 18s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/ada8f4843a5c46b88ba708cac3bccf8f : TIMED_OUT in 2h 34m 21s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f760a86e16644a9f962a1d12e26c18b5 : TIMED_OUT in 1h 11m 51s","accounts_in_message":[],"_revision_number":1},{"id":"ae13cb617e69300b3ad99eb01cbadf20e16fae9f","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-03-11 16:17:48.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"82d02f1b5429e9bc60575a4f570107fd285ee19c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-11 16:31:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a63d4d887dd24c1e9b4f5f414d034b7c\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/dd2e10747b104030ac63f95b2189a8fb : SUCCESS in 9m 53s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7308651d28414ff58cf9c4ed9d7c5ccd : SUCCESS in 11m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"58cb4b1d464fe58f4988f9fb6e8463c5f5ff9453","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-11 18:23:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4b303349bd1645e4b81b8c1d397e0e04\n\n- grenade https://zuul.opendev.org/t/openstack/build/48e1f24295e349be85e90a277ef3f82f : SUCCESS in 1h 06m 26s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/22e5ee9b746e4c4cbb94cbfb9060f22f : SUCCESS in 1h 00m 40s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a8279830902c45d39e897c462e446d7a : SUCCESS in 2h 05m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b49c828ef4984bb7aa473bb6443ed542 : SUCCESS in 37m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cdae4f9306284d8d9a32115bd6a9290a : SUCCESS in 3m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ced199766aca489ea4d90a68f209bf7f : SUCCESS in 2m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/caf88f41f45b4228a230b93d61144ada : SUCCESS in 7m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b7b913728cd47a1924acd6fd0e78eed : SUCCESS in 6m 14s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/de0900b8b44f4b898ef01d902d4205e8 : SUCCESS in 8m 12s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d1e9ce85dd7942e0a2f6d430a299550f : SUCCESS in 8m 58s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/37b7533a85464034aac2a578758085d3 : SUCCESS in 10m 34s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/b1fee55e4806418a9afa55e4b5d720a6 : SUCCESS in 1h 49m 20s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/aa4eade0898d42988652fd75e9125f8a : SUCCESS in 1h 50m 45s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/8649a0c8a64e4f228679f2f89dd57c99 : SUCCESS in 1h 07m 34s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/2ce1d57d15024c96899798ec488c2026 : SUCCESS in 1h 11m 02s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/3ece1051de1e464d9af9b675da29f968 : SUCCESS in 1h 30m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7f834aa465214cf784430c492fdbc395 : SUCCESS in 34m 37s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d5ccdfdf752945eb9414605214368e0c : SUCCESS in 1h 57m 12s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/3ea7fce74b7b49feb5a75c7c3f32e917 : SUCCESS in 1h 10m 59s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/2bdaaab215f34a4cb3556b80fc3edb0e : SUCCESS in 1h 10m 37s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/3934a183869a43bcb3e29fe4f12f1256 : SUCCESS in 1h 45m 20s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/62d46ff5b21849b0b188c02b6fa4a3a6 : SUCCESS in 33m 41s","accounts_in_message":[],"_revision_number":1},{"id":"2489141a13501f4422e1306df2a9a92b9f50733d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-11 18:24:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"34c407d86f229ee2a099b687d96fb06eb3b064be","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-11 20:56:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3c88bac89b294801b5b911ca94af40cc\n\n- grenade https://zuul.opendev.org/t/openstack/build/752207fb2ac943ca87267571c45740c5 : SUCCESS in 58m 12s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f39eaeecb8a44b6492097776914a22d6 : SUCCESS in 51m 46s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ba6d4ae5208543179bd3d36c4cc529d9 : SUCCESS in 1h 47m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2def163baef4404890531b9b68fec705 : SUCCESS in 50m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f051e83ff354dca84f85f2d23e9853a : SUCCESS in 6m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4489649faa2446889a822d5a91b0871f : SUCCESS in 5m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/60fee31d8ab64a4bb0e282fc75c51b50 : SUCCESS in 3m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35c464b701fc4f459d4d72a332831c4f : SUCCESS in 5m 45s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f9695f8c3cec486a99d7e296a262b7a3 : SUCCESS in 8m 43s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/788970284a3b4cd4b219a0e67b766d48 : SUCCESS in 8m 15s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/0ac17d6d29a7492a906a67566d93cfe8 : SUCCESS in 12m 45s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/173ca15e80a844aab79ec704df31483a : SUCCESS in 1h 45m 48s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/f5232d19f91a40c29cbae80bc528aee6 : SUCCESS in 1h 46m 00s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/929cad65d7334ce5979bd792a5628fc9 : SUCCESS in 1h 06m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7f6794e2e4f1440eafb56c0b637ffadc : SUCCESS in 54m 35s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/58ae3d106bbc4cc0af58e87b12e8c0bb : SUCCESS in 1h 43m 35s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/b6ab19bd21f34c7bac41b1115c2393f4 : SUCCESS in 19m 56s","accounts_in_message":[],"_revision_number":1},{"id":"42938685d428215ebdd7ab12efc1f725fdedeb6d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-11 20:56:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7f4dbf9e3541cc0ac71308048608f7399fc9b59a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-11 20:57:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a8d82168f0694f999f82f00c6da0ab2e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96a110e4f59d4b40b006676f31fee523 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1f4f380bd7e08f3219ff06c8cd168e28dc29519d","revisions":{"1f4f380bd7e08f3219ff06c8cd168e28dc29519d":{"kind":"REWORK","_number":1,"created":"2025-03-10 13:03:39.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/32/943932/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/32/943932/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/943932/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/943932/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/943932/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/32/943932/1"}}},"commit":{"parents":[{"commit":"f54f6b3b12a6eb6c50a58de5f40d9b96f1da5184","subject":"Safety check output of image conversion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f54f6b3b12a6eb6c50a58de5f40d9b96f1da5184"}]}],"author":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2025-03-10 13:01:56.000000000","tz":60},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2025-03-10 13:01:56.000000000","tz":60},"subject":"Refresh Glance example configs for Epoxy rc1","message":"Refresh Glance example configs for Epoxy rc1\n\nChange-Id: Idabe0164a82b6aacaab313aa16e96799d847b074\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1f4f380bd7e08f3219ff06c8cd168e28dc29519d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1f4f380bd7e08f3219ff06c8cd168e28dc29519d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
