)]}'
{"id":"openstack%2Fglance~944002","triplet_id":"openstack%2Fglance~master~I5a405e2fada0da7071e835af9006a45d524c4c1b","project":"openstack/glance","branch":"master","topic":"migrate-to-noble","hashtags":[],"change_id":"I5a405e2fada0da7071e835af9006a45d524c4c1b","subject":"Testing doc job on Ubuntu Noble","status":"ABANDONED","created":"2025-03-11 05:27:44.000000000","updated":"2025-03-11 10:58:16.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"dab1cea73009fda458354209c476e9539b23c8c2","_number":944002,"virtual_id_number":944002,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-03-11 08:17:21.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-11 05:41:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-03-11 08:17:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b51df38155af7391c53602c7e22f28915d12168a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-03-11 05:27:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9ffb41930bebad9bbcf09dfec6403a24fb4acdce","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-11 05:41:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da1a2cd521fc4099a97d6d7b51f0bd65\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/63f8a90678c540cbb841a15d18a58745 : SUCCESS in 11m 12s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a3311a89c1ec4be3bb7b5ce31a4d39fe : SUCCESS in 9m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"92aabc83553066d1d40edb2b51a6947238226a4b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-11 08:17:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1f59b3c8b49f41d6bcac9ed14be2d804\n\n- grenade https://zuul.opendev.org/t/openstack/build/21502ef981ea451db1f17ed6904e6868 : SUCCESS in 48m 56s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a7d56361a28843e18fee724230179446 : SUCCESS in 59m 12s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/eb78ec33d16443d3920a86e7505ff123 : SUCCESS in 1h 13m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/14ad5bc48e734b948d68b4e818b63255 : SUCCESS in 48m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c73ada8fb4a4ca2851ca2096d8a41c6 : SUCCESS in 5m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a511d56cb43d433fba7b16393d524b3a : SUCCESS in 5m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7f457602cdc94fd981d2b26e458b4b13 : SUCCESS in 6m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85154d67026c42caa557732b27d5b7a6 : SUCCESS in 5m 13s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a036d28740b64ef08a969ea3f2b228a5 : SUCCESS in 8m 54s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f5c1d955d0e0429e96ec14d3b3c8aabc : SUCCESS in 7m 46s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/6f0ce77228934fb4a29541e4a93a9920 : SUCCESS in 11m 39s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/e2cfa1a7bdf946439e7a090f0c04d43a : SUCCESS in 2h 03m 37s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/da7466cc8a0341e28948fee0fe7e97c2 : SUCCESS in 2h 02m 26s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/e61ffe482ed2467bb99009547fd51dba : SUCCESS in 2h 12m 28s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/b2e75605b0894962a0f4e65be6c4aaed : SUCCESS in 1h 57m 32s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/102a94905d83403fbd42e34b8cb9af9e : SUCCESS in 2h 20m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d9022e80d9bc4087b8d57db318ae46d8 : SUCCESS in 57m 12s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d080c87aeae7418b81de38a47784fadb : SUCCESS in 1h 41m 31s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/5746bc06ee7541a690992ed34e8da624 : SUCCESS in 2h 18m 20s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/dcaa957d9d884a3ca207523cd32a76a9 : SUCCESS in 1h 00m 24s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/2dcce75264ca4972b6d171637a2b2d15 : SUCCESS in 1h 05m 50s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/9920464cab0549db87c5a9ae5b87fc79 : SUCCESS in 38m 44s","accounts_in_message":[],"_revision_number":1},{"id":"dab1cea73009fda458354209c476e9539b23c8c2","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-03-11 10:58:16.000000000","message":"Abandoned\n\nall good, doc job is passing on Noble","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d3ceb7342af941087ad3ab9747e1ccf26bf94011","revisions":{"d3ceb7342af941087ad3ab9747e1ccf26bf94011":{"kind":"REWORK","_number":1,"created":"2025-03-11 05:27:44.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/02/944002/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/02/944002/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/02/944002/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/02/944002/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/02/944002/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/02/944002/1"}}},"commit":{"parents":[{"commit":"f33f55661a9466c97d4558e494cafec0d6a5f691","subject":"Merge \"Add support for service role in discovery API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f33f55661a9466c97d4558e494cafec0d6a5f691"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2025-03-11 05:27:32.000000000","tz":-420},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2025-03-11 05:27:32.000000000","tz":-420},"subject":"Testing doc job on Ubuntu Noble","message":"Testing doc job on Ubuntu Noble\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-zuul-jobs/+/935459\nChange-Id: I5a405e2fada0da7071e835af9006a45d524c4c1b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d3ceb7342af941087ad3ab9747e1ccf26bf94011"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d3ceb7342af941087ad3ab9747e1ccf26bf94011"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
