)]}'
{"id":"openstack%2Fglance~944007","triplet_id":"openstack%2Fglance~stable%2F2024.2~Id40ceab2a84bb7047dfd130bf8c1ac4c8073b79b","project":"openstack/glance","branch":"stable/2024.2","topic":"bug/2081009-stable/2024.2","attention_set":{},"removed_from_attention_set":{"12404":{"account":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"last_update":"2026-04-21 09:22:28.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Id40ceab2a84bb7047dfd130bf8c1ac4c8073b79b","subject":"Do not call Enforcer.__call__ at module level","status":"ABANDONED","created":"2025-03-11 06:07:29.000000000","updated":"2026-04-21 09:22:28.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":929720,"cherry_pick_of_patch_set":2,"meta_rev_id":"dacacf7ee757d40b14eee7edfc9fc6e6e1abaa0a","_number":944007,"virtual_id_number":944007,"owner":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-03-11 08:28:07.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":-1,"date":"2025-03-11 11:21:31.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-11 06:07:29.000000000","updated_by":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"},{"updated":"2025-03-11 06:25:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-03-11 08:28:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-11 11:21:31.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"3d47a189ba2570a91fef4dbaa06de6be99a02b35","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2025-03-11 06:07:29.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"aabff1f77a4f423edc553764e132a76e98092999","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-11 06:25:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/302b8571099f40ed94a32f5a22e7871c\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/859d9be0b27b43eabfe88464257c71da : SUCCESS in 15m 23s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/5ef74d3411c64c0f81e7f5ba611df653 : SUCCESS in 9m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d630abc1370977be3ee54701911938424cca0d42","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-11 08:28:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2b84aa99c95b4212ae35cbd8361f9a9d\n\n- grenade https://zuul.opendev.org/t/openstack/build/162baf786d4541058642096755a760ca : SUCCESS in 57m 31s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5d94639447eb4b64b294794919262185 : SUCCESS in 1h 25m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1372f113b47e44a8afd55b021419a4b7 : SUCCESS in 56m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e53ad252461648a0a9ffa6d742d36420 : SUCCESS in 2m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0688ea49b660464cb0a68592455cb5db : SUCCESS in 5m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/103776fe07d84fd59a6f567e909299ac : SUCCESS in 5m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9e093ea64a1941b1ae1ee027fb8a5f7c : SUCCESS in 7m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e364fbc35a24199aa9121632a26d27b : SUCCESS in 5m 07s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/fc6dd7ca7ce34d77a7dd8ad0984566bb : SUCCESS in 8m 08s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e15af4addde5495094ecce0c22dcb306 : SUCCESS in 6m 47s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/95b38279f6b3460c9a8d5c84a8ee3a41 : SUCCESS in 11m 31s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/84324de34a704169a13d5f68eebeb983 : SUCCESS in 1h 19m 05s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/708f797c3f4b45fbade701fa83994485 : SUCCESS in 1h 49m 27s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/16446f0bf4234a9bb50e5b6d5b4c094f : SUCCESS in 1h 14m 39s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/e4eb0bdc5e6e47af9136db1dcafcc18b : SUCCESS in 2h 09m 33s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/74338adfe7844773ade402ac873c3e55 : SUCCESS in 1h 08m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7fa58cbfa8c94d588010b68b46dd2b25 : SUCCESS in 58m 20s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/4458905cc4e648fcb6f809df99ce9be5 : SUCCESS in 1h 19m 03s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/1584084c6aed4a608d3555afbfe4ff20 : SUCCESS in 1h 58m 15s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/d2fb893817924fffbe707dfd06e60b3a : SUCCESS in 28m 07s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/3e9c2de1e4994944a87b552c74dee468 : SUCCESS in 38m 13s","accounts_in_message":[],"_revision_number":1},{"id":"a9bfe6dbe1c266e01b2e13c8089fbc3f1878c174","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-03-11 11:21:31.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dacacf7ee757d40b14eee7edfc9fc6e6e1abaa0a","tag":"autogenerated:gerrit:abandon","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2026-04-21 09:22:28.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7694364909d09eb4c9591e70d01ed887d7f2ba07","revisions":{"7694364909d09eb4c9591e70d01ed887d7f2ba07":{"kind":"REWORK","_number":1,"created":"2025-03-11 06:07:29.000000000","uploader":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"ref":"refs/changes/07/944007/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/07/944007/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/07/944007/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/07/944007/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/07/944007/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/07/944007/1"}}},"commit":{"parents":[{"commit":"d1cc917a29c9d2e87b1bad51a33a8a2500eb69c6","subject":"Remove S3 credentials from debug log","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d1cc917a29c9d2e87b1bad51a33a8a2500eb69c6"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-09-18 04:49:47.000000000","tz":540},"committer":{"name":"Rico Lin","email":"ricolin@ricolky.com","date":"2025-03-11 06:07:29.000000000","tz":0},"subject":"Do not call Enforcer.__call__ at module level","message":"Do not call Enforcer.__call__ at module level\n\n... because the method may need to user some functionalities which can\nbe used after CONF instance is initialized and module level import\nmakes it difficult to guarantee the order.\n\nCloses-Bug: #2081009\nChange-Id: Id40ceab2a84bb7047dfd130bf8c1ac4c8073b79b\n(cherry picked from commit 562a2eb48b4fc10cf97bd65d9a2f2e6d2a739eba)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7694364909d09eb4c9591e70d01ed887d7f2ba07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7694364909d09eb4c9591e70d01ed887d7f2ba07"}]},"branch":"refs/heads/stable/2024.2"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
