)]}'
{"id":"openstack%2Fglance~944279","triplet_id":"openstack%2Fglance~master~I2a57a39a9edcebaec8a9369013c7010775ee034e","project":"openstack/glance","branch":"master","topic":"pyenv-py313","attention_set":{},"removed_from_attention_set":{"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2025-04-03 08:48:02.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2a57a39a9edcebaec8a9369013c7010775ee034e","subject":"Use pyenv for the periodic jobs running on Python 3.13","status":"MERGED","created":"2025-03-13 15:47:31.000000000","updated":"2025-04-03 08:50:06.000000000","submitted":"2025-04-03 08:48:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"944279-pyenv-py313","meta_rev_id":"c12929a03ca712f8e5d1fe6246a8a2b33fabd3da","_number":944279,"virtual_id_number":944279,"owner":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-04-03 08:48:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2025-04-03 06:32:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2025-04-03 06:32:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-13 16:10:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-03-13 19:01:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-04-03 06:32:53.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"f2b113181125dff5b6a3e16ddfaa39391c739089","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-03-13 15:47:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3a7f7635804cdad9440651a221f142a863d122e4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-13 16:10:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff28cb2ff803435a829fc5cffd55da8b\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/37e293ff582f43cbbe14ba7e247d6136 : SUCCESS in 5m 31s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/44cb11b4bbb24d87ab62ffc22edfe791 : SUCCESS in 8m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d909b0db5212dcff5b62034bdd1f564c5d78dc76","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-13 19:01:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0354911e504f4768b2cbce4171982879\n\n- grenade https://zuul.opendev.org/t/openstack/build/e605b2bddffc4aac86dd7545366288c3 : SUCCESS in 56m 38s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/fb6a9744ccad452593868e5486f8e851 : SUCCESS in 54m 23s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9511fcf3fe124ff884bac686ab709485 : SUCCESS in 1h 30m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/666a2a3c9a074666b2b5595285fb7533 : SUCCESS in 34m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2a01fd2edad40eab5e0e04087cece9f : SUCCESS in 2m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4dae176056ad4f97b22eb5f348e8d82e : SUCCESS in 5m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ee82a7d5e89046ddba97ff59e54cb6c2 : SUCCESS in 6m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e1bff42cea4448698e0857756d80bfc : SUCCESS in 22m 04s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ab6cb0e8e11f4d5591e24925ad97a63c : FAILURE in 6m 39s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0bc8cad475fb4ad196d5d2fe88ef78df : SUCCESS in 8m 34s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/e2f73851e2c84d2da6517cbf489575e7 : SUCCESS in 9m 03s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/9cb41e9cb39d4434bb4f884e348e57ce : SUCCESS in 2h 00m 01s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/fa77e48bed584967b1e4c0b5b67ee376 : SUCCESS in 2h 21m 34s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/5e60448c8c354b3e8a7a75b0b0bc128a : FAILURE in 1h 50m 40s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/7c572fb48d77430a80c101ba75cf6590 : SUCCESS in 2h 12m 55s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/062e618dea3c45c788e93419a98d4e2b : SUCCESS in 2h 04m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/be0aa2ee99b84182a6a708996dcf1f22 : SUCCESS in 1h 28m 37s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/8765e13ce27e4fcdb2f723fb58c863b3 : SUCCESS in 1h 50m 11s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/b90065dce14f4bfcb4cfa4d3da474c7b : SUCCESS in 2h 22m 52s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/ad9dc4b4dcce4735bd8e4f1844b1b0e5 : SUCCESS in 59m 44s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/2166fd53d5b747778ebd02ab7b3cdc42 : SUCCESS in 2h 02m 23s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/89b737720be44108bc968ef032d91b29 : SUCCESS in 34m 12s","accounts_in_message":[],"_revision_number":1},{"id":"27b7fa825ed7480d671536cb23c03ce602442295","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-03-28 15:20:39.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f5fa543822db436804fc8d70138109b8c05b6f9c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-28 15:30:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/18f8aa229f1a4059ae736986cc4d54d0\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b2f255bcfb8c4d6bbe5dbffd3f759576 : SUCCESS in 7m 52s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/bf83aa0762d84eae8bc3e91e24a817d1 : SUCCESS in 7m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0d1771757e58def66476003d219c66aded70cf5b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-28 17:42:44.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3840e8c03e0547dca35b4df29eaa5929\n\n- grenade https://zuul.opendev.org/t/openstack/build/ffd809a5a80c4ff786a7d0d9eae5200b : FAILURE in 37m 07s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/25b939e67e014cc9b0f94e5aa8876385 : SUCCESS in 59m 52s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/134a517c99a44d348487b7e83288e8b6 : SUCCESS in 1h 39m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f6c8abf097a74f79a4e6a35dbf0b40d0 : SUCCESS in 52m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef9f9e9cdd984c4185b5e09a20dcf456 : SUCCESS in 4m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c1f1890cc5fb46e0bd766a15698d276b : SUCCESS in 4m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b21f3ad29b1d446f955b670f37f5a7a5 : SUCCESS in 5m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d5e175f17ac4df79c1a6c95d79467e4 : SUCCESS in 5m 13s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b843644dcfec4baca86561df9649207f : FAILURE in 7m 03s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/79031277dbd44f72aa451fa0a0e7c226 : SUCCESS in 8m 18s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/a81e872ec39f48a2b7ae442590f65877 : SUCCESS in 13m 54s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/903d48d5ea184f4c89e97ad4b8e0ffa5 : SUCCESS in 1h 55m 26s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/ce524fd5e97e429899d46655d93b92c4 : SUCCESS in 2h 15m 37s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/87e9924a13054d64b11906c43cb67a30 : SUCCESS in 1h 56m 30s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/b36518269a7f4f6ea0f56338763e8594 : SUCCESS in 1h 58m 03s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/6c1377645744451bad9bcb59b9c39329 : SUCCESS in 2h 09m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f2184a2095494e99bff256fc491d96e1 : SUCCESS in 52m 45s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/36293375b8e942c19b665aaa38e16ab7 : SUCCESS in 1h 44m 02s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/1e4e5d8e69354677a6ab558d5f0289b0 : SUCCESS in 1h 58m 05s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/fcb19401388644d3a07ec9acf1b3f8e9 : SUCCESS in 56m 18s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/fb0a1201b0fb4f78b385db42ff0ec632 : SUCCESS in 1h 02m 20s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/a732ff929180404780c3d975a03f3b62 : SUCCESS in 31m 40s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/6d305d9d7fd841ad9c8779c4df5de331 : SUCCESS in 51m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"93a30ed715743db9ef7d03966e3aac030c5b617f","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-04-01 14:37:49.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"dad97540b000da054c553ea762be5105395e05eb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-01 14:45:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/64b23a6a466d4c85aab044607e722240\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ff0cecf8b2604a17bc6b0d38ef9d9259 : SUCCESS in 5m 22s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d12c7472176740158e02f10f47690068 : SUCCESS in 5m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bd3f92894d6e69e8ca68b55f99ecd35533a466e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-01 17:02:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b7fc037195b142108b7063751dbb36ff\n\n- grenade https://zuul.opendev.org/t/openstack/build/ff4b68a01bc641118ab8c21115d64592 : SUCCESS in 51m 19s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d8e52f921bfc472cb1fc21a386d64c24 : SUCCESS in 55m 13s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/aee957006eb94d30ada6aa34be00d029 : SUCCESS in 1h 27m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8dc3a0b892054754b1734f983bca9390 : SUCCESS in 51m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d692b92e8b8494da37f6a1793dfe5aa : SUCCESS in 5m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e788bf08a5f64e9da370b01b983c1e8c : SUCCESS in 4m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d2bac2e263e2455084d5394fc68bb50a : SUCCESS in 5m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ccdf8f5e45d44f0a8af1da128b2558e1 : SUCCESS in 5m 50s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b8c958af89e847dc973100d011ffb326 : SUCCESS in 8m 53s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/eb3e22723f9148bc9936f81ffcbd93d8 : SUCCESS in 9m 16s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/338bf714698f448595e33e5e8d0f9de1 : SUCCESS in 13m 18s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/0cb0c2a93ddb46e48c96f3ce39712eb2 : SUCCESS in 1h 44m 15s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/a53430cbe01e4e91b7723ed785c81499 : SUCCESS in 1h 46m 19s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/beb6fceb43ad42f7a615209f14d3f7d0 : SUCCESS in 2h 16m 21s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/013694b75f124bdbbf5b5568243bf577 : SUCCESS in 2h 05m 32s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/c0d369eb23c44dc6856a65e4209234c3 : SUCCESS in 2h 21m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/04d92ff78b4243b3ad45de68d8fa23f8 : SUCCESS in 46m 48s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d5f3a0cd70fe4b46b46b29213d5d8783 : SUCCESS in 2h 18m 58s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/bc2fa525a6464f2fad68fa333e82ea67 : SUCCESS in 1h 55m 02s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/38cc5a0ea75f4da0a51b7e28de44d1b0 : SUCCESS in 1h 08m 18s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/7556ce01611945d8b7883b2205ac8e9b : SUCCESS in 1h 23m 30s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/857247e1ca104b838ad10ac6ba15e094 : SUCCESS in 24m 19s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/230616628c4b49389701dd543b6a9bd8 : SUCCESS in 56m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"319d87f6810c0e0b522487ae92172d0c4125909a","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-04-03 06:32:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"20bae2355da4c282c6250e43008466acf9fbde38","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-03 06:33:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"24ba7d8384b3a40feb4bb15bf1200799de07ba1e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-03 08:48:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/493b844c0d374fa8990625386316dfeb\n\n- grenade https://zuul.opendev.org/t/openstack/build/374d39b69a7d4674928b6dd6d89c4dce : SUCCESS in 44m 29s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/83e1aae78991409583ee5dfb39b09c80 : SUCCESS in 44m 30s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3ca3a29282c64b089c3b36524a881ef7 : SUCCESS in 1h 41m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a472fc6bc3454912ab9f69b89dc3afce : SUCCESS in 1h 00m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35aed72afaaf4c57b72d4493702596d0 : SUCCESS in 4m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6f7fd73659bb414a9f8439435c3ad0a7 : SUCCESS in 4m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ff761fac6f164a538b033c9d88dfa918 : SUCCESS in 5m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81d8b1b1491345c39b883ec3f569b281 : SUCCESS in 6m 36s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c174018c6b6f4f908770e5158baefc94 : SUCCESS in 7m 41s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/dba007d769a04e75ba7c2fa0af74c81d : SUCCESS in 8m 08s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/b2bbd2ba5ae24912b1bd1477576fa307 : SUCCESS in 12m 17s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/7b3422b030694561b0f816e4b13478c2 : SUCCESS in 1h 47m 33s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/31c5671388d242e2af3fa643e7abd2a2 : SUCCESS in 1h 02m 36s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/d50fd8d0ea5b4bfdab1f50ac3612891f : SUCCESS in 2h 02m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1361c7759523469782751cf4b9ce4d43 : SUCCESS in 45m 13s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/479f4a145eb54e23b42b4e8dfa631484 : SUCCESS in 2h 09m 49s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/c1811e4e9541404ea5bcec662ab02136 : SUCCESS in 33m 08s","accounts_in_message":[],"_revision_number":1},{"id":"ec6e1c6c0358726cfdf6999921b37b980be49911","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-03 08:48:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c12929a03ca712f8e5d1fe6246a8a2b33fabd3da","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-03 08:50:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a489dc7b8a4046b9a7836db2ea8232f5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07f24126480e4405925096504183971d : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8546ce00e999d9653aa1bc1a08889afce6631720","revisions":{"8546ce00e999d9653aa1bc1a08889afce6631720":{"kind":"REWORK","_number":1,"created":"2025-03-13 15:47:31.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/79/944279/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/79/944279/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/79/944279/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/79/944279/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/79/944279/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/79/944279/1"}}},"commit":{"parents":[{"commit":"5b12f046e96bb237a6e2c004decbffeee24e11fb","subject":"Merge \"Refresh Glance example configs for Epoxy rc1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5b12f046e96bb237a6e2c004decbffeee24e11fb"}]}],"author":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2025-03-13 15:44:59.000000000","tz":60},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2025-03-13 15:45:37.000000000","tz":60},"subject":"Use pyenv for the periodic jobs running on Python 3.13","message":"Use pyenv for the periodic jobs running on Python 3.13\n\nPython 3.13 is not available yet, so use pyenv to work around that.\n\nChange-Id: I2a57a39a9edcebaec8a9369013c7010775ee034e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8546ce00e999d9653aa1bc1a08889afce6631720"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8546ce00e999d9653aa1bc1a08889afce6631720"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
