)]}'
{"id":"openstack%2Fglance~946343","triplet_id":"openstack%2Fglance~master~I698dbd2946555c0a6455b1be17a6bf4a8d407ee7","project":"openstack/glance","branch":"master","topic":"bug/2106223","attention_set":{},"removed_from_attention_set":{"34034":{"account":{"_account_id":34034,"name":"Franciszek Przewoźny","display_name":"Franciszek Przewozny","email":"przewozny.franciszek@gmail.com","username":"fprzewozn","status":"OpenStack Engineer / SRE @ Opera Software"},"last_update":"2025-04-29 22:43:56.000000000","reason":"Change was submitted"},"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2025-04-29 22:43:56.000000000","reason":"Change was submitted"},"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2025-04-29 22:43:56.000000000","reason":"Change was submitted"},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2025-04-28 15:57:02.000000000","reason":"\u003cGERRIT_ACCOUNT_8122\u003e replied on the change","reason_account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}}},"hashtags":[],"change_id":"I698dbd2946555c0a6455b1be17a6bf4a8d407ee7","subject":"Fix vTPM metadefs.","status":"MERGED","created":"2025-04-04 12:36:52.000000000","updated":"2025-04-29 22:44:59.000000000","submitted":"2025-04-29 22:43:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":15,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"946343-bug/2106223","meta_rev_id":"e580345972498cc43791a8606761b72facbb2010","_number":946343,"virtual_id_number":946343,"owner":{"_account_id":34034,"name":"Franciszek Przewoźny","display_name":"Franciszek Przewozny","email":"przewozny.franciszek@gmail.com","username":"fprzewozn","status":"OpenStack Engineer / SRE @ Opera Software"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-04-29 22:43:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":2,"date":"2025-04-28 15:57:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":1,"date":"2025-04-28 15:57:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-04 12:47:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-04-04 14:54:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-04-04 15:15:54.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"CC"},{"updated":"2025-04-11 14:07:24.000000000","updated_by":{"_account_id":34034,"name":"Franciszek Przewoźny","display_name":"Franciszek Przewozny","email":"przewozny.franciszek@gmail.com","username":"fprzewozn","status":"OpenStack Engineer / SRE @ Opera Software"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2025-04-25 06:37:16.000000000","updated_by":{"_account_id":34034,"name":"Franciszek Przewoźny","display_name":"Franciszek Przewozny","email":"przewozny.franciszek@gmail.com","username":"fprzewozn","status":"OpenStack Engineer / SRE @ Opera Software"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2025-04-25 08:58:33.000000000","updated_by":{"_account_id":34034,"name":"Franciszek Przewoźny","display_name":"Franciszek Przewozny","email":"przewozny.franciszek@gmail.com","username":"fprzewozn","status":"OpenStack Engineer / SRE @ Opera Software"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"}],"messages":[{"id":"ff6f7b3db3071df53204172de17f3c9351ff9e89","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34034,"name":"Franciszek Przewoźny","display_name":"Franciszek Przewozny","email":"przewozny.franciszek@gmail.com","username":"fprzewozn","status":"OpenStack Engineer / SRE @ Opera Software"},"date":"2025-04-04 12:36:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5faea17b3cc3b6e495532bb6166f82f2ec88b970","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-04 12:47:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/adb6a8fcac7043faa09b19691d81005a\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a0998d45c8a94360a16f8d51397ab9fc : SUCCESS in 6m 06s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/eb6cd0a89dbf4964a1a8a1414b7aa990 : SUCCESS in 6m 11s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/149b345119c749b4b516ceae161f7f46 : SUCCESS in 8m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1a584884f8a1e038c62925a70fcd73a9d570f5f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-04 14:54:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/132a51e402e24cf192057922c9cf548e\n\n- grenade https://zuul.opendev.org/t/openstack/build/3386d5c3c16b49acb47bdf9e0d4fb30a : SUCCESS in 59m 27s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8434e3878dfe4517929ff59f9bce7827 : SUCCESS in 1h 00m 53s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c1937a357fb14841ba1418f999312a17 : SUCCESS in 1h 39m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e49ee865de4c404e8a1245af85062256 : SUCCESS in 55m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f6d24e0aee54367b06dd93553c90740 : SUCCESS in 5m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/208a2f24ca2f407c80faf2af384adfc3 : SUCCESS in 6m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/eac26de7e99a4cb68fda070c22e1f417 : SUCCESS in 5m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d61d22c7670c4ea7adb993574eaf0481 : SUCCESS in 10m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d827f59498e48a8b54be2b7f6b4708b : SUCCESS in 5m 20s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c66bda0fbb75450aa56cd4153cb33410 : FAILURE in 10m 42s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8c6164e4ba2b4482845e58a9704bc6dd : FAILURE in 8m 43s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/adbdb8b611714794906b1a9f5823c9bb : FAILURE in 10m 27s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/0eb5486dd1f1487e90954db7b0d074e0 : SUCCESS in 1h 40m 35s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/d41cb61a5b1a42eb91890ef7092554cf : SUCCESS in 2h 01m 45s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/7246b157e2894528b37ecd797daa2c5d : SUCCESS in 1h 50m 17s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/4122c92036df4f1ba033defd70d84842 : SUCCESS in 2h 07m 15s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/01f3b8718e3242aeb643a24977ab9fd2 : SUCCESS in 2h 15m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3ffb60cf64bf4f38b087ad666b339203 : SUCCESS in 52m 24s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9da123a80ad5495c8a5a82ecfe652969 : SUCCESS in 1h 41m 52s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/97430574a04f45d898a0c653eead9503 : SUCCESS in 1h 43m 50s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/06a640c6ad5649dfb9aa6c85e23d647c : SUCCESS in 46m 40s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/57408f01f54d41658eddcbcf2cb256ce : SUCCESS in 1h 33m 41s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/c2080c8fc87940838de1f3c1a995b532 : SUCCESS in 24m 35s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/070d4d9dcf984f8d83015172e20708e3 : SUCCESS in 47m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b8894152bf701a54b6c39b00b3697d82407f73fe","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-04-04 15:15:54.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"70ae881fdc7155988ffca2925d62dd9face1aba9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34034,"name":"Franciszek Przewoźny","display_name":"Franciszek Przewozny","email":"przewozny.franciszek@gmail.com","username":"fprzewozn","status":"OpenStack Engineer / SRE @ Opera Software"},"date":"2025-04-04 16:51:01.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"c2c7ef7a4857c45a38c5cafbfe75eeebba34442f","author":{"_account_id":34034,"name":"Franciszek Przewoźny","display_name":"Franciszek Przewozny","email":"przewozny.franciszek@gmail.com","username":"fprzewozn","status":"OpenStack Engineer / SRE @ Opera Software"},"date":"2025-04-04 16:52:18.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"406c2924f1256ae81955aa24c87fe25d536ab87c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-04 17:06:32.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/07935078cdac40dd93dcf999b953c3d7\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/fc6fff1c32ea4e3cb8180cc8ab6d1afe : SUCCESS in 7m 07s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/364c4eff0747402c91abaac6b1abeaeb : SUCCESS in 7m 04s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/5d96cdc02c7c47d68b8608432fec4686 : SUCCESS in 13m 26s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"83e52b855ba23d3f4bde5d6e6b8543e55fd8a93e","author":{"_account_id":34034,"name":"Franciszek Przewoźny","display_name":"Franciszek Przewozny","email":"przewozny.franciszek@gmail.com","username":"fprzewozn","status":"OpenStack Engineer / SRE @ Opera Software"},"date":"2025-04-04 18:51:31.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c1a34573462e48c15cca593b180f4402870a6609","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-04 19:03:10.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bbfeb7e90b7f4b59b1f22e7e8277a0bb\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1dccfe4f19fd4a2ea0ed72d8b7936aff : SUCCESS in 6m 30s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/18dfcf43f5d7402f9fd7cb79b9a36453 : SUCCESS in 8m 08s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/2f216b55b5794388bbf3486ccce502ec : SUCCESS in 9m 29s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"402df2511a03a6b0a37b092642decc886ebd3f59","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-04 19:13:03.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/16869288b9b24c6cbce766ca05bc02c7\n\n- grenade https://zuul.opendev.org/t/openstack/build/0194c6a1fd524f9da6645e7c25247688 : SUCCESS in 41m 08s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0f2c5675db204292afc20a05f11c893d : SUCCESS in 44m 05s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0aaab493b6904a2c9a37a54065994f64 : SUCCESS in 1h 11m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/35b7a237d2884a5cb099e4c893fa03ac : SUCCESS in 47m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45407e1a48b940d4ae6ee6a0c9947601 : SUCCESS in 4m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5c3b282dcb5942328b5b31eecb296f46 : SUCCESS in 4m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/35cf8312c8334a9797b067600b2bece2 : SUCCESS in 6m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d21e978d4c1241ce9dbdc8ab31aa5988 : SUCCESS in 8m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84bfbf85590c42b1aa9b14e63ed8921b : SUCCESS in 5m 18s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/803ba9bb0ba04cfe9e9548f3ab87398d : SUCCESS in 8m 23s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9953b735280d45278e84f54fcd623c78 : FAILURE in 7m 52s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/a1a0ed2233cd41d79077e5875ad500ea : SUCCESS in 12m 08s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/739a996e5cf34cc8bf3049e764a3fae9 : SUCCESS in 1h 56m 54s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/15eb3eb995ec4e2294f3836d52c8776a : SUCCESS in 1h 40m 14s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/5deac4cb7f234c10b7d5467a672529b5 : SUCCESS in 1h 50m 08s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/3bb778d6b90b48e5b1fcaf924bdecd47 : SUCCESS in 2h 01m 23s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/76979391b3334de882999bf3b6070e1d : SUCCESS in 2h 17m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bb1cc92825204902971b132321e7f319 : SUCCESS in 59m 34s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/069ededab1614a0c8786d2c2d6c59d79 : SUCCESS in 2h 09m 30s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/62d3b517d1424debba301fe46024793d : SUCCESS in 2h 00m 26s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/9ec3123f9e054bdf9d58b6e4877a4195 : SUCCESS in 1h 00m 31s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/7dab036004634f8aa6823a4b95b893ba : SUCCESS in 1h 57m 32s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/06953731213442db8f6976ab3dc6f9d1 : SUCCESS in 30m 21s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/7cf7cb93b27e49e69e04d18017d2fe44 : SUCCESS in 58m 21s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"95e63a6a749b9e749fa4e00f51a94b56fbeb1e01","author":{"_account_id":34034,"name":"Franciszek Przewoźny","display_name":"Franciszek Przewozny","email":"przewozny.franciszek@gmail.com","username":"fprzewozn","status":"OpenStack Engineer / SRE @ Opera Software"},"date":"2025-04-04 19:16:30.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f523f6458bc60d9d2c61e50b248c6cb8a873c6a4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-04 19:28:17.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/acca30d75ed24c829d64d3ad7aeec843\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4f4715709ed54701bf8b96b0d917f753 : SUCCESS in 6m 12s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2c51f03046da49f185901aabcd858c2e : SUCCESS in 6m 43s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/110638d63825499e9db472321c47815d : SUCCESS in 9m 54s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"72429a4150b55976016086d35ef709511634f7bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-04 21:42:44.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/455714262d414df1afa9225292603cba\n\n- grenade https://zuul.opendev.org/t/openstack/build/5fffe682b4ca44cd8b1cff8f6b9c91b6 : SUCCESS in 53m 04s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/477d826fb16e4ecfb9e9abb7f588a854 : SUCCESS in 54m 29s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7c96d9153ae24d58a366f1b11573cf6c : SUCCESS in 1h 26m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e2abda3ca90541168c419d78a3fd5539 : SUCCESS in 59m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da8e6cd2db524aea906a5e51ce8864d0 : SUCCESS in 4m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a65135507ff64478b7ccf5f7872d4c59 : SUCCESS in 4m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3f7bd7d6995540aea7e3fe06b762c7bc : SUCCESS in 4m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b932879f03f740ad91b18b16137e054f : SUCCESS in 6m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/417d5f82f0e747628f48d3a1310e7254 : SUCCESS in 5m 43s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/89b37679e060425ba421429eb98c9efb : SUCCESS in 8m 57s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/fc90e5434bd94aac9a84f6ebfe0623ee : SUCCESS in 7m 28s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/3838957437364673a3070494f6dfb286 : FAILURE in 7m 19s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/1a052edd645349289c6abec13edbde69 : SUCCESS in 1h 56m 12s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/3373250528ce468481477cf196f0baa4 : SUCCESS in 1h 39m 42s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/75e72a1f2d6d4a1f8ee33b49c597c634 : SUCCESS in 1h 12m 43s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/2900f4ed0e334d5cb28751227fd0ca14 : SUCCESS in 1h 39m 21s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/ebf6992d972a433389f63dc53d5da81e : SUCCESS in 2h 13m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/78e18f49ff354d86ba932e55929190d3 : SUCCESS in 42m 12s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/502f74b0d42940bca17900b053359cb9 : SUCCESS in 1h 42m 17s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/21181cb96af944d5919f34618552077a : SUCCESS in 2h 23m 39s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/29e1b1a17aef4207983b14294da796c4 : SUCCESS in 1h 02m 38s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/98b5a9857c964052a2c8331d3fce46e1 : SUCCESS in 2h 00m 05s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/30bd6d3e2828413494cc0cb349a276c7 : SUCCESS in 29m 44s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/dded7544cb9c423e800bf3bb03b6d997 : SUCCESS in 1h 13m 06s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9cc5cd30bf31864f393a54530d25e362601b9e80","author":{"_account_id":34034,"name":"Franciszek Przewoźny","display_name":"Franciszek Przewozny","email":"przewozny.franciszek@gmail.com","username":"fprzewozn","status":"OpenStack Engineer / SRE @ Opera Software"},"date":"2025-04-05 04:41:38.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6b696909e9afac21875de020daa6fb7d68b58407","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-05 04:51:49.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d88e9cb9116f469da13f58d323aa91e3\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/853fd3f022b94c528509992105afff86 : SUCCESS in 6m 16s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d0de15e34cc74158a0c474add50f55a1 : SUCCESS in 6m 58s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/30941c38eb474389a238a9e3def5cf14 : SUCCESS in 8m 19s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c7fada48dfa146c97e45f50224b025eb40316d5b","author":{"_account_id":34034,"name":"Franciszek Przewoźny","display_name":"Franciszek Przewozny","email":"przewozny.franciszek@gmail.com","username":"fprzewozn","status":"OpenStack Engineer / SRE @ Opera Software"},"date":"2025-04-05 04:59:54.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b6056f4c39d6544a01af1d9497d09a8fdf15308b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-05 07:02:48.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/936aa5666c10406ca9eed404affee161\n\n- grenade https://zuul.opendev.org/t/openstack/build/287bf5b569cb4470b0a3011761b0f45e : SUCCESS in 55m 17s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/fc867d406cf642b2a7fa5b144fcadca1 : SUCCESS in 41m 49s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1a2c58fc85694023a2f4d53d199b19de : SUCCESS in 1h 11m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3a9e86eed4c14280b095478e7809a1e9 : SUCCESS in 45m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb32ebda407241ec93f8e83aad137d77 : SUCCESS in 3m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9f6d7b4603a44997b3e557de8650858e : SUCCESS in 3m 28s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/adb79b7bd6594b30a7181635af7a6555 : SUCCESS in 3m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1da46ee3c0ef47a9b6950ebc8edee642 : SUCCESS in 6m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c8f0de85e2b404d9fe8a387e9995b84 : SUCCESS in 6m 42s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/67cf82c8bca74923998b8f415b28d2fb : SUCCESS in 7m 47s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/edabeebb4996422bb1e44c306e5b2aa3 : FAILURE in 5m 48s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/db5b6f21628c4cc8987396e3744a5523 : SUCCESS in 11m 49s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/a9be60de4a9348fbbe287e8b34c3e893 : SUCCESS in 2h 07m 02s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/57daec097b1c4f7eb8bb94aa31b12463 : SUCCESS in 1h 24m 32s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/6d15bf4a910049668faa10db645db452 : SUCCESS in 1h 04m 29s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/34c1dda880b84d1da16aa02d76fe3699 : SUCCESS in 2h 01m 11s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/ae0eae603cad40c7913fc089aa3d20d1 : SUCCESS in 2h 14m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/207e6716041b400787397cfb93dea204 : SUCCESS in 37m 33s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/68e73fe94b4244c8be94856464359d61 : SUCCESS in 1h 51m 52s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/ce953e158bcd471a9725e7bbbac84c84 : SUCCESS in 1h 51m 41s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/10124bc25136493d908bd5f84258e60e : SUCCESS in 1h 02m 10s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/17c7e8e73d944794bc29a1ad6c85c47e : SUCCESS in 1h 00m 24s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/d9b87bf712c04a41b51b545331a9eebc : SUCCESS in 19m 31s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/1257ca82ea944da29008be58e36f5a9f : SUCCESS in 57m 11s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ea65eb0b050ee22b474a25dc5dbf016649c215f8","author":{"_account_id":34034,"name":"Franciszek Przewoźny","display_name":"Franciszek Przewozny","email":"przewozny.franciszek@gmail.com","username":"fprzewozn","status":"OpenStack Engineer / SRE @ Opera Software"},"date":"2025-04-05 07:06:51.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0367956ef036d5721256e27c06fecb998617aee6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-05 07:17:08.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e7eb0eb789634d18a808578282ab91c8\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/394f5ce1f82e443b8755f6a6e065384c : SUCCESS in 7m 31s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a02709d570c34bd7ada3ffdd79c500d4 : SUCCESS in 7m 01s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/65beeb37ed8a4d04b450013c6bdf7b50 : SUCCESS in 8m 20s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ee5f4740d134b219951a09245c641c793a2548f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-05 09:19:30.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/89d4c497a04d4956b0698c707a337d5b\n\n- grenade https://zuul.opendev.org/t/openstack/build/7594989b6e394e08bea5e117cac30b38 : SUCCESS in 42m 33s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d669a8eef9b04ca5b846f20ec0247eeb : SUCCESS in 38m 49s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0eee106aefdd474492d76b5a3e5aabd4 : SUCCESS in 1h 49m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/db21380c8a65403792d29102ccb994f3 : SUCCESS in 38m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb22e07b3145467d8d12b1b4f999cd1c : SUCCESS in 3m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/06664a85718f407b983da3dfc66c5b25 : SUCCESS in 4m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a013582bf3c74ff2b5b2fac2991ef06a : SUCCESS in 4m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c6d62a9f93fe4c97b9145a10d9b3e116 : SUCCESS in 5m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8145e002c564406789453b2cbaa7638a : SUCCESS in 5m 12s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0fff74316c23444d8c6f2d4a18b8cd43 : SUCCESS in 7m 27s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4539ec82330e4fcba16cdb9d079e9567 : SUCCESS in 8m 15s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/0da8aeb3761741b4b9533bc30685bee4 : SUCCESS in 7m 18s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/d7382dd51042496c89d6868dd9f4fdb0 : SUCCESS in 2h 00m 56s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/2ebd9fb72a404084804db9e9c71ed7d9 : SUCCESS in 1h 14m 39s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/32038c9bb5544963afb647714d3fd005 : SUCCESS in 1h 59m 50s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/73af3ee6a4054eedad8e36efe66337d2 : SUCCESS in 1h 13m 10s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/d1b94d21c08a4d518b6175b4368a850c : SUCCESS in 1h 16m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3254e2bbf29c486ca69f4cca89617226 : SUCCESS in 27m 28s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/407cfad8166e41cbac8ef981dc67fa72 : SUCCESS in 1h 58m 51s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/437d5662249342e4a8e4c005632750f8 : SUCCESS in 1h 33m 17s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/484ee3162c4642ee951d3c5697565b87 : SUCCESS in 58m 16s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/f5b7c6f275de48ed959bf750f4695c6d : SUCCESS in 1h 02m 12s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/ec78c046e2dd4a93bb60c604545e0035 : FAILURE in 21m 27s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/3567135fafe64f3d9f8bec9f6709f7d2 : SUCCESS in 42m 44s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ca8d337093df5e69500debb2679b9ad2997bb043","author":{"_account_id":34034,"name":"Franciszek Przewoźny","display_name":"Franciszek Przewozny","email":"przewozny.franciszek@gmail.com","username":"fprzewozn","status":"OpenStack Engineer / SRE @ Opera Software"},"date":"2025-04-05 10:51:47.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"69d580e9b8ea52da73557e08b1e14b50af006438","author":{"_account_id":34034,"name":"Franciszek Przewoźny","display_name":"Franciszek Przewozny","email":"przewozny.franciszek@gmail.com","username":"fprzewozn","status":"OpenStack Engineer / SRE @ Opera Software"},"date":"2025-04-06 11:50:39.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"936192d078629daeb6a4ddc7d9d593fa83fbcfd2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-06 12:04:10.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/59907c342ab4434ba1888d6fc1a9c1ff\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8f64382fd0e14520a7d2c7bf709b8bee : SUCCESS in 7m 12s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e1840d89b5d642b084e4051b48896e16 : SUCCESS in 8m 01s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/dd7f581054564e18b4869e52680cdaa2 : SUCCESS in 11m 29s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"cd5b20776048a4f0c3ba9f2e9fa9f705d4f796fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-06 13:51:51.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0ab0be0a4b844f49a3cf0adee6b78cfc\n\n- grenade https://zuul.opendev.org/t/openstack/build/1e504ae8f6f6425b900feef03f11b5a5 : SUCCESS in 27m 35s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7508f4f1b2e540789e9a4f3f3f35668b : SUCCESS in 26m 22s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/55e3bc782cc54650a9509c253ea20ed6 : SUCCESS in 1h 30m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cae3d16a124f42c09aa764b9f2909bd0 : SUCCESS in 31m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dacd65fc9e9e48dfb645c8a6a8607f63 : SUCCESS in 2m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ff0de41fc26a4acdba1a8d26d82f907f : SUCCESS in 2m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4888040f19a541719593cc1f57c63acf : SUCCESS in 2m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0f4f742b86c54f9993a9052a5231153b : SUCCESS in 4m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f76d230523e4ac0a9c2e2a02927ce9a : SUCCESS in 5m 41s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2400a7ddbad74a38a6cf16bd625bff16 : FAILURE in 5m 43s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/461ffbbf78c34f4db213cb3eb68e3f8b : SUCCESS in 8m 59s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/cef0871041d74a7b904ba8b17c87f053 : SUCCESS in 9m 46s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/1422e869e1584d3f9dd6eb157d7a1bf8 : SUCCESS in 1h 00m 06s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/f28d1019122b4b6293650fed9362812d : SUCCESS in 1h 44m 44s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/f348e49cf63841eda2c51a5d9a5559b4 : SUCCESS in 1h 31m 43s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/bc5b3a4b3c3e4d54940a21ba7aa1b7e7 : SUCCESS in 1h 58m 27s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/2f11963e99614ec2b6a87607dda533a5 : FAILURE in 2h 00m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7fe5d2a94b244cc2a7c7ab77bbdda8c3 : SUCCESS in 48m 16s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/aed86b2a67bb4f18bfd4e689d5e51631 : SUCCESS in 1h 36m 23s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/1d2136440cf04c5e950cddb10be1f52e : SUCCESS in 1h 57m 39s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/f2c293bac61b491d8c149f78dd8ebf57 : SUCCESS in 53m 57s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/50dec5ae1d48447f80b1e26e83295476 : SUCCESS in 1h 38m 26s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/d8292bf7245740be824e6d35aec3ff79 : SUCCESS in 20m 57s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/bdd9fad337d94d74aeb88d7dfa7a3ae0 : SUCCESS in 50m 39s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f6e32a51fbd8844cc391611b579bb4ff81c5749a","author":{"_account_id":34034,"name":"Franciszek Przewoźny","display_name":"Franciszek Przewozny","email":"przewozny.franciszek@gmail.com","username":"fprzewozn","status":"OpenStack Engineer / SRE @ Opera Software"},"date":"2025-04-25 06:40:09.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2f0460c9f02de526f3f4fbc26bef70812f917cb7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-25 06:55:07.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9c717045ce1f401185829358fee874f1\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b99ad361b8414befa8d8b5661e28b445 : SUCCESS in 7m 18s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5f11c2d6882744d88be5c0bb8dd9a6c7 : SUCCESS in 9m 24s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/9e2e0e2e91044d628f86f22efcc427d7 : SUCCESS in 13m 21s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d642e59aef45299b2bbe2e256203f0ddd8601577","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-25 08:56:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/92db65511bd946b3816c9cfbd050d0e0\n\n- grenade https://zuul.opendev.org/t/openstack/build/e2e228e13ec547b8abf24fd421f9f5df : SUCCESS in 52m 07s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6e408392eb914f4aa585c3dacb7c4a92 : SUCCESS in 1h 08m 24s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/32a67693611e416fa206fe12db5e5940 : SUCCESS in 1h 23m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/217f21acb87145bab5f7f12fa21d3b52 : SUCCESS in 59m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe2f978c1cfd40418ab4bce6949f6ef0 : SUCCESS in 2m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/91a12abd78dc4ee59bee19e6ac9b40bc : SUCCESS in 4m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f447e4a8a89e445fad2404f65e88d1d8 : SUCCESS in 5m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ce1512ea9c3f4bdfb16862911eb0f479 : SUCCESS in 7m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83253af1f91d4749b28d6f2818bf3ff1 : SUCCESS in 6m 10s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4cc07f7e4422459a84c79f69bc7fbf50 : SUCCESS in 8m 14s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e3fcb4e3dd81495ca909d97605cf027f : SUCCESS in 6m 47s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/32ee99b5292646bd9f31477099f02c26 : SUCCESS in 12m 39s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/da81a79e341b4c09b1f949728c7a8e6b : FAILURE in 1h 49m 17s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/b981021ab1374090914cf3cfdcc16a07 : SUCCESS in 1h 54m 18s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/9708606eacd444d995bfdea4a03248e3 : SUCCESS in 2h 10m 49s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/592efe070fa4434ba8331b6a57183b53 : SUCCESS in 1h 37m 17s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/c3fcc20fc6f748d79d1df87b1c654217 : SUCCESS in 2h 01m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9c4a536554504db385a4df653bc46969 : SUCCESS in 38m 47s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/14ff3860df6e42aba862374aa019b981 : SUCCESS in 1h 30m 24s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/df2f97bf393d4075ac98c2353342ce26 : SUCCESS in 2h 01m 50s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/54646c901380458497aa8b4ca4326c80 : SUCCESS in 53m 48s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/0a1ae8fc020c4bb0b5bda8a955f35e68 : SUCCESS in 1h 23m 31s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/8de98aeda7ef47cdb1a9787d0b24bf05 : SUCCESS in 29m 27s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/14b2c52dd76f4f3ba0d5ecd761f9d5a3 : SUCCESS in 59m 49s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"fdd5e4668e55170dae5fa309891b5df380b43888","author":{"_account_id":34034,"name":"Franciszek Przewoźny","display_name":"Franciszek Przewozny","email":"przewozny.franciszek@gmail.com","username":"fprzewozn","status":"OpenStack Engineer / SRE @ Opera Software"},"date":"2025-04-25 08:57:29.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9e559e686cdc28db82f40743be675a11789ab8fc","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-04-28 15:57:02.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"51979e66bd33a532e90e713a4b38f9b0345d5311","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-28 15:57:27.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e4da70a53d6c9110a413c5158afc456d8dade72f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-28 18:10:35.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b57aac83faf34860b86393e1de393ecf\n\n- grenade https://zuul.opendev.org/t/openstack/build/f6b762624b6d4b988a5ae84153b83098 : SUCCESS in 49m 53s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8b9d233bc48b42da8e123b84fe472a0d : SUCCESS in 51m 44s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b3a5e49bfe064063af0d932d539a69e1 : SUCCESS in 1h 16m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a867b3fbdb894235bf127e9e65565bd7 : SUCCESS in 56m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa408ec96e624fa88818398625f50b12 : SUCCESS in 2m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2d4317b1ebe84430a1f08103c1c9a9db : SUCCESS in 4m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/120512b1515b4359b3ae72c25e516be2 : SUCCESS in 2m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7b91275cde643eaa47686a555d0e33e : SUCCESS in 5m 32s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c28511d0db064aabbe6c78ecc323977a : FAILURE in 3m 51s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/632c93b8424b4bc0abe7b70390f887e6 : SUCCESS in 7m 30s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/c3397d26e7cb4fd7aeb94960c036be9f : SUCCESS in 7m 05s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/40098ecdc0b74ce78bfd704d9c065864 : SUCCESS in 1h 36m 41s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/bef2531822664ef685ab4c619c2fba27 : SUCCESS in 2h 03m 00s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/2722a8dea38e451b9de57d7225ce18cf : SUCCESS in 2h 08m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/17dc9f156cdb49a897b5e43a673f3e57 : SUCCESS in 55m 58s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d5f1c44b26744b1eb14b8e1215bcfb2e : SUCCESS in 1h 31m 12s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/788aeac738ce433e9d6f9c5596026832 : SUCCESS in 32m 26s","accounts_in_message":[],"_revision_number":2},{"id":"d54f6999ebbca903680c2bd62a1d92acf903d9aa","author":{"_account_id":34034,"name":"Franciszek Przewoźny","display_name":"Franciszek Przewozny","email":"przewozny.franciszek@gmail.com","username":"fprzewozn","status":"OpenStack Engineer / SRE @ Opera Software"},"date":"2025-04-28 18:11:25.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ab2b610cc5ab19966848ae069444654be55ce723","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-28 18:35:18.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5759b51651de4cbf8c2af73d2f7d476f\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7f48caad86b04ef1a9b9e4913abfe524 : SUCCESS in 5m 53s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8bc144de2d0f40ebb58f2e2a0cf349cd : SUCCESS in 7m 07s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/73a952fc9ae5487da53d702a41e39531 : SUCCESS in 12m 21s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5e359b39ba4da5e63c1e11b4d3be062d0834a98a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-28 20:31:33.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0897eb762f574fd5b33ff67b806e7c4e\n\n- grenade https://zuul.opendev.org/t/openstack/build/97b57f1d3a634a87925e281300bf1b52 : SUCCESS in 55m 21s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f703136575be4a6e8c85a9cd328eea97 : SUCCESS in 53m 29s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9a83ebe33477495789019e76897607e9 : SUCCESS in 1h 18m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ccab3611ca8c446ea644025349ece784 : SUCCESS in 51m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d1c7188d5c445959ea20593122b010c : SUCCESS in 4m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5f026d44816b40bdb9cd68608f9ada50 : SUCCESS in 4m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5b6df69c820c4ca98eb7142d67a0a62e : SUCCESS in 5m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2216e646d10f49e7a5b33493b21d3578 : SUCCESS in 4m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4eecfb3b1c624e399d557db937183cea : SUCCESS in 4m 45s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/31b32483b65e4b3b8b2d772b373b7d6d : FAILURE in 3m 45s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e8e37faad7ca43249b62a534679acf2a : SUCCESS in 7m 16s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/f1e596bb2efa47bb8e9564e8f354a596 : SUCCESS in 9m 50s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/c4aac1ba524549aba5efff5d642afd9d : FAILURE in 1h 59m 24s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/eb6a8179dbb640da8561a5dedba5ffb1 : SUCCESS in 1h 58m 09s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/eb6fbe04bd9749d186f64eef00c60a26 : SUCCESS in 2h 12m 34s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/79216e323ac246b68d1caddba88f620f : SUCCESS in 2h 02m 53s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/41988b4fa4d7403e8c0b9bfc85961ae2 : SUCCESS in 2h 15m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f0d0f183c22346bfaa19857dedf04d1d : SUCCESS in 55m 57s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/4ab9118927864652aef59da0e378bd9f : SUCCESS in 1h 43m 58s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/4de70cde466148ee8640a2f551cd127b : SUCCESS in 1h 32m 03s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/79e4e89c1daa47dbaff0e3e4bda35cc3 : SUCCESS in 51m 37s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/f77cdd9a5893462fa2987a6536599ff3 : SUCCESS in 1h 30m 07s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/c2d69abe818a43a198a385e2de52e6e1 : SUCCESS in 14m 45s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/fa3815236eb94dfd83b574de8b3e78e3 : SUCCESS in 1h 01m 26s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6e0a3251e26d4e470b727fc201a218a4be21b30f","author":{"_account_id":34034,"name":"Franciszek Przewoźny","display_name":"Franciszek Przewozny","email":"przewozny.franciszek@gmail.com","username":"fprzewozn","status":"OpenStack Engineer / SRE @ Opera Software"},"date":"2025-04-28 20:35:10.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9bd2fc5ef512239a817d4f541df44731366580c9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-28 20:46:04.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2d939c792df14ee083639169a68d8570\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/299cf0e4de9246f0aecd3b15e26a4032 : SUCCESS in 7m 10s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/bb1ada05383a4507a287339cfa33dabc : SUCCESS in 8m 49s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e406a2f70756445e88f84242019aa867 : SUCCESS in 8m 34s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"bd6e0898c5b5212742200824365f895332c85473","author":{"_account_id":34034,"name":"Franciszek Przewoźny","display_name":"Franciszek Przewozny","email":"przewozny.franciszek@gmail.com","username":"fprzewozn","status":"OpenStack Engineer / SRE @ Opera Software"},"date":"2025-04-28 21:32:04.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bdd38f201ad42cd5d3d530f625ce51b5951196f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-28 22:52:51.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1dc523be61c04ed394e8aa4e8d73cf6c\n\n- grenade https://zuul.opendev.org/t/openstack/build/c0f574139f50449ca92e9322ad5288be : SUCCESS in 1h 02m 25s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/890e6525448540278a7b3e8873a4543f : SUCCESS in 52m 59s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d9c97889a14742eb8818250554d09e5d : SUCCESS in 1h 43m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/66865b235a6a4615b6665c69996390b3 : SUCCESS in 1h 03m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc2ce7c998c34eddb218470b1f902f56 : SUCCESS in 4m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/be654c9388b74724b830761a66efe6f5 : SUCCESS in 5m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/91fcf790f2e1440dbdc660d1bb58ea08 : SUCCESS in 6m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/669d9495f27f491a9d09b332223907da : SUCCESS in 7m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7f35b25d8d842518299a999bf9e982b : SUCCESS in 4m 33s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/756cc8b336c94424b7aecb321bc8c4f6 : FAILURE in 3m 02s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/efc639fa8a574deca6c02e0827efdbef : SUCCESS in 8m 22s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/b36aed4d4efa4c32a86d50e9ed0a23d9 : SUCCESS in 12m 19s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/59e389ece4a54f7ea494eca2e4db1e85 : FAILURE in 1h 42m 44s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/5c6e708784e045e09b1c78f92b404cc6 : SUCCESS in 1h 32m 51s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/77c42b806ca44caabbc1c873af4b1709 : FAILURE in 1h 46m 43s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/0844cce8e89849f38bdfd0e11838dca8 : SUCCESS in 2h 11m 14s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/1a67f96fce1942b88d4b9e229cdb0936 : SUCCESS in 1h 43m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/67ee22f14f0343e5ac3bf231c5f7ba37 : SUCCESS in 54m 56s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/06f4e538e9de408e87429325e024d26f : FAILURE in 1h 33m 18s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/dc6390466e0c47fcac62921c253ec226 : SUCCESS in 1h 39m 46s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/f62efe634c9b465692e95fb2ee4582cf : SUCCESS in 45m 22s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/2fe770d9929b4ec2afb1e04a9133444a : SUCCESS in 1h 27m 33s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/99023001bde64b92aa9db9c13a8c4c1b : SUCCESS in 30m 53s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/0441070cf5b9434eb2f6743b9998f4ad : SUCCESS in 46m 43s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3f66a6358922552596bd0f42da092377deb934d7","author":{"_account_id":34034,"name":"Franciszek Przewoźny","display_name":"Franciszek Przewozny","email":"przewozny.franciszek@gmail.com","username":"fprzewozn","status":"OpenStack Engineer / SRE @ Opera Software"},"date":"2025-04-29 18:18:23.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bed104fd001a6d763101d169e2e63ce822e6cbc9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-29 18:32:59.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0b6fe86f04854ab48efed050a51b6edb\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/44c1aeb803b24a97b57bd6218a7624b7 : SUCCESS in 7m 45s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b0de961ef60c4e6a9a6ea3ba86925ba1 : SUCCESS in 9m 07s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/481d5f80126444e687d3dc0f9e21214b : SUCCESS in 12m 50s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8b93d74bed4b2bac96e985fcf23f63007e04023a","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-04-29 19:28:24.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"1ffa8f95b982ba611bae7cf112f40025e37e0a26","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-29 19:40:30.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fce0ffaec69f4be29613cb01164028b3\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a83d6153ba3c40209f1b2b7066c25d8d : SUCCESS in 7m 54s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d2290906b8b84ab8a66a2482d9c6fb88 : SUCCESS in 7m 18s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/80328d43ed32478d867f743a06df239d : SUCCESS in 10m 32s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c84e58d6bb5173fc27068e27caaec3ec5f097d96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-29 20:43:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4fe944ce1de447318103ed150808562d\n\n- grenade https://zuul.opendev.org/t/openstack/build/eaa2c683ecbb4367aef67cd4884e8886 : SUCCESS in 49m 53s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/335b0a9c85d5479f8ed286fd599d7768 : SUCCESS in 54m 46s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2141053714374f6ca2adb15935d6c2b3 : SUCCESS in 1h 19m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0263468c84544149bc6dea39c4318fca : SUCCESS in 31m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db221c8904c84ea3a1a88e7ba5e50c11 : SUCCESS in 2m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d3600a5f73c04d7e9769aad882e5fa90 : SUCCESS in 4m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f56f9005e1a04c6da88fc2e680414ab0 : SUCCESS in 2m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/07ce3e03920940d5bb953476a408b297 : SUCCESS in 4m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52744605ff4746f89cc8aeadd763c28a : SUCCESS in 5m 31s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b84d889d08844d4780c277587f35a718 : SUCCESS in 7m 11s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/f20d4e9bf3f840d99cdf1dadc151097b : SUCCESS in 11m 38s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/5cb490ae7e1345ee8a9c4a049dc10085 : FAILURE in 1h 49m 23s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/e43e74bc682940898106e28f093c737c : SUCCESS in 1h 43m 44s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/216d135108aa41ada21cea5bc2f48147 : SUCCESS in 1h 58m 54s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/029ee8291d86487bae79bb9e50d8541c : SUCCESS in 1h 56m 43s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/dc9329e3242542cb85f76b47a023aebb : SUCCESS in 2h 24m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/de45ed3a98b04e729f98fb30f31952f1 : SUCCESS in 58m 47s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/5e874e4a12a0416aac1f3cbd129ae3b4 : SUCCESS in 1h 51m 48s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/28ab4a41ff1546939512f9cf31c9b67d : SUCCESS in 1h 35m 20s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/d80185b28f804df4a35120e4f96dd7eb : SUCCESS in 45m 20s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/457010c1440942d09fcde29a6a07cf3d : SUCCESS in 1h 49m 07s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/8ef1101274f54060b7e8bef65cfab54d : SUCCESS in 30m 13s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/0e66c6194a8b471e860e856ba2fd434c : SUCCESS in 48m 50s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3794dec7306a339e52596f808f6d5aa8e7103796","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-29 20:44:12.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"204fbfdaf2c85ba132008270ec919ece6bb9fab6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-29 22:43:56.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/55890e739bd34470a586a18e8e9c1199\n\n- grenade https://zuul.opendev.org/t/openstack/build/1f7ad30884af48a0b9f0bcb4a1475b16 : SUCCESS in 46m 24s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a9f71a40e3fe4436be03a1929f86bcda : SUCCESS in 47m 18s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/dbbb4af5009e4056a7fac6a45eefb5d8 : SUCCESS in 1h 18m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0dfde101beb140219d4ae14027edb331 : SUCCESS in 52m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3dc519328764f7bbf8a169239e507d6 : SUCCESS in 4m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2cba890e0a0d4d32869af31b011413f9 : SUCCESS in 4m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d187b1d0fbaf4d2fb3be5b03766aa621 : SUCCESS in 2m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77155220b2bf42c6a090926ebdef3517 : SUCCESS in 4m 43s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e8c8c4d1d6a5421abd8d5488b9cf4119 : SUCCESS in 7m 42s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/6e2723b2682a45b3ae7458645f75b31e : SUCCESS in 7m 39s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/c8a6da88779649978fa116ade1b82e8b : SUCCESS in 1h 30m 35s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/a7e9dfe2a87b425c805d6136ec485309 : SUCCESS in 1h 38m 24s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/2abff04ab0bd4a77b7de3e0274a1e2f5 : SUCCESS in 1h 52m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e44b954323864feeb274226c380b07b7 : SUCCESS in 58m 48s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/73e8ac885e2442c59ba1f997f8fdb712 : SUCCESS in 1h 53m 11s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/fc44da78a60d489a9e839698916b786d : SUCCESS in 25m 33s","accounts_in_message":[],"_revision_number":2},{"id":"7512de6f65a71945b5c936facfdb0d877ee239ec","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-29 22:43:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"e580345972498cc43791a8606761b72facbb2010","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-29 22:44:59.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/848c2bb55b53421482fdebd9e7dad922\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eceb229471b74a1f8bcaf885f41f4f5c : SUCCESS in 44s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"df5f2b3c8693ce288209e86bee656e7f9c7b9566","revisions":{"cc4bd87c35d8668fef658ada5bf61815bfb17907":{"kind":"REWORK","_number":1,"created":"2025-04-04 12:36:52.000000000","uploader":{"_account_id":34034,"name":"Franciszek Przewoźny","display_name":"Franciszek Przewozny","email":"przewozny.franciszek@gmail.com","username":"fprzewozn","status":"OpenStack Engineer / SRE @ Opera Software"},"ref":"refs/changes/43/946343/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/43/946343/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/43/946343/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/43/946343/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/43/946343/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/43/946343/1"}}},"commit":{"parents":[{"commit":"7dbb31949e7454c680ec3c1fdad06bd90b48ac9f","subject":"Merge \"Use pyenv for the periodic jobs running on Python 3.13\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7dbb31949e7454c680ec3c1fdad06bd90b48ac9f"}]}],"author":{"name":"Franciszek Przewozny","email":"przewozny.franciszek@gmail.com","date":"2025-04-04 12:36:36.000000000","tz":120},"committer":{"name":"Franciszek Przewozny","email":"przewozny.franciszek@gmail.com","date":"2025-04-04 12:36:36.000000000","tz":120},"subject":"Fix vTPM metadefs.","message":"Fix vTPM metadefs.\n\nAccording to the Nova documentation[1], valid tpm_model properties\nare: tpm-tis and tpm-crb, not TIS and CRB. This patch fixes the\nmisalignment and also adds missing tpm_version property.\n\n[1]https://docs.openstack.org/nova/latest/admin/emulated-tpm.html\n\nCloses-Bug: #2106223\nChange-Id: I698dbd2946555c0a6455b1be17a6bf4a8d407ee7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/cc4bd87c35d8668fef658ada5bf61815bfb17907"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/cc4bd87c35d8668fef658ada5bf61815bfb17907"}]},"branch":"refs/heads/master"},"df5f2b3c8693ce288209e86bee656e7f9c7b9566":{"kind":"REWORK","_number":2,"created":"2025-04-04 16:51:01.000000000","uploader":{"_account_id":34034,"name":"Franciszek Przewoźny","display_name":"Franciszek Przewozny","email":"przewozny.franciszek@gmail.com","username":"fprzewozn","status":"OpenStack Engineer / SRE @ Opera Software"},"ref":"refs/changes/43/946343/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/43/946343/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/43/946343/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/43/946343/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/43/946343/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/43/946343/2"}}},"commit":{"parents":[{"commit":"7dbb31949e7454c680ec3c1fdad06bd90b48ac9f","subject":"Merge \"Use pyenv for the periodic jobs running on Python 3.13\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7dbb31949e7454c680ec3c1fdad06bd90b48ac9f"}]}],"author":{"name":"Franciszek Przewozny","email":"przewozny.franciszek@gmail.com","date":"2025-04-04 12:36:36.000000000","tz":120},"committer":{"name":"Franciszek Przewozny","email":"przewozny.franciszek@gmail.com","date":"2025-04-04 16:50:38.000000000","tz":120},"subject":"Fix vTPM metadefs.","message":"Fix vTPM metadefs.\n\nAccording to the Nova documentation[1], valid tpm_model properties\nare: tpm-tis and tpm-crb, not TIS and CRB. This patch fixes the\nmisalignment and also adds missing tpm_version property.\n\n[1]https://docs.openstack.org/nova/latest/admin/emulated-tpm.html\n\nCloses-Bug: #2106223\nChange-Id: I698dbd2946555c0a6455b1be17a6bf4a8d407ee7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/df5f2b3c8693ce288209e86bee656e7f9c7b9566"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/df5f2b3c8693ce288209e86bee656e7f9c7b9566"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
