)]}'
{"id":"openstack%2Fglance~947637","triplet_id":"openstack%2Fglance~master~Ia6931b4cf81c0fb3acbbae914e0c00ba364dc8e2","project":"openstack/glance","branch":"master","topic":"eventlet-removal-scrubber","attention_set":{},"removed_from_attention_set":{"33765":{"account":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"last_update":"2025-05-15 12:46:02.000000000","reason":"Change was submitted"},"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2025-05-15 08:12:23.000000000","reason":"\u003cGERRIT_ACCOUNT_9303\u003e replied on the change","reason_account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},"25402":{"account":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"last_update":"2025-05-09 08:23:49.000000000","reason":"\u003cGERRIT_ACCOUNT_25402\u003e replied on the change","reason_account":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"}},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2025-04-28 14:33:27.000000000","reason":"\u003cGERRIT_ACCOUNT_8122\u003e replied on the change","reason_account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}}},"hashtags":[],"change_id":"Ia6931b4cf81c0fb3acbbae914e0c00ba364dc8e2","subject":"Refactor scrubber to remove use of eventlet","status":"MERGED","created":"2025-04-17 20:27:06.000000000","updated":"2025-05-15 12:47:10.000000000","submitted":"2025-05-15 12:46:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":15,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"947637-eventlet-removal-scrubber","meta_rev_id":"9856db2ea01168c78f4d26335ce6bdab7579b750","_number":947637,"virtual_id_number":947637,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-15 12:46:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":2,"date":"2025-04-28 14:33:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":1,"date":"2025-05-09 08:23:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":1,"date":"2025-05-14 15:24:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"}],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-17 20:27:45.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"state":"REVIEWER"},{"updated":"2025-04-17 20:27:45.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"state":"REVIEWER"},{"updated":"2025-04-17 20:27:45.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2025-04-17 20:35:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-04-17 22:38:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5ebcbe7c7933d659639ea9ea4ea5402c66a41e14","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-04-17 20:27:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dae3609b796cdffd3eed8151b9b18dc95280f5af","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-17 20:35:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ae207c8711d4d4c98b5cb241f48de52\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/972db168b04044deafa68f9d1a4918c7 : SUCCESS in 5m 48s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7eabf1018fc243d38d5ae9ff442d2865 : SUCCESS in 7m 04s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/cf5c0fb39381439eaf1257f599709cc3 : SUCCESS in 7m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9daee867c5af0761045345e1a84865fc0e640790","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-17 22:38:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f868a28e7087460e8328f39aced9f587\n\n- grenade https://zuul.opendev.org/t/openstack/build/1460fc25e9884f6da8e0a7e55934ad4c : SUCCESS in 1h 01m 40s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/14462d26acd74c22a2889bbcffdd1386 : SUCCESS in 1h 00m 16s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a0bdbb7f725a46d49c9de836511d3adc : SUCCESS in 1h 23m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/74972b339da04c52b33731312f3f85c4 : SUCCESS in 57m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/912d8530f3b64de89cb947e798ec2b89 : SUCCESS in 2m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f8723b037a034c8d88873fcf3ad7debf : SUCCESS in 4m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/de785fe934544fa985e46f09dbc0027f : SUCCESS in 5m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b9260dd0523d481fbbc77c967eb5d319 : SUCCESS in 4m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99996b4f1b264d92adab82eba7978967 : SUCCESS in 5m 15s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/17dce864107d421c89b8e047940a32d3 : SUCCESS in 7m 03s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/abae42cad5b24d26ad09321d5bccceec : SUCCESS in 6m 58s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/ca0a04efa09448b892ca41a5427310e4 : SUCCESS in 9m 35s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/6b15b142b8284bf2b6e823f7af48da00 : SUCCESS in 1h 22m 08s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/9b2590198cc241aa907e1229bd97d3a4 : SUCCESS in 2h 09m 23s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/5839109bce094099a8efa7708f8ea6c4 : SUCCESS in 2h 05m 45s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/5852468efa444fd2be8237f575f112d4 : SUCCESS in 1h 03m 54s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/fc8f7ebd63d9482e8647545ee53334c8 : SUCCESS in 2h 11m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7c3ef0084bb64156ab2fcfd02a8020f3 : SUCCESS in 40m 48s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/73610f87b17146c0a81d078397ebc48d : SUCCESS in 1h 35m 19s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/2f25df945570455e82442d14f4ee0c79 : SUCCESS in 1h 40m 08s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/dd4f6762e8a947248519e4a94871d247 : SUCCESS in 45m 58s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/84f824e6e09a4729afccac8402cbdfc6 : SUCCESS in 1h 11m 48s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/b97f83d6dedf4cefac0f05d4fe9d4631 : SUCCESS in 31m 08s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/d3f0720dfa6b455c92fdadaa8aac7af2 : SUCCESS in 46m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"72ac7e6cd91ef7bb19cc5672fa1e9666f996af2d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-04-21 08:36:07.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"263692c8fac87b072689495aac60607bde5d10c0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-04-21 08:45:10.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"dc94a95acf73817839177b8b6ef4dbc24a469cbc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-21 08:53:54.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f8e3e02c098489abb0906d66698777f\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/18197bd53d1141a9ad61695e7ffb8896 : SUCCESS in 5m 05s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/012d4496fa874651a8131ba3e29ec1d0 : SUCCESS in 5m 12s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4c075a151e5947199c92dd93f112dd26 : SUCCESS in 7m 00s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"16f25b457133ac1f9548b12e13d7cc24e9ca747a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-21 11:07:31.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d4024dd3fae41c283fa1ee3d304d00f\n\n- grenade https://zuul.opendev.org/t/openstack/build/888882a1ca6f4898b985e2c123a3b701 : SUCCESS in 51m 34s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/351e302fe61a407381fe590fed7842d8 : SUCCESS in 42m 34s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/efbe82820d8847f5b6bfa20310ddc3a1 : SUCCESS in 1h 28m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/399d515eca814318aacd9b4b875913c1 : SUCCESS in 1h 02m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c886ff0753a343c486ace3405c133cb4 : SUCCESS in 5m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/550b72a47b144f2fa593f5e3f12534d4 : SUCCESS in 4m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5d24f8f8f23640c8a9e4ab5fa2d88f6c : SUCCESS in 4m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/79559a7941f74e2899361069dbb6929a : SUCCESS in 5m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7051dabd3c2840dda85afe97ed17debf : SUCCESS in 3m 34s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/013f61a66a29431ba86d98b644f56c1d : SUCCESS in 6m 22s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a2d9c8a5d9a44a07b01ab059431ffe4b : SUCCESS in 6m 21s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/aed33d95455447dd94b948ac441577e6 : SUCCESS in 7m 30s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/8a46b5e446e9426ebfc9db41935e054c : SUCCESS in 1h 49m 31s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/ae34260d68bb4d57bae34253b8f83db1 : SUCCESS in 1h 36m 43s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/0de91b0eca90488eb8ecf81a042565a8 : SUCCESS in 1h 40m 24s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/5fbbc7a75de34cecbfbe2302f0f44a1c : SUCCESS in 1h 04m 09s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/a0162bb84e554c0388277b15d03fcfa6 : SUCCESS in 2h 17m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d19bbd1008234cffb0470cc1bfdf0979 : SUCCESS in 32m 19s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a5a0d78294244aa6bca7c4c2baae3aa7 : SUCCESS in 1h 49m 27s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/5f0422cb33ed42ed9ee32d4c33939124 : SUCCESS in 1h 16m 20s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/25cf9a101a474648921c43aaac5376d6 : SUCCESS in 59m 10s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/5c14b8d1e2c64cc188bf0c62880bb92d : SUCCESS in 1h 13m 05s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/24897313be874aa19c390cc1d36a04c2 : SUCCESS in 28m 55s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/2198dc576fe645969812417ecea7b078 : SUCCESS in 32m 26s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"90b47b48eadc5903c29bcd5c27aae7dc54663bd1","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-04-23 17:52:41.000000000","message":"Patch Set 3: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":3},{"id":"fcb4808a14b4553a09a4dda8e77b03120fef12e3","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-04-24 06:35:59.000000000","message":"Patch Set 3:\n\n(5 comments)","accounts_in_message":[],"_revision_number":3},{"id":"1046338d6a22cdb0baccc9bb994d577b2c0c110e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-04-24 06:37:44.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"4c460233df1af53875d06c453857b0518dc572a5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-24 06:51:53.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/512c46f3538844a6a799ee844713229d\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b68595f402d043a9821d4b330fc9b477 : SUCCESS in 7m 37s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/abc8e0d0d430493b89018d9254edb5a5 : SUCCESS in 8m 41s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/fc81692d37c24491a329a3016ab6b8de : SUCCESS in 12m 16s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"2ca9dbe8c606a5097dd28c9c89882a4824d95f3d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-24 08:43:28.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0cd655e1c6c6423b885619906613aa79\n\n- grenade https://zuul.opendev.org/t/openstack/build/020acde69d6c4073ac5436d7a698c917 : SUCCESS in 53m 08s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f3bf5284b60f4048878733fff76b6b7f : SUCCESS in 56m 22s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b679893db1fe4e0f89ff373622c4b3b0 : SUCCESS in 1h 31m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ee7128ed98e04ca9a582cb8a607b548f : SUCCESS in 1h 02m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f066499be3bf40369fdfabf06559ae54 : SUCCESS in 2m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ee2c615162614889a13adacd318fe396 : SUCCESS in 4m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ff357159083c4dd79730a56f31677017 : SUCCESS in 5m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/61b001ee7ba84b0890ce7a6d1e64f9b5 : SUCCESS in 7m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0279e265472943ac92e50052c5fbe4c9 : SUCCESS in 5m 47s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/88c06a7f1f634475a2e8d89e30af1174 : FAILURE in 7m 59s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3b1fa6dffd91494ea578294b22c0d14a : SUCCESS in 7m 53s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/dab16c4b5cb64272b491c966f76e1cff : SUCCESS in 9m 36s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/789700c3b7e6435eac1e133a911d403b : SUCCESS in 2h 00m 30s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/4ffc9c10717048c98155fed903581d2e : SUCCESS in 1h 36m 23s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/7503e9161aa8464b87fdd9ab88ccf796 : SUCCESS in 1h 36m 55s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/bdb54bfd55704900b2bfb9319119d810 : SUCCESS in 1h 35m 50s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/a25ef0afa65946ffb9649e230f8ea494 : SUCCESS in 1h 55m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7cf431387964445bb16aecc72104783b : SUCCESS in 27m 15s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/194a4fcf1a7e4d9b840320f97eea6d35 : SUCCESS in 1h 52m 15s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/7c82c6a0f2a2431296dafa40329cf36f : SUCCESS in 1h 51m 27s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/f5399f6c17324acd8465a6e5e08af3be : SUCCESS in 48m 37s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/57caf73e7d5242538d2d5e7c417e45d3 : SUCCESS in 1h 31m 32s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/24f13653cbca462787d5b00db9a25de2 : SUCCESS in 25m 01s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/6d26b7eecc4a4b368149c09ffda4df5c : SUCCESS in 56m 48s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"eb7a7f15729738da9401816b5d060f5f5425ffbf","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-04-24 08:54:37.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"b3dcc4e6882f9a073e1b2d7d882b10800cb5b69a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-24 09:09:19.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/71800e3f5c0b48ff967710e1e7848696\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c0727d568b874e2e96d753d60dccd3e3 : SUCCESS in 8m 00s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/814de665a2974924aa203147efeca022 : SUCCESS in 9m 38s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/6359f891249a420586ece5c8ba19ab75 : SUCCESS in 12m 57s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"e90c401b1744b8e20773c98c9c22d85f98bec746","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-24 11:03:38.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/269841d521d741a7a0e2b8fa381d5588\n\n- grenade https://zuul.opendev.org/t/openstack/build/a3f095ad2cf2492b8208d99f260faf2f : SUCCESS in 52m 34s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0bbb6552b40d4fc69e7be39cdc32cd82 : SUCCESS in 1h 04m 08s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c5fbd5cf646c41f09ddbc474d07d8ccf : SUCCESS in 1h 13m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ee12acaf50a04d7aaac00647fce08aad : SUCCESS in 31m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41a60dbdb0984af9ac7e125e79a8bfb2 : SUCCESS in 2m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ff9bc7140930423fb56de1419eb327fa : SUCCESS in 4m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/773059a6871d4d0197f8e21131f6fc8b : SUCCESS in 5m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9bf07f91335d449a909e9bccdff51411 : SUCCESS in 8m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88321da388784a5ba176f1b938412327 : SUCCESS in 4m 40s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b0dabae79913438d9575a73e385e5c7e : SUCCESS in 7m 35s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/92ff36e965844e46b49821b1421c2595 : SUCCESS in 6m 53s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/80aafa67c54d4b818e1fc0d622db13d4 : SUCCESS in 12m 24s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/2fd41d94bfb548a88240e7caa295a10b : SUCCESS in 1h 51m 06s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/5a130e702a0b4b0481508bf507c9936e : SUCCESS in 1h 58m 03s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/c2e97c512fc0455ebfac071d8da8a0ce : SUCCESS in 1h 30m 31s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/783ce3437af6471ea3b996082cbc517d : SUCCESS in 2h 05m 02s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/34ac34dd369c41bea20832818bb6701f : SUCCESS in 1h 56m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0bf3ffdb16d24ca2841da7154904098d : SUCCESS in 53m 14s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/496a4c56c5fe4bca9d32444d1d324e07 : SUCCESS in 1h 49m 18s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/5df2b9c06f3744849ccb54f02555f2fb : SUCCESS in 2h 00m 40s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/d7d828919508425c8892064203d4b3dd : SUCCESS in 54m 21s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/781258b26ae546fdb6b8294abfb62354 : FAILURE in 1h 26m 17s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/0ce6d4f979654eaeac2803011edd9d67 : SUCCESS in 28m 15s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/f0c9c7f1afc643e3b7245d282780de13 : POST_FAILURE in 7m 27s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"1667ce104c0f2fb58349117b9e2895f3ce772381","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-04-28 14:33:27.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"ab57408ad9ae4c7468760fed049501e889b0d2ac","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2025-05-09 08:23:49.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"e92a4b628127cfecdc92c21fe6c42878af7c4124","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-05-14 15:24:50.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"64f9b0e96a0a55ebadfb0c2ac93159876b8c0306","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-14 15:25:29.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"7fa03a724a67d2f3759df45cbac78379b8bdd07a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-14 18:33:55.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c560d43d3f5e46db9d4d53915562b87c\n\n- grenade https://zuul.opendev.org/t/openstack/build/a43905144e8e4e1cbfd780f92b80d652 : SUCCESS in 1h 01m 27s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b63df21aa7e34c0d82346655031b9bff : TIMED_OUT in 3h 03m 52s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2a54c3fb1a5340cd82a6a7c19f39f3e5 : SUCCESS in 1h 25m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4d9c4b4c74bf471db9516244ea53e698 : SUCCESS in 54m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1fd2883e6ca4568a75efbbffa6bdfaa : SUCCESS in 3m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b2b8c0b227604d808b5c2909216f5634 : SUCCESS in 5m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5b2fb5865fe24967884c15050e2b7ef2 : SUCCESS in 3m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6ffb8a075f74d61ab940ee4dec1f54e : SUCCESS in 7m 10s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c32a4a9385bb45a0b2e844fdd80a7a32 : FAILURE in 8m 20s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/14c143bccfe54454bb8fb77a659c8173 : SUCCESS in 9m 02s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/81aa2900094a4754bd486e5eed50613c : SUCCESS in 1h 55m 55s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/4213d4bf4bc94f35860ef359601a8b26 : SUCCESS in 1h 59m 35s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/0e41c0a1b60c44daa8a7c31e887517b2 : SUCCESS in 2h 13m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8f60ad2b99b94d65a037c8cd1ef88b3c : SUCCESS in 51m 41s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/699d377fc597481ea1affa865a4fc461 : FAILURE in 1h 36m 25s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/ff264d05815f40469dd0850efef6bd57 : SUCCESS in 27m 53s","accounts_in_message":[],"_revision_number":4},{"id":"173902a4f033c6288ceaf1f39e3af5db8e3ef105","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-05-14 18:54:56.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"f832c5048e224ab7ecccb922e46c798603cb521b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-14 19:07:57.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e12a8b83370c46409708b8f28034e833\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4f4a02324e254d1e9e8f36440e6f374f : SUCCESS in 8m 07s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6a372c2d5a1846f1ab77e4ab90f1003f : SUCCESS in 7m 07s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c7ec956914884251a1f55fac274e79cf : SUCCESS in 9m 21s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"faea1701b8a4723bf560fee924ac5592726e0307","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-14 21:08:05.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0eba6ef584d3430487c0d31aca5acebb\n\n- grenade https://zuul.opendev.org/t/openstack/build/b3b98167184144e8a52997d975317b5e : SUCCESS in 1h 00m 06s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/fba7e0675e4b42b9ab5b26bbde96c87c : SUCCESS in 1h 02m 40s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/81adec8dc7ed4734b20aa1f8a50a8b4a : SUCCESS in 1h 25m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cb51150419cb4ce08c3521049828fc6b : SUCCESS in 36m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/022e699e8a2a4cef86bcca16ea48a8a9 : SUCCESS in 6m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9b71fe8bedd9481aa1af9a6665ca210e : SUCCESS in 4m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/af2ea428ee36473682bd9ea0c54267fa : SUCCESS in 3m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2805d0f7ea9046d584c1c306d9a756d3 : SUCCESS in 9m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f05d59716bc429f85f5fe867b8d03dd : SUCCESS in 3m 42s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/bfa7521e056d468c8f125c53e4f6bf44 : SUCCESS in 6m 42s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/cb3224fc11a849d789caf9f23bbd3823 : SUCCESS in 10m 31s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/a86d15668014479598786b2e843e6723 : FAILURE in 2h 00m 26s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/9bf7859454bf4a1ca8757c85bd5fc879 : SUCCESS in 1h 43m 23s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/e4c56aab29a547729ca7f56bca42aa91 : SUCCESS in 1h 41m 30s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/8e9f1605caa748b78dba2ed85af48213 : SUCCESS in 1h 51m 26s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/832189a25e28404293693ae93d6dfe9e : SUCCESS in 2h 10m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/862803720ff44180ae1be6716817ecf5 : SUCCESS in 57m 41s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/add0fe55fd7141ef893e7e5d5996fe90 : SUCCESS in 2h 07m 17s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/3adf11ac78f9475881e0d1179a0881d0 : SUCCESS in 1h 41m 26s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/7f876da9ffd842788a90fe7e0da59cff : FAILURE in 32m 03s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/00d6d419ddac480bb14dfe14a5c68135 : SUCCESS in 1h 10m 36s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/26455180416545ba9522a857d3b3d155 : SUCCESS in 15m 14s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/3d07bf87fb6c4b83bb083141584efc73 : SUCCESS in 59m 39s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"7cb9f35754b02b26fd597cf202e452c7fe0f7219","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-14 21:08:30.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"429384f72163eeb86a7d473f23c88ba549f56ff2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-14 22:01:00.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6872f191ab7e44e9b809b624ff930548\n\n- grenade https://zuul.opendev.org/t/openstack/build/f2dbbec93c534483804d057024d43e0f : FAILURE in 5m 53s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7d693389654e412ea2425e90b934b41a : SUCCESS in 49m 48s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f5a829fb6a874f0eb8145801381e88c8 : FAILURE in 5m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c4d3c946b7e04333830cca7dbeb74c30 : POST_FAILURE in 5m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88510e7cd1d94078bfd9fa310097f3f6 : SUCCESS in 2m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f0128b5272d447d49cbce5a02cb96ed4 : SUCCESS in 2m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/31417764f1d74a32950d976916aeb9f5 : SUCCESS in 2m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01a3319f809444e69b815a7e4d16ebdb : SUCCESS in 3m 45s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7f44db9301734ebc9212dca427341951 : SUCCESS in 5m 09s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/85deed05218243e5869b00549546d69a : SUCCESS in 10m 10s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/852aa9ce4c894a8eb9c393312ec8cae8 : FAILURE in 10m 48s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/a7c85b997e9f4cf4a5e94e6727d1f531 : FAILURE in 5m 51s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/6d705212ebb9420a902d6bb2e4d49203 : FAILURE in 11m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/05522b4f513e4cb68b901c090abcb9ec : FAILURE in 11m 00s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/65b7197eb0d040c0b01eee8bc09b19b6 : FAILURE in 5m 28s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/44f90312e03d43a281e9026b9fe271c1 : FAILURE in 11m 08s","accounts_in_message":[],"_revision_number":4},{"id":"e8f549ac9f20808a66bc8c6f43fe00c4e82ab347","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-05-15 02:14:24.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"1e72c0796838e0f25978d28ea792962ce63de5c3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-15 03:00:35.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ddf1ab58cc44dcaa1103b68d42b509b\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/dbeded490a8449c38e46b51c5d4ac1fb : SUCCESS in 11m 25s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5660f9bc875b4c2e9d7df3c14dd61b63 : SUCCESS in 11m 46s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/9bedd319f4634460806bea05b284f365 : SUCCESS in 14m 06s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"0fb19eb511c25519f320da43e9b7b671ca226bb1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-15 04:46:06.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4a6e92a87e7541158c6165c69379c8f2\n\n- grenade https://zuul.opendev.org/t/openstack/build/1334a54d1e6f42569a3f21762735e85c : SUCCESS in 49m 38s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5c5afbd1d25a4c5abf43c247be3071a7 : SUCCESS in 55m 00s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/bd32795efc2041a4b63c02616a1a2eb4 : SUCCESS in 1h 32m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cef108fee2f140dd9d8a9d74f68f80cf : SUCCESS in 1h 05m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e4fb313a75b4ec7a53574824ba18545 : SUCCESS in 6m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9cdad023f6294c5f8c875e2dcf99cb45 : SUCCESS in 4m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/aa84586db10249189ff1975453cde2e6 : SUCCESS in 5m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4e1ca2f7add14324b67cc9a2dbded144 : SUCCESS in 8m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b886df890334c39a8ee37d0376047d9 : SUCCESS in 5m 27s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c6a86f269349497e87c8a19e7dd287c6 : SUCCESS in 8m 54s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/dbca50b4482944beb6a0cb1e651c8c6e : FAILURE in 12m 38s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/4c7aa2828d60474596b9b1ed423f1a5e : FAILURE in 2h 14m 28s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/a7d24cd408ac4a2b86f685298d6458f4 : SUCCESS in 1h 46m 54s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/cab50392534d43fc9aa250375e880b06 : SUCCESS in 1h 51m 00s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/3cde47959d654336b78b06e818bf0b81 : FAILURE in 1h 45m 05s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/df4be2adc13345ca934fa523d2c20d8b : SUCCESS in 2h 13m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8930020577ad466ca0f0d3f0f593f62d : SUCCESS in 32m 46s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/de17fd4f8f914dc6a797c7b0f730d03a : SUCCESS in 1h 49m 06s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/a24897ce540f45acb99b335af955e6ae : SUCCESS in 1h 46m 47s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/147b09f2e190478684884db4b5d8ec46 : FAILURE in 16m 07s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/75a913fc2e644620813bc6d5d87ada85 : SUCCESS in 1h 47m 13s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/10ba56b6e608411985f1a3346231c52e : SUCCESS in 28m 26s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/e87251489fe149319cb884f35371d9e7 : SUCCESS in 30m 57s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"0b6fc10114e0dc44a660d15bc57e62290226980b","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-05-15 06:03:25.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"95a2dedea3f53304d86218677dc31c648d4c2a96","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-15 06:18:34.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/752225f75faa45019a98b7cc2b3c76a6\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/42cff2f7a3914e83b330b5538318a565 : SUCCESS in 8m 55s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e7ba7ecceaa549c78e229c2db53be82b : SUCCESS in 9m 59s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/782425dbd738479ea8fcf6b2b1e40480 : SUCCESS in 13m 09s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"a013fe4363a4b0f4a780ac57f3824970542aa5d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-15 08:11:28.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/46d4ccabf2dc477a8248352c10e3f3ea\n\n- grenade https://zuul.opendev.org/t/openstack/build/0c887753fc434014b588f17dfa135904 : SUCCESS in 28m 49s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a679f0ea64a34a4bac053ac9cf40ec06 : SUCCESS in 32m 08s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/598948466a514db1a7a0c23e72330d69 : SUCCESS in 1h 23m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/90b6600741e2496e8cdc01eef79b1770 : SUCCESS in 51m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bde7328ea6cb49a4980c50b60809caea : SUCCESS in 4m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4b0df32bd4824847995ef331920fa2d3 : SUCCESS in 5m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8badc6ab08794977897bd944300dbda3 : SUCCESS in 5m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fbc8c721d94c4e22aab49dfe99e08606 : SUCCESS in 4m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/772ed47193cc45468884470ebafe3a4b : SUCCESS in 5m 21s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8d0dfd69b6fd4009a26b7cb5da0ddf31 : FAILURE in 7m 09s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/91e0ca579122417a90d7bad33812c9b9 : SUCCESS in 7m 33s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/5d31558ce33b488b809d53f8b9f6a456 : FAILURE in 1h 35m 20s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/7aeb5aef15b640d28e25c229ab1f9885 : SUCCESS in 1h 50m 50s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/ef0610ad9ce24ec78184137aa99cba7d : SUCCESS in 1h 18m 39s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/8c5693256eff40c5be37860a1b092cb2 : SUCCESS in 1h 35m 37s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/84cdabcb46b4458aa7f25e29dd566dd5 : SUCCESS in 2h 07m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/658441445f884c9ca48601a6c5d15729 : SUCCESS in 1h 01m 43s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/b95d28e208a144d8acc6b0d8ffba6360 : FAILURE in 1h 44m 08s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/14bfbac329e2434d965e21d756a15093 : SUCCESS in 1h 17m 00s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/e64849cb65104844af4d2a14c346bb64 : FAILURE in 31m 33s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/e1a32d02093846cfa2e5df59873de7b4 : SUCCESS in 1h 13m 21s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/053e1f1cfec243b78585949bd3c9ed93 : SUCCESS in 29m 55s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/6f5e9a85f9004a6bae353265fef24af9 : SUCCESS in 32m 44s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"ffeb4ed9e29902a02d01a70925622eeb4fc8ca63","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-05-15 08:12:23.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"3f904d99b0d01885f5c51469c211d822c557745e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-15 08:24:57.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/abc681fe164b4a62ba355143613ae61b\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7fd5220270fc4a52b11fac841d3fbed7 : SUCCESS in 7m 33s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/109d52c8186e4376bcd5150ca0535336 : SUCCESS in 8m 45s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/fb3b3525271346a09cd062db11afcae8 : SUCCESS in 11m 01s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"193f1ff13e889b778757bac4541b736aae4ad7a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-15 10:35:11.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/24da25b701724e75bf02132472ba355d\n\n- grenade https://zuul.opendev.org/t/openstack/build/0ebb35d29f7b4b60bf6af02e93c2e582 : SUCCESS in 56m 28s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ed5c19234c5e4047ae4ce4653e8fb824 : SUCCESS in 50m 18s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8098c759b7fd4ee98a81d44193c413af : SUCCESS in 1h 22m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/95ea8d82e40242b39d5139f896a3f30b : SUCCESS in 1h 00m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/572c3cef18ba41918fdd8380612d3387 : SUCCESS in 4m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/85ad765ca5d14083ac9be4f04a3b715e : SUCCESS in 4m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/aa4e6669f6e34492b2a8569a21d2546e : SUCCESS in 5m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/48ceb187405c45a09bb9d5babc161c38 : SUCCESS in 7m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b9a42935c8b4defb931095277f3657d : SUCCESS in 6m 57s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/812ad669d910471bb73ba68cf345b012 : SUCCESS in 7m 34s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/44387714527043439542c9dd74b47123 : SUCCESS in 10m 46s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/dc1d162fcc2d453da7e6648f2e425849 : FAILURE in 1h 16m 05s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/d26e3913c5c44720b1614dcefbc2f1eb : SUCCESS in 1h 03m 19s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/8b9f2248fbd240628afe5b832663d197 : SUCCESS in 2h 15m 57s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/ec9e6810357c434aade5664260d42454 : SUCCESS in 1h 41m 30s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/0754c105a3144c3ea5e85d49e36714e4 : SUCCESS in 2h 17m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/72882190fadb4f14876afe175ebf4568 : SUCCESS in 53m 24s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/118202d3a15749889c1f388d7a81abd0 : SUCCESS in 1h 02m 58s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/731aa49f835541d3b7f2c3605375fbf9 : SUCCESS in 1h 24m 13s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/eed1a9e9aeb44573929b2361b5a532d0 : FAILURE in 31m 34s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/dc6b66abd17843ae8a7fa30c19758500 : SUCCESS in 1h 44m 42s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/310c54b752f548d290b1d014b485ffe0 : SUCCESS in 13m 49s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/5e6e8f207d4f4abe8541350ee3d9c037 : SUCCESS in 1h 00m 42s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"616e6120b89e8c788ebcf63aa8c6ccdb6dcf4b43","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-15 10:35:39.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"d90b34adb8a8ff776ec6a06ded7f52c2cc8d871e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-15 12:46:01.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c3eee97b74024fc493dcd984703cdc0b\n\n- grenade https://zuul.opendev.org/t/openstack/build/656d6393858d41cdb7af45e34f48e0c2 : SUCCESS in 49m 29s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d132745e852949099f25a240a13b0f0e : SUCCESS in 45m 26s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/89539db4fecf4624a3d0cc9eb2779d5e : SUCCESS in 1h 29m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bbeb348e0831431aa3734552be4b5000 : SUCCESS in 52m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0bef6ea16bf346269d526506ac60f994 : SUCCESS in 7m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e66d0ecba0b34c3dbc543b796d4bb36e : SUCCESS in 6m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ad553686d8df4c579a1ed96ab4a347f8 : SUCCESS in 6m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e209880baa8b4aaba8631966ed7bfcae : SUCCESS in 5m 14s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f969a6b615ea4ecfbfcaadb98f236272 : SUCCESS in 7m 19s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/0c945ad5a63a49b29c044706e3170b67 : SUCCESS in 7m 04s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/df1248419d184ee0a8c2ef9599133da5 : SUCCESS in 1h 52m 49s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/a39c37b41e464d1b9eb69d3b53283340 : SUCCESS in 2h 05m 57s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/1062877b9141454095b894b7daea8b30 : SUCCESS in 1h 41m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/137e6909a24a484395e1dc7a8c005b1a : SUCCESS in 50m 37s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/2127ab90d61e4e038fb0b5c33be36e31 : SUCCESS in 59m 33s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/9e89e55f96174eefabafe7ae19d90f08 : SUCCESS in 17m 15s","accounts_in_message":[],"_revision_number":4},{"id":"99f2541d9e0653b0bb178ef01436fc6845b890a0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-15 12:46:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"9856db2ea01168c78f4d26335ce6bdab7579b750","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-15 12:47:10.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/45ee983c10bd43e595ae8e9e33675725\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6efc5a4d5d904ca78d7dde2cd7779424 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"0660696663f26f09eca12d6e18cab29dd9abb699","revisions":{"2655a0f379b2ab030a55361ba0495118fc2b2f5e":{"kind":"REWORK","_number":1,"created":"2025-04-17 20:27:06.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/37/947637/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/37/947637/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/37/947637/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/37/947637/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/37/947637/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/37/947637/1"}}},"commit":{"parents":[{"commit":"7dbb31949e7454c680ec3c1fdad06bd90b48ac9f","subject":"Merge \"Use pyenv for the periodic jobs running on Python 3.13\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7dbb31949e7454c680ec3c1fdad06bd90b48ac9f"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-04-16 06:41:11.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-04-17 20:05:01.000000000","tz":0},"subject":"Refactor scrubber to remove use of eventlet","message":"Refactor scrubber to remove use of eventlet\n\nModified glance-scrubber and related functional/unit tests to\nmove away from eventlet. Used ThreadPoolExecutor instead of\neventlet.\n\nNOTE: Removed test \u0027test_scrubber_app_queue_errors_not_daemon\u0027 as\nit is no longer reproducible.\n\nChange-Id: Ia6931b4cf81c0fb3acbbae914e0c00ba364dc8e2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2655a0f379b2ab030a55361ba0495118fc2b2f5e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2655a0f379b2ab030a55361ba0495118fc2b2f5e"}]},"branch":"refs/heads/master"},"b6b3586d7f7716e4b611e6727bc366c1bc4a8eac":{"kind":"REWORK","_number":2,"created":"2025-04-21 08:36:07.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/37/947637/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/37/947637/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/37/947637/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/37/947637/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/37/947637/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/37/947637/2"}}},"commit":{"parents":[{"commit":"7dbb31949e7454c680ec3c1fdad06bd90b48ac9f","subject":"Merge \"Use pyenv for the periodic jobs running on Python 3.13\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7dbb31949e7454c680ec3c1fdad06bd90b48ac9f"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-04-16 06:41:11.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-04-21 08:16:40.000000000","tz":0},"subject":"Refactor scrubber to remove use of eventlet","message":"Refactor scrubber to remove use of eventlet\n\nModified glance-scrubber and related functional/unit tests to\nmove away from eventlet. Used ThreadPoolExecutor instead of\neventlet.\n\nNOTE: Removed test \u0027test_scrubber_app_queue_errors_not_daemon\u0027 as\nit is no longer reproducible.\n\nChange-Id: Ia6931b4cf81c0fb3acbbae914e0c00ba364dc8e2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b6b3586d7f7716e4b611e6727bc366c1bc4a8eac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b6b3586d7f7716e4b611e6727bc366c1bc4a8eac"}]},"branch":"refs/heads/master"},"5badf74f78d0636a2daf352f651d108051af3cd8":{"kind":"REWORK","_number":3,"created":"2025-04-21 08:45:10.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/37/947637/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/37/947637/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/37/947637/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/37/947637/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/37/947637/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/37/947637/3"}}},"commit":{"parents":[{"commit":"7dbb31949e7454c680ec3c1fdad06bd90b48ac9f","subject":"Merge \"Use pyenv for the periodic jobs running on Python 3.13\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7dbb31949e7454c680ec3c1fdad06bd90b48ac9f"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-04-16 06:41:11.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-04-21 08:25:54.000000000","tz":0},"subject":"Refactor scrubber to remove use of eventlet","message":"Refactor scrubber to remove use of eventlet\n\nModified glance-scrubber and related functional/unit tests to\nmove away from eventlet. Used ThreadPoolExecutor instead of\neventlet.\n\nNOTE: Removed test \u0027test_scrubber_app_queue_errors_not_daemon\u0027 as\nit is no longer reproducible.\n\nChange-Id: Ia6931b4cf81c0fb3acbbae914e0c00ba364dc8e2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5badf74f78d0636a2daf352f651d108051af3cd8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5badf74f78d0636a2daf352f651d108051af3cd8"}]},"branch":"refs/heads/master"},"0660696663f26f09eca12d6e18cab29dd9abb699":{"kind":"REWORK","_number":4,"created":"2025-04-24 06:37:44.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/37/947637/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/37/947637/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/37/947637/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/37/947637/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/37/947637/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/37/947637/4"}}},"commit":{"parents":[{"commit":"d2c3e102fdff35c1dd15495c0445870e701d7e4b","subject":"Update CLI doc header for 30.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d2c3e102fdff35c1dd15495c0445870e701d7e4b"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-04-16 06:41:11.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-04-24 06:17:59.000000000","tz":0},"subject":"Refactor scrubber to remove use of eventlet","message":"Refactor scrubber to remove use of eventlet\n\nModified glance-scrubber and related functional/unit tests to\nmove away from eventlet. Used ThreadPoolExecutor instead of\neventlet.\n\nNOTE: Removed test \u0027test_scrubber_app_queue_errors_not_daemon\u0027 as\nit is no longer reproducible.\n\nChange-Id: Ia6931b4cf81c0fb3acbbae914e0c00ba364dc8e2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0660696663f26f09eca12d6e18cab29dd9abb699"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0660696663f26f09eca12d6e18cab29dd9abb699"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
