)]}'
{"id":"openstack%2Fglance~947845","triplet_id":"openstack%2Fglance~master~I03f5eed4aa650c6e115ec32e38d79a61fd81de84","project":"openstack/glance","branch":"master","topic":"eventlet-removal","attention_set":{},"removed_from_attention_set":{"33765":{"account":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"last_update":"2025-07-17 20:59:39.000000000","reason":"Change was submitted"},"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2025-07-17 20:59:39.000000000","reason":"Change was submitted"},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2025-07-17 15:11:32.000000000","reason":"\u003cGERRIT_ACCOUNT_8122\u003e replied on the change","reason_account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}}},"hashtags":[],"change_id":"I03f5eed4aa650c6e115ec32e38d79a61fd81de84","subject":"Functional tests migration","status":"MERGED","created":"2025-04-22 16:02:05.000000000","updated":"2025-07-17 21:00:51.000000000","submitted":"2025-07-17 20:59:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"947845-eventlet-removal","meta_rev_id":"0495022ad1e383f32001e2cfa7e9ba88bb9e798b","_number":947845,"virtual_id_number":947845,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-17 20:59:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2025-07-17 15:11:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2025-07-17 15:12:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"}],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-22 16:02:42.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"state":"REVIEWER"},{"updated":"2025-04-22 16:02:42.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2025-04-22 16:31:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-04-22 17:16:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ae4c67d27a91710755390beba943ebd002f48930","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-04-22 16:02:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2616c88da5dace52af6944d22fcb847ee2919328","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-04-22 16:03:56.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6305613fe3f5be5991516599207ffe10cc56c584","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-04-22 16:11:31.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"35b24b9355cbd8f25c491fb14f11407ac4ab2341","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-22 16:31:39.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7e0efb6aeffa435eab74ede2a022db8a\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4b38a9b73a07483284e7dcc1b455c8d2 : SUCCESS in 8m 24s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6c6fc947161c49f8858be03579464a39 : SUCCESS in 9m 20s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/7de26b50f41144ba9955d8889f44d51a : SUCCESS in 14m 14s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a7c2da020b25f2d1bd8d01a646f1a8faae81a32d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-22 17:16:09.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/dc78708cff7a440ca7471e23c40f7839\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6a68ad1511b445cbaec07a8015e96f37 : SUCCESS in 52m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6ed949a9dafa48508a64c684be3e9d61 : SUCCESS in 59m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b29fe13371340129a66fd19c524701f : SUCCESS in 5m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dc9f733c194d4a87b948cfb7065ce75a : SUCCESS in 2m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5eb65570b620452d9fa6ff116402811d : SUCCESS in 5m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/be6155b392d6425287af0766605da3d4 : SUCCESS in 3m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b5453eee6fc47db9026992e1acb8603 : SUCCESS in 6m 00s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/da29e4d91b69485c9474932e86b0c7a6 : SUCCESS in 7m 58s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/782efa90c7ef40228f945136af900b02 : FAILURE in 5m 31s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/7ff20956852b4f3780db6a2c16b0b602 : SUCCESS in 12m 09s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/3ddd5b9a21c9486aba58f4deb2b4a576 : SUCCESS in 26m 55s","accounts_in_message":[],"_revision_number":2},{"id":"efa64f632583979409b732158286e8588966fdd9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-04-22 18:12:43.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"f86e5b606d20796f8c6f3daa11bd5c285a971d50","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-22 18:35:08.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3c4fbcf320684906b142d4ebabbfd3ce\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/32c88e414f8f4dc2b47172f4bab2dfba : SUCCESS in 7m 58s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/79160a6df53047cd9f1f240c65ec6624 : SUCCESS in 8m 54s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/7bf9a1689db94522abf79a70c99f88d7 : SUCCESS in 12m 38s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"48aec96559821abdb349493230fb52175d0f7fce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-22 19:17:11.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b047853e6b35413c9702286bccb44c44\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1968c727d08543f1846de7350d12256b : SUCCESS in 52m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/03a79eaaf4644f2886edabf5bf0c5567 : SUCCESS in 59m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b473574aea40451fbb661d0d459b9403 : SUCCESS in 4m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d865f0b857bc414e827814909163c7e9 : SUCCESS in 2m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1773f5690a7a4eb9b3b4e7a0a45a9a43 : SUCCESS in 5m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/95bcdcd4974348278c8f8a074a4eda38 : SUCCESS in 4m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c26dfe186c04444ebbcdc8f7e2810b61 : SUCCESS in 4m 29s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/498006bf18024a04a74113c34847fbae : SUCCESS in 8m 03s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d604e46841b14dc3b4a4530e9de0ad59 : SUCCESS in 8m 21s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/de691aab42db4d5b83eddacd637ce5cc : SUCCESS in 11m 13s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/5899aa854d334730ab065cad012e4ac2 : SUCCESS in 13m 35s","accounts_in_message":[],"_revision_number":3},{"id":"e99015471343d591ab5272ffdd9e3ddbfd47aa5f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-05-20 06:58:29.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"8648cfba9c337b313ee82cda2be0cfa7740e8992","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 07:08:51.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ce228954f1e64d3b926c0144377a1572\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/33440cd2c341490c9d31b608b02ea054 : SUCCESS in 6m 32s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c76beee538b44bc0b076743fd9ffc061 : SUCCESS in 7m 03s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/81b1132f00904bdfa01f0fd73086700c : SUCCESS in 8m 49s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"cfc6054bc69e622a16a012ec2d27fe06c237633c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 08:05:08.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c6450b96c224c79be77906418ab918a\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/aa3443fa22524fabbeb82615334db230 : SUCCESS in 56m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/069609ec568d439c87d564cc8cab75d3 : SUCCESS in 1h 01m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd0218cece8c457f9ab15ea1039c5295 : SUCCESS in 4m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fe173f3fbb174390b45902e4b749e5d8 : SUCCESS in 4m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5391c7a217ef487aa4cdbed956639845 : SUCCESS in 3m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9f39e075770a414f9bd09993236cae94 : SUCCESS in 8m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b6a387c3793485fb53310e8a01fb6a2 : SUCCESS in 6m 05s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/88bca58461b8463fb2a381bfff0b15fc : SUCCESS in 7m 04s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/9e950b5dc84345db96d28abf32ce7d7d : SUCCESS in 9m 50s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/6a1463de22b847b4b8d464c453cb3975 : SUCCESS in 34m 04s","accounts_in_message":[],"_revision_number":4},{"id":"720c5368c6a8c1446fef59a218a67395c324dd6c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-06-20 14:01:57.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"5c61209bf091aeeb84c6b71fcd310d735b84b0c0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-20 14:14:36.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/802120869a45403cb91d6ce59666a4b3\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a3f847c8f093481b8e5299b7df8cd7d4 : SUCCESS in 5m 30s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/84fd701b9dd84a5ca61a63ed78e91cc8 : SUCCESS in 6m 20s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/6d1c3af0063c4651ac48b6ee0504d96c : SUCCESS in 10m 38s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"c7a9f342044c3189292192dacf1e80eac2725e28","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-20 15:15:51.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1abe1e1d3516412787234d6180521c6d\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f16572d59b1f40048865843cc6e05275 : SUCCESS in 56m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/457193ced2e04631ab15342c641fc942 : SUCCESS in 1h 00m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e953ceea14464743832eb014c52b05d6 : SUCCESS in 4m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b13a675f1abb42e7a78960e42c120c13 : SUCCESS in 6m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8eefad7916bc43618349a2c26bdb90f0 : SUCCESS in 7m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/34ab1c6d646f4ade8478d17cc94c08bb : SUCCESS in 8m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df90fb220fb440138ff281de045eea3b : SUCCESS in 5m 04s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a70dcb78402d42d58d877c3aab1495d8 : SUCCESS in 6m 35s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/ef145c9797144d7bb19827aa64084539 : SUCCESS in 9m 38s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/5f0e1424394a4e76b8c1555325f5da6a : SUCCESS in 29m 22s","accounts_in_message":[],"_revision_number":5},{"id":"903a49b0262ab1e10d4e01141d23df9a43323c84","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-07-17 15:11:32.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"3a4506955d11744a45cd3b8bdcca23413628d55d","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-07-17 15:12:06.000000000","message":"Patch Set 5: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"8f48e1c9d4dfdf0f08e2b3208eeda0d407db0007","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-17 15:12:36.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"580a35a4a8d316ca3bd7f46382b5ecaf734e679a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-17 20:59:38.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bcd2e2acdbb144cc858aa11880195993\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8a8cbe4949a143beb8610ff6e050141c : SUCCESS in 1h 02m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/423b06fa31384ee1a4087c5d0c8eb5af : SUCCESS in 1h 00m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a50c983acaa344c1bb17b8b0b3bcfa1a : SUCCESS in 4m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0b28b686b0c34a1683438e8f9e20982f : SUCCESS in 4m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ea9f51c93b9341f990bee28e9d9b25df : SUCCESS in 9m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac39ad5653de4fb893771c11281a7ad8 : SUCCESS in 5m 01s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4585accc7f1947229d7ee01905a9bf45 : SUCCESS in 7m 29s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/b871ed02bc2343abbbde3a5bbd4a2923 : SUCCESS in 7m 53s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/09e87a58c3404a5eb745d738c267c932 : SUCCESS in 25m 42s","accounts_in_message":[],"_revision_number":5},{"id":"d084d95fd3fe73996897186a9ee9d2218cc1ed32","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-17 20:59:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"0495022ad1e383f32001e2cfa7e9ba88bb9e798b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-17 21:00:51.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8ca983edc9b7440b87d31b9960ea765d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60e72318650241449155eaabf213fae0 : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"ec28aa136cf09b6095ba5c745d9c961eec4c04eb","revisions":{"a18f5aa13e8701a56415566ca30bd0f71fd53642":{"kind":"REWORK","_number":1,"created":"2025-04-22 16:02:05.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/45/947845/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/45/947845/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/45/947845/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/45/947845/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/45/947845/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/45/947845/1"}}},"commit":{"parents":[{"commit":"d2c3e102fdff35c1dd15495c0445870e701d7e4b","subject":"Update CLI doc header for 30.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d2c3e102fdff35c1dd15495c0445870e701d7e4b"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-04-22 15:41:53.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-04-22 15:41:53.000000000","tz":0},"subject":"Migragte Quoatas and Cors middleware functional tests","message":"Migragte Quoatas and Cors middleware functional tests\n\nChange-Id: I03f5eed4aa650c6e115ec32e38d79a61fd81de84\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a18f5aa13e8701a56415566ca30bd0f71fd53642"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a18f5aa13e8701a56415566ca30bd0f71fd53642"}]},"branch":"refs/heads/master"},"14955e302d4a6a1a56c91605a3e229955587726b":{"kind":"REWORK","_number":2,"created":"2025-04-22 16:11:31.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/45/947845/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/45/947845/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/45/947845/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/45/947845/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/45/947845/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/45/947845/2"}}},"commit":{"parents":[{"commit":"d2c3e102fdff35c1dd15495c0445870e701d7e4b","subject":"Update CLI doc header for 30.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d2c3e102fdff35c1dd15495c0445870e701d7e4b"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-04-22 15:41:53.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-04-22 15:52:05.000000000","tz":0},"subject":"Migragte Quoatas and Cors middleware functional tests","message":"Migragte Quoatas and Cors middleware functional tests\n\nChange-Id: I03f5eed4aa650c6e115ec32e38d79a61fd81de84\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/14955e302d4a6a1a56c91605a3e229955587726b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/14955e302d4a6a1a56c91605a3e229955587726b"}]},"branch":"refs/heads/master"},"7e7bde2e82e42d62c32d5f29e4e2d091e2f59aaa":{"kind":"REWORK","_number":3,"created":"2025-04-22 18:12:43.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/45/947845/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/45/947845/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/45/947845/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/45/947845/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/45/947845/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/45/947845/3"}}},"commit":{"parents":[{"commit":"d2c3e102fdff35c1dd15495c0445870e701d7e4b","subject":"Update CLI doc header for 30.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d2c3e102fdff35c1dd15495c0445870e701d7e4b"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-04-22 15:41:53.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-04-22 17:51:42.000000000","tz":0},"subject":"Functional tests migration","message":"Functional tests migration\n\n- test_cors_middleware\n- test_cache_middleware\n- test_images.TestQuotas\n\nChange-Id: I03f5eed4aa650c6e115ec32e38d79a61fd81de84\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7e7bde2e82e42d62c32d5f29e4e2d091e2f59aaa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7e7bde2e82e42d62c32d5f29e4e2d091e2f59aaa"}]},"branch":"refs/heads/master"},"90d5e36cad43efd4a24789e81e721892975015d6":{"kind":"REWORK","_number":4,"created":"2025-05-20 06:58:29.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/45/947845/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/45/947845/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/45/947845/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/45/947845/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/45/947845/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/45/947845/4"}}},"commit":{"parents":[{"commit":"7cf7a1445a222590027b3fca7c9ad026d3781083","subject":"Merge \"Revert \"Deprecate glance scrubber\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7cf7a1445a222590027b3fca7c9ad026d3781083"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-04-22 15:41:53.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-05-20 06:36:04.000000000","tz":0},"subject":"Functional tests migration","message":"Functional tests migration\n\n- test_cors_middleware\n- test_cache_middleware\n- test_images.TestQuotas\n\nChange-Id: I03f5eed4aa650c6e115ec32e38d79a61fd81de84\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/90d5e36cad43efd4a24789e81e721892975015d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/90d5e36cad43efd4a24789e81e721892975015d6"}]},"branch":"refs/heads/master"},"ec28aa136cf09b6095ba5c745d9c961eec4c04eb":{"kind":"REWORK","_number":5,"created":"2025-06-20 14:01:57.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/45/947845/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/45/947845/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/45/947845/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/45/947845/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/45/947845/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/45/947845/5"}}},"commit":{"parents":[{"commit":"c41b6ff3c2d07d9b964cf26a77254c5b10ba4ab8","subject":"Merge \"Functional test migration test_wsgi.py\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c41b6ff3c2d07d9b964cf26a77254c5b10ba4ab8"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-04-22 15:41:53.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-06-20 13:38:43.000000000","tz":0},"subject":"Functional tests migration","message":"Functional tests migration\n\n- test_cors_middleware\n- test_cache_middleware\n- test_images.TestQuotas\n\nChange-Id: I03f5eed4aa650c6e115ec32e38d79a61fd81de84\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ec28aa136cf09b6095ba5c745d9c961eec4c04eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ec28aa136cf09b6095ba5c745d9c961eec4c04eb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
