)]}'
{"id":"openstack%2Fglance~948292","triplet_id":"openstack%2Fglance~master~Ie893070d9974c71c4f121e57db72a781c228420a","project":"openstack/glance","branch":"master","topic":"eventlet-removal","attention_set":{},"removed_from_attention_set":{"33765":{"account":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"last_update":"2025-05-16 17:23:07.000000000","reason":"Change was submitted"},"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2025-05-16 17:23:07.000000000","reason":"Change was submitted"},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2025-05-05 14:34:40.000000000","reason":"\u003cGERRIT_ACCOUNT_8122\u003e replied on the change","reason_account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}}},"hashtags":[],"change_id":"Ie893070d9974c71c4f121e57db72a781c228420a","subject":"Functional tests migration test_images.TestImages","status":"MERGED","created":"2025-04-27 10:31:53.000000000","updated":"2025-05-16 17:24:39.000000000","submitted":"2025-05-16 17:23:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":16,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"948292-eventlet-removal","meta_rev_id":"38dcdc13033f1e2a583884d9b63662d27e701ef6","_number":948292,"virtual_id_number":948292,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-16 17:23:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2025-05-05 14:34:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2025-05-16 15:41:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":1,"date":"2025-04-30 17:32:26.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2025-05-05 14:34:40.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"}],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-27 10:32:26.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"state":"REVIEWER"},{"updated":"2025-04-27 10:32:26.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2025-04-27 10:44:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-04-27 11:17:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5de632b7079f8a31107534207c7f7ad455b0d52c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-04-27 10:31:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8641d73704bf5c9681864721050bd18d3edde39d","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-04-27 10:32:26.000000000","message":"Patch Set 1: Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"89332523449720ca8cb8e492b5a3f0785ab0ea2a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-27 10:44:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f6b71856403b484ab383afc76cfdef55\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8d1fb5363a844d1e96ef7e7fa825c42b : SUCCESS in 9m 05s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8d8b4aceafa5481793526161366bb807 : SUCCESS in 9m 43s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/cd343fd965a947ac8954a3b708a4945e : SUCCESS in 10m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"33fe2f7ba50939185c9cdc557e43a0a3a3527e17","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-27 11:17:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6911a75f6e7b4afb8ae6d1529efbda13\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e13762f450f045f5a028c53a25c243c5 : SUCCESS in 37m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6b23a05f8a844cb9b8ce22f96e0e2f9d : SUCCESS in 44m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9611a078ee4742f8855e533f3fc69590 : SUCCESS in 3m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/89a300e92ed44bc7bfcbd14fc1bbb9b1 : SUCCESS in 2m 27s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/91466ef04a094757ad74f7e25f9ec3eb : SUCCESS in 3m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6c8f1070483644749f64c371910ff828 : SUCCESS in 4m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3646ba485764b7593dd072b8e47c355 : SUCCESS in 4m 42s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/81d8dc9b6fa0467fb556406df59bcb70 : FAILURE in 5m 26s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/034458433c73491db29b14c13404f31d : SUCCESS in 6m 16s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/c1e23eb31088409c8a9391df65eeecfa : FAILURE in 6m 46s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/127e119d20754b2c9de0eb3a53e313a5 : SUCCESS in 13m 32s","accounts_in_message":[],"_revision_number":1},{"id":"da02b42750fc090590d2428cd7d25912bd6c71ac","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-04-27 12:35:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"91617cca287b65fe65b0ddeaee9e2c30f7323d1f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-27 12:50:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f5d37d388f0a45c895f4dac0c1e823ab\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/99ef35df14164c4dace1892d9a634c6a : SUCCESS in 7m 41s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/612921dc13934b559d8402d5ed0897bb : SUCCESS in 9m 31s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/49458d879412463baccfc75cda8f2090 : SUCCESS in 13m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6c7f5d1ccb1d15d56129e9a4e811704123d275da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-04-27 13:13:13.000000000","message":"Uploaded patch set 2.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"836eddaac64c275546fb66230a55166a42f1bfc3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-27 13:29:18.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/18324ced3eea41d58f04ce2d5350892c\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9375f8236fd144e68e654a409cd8e610 : SUCCESS in 8m 16s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f756c3983cbd4bf29395655edb6d9893 : SUCCESS in 9m 33s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/3efc37f3fb3148f1affe98aa254b0f73 : SUCCESS in 14m 18s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8aae30d4ca9943705b39bb9bd13cd3ca5a958401","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-27 14:01:33.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2eb84ffff6e7455b92c59b8e2538e7d3\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/55acec22442a4e56a1bc2531ce1f0340 : SUCCESS in 40m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1a9ea7d7e85249bcb7eca370e4eff972 : SUCCESS in 47m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13e24bf7e22a486fb968dc02041d6618 : SUCCESS in 4m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/16166fe03f3b4d308786d16dfaa27488 : SUCCESS in 3m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6194ed01c6514a7dbb8d08c0f0e814cb : SUCCESS in 4m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f5abd00af0c0477d9ab6f5b784356b47 : SUCCESS in 6m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3bd9b923fd3f4854aa665377210ab9fc : SUCCESS in 5m 19s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2e7fb4822a9745ef8bcc750b71d67428 : SUCCESS in 7m 41s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/83d1bd3d97da4017906e940aaae9aadc : SUCCESS in 7m 34s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/98c01b67b64044108a3ed48a2c0fdf6e : SUCCESS in 10m 05s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/8abae453b6894969b40dc7a0fb388833 : SUCCESS in 29m 44s","accounts_in_message":[],"_revision_number":2},{"id":"f314a717d2e07c514720ec2a1fb4eea0f8e441a8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-04-30 09:01:47.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"7d5c064fb7506ff0ee63b065dea9ed94480a951d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-30 09:35:34.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b02f3c4d57a24296a3fd24a308010c5c\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/de114444cb804219835143202cd7f5b4 : SUCCESS in 6m 25s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/be6260b189c1424d8b4e2f77ffa2aeb9 : SUCCESS in 6m 58s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/792797a18b034c3299035668940b21df : SUCCESS in 14m 38s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"752c2321c30fb61ddb9a126f716ae26a37b4e3ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-30 10:23:49.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0c5833323a054944bc12da1860c9b2fd\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/fefd1a5aa85542ad8bbae075bc178c4e : SUCCESS in 54m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b5d5ef927dc34de691275c0b0df0bad0 : SUCCESS in 1h 05m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/beb13d3da6314095ab8c59b2e669e68b : SUCCESS in 4m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bbb545e908014861aba5e6681003d9d8 : SUCCESS in 3m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/767831a0b4f243f6a21f013b959dec2a : SUCCESS in 2m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/dcca45cdd4254417b826b27dc72f45b4 : SUCCESS in 8m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7acde0697e8d479c8c5db4acbeaa05ee : SUCCESS in 4m 39s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6b4ceea05f7946dea300de6f0cabe661 : SUCCESS in 7m 29s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/e1be9b32ebbe41d18c519013ddd65f26 : SUCCESS in 9m 41s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/a0d070d8f4ca4102875e02238ce1da3a : SUCCESS in 13m 29s","accounts_in_message":[],"_revision_number":3},{"id":"2bfe5ea56db3c66586cac9a1fcc09fe3dd5d13a6","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-04-30 15:42:36.000000000","message":"Patch Set 3: Code-Review-1 Review-Priority+1\n\n(5 comments)","accounts_in_message":[],"_revision_number":3},{"id":"c0bb833d1a7c7072711aec5257db110b4dce0679","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-04-30 16:05:26.000000000","message":"Patch Set 3: Review-Priority+1\n\n(5 comments)","accounts_in_message":[],"_revision_number":3},{"id":"14089e9088a2d103d9f47c0e309eb3928bb53bb4","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-04-30 16:21:04.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"b0869ff27ae6e58d9fe9ea16d5cb9d9d507179f7","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-04-30 16:30:15.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d542631aa48feaf26dde6c9134936625f3339911","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-04-30 17:31:02.000000000","message":"Uploaded patch set 4.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"310e14c400c5e16136180f3203a6a7a411594860","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-04-30 17:32:26.000000000","message":"Patch Set 4: Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"f1fa0311933a410912a3c8c10890cda20fb8473b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-30 18:25:49.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/53979f3247a849b4a7f2dc11ff71ba49\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/69051b5b4b814cb59969c76131507b90 : SUCCESS in 7m 59s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/097a379f400d47009fa5c0901f794cb2 : SUCCESS in 8m 56s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e37e4dfcae38473cbacc30c94859e9c4 : SUCCESS in 15m 20s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"8144adeb35a8a19787ea6ca337cca10e223cfd3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-30 18:36:05.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9548ad9913c047e0a9cb83eb62b1dd98\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1287cfbbc52e403eba52d18ad856278a : SUCCESS in 28m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ec476deea6ff488bbe3e7b7e1b977b15 : SUCCESS in 59m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c689391014f459eab9c38ab1facc399 : SUCCESS in 2m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/185442ea4b04490b951bdb6b8a5b33cd : SUCCESS in 2m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/427e06d335d746969ea7a40d64c41c70 : SUCCESS in 5m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c257554bc07642aa9aa43e724145e3e9 : SUCCESS in 6m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fca8026e2e7541c4a15c6fbe95485445 : SUCCESS in 4m 40s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/69cb0d6eec484237bb74eff526c81f92 : SUCCESS in 7m 53s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/ab7934fec40e4954bfedc5a31d5df6a0 : SUCCESS in 11m 35s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/6cdac92d146a4fe29a537d0b8d245b6a : SUCCESS in 31m 52s","accounts_in_message":[],"_revision_number":4},{"id":"5ec665eed317f19a50317f99fad599c339fb45b4","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-05-05 14:34:40.000000000","message":"Patch Set 4: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":4},{"id":"b758e237e12c50855790a0fa3e464b234d1720d3","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-05-16 15:34:11.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d96a592a4ed23346e7ca559f871468c0f51e1da6","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-05-16 15:41:45.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"e2a38a39319364e464ef0fc0185a6bfbe172a7f9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-16 15:42:10.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"9351fb9560e6556e5d142029ca86b2c726a9b962","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-16 17:23:07.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e8f82d0682e4ec19f76fc8af9df09f6\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/010c8fbb29a14fa281b5fbd3cdf48c51 : SUCCESS in 31m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8d82bc6cd83943e686fe84e18894f815 : SUCCESS in 31m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/469b0a6620e24233bb5b4f8100772852 : SUCCESS in 6m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f2764b3ca89547afb4f78e58d1e00c45 : SUCCESS in 5m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a2aa4f90d5ec458aa7229657417af7ac : SUCCESS in 7m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ebc6ad75cc1941a4a1ba2371ba7f9d20 : SUCCESS in 7m 19s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c85b40f1467c4f679fab32d841a3adef : SUCCESS in 8m 02s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/b4d0d2e1e8e14cdb853efeac59ddd6eb : SUCCESS in 11m 06s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/a73a26c228a342bc9d879dcd799351bb : SUCCESS in 29m 29s","accounts_in_message":[],"_revision_number":4},{"id":"290da4e6689c2e1c876657348fd58adb69fc55b9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-16 17:23:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"38dcdc13033f1e2a583884d9b63662d27e701ef6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-16 17:24:39.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/38ae770cd82144858cfa83ec24d8d8bf\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc0625c5cb7d4a36aa37eee9b23dcede : SUCCESS in 50s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"e8da73973531d2a26afe3394faa8482a39787946","revisions":{"5932aa44b4b80646cce81ffc46907f8aa6cc40af":{"kind":"REWORK","_number":1,"created":"2025-04-27 10:31:53.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/92/948292/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/92/948292/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/92/948292/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/92/948292/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/92/948292/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/92/948292/1"}}},"commit":{"parents":[{"commit":"788172a6f7c7a214e87a80a27cb26ea33da67a58","subject":"Merge \"Add note about requirements lower bounds\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/788172a6f7c7a214e87a80a27cb26ea33da67a58"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-04-25 18:53:52.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-04-27 10:09:31.000000000","tz":0},"subject":"Functional tests migration test_images.TestImages","message":"Functional tests migration test_images.TestImages\n\n- Migrated tests to TestImagesSingleStore class\n- Refactored existing tests to reduce duplication\n- 1 Failing test kept as it is in test_images.TestImages to avoid\n  overall failure\n- Fixed other failing tests in test_images_api_policy due to refactoring\n\nChange-Id: Ie893070d9974c71c4f121e57db72a781c228420a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5932aa44b4b80646cce81ffc46907f8aa6cc40af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5932aa44b4b80646cce81ffc46907f8aa6cc40af"}]},"branch":"refs/heads/master"},"b3d13d3b0b1bf6e25f67ba39b152c7bc9dc994a7":{"kind":"REWORK","_number":2,"created":"2025-04-27 13:13:13.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/92/948292/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/92/948292/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/92/948292/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/92/948292/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/92/948292/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/92/948292/2"}}},"commit":{"parents":[{"commit":"788172a6f7c7a214e87a80a27cb26ea33da67a58","subject":"Merge \"Add note about requirements lower bounds\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/788172a6f7c7a214e87a80a27cb26ea33da67a58"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-04-25 18:53:52.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-04-27 12:44:11.000000000","tz":0},"subject":"Functional tests migration test_images.TestImages","message":"Functional tests migration test_images.TestImages\n\n- Migrated tests to TestImagesSingleStore class\n- Refactored existing tests to reduce duplication\n- 5 Failing test kept as it is in test_images.TestImages to avoid\n  overall failure\n- Fixed other failing tests in test_images_api_policy due to refactoring\n\nChange-Id: Ie893070d9974c71c4f121e57db72a781c228420a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b3d13d3b0b1bf6e25f67ba39b152c7bc9dc994a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b3d13d3b0b1bf6e25f67ba39b152c7bc9dc994a7"}]},"branch":"refs/heads/master"},"34d2de82c9d2406d39a3a319f2a53ea914b4997d":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2025-04-30 09:01:47.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/92/948292/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/92/948292/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/92/948292/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/92/948292/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/92/948292/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/92/948292/3"}}},"commit":{"parents":[{"commit":"55158e2ac1d2954782396adda2ddb0c78c233e57","subject":"Merge \"Fix vTPM metadefs.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/55158e2ac1d2954782396adda2ddb0c78c233e57"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-04-25 18:53:52.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-04-30 08:20:28.000000000","tz":0},"subject":"Functional tests migration test_images.TestImages","message":"Functional tests migration test_images.TestImages\n\n- Migrated tests to TestImagesSingleStore class\n- Refactored existing tests to reduce duplication\n- 5 Failing test kept as it is in test_images.TestImages to avoid\n  overall failure\n- Fixed other failing tests in test_images_api_policy due to refactoring\n\nChange-Id: Ie893070d9974c71c4f121e57db72a781c228420a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/34d2de82c9d2406d39a3a319f2a53ea914b4997d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/34d2de82c9d2406d39a3a319f2a53ea914b4997d"}]},"branch":"refs/heads/master"},"e8da73973531d2a26afe3394faa8482a39787946":{"kind":"REWORK","_number":4,"created":"2025-04-30 17:31:02.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/92/948292/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/92/948292/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/92/948292/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/92/948292/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/92/948292/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/92/948292/4"}}},"commit":{"parents":[{"commit":"55158e2ac1d2954782396adda2ddb0c78c233e57","subject":"Merge \"Fix vTPM metadefs.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/55158e2ac1d2954782396adda2ddb0c78c233e57"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-04-25 18:53:52.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-04-30 16:10:15.000000000","tz":0},"subject":"Functional tests migration test_images.TestImages","message":"Functional tests migration test_images.TestImages\n\n- Migrated tests to TestImagesSingleStore class\n- Refactored existing tests to reduce duplication\n- 5 Failing test kept as it is in test_images.TestImages to avoid\n  overall failure\n- Fixed other failing tests in test_images_api_policy due to refactoring\n\nChange-Id: Ie893070d9974c71c4f121e57db72a781c228420a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e8da73973531d2a26afe3394faa8482a39787946"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e8da73973531d2a26afe3394faa8482a39787946"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
