)]}'
{"id":"openstack%2Fglance~948295","triplet_id":"openstack%2Fglance~master~Ib0357b03f1cba668e954895d2c66ad1edc0b943b","project":"openstack/glance","branch":"master","topic":"eventlet-removal","attention_set":{},"removed_from_attention_set":{"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2025-04-30 10:29:37.000000000","reason":"\u003cGERRIT_ACCOUNT_9303\u003e replied on the change","reason_account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}}},"hashtags":[],"change_id":"Ib0357b03f1cba668e954895d2c66ad1edc0b943b","subject":"Functional tests migration test_images","status":"ABANDONED","created":"2025-04-27 12:16:33.000000000","updated":"2025-06-20 05:51:36.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"0e66cd3b834e3a7b65b68211616cd90b3b9aa45b","_number":948295,"virtual_id_number":948295,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-04-30 11:30:32.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-27 12:32:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-04-27 13:06:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f149a658977d18f856bfc6cfc1ab3c4e482e1239","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-04-27 12:16:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c528f2941d9126724a9b7028bd9f515a8afb12f8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-27 12:32:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/71510a8ff38f4770b9629eae9ad1ca38\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/aca9887df3c34c7681c5559290e210a3 : SUCCESS in 8m 05s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/166a32a43dd14295bbbe323b67361425 : SUCCESS in 10m 14s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f047b981691346feb3c75f3b9035b782 : SUCCESS in 13m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9f937708898ab65d3674c0515d57b1b9e65c226e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-27 13:06:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4d6e6cee8b1a4f539eb0a354b8e31346\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/81124cdbfca740af9c74104a6e9f224e : SUCCESS in 39m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8d5e2700fbbf4583a01796df943c57e3 : SUCCESS in 47m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41b5c02b2fec4dfc974f7f1807b310bc : SUCCESS in 4m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0e3c14c5a2234e9c82344c15e8b98012 : SUCCESS in 4m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fd6cb05fffc14aa2926acbc1c85f17a7 : SUCCESS in 5m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b723015a42214b4ba2f1c25a76dca709 : SUCCESS in 9m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4ed36616754451e8fc5471f05742732 : SUCCESS in 5m 23s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7293e92dd7eb48aa8088fbdda3e2573b : SUCCESS in 7m 45s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/093d860910a24a7ca4f1df42867d9490 : SUCCESS in 5m 28s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/cd5cd275b79a47bda5e85d6ace5a6b9b : FAILURE in 10m 41s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/239f921abb944464aa7eba4053f594d4 : SUCCESS in 20m 08s","accounts_in_message":[],"_revision_number":1},{"id":"3066ff96cf93cb7f72353ad18f41d179eee0db8e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-04-27 13:13:13.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"b176d4acdc87c7415bc48cc5ca54fd58325560e6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-27 13:27:56.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/528d358b052d4cf5b57e5f0e2bd77f9f\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/db54fb4f9dea4690a87affd613bb75b2 : SUCCESS in 6m 55s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/cf60ddc2943b405fbb23244d17492ec6 : SUCCESS in 7m 05s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/0c1bd856a2124f6a84426c035182ad1b : SUCCESS in 12m 52s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9e121a173e5e24650657627f22ed955655143cab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-27 14:11:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d32f6c47860f44bda73e7b562e90f1d2\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/bd3ace0f19f44ab1a645bcbce6431945 : SUCCESS in 41m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/39fde71788d647dfa9bcc9174489d1e1 : SUCCESS in 48m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00bfd9bd1ab9467c9d0287ee50385b76 : SUCCESS in 4m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/394313df6566466a8e96d1a11333b8c0 : SUCCESS in 4m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/022a3ad98f024e32bf867598b5915f4f : SUCCESS in 4m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/061b30375e544ee3ba977eadebb9c2d7 : SUCCESS in 6m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83cb71c767d541eeaf3fe8af91e3e00e : SUCCESS in 5m 27s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e296615c3c6f4705a6a1c5797209cb96 : SUCCESS in 8m 54s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ad4ba974d913400fb11bfd5785300466 : SUCCESS in 7m 47s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/bbceb763e3e84c0ab7f99c388d662553 : SUCCESS in 10m 11s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/c08a7504c7d24a519f2fe17489fb9fe4 : SUCCESS in 23m 58s","accounts_in_message":[],"_revision_number":2},{"id":"0d407e106029ba696ded3f0cf2b327b88c324b7f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-04-30 09:01:47.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"9200b081911bd4aef399ca324ee23b04ac4796d9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-30 09:35:42.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac1d87731a0448baa272f530a1dab4ca\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/47dafe3b33364da394c6054ffe9a2eff : SUCCESS in 8m 44s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/79597f055fea4c30afcaa3be38af9018 : SUCCESS in 6m 31s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/5194de1f00c94091af49ff70714f36a7 : SUCCESS in 9m 34s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"915cd4fd52fff7cc05d83deda6b89fc504f7684e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-30 09:37:09.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cf45849764fe422bafd99b1db390cc3f\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e2acf18c12d947dbbc390fe1752886b2 : SUCCESS in 26m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e4295f9a56e64c39afd8780c0d90d654 : POST_FAILURE in 11m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb6bba0a54be48d59a5f8db7a05570e5 : SUCCESS in 4m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ce9356386321456895c4724809034ebb : SUCCESS in 3m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/eb137000e26e43b9bb32e487cb453614 : SUCCESS in 5m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cb23a8e465f344028e8c26cd27f32948 : SUCCESS in 7m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a272e799de5e4665b6d9255e83405511 : SUCCESS in 4m 49s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ff090385079845c9aca9a9ff298bb886 : SUCCESS in 7m 54s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/9e6631ecac414cc999eae43495161e78 : SUCCESS in 10m 04s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/11718f6918f440ae8a19f7b0525b19d5 : SUCCESS in 30m 23s","accounts_in_message":[],"_revision_number":3},{"id":"87b60de9906bed6a908a04ffe4da4dd5d9297ca5","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-04-30 10:29:37.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"56f4ee94243686865ff1ea1ff086f5766c79d96d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-30 11:09:11.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3bd5a275b3d94462a80bdd13eff3c4b8\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6227affe887743f782964136d55a2a98 : SUCCESS in 9m 59s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/56d2cf5edba54f92909c41a956cebc01 : SUCCESS in 11m 33s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/766bc8173b9a4496b29e59cc1e27f1a4 : SUCCESS in 14m 29s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"55e425b52f131f265494868289b4349d2d5b5dd3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-30 11:30:32.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b6733e3502664e8998e68bac645709f6\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a1c45a3437054cf1be3624d869fb4e97 : SUCCESS in 51m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1fe007c73aa74ba3870812e922e6a874 : SUCCESS in 52m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f45171ab2121441b9c59026d9895d139 : SUCCESS in 4m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d0fa0a89e3ee4b64bc994b0d4ed1367b : SUCCESS in 4m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/23f2ae15521b4a3994672787e889b57c : SUCCESS in 5m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3fc5c9d9338e486fbd799c6ea4132a82 : SUCCESS in 6m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ce30a5e4d7d425caea5ea5b4900e207 : SUCCESS in 4m 43s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/93787171537d44b9a24cc2aa79334f24 : SUCCESS in 8m 12s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/e6d63fce41534ebdbec54d8ec1a5dbb8 : SUCCESS in 9m 23s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/6b6ac1890d534f65a916ac05244b015c : SUCCESS in 29m 54s","accounts_in_message":[],"_revision_number":3},{"id":"0e66cd3b834e3a7b65b68211616cd90b3b9aa45b","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-06-20 05:51:36.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"1e8ffaef0dccf202267ccca9d32235504323b42d","revisions":{"1a60fc5eac446678e8011605e5f0d1da7989122f":{"kind":"REWORK","_number":1,"created":"2025-04-27 12:16:33.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/95/948295/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/95/948295/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/948295/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/948295/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/948295/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/95/948295/1"}}},"commit":{"parents":[{"commit":"5932aa44b4b80646cce81ffc46907f8aa6cc40af","subject":"Functional tests migration test_images.TestImages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5932aa44b4b80646cce81ffc46907f8aa6cc40af"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-04-27 11:55:46.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-04-27 11:55:46.000000000","tz":0},"subject":"Functional tests migration test_images","message":"Functional tests migration test_images\n\n- TestImageDirectURLVisibility\n- TestImageLocationSelectionStrategy\n- TestCopyImagePermissions\n\nChange-Id: Ib0357b03f1cba668e954895d2c66ad1edc0b943b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1a60fc5eac446678e8011605e5f0d1da7989122f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1a60fc5eac446678e8011605e5f0d1da7989122f"}]},"branch":"refs/heads/master"},"7ab526f2b487cd72a45eb9306f0811b42b3030ea":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-04-27 13:13:13.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/95/948295/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/95/948295/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/948295/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/948295/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/948295/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/95/948295/2"}}},"commit":{"parents":[{"commit":"b3d13d3b0b1bf6e25f67ba39b152c7bc9dc994a7","subject":"Functional tests migration test_images.TestImages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b3d13d3b0b1bf6e25f67ba39b152c7bc9dc994a7"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-04-27 12:52:55.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-04-27 12:52:55.000000000","tz":0},"subject":"Functional tests migration test_images","message":"Functional tests migration test_images\n\n- TestImageDirectURLVisibility\n- TestImageLocationSelectionStrategy\n- TestCopyImagePermissions\n\nChange-Id: Ib0357b03f1cba668e954895d2c66ad1edc0b943b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7ab526f2b487cd72a45eb9306f0811b42b3030ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7ab526f2b487cd72a45eb9306f0811b42b3030ea"}]},"branch":"refs/heads/master"},"1e8ffaef0dccf202267ccca9d32235504323b42d":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2025-04-30 09:01:47.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/95/948295/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/95/948295/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/948295/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/948295/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/948295/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/95/948295/3"}}},"commit":{"parents":[{"commit":"34d2de82c9d2406d39a3a319f2a53ea914b4997d","subject":"Functional tests migration test_images.TestImages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/34d2de82c9d2406d39a3a319f2a53ea914b4997d"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-04-27 12:52:55.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-04-30 08:20:57.000000000","tz":0},"subject":"Functional tests migration test_images","message":"Functional tests migration test_images\n\n- TestImageDirectURLVisibility\n- TestImageLocationSelectionStrategy\n- TestCopyImagePermissions\n\nChange-Id: Ib0357b03f1cba668e954895d2c66ad1edc0b943b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1e8ffaef0dccf202267ccca9d32235504323b42d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1e8ffaef0dccf202267ccca9d32235504323b42d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
