)]}'
{"id":"openstack%2Fglance~948455","triplet_id":"openstack%2Fglance~master~If5edfb9e8ce7f28fcc60b86d431918a973c5c689","project":"openstack/glance","branch":"master","topic":"eventlet-removal-multi","attention_set":{},"removed_from_attention_set":{"33765":{"account":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"last_update":"2025-05-27 17:09:50.000000000","reason":"Change was submitted"},"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2025-05-27 17:09:50.000000000","reason":"Change was submitted"},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2025-05-27 16:04:11.000000000","reason":"\u003cGERRIT_ACCOUNT_8122\u003e replied on the change","reason_account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}}},"hashtags":[],"change_id":"If5edfb9e8ce7f28fcc60b86d431918a973c5c689","subject":"Functional tests migration of TestImageMultipleBackend","status":"MERGED","created":"2025-04-29 15:57:32.000000000","updated":"2025-05-27 17:10:48.000000000","submitted":"2025-05-27 17:09:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"948455-eventlet-removal-multi","meta_rev_id":"276285d702c18b0f708c584135029e179c6aaff4","_number":948455,"virtual_id_number":948455,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-27 17:09:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2025-05-27 16:04:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2025-05-27 16:04:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2025-05-21 16:19:53.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"}],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-29 16:12:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-04-29 17:05:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-04-29 18:30:08.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"state":"REVIEWER"},{"updated":"2025-04-29 18:30:08.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"}],"messages":[{"id":"41a1601c3d3a549ad907d215bcebcc9c4438b995","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-04-29 15:57:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3c938bce2bce6f0370c45d7d3bfb82b3f50b8935","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-29 16:12:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6877cae4e69749d98d96cf0cea5d4557\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1eaebab80bec4a01ac715c4c926d544a : SUCCESS in 8m 14s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/26a5f3d747754527b18fe91a78349128 : SUCCESS in 6m 44s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/29789b23feba4e7e86c9cc80516b8ca2 : SUCCESS in 13m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"92255e5c918cfe31eadce75b0350b973772b0ccc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-29 17:05:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8da574f6d10a4feca1fba857d8d4d6f6\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/812b2f1047914909b7e6b58b803cd242 : SUCCESS in 57m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ed86a159623c42e6a8cb5c6a5c567d08 : SUCCESS in 1h 04m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5d4f20786cc488a986f0362ce05d55f : SUCCESS in 5m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2c15be4690a24e43a81df6325de2e0c7 : SUCCESS in 4m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/506c7294f898472a96d7c5445c48c98e : SUCCESS in 5m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1f0fef421a3d4ff4b268a371a1207283 : SUCCESS in 7m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5ecc6f4df3247cb93902dca92360f54 : SUCCESS in 5m 13s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a514c82b9e3c4613b8458c0a8c5e688c : FAILURE in 2m 55s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8003225781834ba0990596b1bac0b80c : SUCCESS in 6m 56s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/fc2cac576a24430cb2d868adad3a3007 : SUCCESS in 11m 19s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/48ff7059b4734645bd0f80c7c58a8be0 : SUCCESS in 32m 17s","accounts_in_message":[],"_revision_number":1},{"id":"982ae0320849151c61dad98bd53a255c20906ae6","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-04-29 18:30:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f304a9374d914759c055c0ab43617a915cb1fdb9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-29 18:42:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ae30c0b8858f4606a03fa91e33ed842c\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/96bdd652acaa48b2aae29f021e2539af : SUCCESS in 6m 30s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5422bed1066f47369dd3bac13d0f436b : SUCCESS in 6m 54s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/9869de03798441aea10f78b7dcfa05aa : SUCCESS in 9m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"53efd01447325349db19825e610cb20f518b230b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-29 19:33:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/747226d2678549ae8668d54a989fffe5\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d1b5a74f5ac74a82bbc2dff063676ef4 : SUCCESS in 28m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b249c510090e4aed94d1349966275d5b : SUCCESS in 58m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/208e1dd96a7e4b86b8314b833fbe1164 : SUCCESS in 4m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0e145cac5da5484bb6af08eb7bd58035 : SUCCESS in 4m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a6df4a5971244b029528d72ca9b91b70 : SUCCESS in 4m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9540e3c013dd4095b3761f5f25094cf6 : SUCCESS in 7m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b50fb8111e5f4e41bed1558251093b69 : SUCCESS in 5m 19s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5f8c400d4ee54be9a23a5391721ea46d : SUCCESS in 5m 37s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/5e629bbc034545848954798b5a62ea6b : SUCCESS in 9m 43s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/04e5ccfca97c47cc9b6ffcb99766e59d : SUCCESS in 24m 25s","accounts_in_message":[],"_revision_number":1},{"id":"697a8e19b717a6a4ce517050d5e0981af94a22c9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-04-30 09:01:47.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"f452a8949aee6768ea4274054d0571ad6d70d058","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-30 09:55:11.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c657604d5910457b8a16df638189a5d3\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/42f77a0eb1dd4a60b6cedbc76a7c2b3b : SUCCESS in 6m 00s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/85f4c79ff8e64d54a628f96bab02e709 : SUCCESS in 5m 24s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/7fa16c49ed694a4f8ddaf4529dd23dca : SUCCESS in 10m 10s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"44945f81b69c7a338fa1ec25af6122d25cf89bd2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-30 10:10:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/778eccc326dc4ed5ab3c1ac7741a5b1d\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b6d125473bff41ac948840cd26eccdab : SUCCESS in 1h 00m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ad321386b3624b549792bd614b86f5b8 : SUCCESS in 1h 01m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a294bf83bd5471b99fb0c3701de1b3c : SUCCESS in 4m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b95d351509814a81b0f40d4dbf7f8bf4 : SUCCESS in 4m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/84adeca1135d43a397e9b84087758f6d : SUCCESS in 6m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6ee550ff7e1941a594e76a4cd4b6eaec : SUCCESS in 7m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42d51dfaeb404710be4d430f51ccb5bc : SUCCESS in 5m 39s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ec901e6807f34c348ad4b205111fa1be : SUCCESS in 7m 36s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/3999c62789a84d13b6af20b047de707b : SUCCESS in 9m 19s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f7a7dc2d586f4866afbafcdb3760a577 : SUCCESS in 32m 22s","accounts_in_message":[],"_revision_number":2},{"id":"a2672c3351efb4629174396e3cd2272ce12a03c8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-05-20 07:03:16.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"cb6d9e128bbf513b10bfedb0a17343d12d1b21c3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 07:19:56.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/13434029c8704cc8a2d35c69f7558dca\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/19632579659a4428bfb66d4fdd1acddb : SUCCESS in 8m 12s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/063c1a19addb4b009cda4ac20ee6fdb6 : SUCCESS in 10m 21s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/30aee193c42a4389a3ee10a1cc0a77a4 : SUCCESS in 14m 37s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"2dd373719af129fa9bfcfcfebe0a081532c20507","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-05-20 07:43:57.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"8c77beb90f0e73acf1c4296088bba4a02ff1e427","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 07:58:16.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/555ebf35002b4f6abdba0b5799999527\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6345e224e4134869b57d89c0f39ee822 : SUCCESS in 5m 21s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9dcd1fd76aec457093e5aeb1f46281e8 : SUCCESS in 6m 19s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b25baf50ca54486fbbecc7e174b63201 : SUCCESS in 14m 06s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"9161ba3b239a93689310f7740576a6d32fdacb71","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 08:47:50.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/677245c916694522b79414c69255f9f7\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8053f9651a864f6280490f2154688aba : SUCCESS in 59m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ed1cb3a4a6694712975de5aeede8732e : SUCCESS in 58m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0355f5e59b11479e85efab3d78fe1eb5 : SUCCESS in 4m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7bfb20a2e7ac483eba78c2c3ff542407 : SUCCESS in 4m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5d4859d20bdb4faf895785e6e398598d : SUCCESS in 5m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e2199c0fa4014f85aa8ea0fc84fdc57b : SUCCESS in 9m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73e71f93ae00491aace5a4b934bb4703 : SUCCESS in 6m 50s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8f1404dc48af416d93b6e2f57c3e9cb2 : SUCCESS in 7m 56s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/eb09c9ed341d4c3eb431f9633d6f5be5 : SUCCESS in 11m 43s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/21bb14ff49af460ab7409db5c2a5842e : SUCCESS in 30m 10s","accounts_in_message":[],"_revision_number":4},{"id":"d231728e597e6d1546b3bd817dfe641bfbf91967","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-05-21 16:19:53.000000000","message":"Patch Set 4: Review-Priority+1","accounts_in_message":[],"_revision_number":4},{"id":"348537f51051f0bd806c6cbddaa47c1b6a232c27","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-05-27 15:54:31.000000000","message":"Patch Set 4:\n\n(4 comments)","accounts_in_message":[],"_revision_number":4},{"id":"9f598564a03ff40cc04822ab1720fbc502f06749","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-05-27 15:57:07.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"82b8da47e5dc3d22460b3a70fb63cdd789307803","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-05-27 15:57:53.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"cd5f03c7ef5f46f275c55f4cc5efd951b3cfba9f","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-05-27 16:04:11.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"2b47bf32d63f1fbc8f6f9b57a9f15a6b6465a2fb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-27 16:04:36.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"42d219245709f996f5182a5be15f26d70f515cd8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-27 17:09:50.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3bb60766e7394c61a43a0f23431fdd59\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4328a05dcdd443bda1a4152c80904d97 : SUCCESS in 1h 00m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b64aa026a45342cb89d4fa0cac8a432b : SUCCESS in 55m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/731b47b7e8274818952c05c1c7c1aef8 : SUCCESS in 5m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/79de8482edd2453a81b746a5b6e4b24c : SUCCESS in 6m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c2adab8d5c20460b9b596fdde0ecc63f : SUCCESS in 5m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c01b669f1e4941edacc02df33aa78692 : SUCCESS in 4m 47s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f1e43fc768c242b0be6b1a9b88d9eb62 : SUCCESS in 7m 00s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/9b401dda39614b24b7c9bb866ff672b0 : SUCCESS in 11m 47s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/d68dc8389fac4d449c16289d17d20c50 : SUCCESS in 37m 18s","accounts_in_message":[],"_revision_number":4},{"id":"0267db8f72eb79e0e962bd43656ff4d8c8ec092f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-27 17:09:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"276285d702c18b0f708c584135029e179c6aaff4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-27 17:10:48.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/394895cbaba6440eba58c4f923dda979\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03fcb09db5434157b6be61972c6e7297 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"7747b2d40a0b6cfac0dd9b9c5456d315814d2d90","revisions":{"aac9706c70b5385817ebd296826e4fa7364248f3":{"kind":"REWORK","_number":1,"created":"2025-04-29 15:57:32.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/55/948455/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/55/948455/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/948455/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/948455/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/948455/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/55/948455/1"}}},"commit":{"parents":[{"commit":"d6afd0e54e78ced4ad770cb63b722d0cec20f91e","subject":"Functional tests migration of image members tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d6afd0e54e78ced4ad770cb63b722d0cec20f91e"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-04-29 09:01:38.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-04-29 15:37:16.000000000","tz":0},"subject":"Functional tests migration of TestImageMultipleBackend","message":"Functional tests migration of TestImageMultipleBackend\n\n- Moved all tests from above class to test_multi_store_images.py\n- Refactored tests to use common helper methods\n- Later we can move other multi store related tests here\n\nChange-Id: If5edfb9e8ce7f28fcc60b86d431918a973c5c689\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/aac9706c70b5385817ebd296826e4fa7364248f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/aac9706c70b5385817ebd296826e4fa7364248f3"}]},"branch":"refs/heads/master"},"6bca700b241e2a0032cd7de1661ab7e59a34a124":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-04-30 09:01:47.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/55/948455/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/55/948455/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/948455/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/948455/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/948455/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/55/948455/2"}}},"commit":{"parents":[{"commit":"fbaabc452d277844801811efca762fe21ea430b2","subject":"Functional tests migration of image members tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fbaabc452d277844801811efca762fe21ea430b2"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-04-29 09:01:38.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-04-30 08:22:01.000000000","tz":0},"subject":"Functional tests migration of TestImageMultipleBackend","message":"Functional tests migration of TestImageMultipleBackend\n\n- Moved all tests from above class to test_multi_store_images.py\n- Refactored tests to use common helper methods\n- Later we can move other multi store related tests here\n\nChange-Id: If5edfb9e8ce7f28fcc60b86d431918a973c5c689\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6bca700b241e2a0032cd7de1661ab7e59a34a124"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6bca700b241e2a0032cd7de1661ab7e59a34a124"}]},"branch":"refs/heads/master"},"8905e511268ed31a08a392d9c13867ca6c34eb75":{"kind":"REWORK","_number":3,"created":"2025-05-20 07:03:16.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/55/948455/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/55/948455/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/948455/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/948455/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/948455/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/55/948455/3"}}},"commit":{"parents":[{"commit":"faccbed1d9291808be563c22784467a65f1f8bd1","subject":"Functional tests migration of image members tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/faccbed1d9291808be563c22784467a65f1f8bd1"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-04-29 09:01:38.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-05-20 06:41:53.000000000","tz":0},"subject":"Functional tests migration of TestImageMultipleBackend","message":"Functional tests migration of TestImageMultipleBackend\n\n- Moved all tests from above class to test_multi_store_images.py\n- Refactored tests to use common helper methods\n- Later we can move other multi store related tests here\n\nChange-Id: If5edfb9e8ce7f28fcc60b86d431918a973c5c689\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8905e511268ed31a08a392d9c13867ca6c34eb75"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8905e511268ed31a08a392d9c13867ca6c34eb75"}]},"branch":"refs/heads/master"},"7747b2d40a0b6cfac0dd9b9c5456d315814d2d90":{"kind":"REWORK","_number":4,"created":"2025-05-20 07:43:57.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/55/948455/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/55/948455/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/948455/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/948455/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/948455/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/55/948455/4"}}},"commit":{"parents":[{"commit":"faccbed1d9291808be563c22784467a65f1f8bd1","subject":"Functional tests migration of image members tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/faccbed1d9291808be563c22784467a65f1f8bd1"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-04-29 09:01:38.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-05-20 07:22:34.000000000","tz":0},"subject":"Functional tests migration of TestImageMultipleBackend","message":"Functional tests migration of TestImageMultipleBackend\n\n- Moved all tests from above class to test_multi_store_images.py\n- Refactored tests to use common helper methods\n- Later we can move other multi store related tests here\n\nChange-Id: If5edfb9e8ce7f28fcc60b86d431918a973c5c689\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7747b2d40a0b6cfac0dd9b9c5456d315814d2d90"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7747b2d40a0b6cfac0dd9b9c5456d315814d2d90"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
