)]}'
{"id":"openstack%2Fglance~948515","triplet_id":"openstack%2Fglance~stable%2F2025.1~I698dbd2946555c0a6455b1be17a6bf4a8d407ee7","project":"openstack/glance","branch":"stable/2025.1","topic":"bug/2106223-stable/2025.1","attention_set":{},"removed_from_attention_set":{"34034":{"account":{"_account_id":34034,"name":"Franciszek Przewoźny","display_name":"Franciszek Przewozny","email":"przewozny.franciszek@gmail.com","username":"fprzewozn","status":"OpenStack Engineer / SRE @ Opera Software"},"last_update":"2025-05-22 18:07:17.000000000","reason":"Change was submitted"},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2025-05-20 20:58:38.000000000","reason":"\u003cGERRIT_ACCOUNT_8122\u003e replied on the change","reason_account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}}},"hashtags":[],"change_id":"I698dbd2946555c0a6455b1be17a6bf4a8d407ee7","subject":"Fix vTPM metadefs.","status":"MERGED","created":"2025-04-30 06:16:41.000000000","updated":"2025-05-22 18:08:08.000000000","submitted":"2025-05-22 18:07:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"948515-bug/2106223-stable/2025.1","cherry_pick_of_change":946343,"cherry_pick_of_patch_set":2,"meta_rev_id":"839aae31506981ca839238abbd24d2ec72909989","_number":948515,"virtual_id_number":948515,"owner":{"_account_id":34034,"name":"Franciszek Przewoźny","display_name":"Franciszek Przewozny","email":"przewozny.franciszek@gmail.com","username":"fprzewozn","status":"OpenStack Engineer / SRE @ Opera Software"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-22 18:07:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":2,"date":"2025-05-20 20:58:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":1,"date":"2025-05-20 20:58:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-30 06:17:07.000000000","updated_by":{"_account_id":34034,"name":"Franciszek Przewoźny","display_name":"Franciszek Przewozny","email":"przewozny.franciszek@gmail.com","username":"fprzewozn","status":"OpenStack Engineer / SRE @ Opera Software"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2025-04-30 06:29:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-04-30 08:45:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a0225181372a0bd0f0db8ec2818c055df53dd6df","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34034,"name":"Franciszek Przewoźny","display_name":"Franciszek Przewozny","email":"przewozny.franciszek@gmail.com","username":"fprzewozn","status":"OpenStack Engineer / SRE @ Opera Software"},"date":"2025-04-30 06:16:41.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"8f7be964babd0a1068186b80e3bf1cdde4639b95","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-30 06:29:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/38f2dde1de5c4e7a8279ebfb8b66d8fc\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b6b006566712481288c5e9d3e72ff608 : SUCCESS in 10m 33s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8311644d7f614a3bba481de3351bab38 : SUCCESS in 10m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d2e2e05772ddfea2b18afce915485d6d15e4e4db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-30 08:45:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8eab65aefc9b451585e3e6aaffdab542\n\n- grenade https://zuul.opendev.org/t/openstack/build/67556dd07eb346c6bdd5b3b8a068e774 : SUCCESS in 1h 06m 19s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7efe16885d034fdebaf65907018f92cc : SUCCESS in 57m 25s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f844236f33a34edc9350003f337d5603 : SUCCESS in 1h 31m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b94bd1134eeb4919ad0cfa6112411282 : SUCCESS in 1h 06m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73c12786c9ef4df88335f5a1a77d8f2b : SUCCESS in 4m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ace3c805838c4f5586007e86d3378e7d : SUCCESS in 4m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3a39fb20b6cc4bb1bb512e49ce831c8c : SUCCESS in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09eedacfbe704ca5adce11dc43d5878d : SUCCESS in 4m 29s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/38be084873a94cdab9dee747f8017968 : SUCCESS in 9m 09s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8e4fe6e867ce455f945f437ba55c141e : SUCCESS in 9m 17s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/ac190bbcb6e144b0a005a91cb4545f72 : SUCCESS in 11m 41s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/c4b74ff0fbe6425da4823210657b2e9b : SUCCESS in 1h 55m 21s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/975a8fba108a421092c0a5476af91dcd : SUCCESS in 1h 50m 58s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/d76a670d6b5a45b287667d479b0ed076 : SUCCESS in 2h 08m 27s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/5b2ef5aa99784b39b642d7fa74a303c3 : SUCCESS in 2h 03m 09s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/dd3d5a5145ca40f4a2f844b9a23a4c5f : SUCCESS in 2h 23m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/710680a11fea4de796a6772891c47642 : SUCCESS in 1h 03m 48s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f89df8d061d64ba2a6bdc1d3bc54db3e : SUCCESS in 1h 45m 59s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/d198d72134144ef084ce480be897fe1e : SUCCESS in 1h 49m 07s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/12408670da5f466ba28e8d11570761c4 : SUCCESS in 55m 33s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/2a21da330c0743b5a859c95d61dcd561 : SUCCESS in 1h 47m 42s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/62ac85065c6640edabf5d1cccca095ed : SUCCESS in 30m 11s","accounts_in_message":[],"_revision_number":1},{"id":"aa33f2cca60bfe1109879d141372555b6bd33187","author":{"_account_id":34034,"name":"Franciszek Przewoźny","display_name":"Franciszek Przewozny","email":"przewozny.franciszek@gmail.com","username":"fprzewozn","status":"OpenStack Engineer / SRE @ Opera Software"},"date":"2025-05-09 11:12:12.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4c0fc249135ee9a945782a79103880286181ffe7","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-05-20 20:58:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d9d0b977e643a67988f057475993dd0b207c89f1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 20:59:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c5a28073038ff9f40b65f681f133fa1ff30c9532","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 23:08:17.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fed21a41fcd24a1a93aa273391c32790\n\n- grenade https://zuul.opendev.org/t/openstack/build/0f3bcfa5318244d09d0f83785d7acc64 : SUCCESS in 52m 32s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/fa66412bffb84857b264ea1f19da6323 : SUCCESS in 25m 31s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/286c79f3de044b4dabeb852e973f4291 : SUCCESS in 1h 27m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/52c7220999d64387b60494f6e4d941ea : SUCCESS in 58m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4acc972ee454873a734b26e0738d3c6 : SUCCESS in 3m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/09495787cf6d4f7fa528a08d82e1d597 : SUCCESS in 4m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/22e32ad48c73434b92ba471debc7c321 : SUCCESS in 4m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a7432928d3e4b4fb09598ed0e2e8df2 : SUCCESS in 4m 47s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/cd65e806d33045e19ca743558a6c0a18 : SUCCESS in 8m 05s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3119a577b0544876a1ae649563d0aaca : FAILURE in 8m 27s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/4977616b7bcb42adbb504c1f5a7a56e4 : SUCCESS in 14m 18s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/449859c7d8fa4b4ea62e92dbb8f74aca : SUCCESS in 2h 00m 51s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/527cfac62294403f9911e8e1f8f9b8e2 : SUCCESS in 2h 04m 49s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/e73f50f803e54100b19f111759810423 : SUCCESS in 1h 55m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5afb65a54c284d42a2acaa40941efd48 : SUCCESS in 29m 31s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/cbc32162b4c34941948b40675fe47688 : SUCCESS in 1h 47m 31s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/c1946b2188f1467db78d126f1d216da6 : SUCCESS in 31m 59s","accounts_in_message":[],"_revision_number":1},{"id":"cdcec3e96782480cfe49a8617a83591f1a645a9e","author":{"_account_id":34034,"name":"Franciszek Przewoźny","display_name":"Franciszek Przewozny","email":"przewozny.franciszek@gmail.com","username":"fprzewozn","status":"OpenStack Engineer / SRE @ Opera Software"},"date":"2025-05-21 04:25:18.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"76711fdc43eb5e25a3c26dd88e3a1de7668a0847","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-21 04:35:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be28b85e56114ee7bd973d7aca8cda7e\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7185f80442904e0d9988971245de1fdd : SUCCESS in 6m 37s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8e6ddf57bdcc4c0381f2d1db94d95e66 : SUCCESS in 8m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3cf9ea238acf196f9599df7bfdbc8edc47be32b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-21 06:26:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d2d51f9b4984183a11f5f2d121ca069\n\n- grenade https://zuul.opendev.org/t/openstack/build/decd8714d7864e63a5850759e5690dda : SUCCESS in 50m 36s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c58340149c49414a997a7ca0dd332845 : SUCCESS in 45m 27s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c29cba689fe149ed893ab09fe1bbfd6e : SUCCESS in 1h 23m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/650c6a83ab4641d8a7acec1a3185969f : SUCCESS in 55m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/906e9db204e64235a21a68cf405a2d05 : SUCCESS in 5m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0e3c7fd39ca6457e8abd664c8e0262a0 : SUCCESS in 4m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/200e3e93b6e9411cb28ec53721e864d4 : SUCCESS in 5m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cde735975c574b91af46ef7d07baac37 : SUCCESS in 5m 58s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e665150654f94d10b19afac04d77f2b5 : SUCCESS in 8m 30s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a21cff089af94c7aa01a2c434ab1fd03 : SUCCESS in 5m 44s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/0e6bd8a3d0854088809cebdb88ec55f2 : SUCCESS in 10m 14s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/7756b92ed51f4188895c39dd07d7573e : SUCCESS in 1h 46m 17s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/7a29befa70af45239082674a91466404 : SUCCESS in 1h 50m 17s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/37b4f1b42ced48c28d79b02672ed89ce : SUCCESS in 1h 08m 29s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/9f8d8421c8fa4ee389b3b4c35c3878f5 : SUCCESS in 1h 56m 31s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/78ebe07df8b742cbb781803d38806188 : SUCCESS in 1h 54m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6b7fc88cbb904c14905ffa350b624417 : SUCCESS in 28m 23s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/0c722129e75d4d7b9b09132bf03d9399 : SUCCESS in 1h 53m 48s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/3c76afb9a4c74d8c9bf473d88cfcdde1 : SUCCESS in 1h 44m 05s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/8e8bf87c0b064f09b2f864eecc9ee995 : SUCCESS in 55m 47s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/495a9cb42b4e4a09a9697520e6ec926e : SUCCESS in 1h 46m 52s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/31f765932ccb4ca681f6017715def01f : SUCCESS in 28m 49s","accounts_in_message":[],"_revision_number":1},{"id":"29f086193bcdeb5c4e43617e17de6df030c615c5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-21 06:27:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2a324f96d0ccb290300811b84e909bf382a4cf1a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-21 08:42:53.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d7e04f6596e64b3d9b06acdc08d2e5fc\n\n- grenade https://zuul.opendev.org/t/openstack/build/a91b3af2ac79470ebc46f2daaecfc20d : SUCCESS in 30m 32s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e26379e8332e4f9b99180ec4e0f91706 : SUCCESS in 25m 53s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a70f7cc12edc4a309276d902cb18a921 : SUCCESS in 1h 18m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3889ce90a65d4797b4f2d276ddafa407 : SUCCESS in 31m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb742da85a0f4088962f885c1026facf : SUCCESS in 3m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/75e18a84f62d43a893bc0d7a0d7702f2 : SUCCESS in 4m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e8de9e045be14cba97f206b5a3054819 : SUCCESS in 5m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9e1d8dc20494dd5b2af80635b2c6c50 : SUCCESS in 5m 26s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/fc8249e76fba488c8d30614a5da7154d : SUCCESS in 8m 05s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e9a7418c15a94c2592435c1f7bef054c : SUCCESS in 8m 34s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/42774e00829d4c9b8da7ae7c3ec0d7bb : SUCCESS in 11m 26s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/54a67a8d4c4c467b9fe955bcde0cf85e : SUCCESS in 2h 09m 23s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/f0ec91a16c514efc9d1cc74a0682c342 : SUCCESS in 1h 54m 41s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/16df37400a85469e84821d03f5f7e6e1 : SUCCESS in 2h 15m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9e3f7fc321134727885ae487898f3297 : SUCCESS in 49m 57s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/2df1bbb45fff4213b07b1af65dd6df4b : FAILURE in 12m 30s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f097729cfdef4c8d84bd14d46037c6df : SUCCESS in 21m 19s","accounts_in_message":[],"_revision_number":1},{"id":"3423a19e7afa28b044023ee1803d31cea9ec5f88","author":{"_account_id":34034,"name":"Franciszek Przewoźny","display_name":"Franciszek Przewozny","email":"przewozny.franciszek@gmail.com","username":"fprzewozn","status":"OpenStack Engineer / SRE @ Opera Software"},"date":"2025-05-21 08:43:23.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b72e365081a7ce95eac354822363e22f73823dd9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-21 08:54:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7b85a538c6304e5f888fd66af1aac7eb\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2ced6073006e45bd8c5170b7cadd4f0b : SUCCESS in 9m 39s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2e7da207052c41c4aba0d47e665a3f93 : SUCCESS in 7m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1ea5e7d9a6d4ed8f1914b0e3a10baa924ac84615","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-21 10:58:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4184dbeff2b74ddfb97f00a986dda032\n\n- grenade https://zuul.opendev.org/t/openstack/build/36544ce73a6d4658bb436eb14632750d : SUCCESS in 26m 13s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e52acaf85f874533bef6ff8871d27b3f : SUCCESS in 55m 45s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5fddbc663130449dbce320d1714a4999 : SUCCESS in 1h 36m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fbc1063540a740f88d2a8c829da0a2f8 : SUCCESS in 1h 12m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42b6ae32fbb14518840139c305b1e50e : SUCCESS in 4m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cb3fdc976f4944e2bf940fcaa48a5f02 : SUCCESS in 5m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/80aebb8faac346edb686c4372e642ec7 : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/903c3fcac9e24911b555b8aa35531b4b : SUCCESS in 5m 16s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d0f3c6a968464a5dae2d54454e1a4114 : SUCCESS in 8m 28s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6bc5e2b1c43a46479fd4bdcfaea0ee5d : SUCCESS in 7m 41s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/3c6f80d97b8c4ea9926dd28732e2fbb8 : SUCCESS in 13m 52s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/465b5231da0b49eda8eaa346995565af : SUCCESS in 2h 00m 57s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/df5cf7e563dd4b4c8b7a73c34517c042 : SUCCESS in 1h 49m 55s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/f1f844ad72594530ba27ca3762c97dd5 : SUCCESS in 2h 13m 10s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/2ae1f02afe8547fe988cb4529cd8c491 : SUCCESS in 1h 56m 58s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/8c560e7e4c144cb8bcd85df7f3b61c33 : FAILURE in 2h 05m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a4b354480c8845e0a99cb0d1db5e0dc3 : SUCCESS in 59m 36s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/4b1de2487fd4460a8455bcec1f51267f : SUCCESS in 2h 01m 37s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/999a8222422945aca35fb13434a2f709 : SUCCESS in 1h 12m 31s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/5d4a2eda169d4c708d77a95e866c63dc : SUCCESS in 55m 06s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/48fce315d8ea4d59934985d360262c88 : SUCCESS in 1h 51m 30s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/dc8a5fa4242640fdb54bd813faee5039 : SUCCESS in 27m 21s","accounts_in_message":[],"_revision_number":1},{"id":"58dd5746b6dc02b040b5dc14a6fb2587bd4b53ef","author":{"_account_id":34034,"name":"Franciszek Przewoźny","display_name":"Franciszek Przewozny","email":"przewozny.franciszek@gmail.com","username":"fprzewozn","status":"OpenStack Engineer / SRE @ Opera Software"},"date":"2025-05-21 11:00:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"579596a95026b36f5372e051cb46b1a95a7ab18b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-21 11:09:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8020d6f192624359a9c44a92a1bbff97\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/16410808429e4efa82103b33f045e3aa : SUCCESS in 7m 28s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/fe6fff5867364e5fba2d59507761e46f : SUCCESS in 7m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5dfa99f40b9d9c745ba6ba9c7e5edafca513ed41","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-21 13:18:34.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/da30477813b2498eb0da0f7208f237e6\n\n- grenade https://zuul.opendev.org/t/openstack/build/3bef37923617494ba870ebd4815f7db3 : SUCCESS in 45m 37s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/186814ac9e2c4103ac99e812cf9f95ae : SUCCESS in 50m 27s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/80ea520ef89d4e83ae5e4717fad52c12 : SUCCESS in 1h 27m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d7fdb03715d547118ff78ab941565497 : SUCCESS in 1h 04m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84b12162b9e7448bb6a4e2b27db81d85 : SUCCESS in 4m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d89bc8810cd54f7a98bf40ba76838771 : SUCCESS in 5m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6e016996b706494dbeb44b9cba9ec57d : SUCCESS in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb233500ecce403c96085667d104d213 : SUCCESS in 5m 49s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/42028aae6eb644c499bb2f3406c843bb : SUCCESS in 8m 09s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1c213c87b1d0479a8d280533becadddb : SUCCESS in 7m 54s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/808b4c22343d48fabb3d6ee79886cc6e : SUCCESS in 11m 39s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/e1c47158059d4579a4e8109af0bc8c36 : SUCCESS in 2h 12m 57s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/e318aaba570c40c59c2538a4e6f3d87a : SUCCESS in 1h 53m 25s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/da2ffae44ade4249b3b9a7fda7aa26f1 : SUCCESS in 1h 45m 05s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/dda2ac319a944ad5a4300a694179a4db : SUCCESS in 2h 05m 58s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/727755b6d51c4468b6f9cf98ff4ddfe6 : FAILURE in 2h 06m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6d14f3851ebb4278b2e2dac5e0cadbda : SUCCESS in 1h 00m 48s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/37f4ce4533c9469eaeb540b34d51ff21 : SUCCESS in 1h 53m 04s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/cd235679826349e6aa7038e61a967e3a : SUCCESS in 2h 01m 50s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/e9bc4a93bc484bd3a417f3e608637d53 : SUCCESS in 26m 59s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/329dd1f38e4747ec9199b3927934e472 : SUCCESS in 1h 54m 08s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/6c60d6b530fd45cca0e5ad90a170d35d : SUCCESS in 13m 40s","accounts_in_message":[],"_revision_number":1},{"id":"c0a065aa8cadc973c7656387e1d131f8934aaa8b","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-05-21 14:23:48.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"337a59a46c3f1b13d91af1e00f2292b22f53d2d6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-21 14:56:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc4f4bc35df44fecb5022c0a19274d4d\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8536de306a5e49e9a376c328f20fb324 : SUCCESS in 9m 57s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/db9477c6d8274a99841bda1166bed31a : SUCCESS in 9m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3d963b4bbeb09168afa04086244440745873f6c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-21 16:47:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1bed9044ef774d118418bc91db4d0371\n\n- grenade https://zuul.opendev.org/t/openstack/build/1f1b0a22f7f44f088e22279602c38cd4 : SUCCESS in 57m 48s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e8cda63f71d04ece919325fddb010fc1 : SUCCESS in 48m 05s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4e81dd684e5148e89f1082d6b487822c : SUCCESS in 1h 26m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/51153f5c48b244119088ac586a7472d5 : SUCCESS in 57m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d43a37294cf646aaaa7200eb706c91c6 : SUCCESS in 7m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5d32f0ecf1cf4266867e62c2c1d13675 : SUCCESS in 4m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/52f1a9145f6f4f0685a5cbaeb4a8ef55 : SUCCESS in 7m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd3d9f8033f74d31b95851a46e4046e7 : SUCCESS in 6m 53s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4b980d34630b4ffcad1a88fcabf95e71 : SUCCESS in 8m 42s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f13103b1ed96483cbe19f8022846cb15 : SUCCESS in 8m 21s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/4470f2db13d24025b661ae15d80b2f4a : SUCCESS in 11m 40s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/c8dffc85fb154e56a51c70f149afb99e : SUCCESS in 1h 45m 26s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/1d5eacb8419c4e67b47d858c1b7152fa : SUCCESS in 1h 03m 25s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/7dd5a06226c14598af81e2b220d340b7 : SUCCESS in 1h 50m 54s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/1d69c018d9b3426ba419a9edd82d7d51 : SUCCESS in 1h 55m 49s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/8cd5a2cfb348424caa01f5b9da01905c : SUCCESS in 2h 15m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ea2c6cc4de4b429d996feb5943fc6792 : SUCCESS in 57m 57s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/2b9636c921ba475da47575f37e6349d8 : SUCCESS in 1h 52m 41s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/7a6b48544aeb40cd91a075f4ad42b44f : FAILURE in 1h 49m 12s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/3dd198edd9894311b73bd2be4004fc9f : SUCCESS in 59m 04s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/7cbcc3b26c4743e9a1a880c00bb9fcf4 : SUCCESS in 2h 00m 38s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/7ffe4820bd80496e894f2e9b44ebfea6 : SUCCESS in 33m 49s","accounts_in_message":[],"_revision_number":1},{"id":"2ca9fbe324aa7490c9b5a6ca91107edf70cd2008","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-21 16:47:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b74fdac42fe3b27946d8adf58cda322c71a2d958","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-21 19:17:19.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d13ef76bfd61447e8a5dedc005ed5d78\n\n- grenade https://zuul.opendev.org/t/openstack/build/e31143dd921449c2afb36c70b24fd61a : SUCCESS in 27m 40s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8fb369651fc3455f80bb8ead239ef05a : SUCCESS in 53m 03s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/47296912a57647d79fd3bc279f795f34 : TIMED_OUT in 2h 05m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5913f05d94514583939dd365582ee2b8 : SUCCESS in 1h 00m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/690449399d5f4c77b773a4d202e7f466 : SUCCESS in 4m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5831594a9230438f8ab609fe5cbafae4 : SUCCESS in 7m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/169e430dd742481a9658068661e4c856 : SUCCESS in 5m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6bd64981509d4849b6284b985ade6617 : SUCCESS in 5m 24s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/22a747d7c127435baed4f0159b2647c4 : SUCCESS in 8m 48s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1cfaad1c881749129a1611886b2e6a00 : SUCCESS in 8m 52s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/fd993cb4c0d440d58baa0eeb8a8bad3f : SUCCESS in 12m 18s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/68298f5e8f6642929ca9f36bff1376ad : SUCCESS in 2h 05m 58s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/9b1dc35d76bf488296f3fab0f63f3016 : SUCCESS in 1h 51m 54s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/7004b531eb674e52b9ff17b90abacf7d : SUCCESS in 1h 58m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/002b6073ef58413fb8acaf09dd218841 : SUCCESS in 58m 34s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/7f8339545585483cb91f6fadb2964943 : SUCCESS in 1h 48m 54s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/250621f02dd94a569fc0985eb8fce8b6 : SUCCESS in 33m 24s","accounts_in_message":[],"_revision_number":1},{"id":"47793856a133e016e32719e91fa7e32bd821815e","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-05-21 20:34:58.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"80924f8185e3ed77f905cb2aee4772476a6e5850","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-21 20:43:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/874a391255634cd9b9e02347acfcc430\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d4f4e119cca2442a97a2c0ab6f42817b : SUCCESS in 6m 58s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e7c5ef23a3ad442484a5d0b29fce2b70 : SUCCESS in 5m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"410e6cf2ad89ddb78a59c97c3fb2f8fc87bc7972","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-21 22:53:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a5a0b02f2ec34582bccdd4428a3dbf0c\n\n- grenade https://zuul.opendev.org/t/openstack/build/be28ff2123d4476b80e5329067a443c4 : FAILURE in 1h 13m 16s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/91ac375bd03f457b88d32459319514d5 : SUCCESS in 57m 48s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/dcc74a45da2540ea87e35738f05e2d86 : SUCCESS in 1h 45m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6ac33c90f6694699becc326ca8fe72b6 : SUCCESS in 1h 15m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8378964a5a442a7a5a3d7fb0954c363 : SUCCESS in 2m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ac31883c1e7041028b8f063bc42c0e70 : SUCCESS in 2m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8a7852fb0db54f38a649f457cd9e4f08 : SUCCESS in 10m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b78e8e0ecf7f4dbca58948f0de334ce2 : SUCCESS in 7m 43s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/66c87291065e468e830088a359cc5d98 : SUCCESS in 12m 55s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f2cb1fc15d0346028ff37ef7a218ecf9 : SUCCESS in 7m 28s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/a678ac36c69d4e618b6b5cbaf0e52e9c : SUCCESS in 7m 38s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/91571c3b540c445e98ac938b73c46a9e : SUCCESS in 1h 48m 49s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/4b5171e5eb5e4cdbbc4ceed834c211c1 : SUCCESS in 1h 57m 32s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/6268f9a55eb647f79f0eee75d49bb9b8 : SUCCESS in 1h 48m 07s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/8546c02d66d7409eb94b4ab5a5112205 : SUCCESS in 1h 54m 43s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/05d5401240ce4c5e8720629fb5c7d394 : SUCCESS in 2h 13m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/75c233b69abd4def84ac97ab666e8ee1 : SUCCESS in 58m 30s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/4094605adbf74efe88dbc8593d37391c : SUCCESS in 1h 59m 35s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/3b8809efc58f4d3180fd8f4d301e195b : SUCCESS in 2h 07m 30s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/3cbdf1fd51024fbfa512883404cb50e7 : SUCCESS in 59m 32s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/a58b042d548a4764ac64e4ba4dcc86b1 : SUCCESS in 1h 39m 11s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/cdce2dda5b8647cca33c87b33d4ac4c2 : SUCCESS in 27m 17s","accounts_in_message":[],"_revision_number":1},{"id":"fde619f3aece212d5da5c30729c5106614dcca02","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-05-21 23:46:22.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"75109794c6947f03db61e1924772f7d39f6eab2a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-21 23:55:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a20b1104434944e59f034924658816af\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/34fae2456b5c48ad92522631a6be3e72 : SUCCESS in 7m 25s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1e9ef1ce88c9452fad23999e324d7ac4 : SUCCESS in 7m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2a57c8429aae9c040c5993fa6dcbb64fd4d0ef64","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-22 01:50:40.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8a2c23f904f5445c88e51a51a46fb26f\n\n- grenade https://zuul.opendev.org/t/openstack/build/9199d6c890b7489f9ca22ed797c5728a : SUCCESS in 56m 50s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8905f66426194e2e9db53d17157b4c2d : SUCCESS in 42m 28s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4cc0dd428c60446bbf91d12c7735555f : SUCCESS in 1h 06m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2f36b99cc9324e1fbfe9d2fac31f79e1 : FAILURE in 20m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a607fe0011b4e50b4b90a4ee52360e7 : SUCCESS in 3m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b618eb9f028f4cc49735aeef7ffd8906 : SUCCESS in 2m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9ae57d5a0f6641fdabe436281dde0ad2 : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32321256773b4d6f9efe3a4f653c8f97 : SUCCESS in 3m 23s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4b6246f2375f4d63ba1433f5b82c0c5c : SUCCESS in 5m 34s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1a420ad079ab4c90b7dc9cac01743ae2 : SUCCESS in 5m 50s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/e0a6719ebac5422fa55727d80f688762 : SUCCESS in 7m 46s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/ddca4c9b8e9e4a0ebaa0c4bb6cb1125f : SUCCESS in 1h 35m 16s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/dc9539aba0a342c087caaedee91a2d99 : SUCCESS in 1h 10m 13s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/6679f8efcabd43dea9295ef8ef7105a6 : SUCCESS in 1h 57m 23s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/cf279a3c6f524689b43d8c8cd0387ad1 : SUCCESS in 1h 41m 01s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/a66bf149fbdf4c46a41a1f11c8bff1a0 : SUCCESS in 1h 14m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ac0f742a2c4344fb9355d13cb16862b8 : SUCCESS in 56m 40s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/423bf0ca1b594b0e97c5e1545e145107 : SUCCESS in 1h 25m 39s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/43affde7d59143b59dd1815e6fb98389 : SUCCESS in 2h 02m 47s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/7121f40a98df4f59835ccfe0efed7fed : SUCCESS in 59m 26s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/e3d8c36d320d4bb7ba08eeaf6da416aa : SUCCESS in 1h 50m 06s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/657e7c03ce304f688155eeb54282e305 : SUCCESS in 23m 01s","accounts_in_message":[],"_revision_number":1},{"id":"09351a92f1bce2940af33c465443f3b7382cb7a6","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-05-22 13:24:18.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1d8fcf4d893e750f59cd91f25d04a12ea93b37b8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-22 14:03:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a6323a580cc48bb8c8845d482ea4e34\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/339066f611774dfd8907e6d5b861271a : SUCCESS in 7m 19s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c856e7715e774f6eb3bfff08cbb13555 : SUCCESS in 8m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"92550191a37e76bc90bf4114ee6f6b380e399aea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-22 16:00:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7cf4ca95ead34412a2c0d773a71c4df9\n\n- grenade https://zuul.opendev.org/t/openstack/build/1c94710ea7ff49988df91ce489d81d72 : SUCCESS in 1h 00m 56s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4c6781fa5d5340caa1b1521feb0649cd : SUCCESS in 45m 04s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ffdc3d453e694252a1386ecc3b6be48f : SUCCESS in 1h 18m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0d250a4e764f4bd6897f59844f670e7e : SUCCESS in 1h 00m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb3906dc339e42a1911dc7968c5a0e79 : SUCCESS in 4m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/66f04561fab3464c98cccb5621a6bce2 : SUCCESS in 5m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/244ad6bd649441d49ee652bfb99a2c3d : SUCCESS in 5m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f9ff2abe32843d8b44a5e571fcf875b : SUCCESS in 5m 17s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4508c57a71164aafa9a0281db087eb2f : SUCCESS in 8m 13s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/eea49d48929d481caa42841deca5355b : SUCCESS in 9m 51s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/6723c7c05e9a4510a7dfb21e92edb568 : SUCCESS in 12m 15s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/0a8b5ac4801a48b681cc84922404e21c : FAILURE in 2h 01m 11s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/88405e17129f42d7adb1d2d7333b22d2 : SUCCESS in 1h 56m 25s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/0087cd6c7a11411b9af1c0fd5e85092a : SUCCESS in 2h 03m 25s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/81cf37b12b4c492ca319c6e27f22e526 : SUCCESS in 2h 02m 26s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/879c63c291bb4c8f868a2ef00bc3cb0d : SUCCESS in 2h 24m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b71a29f92c84416ba7e6ccb7e66aec61 : SUCCESS in 1h 10m 00s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/4129a2a0aae142c19eb2c85b610d6ccd : SUCCESS in 1h 49m 09s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/feecb44363ea482690309ef5f7a43a5f : SUCCESS in 2h 07m 37s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/86db450d03d5417b8f58080ca598985a : SUCCESS in 59m 36s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/36b5ea3fa95d4db08d634ccc75bdde17 : SUCCESS in 1h 50m 45s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/0e35b68f8a4641d7bed9ee3959104799 : SUCCESS in 36m 01s","accounts_in_message":[],"_revision_number":1},{"id":"a6d3570598e7d27b487a3f874f68b07df80b5f4f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-22 16:01:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"90696e5c1a8281bbe4df267684b5839dddab669c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-22 18:07:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/69f461bdac824735ad26cd17ec343a4f\n\n- grenade https://zuul.opendev.org/t/openstack/build/ed11cf7223264a3c9db1370028b2d879 : SUCCESS in 52m 06s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/edd7abd7c23e49db8ec043dfe1ca6d9b : SUCCESS in 55m 01s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9930e8c82cfa49eab0b166c1a0fca976 : SUCCESS in 1h 24m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/304c2b648aa54ea3808501d06b8e6a1e : SUCCESS in 59m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3cbe3e2110054bacb7e6f6c5e1fe1efd : SUCCESS in 4m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/07810551eeec43818e9d4e1519085591 : SUCCESS in 5m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/12bb8ccf41d74e949d437fbc0818fe00 : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8477348f0f10434882347e7168899e21 : SUCCESS in 5m 58s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9918b8521f6e48ffa76dd8a3afbb9a2f : SUCCESS in 9m 43s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ce1d6b42110c4e0f8b41954cfb396c82 : SUCCESS in 8m 29s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/03a33c7f3f884194a71e825d34128e56 : SUCCESS in 13m 55s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/16a5aad6aacc4c1e887ba9959c01216e : SUCCESS in 1h 48m 20s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/3b02dbfac0134deb9a93ad8fc6a3916c : SUCCESS in 1h 57m 13s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/4bd4a9836ca9495db63683971e6d8b16 : SUCCESS in 2h 03m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/42e2545b282a4999b7a943f4689fcecb : SUCCESS in 54m 06s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f97601ed25b440f19a478b8dbbf11b0a : SUCCESS in 1h 54m 07s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f2bdea7ee16149a281c26d5371a06b2f : SUCCESS in 34m 09s","accounts_in_message":[],"_revision_number":1},{"id":"63071a22fbe1b4dc4502c7e818364effb432f3eb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-22 18:07:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"839aae31506981ca839238abbd24d2ec72909989","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-22 18:08:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/48fa1eb2773a481dacf7618bf18abf0c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b435e68deb324ce5a7856ab740e242ef : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"08ec545b6cdfdba7ee38e97da924ff7c021a98f1","revisions":{"08ec545b6cdfdba7ee38e97da924ff7c021a98f1":{"kind":"REWORK","_number":1,"created":"2025-04-30 06:16:41.000000000","uploader":{"_account_id":34034,"name":"Franciszek Przewoźny","display_name":"Franciszek Przewozny","email":"przewozny.franciszek@gmail.com","username":"fprzewozn","status":"OpenStack Engineer / SRE @ Opera Software"},"ref":"refs/changes/15/948515/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/15/948515/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/15/948515/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/15/948515/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/15/948515/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/15/948515/1"}}},"commit":{"parents":[{"commit":"0e6443944e4594c5affe78633b449722a665c01b","subject":"Update TOX_CONSTRAINTS_FILE for stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0e6443944e4594c5affe78633b449722a665c01b"}]}],"author":{"name":"Franciszek Przewozny","email":"przewozny.franciszek@gmail.com","date":"2025-04-04 12:36:36.000000000","tz":120},"committer":{"name":"Franciszek Przewoźny","email":"przewozny.franciszek@gmail.com","date":"2025-04-30 06:16:41.000000000","tz":0},"subject":"Fix vTPM metadefs.","message":"Fix vTPM metadefs.\n\nAccording to the Nova documentation[1], valid tpm_model properties\nare: tpm-tis and tpm-crb, not TIS and CRB. This patch fixes the\nmisalignment and also adds missing tpm_version property.\n\n[1]https://docs.openstack.org/nova/latest/admin/emulated-tpm.html\n\nCloses-Bug: #2106223\nChange-Id: I698dbd2946555c0a6455b1be17a6bf4a8d407ee7\n(cherry picked from commit df5f2b3c8693ce288209e86bee656e7f9c7b9566)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/08ec545b6cdfdba7ee38e97da924ff7c021a98f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/08ec545b6cdfdba7ee38e97da924ff7c021a98f1"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
