)]}'
{"id":"openstack%2Fglance~948516","triplet_id":"openstack%2Fglance~stable%2F2024.2~I698dbd2946555c0a6455b1be17a6bf4a8d407ee7","project":"openstack/glance","branch":"stable/2024.2","topic":"bug/2106223-stable/2024.2","attention_set":{},"removed_from_attention_set":{"34034":{"account":{"_account_id":34034,"name":"Franciszek Przewoźny","display_name":"Franciszek Przewozny","email":"przewozny.franciszek@gmail.com","username":"fprzewozn","status":"OpenStack Engineer / SRE @ Opera Software"},"last_update":"2025-05-27 16:46:36.000000000","reason":"Change was submitted"},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2025-05-27 14:11:55.000000000","reason":"\u003cGERRIT_ACCOUNT_8122\u003e replied on the change","reason_account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}}},"hashtags":[],"change_id":"I698dbd2946555c0a6455b1be17a6bf4a8d407ee7","subject":"Fix vTPM metadefs.","status":"MERGED","created":"2025-04-30 06:17:23.000000000","updated":"2025-05-27 16:47:29.000000000","submitted":"2025-05-27 16:46:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"948516-bug/2106223-stable/2024.2","cherry_pick_of_change":946343,"cherry_pick_of_patch_set":2,"meta_rev_id":"79d16d7af3928a488047014fb89fd835c6c68c90","_number":948516,"virtual_id_number":948516,"owner":{"_account_id":34034,"name":"Franciszek Przewoźny","display_name":"Franciszek Przewozny","email":"przewozny.franciszek@gmail.com","username":"fprzewozn","status":"OpenStack Engineer / SRE @ Opera Software"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-27 16:46:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":2,"date":"2025-05-27 14:11:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":1,"date":"2025-05-27 14:11:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-30 06:18:41.000000000","updated_by":{"_account_id":34034,"name":"Franciszek Przewoźny","display_name":"Franciszek Przewozny","email":"przewozny.franciszek@gmail.com","username":"fprzewozn","status":"OpenStack Engineer / SRE @ Opera Software"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2025-04-30 06:32:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-04-30 08:23:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ebb15dd7a9106c5033d47be7e3ba2fc9f466f1bf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34034,"name":"Franciszek Przewoźny","display_name":"Franciszek Przewozny","email":"przewozny.franciszek@gmail.com","username":"fprzewozn","status":"OpenStack Engineer / SRE @ Opera Software"},"date":"2025-04-30 06:17:23.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"aa9af0bc2580e481af94f8bb4a8f1b0822b7f927","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-30 06:32:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d1e80bc528f48ddb48272c086e04993\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3c6d5d5b50d14a639d237006f7d109da : SUCCESS in 11m 25s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/67797651e66d4ba8b419575bb0560853 : SUCCESS in 7m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d0bde664777695b2e96e953761e87426e534a48c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-30 08:23:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/65090606f7d74cfd9511b0bb6bb7854a\n\n- grenade https://zuul.opendev.org/t/openstack/build/ec68d163c8da4a60838b53537ac52b4e : SUCCESS in 1h 00m 31s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/44bb861f4c7e4b09ae64d651f7be1852 : SUCCESS in 1h 33m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5214b6cde69a473d99a766df93c3bb34 : SUCCESS in 50m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2479750ea4b34e8b9f0bd0cb7076297a : SUCCESS in 3m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/47bb8466f5554bc893887e2b68f58d2f : SUCCESS in 5m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9431258f2242448c85b0707e87aae6cf : SUCCESS in 4m 28s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ae22e4985bf04fd68d6b60b812693d31 : SUCCESS in 5m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6da6794a491457db1d09ca5c9583d8b : SUCCESS in 4m 27s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/845ac66634bb460ab3646b4533371d88 : SUCCESS in 7m 56s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f83644e9dcaa40b6b713d718f1fb01d4 : SUCCESS in 8m 53s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/54b16c93ae624973bfa42802c8c35b2c : SUCCESS in 11m 40s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/b1255135f3e2434e850bafc268f5d33b : SUCCESS in 1h 44m 44s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/576760ef1b8b40d184c5f6753b0e6db4 : SUCCESS in 1h 30m 57s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/739023349cf841cf83371c5decce02a8 : SUCCESS in 1h 49m 47s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/0476b39208934ef78151a02b69775da5 : SUCCESS in 2h 00m 09s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/1f618118491d458aa11332fb810888ae : SUCCESS in 2h 00m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bb2759c927d64fea8f287143ffc40e76 : SUCCESS in 55m 10s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/38b843980abc49a4a89122a66ce21eb6 : SUCCESS in 1h 37m 01s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/c95e95025fd94589b96c0ddf8448aeaa : SUCCESS in 1h 50m 53s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/3203048f1dba484b98841f9146f7ac18 : SUCCESS in 59m 59s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/51e8c170dc754d66bc8c20f921f13f62 : SUCCESS in 25m 48s","accounts_in_message":[],"_revision_number":1},{"id":"1a7adbf9d604c852e3302882003ea50da2e4d20a","author":{"_account_id":34034,"name":"Franciszek Przewoźny","display_name":"Franciszek Przewozny","email":"przewozny.franciszek@gmail.com","username":"fprzewozn","status":"OpenStack Engineer / SRE @ Opera Software"},"date":"2025-05-09 11:12:17.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a1fd4308dae8d4b0fcfe633c860c3c07a20fe5fa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-05-26 15:45:07.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"d5c91ef4ce26d2a419a54b7a0f4bd54e7ef6fdc9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-26 15:56:40.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/32716f9d0c054499984220c83ea11749\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/03fc4be36f78475abd55d4a278f2ae66 : SUCCESS in 8m 21s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/7f9b8d9360644ae88e13fb455ae17ff7 : SUCCESS in 9m 11s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3b715be237fc49d4f1d999463b1565e8239e5f37","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-26 18:06:39.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/10a16eb60719431088eda3dbec2be937\n\n- grenade https://zuul.opendev.org/t/openstack/build/b5c404071f9047238f25372889476250 : SUCCESS in 57m 55s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/71916ef39fa04096a9072fe50922e580 : SUCCESS in 1h 20m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/01f6089f58314b3e836fdd9bec6b152c : SUCCESS in 46m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e908a6122ce43499a3c00cbcd304048 : SUCCESS in 4m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e4234fd902a94fc983e9b3ba05289941 : SUCCESS in 5m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/aa6e185ae4fb4a1fbd109c7e77b51648 : SUCCESS in 6m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d29a69617f1645afa4e358f7f82d3ae5 : SUCCESS in 4m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59d4e915ae3e46639468e2cb797f97a7 : SUCCESS in 5m 45s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e2408049084e4c01b35331736f5d10b2 : SUCCESS in 8m 40s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7ef2c516d6ea48079f1497e861fa0135 : SUCCESS in 7m 42s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/0fe5150a5511493faf76dce0deb95e8a : SUCCESS in 11m 32s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/f298bb66a78141cbb532a99c00a282e4 : SUCCESS in 1h 43m 47s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/ac0cc4dd7b7146fca26d0db02905bfb1 : SUCCESS in 1h 55m 11s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/a5eda495b3134073ac6d7318fb9d33b9 : SUCCESS in 1h 32m 51s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/8921f1ca64a7474eb262a130a3694fa0 : SUCCESS in 1h 56m 41s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/69a71e5e929d41978c21649319a05a24 : SUCCESS in 2h 16m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1299a45f51dd49e6b810370b11a49490 : SUCCESS in 54m 22s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/659a92f8dccc4dedbe50cbb43ef76b44 : SUCCESS in 1h 45m 38s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/b3ad6586843c4c4c88b8c11eaf7f9ab7 : SUCCESS in 1h 52m 02s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/e45945c89a1b46ffa979a6a7d7bfe622 : SUCCESS in 56m 11s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/59621905237e4768bbb13674cdcbe0e1 : SUCCESS in 28m 35s","accounts_in_message":[],"_revision_number":2},{"id":"e72d1ed9556bd76828ff667140376c8ea35a78bd","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-05-27 14:11:55.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"53927889225a2364d8b942a23f2c8383a45f1ec2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-27 14:12:24.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"01fd0b3158e80864250bed8d16681b887440499e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-27 16:46:36.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/386d3e9760054e4089763aebd60fbc04\n\n- grenade https://zuul.opendev.org/t/openstack/build/a0296550fcc34e3b9795a897e023d70a : SUCCESS in 59m 11s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e00d166c9f5b424d8bca644a16723fe4 : SUCCESS in 1h 11m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f164e2a44acb4a17b8183b39c44c877b : SUCCESS in 57m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59e58cf497394926a66b51aa56c21f07 : SUCCESS in 4m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/81ea2252a3584e2994e0d00bac2b5f0e : SUCCESS in 5m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/aa08c78ca9684e509138efb3236c8d42 : SUCCESS in 6m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8726e7f82681461e97efa69c613cc7a2 : SUCCESS in 5m 55s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3ef2b8cff1f941628aaeae52748e5903 : SUCCESS in 8m 32s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/14adb36980bc4eaf9dfab2dc7875af45 : SUCCESS in 9m 15s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/46b3021b9e41440ea6fedc21067c2c88 : SUCCESS in 12m 40s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/0fa4fc6d08374e28895ac47181851295 : SUCCESS in 1h 46m 22s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/0b135150fa454821bfb7a08c44005364 : SUCCESS in 2h 04m 05s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/48e37e53b870453b8ae8e5a531cd38c2 : SUCCESS in 1h 47m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fdeafd671a9541e9ba94935521f3f320 : SUCCESS in 46m 03s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/fe244889411c4e51bdb44dbc3784250e : SUCCESS in 1h 52m 50s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/8fc112a09a5d4cf58cfce917dc87b80d : SUCCESS in 30m 07s","accounts_in_message":[],"_revision_number":2},{"id":"0a187e03f7779bba7400cdc8a06decd58f1e5f24","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-27 16:46:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"79d16d7af3928a488047014fb89fd835c6c68c90","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-27 16:47:29.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/41eecd2717274f0a81306586846dd801\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05b52b9b83fa473c86f9a3e0108633ae : SUCCESS in 41s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"097f38158189efc5c7d52fd56467455a58442c73","revisions":{"5c695a84b888f13a8570699a98015313d3968ad2":{"kind":"REWORK","_number":1,"created":"2025-04-30 06:17:23.000000000","uploader":{"_account_id":34034,"name":"Franciszek Przewoźny","display_name":"Franciszek Przewozny","email":"przewozny.franciszek@gmail.com","username":"fprzewozn","status":"OpenStack Engineer / SRE @ Opera Software"},"ref":"refs/changes/16/948516/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/16/948516/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/16/948516/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/16/948516/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/16/948516/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/16/948516/1"}}},"commit":{"parents":[{"commit":"e74b91d24de43836fe1269fd8d3ea11afd0caa58","subject":"Fix broken metadata function calls","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e74b91d24de43836fe1269fd8d3ea11afd0caa58"}]}],"author":{"name":"Franciszek Przewozny","email":"przewozny.franciszek@gmail.com","date":"2025-04-04 12:36:36.000000000","tz":120},"committer":{"name":"Franciszek Przewoźny","email":"przewozny.franciszek@gmail.com","date":"2025-04-30 06:17:23.000000000","tz":0},"subject":"Fix vTPM metadefs.","message":"Fix vTPM metadefs.\n\nAccording to the Nova documentation[1], valid tpm_model properties\nare: tpm-tis and tpm-crb, not TIS and CRB. This patch fixes the\nmisalignment and also adds missing tpm_version property.\n\n[1]https://docs.openstack.org/nova/latest/admin/emulated-tpm.html\n\nCloses-Bug: #2106223\nChange-Id: I698dbd2946555c0a6455b1be17a6bf4a8d407ee7\n(cherry picked from commit df5f2b3c8693ce288209e86bee656e7f9c7b9566)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5c695a84b888f13a8570699a98015313d3968ad2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5c695a84b888f13a8570699a98015313d3968ad2"}]},"branch":"refs/heads/stable/2024.2"},"097f38158189efc5c7d52fd56467455a58442c73":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":2,"created":"2025-05-26 15:45:07.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/16/948516/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/16/948516/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/16/948516/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/16/948516/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/16/948516/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/16/948516/2"}}},"commit":{"parents":[{"commit":"7af1ddb0db23f0091bee4ad0d34c217f70ccea51","subject":"Harden copy bust lock test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7af1ddb0db23f0091bee4ad0d34c217f70ccea51"}]}],"author":{"name":"Franciszek Przewozny","email":"przewozny.franciszek@gmail.com","date":"2025-04-04 12:36:36.000000000","tz":120},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2025-05-26 15:44:46.000000000","tz":120},"subject":"Fix vTPM metadefs.","message":"Fix vTPM metadefs.\n\nAccording to the Nova documentation[1], valid tpm_model properties\nare: tpm-tis and tpm-crb, not TIS and CRB. This patch fixes the\nmisalignment and also adds missing tpm_version property.\n\n[1]https://docs.openstack.org/nova/latest/admin/emulated-tpm.html\n\nCloses-Bug: #2106223\nChange-Id: I698dbd2946555c0a6455b1be17a6bf4a8d407ee7\n(cherry picked from commit df5f2b3c8693ce288209e86bee656e7f9c7b9566)\n(cherry picked from commit 08ec545b6cdfdba7ee38e97da924ff7c021a98f1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/097f38158189efc5c7d52fd56467455a58442c73"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/097f38158189efc5c7d52fd56467455a58442c73"}]},"branch":"refs/heads/stable/2024.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
