)]}'
{"id":"openstack%2Fglance~948518","triplet_id":"openstack%2Fglance~stable%2F2024.1~I698dbd2946555c0a6455b1be17a6bf4a8d407ee7","project":"openstack/glance","branch":"stable/2024.1","topic":"bug/2106223-stable/2024.1","attention_set":{},"removed_from_attention_set":{"34034":{"account":{"_account_id":34034,"name":"Franciszek Przewoźny","display_name":"Franciszek Przewozny","email":"przewozny.franciszek@gmail.com","username":"fprzewozn","status":"OpenStack Engineer / SRE @ Opera Software"},"last_update":"2025-10-17 09:12:00.000000000","reason":"\u003cGERRIT_ACCOUNT_34034\u003e replied on the change","reason_account":{"_account_id":34034,"name":"Franciszek Przewoźny","display_name":"Franciszek Przewozny","email":"przewozny.franciszek@gmail.com","username":"fprzewozn","status":"OpenStack Engineer / SRE @ Opera Software"}},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2025-05-12 09:15:15.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_34034\u003e by clicking the attention icon","reason_account":{"_account_id":34034,"name":"Franciszek Przewoźny","display_name":"Franciszek Przewozny","email":"przewozny.franciszek@gmail.com","username":"fprzewozn","status":"OpenStack Engineer / SRE @ Opera Software"}}},"hashtags":[],"change_id":"I698dbd2946555c0a6455b1be17a6bf4a8d407ee7","subject":"Fix vTPM metadefs.","status":"MERGED","created":"2025-04-30 06:19:56.000000000","updated":"2026-09-16 05:40:15.000000000","submitted":"2025-10-17 13:32:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"948518-bug/2106223-stable/2024.1","cherry_pick_of_change":946343,"cherry_pick_of_patch_set":2,"meta_rev_id":"9d16d9975af32596fda98caaa46300bfccc8e06e","_number":948518,"virtual_id_number":948518,"owner":{"_account_id":34034,"name":"Franciszek Przewoźny","display_name":"Franciszek Przewozny","email":"przewozny.franciszek@gmail.com","username":"fprzewozn","status":"OpenStack Engineer / SRE @ Opera Software"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-17 13:32:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2025-10-16 14:16:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2025-10-16 15:13:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2025-10-16 15:13:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-30 06:20:19.000000000","updated_by":{"_account_id":34034,"name":"Franciszek Przewoźny","display_name":"Franciszek Przewozny","email":"przewozny.franciszek@gmail.com","username":"fprzewozn","status":"OpenStack Engineer / SRE @ Opera Software"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2025-04-30 06:43:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-04-30 08:24:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-16 14:16:27.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2026-09-16 05:40:15.000000000","updated_by":{"_account_id":34034,"name":"Franciszek Przewoźny","display_name":"Franciszek Przewozny","email":"przewozny.franciszek@gmail.com","username":"fprzewozn","status":"OpenStack Engineer / SRE @ Opera Software"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"}],"messages":[{"id":"0aff11fadfcdb16d943c13a5a3c93f1b1da52f15","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34034,"name":"Franciszek Przewoźny","display_name":"Franciszek Przewozny","email":"przewozny.franciszek@gmail.com","username":"fprzewozn","status":"OpenStack Engineer / SRE @ Opera Software"},"date":"2025-04-30 06:19:56.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"e6d73ad43e3d6ddda283fb1b97046a18ddb32ee3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-30 06:43:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b99679a8efd74b08b01a0fc5c0490b28\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8587b9f89ead401090a51cb086918ebc : SUCCESS in 7m 13s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/95b2c76a891448488e68c80c06edd33c : SUCCESS in 10m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3e2fa132ead270db3378bdc8713a40483e2bebff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-30 08:24:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bafc0a7b52ae4b049ff147d0fba0c36e\n\n- grenade https://zuul.opendev.org/t/openstack/build/d210b9b4d6d042c1875c9d3ebc3680a8 : SUCCESS in 1h 03m 03s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/4893772219d04ee1b63428f5a0113258 : SUCCESS in 50m 15s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0509de8f32284912b1428f6a6bb6940b : SUCCESS in 1h 22m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8c20f848912145f3a5242095920f8a5f : SUCCESS in 50m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c12877723fd0413791e55865e84c1ddf : SUCCESS in 3m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f1bf7cf035c742d6be500c17130f9ce8 : SUCCESS in 4m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/364fb9121df6410db006ead7559a540d : SUCCESS in 4m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/680aedf5b7404a2d8d72feb80611df38 : SUCCESS in 5m 17s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a43cd6f2eca44d40ad9b6d7916f82c13 : SUCCESS in 8m 05s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/04be31d582564e0a8194db2c9fb3d5d3 : SUCCESS in 7m 18s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/8c12a0400b6f4767bb90db722b5d0059 : SUCCESS in 7m 18s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/f784f02c7dd641418ef7a9aa75fe1be6 : SUCCESS in 1h 54m 24s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/6a9e9ff43f75480796f1bdd5be60b3fa : SUCCESS in 1h 43m 39s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/18b15bdfffe34c1183ce3239c2f946e7 : SUCCESS in 1h 38m 30s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/ee275454cd13409b87b9e371d4e26853 : SUCCESS in 1h 41m 27s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/51e0c47cdd104ac9bfd730c08e75f5c2 : FAILURE in 1h 52m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/af18b22f5a774dcfbddbef18d7f64024 : SUCCESS in 51m 29s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/04f72dde4564427e836f493f4f731cd2 : SUCCESS in 1h 39m 04s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/f2b58d6bebc14df8acf7a2342b91f562 : SUCCESS in 1h 40m 07s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/3dbbae4a02204a26abe195fdbbd7d2a8 : SUCCESS in 1h 01m 38s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/eaaaf79b01004296b27d6dacaf19902f : SUCCESS in 27m 12s","accounts_in_message":[],"_revision_number":1},{"id":"77c6ba7c0890f1274fee656ef4e031365ba8922b","author":{"_account_id":34034,"name":"Franciszek Przewoźny","display_name":"Franciszek Przewozny","email":"przewozny.franciszek@gmail.com","username":"fprzewozn","status":"OpenStack Engineer / SRE @ Opera Software"},"date":"2025-04-30 08:25:50.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"392f9f95c287a60af1944e061267dc71347535de","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-30 08:47:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/46c6e46f37be4ee0b65583b1a824712d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/713014d5e915431a9f32d1aecf7460ce : SUCCESS in 6m 01s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/5adeb767293443e8876e202ae6446952 : SUCCESS in 10m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"01469bdd3d64431114aab3af959cae1161932650","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-30 10:26:02.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1b9a20bd995347aeb2e2e7c51cb008c8\n\n- grenade https://zuul.opendev.org/t/openstack/build/c6421cf338704ee584f069f485c68103 : SUCCESS in 40m 41s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/91a254fe0ed94625ac9a1893319f686d : SUCCESS in 46m 07s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a178a1f2c61a4fd3b4c46138d2b0c8d5 : SUCCESS in 1h 07m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8c4ac83508c6493dbe95c5577a038fba : SUCCESS in 24m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e502c236e804dbf9ff9bb90f7f8f731 : SUCCESS in 3m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6c1cce743e144957b2029dacf151c274 : SUCCESS in 4m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/689e90342c024d90b99dfb11256cc2f7 : SUCCESS in 2m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f08cecfa176a4d91a719f48f471b0571 : SUCCESS in 4m 36s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4ff0e7ef5c2a4f50866b2ea7fcececca : SUCCESS in 9m 25s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9c030996c24d472dbd86eb78d0e393e9 : FAILURE in 5m 13s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/fb8bfef003824c8b9ed8a84b77d6c435 : SUCCESS in 9m 05s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/65f6a8301a494c6293a30d09ed13fb02 : SUCCESS in 1h 52m 18s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/7087367bd5c541d5b587589775cd0863 : SUCCESS in 1h 23m 21s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/d16f212a748d4961aef750ecee4e8a84 : SUCCESS in 46m 49s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/9a44d454057a4e08887c4668c28b1766 : SUCCESS in 1h 37m 44s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/36f6679e96a74167bc3e1c71fab09ca9 : SUCCESS in 1h 32m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1c2c78d7d67f483180b3d77bbd384996 : SUCCESS in 23m 39s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/bc568057c79d40e597d66c46ade14cd9 : SUCCESS in 1h 33m 17s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/2e80d94f01bd4484acdbb2124bb497fb : SUCCESS in 1h 46m 55s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/36af2b81d8754bb883528f2a40177502 : SUCCESS in 26m 16s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/d3479f0e4f2e4df19b8345b6574b07f0 : SUCCESS in 31m 08s","accounts_in_message":[],"_revision_number":1},{"id":"5e79350ee8e138d692a91618387893a9cbc71310","author":{"_account_id":34034,"name":"Franciszek Przewoźny","display_name":"Franciszek Przewozny","email":"przewozny.franciszek@gmail.com","username":"fprzewozn","status":"OpenStack Engineer / SRE @ Opera Software"},"date":"2025-05-09 21:07:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bbb8574b78d88b240027113ec34fe326c0fd985b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-09 21:16:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/12be521281ca488ba6ac30e8c422c713\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0dac5508450243bb828ea35b95f8814d : SUCCESS in 8m 00s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d83a91438e394a42864a0a30524e57a6 : SUCCESS in 6m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fd17d8f5ec24ec1ff431f638c35be50a78eb8808","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-09 23:04:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1dd646729cd54f39adb0cf7fda1ed435\n\n- grenade https://zuul.opendev.org/t/openstack/build/293ce83d9ead43fe8c780fbfc14fd487 : SUCCESS in 52m 53s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/5aa8708348d74de7999dd28d5c1799f3 : SUCCESS in 47m 22s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a786d180826f4c5a9f89adeff26cddce : SUCCESS in 1h 21m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f344ee3244a948b4912140081303da6f : SUCCESS in 48m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b86a96aaa8b141b49d31156fd08f1d1f : SUCCESS in 4m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d29593eb22e84cc59d5269f689895e13 : SUCCESS in 3m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c1f502d49c8c40128c4e91781a7f39b7 : SUCCESS in 7m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b208131a1fa45259d97ebaf7b52049b : SUCCESS in 6m 17s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ae459263f1094dec874ec8d8adf531eb : SUCCESS in 7m 09s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ebf21c7504504eca9fb68a6dbce1bbdb : SUCCESS in 7m 12s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/fbd762c5f2cd4d9d82d0d2929426fff9 : SUCCESS in 9m 04s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/06e9b91364f0482da90195a617e1a273 : SUCCESS in 1h 50m 42s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/af6c17967ce14eddae6160a1bad9b051 : SUCCESS in 1h 30m 11s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/01994fb8ef324bc0992cebd316040384 : SUCCESS in 1h 28m 19s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/68b718e737144d79be95748588f8cbde : SUCCESS in 1h 32m 00s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/46c884d4cd0b411d91056a43573da864 : SUCCESS in 1h 04m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b96a6d10b1df49bc9af97427c2ae6d1c : SUCCESS in 45m 06s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/7e1ed612155f4cf9879d32fbf192f511 : SUCCESS in 1h 32m 41s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/3209a74657f34a85bbfcb1d6601dc8ba : SUCCESS in 1h 30m 47s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/73f0ca6ede924e1cb2d83a24f0831c47 : SUCCESS in 58m 47s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/d9b3bf873776439a9f8b348a258ed1a7 : SUCCESS in 32m 42s","accounts_in_message":[],"_revision_number":1},{"id":"2e2b1bf0f80ec15052f4c6d00127f3c82308961e","author":{"_account_id":34034,"name":"Franciszek Przewoźny","display_name":"Franciszek Przewozny","email":"przewozny.franciszek@gmail.com","username":"fprzewozn","status":"OpenStack Engineer / SRE @ Opera Software"},"date":"2025-07-31 08:36:25.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d0814b4253e886f13b7786892eb8553f144987ef","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-07-31 17:21:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b0e921299e8011a067dfcb467d40430463d7c844","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-10-16 14:16:27.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"addfda05c9da052c0cb31d78e5da571ca9271958","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-10-16 15:13:33.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c4d591585a3163d1270fd5a865b110c5c22b3ce9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-16 15:13:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"892b46b619b5550da7dab0aee35bf66378031830","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-16 17:38:46.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/246cf3a4b35448c99a0c91cbe754b972\n\n- grenade https://zuul.opendev.org/t/openstack/build/2bf33c2af5a44e72911a58c1e20c9fa3 : SUCCESS in 58m 50s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/0f5d5b82982d4b2fba6c5034b4122cf1 : SUCCESS in 54m 54s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b37c57d518574ef6b3228c54a38eaef8 : POST_FAILURE in 1h 21m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6b4ed7ace75d4719b32c0b909d978337 : SUCCESS in 40m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/912bb78e00fc41288f404e34c4ce1f71 : SUCCESS in 4m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f9ac3e9006fe422eac38c3999706044e : SUCCESS in 4m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e74f85cf14844920824164f12b6964a2 : SUCCESS in 6m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7a7d8608bcf4b548584242e8ffb24ae : SUCCESS in 5m 22s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e313501ae2194a149c577a20f66e3020 : SUCCESS in 9m 20s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/80896965fb804d959609e44e8808f682 : SUCCESS in 9m 08s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/ebc2f3b410c74f74af4510ff0d2eeb4a : SUCCESS in 2h 07m 37s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/b72f6fd506b04a86b02f3267d3e4a3b2 : SUCCESS in 1h 42m 31s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/4b2310e078ee4dbcaf31f2dff03c14f1 : SUCCESS in 1h 50m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a8e194ef1fe24478af34367235de4e54 : SUCCESS in 49m 13s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a3758b19b14d49de9a03632e282023a3 : SUCCESS in 1h 43m 19s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/47327d5826b34e1eae62959e59a0b3c1 : SUCCESS in 24m 16s","accounts_in_message":[],"_revision_number":1},{"id":"f736e25bfb589e9f4d42319f19e9be71aa24bf02","author":{"_account_id":34034,"name":"Franciszek Przewoźny","display_name":"Franciszek Przewozny","email":"przewozny.franciszek@gmail.com","username":"fprzewozn","status":"OpenStack Engineer / SRE @ Opera Software"},"date":"2025-10-16 17:57:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5ef66621f5ef3231ce7f5376318ab4825ba1e339","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-16 18:09:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c521ea48bbea40ae9b5f106321971838\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/80daef45b74f4c00b1574488795447e7 : SUCCESS in 10m 20s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/46d5e75da9f24b2a9c9619165ff6d46c : SUCCESS in 9m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5193a1d1e8e9382c602b024cdd449985ec0a2667","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-16 20:04:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/01f833ff4a9b4ddcbf030dab5372b834\n\n- grenade https://zuul.opendev.org/t/openstack/build/49c22f92bf034c07b685d523cdc7c765 : SUCCESS in 1h 00m 10s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/905c5e975ffa4afd9e59e222d8011278 : SUCCESS in 38m 43s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7886d5abd5654a26a94fc9d13706f7a3 : SUCCESS in 1h 32m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/856ebc856b294740ad3e57f985b12ff0 : SUCCESS in 46m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6332600454e483bb317e1f1b5189629 : SUCCESS in 4m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2a45e8746e1740d0be30f3992b173c72 : SUCCESS in 5m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0a53d0ebe6e348338d052ab0d873d01e : SUCCESS in 7m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8511b7d7c294ee8954b64d8cd11ffb0 : SUCCESS in 4m 08s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/dd4180471b2b45428918e13512f59260 : SUCCESS in 7m 41s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/add74d585ec44213976f07b1401cd14d : SUCCESS in 9m 07s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/8b6f0b7e6a574c8f91299cd2e6088251 : SUCCESS in 7m 27s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/e7c1cd94b49c4a98b695220d2006af09 : SUCCESS in 1h 42m 17s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/392b225e62374c28b08c65921229e3e0 : SUCCESS in 1h 40m 53s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/ba3e31df1f094ebbb47283fd3e171f6c : SUCCESS in 1h 54m 08s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/9d9efe38baea4e05b4969367abbad688 : SUCCESS in 1h 20m 39s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/3acd44f7f81943e6940a3ada1cf93765 : SUCCESS in 1h 59m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/771682d73ad1437e8ec0c3e7719fd03a : SUCCESS in 52m 19s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/78317efe80574d34b5f2715ff81c644c : SUCCESS in 1h 30m 57s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/58ee00dec14f4350baf0e128a9daddb7 : SUCCESS in 1h 39m 21s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/a5b67fe7fcdd47629c17ac55d687b9a8 : SUCCESS in 56m 15s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/65d50670978641819696716889cd18b1 : SUCCESS in 32m 02s","accounts_in_message":[],"_revision_number":1},{"id":"9b3b56ff95b14e8e1db31072ec55dd8c5c74cf9c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-16 20:04:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"37e85528cd6c6ddfaa5e2a16993fb3ab6256fcc5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-16 22:07:20.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9812d9ababf34535aea70ded17ade6be\n\n- grenade https://zuul.opendev.org/t/openstack/build/c1636bb39f07486da2109b201d35a487 : SUCCESS in 1h 03m 34s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/2e319a7029684f67b61a4588d973f555 : SUCCESS in 54m 34s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/21608bd8e1214e94bb4ba320e004f6f1 : SUCCESS in 53m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f80675022f5f4ab181205626dd99b58f : SUCCESS in 38m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42c72624dea5424599a4c74e4f6cc773 : SUCCESS in 2m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d6901a6770f141e984adddb00c737d91 : SUCCESS in 3m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d4dafd38c33a4fcabbbf279e94ed9071 : POST_FAILURE in 7m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea957becca19470c99e3cf57ae42ea43 : SUCCESS in 5m 28s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1d6bd6e3c8b94e89a3d55df41d25fc66 : SUCCESS in 9m 06s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/909a528d0f57487c866fe4dd506a77f6 : SUCCESS in 8m 02s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/fa022e9e866a4b3c858f596175e6583c : SUCCESS in 1h 44m 43s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/f19fc2886dc34ebf9ae29db1924ce279 : SUCCESS in 1h 52m 23s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/edcccb3ab8ec4ea8b8380baff140a341 : SUCCESS in 1h 49m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bf1f5b8efba942b78445477f0cd8b94e : SUCCESS in 36m 20s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/32c0275324e24a7ba6d9189af2e68384 : FAILURE in 1h 24m 05s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/d7a795183360461588b3072f47c9aee4 : SUCCESS in 20m 07s","accounts_in_message":[],"_revision_number":1},{"id":"79fc106a4c170b1f2c9faa68a624b76454b7c6a2","author":{"_account_id":34034,"name":"Franciszek Przewoźny","display_name":"Franciszek Przewozny","email":"przewozny.franciszek@gmail.com","username":"fprzewozn","status":"OpenStack Engineer / SRE @ Opera Software"},"date":"2025-10-17 09:12:00.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5b3065f54b8c18a5bdec693c6832d76cabe8793e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-17 10:37:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ad6afafb0b9e4857ab08eb856935caf2\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a8faab9eab3740a99359c73c27ab6213 : SUCCESS in 11m 12s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0f4cf2b67bf44633a727573d5abf5ada : SUCCESS in 12m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5e705a940bf88bdd9f6b35bb70c7f867ecfb203f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-17 11:25:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/476e6d9285054dc7af23ae1c688fa7fd\n\n- grenade https://zuul.opendev.org/t/openstack/build/fe6d6862592d4d3ba54df34d47e6871e : SUCCESS in 45m 06s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/bbfb8bf346c04c8cb2fa60c3bd729188 : SUCCESS in 53m 28s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/22ad41582d274560b41d8b3da5067a9c : SUCCESS in 2h 01m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/05b1ae52234e4cd7a9f739dfbe7666be : SUCCESS in 57m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab21989c6d084765b6eca2be2a2ed3df : SUCCESS in 2m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2d07e800e0d24baab40de74899cb89ca : SUCCESS in 3m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bbd95da10d4f4cc58be2f2380ecb89ae : SUCCESS in 3m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef442a77b16f48b7b0bfdcf5c22028f6 : SUCCESS in 3m 30s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/031bd4af490b45dab2768ebecde67642 : SUCCESS in 9m 12s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/40ab01f9aa8b4dd9a72efc5b2ef7382d : SUCCESS in 8m 37s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/165186463f7245de99a76ecc15d0b70f : SUCCESS in 8m 32s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/456a0a52651a4422ba1773d771ce457e : SUCCESS in 1h 46m 11s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/a01aa06c0fc04bb49da92857aa1cbe8a : SUCCESS in 1h 22m 48s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/71dd7c7bb0454f809015cd9ed204e514 : SUCCESS in 1h 46m 01s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/18ac85d276134805ae936cafd703ac00 : SUCCESS in 55m 07s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/a46f89cca6a14516a016e93ea2956688 : SUCCESS in 2h 08m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/97d92f1022874c08b49c2f5de4d87393 : SUCCESS in 54m 20s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/2b55b77f331f4bc08f2a860f16cc57d9 : SUCCESS in 1h 13m 45s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/14a80d39715646f481d7c70aa4f0156a : SUCCESS in 1h 48m 52s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/c3652ea242d342bea14494542c51abb0 : SUCCESS in 1h 04m 22s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f6b39ecf3f3947bc8e9b454cf91d2391 : SUCCESS in 23m 32s","accounts_in_message":[],"_revision_number":1},{"id":"a10b2d479ff9cb111e6b3823d7383016aa754743","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-17 11:26:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cab277cb669402d8e8a79a431a2e4e1256bf9239","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-17 13:32:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d20529720f6d4ca0a95295f70d8f9b57\n\n- grenade https://zuul.opendev.org/t/openstack/build/e78b0869e9e540f2bd6f0f1cb4a45e63 : SUCCESS in 1h 04m 37s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/e08df97f32514509a1f3d3eb5d95fee1 : SUCCESS in 59m 54s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1587766b032e499faf958ea5ad5fbd39 : SUCCESS in 1h 35m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/580634dbbd514688ab917985ecfa3ae2 : SUCCESS in 58m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d5c07a03e2e46798d4086d145fdca33 : SUCCESS in 4m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4a161f57f87f47c79abcc106606fa0c1 : SUCCESS in 5m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4ad3d7ea840446c7b0b4f3b60691d7f5 : SUCCESS in 6m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ac9620410884e85ad071aa8501c5579 : SUCCESS in 3m 21s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/061f7c8d3a434903be9ad7797fecde0a : SUCCESS in 8m 38s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c7abfe8813de4ae7b88b89b387277b3f : SUCCESS in 6m 14s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/578cbe9b3e544cab98bf6f4a36f9293c : SUCCESS in 51m 00s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/e1f11bf304014916830c1117d3e4334d : SUCCESS in 48m 52s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/8361590d47394532a3a687aed50ed9e7 : SUCCESS in 2h 01m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0a920beb119649a0ac5bf1c58aaea4c8 : SUCCESS in 50m 20s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/b9fc8effd11345b18588ce3149f28089 : SUCCESS in 1h 37m 00s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/7f39289ae8a142d88d4bd4bbede61464 : SUCCESS in 30m 09s","accounts_in_message":[],"_revision_number":1},{"id":"7079da273935fc3be364cc0565792ba76b8fb122","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-17 13:32:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"8374ff0b4415cff9ba1a2132afe619d1cc400635","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-17 13:33:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/198e1a15f17c4e40acdb23ebdac3b001\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37c6e27264ac435096691ffb732e6949 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"172cd49f3a5d9b9761a11a13b39104ec94ccb159","revisions":{"172cd49f3a5d9b9761a11a13b39104ec94ccb159":{"kind":"REWORK","_number":1,"created":"2025-04-30 06:19:56.000000000","uploader":{"_account_id":34034,"name":"Franciszek Przewoźny","display_name":"Franciszek Przewozny","email":"przewozny.franciszek@gmail.com","username":"fprzewozn","status":"OpenStack Engineer / SRE @ Opera Software"},"ref":"refs/changes/18/948518/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/18/948518/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/948518/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/948518/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/948518/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/18/948518/1"}}},"commit":{"parents":[{"commit":"265ad47616f1ef8d762448705a0637931e04c33a","subject":"Remove S3 credentials from debug log","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/265ad47616f1ef8d762448705a0637931e04c33a"}]}],"author":{"name":"Franciszek Przewozny","email":"przewozny.franciszek@gmail.com","date":"2025-04-04 12:36:36.000000000","tz":120},"committer":{"name":"Franciszek Przewoźny","email":"przewozny.franciszek@gmail.com","date":"2025-04-30 06:19:56.000000000","tz":0},"subject":"Fix vTPM metadefs.","message":"Fix vTPM metadefs.\n\nAccording to the Nova documentation[1], valid tpm_model properties\nare: tpm-tis and tpm-crb, not TIS and CRB. This patch fixes the\nmisalignment and also adds missing tpm_version property.\n\n[1]https://docs.openstack.org/nova/latest/admin/emulated-tpm.html\n\nCloses-Bug: #2106223\nChange-Id: I698dbd2946555c0a6455b1be17a6bf4a8d407ee7\n(cherry picked from commit df5f2b3c8693ce288209e86bee656e7f9c7b9566)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/172cd49f3a5d9b9761a11a13b39104ec94ccb159"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/172cd49f3a5d9b9761a11a13b39104ec94ccb159"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
